)]}'
{"id":"openstack%2Fnova~955687","triplet_id":"openstack%2Fnova~master~Ifa64e02eec8c4f08258450e39844e548b5c78e34","project":"openstack/nova","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-09-22 14:31:28.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-07-30 15:21:15.000000000","reason":"\u003cGERRIT_ACCOUNT_38234\u003e replied on the change","reason_account":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"}},"38234":{"account":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"last_update":"2025-09-22 16:57:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ifa64e02eec8c4f08258450e39844e548b5c78e34","subject":"Changing how placement updates are triggered in RT","status":"NEW","created":"2025-07-23 11:54:06.000000000","updated":"2025-12-12 06:18:49.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"3ad40ac58b1bf749ed3548936ffff918e8de250b","_number":955687,"virtual_id_number":955687,"owner":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-10-03 16:07:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 13:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-23 13:40:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-23 13:57:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-07-30 13:41:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2025-09-22 10:09:43.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2025-12-11 14:29:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-12-12 04:47:44.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d4dd46380ce796f0b156f36a453382321f6700c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-07-23 11:54:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0d2d57a173845079af42b24fb8191d0f6067f55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 13:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a57d4ff9f57448ab2aad765ceead446\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a15882075cd45b9b119fd65c8d9365b : FAILURE in 39m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8fff4f2722a4b178c8c845ed3fffb89 : SUCCESS in 37m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6b58220d9084619b83fa17157229663 : SUCCESS in 44m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e5eaac113497edbe105532c79713f0b6c05f58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 13:40:44.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13f967be64bd463cac81cf6c4986bd6e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/24d71b396c554bbba8738b89208d5d8c : SUCCESS in 37m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/316527b6bc404bc9afd1ae029b7276ea : SUCCESS in 1h 02m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a33f39ecbe145f1bbdd0eb0ad63ccdc : SUCCESS in 37m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/962abc6daa5246a2ba0c23dc4a6d8933 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10728ef7f8004c7e8381c416f9200be1 : FAILURE in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/705702a027dc4e7aba507ed29e84dfb4 : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a9a725b5b524d1983d44ce196582f09 : SUCCESS in 8m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2e14498f7077467abe41793a212125c7 : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31be20804a3548cfa8b84b211fb5651f : SUCCESS in 7m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/000c4ca25b7745c0add67e2471d62924 : SUCCESS in 29m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/affd991cd76f4eccb82cdf5c7656a7da : SUCCESS in 1h 04m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cd3dfd8611ce42df8f670f3a819cc8e5 : SUCCESS in 26m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/74012ce371ec4be3a7d18acd5f0b6ac2 : SUCCESS in 28m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a1adfd32108c4c4e9b5cb450fd90f7f5 : SUCCESS in 1h 34m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5a1f5d037744d5181d420f391e3b117 : SUCCESS in 1h 45m 27s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5466b9eaea50460da962a9386dece1da : SUCCESS in 30m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/de9aefc780804bb1af9a013ced96905d : SUCCESS in 4m 41s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12e166a21699439482854e66ae64bae5 : SUCCESS in 24m 30s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c59167a3972848748bb784a4d25e235b : SUCCESS in 26m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e56bb1e75cc443af90887d7e8ee1b2cc : SUCCESS in 1h 16m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0046146886f4b08b4a47aeaad72eb44 : SUCCESS in 36m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a17cd51406694eec93ad2644abb223e6 : SUCCESS in 26m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4dd9ad4e9c7d4ebba997b8b3ab945581 : SUCCESS in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e7fd0e7cd2102b18d73dcb621eb8fa289bab4df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-07-23 13:57:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"182e7a2a554d3a321e1362ab414d0283084c83c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-07-24 09:23:37.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8f1b015738b4fc2769491f02924c03a10cfe9dc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 11:19:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dc80832865c4498a1883a20e514285e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e94c8978c684d5ab4d090a7e58a0430 : SUCCESS in 31m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/922f141de9e547b3b073a51c37db9765 : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/307a0d3c957b439ca707a2d36dd4a4a5 : SUCCESS in 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6123c4417ecc799d0e50d67a5a7fea0b4e88dd61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-24 11:35:27.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26d25291b7e046dda60625da5e3602d3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2143c9e63eff4717ae9fe2d9ce458a96 : SUCCESS in 47m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0b19a155d5cf47b09596953e1f4ce8a8 : SUCCESS in 1h 19m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f00bcaa4b767487cbcfc1b3af8289de3 : SUCCESS in 1h 01m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8eef9249471c40f2988d8a0b0f79c4c8 : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf95a22bd3844b7eb3a0bd62885d8329 : FAILURE in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cfae3cf847c451cb057f243d2dceb33 : SUCCESS in 13m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d09942139fc549b997924d14ab3afd23 : SUCCESS in 15m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/501142830b454623bed8d64dc54a6d93 : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21153353dd714edaa9cc955c600c5821 : SUCCESS in 12m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5248ef69023a450cb779c4a25da34041 : SUCCESS in 48m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2630bfdd168b4bd6ba9d9442059ae89a : SUCCESS in 1h 40m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/abb3ec28979b4670baf286ce015f6179 : SUCCESS in 38m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ef7f28392c694b16883909181ffbba91 : SUCCESS in 59m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9b6a4be8a31142649beb5fda591a2efc : SUCCESS in 1h 50m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/441554f5a58147e5826542bde9b3e8bd : SUCCESS in 1h 45m 39s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f271872e769d4b4ca5fb75c58c102f61 : SUCCESS in 39m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6e961438c51848a5b4d90a07e7406b09 : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cdca000960854cf5a88fe11cf8bd35c5 : SUCCESS in 23m 42s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ad59285023064d228cf1c6559a928d1a : SUCCESS in 25m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d776c37aaa54244ac8d89a3c51a848b : SUCCESS in 1h 05m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b6c6eaa04dc49ce9fd1e250f95af090 : SUCCESS in 1h 11m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b738a9b5a7004298be405363b073557c : SUCCESS in 14m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f01b39b20e6a41f79ead9635e6831e1a : SUCCESS in 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f20f11ca78d84167b4d2f0b53f512ecd0f939d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-07-30 09:47:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"94ec791e0e7a22ef82358167d30cc1b36fcea65e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-07-30 09:48:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"323226f053509afc8909517393ffc202be72fd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 13:30:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3725fdfad144b929816b867b08bd0ab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/121d43de6c0d42f0bbeef5acc4a25a0a : SUCCESS in 1h 03m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f06ae02d6ad94ec7b9c76eefc9a55abe : SUCCESS in 1h 39m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/283b5c9504a84384932aeca13f501d07 : SUCCESS in 1h 02m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/183ba17359c742e79c50e60dcde76101 : SUCCESS in 23m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49707007e4074be9906747ecbf894e57 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ee2383f1a104cdfa11cb2b472ba7668 : SUCCESS in 13m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39163b858ff54cbab186863ec125ed3d : SUCCESS in 12m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/889243d90cf34f91a2a4bea97f5505d5 : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc0d0bfaddf4f01800c924f1294d87a : SUCCESS in 11m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/461ea2c1b64848508efa2c8b6dc7c0ea : SUCCESS in 53m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9904931557744563afcbaf037858b96e : SUCCESS in 1h 35m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/03ad487e954940c182b497277b1054b1 : SUCCESS in 45m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1ed6705fd8e2458d9ed6224ada346619 : SUCCESS in 50m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c6d9deee585a45b6a45a48991eb89500 : SUCCESS in 2h 07m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/d68113d37bb1438e8e4d2a5bcff74b88 : SUCCESS in 2h 41m 58s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/50aa60d028a14720a3ba02db2ee5a30f : SUCCESS in 45m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ebe82fc7ee1e40449880e18fbb04a67b : SUCCESS in 5m 09s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/725e71a7d64644779757655a20be88d4 : SUCCESS in 24m 01s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a85a0e057cdf4087acb1fc82683f94ee : SUCCESS in 34m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ff545b591d34cd2bd399b207df406d1 : SUCCESS in 1h 23m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c271e18be5124dfb8585e0f05131fd59 : SUCCESS in 56m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/00f35c683d2f44b8af5726c8a9b452f6 : SUCCESS in 32m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/acc4ef79ddef47fe87b0a258883f9525 : SUCCESS in 14m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ae34e2f48566d2231694597383b4bb1b2297214","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-07-30 13:41:12.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":4},{"id":"a177423b78db069e13b57ee1cb2f6d838516fa62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 15:15:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2ebc72b4a3348b783e18847d52ca0c4\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/752ae328ed4a4bc38f0cc67ac696bd98 : SUCCESS in 36m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1454be5b325742be8317686bedcc0236 : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0e0ac844d6449cbb774f9df1f8565da : FAILURE in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"07d324dd7ad6216a46f589683b86d63a7deeff58","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-07-30 15:21:15.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"45e91ddb555ded22ad1b6d41fc469a88cd4ead5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 17:36:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/866678b30d904531b6c00839520f0291\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dd1698535eb4778aabcada4540edee3 : SUCCESS in 58m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3965cf0bd412437fb7d452f7156a252e : SUCCESS in 1h 49m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/723f578d0255443d885dc7dcfba86cd0 : SUCCESS in 1h 04m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af12b338bdcc43f2b9d3231c773852bd : SUCCESS in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c04972b1234dbcabc5c2e90669c362 : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/141959b10ed14be6bc2e3cf7dff94130 : SUCCESS in 15m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d81dd943c9b1411d89a5df3b0f057108 : SUCCESS in 13m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c12571d9a37c493081318f89c69f3314 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278305a6a74d49049be688bbcd4655cb : SUCCESS in 10m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/896f2303cee048608fd7dc72665540da : SUCCESS in 1h 09m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f6f8f25830dc4bb4959f2e2f5f6e8abf : SUCCESS in 1h 56m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d63ab1fe918b437d825de57bf0b7b3db : SUCCESS in 43m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d3c2bac540364cae8a3200617aadef33 : SUCCESS in 50m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b2a521da080c435abb436c66cd256d48 : SUCCESS in 1h 58m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/a035744b966e4c638a0a2ec47d38524b : SUCCESS in 2h 02m 32s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b34deaef391f4e5ead26e6686446ea66 : SUCCESS in 36m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/00f8bab8b6f04d20add62ca4b4674930 : SUCCESS in 4m 48s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b98234b30274a0fb3523124cb4a73e8 : SUCCESS in 25m 33s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7ff639f2ea7a4d5f8683e5bf8210899d : SUCCESS in 54m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5afb4dd490574ea7ae84051db902a191 : SUCCESS in 1h 24m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da2f893c9d08460cb7e607ab56b2d5c0 : SUCCESS in 57m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ffaba889ecbd4524855c6801a99c66e5 : SUCCESS in 26m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b5f7d8b9d79e44ae94ebfb5e3d45f67a : SUCCESS in 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3bf2c1ab6201ebe4be66677ea2b3904f18e73356","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 18:03:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea9c4c732abd4aa6b47d0ce154128da2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30bdd4775ee144ad9e274fa304187467 : FAILURE in 37m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ddaeb3da65de43f0a4c6e1ad98b53b2f : SUCCESS in 46m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4fdf52f7c57643d490ddf2831f259025 : SUCCESS in 49m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c84a7c68b5083fcd501b5f5ecb446a2dbe0cf30","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-22 10:09:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8ec3cedb005264753249f4fcb16bdb60e5395ecf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-09-22 14:31:28.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0f2b38a030ff49ef8b3405d9a6df297eab332461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 15:24:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b8dabc676964919a46cc7e8cc0c55b8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fbef9dcbb0be483fb6887435f998d09c : FAILURE in 33m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d34de35e0fa64c9aa3a5e0bcbce8163f : FAILURE in 28m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/10aacb6b069544f0a62f0cb859c8b7f4 : FAILURE in 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a82a1ae0b4ed84a62f1a4eff30e01ce763e46547","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 16:57:23.000000000","message":"Patch Set 5: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/459987e0528f41c9bd55cafc4e8e740a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b4b5b45f8be4e59b43243b3cf1c5d10 : SUCCESS in 48m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/295b4216c21c42738a6c65fa1d22c6cc : SUCCESS in 1h 30m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434edbe980094da9a3b00d2b508800d8 : SUCCESS in 52m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb11e93778564c529da64e822a63b8df : FAILURE in 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7eed297d90845c8870666567050286b : FAILURE in 8m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a92b4236aea04d7ba15cd756a2e26a4f : FAILURE in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a00ac1424a94404891c2392d8da0c674 : FAILURE in 13m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27978335d8a240f6b55b16e92cc33035 : FAILURE in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097195719e7145de90e9bb2dfccab174 : SUCCESS in 11m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3c055f2fe017403b893c55e280187c1d : SUCCESS in 1h 00m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e46dd64b30d646658b3c3ba23c70d9dd : SUCCESS in 2h 13m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3d5ed362e0484b28b423a212df1b4d15 : SUCCESS in 44m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c17f3387cf3f4e7faa7924f5657894bc : SUCCESS in 51m 14s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/139bac2174ec425c899a84fb313bd59c : SUCCESS in 54m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a47bcfa979194dc1915fc0f19b91fa1f : SUCCESS in 1h 48m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/cccf6c29b84a421a9dc65672c633543d : SUCCESS in 2h 04m 49s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b579d6a4461c4adfbe4222156db8bfd3 : SUCCESS in 41m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a1ec7cc506a443bb99c0c92235d83500 : SUCCESS in 5m 49s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/891ef7e8ed98437894a270a2ef092117 : SUCCESS in 29m 05s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/04bd6265dc1b401b9a1fa9d8ec66adff : SUCCESS in 33m 54s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/31b7a721abdc4e4d913c20baa60c7d89 : SUCCESS in 1h 25m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1fa938416945339209264318909426 : SUCCESS in 1h 05m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/173e249e76f946cbaf05da1c6ccbf1ed : SUCCESS in 30m 12s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b44f60ef46c042db992278a00a23904f : SUCCESS in 18m 33s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/0a90595b8a44434389181b8d6417a144 : FAILURE in 14m 09s","accounts_in_message":[],"_revision_number":5},{"id":"46552c89e049341a6c44adad9b2842dc4c98f53e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"date":"2025-10-03 13:35:31.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"ddf75f5c8b9446df64b57f9a25b0b6b273ebadcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 14:13:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dfe3c983ae84230b480596210070a44\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b866c7717d62455ebb216e94bc8a1f01 : SUCCESS in 32m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/289b86f108ca4ff382c69a52ff5b985a : SUCCESS in 33m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95931378921f4befbb0433254c3dabdb : SUCCESS in 18m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"54cc08ff28b4e09b6fb251fddfd59b87f8eec750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 16:07:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf542fde3c64e4c89e37641bd734998\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94c958c4d144448ebdf66f87f3b52d6d : SUCCESS in 59m 05s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f6d02f3c00954474a03272574aca47a9 : SUCCESS in 1h 46m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78a35e3cba0e454aa870731715368449 : SUCCESS in 56m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f68840007ad44a2faeb47c219397857b : SUCCESS in 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d108060bb76c48da8d9c1b5068e7e0a2 : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c23c62b6ef34bfc974db4ae8e5732d8 : SUCCESS in 12m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26f13762bf154e9a8cb19f6f4ad16ba7 : SUCCESS in 13m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/496b3e77beaf4342b5c153bf4bf7a4d2 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13036031a64e4773a8c7eec239fee879 : SUCCESS in 13m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ddefdc03d4d9445eabcfab1e73a0a0c3 : SUCCESS in 54m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0d60c45e75144e2a2d0b8852bdd050d : SUCCESS in 1h 44m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/eed2eb90513a40e3a4bc6c70806f1a27 : SUCCESS in 46m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d5361923e3a54b8ab1b1c7e8412e7ac8 : SUCCESS in 55m 50s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/48b413b98cd740c39e6786f8ddb4dabf : SUCCESS in 1h 13m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/27be991ae8d74b54b3bfa7453558bf49 : SUCCESS in 2h 17m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f85abef96fd4bcf813d6ce29a846a55 : SUCCESS in 2h 27m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9342d974325c48808a7e8bf74e326fd4 : SUCCESS in 50m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a0a395efc26e4b52b2d644b78d0cf218 : SUCCESS in 5m 39s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3cfeb96ce05d4a0090c6a09c2e04c0c6 : SUCCESS in 31m 01s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c50d56f370ad4441b5fed8d2f6c581b5 : SUCCESS in 33m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd4e3570c1b34b88bbdd8ecee7e801e3 : SUCCESS in 1h 28m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34a5e21699f940dbb77c97b926c54865 : SUCCESS in 48m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9f16d30071c5423fbd808ae0f182a9b0 : SUCCESS in 23m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/99c5f5d758d94800b8e5d7ebf022c20d : SUCCESS in 27m 24s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/3b2ab0e9cdf4439986656b4dc6b9f6d9 : SUCCESS in 17m 02s","accounts_in_message":[],"_revision_number":6},{"id":"dc45d195da68db6d854fe7ae2e3e56d57c74ee53","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-12 04:47:44.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-955687-j9wfh/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"3ad40ac58b1bf749ed3548936ffff918e8de250b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-12 06:18:49.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-955687-czwzz/index.html\u003e","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"99107038aaac513de2c151a4d244dadcccfb4934","revisions":{"b07a020a8093c0fddfbe727a12c703de62f7706e":{"kind":"REWORK","_number":1,"created":"2025-07-23 11:54:06.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/87/955687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/955687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/955687/1"}}},"commit":{"parents":[{"commit":"6053bb27228d5f43a3112fcdece5c01229feef31","subject":"Merge \"api: Address issues with server group APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6053bb27228d5f43a3112fcdece5c01229feef31"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"subject":"Changing how placement updates are trigger by seeing if a full reshape has occured and whether any changes to the provider tree has taken place. Closes-Bug: #2117701","message":"Changing how placement updates are trigger by seeing if a full reshape has occured and whether any changes to the provider tree has taken place. Closes-Bug: #2117701\n\nChange-Id: Ifa64e02eec8c4f08258450e39844e548b5c78e34\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b07a020a8093c0fddfbe727a12c703de62f7706e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b07a020a8093c0fddfbe727a12c703de62f7706e"}]},"branch":"refs/heads/master"},"756c4bc6939d49993c72a2268912e9a5e793a524":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-07-24 09:23:37.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/87/955687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/955687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/955687/2"}}},"commit":{"parents":[{"commit":"6053bb27228d5f43a3112fcdece5c01229feef31","subject":"Merge \"api: Address issues with server group APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6053bb27228d5f43a3112fcdece5c01229feef31"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-24 09:18:33.000000000","tz":60},"subject":"Changing how placement updates are triggered in RT","message":"Changing how placement updates are triggered in RT\n\nProvides a more efficient way of triggering placement updates by\nusing checking if a Reshape is needed or if the driver has\nupdated the in memory provider tree resulting in a speed up to the\nupdate_available_resource periodic task due to reduced API activity.\n\nCloses-Bug: #2117701\nChange-Id: Ifa64e02eec8c4f08258450e39844e548b5c78e34\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/756c4bc6939d49993c72a2268912e9a5e793a524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/756c4bc6939d49993c72a2268912e9a5e793a524"}]},"branch":"refs/heads/master"},"1f7d6c25624e3a8d0291a2377ac4acd7fa03026e":{"kind":"REWORK","_number":3,"created":"2025-07-30 09:47:06.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/87/955687/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/955687/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/955687/3"}}},"commit":{"parents":[{"commit":"6053bb27228d5f43a3112fcdece5c01229feef31","subject":"Merge \"api: Address issues with server group APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6053bb27228d5f43a3112fcdece5c01229feef31"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-30 09:45:41.000000000","tz":60},"subject":"Changing how placement updates are triggered in RT","message":"Changing how placement updates are triggered in RT\n\nProvides a more efficient way of triggering placement updates by\nusing checking if a Reshape is needed or if the driver has\nupdated the in memory provider tree resulting in a speed up to the\nupdate_available_resource periodic task due to reduced API activity.\n\nCloses-Bug: #2117701\nChange-Id: Ifa64e02eec8c4f08258450e39844e548b5c78e34\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f7d6c25624e3a8d0291a2377ac4acd7fa03026e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f7d6c25624e3a8d0291a2377ac4acd7fa03026e"}]},"branch":"refs/heads/master"},"64d53ae42d86cb4c704d50d8f3b881bf31962fa0":{"kind":"REWORK","_number":4,"created":"2025-07-30 09:48:54.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/87/955687/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/955687/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/955687/4"}}},"commit":{"parents":[{"commit":"6053bb27228d5f43a3112fcdece5c01229feef31","subject":"Merge \"api: Address issues with server group APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6053bb27228d5f43a3112fcdece5c01229feef31"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-30 09:48:41.000000000","tz":60},"subject":"Changing how placement updates are triggered in RT","message":"Changing how placement updates are triggered in RT\n\nProvides a more efficient way of triggering placement updates by\nusing checking if a Reshape is needed or if the driver has\nupdated the in memory provider tree resulting in a speed up to the\nupdate_available_resource periodic task due to reduced API activity.\n\nCloses-Bug: #2117701\nChange-Id: Ifa64e02eec8c4f08258450e39844e548b5c78e34\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64d53ae42d86cb4c704d50d8f3b881bf31962fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64d53ae42d86cb4c704d50d8f3b881bf31962fa0"}]},"branch":"refs/heads/master"},"e1c79b47cdeb37ca0152fdb0d9eeb10c282c8a84":{"kind":"REWORK","_number":5,"created":"2025-09-22 14:31:28.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/87/955687/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/955687/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/955687/5"}}},"commit":{"parents":[{"commit":"6053bb27228d5f43a3112fcdece5c01229feef31","subject":"Merge \"api: Address issues with server group APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6053bb27228d5f43a3112fcdece5c01229feef31"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-09-22 14:30:49.000000000","tz":60},"subject":"Changing how placement updates are triggered in RT","message":"Changing how placement updates are triggered in RT\n\nProvides a more efficient way of triggering placement updates by\nusing checking if a Reshape is needed or if the driver has\nupdated the in memory provider tree resulting in a speed up to the\nupdate_available_resource periodic task due to reduced API activity.\n\nCloses-Bug: #2117701\nChange-Id: Ifa64e02eec8c4f08258450e39844e548b5c78e34\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1c79b47cdeb37ca0152fdb0d9eeb10c282c8a84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1c79b47cdeb37ca0152fdb0d9eeb10c282c8a84"}]},"branch":"refs/heads/master"},"99107038aaac513de2c151a4d244dadcccfb4934":{"kind":"REWORK","_number":6,"created":"2025-10-03 13:35:31.000000000","uploader":{"_account_id":38234,"name":"Ian Watson","display_name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","username":"ianwatsonit"},"ref":"refs/changes/87/955687/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/87/955687/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/87/955687/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/87/955687/6"}}},"commit":{"parents":[{"commit":"6053bb27228d5f43a3112fcdece5c01229feef31","subject":"Merge \"api: Address issues with server group APIs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6053bb27228d5f43a3112fcdece5c01229feef31"}]}],"author":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-07-23 11:53:58.000000000","tz":60},"committer":{"name":"Ian Watson","email":"ianwatson92@hotmail.co.uk","date":"2025-10-03 13:35:14.000000000","tz":0},"subject":"Changing how placement updates are triggered in RT","message":"Changing how placement updates are triggered in RT\n\nProvides a more efficient way of triggering placement updates by\nusing checking if a Reshape is needed or if the driver has\nupdated the in memory provider tree resulting in a speed up to the\nupdate_available_resource periodic task due to reduced API activity.\n\nCloses-Bug: #2117701\nChange-Id: Ifa64e02eec8c4f08258450e39844e548b5c78e34\nSigned-off-by: Ian Watson \u003cianwatson92@hotmail.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/99107038aaac513de2c151a4d244dadcccfb4934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/99107038aaac513de2c151a4d244dadcccfb4934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
