)]}'
{"id":"openstack%2Fnova~959181","triplet_id":"openstack%2Fnova~master~I2b1e76eee52dc9a15d1938a3384ebe30786ceb43","project":"openstack/nova","branch":"master","topic":"flamingo_rpc_alias","attention_set":{},"removed_from_attention_set":{"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2025-09-08 13:27:33.000000000","reason":"\u003cGERRIT_ACCOUNT_16207\u003e replied on the change","reason_account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}}},"hashtags":[],"change_id":"I2b1e76eee52dc9a15d1938a3384ebe30786ceb43","subject":"Update compute rpc alias for epoxy","status":"MERGED","created":"2025-09-02 13:40:40.000000000","updated":"2025-09-08 19:20:48.000000000","submitted":"2025-09-08 19:17:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959181-flamingo_rpc_alias","meta_rev_id":"f8f325e1605e0d9e77f5509d8e9155a44f0ce66c","_number":959181,"virtual_id_number":959181,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-08 19:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":2,"date":"2025-09-02 13:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2025-09-02 15:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-02 15:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2025-09-02 15:32:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-09-02 15:40:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-02 13:41:30.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2025-09-02 14:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-02 15:32:11.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2025-09-02 15:40:15.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2025-09-02 15:40:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08220a894edc0888f650b74337bfbe8dcd3734f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-09-02 13:40:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92b5bc6118aa7d75fefc1d235d10f75443d6d9aa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-09-02 13:41:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"501b547192db3bf431b20c4a36a3cd2dc316a33a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 14:56:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f89ddd654c34f0d81bb9a992015cbb3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dea336ef84e4442bb0387bdb8bbfa7bd : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a9347f9fbf140f6baaaed005fa411e4 : SUCCESS in 32m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5e0cb6931a7d4454a1aa754ab7fea740 : SUCCESS in 41m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5915f8117b93802bf290a433223484093dc8dfb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-09-02 15:32:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d2855a75f504ea406d81a12c31b0a0597a6f673","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-09-02 15:40:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2082172ef584880b02c3b18655a1749026a82d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 15:40:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d9d8b830dd2479692a314ff19f8ce7a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f634e478f77e4c9d8767236e14fb02cd : SUCCESS in 1h 08m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1e7ff4f899f248449aef0a09391e7619 : SUCCESS in 1h 57m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f316219185cc44778d54c87d94f40057 : SUCCESS in 50m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff4cc750749c4fe1ab750f290c7c7621 : SUCCESS in 24m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d1783bcf0c441acb907616e1613d2be : SUCCESS in 10m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/216aa28e6a604ed7ae7937601f709ed8 : SUCCESS in 12m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f7af3e4357240588b329f6700af471f : SUCCESS in 13m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4dd9bbc698bb4ac98ac8766a0495ecda : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b94d4fa6b9f40f5b7d9ef9f21ea4b78 : SUCCESS in 12m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d5a38e45eadd4dfcbd47f555cd9374b0 : SUCCESS in 1h 40m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad25337d7ca94146948e25ec914e378f : SUCCESS in 1h 41m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6cb5fd17be3c46c69492928e617edc9e : SUCCESS in 45m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9a13d7d0a6914e4c904ba34765f5fd24 : SUCCESS in 42m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bcbec23cbb704cca86fc48725a398486 : SUCCESS in 1h 03m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/dcd0970413ef454c8d9f82204d1d6965 : SUCCESS in 1h 10m 42s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/8556d8e4ab1a42709cff0e08d0e91c05 : SUCCESS in 30m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/67cd86ccbee24fd889f5205257a78b1a : SUCCESS in 3m 08s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f733433f9a3a4f03b3ac4aa9899413ec : SUCCESS in 26m 57s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1b62299f17994f1386d1691957e5ead3 : SUCCESS in 31m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cab083c8c304ecab75dc489960391b7 : SUCCESS in 1h 13m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0f8833eda60d4de18f1f37f420d1fbb5 : SUCCESS in 1h 08m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f14faa2bf12544f9a8809b054aec23a9 : SUCCESS in 26m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c4dfc48f38f8453cb1593c528fdac161 : SUCCESS in 33m 40s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8b914fa2731f4980bbb78d7558cec7b4 : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"126e3eac5945995ecb5005de116603389372a1f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 15:41:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34e90e80b1ad15838f4f0a2babffe208534bbf5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-02 17:58:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/441f0132fda14d24a505380839c6c316\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85aeda3dc4f34fcf8e17e5eab73820b6 : SUCCESS in 32m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3df69e7faf1a4668ad55237981538160 : SUCCESS in 1h 46m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cc5bd77c19a4aa1ae464d945cbce349 : SUCCESS in 1h 00m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18c3ca5da174d89afc91dca6c7892b4 : SUCCESS in 8m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7bb4a1413d9644f4beec980eef3d3a75 : SUCCESS in 8m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81888cabc25947c19d7a3175364efe60 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70eb1f67c3e4a0b9aa1d124f02552d8 : SUCCESS in 11m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1bbcc16011d1488bb1bd80a6707ec6ba : SUCCESS in 42m 59s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03000870a6934a538fd97194ee819a8d : SUCCESS in 28m 29s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/69a1034175ff4ccdb5a06737114926d1 : SUCCESS in 43m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c22344beba5a4f659a75ad28d3b5b9ff : FAILURE in 2h 12m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f44be9304e9476a910d3ad17990ec0c : SUCCESS in 1h 49m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/201cac99e04745589d8a20313e206aaf : SUCCESS in 6m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc9eddda8a004d428ea328cd3a973ed1 : SUCCESS in 2h 00m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0373d52d228e42e6ab7ba7939ab0919e : SUCCESS in 52m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fee165155e846778be2d2a8490cf554 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ba150453b54f93982712988f7d62fce731228aed","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2025-09-08 13:27:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67dd9577dc55730b7ae05726f030f8c742ce7f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:34:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ff5a6505d04fa1a4a33739f3d9c472\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d48809f46fac4d4a852352936494dd10 : SUCCESS in 1h 00m 05s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/49b8cba633784c6ea3bc023b652b7dba : SUCCESS in 1h 01m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf4dca378d3a4ce891138c224c9e0e55 : SUCCESS in 34m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd053fec037644dcb2703b56ac34fe65 : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6528ddc4498e45ffbcd126f513812a0f : SUCCESS in 8m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26814eaeed7b422993fa903c71c0d80f : SUCCESS in 11m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1c96e057cd34226ba085c627c289528 : SUCCESS in 12m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfd72e56f4bb47c7a4f672ef3d0e3bed : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6751ff4d3944476c9e50245fd5d117f1 : SUCCESS in 12m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d86c4de238544dfd9808fcfd8d0daa82 : SUCCESS in 1h 10m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fefd675e3d104c2eab5453dace02031e : SUCCESS in 1h 49m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b4ef4d4620ec4733ba4456925050b699 : SUCCESS in 47m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/03eb7185554e43a682f3cdc9e790b83a : SUCCESS in 50m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d9ebc288193a41089cfa3fb36ff9c39c : SUCCESS in 1h 31m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/c7169d9125064cd2aaa464dde49ee6b2 : SUCCESS in 1h 44m 17s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9f896df962c64f5ea59df9592de6fc1a : SUCCESS in 53m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/166d548ac7864e41af1dc1191e526cfc : SUCCESS in 5m 21s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88abae0a47db483186d8b445001a0cab : SUCCESS in 21m 24s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/05f1ce7116904f19a287c7c66f1f2f7a : SUCCESS in 33m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/62d9ee8c5f3948aca09eb7e2733e32ee : SUCCESS in 1h 30m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27c808c8c52345fa9f51090b91a1f064 : SUCCESS in 31m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8bbf498614644d6cb01d219d8a736cb6 : SUCCESS in 29m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5a351dde052341a4ab99daa3756113c6 : SUCCESS in 15m 34s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/be24a2f0a66549f8b7fb9d1681c655df : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e1eab4f857826104e0d377ac39bca6b8e5faf707","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a48ef091c249a90b60cf1513c54f966c05d78ff3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 17:05:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2ef33e4ee4a4f33b2731b728a0e8276\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eb4246b07b834cb58157190f00623060 : FAILURE in 38m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d39ff38637e4d9b8082671a680bb87c : SUCCESS in 37m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17d8b90398794439a63cc9e02e21d85a : SUCCESS in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e013b5e0ff6667f3a05167a724132068ef02983","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 19:17:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06be413c588d48c1a1eb3b7708369cab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18ce602be31b4b50b14d428f716f576b : SUCCESS in 1h 02m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e5a2eb94d338483598b501d4aec6c8bc : SUCCESS in 1h 53m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f2efc5dcfa14355aec082e829301f61 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a4ab58a99d4632a648615ccf1920bc : SUCCESS in 8m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42bf96a1d54b47e69bdd412b465fa4e7 : SUCCESS in 13m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8fee2b52d67f4873a594b3b06b59977c : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498e7e91b6bb482a8b21c30aacc23245 : SUCCESS in 13m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1a699add3b464686a294334d81f32813 : SUCCESS in 58m 01s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/427ce8782d4a4959a81554f693c1eb98 : SUCCESS in 29m 06s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/15a9b0dbd818466bad9151635ad61e3d : SUCCESS in 35m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c477aca369104a23826b59485f00b80f : SUCCESS in 1h 01m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/f99bf66abdcb4e6081e166b3b176980a : SUCCESS in 2h 17m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4e501101f1344ab58c50375b9db4c0fb : SUCCESS in 5m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/88983b1ff8594e288cecc9de6eaa6199 : SUCCESS in 1h 46m 54s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/efa704831332425fa87cb07bb2d6dfc7 : SUCCESS in 1h 27m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f666cf941ccb4872ad459d02744d09ae : SUCCESS in 36m 19s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f70ca4da00d54dea90022445d56b9d12 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d8894dda1cda786500bfe4437c63b0c40602bf7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 19:17:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f8f325e1605e0d9e77f5509d8e9155a44f0ce66c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 19:20:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7545614a41a2445f9907189d7cc8885c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d4cbca8aac4eceafc43fd590e2a0d2 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c87ab7573adb3d43b13c174e2f306b21ca7dad8","revisions":{"2c87ab7573adb3d43b13c174e2f306b21ca7dad8":{"kind":"REWORK","_number":1,"created":"2025-09-02 13:40:40.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/81/959181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/81/959181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/959181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/959181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/81/959181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/81/959181/1"}}},"commit":{"parents":[{"commit":"9c1d971f017eb30aa8a7cbf43bf399e7f0b7f0e6","subject":"Merge \"Reproduce that only half of the PCI devs are removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9c1d971f017eb30aa8a7cbf43bf399e7f0b7f0e6"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-09-02 13:40:03.000000000","tz":120},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2025-09-02 13:40:03.000000000","tz":120},"subject":"Update compute rpc alias for epoxy","message":"Update compute rpc alias for epoxy\n\nThis adds an alias for Flamingo\n\nSigned-off-by: René Ribaud \u003crribaud@redhat.com\u003e\nChange-Id: I2b1e76eee52dc9a15d1938a3384ebe30786ceb43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c87ab7573adb3d43b13c174e2f306b21ca7dad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c87ab7573adb3d43b13c174e2f306b21ca7dad8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
