)]}'
{"id":"openstack%2Fnova~961804","triplet_id":"openstack%2Fnova~stable%2F2025.2~Ibbd80324c17be6546ecd8b80f908ac5bbab5abd0","project":"openstack/nova","branch":"stable/2025.2","topic":"bug/2125052","attention_set":{},"removed_from_attention_set":{"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2025-09-22 11:00:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbd80324c17be6546ecd8b80f908ac5bbab5abd0","subject":"Add admin context when filling metadata on ImagePropsWeigher","status":"MERGED","created":"2025-09-19 14:21:38.000000000","updated":"2025-09-22 11:01:08.000000000","submitted":"2025-09-22 11:00:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961804-bug/2125052","meta_rev_id":"1d2392f99f0b1cd04e59a2bcf47b4e03121a7bcb","_number":961804,"virtual_id_number":961804,"owner":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-22 11:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2025-09-19 15:30:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2025-09-22 06:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2025-09-22 06:35:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-19 15:30:19.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2025-09-19 16:53:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-22 06:34:48.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2025-09-22 06:35:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"579490a494b1c0c1dcb3aa013aa2582a99fab7d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-09-19 14:21:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"feebe0a6e488a06c6e2b5ee41f96ca329249ec3b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-09-19 14:23:30.000000000","message":"Topic bug/2125052-epoxy removed","accounts_in_message":[],"_revision_number":1},{"id":"0daaf030733b27e6e91ccacce86e09b3c55cc71b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2025-09-19 14:23:41.000000000","message":"Topic set to bug/2125052","accounts_in_message":[],"_revision_number":1},{"id":"9fc868cf25f842d7780aec9e5f1b9ada1dcde1d3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-19 15:30:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5a42304f0b9a47834d9d49eb226d8706551a012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 16:53:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2efc96965334421a5d4affa64e3c9c3\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8586f7a4ec134cfa8e6eb5a3849a45e8 : SUCCESS in 1h 56m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f95c64cb0a94d0c811ea7e090d7ffb7 : SUCCESS in 58m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e3ebd4bdb7d45f5a5b0990c1a224410 : FAILURE in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea141c4a92b648268e6026add2383f06 : SUCCESS in 8m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c7e69255ba94da686cf53c5295cda60 : SUCCESS in 13m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88ecffad231e47f3b82910146baf2430 : SUCCESS in 12m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c4b85274a7d4013b39a9f13a8af889a : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccffb3b59824988ad7be57f31eee4d7 : SUCCESS in 12m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d1253d4cad494ad6bf32e2dfa7acea3d : SUCCESS in 1h 04m 15s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/27395a0de67b47dc9ea3702c2c341f2a : SUCCESS in 1h 51m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f0b5dc5c2b6941abb551c37af42e0535 : SUCCESS in 55m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1c21e80cb8ca4326aa2ec9a9e7d85991 : SUCCESS in 43m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/92e9eee0b2a54b6eb0b298d654c23a8c : SUCCESS in 2h 10m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/789a988bc3d24eb298fd5812e1d127a8 : SUCCESS in 1h 51m 14s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/9d6479f3f8ab4881a7c31688ea70d626 : SUCCESS in 54m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ade3153f4b13416591fb9f50b9bd8557 : SUCCESS in 6m 30s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70af58c7e2b548cabf07c164f72fb4a0 : SUCCESS in 27m 11s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/adbce17ba26a415aa0c836cdb565bf73 : SUCCESS in 33m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6429b5b9297647b38a3148f46fd08384 : SUCCESS in 1h 43m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cad6e706f0824280891afa06951ddb58 : SUCCESS in 22m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/eacc3438d46f40b997d56091be25417d : SUCCESS in 36m 36s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/9fb06425b9e64e30a5820b31a29dafc0 : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"338ac3e21b70e829076e9793bb498389ea5758d3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-09-22 06:34:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50b818ac66d05f8b648636be51552ddeacd1f6eb","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2025-09-22 06:35:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5d5ecd9b74ba7080aba428f2ef1ffb3210c270f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 08:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4a1c6a316c4425a96008ece619e40b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b0d1cc3eb7f7463aad855d90155ebd12 : SUCCESS in 1h 01m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fb2ef1c751fd4ebb828b01c8c6ef605a : SUCCESS in 1h 48m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e055723ddf8a4986846c85a6c6deb483 : SUCCESS in 1h 02m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c5fa0de63c943428ea07e93a94773a8 : SUCCESS in 25m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e745710762f348d28a394c2eb1ee22c6 : SUCCESS in 9m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2fe840224cb4531b10da63907a3d59e : SUCCESS in 16m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/834f351d1e7748d9929879c22767edcd : SUCCESS in 14m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da0a81a6700b4204b38cb6c45bed23fb : SUCCESS in 14m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1fa22e1d0d64c03bbd44ade1cfc96af : SUCCESS in 11m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f09a2df34a29444b8a84b98f0c94d527 : SUCCESS in 44m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed4efbccdb814060bf962aaee99a9312 : SUCCESS in 1h 38m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/56ea002d95aa4621b69a6d134ea981ef : SUCCESS in 58m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d919e43da6cd49b0b425710b9c52ee5a : SUCCESS in 42m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0bd76c5c31224c0bb26fe93098ff037d : SUCCESS in 1h 44m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/58350fdc2a304371a7db169e1c165ec2 : SUCCESS in 2h 11m 57s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/d1e8f9e6971641628fa600dc04c3268d : SUCCESS in 52m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/203f4100e41c44d2b9f1cc97dde2ff40 : SUCCESS in 5m 45s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/17df17f4838c400893d2fc6e2db415e5 : SUCCESS in 25m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/22fd57ac14c441c98c45163229d50da3 : SUCCESS in 34m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ba7b1d01b5d245bbb14987fa36e160fc : SUCCESS in 1h 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ed06fc003594a8888b0b9de25f4f79d : SUCCESS in 47m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2dc3e804438f4cf49d18f5ed3edbf424 : SUCCESS in 25m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6d569ff8e4694bea88dc50e9a3159fdd : SUCCESS in 31m 25s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/bf5eb7094f4f45fe8cbb1989ac6b7d78 : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6108747d8fcad3ddf7e78d7db6ff5c130cb2d4da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 08:52:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"745f6622918d50817dc9ec9e33ee455f36d5d558","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 11:00:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b174829fc3b647dba955536743d1c79e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0ff2c920d7d45f68478eb0117504797 : SUCCESS in 1h 00m 25s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f9f8f22a941d4bf693a61242cad543fb : SUCCESS in 1h 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b787a70830ef4d4cabd6314fa2b322b6 : SUCCESS in 1h 03m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6fa2496a8ea4e9b83ab86eadc12ac0a : SUCCESS in 7m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5504e23bc04840e2981c325dbebf084b : SUCCESS in 14m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb13bbdbfcd546d98ae0d05b52d3e13f : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e06c3eba9641f9a0088c5aa3c7b15f : SUCCESS in 12m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b113437244aa40c8a185031c023d939a : SUCCESS in 49m 35s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b1d4344c5bc43eca59c5a0095c18a40 : SUCCESS in 22m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/12298ceb15284cb993d64df1dca1ab8b : SUCCESS in 34m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/74f8b30c84a0470ab6abad7d0df195c6 : SUCCESS in 1h 35m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/f2f9758ab3f846daace947c28e3e4200 : SUCCESS in 2h 04m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ebff7986f5a74f81a07738af83fbd555 : SUCCESS in 5m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/534ea51297bd410b840f4715098a1612 : SUCCESS in 1h 44m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb806dd531ef48f08a13e2ad765489e7 : SUCCESS in 1h 22m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e4da0868daf40cda92d67973cf8433d : SUCCESS in 59m 23s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1c29322331ae4e5c814ad7932de36b7f : SUCCESS in 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ad30c97ccb9a781c6c03f8c0e7b1858cc7359e64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 11:00:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1d2392f99f0b1cd04e59a2bcf47b4e03121a7bcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 11:01:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a937ef5fb3124e12866a5aeeaaa60419\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bc326520d2247aeaefbf06bd9cba61e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9649137d61ff39be667fe093a2051fcca323ec8","revisions":{"a9649137d61ff39be667fe093a2051fcca323ec8":{"kind":"REWORK","_number":1,"created":"2025-09-19 14:21:38.000000000","uploader":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"ref":"refs/changes/04/961804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/961804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/961804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/961804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/961804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/961804/1"}}},"commit":{"parents":[{"commit":"1a669a36cecdbf93159ed95ebf05951fb6a99300","subject":"Regression test for ImagePropsWeigher due to missing context","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a669a36cecdbf93159ed95ebf05951fb6a99300"}]}],"author":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2025-09-18 14:25:50.000000000","tz":120},"committer":{"name":"Sylvain Bauza","email":"sbauza@redhat.com","date":"2025-09-19 14:20:18.000000000","tz":120},"subject":"Add admin context when filling metadata on ImagePropsWeigher","message":"Add admin context when filling metadata on ImagePropsWeigher\n\nCreating a new admin context as we can\u0027t reuse the RequestSpec user\ncontext in order to hydrate InstanceList with the full list of instances\nfrom the host and not only the ones from the user.\n\nCloses-Bug: #2125052\n\nChange-Id: Ibbd80324c17be6546ecd8b80f908ac5bbab5abd0\nSigned-off-by: Sylvain Bauza \u003csbauza@redhat.com\u003e\n(cherry picked from commit dedfc305ddc1ab23365698469bb1552edf0613e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9649137d61ff39be667fe093a2051fcca323ec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9649137d61ff39be667fe093a2051fcca323ec8"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
