)]}'
{"id":"openstack%2Fnova~966106","triplet_id":"openstack%2Fnova~master~I86fed47197015904b35f70a9e503413b1e9a6cc1","project":"openstack/nova","branch":"master","topic":"pre-copy-activation","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2025-11-05 14:37:51.000000000","reason":"\u003cGERRIT_ACCOUNT_38560\u003e replied on the change","reason_account":{"_account_id":38560,"name":"Zhan Zhang","display_name":"Zhan","email":"zhanz1.ius@proton.me","username":"zhanz1","status":"Software Engineer @ Bloomberg"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I86fed47197015904b35f70a9e503413b1e9a6cc1","subject":"[DNM] testing early activation","status":"NEW","created":"2025-11-04 19:52:30.000000000","updated":"2025-11-12 12:48:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"19cf1799d460698b21af58c77445c87a868594bf","_number":966106,"virtual_id_number":966106,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-04 22:20:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},{"_account_id":38560,"name":"Zhan Zhang","display_name":"Zhan","email":"zhanz1.ius@proton.me","username":"zhanz1","status":"Software Engineer @ Bloomberg"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 20:25:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-04 22:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-05 14:37:51.000000000","updated_by":{"_account_id":38560,"name":"Zhan Zhang","display_name":"Zhan","email":"zhanz1.ius@proton.me","username":"zhanz1","status":"Software Engineer @ Bloomberg"},"reviewer":{"_account_id":38560,"name":"Zhan Zhang","display_name":"Zhan","email":"zhanz1.ius@proton.me","username":"zhanz1","status":"Software Engineer @ Bloomberg"},"state":"CC"},{"updated":"2025-11-12 12:48:02.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b726071a5e97eb0f0e07460379fd0318fb99fbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-04 19:52:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"193c95ee78c45f6fbcc61814030636c8ca515295","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-04 19:54:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fac3608f6a4ccefc9133e8a1820dae5f2a9e8b0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 20:25:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a42c5fa0574f97be0e8928af59451e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01f7d8dd79914746b8326cbbbfd6c7c5 : FAILURE in 25m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/987f90eb10ab4eada75ebcd04666ead1 : FAILURE in 24m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3433e2920fee46ffb73ba2ab6b8fb1d1 : FAILURE in 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9e347d9419a5873fb2d0ac96cd07300907a0cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 22:20:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2037d7e68b14b19afd50d1990a3cb11\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea432d0b630845929f40eacc53ac76a6 : SUCCESS in 56m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/12a5658eb33f438988dc8bb71799c333 : SUCCESS in 1h 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2568ff416f6e4bb5b5863be5c63b3c74 : SUCCESS in 1h 06m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db56084fcce34504b08fdffd4ec9c6ac : FAILURE in 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/387346848fc54439943ad18f4a2fd618 : SUCCESS in 7m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82d19dfc447048208026acb4c0c27ffd : FAILURE in 14m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/386626b4b84a42fba2a694aeb229e294 : FAILURE in 13m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/545ebddf2eea4b0db4b13a17886b5633 : FAILURE in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/668ed5819b4e46e28a07fd93ed9f84f8 : SUCCESS in 10m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/03fb2e4cd5c14ac98a00d108137e397f : SUCCESS in 1h 07m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f335659eb1b1447b8af1d4a9fb7bbcfd : SUCCESS in 1h 39m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2025c3065dee422f8b8fdaf419983a38 : SUCCESS in 46m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8208e20897a344e18df3a59914fa794d : SUCCESS in 51m 18s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e5ee2e4b157740dc9aaae4aabcfc6c6d : SUCCESS in 1h 10m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5bd0398f358e4a40ba5bb0c726c59524 : SUCCESS in 1h 56m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/b90ac742457d4a00a9133bd092304b66 : SUCCESS in 2h 21m 35s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/f45fddd068484c218c3f119a71d9fb02 : SUCCESS in 52m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5d6770ddf503425c97b0e0bddc56713c : SUCCESS in 5m 34s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f7a13455cb54eb8b3129ddf47824d51 : SUCCESS in 28m 07s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f9749102914a4d65a1794e4143fc150b : SUCCESS in 27m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b600d79e299240e4802be51ca302ca63 : SUCCESS in 1h 03m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7196302bd9de448a8383a5f4f08cf4b8 : SUCCESS in 44m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/39132585a6544dc6b2620322312324d4 : SUCCESS in 32m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7030742c6dc843549c41cb22ba029327 : SUCCESS in 30m 27s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/36aeae0a3cf1428eb743c9148f3946d0 : FAILURE in 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a327c2fddd2fd829c96224c2961cd7a2abf1bf98","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-05 12:25:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39f2f491e09ee570fc46648cb560d30fe708907f","author":{"_account_id":38560,"name":"Zhan Zhang","display_name":"Zhan","email":"zhanz1.ius@proton.me","username":"zhanz1","status":"Software Engineer @ Bloomberg"},"date":"2025-11-05 14:37:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4052077f74e31b6350d2995e34e1e63132430ac2","author":{"_account_id":38560,"name":"Zhan Zhang","display_name":"Zhan","email":"zhanz1.ius@proton.me","username":"zhanz1","status":"Software Engineer @ Bloomberg"},"date":"2025-11-07 20:32:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19cf1799d460698b21af58c77445c87a868594bf","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-11-12 12:48:02.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-966106-rxj8c/index.html\u003e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f16ce085b35ee110953923d3d38e386315282394","revisions":{"f16ce085b35ee110953923d3d38e386315282394":{"kind":"REWORK","_number":1,"created":"2025-11-04 19:52:30.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/06/966106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/06/966106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/966106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/966106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/06/966106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/06/966106/1"}}},"commit":{"parents":[{"commit":"32f58e8ad6a7ff896cc6ae8a361e3a18f5b35c9a","subject":"Merge \"[func]Test with optimize_for_wide_provider_trees\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32f58e8ad6a7ff896cc6ae8a361e3a18f5b35c9a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-04 19:51:08.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-04 19:51:12.000000000","tz":0},"subject":"[DNM] testing early activation","message":"[DNM] testing early activation\n\nthis si related to bug/2128665 but proably not a correct\nfix. this is just testing the behvior in the ci.\nChange-Id: I86fed47197015904b35f70a9e503413b1e9a6cc1\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f16ce085b35ee110953923d3d38e386315282394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f16ce085b35ee110953923d3d38e386315282394"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
