)]}'
{"id":"openstack%2Fnova~966223","triplet_id":"openstack%2Fnova~master~Ib47890087110d460504df64aeed5206ded2e70b0","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-11-07 19:27:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib47890087110d460504df64aeed5206ded2e70b0","subject":"Fix test_simple_tenant_usage test","status":"MERGED","created":"2025-11-05 20:29:00.000000000","updated":"2025-11-07 19:29:04.000000000","submitted":"2025-11-07 19:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966223","meta_rev_id":"7815e0f59ba28af20d737ece36909f23b3d3aefc","_number":966223,"virtual_id_number":966223,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-07 19:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-07 17:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-07 17:28:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-05 21:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 17:28:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8206ae513e9d98349578f7a7192fd9425df91c8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-05 20:29:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9891f0316b569beb3d5187f3f7e07cc22b18a7ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 21:08:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da81b9b0736f4938ac4533166bf61dd2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d49c2dead2e345f891a8bbdc6ec88e6b : SUCCESS in 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e4d4bc5ef3546d487183d3123832f1a : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2457d84ac1624788a0d44b057f9d6c91 : SUCCESS in 10m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/637b0d8058f34e9abcff6bfa91aaeacd : SUCCESS in 10m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8e80cb6f85b4dc28bab2e91ce30aca5 : SUCCESS in 11m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd8c3f26d5e45238d7c7ad66113e980 : SUCCESS in 7m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a178b91cc596489cb4be0286f158dc50 : SUCCESS in 5m 14s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2cbe245234b428e871265c10123b5f7 : SUCCESS in 33m 28s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/591f693001f14be5b71d03a6f32837bf : SUCCESS in 30m 14s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b064cff446b347459b19b427276fe9a3 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dfce9b5910dbb134181c11775efebb913a22e626","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 21:32:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2878e50c6a41a0932b1b4153999a6c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/749bc742755c4f98a558117294ba7379 : SUCCESS in 14m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c85b5aa8c84472693b078c3609abdd2 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a1fd2318fc24441847aaa6bbf1103c7 : SUCCESS in 21m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f690d4af0e39ed0cfc1423cf207f3c6fd700169","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-07 17:28:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"769e16c417562ff90e6b54014363fca153eebe62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 17:29:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6764eb2baa04151a562bf8c93a5320928cd3948d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 19:27:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b95c7938303468b81ac2ca76d2736d1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65974c4a85ba4168916ebfc6c938990f : SUCCESS in 7m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb93d62efdf5493e9e64f0b257d6691e : SUCCESS in 13m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4887e0b79edd44b8be7b919b6b543ea5 : SUCCESS in 18m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277ede9c4a654debbab239386fe23b78 : SUCCESS in 12m 48s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca7289963aa44f0ba951c6ca02be14d1 : SUCCESS in 23m 46s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/28c1cbda5b194e5b986b54b1b4bd739b : SUCCESS in 33m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/81ac3dda4881465ab871fc5726d6d07b : SUCCESS in 5m 32s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/6ea62ee5eddb4a07b63c33ea5296cea8 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":1},{"id":"93983ca4dfe2e5c12a74e64fe5bfdc0787b8526e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 19:27:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7815e0f59ba28af20d737ece36909f23b3d3aefc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 19:29:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4593a2292a94799b04f836e69598eb9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299a987ec7984819bfed44cbf0f44a75 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4ddc76bd6541302b9d5fba3fa1edb47561fe747","revisions":{"c4ddc76bd6541302b9d5fba3fa1edb47561fe747":{"kind":"REWORK","_number":1,"created":"2025-11-05 20:29:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/23/966223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/966223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/966223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/966223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/966223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/966223/1"}}},"commit":{"parents":[{"commit":"9e5ad07aeeb9f14eba37e2cdea9377e7af48ef88","subject":"Merge \"setup: Remove pbr\u0027s wsgi_scripts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9e5ad07aeeb9f14eba37e2cdea9377e7af48ef88"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-05 20:23:48.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-05 20:23:48.000000000","tz":0},"subject":"Fix test_simple_tenant_usage test","message":"Fix test_simple_tenant_usage test\n\nAPI policy test_simple_tenant_usage test does\nnot send the start and end time in request\u0027s query\nstring. In that case, API set the current time to\nboth start and end times. So there is a chance that\nboth start and end times can be the same, and Nova\nraises an error:\n- https://github.com/openstack/nova/blob/9e5ad07aeeb9f14eba37e2cdea9377e7af48ef88/nova/api/openstack/compute/simple_tenant_usage.py#L258\n\nCloses-Bug: 2130703\n\nChange-Id: Ib47890087110d460504df64aeed5206ded2e70b0\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4ddc76bd6541302b9d5fba3fa1edb47561fe747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4ddc76bd6541302b9d5fba3fa1edb47561fe747"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
