)]}'
{"id":"openstack%2Fnova~966479","triplet_id":"openstack%2Fnova~master~I78270c51e1216178e758615380c005bf147b6614","project":"openstack/nova","branch":"master","topic":"nova-alt-config","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-28 20:37:24.000000000","reason":"Someone else replied on the change"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-05-12 17:26:52.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":["trixie"],"change_id":"I78270c51e1216178e758615380c005bf147b6614","subject":"move nova-alt-config to debian-13","status":"NEW","created":"2025-11-07 18:22:14.000000000","updated":"2026-09-28 20:47:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"503c5e720f93066d3c17d6a0b15d6f74934606ab","_number":966479,"virtual_id_number":966479,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-28 20:37:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-09-28 16:42:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-07 18:49:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-11-07 20:55:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-12 06:32:51.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 11:53:45.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-12 14:30:25.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-05-12 17:07:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2026-09-28 16:42:46.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-09-28 20:37:24.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"b88cc01d214cf89b1429748ba63be8de6e5de29a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-07 18:22:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d103d072027f3268bb64632902377c60d44e3e22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 18:49:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9192e0fcaeac4b79b21e7fe1558aed00\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f480ae89d2b4e708f2e6f0e12fcd572 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e08378e6b5c84128aa06e03ad2e7c6b7 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/424d46efa4cd48d2aac7153ccd426c47 : SUCCESS in 17m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8794a42b831e4f9fbea50f4b2e68d935cc98677","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-07 19:55:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4aa3f3d1470f35c6720fd9d16d478e5dc9a66b79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-07 20:55:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/868efd9145464e8990b838e7ed0616d0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba15dc1efd0a46e0ba14e2815bbb20ce : SUCCESS in 31m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/50f44712427f4bef83be9828c908ac3f : SUCCESS in 2h 04m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cb9e27ed2b6459fbd9863239e340de0 : SUCCESS in 58m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be329b541504f109e2e35143ea6f3ae : SUCCESS in 27m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e4d998337e488cae10c4eb02f8606f : SUCCESS in 7m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a0d42a60c45480eb7865bb29a66ce29 : SUCCESS in 13m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/beefe2b9a06b40aea35c96f838f2d281 : SUCCESS in 17m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e8f01cbe33148fa9dd207be7082cd6f : SUCCESS in 18m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc631e1c2b9466b9934262e063102d0 : SUCCESS in 13m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bceb406c20834eeab5e2999636f2993c : SUCCESS in 1h 03m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59e1403fa75844179056112ea598527c : SUCCESS in 1h 50m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5c1112acbb114b6c8ca820a873a12f92 : SUCCESS in 52m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b97bd3b604914d8183974fbb9c0f2ae4 : SUCCESS in 55m 04s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/52188d712bdc47febdabf2b7224a7645 : SUCCESS in 1h 08m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/32213b7e489c4a1baeea6331033950c4 : SUCCESS in 1h 57m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/466ccdfe868643049192e03d1fabf285 : SUCCESS in 2h 27m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4fd9c47779d54871b284500afbaccf35 : POST_FAILURE in 38m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fc4828566ba34f28b4613627f124dae9 : SUCCESS in 6m 46s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8676b4e04b54420828824ea6098e487 : SUCCESS in 26m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d6ffdb76654d4d72bdbb3212dfc9f5d7 : SUCCESS in 33m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/af3f3da2513246f5a18f0cd0eb597368 : SUCCESS in 1h 00m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/867808c5e9c24ca5898a18a9c9931483 : SUCCESS in 56m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a30d6ddc56c644de91a4e1708d7f892a : SUCCESS in 28m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/425e1478a926444ba09c4dcc64a9187f : SUCCESS in 33m 42s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b0044e5e0c734a0a8b5aa88d7b63c89a : SUCCESS in 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4c8208b15c4848505b1b62fde671da606f1fbffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-10 18:09:48.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca7556c7190a0c68c88cfa02848d945772d16ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-10 18:10:25.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"137171b74a1a0f4aeb1ae3f56f332a7b7214abda","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-10 18:10:44.000000000","message":"Hashtag added: trixie","accounts_in_message":[],"_revision_number":3},{"id":"a660ef0f482f1988b6e102242dbafc4c8c4f4319","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-11-10 18:22:31.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d2370f9bc92b3e3652a81858f29e86c034f09a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 18:41:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0a6ff7b9e8449f1b36af14b5ea5836b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/129f65a495104072935b21a754298368 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6752b680666e46c3beacace2540f9a78 : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd8e94b777374bee8cba886c024157ac : SUCCESS in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"159d458523be381e81a3bda041419847b99d13ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 20:36:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28ad31caa0d446ac8c0dd426bcb30870\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71e075b791d9464082601adb8a18b663 : SUCCESS in 41m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d2283055d1c14a19852903b5b738d863 : SUCCESS in 1h 39m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81ca5119e6bc473ca6995f85c601443c : SUCCESS in 1h 07m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07fb22962f1046158306261a0a2f6ed8 : SUCCESS in 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/507ced613bb940c3b11cabe962f46096 : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5916a87c00a24da588a9aadda8f80950 : SUCCESS in 13m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/10288aaa3b2c4b9dbd11cf8b6d4f14dc : SUCCESS in 13m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff0f0b21938e4b4ba8ddc82f975ece17 : SUCCESS in 16m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4327345e7cfd42d1a1de85a3c84f2dc0 : SUCCESS in 12m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1d8aa2b4989d4ed5bbe0be437d7a618c : SUCCESS in 1h 01m 15s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3efe527aaaef48389196dedafe129a0b : SUCCESS in 1h 57m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/93d2a0e344ed428384d0b1707997bf9e : SUCCESS in 52m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a9926330b99941acad7d7f4779b4d02b : SUCCESS in 41m 28s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e65c7795a93442d7a223a7303480ced5 : SUCCESS in 1h 09m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8188d8149d504ab48f779b22428a9401 : SUCCESS in 2h 01m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/f4b98091b004467fbda286f962366b10 : SUCCESS in 2h 22m 41s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/39fbd33bcd644d27bdf17d229a0e6660 : POST_FAILURE in 32m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9756b21faff54fdfb29a61e62df98a2b : SUCCESS in 3m 56s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a165097cd074748b94e10d0e1a16135 : SUCCESS in 21m 00s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/616aa704274c46a8b2759e36d4cd2367 : SUCCESS in 29m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f97753a1af904a94b57a303f7a070db4 : SUCCESS in 1h 05m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c696944cb5ec408e8ebe7f5176836191 : SUCCESS in 56m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/32c2721275fb488c8599c01b415b39ce : SUCCESS in 26m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c9bcef713fdd481b86bb9f6d6816af90 : SUCCESS in 32m 10s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8b9c8fe5c26e46a4892814a00455a2a4 : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":3},{"id":"3e562ab55d99535982699d40f2d9d35ed78609c5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-11-12 06:32:51.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-966479-7ffnf/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"0c0075ce4c1f7b7096879530c67d770b12e70b8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-12 10:58:10.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1824b04bc3171f09713232a00241d365c3b429fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 11:32:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7afdd681d6214b1da72b6178cac1971b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02a24e410aa04544bbd1a4e250b2c673 : FAILURE in 7m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fc7df2a27c3044dc9f2857c889a7b468 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b8f8751d8da54684a50b260bcec0cae7 : SUCCESS in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c20235aaeebfbb02a0f18c8f912d0c5fb0f6439","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-12 11:53:45.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I78270c51e1216178e758615380c005bf147b6614 .","accounts_in_message":[],"_revision_number":4},{"id":"50e54e16646f70ceb07c5401cdb3df66e15d9983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 13:31:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64759168c32c429c9d25e58613946305\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d5fa4053d4f401aa88192804fb79546 : SUCCESS in 1h 14m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5f5fac45d2a5489c95825005dab69387 : SUCCESS in 2h 04m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a92f364fcce4bf1bc03a2d08ced9d4c : SUCCESS in 1h 06m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a67b5e9a4a24680a7b2d1adf81b253a : SUCCESS in 1h 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b56862ec4f42d2adbbdb0ed1c288c1 : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f0b934c234544fd8d92c6bed69c5938 : SUCCESS in 14m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e4255ca7b97446392ea5e3be8b55084 : SUCCESS in 14m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/edef29341f9244e4a648725c795cb9ad : SUCCESS in 20m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc1380e4cf48468ab202756c4667d149 : SUCCESS in 12m 36s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3e51dc22cf0645d9a33f397bba6d88c3 : SUCCESS in 1h 05m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/61ab17d6510d4a119b50e9c529c20e33 : SUCCESS in 1h 46m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bb42281fb8554bb696dcc6b476a7f03d : SUCCESS in 59m 13s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/cb63d169b4684c58bac47d4f11089163 : SUCCESS in 32m 02s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/91a2c9dddad1471089ca83794ab871d2 : SUCCESS in 1h 02m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ab140c09a3204624a78fb201a1a00b62 : SUCCESS in 42m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/779ba54727bb49af9b1d3e60f3e018e0 : SUCCESS in 1h 12m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4324ef25ccd44cdf92adcaf47f1f89e3 : SUCCESS in 2h 08m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/65a6897cd514438ea9329c5e99723b04 : SUCCESS in 2h 22m 40s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/69b8847a340b4de9a71871141738b851 : FAILURE in 51m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/503ae215f7cb429b8f7b65146f287d0e : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e108d4ff68cd42f9837ff516f9d2c647 : SUCCESS in 25m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/177fbe85fd5a4945b238a167643797e3 : SUCCESS in 27m 39s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4b3ab3b3bcd54cef957279cde27041cd : FAILURE in 49m 46s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/43b592712bb64bb1bf29b0cd25c08765 : SUCCESS in 1h 27m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a712e9d469a2436fac20ae4862548db9 : SUCCESS in 53m 33s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dac4f7a3df054519a65b6f6729bca489 : SUCCESS in 30m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a49121c4137043d4a452a5402a4b7cc7 : SUCCESS in 33m 02s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2fd868cdd1b44aac869026a889d9ec70 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":4},{"id":"d8ff55a6680fca9ba2624524d42e64efba13d868","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-05-12 14:30:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8ff29d75a80511acdffbf4275ff182a5d54d7c14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 14:51:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c81b26d2810421aa527270988d5974d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0084ab4ebfa94ca1a0bbc9491ce0b9af : FAILURE in 5m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7220c8d602ea4b41879cd685973ac810 : SUCCESS in 16m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9ba8efdf568f4e678301ea6a48637b04 : SUCCESS in 18m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"86036d6a4b76a718acbd30c11b2e9f9c37a179fc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-05-12 17:07:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7ba8c72a8614800db82163783c49339f2419411a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-12 17:26:52.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42f88e72d52641c18d6a20ebf2b6922f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e2ebfe3377f94ae98be7c20efb27134c : SUCCESS in 31m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/07b242745bd7472a92b0601d4960a28b : SUCCESS in 1h 56m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64805d9615f446aeb4cffe2ceab761f8 : SUCCESS in 40m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3391d9d8f2f6411fbfa2673c372f296c : FAILURE in 46m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e185fc5a57c4552a57e16d36202f6f6 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dff5f2a118eb436f92d4def7ae75c6bc : SUCCESS in 12m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ec1f0f816ec416ca057db855514f379 : SUCCESS in 12m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de587156b677412092e09c9b8c916e7c : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0d1d5f53964aa5ac08d363dfb3dd9a : SUCCESS in 11m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6bcbfa807c994b48af4b616b2e326fd2 : SUCCESS in 49m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/71138087fc19427b9d43dd2d0760012b : SUCCESS in 1h 08m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1b007a233c574785992c09287d102397 : SUCCESS in 48m 38s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/20f2985d0fcb4050bd120a7daf1abae3 : SUCCESS in 33m 59s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1a02c74a145b4c6db15b535a85c61e0a : SUCCESS in 1h 07m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/186fbab207d3434089ab2eecf0fb13e1 : SUCCESS in 33m 13s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fb507a965aea48678e6e3114462aa887 : SUCCESS in 52m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ae3cfe21cf104648bbe4004f3fbabec5 : SUCCESS in 2h 49m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/a7ecae6bf7db4afd8f8206773d66a93b : SUCCESS in 1h 51m 06s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8ba07fdc336b43b9855a6abd3ac25cd1 : SUCCESS in 36m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ca40b9f164834b9da354e79bd76cb21d : SUCCESS in 7m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/03a56202406149d3b28b1ca237fb1f23 : SUCCESS in 30m 48s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d68223f7dcd44d57aad70af13e275cac : SUCCESS in 29m 43s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/fbfea6a957a1423199b9f16313fcde12 : FAILURE in 1h 00m 57s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6888e947f7f54dbaabcd66a2d45f89a2 : SUCCESS in 1h 26m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f617f37876cf4c858705106d2f033a34 : SUCCESS in 44m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c7777a1f343c49c9941bce14acb11f50 : SUCCESS in 33m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0c4b53315540408b8a970bfa9f91c4d0 : SUCCESS in 32m 51s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/beeedb9c48b4442495558d9b96a68bc1 : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":4},{"id":"043dc0b9f66e2f5f77fb1c296dd6097381ace2d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-09 12:46:08.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"7e80db6c7609c166e69f55cdd5d572878af955b3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-09 13:18:07.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-966479-knpbj/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"cfb527950682c966df167bc0e46babac7c6f9825","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 13:23:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05ffce63335340bab9610094f6fde306\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aec4d55fc27f4ff9ab85f394381b99e8 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/68bf9582413f43cea49ee611d3bb4f8b : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0d2cb84494524e7098880f377e21a23a : SUCCESS in 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c98bd7270be0b5a20a7640429495672467d3520f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-09 13:40:26.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I78270c51e1216178e758615380c005bf147b6614 .","accounts_in_message":[],"_revision_number":5},{"id":"3637120ae175b3a41a448f8ae4e7823098721332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-09 14:58:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/143a80327cd04fe7b93aa698dce232f9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54049b074ff84eeb8410bb3d7e8016a6 : SUCCESS in 1h 04m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ed7ea043f227412b8daeaf4d4faee165 : SUCCESS in 1h 52m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67e5b1d80da346a28bdef95dde49dc59 : SUCCESS in 1h 02m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39bb6fbb90f447089df39f991d10b1e0 : TIMED_OUT in 1h 41m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae7794d292343389310d85153ea297c : SUCCESS in 9m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4636e35483445c19abaef548fd01e45 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b8494af5f234ab5be5cb84fafcfd42e : SUCCESS in 13m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71064ab21a9a470b93247fff430e3e98 : SUCCESS in 18m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b9d3f1bb1f4133877e6592e0f0a216 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1ec36e77f2f4559afa7e73ea78bfe17 : SUCCESS in 9m 40s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/dcac485c716c4c75b6a01a0a4a008990 : SUCCESS in 46m 48s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ccdf7996fc19473a9e0fe22a63ece03a : SUCCESS in 46m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/70575dac0e0b412aaef20948b0358213 : SUCCESS in 52m 43s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/474e48d9ccbc45b68070e7ebea64c3bf : SUCCESS in 37m 17s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/117fb3d1b32248f6962541aa3d85d554 : SUCCESS in 37m 49s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ed73477bb0104fd99880efd183e2e470 : SUCCESS in 37m 10s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6f85d76c97934065b8ebb0b5ddea72e0 : SUCCESS in 1h 04m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f49445931a87451badd5c512ecc8da05 : SUCCESS in 1h 45m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/04c1cf59e24f4679bd5da069937ba9f5 : SUCCESS in 2h 02m 38s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b8b79d01e65b4d17a3fd84355bc6bbf0 : SUCCESS in 49m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/44cdb6e99c294ddeaad508366170fcc8 : SUCCESS in 5m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5b480b2ef7974b6ca68f93ab790ab0da : SUCCESS in 31m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c283c908c0424a8a8ee164152bbd0285 : SUCCESS in 27m 05s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e0270ef666624fd6b74d6069276889b6 : SUCCESS in 35m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/688e17583bef4299bdfcdc4958e3b5bc : SUCCESS in 1h 37m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8120a891b4634a81be85fcb799d00d03 : SUCCESS in 1h 01m 26s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d63a1cb89acf4cad81e6e87335eceb2c : SUCCESS in 36m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c566c482f52c4c2ba022dcdfa614c285 : SUCCESS in 37m 38s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3a44747a9a4d4d3298b80704440c87c6 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":5},{"id":"2fdaf3fa1f7b7289a168bd4df4c066d8938387bd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-10 12:24:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73f4aad7776f39522526381563349303ce2ebfb7","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-10 12:55:51.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-966479-xnfdh/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"a3a9b1161dc4dbefa7d96d963ee641ef88483efb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:11:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40b44526adeb4473b836f3fd82ec7a94\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/62d194bd3bb24ca2bc9b4573f39cf189 : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3c7b86ab4874af89810dccf17c64349 : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/62e3ec1c3f644d5b99f5ad6962cf8ca0 : SUCCESS in 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"16114f8597145d6350ac865e7317a253b405ab00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 14:26:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34e6b01c38204335ba08c5faa5eb29a1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17939a9f5a5c4340836338aab3a20c8d : SUCCESS in 33m 47s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/32510ee4f27741bf995d9ec36b70d5b5 : SUCCESS in 50m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/004f08758f1b40d9829629781665955e : SUCCESS in 33m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffb486d78595445fa330fc50701ee007 : SUCCESS in 55m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c120a557fc4911b95afa73c82f0ea2 : SUCCESS in 7m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aabbdac8d5f14c808a111ac13502c55b : SUCCESS in 11m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8973fd44937b4b2fa991df4190bd7ba9 : SUCCESS in 9m 37s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b189076ba85e4e53b26badbd3e68d4ad : SUCCESS in 20m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21fc46d52ff740c7a1f67a018c7684f0 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6636ac55414c4706ba51b91a464794da : SUCCESS in 8m 32s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/09336c99f8d04d4eb002a10cb2658fba : SUCCESS in 1h 15m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83a9032d7fd84f1b992a5da8cc9c25a1 : SUCCESS in 1h 55m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/53894c4584c7456db20c9957e47f6247 : SUCCESS in 29m 32s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/95dcd264c78a4fd6a7db2636e1f5b4d3 : SUCCESS in 43m 55s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/516dc72557674c95bbebe83b7184b925 : SUCCESS in 40m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7e57a5d76de8424ebd6c6c57c84c7592 : SUCCESS in 59m 50s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/bedbd231f56c42169d56a913c8cc447b : SUCCESS in 1h 10m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/46e4725a9ba34349944a0fa5ba7ec5f6 : SUCCESS in 1h 52m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/ddff9c0be26f4c38b9d4ee6880344197 : FAILURE in 1h 55m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/231138214005408b8f2e88f96bca2622 : SUCCESS in 57m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b2fb18baaa2b4088846c581dbe23bb72 : SUCCESS in 3m 50s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3c06036a42f5468387a062890e11114f : SUCCESS in 29m 37s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c2feba9e3cdd4887a0d60504d21fd4e3 : SUCCESS in 29m 05s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/429c91490024438c8a77890d6824458c : SUCCESS in 39m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a759f5ff9e74f9885d6f100f47118a6 : SUCCESS in 1h 39m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4420cf7e5a14d8ea21f5b10d708dcf8 : SUCCESS in 1h 06m 31s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4ca5655b8cdc4b7191c8829b66a2d2a0 : SUCCESS in 55m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3f34aedadd654bd78d8f364e07ceb47a : SUCCESS in 45m 06s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e5c29e9699af4186a445214d516b294a : SUCCESS in 10m 01s","accounts_in_message":[],"_revision_number":5},{"id":"50265b9c6c1798d2c3be29f526760d6e2fabd0f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-09-28 16:42:46.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"419680fde853f8d0c082bda89f8d1df3fa9f3a20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 17:34:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d61773111aa4fd3ab8e935eba8549a3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d08cc30e4cb84770860e191bc657025c : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/154c307f969b43f3886a21decf05f534 : SUCCESS in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"daa6bfea06ddc54ba0bba6af677dc56ff895853e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-28 19:20:31.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-966479-cstrq/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"b9fbedd4da7d918438d8d676fa77a1f6c21df015","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 20:37:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79ac97241caa47e6a664aec16611533c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/348abd4e5d604e6a96cec6ad6bf2f598 : SUCCESS in 1h 06m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f257a2bc0a3a497887aa7366ff49dbbe : SUCCESS in 1h 50m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/300f44d7f5ab4f67951aae442b95b312 : SUCCESS in 58m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/461b95a044774e5589a60b84c3c34d7b : SUCCESS in 1h 15m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ec8d736253643bd9b791dd01cd58368 : SUCCESS in 9m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa54caa2d8ee422fa19705320a4c9860 : SUCCESS in 14m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f0eefc0187f74eaf828bae0fdb04a4ae : SUCCESS in 21m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/235f8ce9c21a469a9c2873343cfaed72 : SUCCESS in 12m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd01f50259b34294a1d37c228a7a3f3d : SUCCESS in 10m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/139866d4a7a743fbae6be2f5efa3f250 : SUCCESS in 1h 18m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/079004cae49145f4801b5143e63bbd87 : SUCCESS in 1h 47m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0652b1152ba44f8f88b632b3597aab14 : SUCCESS in 59m 28s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0bb94ae5ec8f4c59a69dcbf8a85c656b : SUCCESS in 42m 38s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6812b0492c17470894945813bbb971ca : SUCCESS in 48m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f978113221d34f6e8f4ded8c2738ed01 : SUCCESS in 49m 49s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/76229b8611ff41f4b144ab59024c79a4 : SUCCESS in 1h 06m 18s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/852337e59a7b40d0875613e6930462de : SUCCESS in 1h 59m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/a4392f1b2c62489d9d30325dcccffcd9 : SUCCESS in 1h 55m 22s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/dabbe3e57e824e8f84355bbebe66854e : SUCCESS in 30m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f4668bbf5619489ca1ba7109501c1828 : SUCCESS in 5m 23s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fdd710a0491343478eacbc6a8d935717 : SUCCESS in 29m 31s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b3efae9b7bfe403aafe5a43015d8d067 : SUCCESS in 30m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/aaf73779444041179ebc6ab9d0544f74 : SUCCESS in 33m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4c30e1ae44d443108d2a5b1015b7af2c : SUCCESS in 1h 02m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9ac5bcb9ac54b449da14d9b7baa33f8 : SUCCESS in 1h 02m 39s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2840fd1727704d449af72f44271db4a2 : SUCCESS in 36m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/203c2a06ecaf4cd5945d848f323b2baa : SUCCESS in 37m 54s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/90f5a11a1467442ba4e992f12ea09715 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":5},{"id":"503c5e720f93066d3c17d6a0b15d6f74934606ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-28 20:47:05.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bfa39fae0f5ee4d3be9dc39341c53255eb725ba4","revisions":{"da0b20a525ad400c778cb52e39e3d9fe287a225e":{"kind":"REWORK","_number":1,"created":"2025-11-07 18:22:14.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/966479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/966479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/966479/1"}}},"commit":{"parents":[{"commit":"44c9d18b08597fcff49549950e2c840e1b627299","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44c9d18b08597fcff49549950e2c840e1b627299"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-07 18:17:50.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-07 18:17:50.000000000","tz":0},"subject":"move nova-alt-config to debian-13","message":"move nova-alt-config to debian-13\n\nThis change updates the nodeset used in the nova-alt-config job\nto use debian 13. This expands our tempest coverge to using\npython 3.13 in addtion to newer mariadb versions.\n\nAs mirrors are not currently aviable for debian-13 in our ci\nproiders we directly use the upstream mirror network in this\njob until older releases can be retired to make space.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/954653\nChange-Id: I78270c51e1216178e758615380c005bf147b6614\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da0b20a525ad400c778cb52e39e3d9fe287a225e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da0b20a525ad400c778cb52e39e3d9fe287a225e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"44c9d18b08597fcff49549950e2c840e1b627299","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8233a2490fcb6a7fe4c821c12095e7c35dfd3e63":{"kind":"REWORK","_number":2,"created":"2025-11-10 18:09:48.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/966479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/966479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/966479/2"}}},"commit":{"parents":[{"commit":"44c9d18b08597fcff49549950e2c840e1b627299","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44c9d18b08597fcff49549950e2c840e1b627299"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-07 18:17:50.000000000","tz":0},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2025-11-10 18:09:12.000000000","tz":0},"subject":"move nova-alt-config to debian-13","message":"move nova-alt-config to debian-13\n\nThis change updates the nodeset used in the nova-alt-config job\nto use debian 13. This expands our tempest coverge to using\npython 3.13 in addtion to newer mariadb versions.\n\nAs mirrors are not currently aviable for debian-13 in our ci\nproiders we directly use the upstream mirror network in this\njob until older releases can be retired to make space.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/954653\nChange-Id: I78270c51e1216178e758615380c005bf147b6614\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8233a2490fcb6a7fe4c821c12095e7c35dfd3e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8233a2490fcb6a7fe4c821c12095e7c35dfd3e63"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"44c9d18b08597fcff49549950e2c840e1b627299","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8f28cccd19269bc999f905130539edf734dc6f87":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-10 18:10:25.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/966479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/966479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/966479/3"}}},"commit":{"parents":[{"commit":"44c9d18b08597fcff49549950e2c840e1b627299","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44c9d18b08597fcff49549950e2c840e1b627299"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-07 18:17:50.000000000","tz":0},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2025-11-10 18:10:25.000000000","tz":0},"subject":"move nova-alt-config to debian-13","message":"move nova-alt-config to debian-13\n\nThis change updates the nodeset used in the nova-alt-config job\nto use debian 13. This expands our tempest coverge to using\npython 3.13 in addtion to newer mariadb versions.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/954653\nChange-Id: I78270c51e1216178e758615380c005bf147b6614\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f28cccd19269bc999f905130539edf734dc6f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f28cccd19269bc999f905130539edf734dc6f87"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"44c9d18b08597fcff49549950e2c840e1b627299","is_merged_in_target_branch":true}],"branch":"refs/heads/master","description":"Edit commit message"},"563f433a5e28b566b1ede410095ef670ecf85d92":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-05-12 10:58:10.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/79/966479/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/966479/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/966479/4"}}},"commit":{"parents":[{"commit":"44c9d18b08597fcff49549950e2c840e1b627299","subject":"Merge \"Migrate setup configuration to pyproject.toml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44c9d18b08597fcff49549950e2c840e1b627299"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-07 18:17:50.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibi@redhat.com","date":"2026-05-12 10:56:54.000000000","tz":120},"subject":"move nova-alt-config to debian-13","message":"move nova-alt-config to debian-13\n\nThis change updates the nodeset used in the nova-alt-config job\nto use debian 13. This expands our tempest coverge to using\npython 3.13 in addtion to newer mariadb versions.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/988214\nChange-Id: I78270c51e1216178e758615380c005bf147b6614\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/563f433a5e28b566b1ede410095ef670ecf85d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/563f433a5e28b566b1ede410095ef670ecf85d92"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"44c9d18b08597fcff49549950e2c840e1b627299","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"bfa39fae0f5ee4d3be9dc39341c53255eb725ba4":{"kind":"REWORK","_number":5,"created":"2026-09-09 12:46:08.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/966479/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/966479/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/966479/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/966479/5"}}},"commit":{"parents":[{"commit":"2a1a05dcf4acd0749a3db64a4ebe50c4134c0742","subject":"move nova-next to ubuntu 26.04","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a1a05dcf4acd0749a3db64a4ebe50c4134c0742"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2025-11-07 18:17:50.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-09-09 12:45:09.000000000","tz":60},"subject":"move nova-alt-config to debian-13","message":"move nova-alt-config to debian-13\n\nThis change updates the nodeset used in the nova-alt-config job\nto use debian 13. This expands our tempest coverge to using\npython 3.13 in addtion to newer mariadb versions.\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/988214\nChange-Id: I78270c51e1216178e758615380c005bf147b6614\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bfa39fae0f5ee4d3be9dc39341c53255eb725ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bfa39fae0f5ee4d3be9dc39341c53255eb725ba4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2a1a05dcf4acd0749a3db64a4ebe50c4134c0742","is_merged_in_target_branch":false,"change_id":"I6f5bcf88684e50ab9941e93a63720b998f6a4e6b","change_number":1004687,"patch_set_number":1,"change_status":"NEW"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
