)]}'
{"id":"openstack%2Fnova~969298","triplet_id":"openstack%2Fnova~master~I1e9a0d871398ef78fded868154e61f2c156d7c41","project":"openstack/nova","branch":"master","topic":"sdk_for_neutron","attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-06-15 19:46:21.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-06-15 16:47:55.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I1e9a0d871398ef78fded868154e61f2c156d7c41","subject":"Use SDK for Neutron Ports","status":"NEW","created":"2025-12-02 19:01:43.000000000","updated":"2026-06-17 05:30:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":13,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"5f2b967503cb1003ced9aaafda43d5757e7257e8","_number":969298,"virtual_id_number":969298,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-15 19:46:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35876,"name":"Max Harmathy","display_name":"Max Harmathy","email":"maxh@uhurutec.com","username":"harmathy","status":"Developer at UhuruTec AG working with Kubernetes and OpenStack. Open Source and Free Software enthusiast. Vintage Computer hobbyist."},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-02 20:00:13.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-02 21:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-09 08:09:18.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-01-14 10:47:56.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-03-24 13:33:10.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2026-06-03 11:55:16.000000000","updated_by":{"_account_id":35876,"name":"Max Harmathy","display_name":"Max Harmathy","email":"maxh@uhurutec.com","username":"harmathy","status":"Developer at UhuruTec AG working with Kubernetes and OpenStack. Open Source and Free Software enthusiast. Vintage Computer hobbyist."},"reviewer":{"_account_id":35876,"name":"Max Harmathy","display_name":"Max Harmathy","email":"maxh@uhurutec.com","username":"harmathy","status":"Developer at UhuruTec AG working with Kubernetes and OpenStack. Open Source and Free Software enthusiast. Vintage Computer hobbyist."},"state":"CC"},{"updated":"2026-06-05 21:17:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c631bb095e4181e73e0b6353cb73c1c353190b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-02 19:01:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"787bb9aadafee9eb82743dc44712b6fa602cf8cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-02 20:00:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":1},{"id":"8f049ba76a249a1d6edd530d964b2f39d36e5d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-02 21:30:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ae2c1b83cb3413085fe004e412d5c30\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b55dced4aca847688bb7fbe6ab7768cf : SUCCESS in 1h 01m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7b3ef45303dc409ba1bd71c3a3b87577 : FAILURE in 52m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40d1b7adb1f8453ea6cef9456f462638 : SUCCESS in 58m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/762f76f537674c05b3f7606242180c40 : FAILURE in 20m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e779bdee0104d078dee4e5a39181a74 : SUCCESS in 9m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2462d36bfe17468ea55bc70f0ab51926 : FAILURE in 14m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/902419a8f43549f382850911c11b461b : FAILURE in 11m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f3c4e713c2641489f7d486915501ba8 : FAILURE in 18m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e4598d9c34413f97ef0f8b1fd283ac : SUCCESS in 13m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5c04c0eea5b041318d708fb41847be96 : SUCCESS in 1h 16m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07575f53170c43fd9b9e0796c39fa3d1 : FAILURE in 1h 42m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6f738cb73eee4883991fe16092ebde96 : SUCCESS in 58m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5c36d779119b49a0b6727e5825b1668e : SUCCESS in 54m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/37f7e0118b3f44f1a741913a76265a20 : FAILURE in 2h 14m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/d03532c8767f4d8b867bbfe565da5563 : POST_FAILURE in 2h 10m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8a580a7818ca4bc287c3e5a82a785f01 : SUCCESS in 50m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/18960a2dfe99429eb41786145ccbb0e9 : SUCCESS in 5m 21s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5aa2be3e26394e5293cda84d4bb7f6b6 : TIMED_OUT in 1h 00m 58s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/89d9f2efbdee4176a6dfdadfe46862c4 : TIMED_OUT in 1h 01m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ed7004e7c3b54a869601b062daa2540b : FAILURE in 1h 26m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81074bef7f3c499db61dd11718cce335 : SUCCESS in 42m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/afebe016afab4a89a653a4035995baa8 : SUCCESS in 19m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ac23ec2911614320bdeb4e960d3a0e91 : SUCCESS in 30m 11s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/31f801165f6544579ba8ccc57c5fb14a : FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":1},{"id":"798fe953030d17d10fd131d51a7524059e226cbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 01:25:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f14d440019641bd81580c1c8346512c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2dda775cbfec48c7a4dc068c66ec45ad : FAILURE in 14m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb0f6156d96a4f4299ca9fa39237874e : FAILURE in 31m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5769e62eac174c75b25a3e6ae4350841 : FAILURE in 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0dd53453e48bfce9f56f01aa41938617992c77c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-05 10:54:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"aba0ffbcea00e5dcef05bde19b5acb6e2fa8a7a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 11:44:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/409385e3ffd343949fc7918dfc03081a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9cd8aca8c224eae973dca67083d4efd : FAILURE in 40m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d280fdb98c64b159423aff473fc10b1 : FAILURE in 41m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad836053b3ae4af18ebd05d559098058 : FAILURE in 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"292d2f647482feeb7f8952dd763767c264a4a9d9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-05 11:48:09.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":2},{"id":"115c35ca26fbe968cd664e753476fd1aa3b64ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 13:13:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e316fea3f814080adee3848aa0615b5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d52936a829e24c25bd3995a6e1312fa4 : FAILURE in 31m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f3d2b4344c284c95abbaafea7d93c8c2 : FAILURE in 54m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a7e37ad15514361ab9b1d8a96b036b9 : SUCCESS in 35m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0c040199dca4340a7f6fed737c7659e : FAILURE in 24m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b273ecd137c1496cb82e0f76b2109d29 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22b09cfef93a45fa88efa20ed57dd37f : FAILURE in 8m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5e2f1fc92934c1ba41fa687c9e35cc2 : FAILURE in 14m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ae77aad418004d7cbe0c5a16d0a2bc3f : FAILURE in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3719d4069f5c4f02a91af7ca0eefdd20 : SUCCESS in 7m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d4440cbc9264b38813d3fec15a6290a : FAILURE in 18m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f8a7c69ed3343fc9e090ab7fd3f4d9d : FAILURE in 2h 12m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cc7042f162ca42d8be64800f39f17e4f : SUCCESS in 46m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c7f0ad51942549f6864b589fc4eae7bc : SUCCESS in 27m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/183e72868e2c4ddf80ce669929b9d2c3 : FAILURE in 2h 02m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/eaa2d7497b934094863ee8950d2d6fb5 : POST_FAILURE in 1h 43m 33s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f4b84a4ea5a04ed6bf4dec9b25efdf50 : SUCCESS in 33m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/94462258e1194367ab16a20992ee75af : SUCCESS in 5m 51s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b78f97ccb7844b7cb58c9dff83efece7 : FAILURE in 39m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2f245aedd90346bd87c3e20d25ceaaa4 : FAILURE in 41m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c19d66fae114879a597ff15b90166d7 : FAILURE in 1h 24m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/544fd9d2a90e47e1ae827cb32a9abc15 : FAILURE in 48m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b3f6d9aeeedc49abaf9f64e5d3fb2b0b : SUCCESS in 26m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0767b45c0c8241a2979e2a06291a3fd3 : SUCCESS in 33m 10s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/5250928c5a334ad096168449a8121dab : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"98ec48938fa9c2c71c3e4370d5fbc0bdcf7e2059","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-09 08:09:18.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-ps9jz/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"14babe9b803ccb2847ed6fa2e54dc72e02cd5866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-09 12:06:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b22bdac18666b610b342d62c09062dd5ae23b6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 12:25:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a33907e4b6a46d48c21c05b1f55146e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f05a7dd7d9c4acdaf22d9a8ffda5518 : FAILURE in 13m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/62eae8af1cbf4a3e9e3ba8bbbe631398 : FAILURE in 15m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/172e0879ec114e01a599093b40f59591 : FAILURE in 16m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d231c822344f266cb9048ba136cd6e03f4a0e123","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-09 13:09:27.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":3},{"id":"81de7e75560ee5432f2fd9a1524ba483c9c14c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 14:18:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6695bfef1fda471b9ab8fb54876cd28a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7387e8d200184b54a62d00a8827b0093 : SUCCESS in 57m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7bef43ec3e3c419bb08cf6bf59c1310e : FAILURE in 1h 34m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/610abd1079eb4df4a40151ef10734b70 : SUCCESS in 59m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1677d52e3d349459e2ad23cce472b83 : FAILURE in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/830f6e8dec4c4bc7820538282085a22f : SUCCESS in 7m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/703714aeb8d84b3381d4a28ce4bc5a7c : FAILURE in 13m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d63ba12a1c0042cdacaddd4f9f4b01eb : FAILURE in 8m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57b8c25d74cc47728848c5830836a1ac : FAILURE in 18m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/070c6a627b2e4119ae41fbe1ba29cd5f : SUCCESS in 12m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/27000f69f09743a583f2923461660abb : SUCCESS in 1h 18m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb402e1705f1406cacb81d3a75b506a9 : FAILURE in 1h 04m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b514d874dbdc459e881fd158d59e9767 : SUCCESS in 53m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e214305a67bf45f3844d2f6c7cd2e67b : SUCCESS in 53m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/54ba153d78fa4bd1a72f56aaa2fcf2dd : FAILURE in 2h 05m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/ce81c431200942f7adcea914fe54fa03 : POST_FAILURE in 1h 13m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ee8c3828e0f94a04a579012e3d2c458f : SUCCESS in 31m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c2d75ca7bc8d4bc39b0e150025c5ef77 : SUCCESS in 5m 10s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fdd71dc338fb47588505acb46b1959d7 : FAILURE in 29m 55s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d6cd335570948f599c46dc14e4f27e6 : FAILURE in 34m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/38e0e09c8ddf42058f0b154992020f11 : FAILURE in 52m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c418eb15c924147ba0895a491191477 : SUCCESS in 59m 43s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4196f8c657424c719e97479c3391d459 : SUCCESS in 28m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/aad2c05531ff4682ac84f60d7e3d5b07 : SUCCESS in 32m 46s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/3a907454518c4a43b10b72740c552819 : FAILURE in 14m 39s","accounts_in_message":[],"_revision_number":3},{"id":"52560b591ca6cb81c4000a23b7df5cd01dc42023","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-09 22:55:12.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-nfdpb/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"5c791810114d3d908beb9087ce7af58b77388f35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-12 16:23:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"19efe0cf354552b2bef1a238001f5fc3d5f9497b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-12 16:24:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6d4afe3c62ca582f62fc4ed92d9be1b6aeb1170f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-12 17:27:07.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":4},{"id":"812269d708d569000dec50c5ee99ca1d761302d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 18:47:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/416b90d3eccb48c093fac49c3ffc4e8f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7308b873a0a4fb3966e099b2f7e0bcd : SUCCESS in 45m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b21741a3f5a04b6299a3bb58b61ceb9e : FAILURE in 56m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc5be3aaa2434768b6d0db2c8f475bc2 : SUCCESS in 59m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f5192c121b84a9bb83056037d89f25d : FAILURE in 29m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd9381119fb438d96b86090ad23032a : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f446e6a87985477eb86be2006e7af19e : FAILURE in 16m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46290260f14a42618e024cddbc3bac17 : FAILURE in 12m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1d7f2552fd294b04bb9ad163dabef56e : FAILURE in 18m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff6649fb289d47cba9ea4e79f00ddacd : SUCCESS in 9m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7afaa6980171463f9629e2c10499fd3a : FAILURE in 1h 43m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1db414f91da6404d87437e71d7aacb5b : FAILURE in 1h 33m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/997122fb29664ea78e7a56fb9aee47a0 : SUCCESS in 57m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/39e453cc17184f6e89ad788468bcaacf : SUCCESS in 51m 17s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/70f03625cfb6476a8f0b0d57ac947e3f : FAILURE in 2h 05m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/2d7d0c47a7d747e29cea5fb5493f3fb4 : POST_FAILURE in 2h 07m 28s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6cefb09e29ef49b0b26254d6c99b4af0 : SUCCESS in 54m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9ebf612e1e0d4212bef804fcc9138504 : SUCCESS in 5m 22s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/405d4379d8e8480b9e17e158de0a5dbb : FAILURE in 29m 22s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/677efe664d3d4a1790ae5f89ce5d986a : FAILURE in 28m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea810df2502e4ecbb60235267dd3b7e1 : FAILURE in 1h 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a6f4c2c29ef46bbafb88e6938e2d014 : SUCCESS in 31m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/86fc32a6d2064f39a6916110a9fb7479 : SUCCESS in 28m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/344aa871e67e4f6c9796bf896e4f51d4 : SUCCESS in 15m 59s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/ae25853a299e46f2abc47e033f5a548e : FAILURE in 14m 06s","accounts_in_message":[],"_revision_number":4},{"id":"da7e7793b6328d1830f269206b7f34fb5b0f94c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 19:05:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d8b543d4f4c46d099e5821ca1228d5f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6acddf256534090aed11666f149b30f : FAILURE in 15m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3efb26d100614462853f293c7ad3fe14 : NODE_FAILURE Node(set) request 5b4247c6be70478c883e979ca1e6e7d2 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd042f71104e48518e2fc8c57b398b33 : FAILURE in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5641ed4feb5a1792c920ea22ba242686160f965f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-12 19:48:42.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-bvz85/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"45f02f7ac1b55f8d341a3eba29ed7e8f1b5ba18c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2025-12-15 15:27:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"54c00b403f5d530342f396edb569b0d544590b58","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-15 15:55:11.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-tm2h4/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"901b3d4085f96c92d687dbc96bcdca85400a7178","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-15 16:26:24.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":5},{"id":"405f5d841d17c385d94f0a9242e8634d15f041c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 16:29:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f191f4be57e4371be165957ec0b0448\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0356a0e4be52475db462405bc0be3fb1 : FAILURE in 15m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4ebd2a282d740478ab72711e3996a57 : FAILURE in 36m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94ac45d6c7fe4b4abf156adce195c2eb : FAILURE in 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"41e50002df9122c5736e052553ecf394a9177299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 17:34:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e6420e2bfb0422fb6f75b7f777498ff\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea7fcc2114d3428387baafa125644628 : SUCCESS in 56m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/32522c7791874772aa222deef367afb1 : FAILURE in 1h 24m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90d4969c50a245008a98efd7c3abf70c : SUCCESS in 1h 03m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132a66e8ca5448158b8c8ad55ecfc0d8 : FAILURE in 23m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81ebb9408924c0c9c9918c1c45c9731 : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ac9d20fd5ab4208af4299ee809a543e : FAILURE in 9m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/28e0e64464714fc6bb02958f8b1ab26d : FAILURE in 8m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0516258892a4513a91aebecd4de2d0f : FAILURE in 17m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5189bbafee47b5acb11e588662b743 : SUCCESS in 5m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/10f5d8cff37f4507a598d6bb3c2a53e5 : SUCCESS in 1h 22m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a882acef76f440098cf319b31bcd4bf8 : FAILURE in 1h 48m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ae9f55a65d864c6b94cf87863f21c225 : SUCCESS in 49m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0d74320e2eeb41c5a833b8578cb5060c : SUCCESS in 44m 26s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fcce95da17454367acbdf6dcf5e37875 : FAILURE in 2h 01m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/1c0d624e9e6746e1899b3c9de740a4de : POST_FAILURE in 1h 16m 19s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/75479337c4c74bfdb411a679e43edc9c : SUCCESS in 33m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/92a91e3c1e2048ac80145199b9bbdd16 : SUCCESS in 3m 14s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/750a30e79b5c4677a3e1563e0e561429 : SUCCESS in 30m 10s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9ab37d9a9a8148609e9fe5b8b3c2c99c : SUCCESS in 38m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/016fb66a1cd2415fb48b83d00ae05e2c : FAILURE in 1h 30m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/225896beff834d11b4172f15c0a91e52 : SUCCESS in 1h 08m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fcb9aa741e5a41278926da9f8994e55f : SUCCESS in 32m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1d39f754232a4f41a57b9b1e92e49979 : SUCCESS in 31m 28s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/9b3b01dd5da042b3bf1b1621324bc7b5 : FAILURE in 16m 18s","accounts_in_message":[],"_revision_number":5},{"id":"8c67fb1ec02767f177bb8c292f06803955c35924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-06 17:07:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"974ac02d3aa59407873584ae9de4a30640d6dcbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 17:56:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d37ae6fb684266b67e3260468acba8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/202aa8a36ce44de491f60e18c5837799 : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/35260efcc7c54d84bcb90bd4d5446a96 : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/93a074da7bb94a518fca42e91a591d43 : SUCCESS in 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7cdc391f46e82115a47e3ff99efd03dc9f302ac1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-06 18:09:32.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":6},{"id":"bd731c2c863701b329b38f748cc472262343138e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-06 19:44:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e28c96bc49db4be1a51e18d3fcad8a7a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a8a921f5cda64eb9983534b1da159eb6 : SUCCESS in 58m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6d3a54972c9b42419c30550742d4ecac : SUCCESS in 1h 04m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e9d0945900f4ba8a022a42671758d60 : SUCCESS in 35m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f73686247826467b8e80758f9d7f9f7f : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ffeb373d85b4b37966462111d4ede3c : SUCCESS in 7m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce5f0b76ff9d4e3884e276f61817cc85 : SUCCESS in 10m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08dc599087c745338b634ef7bcc41c3f : SUCCESS in 9m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac67c37b3a8e4c7b8a9f1dd1ebc1a8d6 : SUCCESS in 16m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750334f513a446dd945ef6ccaed66009 : SUCCESS in 11m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d9c65506f1f442a4b3952a22942b5326 : FAILURE in 1h 15m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29716fde876a41e28b5070767584ae50 : SUCCESS in 2h 00m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d737ce5fd1484f69ad0c3036684b7b11 : SUCCESS in 51m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/fecfbd6454034315aeceaae8f9584f8f : SUCCESS in 52m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f537880237bc43b2b5381ccbccc70848 : SUCCESS in 2h 21m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/fc4630307d994cb08a4f304fbb45ca67 : POST_FAILURE in 1h 57m 55s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c44d0a81afaa4b4a8baf92f0dee06a5b : SUCCESS in 49m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9575542e7ad84820abf648f7590d0975 : SUCCESS in 5m 16s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c351d1c21f04b73bec5042765dd7c36 : SUCCESS in 26m 49s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0574c094a50d4db49201de251fb34c7f : SUCCESS in 34m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/94b6200816324affbf79ede3ef982c35 : SUCCESS in 58m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad06a2730e8c4ba1882e6f62847164a6 : SUCCESS in 58m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9d2d099ecc62452f94bbac86626f2b1d : SUCCESS in 15m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d51e9a82205c40f5960d82880cfd12ca : SUCCESS in 37m 04s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/d92b8ed8d35c45ef9fd2ed46cd7050a7 : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":6},{"id":"f3cb5b17e1d90ed42c9b06c1d4141baa1a76fd54","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-01-06 20:33:10.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-4r6tz/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"999e9c578e30922dbd734e4be8b875388a2e5fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-07 08:43:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4d4e9bb578c5ed84eb0d562113643a88e3d3ed25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 09:09:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33d7bda198494bbf9a50daaca945ec22\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57a07b7f0f0d46cf816c75da19d5c397 : SUCCESS in 14m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11e3382cf7ba4eca867bca56c8ed8fcf : SUCCESS in 15m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5a71c4215e1640b3ad050ad3e064b1e2 : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ce805d47be755195c7a0318ba1323a81f0ab7740","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-07 09:45:57.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":7},{"id":"5f50b089d747d0f2efbc7bd29a272b637bedc59d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 10:48:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3a5ca26ed4240d3a0e0467f93a4a5d3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6e9d2617caa4f19ace0f5e6f549316c : SUCCESS in 59m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a03c565f106c40b39c60a17fd85b9706 : SUCCESS in 1h 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc5b898e722c4e61b2af34e4db3bf3d1 : SUCCESS in 1h 01m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3a492cc3b0e4fbd98203d1acb64c78f : SUCCESS in 20m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f82f6ef8d284ee99d620053f461a3ef : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9666e9c432f4a28a5cc2c6df6ef439a : SUCCESS in 11m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b03701c42c6c4202a56a18fa511698e6 : SUCCESS in 13m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6aa4234a8d82444599adf5aeeba306f7 : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e2bdf7bdd1450ebd863e4041380034 : SUCCESS in 13m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e406aa1703bd492fb1dd49107bc779a9 : SUCCESS in 1h 14m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5c3a8d111144d69876a1dddd070245a : FAILURE in 1h 01m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5b62d44f380f44cdba9f76ec850230dd : SUCCESS in 1h 00m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5f273bd23bfe41aa9840e3627eca3cee : SUCCESS in 36m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0129a6fd853a41b2a1e90aa67a9209f9 : SUCCESS in 1h 57m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/36a39772d5894d93b0f14cf59ad651ee : SUCCESS in 1h 15m 20s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/af0b47d89ede429bb1115e0cef9f34c9 : SUCCESS in 52m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8b9c79571d7b4b7c93cde33f319acf04 : SUCCESS in 5m 25s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/452c89b7d3994d7eae9617600ccf99fd : SUCCESS in 23m 49s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dd05d104ae454d00a1bbb09c80db6f62 : SUCCESS in 28m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/eccbd7cbe126470fb22b708d5a6c3103 : SUCCESS in 54m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b957a52a1584942adb9492b82791d73 : SUCCESS in 43m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/35c9651168a942539bb18f84310a1f79 : SUCCESS in 30m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d2ce565b129941eca41b127ced281f22 : SUCCESS in 27m 31s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/f94c35e31ae345d2979dd71a6cc1da67 : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":7},{"id":"25392fc3f41ac839403c1426fc4eabb25d37f6b2","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-01-07 10:49:18.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-wr78l/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"9ec0030c5910f349b5ebbcca77861a3fc8da3638","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-14 08:28:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8d1736596d205f019c7eb24a2e8096c7f7d79062","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 09:18:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d03b8e663e44e01ae5529753332e87b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2327dd9e288642b1945075d7176fc86a : SUCCESS in 20m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfeb3253175f46588d55df1df2ace707 : SUCCESS in 14m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4f5f0ead468c4575950a4c435c5c69b0 : SUCCESS in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bba25256537518f34d72945dc36561ca9c59f429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 10:40:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e894e92fa3d144abb92705305127e23c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1961a2945e2f47d3b86ea3843f45cebb : SUCCESS in 59m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/83af7829fff941dd9fff39dd1bb7a1cc : SUCCESS in 1h 48m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a942553be328452993dfee0a2ab67412 : SUCCESS in 1h 08m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e552bf0ccc644048593191c1b8adb66 : SUCCESS in 26m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dff553531904d4e8580e2b158c40318 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a947f8627e0e49a08ea744ead1096870 : SUCCESS in 13m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9a887f1ae0c4ca4892436cb6ad748a7 : SUCCESS in 13m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5507bf6dece44b20a53d4489a2b71eb5 : SUCCESS in 17m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84a420e3a804d73bf7c4b0858bfd2f6 : SUCCESS in 12m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1158ceee0cc1413aa65f5461a57387d5 : SUCCESS in 1h 14m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ab304554a034bfaab6e172f5a0f43e5 : SUCCESS in 1h 56m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1df46846a97a42ff89b57b480ea6bd63 : SUCCESS in 1h 01m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/eeab3de4f35047c09af352ec6d57a24a : SUCCESS in 48m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b1a6f6444ea14ad68f75477544f4a5e9 : SUCCESS in 1h 48m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/fea10c3ef25147119c536f651a1425c7 : SUCCESS in 2h 04m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7dde0a17200f4331a6d762a17f90be04 : SUCCESS in 54m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f2d0ec3ce8484e86b268473eedd7d672 : SUCCESS in 4m 57s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c13826da4e764780830ed956c98b96df : SUCCESS in 28m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8f7045634fc148b2b025dd548a83de4e : SUCCESS in 26m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/90894898055f4f6887d510eb5638bb22 : SUCCESS in 1h 21m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ceaed95363c94f6489ece31cf9962c8c : SUCCESS in 57m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ac26da0ce3ae47e99192f46eda78ca51 : SUCCESS in 34m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ca8b4a9117ef4cde90d5c3c0d156d471 : SUCCESS in 34m 19s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/9e1e8f1ca5fb4127acc5bb6bc5e3c4c8 : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":7},{"id":"bbbe753f6a3ce9ec0e3849c37ac9993766c4d07a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-01-14 10:47:56.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7},{"id":"c887627acfe21d3b2dc7434e2a48d1cbc4a7c7b5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-01-14 11:54:54.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969298-shjrm/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"2b090111f8284c68f612f1a80beee6313bdc6dd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-01-28 17:11:08.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"8d7192249218c1746ed0b10fa3a3c8ddf4116bdf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-01-28 18:09:15.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":8},{"id":"12bd066acfbc4dd9aeb8a1ab01f04a68575e3c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 19:48:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c141d94ed9b64bc092f1e15c3cf72665\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b432a8ee0d854a32bc14dc6b2af8607b : SUCCESS in 1h 01m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/535ccf1e03b54121bd650ea224bac7d8 : SUCCESS in 2h 13m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7508d68554144dbbe5c07a8d01b5a40 : SUCCESS in 1h 05m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b424f74457d48b1a08a2a71cfe28c47 : SUCCESS in 24m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9210fb9d10a34ecfa6e73d373c050879 : SUCCESS in 8m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b3d62e343f04b72a8d6ca6835e29819 : SUCCESS in 15m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5d91e07ff02470bb08040576ed8264e : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518c96e3ba8d4ee6a1658ef1a7ffd8d3 : SUCCESS in 7m 08s\n- ironic-tempest-bios-ipmi-direct https://zuul.opendev.org/t/openstack/build/ec0c3653df714a938006cbece2a5c1dd : SUCCESS in 1h 18m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7de7640fe27c49fd86ff136759f7aeec : TIMED_OUT in 2h 34m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2562d701180e486da52fd5840356e9b9 : SUCCESS in 1h 06m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6fbf499ca05345d485b65634da4301ac : SUCCESS in 55m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/20543cc7274746899aca3db05fdee843 : FAILURE in 2h 30m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/00eebb4017f04f6a86921fd2c545b03b : POST_FAILURE in 2h 19m 51s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bfbac236b4c24339a42ba28c830dff1b : SUCCESS in 1h 02m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0ed318dc70b4494e8d7bcd520b1eb910 : SUCCESS in 4m 06s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca16abb4827a4d618e3bfd8151f0d9f1 : SUCCESS in 31m 01s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/41f9b5830fe34c1c9fccad215f48cbd6 : SUCCESS in 23m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/74d68ad88c5a4c2b8aeb5096fc4803a6 : SUCCESS in 1h 27m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49ca55bae81f4c528ca43470712bcfae : SUCCESS in 1h 07m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e82b55c90bf547b0b5474d6d875ffd53 : FAILURE in 11m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e69b7e1846bb45cd9a3cae63a9055c12 : SUCCESS in 15m 36s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/db34091809dd4536a38e96430bb26419 : SUCCESS in 15m 33s","accounts_in_message":[],"_revision_number":8},{"id":"3abd6b303f3ef8544aa25ce1b29b2a604b8abe10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 07:47:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90d32e19f2d448599e1370d887ce79ca\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/328f0d0efec74e95bb583b4094dc5140 : SUCCESS in 32m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4e22532afd44b1ebd0f14c31da7e290 : SUCCESS in 28m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0661d5b7ea954998b7c73595aecdde99 : RETRY_LIMIT in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"715c61b126f678d9370b120e95cf070ae11292b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-13 09:04:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ffd1b51fdbb20ca464a13106ca1bc220c9431f9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 09:26:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de524b0727f14a6e8af2fa3d65fed16e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2988bb3282ea4afbac9316c6becf7927 : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81faec6d06344ae7bd9380279ff8aaac : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/97910eeff1af407c8da56e993659d6c9 : FAILURE in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c299763f9fa84e4b2968a18bef48d6f6f10ed8a1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-13 09:28:57.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":9},{"id":"8adb6268adcf9437b2ea14988383484140c9bcbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-13 09:56:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"af8e42e2baaefeafb707ab98fab94f82e8724d12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 10:24:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ba9fdc029e147199ead31f53756af2f\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d69e4fbe8bbe48e5ad1e13315443ebfe : SUCCESS in 16m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f0a45560b4945e3b0c5292d184738f3 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ec58395fcfab4f039fd1234050247aa3 : FAILURE in 13m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e38ff8ddee955c6dd0c05a92f91ab3dc05e6eaf7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-13 10:33:36.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":10},{"id":"a1385cad5ef4a4d213561ef09460ed464cd21a7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-13 12:33:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1839e88b1779427683098e45a0f5f3ef\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f2d1ad752cbe4f88b2a510c3fa3c6f16 : FAILURE in 29m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a497d1f42d5345598b915c78aabe6655 : SUCCESS in 1h 47m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7aa6afd84e324052a335bd95975e3f8f : SUCCESS in 55m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a761d1af5037419b951ca864c2db362a : FAILURE in 1h 23m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5797d3413b9d41cca201cc107f02d3d5 : SUCCESS in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80c6738243b84f9f8558288a1248b426 : SUCCESS in 16m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/572e727e7f744132873db61dde3d87b2 : SUCCESS in 14m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bc01785ad68346a6b89bac2792065079 : FAILURE in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364fb1641f3548238155a72486f32823 : SUCCESS in 12m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5bb3fbe7ae824062a366e7677136f50f : SUCCESS in 1h 35m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3af09535c54e43509ba3fd5df158c82b : SUCCESS in 1h 04m 13s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ea2a18cff5764a12b153836fc400bf99 : SUCCESS in 1h 02m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/102d25663cbc43ae980f035ea51d3cb1 : SUCCESS in 41m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/937730e83444434795fe29d7767e0ee6 : SUCCESS in 1h 35m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/b72ab68b31514f1a91d28dfbfe4645ee : SUCCESS in 2h 28m 43s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d553e30dd483414ea816e1bdc303b173 : SUCCESS in 55m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dbe8a7927d2941d4b0738d0d6ebec1bb : SUCCESS in 5m 18s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5dbfc415a0ca4ed39bd06d089e12c644 : SUCCESS in 29m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0c1d77b48b644cbb88318c3f64bfd657 : SUCCESS in 25m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7b5cccc311249ea8d0ff0cc054739ea : SUCCESS in 1h 22m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ade5291baaf401786f2b5877bb81e90 : SUCCESS in 1h 02m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5d730cce148d43d28c1da7cf780af156 : SUCCESS in 24m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f74b9f003e9340ffa0795188d873ac6b : SUCCESS in 28m 41s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/e15801f26aa44d8ebb4403dc656ff3dc : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":10},{"id":"fa2d248741f353753e91b18cd84e8667f7076f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-03-17 11:06:28.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"703c963ab9d0a6c0d85e9410e0abb478fa6ca0eb","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-17 11:45:27.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":11},{"id":"21f67507fbab535a0e52ad2edcd2eddb44504e86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 13:12:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cba008c8530401d8cc79ed32592604d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4709bb2c8d114622956627df418adfde : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b810b6711294879ad2b7e0e8fec2bbc : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6826367bfe9e4bef9bb3b910db42bf20 : SUCCESS in 51m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"41df15f4238536511afa0134ac09d24e2210eb86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 13:56:29.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17cbf15f075c479c9e889c8230db2425\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2fa893694b914ac9a590e55db6de697d : SUCCESS in 45m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/415a5dd7dee5445398adf562d397ef8b : SUCCESS in 2h 04m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b200d4c82014b34b2d5a605258dc21c : SUCCESS in 47m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b0c54c163b947a99de9ec2b94715515 : SUCCESS in 1h 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4938cfd5856e49efa5825786867752b2 : SUCCESS in 8m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ef7f1029fd740a5aa5d81cdfea65746 : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ee0c52c3f7e46219efadbfd51ff94b3 : SUCCESS in 9m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/828dc282894b425ea7c507a347215443 : SUCCESS in 20m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b898f9ce5512448fabe32e9a0bc7abf5 : SUCCESS in 12m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5deec74e2d314caa9e258bd51f36b168 : SUCCESS in 1h 26m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8f0382977f2942a2a5a7df2810e50263 : SUCCESS in 27m 33s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b507e290cd4a4b498dab2357b77ab221 : SUCCESS in 1h 02m 12s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/53e5f7788ef848258ed0af80b56ec95a : SUCCESS in 56m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/57e8910de1f04bc492a06b15af8b6ff4 : SUCCESS in 2h 09m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/5675e9edd58c4b208f8c2df017d5119d : SUCCESS in 2h 43m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/3f4a721671be4038839fa19756234e0c : SUCCESS in 49m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/33389b4ef11a464d8f99f99d9ae1aabd : SUCCESS in 7m 04s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/899b6fe788ce453d95beca6204cba5e4 : SUCCESS in 29m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e30097bdf55444018ffedb0630160683 : SUCCESS in 31m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cc85f1eb3c1456892c12d443b825df0 : SUCCESS in 1h 05m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53fffe329db24fbd9c42c016e94b6b63 : SUCCESS in 50m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e5dc5cc9546a4252b6d1c59dd4245b02 : SUCCESS in 20m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/82a020e187d443498843c46c375ad3b0 : SUCCESS in 34m 43s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/647f4513064b4feab829a648089fb45b : SUCCESS in 16m 47s","accounts_in_message":[],"_revision_number":11},{"id":"f3b1cade00d81ef54866e823fafe85a94c2da607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-24 11:26:36.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"96f4e58cbfa03cf46ed8fa024568f4a85a03a54b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-04-24 11:43:18.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f598f737ee415b34660b3d87ecb41341fb72801c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-24 12:36:50.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":12},{"id":"8fc0de51ca359f6800c4b3a9139d308a19875028","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-24 12:44:13.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":13},{"id":"f15f240324ebe3a6befd88797f629f8d5e5a213b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 13:41:52.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7debe43dd77b4d9a9d73a35770aa0393\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6033a4cb3cac458d983a0af5e8384185 : FAILURE in 14m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f6838e41d8445d0816fda115c0e3f6b : FAILURE in 12m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3d9969a081804077b140ecb64d36d2da : FAILURE in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"115e6278888b37a25efad34d12c1caf59132ecde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-24 14:40:51.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6ff609890aa4264a867313882ad93ac\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/974946e928a548ae8562dec6cf6ddaec : SUCCESS in 1h 06m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ffb2e0017a72480988e6ac01321aed9c : FAILURE in 1h 33m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5c9b6732bae458a9c3c1cf3e1e6d029 : SUCCESS in 1h 01m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fccd3fe4d4994720bbc852045b0e5a15 : FAILURE in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f2e51fe30c743af9ad094b223063900 : SUCCESS in 7m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bf2ec01009346fa878dd6bbc7515886 : FAILURE in 14m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b8df4efe53584df28acd0bdcabf04cf2 : FAILURE in 13m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/088e4157a3c74662b29f88372b3086f8 : FAILURE in 21m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b4c14ea7404f94971acc5237a8e913 : SUCCESS in 14m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5032bef279114064b3092a74bc616c60 : SUCCESS in 1h 22m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7e2ceb3c64d4a1f91d30f2f587817ef : FAILURE in 1h 50m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/65475d367ddb4c998d1bc38a83de59e4 : SUCCESS in 54m 38s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/d45ee7e70f46406eaca9744dcd4f53ca : SUCCESS in 37m 22s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/7b9368314f6b4461a6a9d075630996cf : SUCCESS in 1h 04m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/894224a7e3634f7c86e7982614134890 : SUCCESS in 51m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/22b538012e3f4819870025f611c5957c : FAILURE in 1h 41m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/2b67a59c21ee424a91b127419c064d58 : SUCCESS in 2h 35m 20s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/35963acbae9244989920662e2a31e859 : SUCCESS in 1h 01m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a89f8a05c7874730bee9cb833cde0d9c : SUCCESS in 5m 04s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0308c8e4c2004850a5ceb7d9af6c9e81 : FAILURE in 28m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/492a85c083ec46edbc53ac17384b85ed : FAILURE in 27m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/72f3be8660a7413786557e2dce3a717e : SUCCESS in 1h 36m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82c67fc1f029472e9f6057be1cffcc4a : SUCCESS in 1h 01m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/846834bc616344d8adf44cf7bd9f4692 : SUCCESS in 28m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f0c643d0362545f38d23ffe10410e1a1 : SUCCESS in 36m 51s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ceb2fa98ea9c4fd8af791ebfdce01a46 : FAILURE in 14m 51s","accounts_in_message":[],"_revision_number":13},{"id":"026b30c2627cd50792e4e66531cacea636969f9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-11 13:26:42.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"d39e44b8b9d4505ca22a28d83e6784ebec9c94ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 13:50:42.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5c67015a99e41f1b6fcaf4f9af068dd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c2ca2cd62bbc4b94a89e05f154b826df : FAILURE in 8m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9e984ffc76b8418cba33bd0a443a8858 : SUCCESS in 14m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b5659a3c10124be39eef029fd1eebff8 : SUCCESS in 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"affcb11b0e8d4adccb3fcc151087343d1f075caf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-11 14:32:02.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":14},{"id":"86cdb301c31527443396df73a1958117900c7cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-11 16:18:49.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee900cd9dccf4ad4b3253b678f405fdb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/facd6319369141dbbb29e7d80fea6e6c : SUCCESS in 38m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1125bb0639704932951ff57cb5d32e4d : SUCCESS in 1h 17m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0b9e12fbf1e4745bf28da3934ae8f64 : SUCCESS in 46m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22013a52dbeb43259a1e732bc577e464 : SUCCESS in 1h 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc713c2fbe1b42d19ef6a73b8abb1cae : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ec5cd5fe48e44748a7f6c6bc22ac194 : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44eee562283d47e3b4b8fe8132f7d8e8 : SUCCESS in 11m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa83b166629a490f9cd7b71f26cd80a6 : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072463db9b9b4dcc8c99415744bc0197 : SUCCESS in 12m 38s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9091374c7059428694b95df54977ea69 : SUCCESS in 33m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f7adc85e1b4446d97961c76935c1866 : SUCCESS in 1h 34m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73a5062240fb4d04a1d86996d1ce4ae8 : SUCCESS in 35m 38s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/6e250a3a5f1342e1aa9343567655713a : SUCCESS in 36m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a4d016f9a34e4a3cacede370c756901e : SUCCESS in 1h 03m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d76780661800481d844c3655352ac713 : SUCCESS in 33m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b79be9e48f134c58b38f77ebf426be34 : SUCCESS in 2h 09m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/57644b3ae3d349cd93b676140bc94a9b : SUCCESS in 2h 44m 50s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0219c96cb89143faa945f07acb0a1854 : POST_FAILURE in 40m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6cded16d415e45e2b82046bd2e585f64 : SUCCESS in 5m 32s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/27fcfc979ed14109880fad7561896cd3 : SUCCESS in 28m 02s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e8ec695dacc84c73b6571016058b35af : SUCCESS in 21m 04s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/44eb372a4e1440f1a7c822d7f05757c0 : FAILURE in 1h 12m 42s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d46e7d155163490987561261e56916b0 : SUCCESS in 1h 36m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f7c591bb967446d8e3778648b699dc1 : SUCCESS in 58m 51s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9fe6bb4776eb4c91a89d84c607eded8c : SUCCESS in 17m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b8f4a9ed4c434423bf5d8a92d75a3c2e : SUCCESS in 31m 42s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/26bdbe94e5c647f1ae5dac2c17a1b327 : SUCCESS in 15m 08s","accounts_in_message":[],"_revision_number":14},{"id":"652b0bc0f01af3f02ad845ab59d87c83d24a61c1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-13 08:16:43.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e415f2c4c2e7b9b606aa686080cb359a94d1e4bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 09:48:33.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68197d1fbbe4c359ca9892b932a2e81\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2962f4b470ee46a8ac9fdb591829401f : FAILURE in 7m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bd97db1899db415b919d8a827d53aced : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3c7790dfe8ed4afc84c52e728ce81605 : SUCCESS in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2e290836b7ea6f390cc6e6d87ba3d5720c56e97c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 11:16:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c13d30dbcf43559e5108dcc04bf698\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99faf47f1eed48c1b71220e71a590665 : SUCCESS in 1h 10m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f0c8016217624e93a7aedde63fef1602 : SUCCESS in 2h 07m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4506e277d8d4c019db62dfac66fac5e : SUCCESS in 1h 08m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/353d2112058c46a799aa0c7fcd5934bc : SUCCESS in 1h 23m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748859a1480448b5b6a365ec1f9112be : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec8e67097d3c4874b6547a5b5ea52385 : SUCCESS in 15m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34858544894249829d0464a795a47b26 : SUCCESS in 8m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cc709713439845c5ad2021ee812f6310 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af0e4c8f601f4ea59a047c17b6b38993 : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c60b698e899247059ab337950e120aa1 : SUCCESS in 15m 40s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e6759e7cd77747058dd92963cac623c2 : SUCCESS in 1h 15m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bf32b81b497a4066a5a2e2c69e5de2e9 : SUCCESS in 2h 01m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3cb726047cb543e9a75905154010209a : SUCCESS in 33m 10s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/082eef11ba314c7e89b60260f8b42520 : SUCCESS in 33m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b4a48d749d614414abee58076a776135 : SUCCESS in 1h 03m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d9d3cc31541848e38c7e6d9d3a9dbc72 : SUCCESS in 1h 04m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/30bb20aacf8c47cca08016b5c1fd955a : SUCCESS in 2h 50m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/4d1f59928ccf4aba8a21792b9da72c25 : SUCCESS in 1h 26m 51s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/39df4e0aae1a4d94b7f663a3b773c353 : SUCCESS in 36m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cf2732d4407444f4a17d87991ef5639a : SUCCESS in 3m 46s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/20673d86dc694f468e25d18b9f46fb28 : SUCCESS in 33m 00s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/977d72701a904256aaba80934ebcff59 : SUCCESS in 32m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1d0253fc80404dc1943be1d9a5a4562d : FAILURE in 1h 13m 47s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3db6eafe417c4a77aa8a4eab69fb3fc9 : SUCCESS in 1h 32m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/641edd047070459d9b3e16bf8b2ab3b3 : SUCCESS in 1h 04m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e6caeceddb7c49769872ede69f138d60 : SUCCESS in 33m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8845ecfc2f1f4415b44aa90a1094fd5f : SUCCESS in 46m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3e356723149d4dfa855c3be215d9b28f : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":14},{"id":"de0851478e277481a74cd0fbacfa45c0d27335b6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-13 11:27:16.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"c4f40c3c012a7ebd494b1b0ff79dc70ebad69b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-26 16:34:21.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"a370073970156ff82c876faa040f8e2e9c470af1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-26 17:32:41.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":15},{"id":"06b8008dbc80009161cc702b0d9d87afcf928d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 19:13:25.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d9f84bf97804fc4a9f815729da74d5e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de2e4767d3264fba97de63e3777afd65 : SUCCESS in 1h 01m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4cccda0058154b9ebcf4e1346ebf00c9 : SUCCESS in 1h 46m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60dbd82e69aa4cf695e8be26092ef0a5 : SUCCESS in 1h 04m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/170d1453b71b4d8197756f499e6a0e0a : FAILURE in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046102bd579b449fa340d7485af5ca42 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a27bfbe3ac741988e530965fa9c3e98 : FAILURE in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a8c53910da643bab44865b172ac6ba3 : FAILURE in 13m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f11c17cf88b487da26e147c60684247 : FAILURE in 17m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f49ac302ef451ca52ba9f8368a7d39 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33e486da59b54eee821c8a3b7b3815f8 : SUCCESS in 12m 57s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d435bf985c434ccf92e9c5660b37bdfa : SUCCESS in 31m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8d2343e81814b22987a41c6407762ed : SUCCESS in 1h 01m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/01c0cfb363e24568861d178b5a40ce2c : SUCCESS in 34m 26s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/173b8e79fdb74b5297fc081eb063efd8 : SUCCESS in 33m 22s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4993630f4f2244d7ad26d01abea32b72 : SUCCESS in 40m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/80b339b09a7c4ad89af85b1fe9b560bc : SUCCESS in 59m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/81d8e1882262418bb9218ca0d188c95d : SUCCESS in 2h 13m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/c7b9f02beb154de6b12bb75184b09101 : SUCCESS in 2h 32m 10s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ab06c5dd63cc44c989fc1930ddb38805 : SUCCESS in 1h 41m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dbf14b3fe0514804b293e196f51e9db4 : SUCCESS in 5m 36s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f7fe972fc5442e081f685cad8344b08 : FAILURE in 29m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cd47ed56f10c45efa9ffc6672a77b9ca : FAILURE in 27m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/947ac8532aff4730bcc86ce46a50b3b7 : FAILURE in 1h 09m 21s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f98ab4923894a2ab674f30e1ed36ad7 : SUCCESS in 1h 37m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44210c8ed6b3460ba79134c40d108174 : SUCCESS in 1h 09m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/22ea12dadd86452fa7fdc2f518a6a802 : SUCCESS in 32m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c6684ebb824d420fbfded8a5fd4ad76e : SUCCESS in 33m 24s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/bd359df4ee534e30bd442bf266cbc6fe : FAILURE in 14m 02s","accounts_in_message":[],"_revision_number":15},{"id":"3ced253e623a405730274b6bba5956e88bf98fe8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 22:08:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44cf2c7d2a042beaeffb174ed1fdbcf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b27240ed289d42b9ab9a54e92579d612 : FAILURE in 20m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59781192737844eda63b8840197e12d1 : FAILURE in 12m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f5ea46cb1a0b4ef987c33488d8aa9c20 : FAILURE in 38m 41s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b32940182796505b925674c8f000987991073e12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-27 15:49:12.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"8c1d2ccc40edbe5a5ce4de75016420f255a9bd66","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-27 16:47:21.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":16},{"id":"682f8a7db7dbf2bb1ec1bbd975bf341b216fdcf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 18:13:26.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/430fe0ec86fc4f90a43c23b0e5d2bf78\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9583429cd6b4ac5b6fb38b1f1223003 : SUCCESS in 54m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/afc30c9d3c2943238c93dffd559be832 : SUCCESS in 1h 08m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc123b4cd61846dea349288514a824c1 : SUCCESS in 1h 04m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb05b006714b44b7bd3cb896167c6f3c : FAILURE in 1h 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1981566c542e4b89bc07c024787b4b23 : SUCCESS in 8m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9387ebf285e04577995ce2b33d790b5b : SUCCESS in 13m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/37b326ff5b2f4c18acc8ec0a1c9e05af : SUCCESS in 12m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a1876cde80e046d2a39e2311f3fac93a : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea7f18af2194433e8a8560d3ec165def : SUCCESS in 14m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/882c5e738e65406284a7624b8805d021 : SUCCESS in 15m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5f0b4d847aea40a882ad62e6a04f91cf : SUCCESS in 1h 01m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/12251db8d20c4989ba878d4581594193 : SUCCESS in 1h 52m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/40a860e9a93946619a2e6d7ac7d0159f : SUCCESS in 57m 44s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/bc7696b601334c47805d56809a74f29c : FAILURE in 49m 18s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/60279feafaa74989a65c06d9a8302078 : SUCCESS in 1h 00m 12s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9a7e630f1cc14b79a57625ed1e57dfa1 : SUCCESS in 1h 08m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8cc259477c25452caf54218fad161e39 : SUCCESS in 2h 09m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/b133ea5e7b46417b9252eb7f34df0e04 : SUCCESS in 2h 12m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8a5929f693da468c842c30c2064d1c7e : SUCCESS in 1h 22m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2abee9f183634736b128f3544326f258 : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/540b870f048549b69def4dbb0c7f3803 : FAILURE in 24m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3cb501a27ed242fe94d21ba260a42c79 : FAILURE in 34m 46s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9cb931f84df045489306b05b724e77ce : FAILURE in 43m 01s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a05cb384a904efbb4e2dac6cecb81ca : SUCCESS in 1h 35m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/396bd96428c049958218a568e9396e36 : SUCCESS in 48m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/29f070c5fe7c4c60a04b852fec7153a7 : SUCCESS in 31m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/655a04465ec448089ff8722abce3abc9 : SUCCESS in 42m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/36a559e4107340d18b718fae1f15e5dd : SUCCESS in 13m 36s","accounts_in_message":[],"_revision_number":16},{"id":"2a430e5361528579768a0d3b74741dd482674a19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 20:10:31.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7375dcaf2a874b55adee6371da8c56f1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a441f9046f8240c7b7d26bed26bde39d : SUCCESS in 19m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94f64512c71f4467a988ff4968e1227f : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7a2ca0d01f5243b9bcef32c2241c3734 : SUCCESS in 39m 20s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6fccdbadba90f3466bb1a8eb9cd4a185895545cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-29 11:02:45.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"8b573265f8620517720eaf1fe57f3106a0e2d268","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-29 11:58:09.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":17},{"id":"e461a0a2f6c68ed46ce3629411f52f89085a624d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 13:55:02.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc974c3984e4a5ab99529bc34fb0179\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a99b55a8877b409e94b2d89c6bdf4492 : SUCCESS in 1h 05m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aec306c936b5441fa391ebf067224fe9 : SUCCESS in 2h 02m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9951178938c544d3828161fbee2361de : SUCCESS in 1h 05m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4665f9d91bf8443689781c056f128e64 : SUCCESS in 1h 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b0b791174446f3babcbfa35856c543 : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/78f5ece34ee84a95bddf9a72cde5cfa2 : SUCCESS in 13m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7981889070634ff3bb97e026eb7c83f5 : SUCCESS in 14m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0ebc6a5677734b5fbe59a6713602d301 : SUCCESS in 22m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac48b2e35d84c06ac712b2914803167 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2329e997cdee47fdb91abcd52dad03a3 : SUCCESS in 21m 00s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6cb2732d5d4f4c398f2553facc086927 : SUCCESS in 1h 03m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75215516a9da48628861375467ec53a9 : SUCCESS in 1h 39m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3b8f3770b7044c1784e22709d2f5734f : SUCCESS in 1h 01m 29s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/e3346846361a4b4baccc9fb92f37ac2a : SUCCESS in 37m 28s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f6584cdc82ca44bdbcd191c5d185e41a : SUCCESS in 1h 05m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/35e19a7f6888470c8c88d2e0c0cdf859 : SUCCESS in 44m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1fa1587698d84ae6bc5f85b0f15b2a14 : SUCCESS in 2h 10m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/d753dfbcf7284a98bfa22d072dab1e99 : SUCCESS in 2h 44m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b9b7804b10a74d19a02272909d356613 : SUCCESS in 1h 41m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/05a7b63868974b1fb78175791019467a : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2751be6e59f54d929c3fb43caa084415 : SUCCESS in 26m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1a92a722d339466792e5be4961b94bdb : SUCCESS in 24m 31s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7fa1a879f7ba49dca3f45afab236fd8b : FAILURE in 48m 24s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e8ea01dc6c4432b955c846d3f966dec : SUCCESS in 1h 29m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a4ad7fdc5d7476ba4e6e5f9c3a9976b : SUCCESS in 49m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc23e68adc854882bdfb652a2e7048f8 : SUCCESS in 38m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bf9410f6f5a949f8ae6dfd3aa28c375f : SUCCESS in 33m 39s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/459061759529458da40482b81c2bea4c : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":17},{"id":"fc9ff663e45eb3f20b740291538e9b9bba0477e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-29 14:06:06.000000000","message":"Patch Set 17:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":17},{"id":"f9b5425310de9195a69e25fbac70f98685022357","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 15:55:49.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51f2860f267741a287d3d4a668822c18\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68666e57e1744530bb72c630e790ead4 : SUCCESS in 21m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83dd49eb8f5e494988c8f671af4be33e : SUCCESS in 17m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4af7f99096a1486c9f85cd566f7f4909 : SUCCESS in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a438665432371a4aca5de9dd054e723dfde0ae8c","author":{"_account_id":35876,"name":"Max Harmathy","display_name":"Max Harmathy","email":"maxh@uhurutec.com","username":"harmathy","status":"Developer at UhuruTec AG working with Kubernetes and OpenStack. Open Source and Free Software enthusiast. Vintage Computer hobbyist."},"date":"2026-06-03 11:55:16.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"ae9cb0c0231b613b2eb603f9e328b4f32858c0c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 21:17:22.000000000","message":"Patch Set 17: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":17},{"id":"fedc4da3405c8542eece6a841575302200af7bfa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-05 22:43:18.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a892e8363881590c4f78531f7dc90cb6f889906d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-09 01:04:41.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"5ef56145bdbf3a5627d213161e79ad4014013af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-12 15:24:39.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":18},{"id":"52ab4bccdd8c6abcaa9185251230b833c1a69669","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-12 15:28:36.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":18},{"id":"bce832d82c037500ea92cd9b9f7fd66b294d8223","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 17:54:51.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/673e4aa61e714f87a050b9d7856b3f99\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb207854df764d178e4d39790b2affc8 : SUCCESS in 51m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/351642411f7741e799c66e364b2b16c4 : SUCCESS in 2h 06m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31174922d15944b1a70dec7c39772e1b : SUCCESS in 1h 05m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab322c648da3484080a37899d83143b8 : FAILURE in 57m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95ed0627479d4fe48166e63b9bb1d1c2 : SUCCESS in 8m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47c671e1255547cb9092a2872dafbd0c : SUCCESS in 11m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/890d86e339f44939bf041907ea7e2ba4 : SUCCESS in 15m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/78dc508a7dad4e3e812308e3342295d4 : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c85586b65d84f309285af488464bac0 : SUCCESS in 11m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52102926d0af493fb2ba4441effc5752 : SUCCESS in 16m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1f16555dbe6b4bcb9c7d253cad84280b : SUCCESS in 1h 04m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b3f50c82490d4493a9d0e84a5c515d9a : SUCCESS in 1h 54m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b6967a8651214908bbebf38536ba29cb : SUCCESS in 43m 21s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/a875ef8cdcf8440b8259161205d069b7 : SUCCESS in 39m 06s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d3629b4c6f1a4e0f9763f49752feb402 : SUCCESS in 1h 05m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/27b7b1b31d384d26b517bb7d5aa2b9fb : SUCCESS in 54m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7cad7e75a0714fc8b77aaaa4cab0e607 : SUCCESS in 1h 19m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/7039215ce70748059fcc8d0310ececfb : FAILURE in 2h 21m 34s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2f6add31b1cf47e08565fa8270730c6c : SUCCESS in 1h 55m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3cd51a413d2b42a6a3236e868290a310 : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/416344c251cc4276ac31d322b998f1d2 : FAILURE in 26m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e5b31be8da214b679f048b3ad2a07d07 : FAILURE in 26m 27s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4b4bef237ce84632adb4c46e42ddcd7d : TIMED_OUT in 2h 01m 28s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5b966b097214bd2a8681fc547dbd6ef : SUCCESS in 1h 48m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6824d739a4674a88a9175d5d77c5ce11 : SUCCESS in 1h 05m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cc26c59809df42a39968e4a87e8688bb : SUCCESS in 28m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/aeeda1a9132948468cffd5be5f2e6fc5 : SUCCESS in 32m 24s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9ca8d1883a8b4bcbbb148fbd8c3af94f : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":18},{"id":"48287e273c1f03629143da559e146ef8e58ac372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-06-15 11:17:42.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"9a9bfca0e58802df486dc668092994d60959181a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-15 11:34:30.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I1e9a0d871398ef78fded868154e61f2c156d7c41 .","accounts_in_message":[],"_revision_number":19},{"id":"9c6df13237109730a45d74e562a5184e789edc54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 13:45:10.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b91f5249f91740688097b9364470444e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9ba43ae8d4d4a279f3605ca02fefbdb : SUCCESS in 56m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c180aada3f95448c8eb6a51803d5e3cc : SUCCESS in 1h 57m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4dfcacc63b44595ac08f4e38ecde2c5 : SUCCESS in 54m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f0df1368de44f74a7f3af7717a71c6f : SUCCESS in 55m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed67d49089e4fdb9d1b8098fb107110 : POST_FAILURE in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b011034d5ce459f8a2f50cbfc43e30b : SUCCESS in 11m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ce325a98ffe4cb9bec26a4015e4e140 : POST_FAILURE in 11m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0c005e81dfb145cbbb31ac3228e57dd9 : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a325e17a7a0418fb7448fa928dc06ef : SUCCESS in 12m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/829ab3c159454595a923e95af478cf1a : POST_FAILURE in 10m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d0e1d16a78804ef6801ff6d414c06121 : POST_FAILURE in 1h 14m 35s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5679fceda614bbcb2ac0107e86fc997 : POST_FAILURE in 1h 49m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b28a6f66d18f4e3bae39f5f87b8246f2 : SUCCESS in 56m 58s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/33910aa3e03b495597bb21e07b8c39ea : POST_FAILURE in 37m 00s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e5b32545095142b6b3269c0e43527451 : SUCCESS in 54m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/957b69f12b1846aaaa9b3960fff169d2 : SUCCESS in 56m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/09fbda82e51841929d5c9b569759bd37 : SUCCESS in 2h 04m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/614d2c3f738b4f1a860c926e43883035 : SUCCESS in 2h 20m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/091777c3f19748658ce75540fbdbb983 : SUCCESS in 1h 42m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fcc66a7b227f4be9868d8aa73e444047 : SUCCESS in 4m 55s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5dc9011773194d5184dfd7faeb98d8d7 : SUCCESS in 23m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/56f49437f83d464d9a12706b77afaeac : POST_FAILURE in 27m 00s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ad02730b9332469a966881aa0951d8cc : TIMED_OUT in 2h 01m 26s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd4f3f2126e24f61a9b2f9bea1c630c0 : SUCCESS in 1h 26m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a17d30d8b9146eba0c036d4c0f97e55 : POST_FAILURE in 48m 59s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7abfc798d1f54d348b842aba7f1ffc18 : POST_FAILURE in 28m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e89e01e6e9f7485a86c23b921260bf59 : SUCCESS in 34m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/28457cfdec0b43fcb1b0a78585f5b593 : POST_FAILURE in 14m 51s","accounts_in_message":[],"_revision_number":19},{"id":"9d5d2089d0edc92c2dea96dce92600b97aefb584","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-06-15 16:47:55.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"c40313010354804a7c8ba9bc9748fa3dd5a556b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 19:46:13.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/270e9b4451dd48e1956eff2122fbec0d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/38c6ec92c31b455e8dda3be2c589d431 : SUCCESS in 1h 06m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/40c8b615cb6c4b0c9ef617bdc1f43082 : SUCCESS in 1h 19m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e1814f547204d6a9000352fe94551c9 : SUCCESS in 1h 11m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaed501a45be4c1aa65770c82c444b66 : SUCCESS in 59m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5362053f3e14d49b4be1f4a59b3dae6 : SUCCESS in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c9d8f44f86440d99474a75ef85869a7 : SUCCESS in 14m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89ba621e47e14f0cae72d2c019247276 : SUCCESS in 15m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/39bcf12362e34a94b09f59ef3b36aecf : SUCCESS in 23m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f3e620d05741c8b7e188a700c1263a : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2d9a2c302d4076b790bb98066954ba : SUCCESS in 9m 39s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/654ca414f18941d692a9909debe7f75b : SUCCESS in 1h 03m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a76e464f60c4415b9a126c8ed4f1af8 : SUCCESS in 1h 44m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c8f706a107a490c8f251f97edbc568e : SUCCESS in 48m 52s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/ff36aba5082d4411be75b7efbe5d1a1e : SUCCESS in 37m 01s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/eb8c8674ece44abfa4ffa59ab2cbad28 : SUCCESS in 1h 04m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9c90bea5727d4290bd6685a7eb32023e : SUCCESS in 35m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e94c0b9a70b042c291ba1c23bae79d2c : SUCCESS in 1h 22m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/c00f29cffccf4d06ad410f0d0f99f9ab : SUCCESS in 2h 48m 38s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bbf2a00035f24dcebe8ecfb58da3f77a : SUCCESS in 57m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a979603fd0c34bd2a48a621f54ad9ca2 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fee1f307ba5e4b38b9b47dc7ecb1529e : SUCCESS in 30m 09s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9d05ce2906b54e8e88397997d3f19748 : SUCCESS in 31m 24s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0f9a184699af48ee96e069fc1f2e9512 : TIMED_OUT in 2h 01m 59s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a1096461a1b422d826ef2e3ee803450 : SUCCESS in 1h 18m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04dd35041ea54d26a82cd73f63d8b8d0 : SUCCESS in 59m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/20bc3a82a59f4ee08c5ca56ce9aa8cbb : SUCCESS in 21m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e15661e626624e82abca9a844d19415f : SUCCESS in 35m 00s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/46e7f55d88ad4349ac7ca8d692683a02 : SUCCESS in 18m 27s","accounts_in_message":[],"_revision_number":19},{"id":"5942a9d2c9acad52edad6d31b4b91b197c59a36f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-15 19:57:20.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":19},{"id":"5f2b967503cb1003ced9aaafda43d5757e7257e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 05:30:38.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2517df09932842b18ddf816bdf6464e1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5ef7918191b7408aa570c1a61a468241 : POST_FAILURE in 16m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c06b255d89744ff399cd28696fefffb0 : POST_FAILURE in 19m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e7da8e0d60fa48b8a6186257ebb6ec85 : SUCCESS in 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"3f721fcd0b89b7443f67bd57435e26f199aebba4","revisions":{"b3106bfcea9933fb13e9a5afa9f68e9d65dac145":{"kind":"REWORK","_number":1,"created":"2025-12-02 19:01:43.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/1"}}},"commit":{"parents":[{"commit":"fb7efeee2df6d1cc6dfcec3eed72db9a9f364f61","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb7efeee2df6d1cc6dfcec3eed72db9a9f364f61"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-02 19:00:29.000000000","tz":60},"subject":"WIP: Use SDK for Neutron Ports","message":"WIP: Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3106bfcea9933fb13e9a5afa9f68e9d65dac145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3106bfcea9933fb13e9a5afa9f68e9d65dac145"}]},"branch":"refs/heads/master"},"7d2207f4f28b14302fb1a0ca0e25c3f7031cb2f0":{"kind":"REWORK","_number":2,"created":"2025-12-05 10:54:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/2"}}},"commit":{"parents":[{"commit":"fb7efeee2df6d1cc6dfcec3eed72db9a9f364f61","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb7efeee2df6d1cc6dfcec3eed72db9a9f364f61"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-05 10:53:04.000000000","tz":60},"subject":"WIP: Use SDK for Neutron Ports","message":"WIP: Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d2207f4f28b14302fb1a0ca0e25c3f7031cb2f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d2207f4f28b14302fb1a0ca0e25c3f7031cb2f0"}]},"branch":"refs/heads/master"},"fa04bfb727f0020c32a8b62c498ba04d1806d1dd":{"kind":"REWORK","_number":3,"created":"2025-12-09 12:06:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/3"}}},"commit":{"parents":[{"commit":"fb7efeee2df6d1cc6dfcec3eed72db9a9f364f61","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fb7efeee2df6d1cc6dfcec3eed72db9a9f364f61"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-09 12:05:45.000000000","tz":60},"subject":"WIP: Use SDK for Neutron Ports","message":"WIP: Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa04bfb727f0020c32a8b62c498ba04d1806d1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa04bfb727f0020c32a8b62c498ba04d1806d1dd"}]},"branch":"refs/heads/master"},"785c4a0599c84846079c9f67735e92a62d4c0e12":{"kind":"REWORK","_number":4,"created":"2025-12-12 16:23:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/4"}}},"commit":{"parents":[{"commit":"944ef31e0198b7201dc1c0a00cdfcb9f3c11d3d8","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/944ef31e0198b7201dc1c0a00cdfcb9f3c11d3d8"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-12 16:22:38.000000000","tz":60},"subject":"WIP: Use SDK for Neutron Ports","message":"WIP: Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/785c4a0599c84846079c9f67735e92a62d4c0e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/785c4a0599c84846079c9f67735e92a62d4c0e12"}]},"branch":"refs/heads/master"},"5ab2b51d40707c848a30f1dfe30ce59540b95ed9":{"kind":"REWORK","_number":5,"created":"2025-12-15 15:27:57.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/5"}}},"commit":{"parents":[{"commit":"944ef31e0198b7201dc1c0a00cdfcb9f3c11d3d8","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/944ef31e0198b7201dc1c0a00cdfcb9f3c11d3d8"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-12-15 15:08:42.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ab2b51d40707c848a30f1dfe30ce59540b95ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ab2b51d40707c848a30f1dfe30ce59540b95ed9"}]},"branch":"refs/heads/master"},"c481a3b0d74130943ea2ad145ba90c311cbb0f3d":{"kind":"REWORK","_number":6,"created":"2026-01-06 17:07:44.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/6"}}},"commit":{"parents":[{"commit":"a4d5922716bae2e5fcffcd718c11e4c115e63c01","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4d5922716bae2e5fcffcd718c11e4c115e63c01"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-06 17:07:13.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c481a3b0d74130943ea2ad145ba90c311cbb0f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c481a3b0d74130943ea2ad145ba90c311cbb0f3d"}]},"branch":"refs/heads/master"},"febf9c41d8b7247acc1b38278d71cd3ce07a5f30":{"kind":"REWORK","_number":7,"created":"2026-01-07 08:43:51.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/7"}}},"commit":{"parents":[{"commit":"a4d5922716bae2e5fcffcd718c11e4c115e63c01","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4d5922716bae2e5fcffcd718c11e4c115e63c01"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-07 08:43:19.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/febf9c41d8b7247acc1b38278d71cd3ce07a5f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/febf9c41d8b7247acc1b38278d71cd3ce07a5f30"}]},"branch":"refs/heads/master"},"4ea2d29cba8ae406cca6fba16e46666494dd5767":{"kind":"REWORK","_number":8,"created":"2026-01-28 17:11:08.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/8"}}},"commit":{"parents":[{"commit":"f91ff73cd083561dea1aeb999dc85ff0a498409d","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f91ff73cd083561dea1aeb999dc85ff0a498409d"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-01-28 17:08:10.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ea2d29cba8ae406cca6fba16e46666494dd5767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4ea2d29cba8ae406cca6fba16e46666494dd5767"}]},"branch":"refs/heads/master"},"93fdc6d8ed826fd0862296f6099549855df7d479":{"kind":"REWORK","_number":9,"created":"2026-03-13 09:04:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/9"}}},"commit":{"parents":[{"commit":"abfee6668084b34e57723979b70364cadd4158c0","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abfee6668084b34e57723979b70364cadd4158c0"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-03-13 09:03:39.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93fdc6d8ed826fd0862296f6099549855df7d479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/93fdc6d8ed826fd0862296f6099549855df7d479"}]},"branch":"refs/heads/master"},"e7cb38360f0b997fe8043f9ca92d8517f40f22bb":{"kind":"REWORK","_number":10,"created":"2026-03-13 09:56:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/10"}}},"commit":{"parents":[{"commit":"abfee6668084b34e57723979b70364cadd4158c0","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abfee6668084b34e57723979b70364cadd4158c0"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-03-13 09:55:50.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7cb38360f0b997fe8043f9ca92d8517f40f22bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7cb38360f0b997fe8043f9ca92d8517f40f22bb"}]},"branch":"refs/heads/master"},"aa86eb3870b20ad4556a218776793f94f290cfa4":{"kind":"REWORK","_number":11,"created":"2026-03-17 11:06:28.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/11"}}},"commit":{"parents":[{"commit":"abfee6668084b34e57723979b70364cadd4158c0","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abfee6668084b34e57723979b70364cadd4158c0"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-03-17 11:06:04.000000000","tz":60},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa86eb3870b20ad4556a218776793f94f290cfa4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa86eb3870b20ad4556a218776793f94f290cfa4"}]},"branch":"refs/heads/master"},"8bafc0e91ca1fbc7beb3f4563a731db432fbf79e":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-04-24 11:26:36.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/12"}}},"commit":{"parents":[{"commit":"1b17c3b5b8e1e2fc5f90c83e966fdaabd3d17325","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b17c3b5b8e1e2fc5f90c83e966fdaabd3d17325"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-04-24 11:17:22.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bafc0e91ca1fbc7beb3f4563a731db432fbf79e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8bafc0e91ca1fbc7beb3f4563a731db432fbf79e"}]},"branch":"refs/heads/master"},"1c522b2d0d7cbca34508b1fb83940b311079aff7":{"kind":"REWORK","_number":13,"created":"2026-04-24 11:43:18.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/13"}}},"commit":{"parents":[{"commit":"1b17c3b5b8e1e2fc5f90c83e966fdaabd3d17325","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1b17c3b5b8e1e2fc5f90c83e966fdaabd3d17325"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-04-24 11:42:51.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c522b2d0d7cbca34508b1fb83940b311079aff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c522b2d0d7cbca34508b1fb83940b311079aff7"}]},"branch":"refs/heads/master"},"2c2151eb955f5d308c1bb7b72d69e7b6f3b73bd0":{"kind":"REWORK","_number":14,"created":"2026-05-11 13:26:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/14"}}},"commit":{"parents":[{"commit":"390a25b1cd38428ec69a3fe95a0f6e1c80ad40f2","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/390a25b1cd38428ec69a3fe95a0f6e1c80ad40f2"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-11 13:07:13.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c2151eb955f5d308c1bb7b72d69e7b6f3b73bd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2c2151eb955f5d308c1bb7b72d69e7b6f3b73bd0"}]},"branch":"refs/heads/master"},"f57895658919db84e4e6ad468b58b9774e7959a6":{"kind":"REWORK","_number":15,"created":"2026-05-26 16:34:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/15"}}},"commit":{"parents":[{"commit":"15485a8e70a9bf47f99158a18a62e42e0ea8d8e8","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/15485a8e70a9bf47f99158a18a62e42e0ea8d8e8"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-26 16:34:02.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f57895658919db84e4e6ad468b58b9774e7959a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f57895658919db84e4e6ad468b58b9774e7959a6"}]},"branch":"refs/heads/master"},"c3164f5f6fec4e3a80148fd6109ca5f9b1eb3dde":{"kind":"REWORK","_number":16,"created":"2026-05-27 15:49:12.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/16"}}},"commit":{"parents":[{"commit":"bc5fe16c319ca59626f8015b8daea699cf86967d","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc5fe16c319ca59626f8015b8daea699cf86967d"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-27 15:48:40.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3164f5f6fec4e3a80148fd6109ca5f9b1eb3dde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3164f5f6fec4e3a80148fd6109ca5f9b1eb3dde"}]},"branch":"refs/heads/master"},"4b10dbd3854f55aab6ce22d168a592072924ea9a":{"kind":"REWORK","_number":17,"created":"2026-05-29 11:02:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/17"}}},"commit":{"parents":[{"commit":"0ae066471085b40875680e633760383fe121adf4","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ae066471085b40875680e633760383fe121adf4"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-29 11:01:17.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b10dbd3854f55aab6ce22d168a592072924ea9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b10dbd3854f55aab6ce22d168a592072924ea9a"}]},"branch":"refs/heads/master"},"ed3bffcb079e6788038681c5bf408e387be0cc9f":{"kind":"REWORK","_number":18,"created":"2026-06-12 15:24:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/18"}}},"commit":{"parents":[{"commit":"7a4be56698e79e6aed543efcece89a07bfa67ba5","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a4be56698e79e6aed543efcece89a07bfa67ba5"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-06-12 15:23:59.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed3bffcb079e6788038681c5bf408e387be0cc9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed3bffcb079e6788038681c5bf408e387be0cc9f"}]},"branch":"refs/heads/master"},"3f721fcd0b89b7443f67bd57435e26f199aebba4":{"kind":"REWORK","_number":19,"created":"2026-06-15 11:17:42.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/98/969298/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/98/969298/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/98/969298/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/98/969298/19"}}},"commit":{"parents":[{"commit":"7a4be56698e79e6aed543efcece89a07bfa67ba5","subject":"Use SDK for Neutron floating IPs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7a4be56698e79e6aed543efcece89a07bfa67ba5"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2025-11-25 10:09:47.000000000","tz":60},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-06-15 11:16:20.000000000","tz":120},"subject":"Use SDK for Neutron Ports","message":"Use SDK for Neutron Ports\n\nChange-Id: I1e9a0d871398ef78fded868154e61f2c156d7c41\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\nAssisted-By: claude-haiku-4.5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3f721fcd0b89b7443f67bd57435e26f199aebba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3f721fcd0b89b7443f67bd57435e26f199aebba4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
