)]}'
{"id":"openstack%2Fnova~969380","triplet_id":"openstack%2Fnova~master~I4bf6294949f9c72ae0bbc111656507ce6603ed88","project":"openstack/nova","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-12-31 20:15:44.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4bf6294949f9c72ae0bbc111656507ce6603ed88","subject":"Adds regression test for bug LP#2132984","status":"NEW","created":"2025-12-03 11:52:11.000000000","updated":"2025-12-31 20:15:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"365d858d01b4560a11500c4eae00a45a169e9615","_number":969380,"virtual_id_number":969380,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-12-31 20:08:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-03 12:50:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-03 12:50:50.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-03 13:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-09 08:36:56.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-31 20:15:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"377710c1b23596af7a7187cc62bba3c6000abfe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-03 11:52:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de5df541a1cbeb7bac811be77cca2b7130031162","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 12:50:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ca45d8b7b54550a010c1bea3406597\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f610040003a405e807c18b0a5e720eb : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a708e6f10fae44a48212f98f32d7110e : SUCCESS in 33m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ba6fc0314f94d51921ee8a4ca2eab05 : SUCCESS in 40m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5f4496b5d4396068d1d1123a4ddb235abfb18a7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-03 12:50:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":1},{"id":"04de10dbbabf896ef2099194d3a6403065cded42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 13:51:47.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4348f6c21148460fbd810084a170b8fe\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcebffaf73ae4d8e89b4ff4aaf923b2c : SUCCESS in 1h 01m 00s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dd7c3f68470e4592993c31531f9c0666 : SUCCESS in 1h 49m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4832e62d2b1845ca81fb63a7e881474a : SUCCESS in 58m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ecf1e69387940eb93714984c06defb3 : SUCCESS in 26m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4f31e803ddc4429b8bb184bad3f4b5c : FAILURE in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bd75950ecc94086a493ccd6b4637957 : SUCCESS in 8m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ed412539712485b8376c20253f26284 : SUCCESS in 9m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f5730f948ff4c008f424aa2eb07ee97 : SUCCESS in 17m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0aa7fba18304101ae7447e6de750c5c : SUCCESS in 13m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c00843783e104fc1a5fc3421a836d109 : SUCCESS in 35m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07ca628b7df745908a17ae27ec732182 : SUCCESS in 1h 51m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7f41d5dc0bf24b2ca7cb37bb24561094 : SUCCESS in 59m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ab028ee18e17471cb4ffaf8a0bbefbb8 : SUCCESS in 26m 44s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e92bca2cd9aa4fe59cac574615adb652 : SUCCESS in 1h 08m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/48e7cd4e9ddd4520b51da820a5dedd91 : SUCCESS in 1h 18m 32s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f134916f4ba44cabbce6b2c199fa8309 : SUCCESS in 30m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/be8ca1ee2de7464bae271226d0b71da7 : SUCCESS in 5m 30s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa32a096d020473da4b01c97e0cbe6dd : FAILURE in 26m 45s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/774453b72c4c4ec5abb85f4806346d7f : FAILURE in 33m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/01d696464da44e8096951cd7f7b08d8b : SUCCESS in 1h 27m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa93fd8c44a64876a5552ebff7a9af4e : SUCCESS in 32m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8fc04f9480a5479abadb6be33bec88a1 : SUCCESS in 29m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b9645a926a254a1892f0e85571bcaba3 : SUCCESS in 33m 13s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1173a132d6274f4c95621e5e11df0c62 : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4536698565c891b0f13c790c528ba835841a9b51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-03 18:21:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0b160be3e98e8fa4514d4b9ccf9b2da634519d1d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-03 19:22:20.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":2},{"id":"13467c4e59f435f08744c6fb709b4615bb51cd2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 20:03:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d6e2e0c7434c82bba0da31b8a835e3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6227ed04014a4686adf82434f6431561 : SUCCESS in 14m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42a37fa8d48548318df423818f960c2b : SUCCESS in 14m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/944f32c5af2a4526aaeeb7c5e633b9f8 : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12bd5c9accd732e960f772be09ad0afaf3914cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-03 20:52:26.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b14d498820fe448fa4fe31987e9cc485\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84e7ddc6639544fab1e4620be2f45c97 : SUCCESS in 1h 00m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4830d184b9604f86b80b2ccccd0e1e8a : SUCCESS in 1h 45m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9912a0f18c8044fc864df559f7009879 : SUCCESS in 1h 03m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ac679ed1c8c4e39817a84be5209a52c : SUCCESS in 34m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0bdad04c7c4d5a8fcce637bb1aa3c1 : FAILURE in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cf2dd533797443da713ed613a32cf75 : SUCCESS in 12m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1dd5846052a84e638f931748f48c09e7 : SUCCESS in 14m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c5d3213333ec49b09f2b37bd05c7dc7e : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e579f5671b47e8b8867d6a935b588d : SUCCESS in 11m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4cf5ffbe8064420f93beb2680d4bb076 : SUCCESS in 1h 11m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef5e886314c149e19aca627645b8ae4e : SUCCESS in 1h 04m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c18c62393574dd2a4b1e174b9dbea63 : SUCCESS in 36m 32s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5312e0a11c8549cfbc0222ec388d2428 : SUCCESS in 26m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6ecbca829088464e97f739d9f355d1c5 : SUCCESS in 1h 51m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/c57aef288c1f4b3896fe254077baa53d : SUCCESS in 2h 23m 14s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f0e48d5222f8482aa08bb357f5503523 : SUCCESS in 59m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6f2a8cb2134d40b4ba891662b854d190 : SUCCESS in 5m 42s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3c2334699f94048b422a30e9ac1b869 : FAILURE in 25m 34s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3c877b7b55674268946c14a4a8b7fb69 : FAILURE in 27m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8b51e17fb274a72bf882b72f4d60713 : SUCCESS in 1h 20m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42704342c7d741bc916fc20fef3b9029 : SUCCESS in 30m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/64779243aaa94f8197378bc5bd00ffe7 : SUCCESS in 13m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/beb0351932c5405c88bf5e1e2b28256f : SUCCESS in 32m 14s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/10186d30cab2433b993cdcb76d6577af : SUCCESS in 16m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e56f56e73c9a4f00aa1e246a59cbbcfb8e02a31c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-05 11:15:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"404688b6a72fc691418cae4d204a94a72fb269a5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-05 11:16:46.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"dbb433152c35026646d8936e5d7af9936556acef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 11:49:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192eb25f473445f782a94a4b3c6bdd6c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ca5b650ed144ff3b0b34df7d25ed9b2 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f25e388d377243cf910941808138f4a5 : SUCCESS in 31m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/97fbe96f829c4ab98f0726faa3659647 : SUCCESS in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"321d9bef7403f36594795636dd36f83914af95c5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-05 12:13:09.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":3},{"id":"b72e160b4405c8cbc880278ca06895a3d1193f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 14:09:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3a9520f72cd494b8c0dd6687361aed6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a258ee7400e42098de96048e518c514 : SUCCESS in 1h 03m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/43fb92e6cf7b448a891c18d1d31ef5ad : SUCCESS in 2h 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8eea1f25989d49278ab26fb4ec582306 : SUCCESS in 34m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4460b397a4964011b39efe10ec334b95 : SUCCESS in 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2000bb4fa3d64845a2601baa751c08ba : SUCCESS in 7m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1469130c1528497bba76635c27814ef3 : SUCCESS in 9m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5554d64d76b54fc995ffb4ddf80e5aa8 : SUCCESS in 13m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/decad223c76249488aa899498618674d : SUCCESS in 11m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62539a9fc4d14088852a1b60fb330d3b : SUCCESS in 6m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/956a4a5acb8641669e37ba5c6b81e485 : SUCCESS in 1h 21m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/742a39211ea1405da4f07e454b13f7f7 : SUCCESS in 1h 50m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/728637873ceb490bb9003cf730fdbcb4 : SUCCESS in 58m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/54fe9a4a88044f37bd6adfaa3e05695d : SUCCESS in 25m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/267b4710d45241fc9fe4b7eff4f8a89c : SUCCESS in 1h 56m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/9ed5dac1431c4452985e9add9135f6a1 : SUCCESS in 2h 37m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c8beee07cc6a4b51b253d4d7bb3e311d : SUCCESS in 49m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7d6e5419397e4282a9301de19334f6ed : SUCCESS in 5m 22s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c77c986ee5db4925bfa569f0956e9673 : FAILURE in 26m 09s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dade8ed38df844808fa784f174fa5f55 : FAILURE in 34m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1291288182f4894bdfc68d6ebf9bc3c : SUCCESS in 1h 09m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dbfb565549743e1aaf0797f4e2ac996 : SUCCESS in 58m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f30fcddba6fe4042b2eaefcf92488d26 : SUCCESS in 29m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/80547377bf424229ba73e284e8cdb21c : SUCCESS in 27m 28s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/01d041dd139e4135a5dd7eae5181fb17 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b6192f66288e8bfaaaa0198214f826f5acf6fc56","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2025-12-09 08:36:56.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-969380-t4lps/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"0f8020a79ff179580932016663b20618045f5d3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-24 10:09:15.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6f3cb67cd73ea255c9f39988649c540265cdad03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 10:39:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f719342d4934acea7640d50aff2b956\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e97d123b4e5840489953a288e1c11149 : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8caf70c6fb734852ab509768fc727dae : SUCCESS in 12m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3bf4633e8259423aab60fc05d46a48ba : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ebb5afd7ffbca3eaeefc85301dc69ead531fabaa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-24 11:14:23.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":4},{"id":"eeae8d8f5452d724543b8fad2b68979516329d07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 12:27:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7cdd63c4f1446fb9184805f4e3edbed\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2010147ed206491299e861142f870f90 : SUCCESS in 59m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b9cc406d337c4be28a7949bc6067dae0 : SUCCESS in 1h 20m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a08d1675a0d94c8aa51aa1c6b436a622 : SUCCESS in 34m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d592d7656a147fb8822ae02d9470dda : SUCCESS in 25m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385ad4d286104c8d94fd780387fed3ae : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28153fb9bf2b4c2baac7f348c0a59a7b : SUCCESS in 13m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e7f0a19314649aca324915f5a97b5fd : SUCCESS in 10m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f4c3a8326c0a404cbce2d95c57deea24 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd06b0f36b7c45fb888383b9c572027e : SUCCESS in 11m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c373aa969fe4e3da6ebd09643c11fb8 : SUCCESS in 30m 51s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b855115633145bb8ecc2d303b5a1475 : SUCCESS in 2h 10m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/405045193958469fbffa2b9e7c0b4aa5 : SUCCESS in 45m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a3b4fd77b68a4d56aa251e980328600b : SUCCESS in 24m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/426a75ffe23a46bc943055e1ba2d74e8 : SUCCESS in 1h 42m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/9fbf5f8d24fc4cb387901b33e54cc3c9 : SUCCESS in 2h 13m 52s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ad4338eb48564949aaf2afd7f8fa1946 : SUCCESS in 50m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3991da1dd338442fb458a5924fd461cf : SUCCESS in 2m 44s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d57685ad0bd4fc2bb0840ac0d95721a : FAILURE in 29m 27s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/187488b8d7c14dbc9c9a478ab73fc046 : FAILURE in 25m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/18dca3e56caa41a38ef5841346aa218a : SUCCESS in 1h 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f2615f0658f4755959a1703605583b7 : SUCCESS in 30m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4501247bd7a24aa4bf3ba45e56214660 : SUCCESS in 20m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a0b75a73b9034d558054d3c6778fbd96 : SUCCESS in 20m 52s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/03df3a620c034efd86490aadacadd79e : SUCCESS in 10m 26s","accounts_in_message":[],"_revision_number":4},{"id":"d5c7d778f0afa4d829395d598fceb3e3d07ef0a9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-24 17:41:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1b748a84112fc56f2549b6269e2828c9180840a6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-24 17:42:26.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"ae14617407d9e54ceaecaad5a4f7dc27de722d04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 18:16:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15c94bc860647c18db717b8e7fca677\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f217ba4ab698475b958d476ad62dba67 : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5db88bc893514757bef4a9b9be145edc : NODE_FAILURE Node(set) request cd21dd1a0ae4422d891cd1226e9163cc failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c2795cf156e4c1bbb5f2d7003c97f96 : NODE_FAILURE Node(set) request 81887f2321344b54943003e4201bbfff failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"48d8073796b9f899ac7dbe458330cdc37ba8c810","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-24 18:40:47.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":5},{"id":"09b848acf3d8592c8130e570622313edec782813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-24 19:58:57.000000000","message":"Patch Set 5: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ece0c9f056a1409d8b8a25535d1f3eb5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de8a2a72d3cf4dd588f6ec0702d44378 : SUCCESS in 1h 00m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0a66209e46e344169fbf063d4e729bd1 : SUCCESS in 1h 02m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecd811e3ad8940fba34cd3c2122a3b6b : SUCCESS in 1h 01m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25106bdb971a4c71beca8eed16c628c5 : SUCCESS in 26m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde45e1074c64ca7bc8abdef5d430ba4 : FAILURE in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad2d94626ffa4291af2adb32bad99c81 : SUCCESS in 12m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f177ea12a1584844a9e85e771bf1718c : SUCCESS in 14m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad0653fdb6e24c2eace8da67c8e6ed7f : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0baffa37d044ab49efc55cd58cb9338 : SUCCESS in 12m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fa340e1f72f5417e835098d03ce3d1cc : POST_FAILURE in 11m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ac0eb3558894d22b228cfe19af659fa : SUCCESS in 1h 52m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8988a05aa71c4bba9ad29d9afabca846 : SUCCESS in 56m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/866b6c7128e7452a80ebb615f973a169 : SUCCESS in 41m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cef3958986c645a2a19d27dd39a1d104 : SUCCESS in 2h 11m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/439ae13121f443bcb183df11e66e4dc0 : SUCCESS in 2h 08m 00s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/13f29cc7ac3348c7b3309eca69f903a5 : SUCCESS in 52m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/27b422a79db641c78e4210befc04c539 : SUCCESS in 3m 12s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1896018038c843bd88b347047ac7b05b : SUCCESS in 28m 39s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/55ca8c8a6c9d4962abafb3e098edc1a7 : SUCCESS in 33m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3293a9d68a3a48aebd8e73ad3193b3fe : SUCCESS in 54m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbdb49aee5da4d0ca75e5b7fbc286192 : SUCCESS in 44m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/cdb201290f494a53acea7e3db0e29790 : SUCCESS in 17m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dd99f7a64de348bdbd36951ad028fbef : SUCCESS in 24m 15s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/1f103616748f47b6aff8a1bf30359f4b : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":5},{"id":"7a5d32d3f8fb97bbf7d34acc58380ef8643c6e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-31 17:56:47.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"41f1d29c9c41120dccbee495a6ece84359ec42e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-12-31 18:10:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a5afb234214cf6d91267216da58d9f91c557a6db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 18:35:18.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0083381b5c294c068b366461c9434aba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9430c734e0a4963954aef009dc4045e : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c8b15f3037824f64aaf08fd45974d9ca : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f602001888504819949bb3cfa6195965 : SUCCESS in 19m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f0146a88b90cdf41a9ed347f3e646a279ca18a18","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-31 18:56:33.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":6},{"id":"c1697fc06006cabb01db1c935f2021a670a7ac58","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-31 19:04:24.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I4bf6294949f9c72ae0bbc111656507ce6603ed88 .","accounts_in_message":[],"_revision_number":7},{"id":"5dd80cfe826956eff1cfd497cb83ef667df86a79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-31 20:08:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0663aa19520048e8a1054cdfffc7ea75\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e853efc2c2cf4a3c9579ba1ce7bcbdf2 : SUCCESS in 1h 01m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/890ce266ca6b45d4a5fe73e5a4743fa2 : SUCCESS in 1h 52m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36f2f49ddf1b4ce7b902d7a29757a73e : SUCCESS in 32m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a4595a0283d4ad493b7fb70eb206c83 : SUCCESS in 24m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774801735ccf43e29ba6569a8ec31213 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6655164ee7ec493e83a842cedb623e65 : SUCCESS in 13m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dccfcb0a66194eeab2a57506831193d9 : SUCCESS in 9m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7f105d953636482cb26a47bee852471d : SUCCESS in 17m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e86a571bfbc4073a5ca80b09db1c43d : SUCCESS in 11m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8492b2173ece4ded9142b6dd6167ac43 : SUCCESS in 1h 09m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ae077b78737d4b7a8989c27992754414 : SUCCESS in 1h 44m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ebe62b5ac5c747c985265884bb9005f3 : SUCCESS in 56m 50s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/02f68dd22f5449b1b90b594c4691bd46 : SUCCESS in 41m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e77e80889cfa408987fab1abbcaac71b : SUCCESS in 1h 31m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/65d4ac4a3ab7487ca76b00ee87128a8d : SUCCESS in 1h 24m 09s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/76a32a69b46f412ca456a14d034b00e0 : SUCCESS in 39m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b32d34baea054c9e99c370c1841e289f : SUCCESS in 2m 42s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/072a2a4064fd4fb0a46facdc0d9606c4 : SUCCESS in 27m 48s\n- nova-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5553118cd84040f691ca49a4f3df0f32 : SUCCESS in 34m 49s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8b030f5664a4805a9343b7d2d160e44 : SUCCESS in 1h 27m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40b64f1bb2e54b1c8a5d7b059428092e : SUCCESS in 41m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f268a22e4b7b46aa84ed99b8ea9e4603 : SUCCESS in 28m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/857db609447140e482e78e81e1110bfe : SUCCESS in 26m 11s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/bbbfdb93126d423594b471a1a96d1498 : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":7},{"id":"365d858d01b4560a11500c4eae00a45a169e9615","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2025-12-31 20:15:44.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with \u0027run-Lightbits CI\u0027 to trigger a recheck. \nFor more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/LightbitsLabs_CI","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"e7bfa40c5f5a43483b500b8b162c6033b71235f6","revisions":{"74924d9e2f1f2bc446bfafdbcf3ae9f32be04b8e":{"kind":"REWORK","_number":1,"created":"2025-12-03 11:52:11.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/1"}}},"commit":{"parents":[{"commit":"23b462d77df1a1d09c43d0918bca853ef3af1e3f","subject":"Merge \"[ironic] Ensure unprovision happens for new states\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23b462d77df1a1d09c43d0918bca853ef3af1e3f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:52:03.000000000","tz":0},"subject":"WIP: Try and reproduce nova scheduler race","message":"WIP: Try and reproduce nova scheduler race\n\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74924d9e2f1f2bc446bfafdbcf3ae9f32be04b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/74924d9e2f1f2bc446bfafdbcf3ae9f32be04b8e"}]},"branch":"refs/heads/master"},"1f04411ab8fc90262a824bc93d8de29ca7a9758c":{"kind":"REWORK","_number":2,"created":"2025-12-03 18:21:25.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/2"}}},"commit":{"parents":[{"commit":"23b462d77df1a1d09c43d0918bca853ef3af1e3f","subject":"Merge \"[ironic] Ensure unprovision happens for new states\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23b462d77df1a1d09c43d0918bca853ef3af1e3f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 18:16:41.000000000","tz":0},"subject":"WIP: Adds regression test for bug LP#2132984","message":"WIP: Adds regression test for bug LP#2132984\n\nRelated-Bug: #2132984\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f04411ab8fc90262a824bc93d8de29ca7a9758c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f04411ab8fc90262a824bc93d8de29ca7a9758c"}]},"branch":"refs/heads/master"},"868093a6309e5de5323a2666dafaae2c81c96bf2":{"kind":"REWORK","_number":3,"created":"2025-12-05 11:15:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/3"}}},"commit":{"parents":[{"commit":"23b462d77df1a1d09c43d0918bca853ef3af1e3f","subject":"Merge \"[ironic] Ensure unprovision happens for new states\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23b462d77df1a1d09c43d0918bca853ef3af1e3f"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-05 11:15:08.000000000","tz":0},"subject":"WIP: Adds regression test for bug LP#2132984","message":"WIP: Adds regression test for bug LP#2132984\n\nRelated-Bug: #2132984\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/868093a6309e5de5323a2666dafaae2c81c96bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/868093a6309e5de5323a2666dafaae2c81c96bf2"}]},"branch":"refs/heads/master"},"cfcfd2d2f2e3414b2c63b0822d61a143edad3c92":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-12-24 10:09:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/4"}}},"commit":{"parents":[{"commit":"4b71dab239b4f82d50297ec5bd887c207b43b7ea","subject":"Merge \"Remove tpool_execute as it is unused\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b71dab239b4f82d50297ec5bd887c207b43b7ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-22 14:16:03.000000000","tz":0},"subject":"WIP: Adds regression test for bug LP#2132984","message":"WIP: Adds regression test for bug LP#2132984\n\nRelated-Bug: #2132984\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfcfd2d2f2e3414b2c63b0822d61a143edad3c92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfcfd2d2f2e3414b2c63b0822d61a143edad3c92"}]},"branch":"refs/heads/master"},"43a6085b86f5342505e91980d60d204093131dd9":{"kind":"REWORK","_number":5,"created":"2025-12-24 17:41:52.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/5"}}},"commit":{"parents":[{"commit":"4b71dab239b4f82d50297ec5bd887c207b43b7ea","subject":"Merge \"Remove tpool_execute as it is unused\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b71dab239b4f82d50297ec5bd887c207b43b7ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-24 17:41:35.000000000","tz":0},"subject":"Adds regression test for bug LP#2132984","message":"Adds regression test for bug LP#2132984\n\nRelated-Bug: #2132984\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43a6085b86f5342505e91980d60d204093131dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43a6085b86f5342505e91980d60d204093131dd9"}]},"branch":"refs/heads/master"},"fd4de2506499420ef6b9f99456bbd42ef22b54d2":{"kind":"REWORK","_number":6,"created":"2025-12-31 17:56:47.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/6"}}},"commit":{"parents":[{"commit":"4b71dab239b4f82d50297ec5bd887c207b43b7ea","subject":"Merge \"Remove tpool_execute as it is unused\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b71dab239b4f82d50297ec5bd887c207b43b7ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-31 17:56:35.000000000","tz":0},"subject":"Adds regression test for bug LP#2132984","message":"Adds regression test for bug LP#2132984\n\nRelated-Bug: #2132984\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd4de2506499420ef6b9f99456bbd42ef22b54d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd4de2506499420ef6b9f99456bbd42ef22b54d2"}]},"branch":"refs/heads/master"},"e7bfa40c5f5a43483b500b8b162c6033b71235f6":{"kind":"REWORK","_number":7,"created":"2025-12-31 18:10:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/80/969380/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/80/969380/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/80/969380/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/80/969380/7"}}},"commit":{"parents":[{"commit":"4b71dab239b4f82d50297ec5bd887c207b43b7ea","subject":"Merge \"Remove tpool_execute as it is unused\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4b71dab239b4f82d50297ec5bd887c207b43b7ea"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-03 11:51:19.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-12-31 18:08:20.000000000","tz":0},"subject":"Adds regression test for bug LP#2132984","message":"Adds regression test for bug LP#2132984\n\nRelated-Bug: #2132984\nChange-Id: I4bf6294949f9c72ae0bbc111656507ce6603ed88\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7bfa40c5f5a43483b500b8b162c6033b71235f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7bfa40c5f5a43483b500b8b162c6033b71235f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
