)]}'
{"id":"openstack%2Fnova~971299","triplet_id":"openstack%2Fnova~unmaintained%2F2023.1~I8f4124c98a4ed81d0048f62d762091afc73a5452","project":"openstack/nova","branch":"unmaintained/2023.1","topic":"storpool-backports-2023.1","hashtags":[],"change_id":"I8f4124c98a4ed81d0048f62d762091afc73a5452","subject":"Hardcode the use of iothreads for KVM.","status":"NEW","created":"2025-12-17 09:35:21.000000000","updated":"2026-01-03 03:11:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":918669,"cherry_pick_of_patch_set":4,"meta_rev_id":"1fc4650ec8a11a5cebccc7b61aadad22c64ec88d","_number":971299,"virtual_id_number":971299,"owner":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-12-17 11:30:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2025-12-17 09:35:21.000000000","updated_by":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"reviewer":{"_account_id":12988,"name":"Peter Penchev","email":"openstack-dev@storpool.com","username":"ppenchev"},"state":"CC"},{"updated":"2025-12-17 10:45:02.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-17 11:30:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-03 03:11:07.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"eda3f13d29f81a062ec1f2300656e9e07f5f4475","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-12-17 09:35:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* nova/tests/unit/virt/libvirt/test_config.py","accounts_in_message":[],"_revision_number":1},{"id":"367f78261b299298d5501bf80972620dcb465339","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"date":"2025-12-17 09:35:50.000000000","message":"Topic set to storpool-backports-2023.1","accounts_in_message":[],"_revision_number":1},{"id":"50712588bbc71d60d2952d4790b3241edbd3d301","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2025-12-17 10:45:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I8f4124c98a4ed81d0048f62d762091afc73a5452 .","accounts_in_message":[],"_revision_number":1},{"id":"f4ece3cb2f4a29f767a4882700cc04731f576d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 11:30:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d064607d7484003a0531b6decde8a28\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/756e3876e7984164b774668de63e914d : SUCCESS in 27m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fc6628468df3438fa2389373c0c9f109 : FAILURE in 34m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07c4d4c7a7ac4ed8adddfeac1be10e94 : SUCCESS in 54m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80392b5a71be43408399e22f855ac723 : FAILURE in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d348006063d748dd8e64f83764cdc8fa : FAILURE in 9m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade0f1e0ba474b7f836fbad1eed55e63 : SUCCESS in 10m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/661200f1df204bbe96a33654da648018 : SUCCESS in 1h 00m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d441933db6c446c8c041d3a5ff0619f : FAILURE in 1h 49m 52s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6191468b1bac4bff80530b21a6193b07 : FAILURE in 1h 00m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/84045579181348bf87dfd9e3e63bb3fa : FAILURE in 4m 55s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97b05940c3bb4c64836278f8e409f323 : SUCCESS in 22m 43s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3ca61d5f8bf42aea7346805618ad866 : SUCCESS in 28m 29s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/35905ce1bdf34ed38ace4c85029dc8ce : SUCCESS in 25m 23s\n- tempest-integrated-compute-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92beb08f7f34001a0d1ddbdc437be15 : FAILURE in 1h 27m 34s\n- tempest-integrated-compute-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88c446e8f4bf4de5902413bfeaa84938 : FAILURE in 1h 33m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/83818e24403b4108bb503e60c75af9da : SUCCESS in 1h 15m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb94d9f0a72543c594b42be3c4be65ce : SUCCESS in 49m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d3d66e38279b4fafb1efdeb5e05b7ecc : SUCCESS in 25m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e0013650dd6d4d89a5d3211d9fbeeae8 : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fc4650ec8a11a5cebccc7b61aadad22c64ec88d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-01-03 03:11:07.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-971299-59gzj/index.html\u003e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a00c657becc021eca4c83964b0ad4a40bd4ebbc5","revisions":{"a00c657becc021eca4c83964b0ad4a40bd4ebbc5":{"kind":"REWORK","_number":1,"created":"2025-12-17 09:35:21.000000000","uploader":{"_account_id":35429,"name":"Biser Milanov","email":"biser.milanov@storpool.com","username":"sp-bmilanov"},"ref":"refs/changes/99/971299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/99/971299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/971299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/971299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/971299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/99/971299/1"}}},"commit":{"parents":[{"commit":"6f8decf0b4f1aa2e96292b6a2ffc28249fe4af5e","subject":"Update Nova bdm with updated swap info","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f8decf0b4f1aa2e96292b6a2ffc28249fe4af5e"}]}],"author":{"name":"Peter Penchev","email":"openstack-dev@storpool.com","date":"2021-07-15 16:32:28.000000000","tz":180},"committer":{"name":"Biser Milanov","email":"biser.milanov@storpool.com","date":"2025-12-17 09:35:21.000000000","tz":0},"subject":"Hardcode the use of iothreads for KVM.","message":"Hardcode the use of iothreads for KVM.\n\nAdd the iothread_count Nova configuration option that controls\nthe libvirt specification for qemu I/O threads to be used.\nIf set to a non-zero value (usually 1), the libvirt definition of\nthe Nova instance will contain the \"iothread\" attribute for\nvirtio-scsi controllers and for /dev/vd* block devices.\n\nChange-Id: I8f4124c98a4ed81d0048f62d762091afc73a5452\nSigned-off-by: Biser Milanov \u003cbiser.milanov@storpool.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a00c657becc021eca4c83964b0ad4a40bd4ebbc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a00c657becc021eca4c83964b0ad4a40bd4ebbc5"}]},"branch":"refs/heads/unmaintained/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
