)]}'
{"id":"openstack%2Fnova~978121","triplet_id":"openstack%2Fnova~master~Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8","project":"openstack/nova","branch":"master","topic":"upstream-livemig-portbinding-cleanup","attention_set":{},"removed_from_attention_set":{"37800":{"account":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"last_update":"2026-09-25 13:06:16.000000000","reason":"\u003cGERRIT_ACCOUNT_37800\u003e replied on the change","reason_account":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"}}},"hashtags":[],"change_id":"Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8","subject":"conductor: Clean up port-bindings for live migrations","status":"NEW","created":"2026-02-26 18:14:20.000000000","updated":"2026-09-25 13:09:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"16b4554ee560bdc47f0aca8720433a2bcff7f233","_number":978121,"virtual_id_number":978121,"owner":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-24 15:56:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},{"_account_id":38925,"name":"Roman Hros","email":"roman.hros@dnation.cloud","username":"rhros"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 18:56:44.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-26 19:10:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-26 20:15:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-27 07:37:32.000000000","updated_by":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"reviewer":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"state":"CC"},{"updated":"2026-02-27 08:34:41.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 16:10:52.000000000","updated_by":{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},"reviewer":{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},"state":"CC"},{"updated":"2026-07-31 13:37:51.000000000","updated_by":{"_account_id":38925,"name":"Roman Hros","email":"roman.hros@dnation.cloud","username":"rhros"},"reviewer":{"_account_id":38925,"name":"Roman Hros","email":"roman.hros@dnation.cloud","username":"rhros"},"state":"CC"},{"updated":"2026-08-05 12:03:10.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"c7bbbab06257dc444a85cff18998c0c4078ec6fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-02-26 18:14:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2435040ef594fd1d4d253c7b449642861e88c09e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-02-26 18:56:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8 .","accounts_in_message":[],"_revision_number":1},{"id":"b5d0a656adf61ca2a9febb6afda6d7240757e10b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 19:10:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e664fcee37364a38906489039a2e7bf6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0f1f83e0c9f94aabbeeb5fb29b89cb60 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a24b6f67c0b84e139081e32c1ccbb920 : RETRY_LIMIT in 5m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2f4803aad29f43398fafe3caaa99fd96 : FAILURE in 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c6492533f563dff2e84254decbc211ec551fc2c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 20:15:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdc75cab435f4331907173a28cdd6b2a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cc1844cfd45540f6b0528549883fa498 : SUCCESS in 48m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c5d19798ab204a3c9f2c1b180ef4b8b0 : SUCCESS in 1h 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0fcdbe175154fa19afa5ad67a20208d : FAILURE in 58m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43190e39cbc542d7946414d0d5c8fcdb : SUCCESS in 26m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e6b1e8314a4326af83b5f5395f97dc : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a019c40fbd240fca69a13516fbf7c6a : SUCCESS in 18m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfb91d11b14e40709b546169e7e89395 : SUCCESS in 15m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/967f403179a34db8af75038e9d5c4268 : FAILURE in 12m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396d44e50a3e4a0fb1dc2232b81a4b2d : SUCCESS in 6m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e6394e3f4c8439d904975429286d849 : SUCCESS in 1h 02m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b8fa55ec48ad450295700c121e7db41e : SUCCESS in 33m 09s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/739775856c004cdd8d0b54af98993fed : SUCCESS in 43m 49s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bba1e8797c5e4760990ddb987c21c577 : SUCCESS in 52m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0a68b26b4a0c440792698887dbc9b3ed : FAILURE in 1h 17m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/caf80b53146e4f57b31e927dc81dd88d : SUCCESS in 1h 59m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/89f6adf55b0a4178868e94a4a78b4ce4 : SUCCESS in 50m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b160055d0c6d471a8bf462682bae81a2 : SUCCESS in 5m 09s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0103316240643e3b8cf323ab8fad399 : SUCCESS in 31m 28s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ddcbeb5b57524577ba3a7383fce13c55 : SUCCESS in 27m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/28f34f7779634c0a888c33abf7296069 : SUCCESS in 1h 26m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d932df7e28e3440da6f73f98276d6199 : SUCCESS in 34m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/509ddc66a42f4a73b22a04477e852b71 : SUCCESS in 21m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7c46c26727d94b99ba8085fb10553fea : SUCCESS in 34m 38s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/2eb50483e98248e0a559f113e0ffa663 : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4005f8eadff46c21a8460f225ad8eab0bc5e1fdc","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-02-27 08:34:41.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-978121-x5dq6/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"8e9c0b609c7549abe770a85b8576a9c9861705b9","author":{"_account_id":26250,"name":"Johannes Kulik","email":"johannes.kulik@sap.com","username":"jkulik"},"date":"2026-03-04 09:10:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a186f2d06bd9fa2dd17d2e8b2748f6cda5b6541","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 10:16:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec8a829995604f94a59175bb2e98a19a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a94df955748348efb7e5919426303d29 : SUCCESS in 45m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4f5cb0cd8da47c2be3ac11b0c776a84 : RETRY_LIMIT in 5m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4e6ebad7fd7346458494c30c00d38529 : FAILURE in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bac8b4a6effcc29aca6b062cff317120b6ebfb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-04 11:38:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c8d791b3b4f429fbc7168c6236a13fc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7cb7a229c054878a43c74ac57741e83 : SUCCESS in 57m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/332969df9bcd472b80f18942a11de545 : SUCCESS in 1h 55m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f848539e387f48719a28a97bf00e8115 : SUCCESS in 44m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdb77119f60b4ef4a3efe37ec5ef6133 : SUCCESS in 1h 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65605fa3f91044ca9baa3c391d617a77 : SUCCESS in 8m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7534c32e258542179a2798de20cb62ea : SUCCESS in 16m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a99d85faf9a3440aae10a28293836d73 : SUCCESS in 13m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1738887425e9426d9ea00691ae4d92fe : FAILURE in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cd9c5475b1b4ea79d9ffd4d48cf4827 : SUCCESS in 8m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e0c7057b230e4aab966cf227b0ae1696 : SUCCESS in 1h 44m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/317d2d86f7dc4372af5b9faa99c89adb : SUCCESS in 58m 15s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/73d823aaf0bd49f5a84ff5448a6d1893 : SUCCESS in 1h 04m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/eadbaaa82b12411f8e5f205845d1190b : POST_FAILURE in 24m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/15940c4fad2e462e99e3dba0b8c2bbfd : SUCCESS in 2h 05m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/95081286a8d740d29da5003cfa7150f1 : SUCCESS in 2h 20m 07s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/60cc258b8b6f4240bf0421c916e3bbf0 : SUCCESS in 53m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e7541fe1374f47b6a212c63ed8eab755 : SUCCESS in 5m 04s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9abcb09a5f81421ab632b8739ac32cea : SUCCESS in 27m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/32f6c40c51ce4fe8809056aeac347747 : SUCCESS in 32m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/005e6b34e9334eae94884207c7993846 : SUCCESS in 1h 22m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14adfc2e027f44799d89d0fc77339499 : SUCCESS in 1h 05m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4679924ff6394c82a4f9c93e68701c2f : SUCCESS in 28m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f21534f6417b4ca9afd45c486bc1376c : SUCCESS in 31m 59s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8b29221e3af24798adc9b135535f008a : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b6c655d64417139a516eb5ffe589eb4c711cb0a9","author":{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},"date":"2026-03-05 16:10:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3825fc7c1e70be78e49fe0833d075a0516fad621","author":{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},"date":"2026-03-11 09:25:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a251cbc0abe57f549cc7c9a156e8f90ea68ce170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 09:48:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3240ea5575949afbb3498ff97290827\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0866e7235b9a4036b30baaf1cac52d70 : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/64a5b42398bb4a51952ae265205914b4 : SUCCESS in 15m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1636cd7d2277479c95b7199577a6176f : FAILURE in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3be0b0a160db3541297ce2a90fadd1fc72e64e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-11 11:43:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3879026af6d04667b1eeff189da7205e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4960fa7fffe0441d9eedc70f20b71edb : SUCCESS in 31m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c5aaf999c1f243e1867f4685642fe7a1 : SUCCESS in 2h 09m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61d00b7539414ee9b9c55f77d47ea776 : SUCCESS in 1h 06m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f67e57effdb441d2b7fb7cf78e53e769 : SUCCESS in 1h 04m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e31ae3565d4047b9665ccbc3c2784d : SUCCESS in 7m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc4dc7ea68d64443bb2a07ae0a6e35b9 : SUCCESS in 10m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4115cc2dd4eb423ea759546e6521b5ba : SUCCESS in 13m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9d9dae3ee9a04e7ea710cca039404b12 : FAILURE in 10m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6019ddc60fe84ce9b7bf6140927c631a : SUCCESS in 12m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f7305b838b04cb7b4b594d0757b16d3 : SUCCESS in 1h 18m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f57ba9d410e942b2a2f72b347539284f : SUCCESS in 30m 59s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/da558986db864510b6b96b0ebb625a76 : SUCCESS in 1h 02m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7a0fb6b12e61482aa10424273bb7a097 : SUCCESS in 41m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/32289177504845eeb184ecda4c259533 : SUCCESS in 1h 54m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/a91de3908c7847b18b6ac8627bcaa271 : SUCCESS in 2h 09m 52s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7b40d1fd52c3408694c6c9032bcc35cc : SUCCESS in 56m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/77fc84a6f6e249f291878f92e4aa8918 : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7c1de2af49f549d38d41f36eb30166cf : SUCCESS in 26m 17s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e8cb06204b5d4cb29e7c9daf81361360 : SUCCESS in 25m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ed7594956ae4bb58331f396d27b21ca : SUCCESS in 1h 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa901045f48f4ecb94cc725ed546b52b : SUCCESS in 1h 08m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/83c3829da25d442ba61c65f0441325ad : SUCCESS in 29m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/206af582c10b4770b38bea30e6b5ddd8 : SUCCESS in 34m 40s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8ab68c488b2f47d8881b6bffb3985b81 : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7dd693b4647ba0914d900eb2125af89f1154928f","author":{"_account_id":38925,"name":"Roman Hros","email":"roman.hros@dnation.cloud","username":"rhros"},"date":"2026-07-31 13:37:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cd4db6dbf32364e04781778f9d6f4bf43c72791c","author":{"_account_id":38925,"name":"Roman Hros","email":"roman.hros@dnation.cloud","username":"rhros"},"date":"2026-07-31 13:39:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a1cb01b0848ce5a4fcc21bfbb9ac254e1419471","author":{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},"date":"2026-08-04 19:08:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9e46d7858dc78b2fa891d0c6b3ee08e9c58d5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-08-04 19:48:52.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"124859cb71831edf865595cdd2f094817a9c23af","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 20:43:56.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8 .","accounts_in_message":[],"_revision_number":2},{"id":"08adc70672dd809c7a684951d5b3d45c03cc2007","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 21:03:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/919ffbe1ca324ea8958b75d5d9f0253f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9b5c7e1ff887443aa5a2d27cfab597d7 : SUCCESS in 13m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f944a6ee022b45f2977de7877a85788b : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4d5d76e37b9a47289ae0dc94f1389ab5 : SUCCESS in 42m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fe6a8b8093a94eba767b5f2ce2cc2c86a4ae1b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 21:43:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f69cf8ed5ecc497583359ea92be05f9c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef4618fbda8a4d02addc52bd198208c1 : SUCCESS in 30m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ba1a0cfd15f74728b59e86ed29411191 : SUCCESS in 1h 25m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1602d568b2aa45ceb58bb0ee50d524a6 : SUCCESS in 55m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18d3679e0cc742bdbbcabf09ad34a817 : TIMED_OUT in 1h 41m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0afd98ef781b4c478840a13b88171a19 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5daf2ddaf58d4f7ba36453d8b1f4612f : SUCCESS in 14m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d376143b61234e6f97e96ddcc51a9277 : SUCCESS in 14m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da655d762bb44ff4a263f984b7e22e1e : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654fb94093a94cfe94e52b050f1f08be : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f48546c56ca48e69a0c84d2c16a3c23 : SUCCESS in 5m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/189ac6a0c5e6442c8a9fa2b036951925 : SUCCESS in 54m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f11c3655d2b74a17b85a75e6d613ec7a : SUCCESS in 1h 12m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9c637e86d37e4a4fa0b60e2ba914fb54 : SUCCESS in 49m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/aafc58953e1c4d53bb1eff00fd3860a2 : SUCCESS in 36m 11s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/aab692c2da614984ace85c7150079515 : SUCCESS in 1h 00m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2eaa134278534c1c8a98a5330591bd3a : SUCCESS in 27m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4637377b2eae45ba980698fed7d8526d : SUCCESS in 52m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/5cf99e14872349c6b795a9aef89f28a0 : SUCCESS in 1h 49m 39s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4a2ba39a863a4608b7a6a5c2512f849f : SUCCESS in 46m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f149e13c638c426db905500ccb78a479 : SUCCESS in 2m 53s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8e8f1cf9b30a44f8b4f153ab14e700bb : SUCCESS in 28m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6ef9bc2f7e4e408d983014097163480c : SUCCESS in 26m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/2322db4b4d684a6f90500439186fbb1b : SUCCESS in 29m 03s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5b3e2e2167394a49beb64b76e6f12508 : SUCCESS in 56m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5422c82a03345f4827f26a0efa69f0f : SUCCESS in 26m 05s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5d9c8299c30b4901aaf7802c63faa557 : SUCCESS in 54m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/664849220890442396e11ec9c38d5d3e : SUCCESS in 33m 05s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1d83a354141c4ab097e453982c1689f0 : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":2},{"id":"b14a591692255ab2846815b5eec0256dde3d1f98","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-05 02:57:22.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-978121-2v8pl/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"f4de942d96c9c884283035a7e7e0235a804ac922","author":{"_account_id":38540,"name":"Sebastian Krott","email":"sebastian.krott@gmail.com","username":"sebkro"},"date":"2026-08-05 09:39:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d44265e89556d35c4fc8835b17f1d6cdb71616fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 10:12:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb116d78e62435192b6d2b1df004f1b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bca01f9a5e349cc9500c884acea85a8 : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7777eaa76de43b8b9bc6af743ea734c : SUCCESS in 17m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d68b3607713e4b9bab2ebd5374e26d0f : SUCCESS in 29m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc742a06e8c62e1037640f190291a7b6f97c3751","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 12:03:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d17d1d4d3b4db087a3816e355d8e6e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2bf7468e85334b0c8ffb1681a06d356d : SUCCESS in 44m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f8712fba53c3426494c295a819922439 : SUCCESS in 1h 33m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61c906b6762046efa1b0282357c410bc : SUCCESS in 1h 01m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5945875e9240e08761ca4b034c0c85 : SUCCESS in 1h 05m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fb3c60c4f04769ba334223b24d3ad6 : SUCCESS in 7m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5f3c24b81f64dcea3670ef3358dd611 : SUCCESS in 10m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a1b8d50d8c534e5cb5f627e9cc8a2258 : SUCCESS in 10m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0bd57a031054d408e8c69a1bc79ce44 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02927981a4324d2cb65667cb89c8f7c0 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49ee24f687b44257afcde55dcc5c0b54 : SUCCESS in 7m 52s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/8ec3b6993f5b456d898357071c48534e : SUCCESS in 1h 00m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2b2fa2df880a4cab99767d38d5420945 : SUCCESS in 1h 10m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f36a632f8e98480e86a97ac9abac7993 : SUCCESS in 22m 25s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d09e414ebd9b41448c1cd5195ff7c126 : SUCCESS in 41m 18s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/01899ba811f34e9eaeb478442a75d46a : SUCCESS in 48m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/68a2c472f3b7427da4ca83bd0118acbf : SUCCESS in 49m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f0c472fdf402441f80943d503264c2c5 : SUCCESS in 58m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/c8587f0947494d90be36f5c2ed612d4f : SUCCESS in 2h 15m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e1099422cd1548dbaaa6965f9c35c8f4 : SUCCESS in 40m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/32131068722d4ac8ae466b58f97fc9e9 : SUCCESS in 4m 49s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ab7fa49acb414d70b2e5a9af6e7a7f13 : SUCCESS in 28m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3b27be87abe4465c99287e030fb98524 : SUCCESS in 24m 21s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5f7ca89a4dce4898972267cf293d034e : SUCCESS in 35m 18s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/34166a085fd74fd8b9340e1ce3998769 : SUCCESS in 1h 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39490e889f3b46f4974cc8f339f59ce8 : SUCCESS in 53m 53s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6110b5ada5764b17a3390e41896f051f : SUCCESS in 38m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d60d21c4018049f983ade683a4e9b01a : SUCCESS in 33m 38s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4d2608001d224e1ca5e948007c4bca30 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3a5123bf5140cc41829f41cbb0a29764cbbd3011","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-05 12:13:05.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"298f1fb1fafddf30fe6ce357b7d9360818943483","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-05 19:33:43.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-978121-7wglq/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"12375c3c6ec5b5f7b9d2fc31ff3ada12a4595f6c","author":{"_account_id":38925,"name":"Roman Hros","email":"roman.hros@dnation.cloud","username":"rhros"},"date":"2026-08-11 11:39:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0599aa568da8600ff5a940751a52bb24458405d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-09-24 13:46:19.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"44853038c0e02921c9ac9d15394ef934f2443f5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 14:15:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d0b73156e7c4db3b598eb96f2b9395f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c15474036c3d4f88bc53d0efd9aed78a : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/39b8bc2975f6429984f6c493bf06cf7c : SUCCESS in 27m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45e5edb0c945a1661894980814e25de45b6500de","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-24 14:32:11.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-978121-j4kfj/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"a9767f4611dd9cd69ae9bd32c115108d9d0bb606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-24 15:56:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae5485650c404ce69387205d83853738\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed39cc997d7a4460aa354067d5a59564 : SUCCESS in 32m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/56eb2b72f3a3484582b2c44b9d4c2926 : SUCCESS in 2h 00m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/083aad732b2c4df99d876ccc77d6b100 : SUCCESS in 37m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43a2e200089244d4bf8548b6fcd533b0 : SUCCESS in 56m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b1e07c944a4695809072012cd7bc51 : SUCCESS in 6m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4501ce1ca5234f43a55bb102cea3d2ec : SUCCESS in 15m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3df2e9acf1a941019891fddcc87b979d : SUCCESS in 11m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65bfb01b310b45549c09b0b7167eab80 : SUCCESS in 13m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7de6663eefb04c24807a000f868e8e38 : SUCCESS in 4m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/85a0e4ea7c4844a8b35d916cfec8fc40 : SUCCESS in 1h 11m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1db5acf69e354eb79384b6711744bdfd : SUCCESS in 1h 00m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5ceaf81320f5492f87609ac9ea207dc5 : SUCCESS in 54m 41s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5eac83aa3c6643cebfcca5a3f5095a08 : SUCCESS in 36m 47s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/040355fae87545fd8d4a0afc91dd2b42 : SUCCESS in 45m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7e171449d0b64d17a0ff01cdb6b57e1a : SUCCESS in 33m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/38b1a4efac2441499ec97f6daa82aa45 : SUCCESS in 51m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/31f98bf8154e4fd2bfdcd8af66004a11 : SUCCESS in 2h 02m 59s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f59ca7f5bbed48c9b8aa7894808b0292 : SUCCESS in 48m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b990a99270d44c52a8529d7dac9ac2ba : SUCCESS in 4m 07s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/10eebf7026e442278bef7aecf42a0631 : SUCCESS in 27m 25s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a708310fea2f4ae195e8e4b47732e733 : SUCCESS in 27m 02s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6ff7c189010f4578b0a13bd0da369eba : SUCCESS in 34m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9c05245e4a4f46b89cde6326741d755a : SUCCESS in 1h 12m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32053d383ff64f30a8fc7619d8ec6fd9 : SUCCESS in 37m 12s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf477b135e674a978e0dd17f298bf5af : SUCCESS in 56m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ce8c18722f464f9ba39c162a1c3b6a02 : SUCCESS in 34m 25s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a2239871d6164456b25a0ba1c0560240 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":3},{"id":"354caa2eda0721a945ff1dce67b0fad656fb7c86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-24 16:06:55.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"c9f8a6fe91dfa46c202170a9f9c9fe00c7a26757","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-25 11:02:32.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8 .","accounts_in_message":[],"_revision_number":3},{"id":"101b2cf475533551775bd10afa6d946c16776998","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-09-25 13:06:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16b4554ee560bdc47f0aca8720433a2bcff7f233","author":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"date":"2026-09-25 13:09:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a9c7a9ac7df9fef0fdf56a9cc81d324705032f8c","revisions":{"bcb174ca0ab83755d3abf1fe2e76bc3b233b0218":{"kind":"REWORK","_number":1,"created":"2026-02-26 18:14:20.000000000","uploader":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"ref":"refs/changes/21/978121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/978121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/978121/1"}}},"commit":{"parents":[{"commit":"6ae54593511ee28e8a5a9c1216b15df31313c6de","subject":"Merge \"Deprecate unlimited compute actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ae54593511ee28e8a5a9c1216b15df31313c6de"}]}],"author":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-02-17 15:53:09.000000000","tz":60},"committer":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-02-26 17:54:26.000000000","tz":60},"subject":"conductor: Clean up port-bindings for live migrations","message":"conductor: Clean up port-bindings for live migrations\n\nIf a live migration fails in `LiveMigrationTask._execute` before\nentering `ComputeManager.live_migration`, then the destination host\nport-bindings created via `_bind_ports_on_destination` will not get\ncleaned up.\n\nWe fix this by adding port-binding cleanup to\n`LiveMigrationTask.rollback`. To avoid live migration errors, we\nadditionally delete orphaned port-bindings (e.g. from failed previous\nlive migrations or other causes) on the destination host before\nattempting to bind ports in `_call_livem_checks_on_host` during a live\nmigration.\n\nCloses-Bug: #2142778\nChange-Id: Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8\nSigned-off-by: Sebastian Krott \u003csebastian.krott@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bcb174ca0ab83755d3abf1fe2e76bc3b233b0218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bcb174ca0ab83755d3abf1fe2e76bc3b233b0218"}]},"branch":"refs/heads/master"},"fe9107302449a41e1d79c893a5727832c27b0c0c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-04 19:48:52.000000000","uploader":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"ref":"refs/changes/21/978121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/978121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/978121/2"}}},"commit":{"parents":[{"commit":"6ae54593511ee28e8a5a9c1216b15df31313c6de","subject":"Merge \"Deprecate unlimited compute actions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ae54593511ee28e8a5a9c1216b15df31313c6de"}]}],"author":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-02-17 15:53:09.000000000","tz":60},"committer":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-08-04 19:47:05.000000000","tz":120},"subject":"conductor: Clean up port-bindings for live migrations","message":"conductor: Clean up port-bindings for live migrations\n\nIf a live migration fails in `LiveMigrationTask._execute` before\nentering `ComputeManager.live_migration`, then the destination host\nport-bindings created via `_bind_ports_on_destination` will not get\ncleaned up.\n\nWe fix this by adding port-binding cleanup to\n`LiveMigrationTask.rollback`. To avoid live migration errors, we\nadditionally delete orphaned port-bindings (e.g. from failed previous\nlive migrations or other causes) on the destination host before\nattempting to bind ports in `_call_livem_checks_on_host` during a live\nmigration.\n\nCloses-Bug: #2142778\nCloses-Bug: #2125456\nChange-Id: Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8\nSigned-off-by: Sebastian Krott \u003csebastian.krott@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe9107302449a41e1d79c893a5727832c27b0c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fe9107302449a41e1d79c893a5727832c27b0c0c"}]},"branch":"refs/heads/master"},"a9c7a9ac7df9fef0fdf56a9cc81d324705032f8c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-09-24 13:46:19.000000000","uploader":{"_account_id":37800,"name":"Sebastian Krott","display_name":"Sebastian Krott","email":"sebastian.krott@sap.com","username":"sebkro_sap"},"ref":"refs/changes/21/978121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/978121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/978121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/978121/3"}}},"commit":{"parents":[{"commit":"cf96eb8e3549e532690afe00f3ce0aa48f37204f","subject":"network: Add host filtering to `cleanup_instance_network_on_host`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cf96eb8e3549e532690afe00f3ce0aa48f37204f"}]}],"author":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-02-17 15:53:09.000000000","tz":60},"committer":{"name":"Sebastian Krott","email":"sebastian.krott@sap.com","date":"2026-09-24 13:45:32.000000000","tz":120},"subject":"conductor: Clean up port-bindings for live migrations","message":"conductor: Clean up port-bindings for live migrations\n\nIf a live migration fails in `LiveMigrationTask._execute` before\nentering `ComputeManager.live_migration`, then the destination host\nport-bindings created via `_bind_ports_on_destination` will not get\ncleaned up.\n\nWe fix this by adding port-binding cleanup to\n`LiveMigrationTask.rollback`. To avoid live migration errors, we\nadditionally delete orphaned port-bindings (e.g. from failed previous\nlive migrations or other causes) on the destination host before\nattempting to bind ports in `_call_livem_checks_on_host` during a live\nmigration.\n\nCloses-Bug: #2142778\nCloses-Bug: #2125456\nChange-Id: Ibc379277fb6e23ca7ee368f2118878a7ecfa8df8\nSigned-off-by: Sebastian Krott \u003csebastian.krott@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9c7a9ac7df9fef0fdf56a9cc81d324705032f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a9c7a9ac7df9fef0fdf56a9cc81d324705032f8c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
