)]}'
{"id":"openstack%2Fnova~979016","triplet_id":"openstack%2Fnova~master~I6668c15051a089fe4ce5808254a831ff3dacaa75","project":"openstack/nova","branch":"master","topic":"gazpacho_service_version","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-03-12 12:08:30.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-03-12 12:08:30.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-03-06 18:15:16.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-03-12 08:41:54.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2026-03-12 12:08:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6668c15051a089fe4ce5808254a831ff3dacaa75","subject":"Add service version for Gazpacho","status":"MERGED","created":"2026-03-05 14:57:10.000000000","updated":"2026-03-12 12:10:36.000000000","submitted":"2026-03-12 12:08:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979016-gazpacho_service_version","meta_rev_id":"f97729c239c67b53e123b07afbe9443f54c2eefd","_number":979016,"virtual_id_number":979016,"owner":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-12 12:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2026-03-06 18:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-12 08:41:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-12 08:41:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 15:23:51.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 15:32:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-05 17:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-06 16:13:36.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-03-06 16:13:36.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-03-06 16:13:36.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-03-06 16:16:46.000000000","updated_by":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"2c5ac780adc5add3993ba0edea8089bc9987b2d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"date":"2026-03-05 14:57:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5a24e69f69d3bc6fadd48dcadb4f03c6b9be762","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-05 15:23:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I6668c15051a089fe4ce5808254a831ff3dacaa75 .","accounts_in_message":[],"_revision_number":1},{"id":"6e800b0e6655602726b863e2984cae28befce7e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 15:32:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb366a5b96da407c85209f18d43dd609\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c50265bad7314486bbd932bfe658a0c4 : SUCCESS in 17m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3f13d74c27e44ca9b4df1a9329719f2 : RETRY_LIMIT in 6m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/95750e7ab5f142e3842ea02417b65d2a : FAILURE in 19m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3e6cef6530f267b9716fb743f5b971bdf2e0246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 17:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad8d4e3e57a649a88a56973314374aa0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0701937be9b4efa9b13bdac89ae32c4 : SUCCESS in 34m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c7bd9d981928424588b1b411f9ab2059 : SUCCESS in 2h 13m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf282310972148a0bdae10a9edccd537 : SUCCESS in 37m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00814aaba4234ffe9b6eb4d561d62b35 : SUCCESS in 49m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb3bd78a38242dda3215ff51132ff91 : SUCCESS in 9m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b17a913a4b64e4d84a35370d9c93503 : SUCCESS in 16m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19966af2143d4c5591fec0399947d701 : SUCCESS in 9m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/62a44b356bf245b8b95743fe8533016a : FAILURE in 12m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc22e93fa2f4669aac9c098d9d00a5c : SUCCESS in 12m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e45c2b1a766f438ab27a0a44c8de0808 : SUCCESS in 1h 47m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/96f3f8cc4f5141a7b2234a6b0bab3620 : SUCCESS in 57m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b26c2ee9355e49bcb11cc8c297af7d11 : SUCCESS in 1h 07m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/01ad888576c741079ab884901242e68b : SUCCESS in 1h 00m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2f388e245fa740beb25ce4f159aae3b9 : SUCCESS in 2h 13m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/121afb2ec952440f9cd4efd1fc3573bf : SUCCESS in 2h 16m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2a7626bcb1804905b8bf3946505d6468 : SUCCESS in 52m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a9b09b95af2544d98b462586effacb9c : SUCCESS in 7m 14s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d28d85d167af44bbba2cc65bf74db409 : SUCCESS in 29m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c6ce05466e54b4ba864a9e671765dbe : SUCCESS in 24m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c17cefc1c159430187017a00a17d7e76 : SUCCESS in 1h 28m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6427f472e6194d1c871c13c6a8ab1926 : SUCCESS in 58m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2195f59be7c74d67ba9d73a1c34eed35 : SUCCESS in 31m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3374900194034dc4943bf8d6794540ca : SUCCESS in 27m 38s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/8b920ae1474d4228bc29281b69c6eace : SUCCESS in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f7265a982ec2193ac344957a7e7d6fe17b85920e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-03-06 18:15:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c13cada19cf0482b2555d9f3a572ba63d06f25b","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-03-12 08:41:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f75bb7546485e7330b1ae0bdadc3ca9e62795023","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 08:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d24fabe3fe332573425b258cc26550dc519f6dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:08:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a0269819cc2487899b80d7bb5f37a38\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca82d605b888473e8e1f5674c3247ed1 : SUCCESS in 56m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ab5d02a993b64ad7b1ad684d50bcf12b : SUCCESS in 2h 00m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49cc0592a89941b7b8e8183510e8e78f : SUCCESS in 40m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd5f112151b4bfb968fa193c9caaba2 : SUCCESS in 7m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4fd2311beca49d6ab3cfc0c11eef6ca : SUCCESS in 10m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c8c19bf4a02442792c12e3f30b23a4f : SUCCESS in 8m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d2498a971804196b2a545952dcfead6 : SUCCESS in 5m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/63e83e060b674d02bc4a0387fd3f70bf : SUCCESS in 51m 38s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d72a8f619ac74ed4a714f14c1ae8dfb9 : SUCCESS in 29m 31s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/abe0db8673324e889c0d7717c404fd5b : SUCCESS in 29m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8cf2ecb07a1843f99a8c7d916e1eb4f9 : SUCCESS in 1h 11m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/8d063ad2da6844e58c0e49f908693aab : SUCCESS in 1h 23m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5d5df05947f34c84abf378ee72f2342b : SUCCESS in 5m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9641a3eb0da4b6ca165c2d51467bf8a : SUCCESS in 1h 55m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e69bc46af424796a51c5a80f9531558 : SUCCESS in 1h 34m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ec0271f3cd040909305ee236f907dcf : SUCCESS in 48m 05s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/c42770d2a2f943648886397905edf3ec : SUCCESS in 18m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/cafc18c5c85741b694f490af66c8b3a7 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"08cf2da493983a52681515ae9aa5da535ef8f5f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:08:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f97729c239c67b53e123b07afbe9443f54c2eefd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-12 12:10:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0a7b1ac1f734eb8a26f0074dfd7879d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ecd12db4f8a422c9b000decc23bad60 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"864acc23ab8df04335d4c5065966be1cd96308a0","revisions":{"864acc23ab8df04335d4c5065966be1cd96308a0":{"kind":"REWORK","_number":1,"created":"2026-03-05 14:57:10.000000000","uploader":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"ref":"refs/changes/16/979016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/16/979016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/979016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/979016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/16/979016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/16/979016/1"}}},"commit":{"parents":[{"commit":"968965cc77c6932eb67cc36a539f108c73d50ea0","subject":"Merge \"Migrate coverage configuration to pyproject\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/968965cc77c6932eb67cc36a539f108c73d50ea0"}]}],"author":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-03-05 14:42:00.000000000","tz":60},"committer":{"name":"René Ribaud","email":"rribaud@redhat.com","date":"2026-03-05 14:56:41.000000000","tz":60},"subject":"Add service version for Gazpacho","message":"Add service version for Gazpacho\n\nWe agreed by I2dd906f34118da02783bb7755e0d6c2a2b88eb5d  on the support\nenvelope.\nPre-RC1, we need to add a service version in the object.\nPost-RC1, depending on whether it\u0027s SLURP or not SLURP, we need to bump\nthe minimum version or not.\n\nThis patch only focuses on pre-RC1 stage.\nGiven Hibiscus will be skippable, we will need a post-RC1 patch for\nupdating the min that will bump to Gazpacho.\n\nSigned-off-by: René Ribaud \u003crene.ribaud@gmail.com\u003e\nChange-Id: I6668c15051a089fe4ce5808254a831ff3dacaa75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/864acc23ab8df04335d4c5065966be1cd96308a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/864acc23ab8df04335d4c5065966be1cd96308a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
