)]}'
{"id":"openstack%2Fnova~980676","triplet_id":"openstack%2Fnova~master~I334f80ed76ae869fe241d468c0bf5dd3d0634bbd","project":"openstack/nova","branch":"master","topic":"startup-time","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-08-27 14:51:50.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-27 21:33:57.000000000","reason":"Change was submitted"},"14228":{"account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"last_update":"2026-08-27 21:33:57.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-08-27 21:33:57.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-27 21:33:57.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-27 16:05:14.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I334f80ed76ae869fe241d468c0bf5dd3d0634bbd","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","status":"MERGED","created":"2026-03-16 10:37:08.000000000","updated":"2026-08-27 23:09:41.000000000","submitted":"2026-08-27 21:33:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":56,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"980676-startup-time","meta_rev_id":"a79ca8e5c899f554133e29f6c098097e7e2c8aed","_number":980676,"virtual_id_number":980676,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-27 21:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-08-27 16:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-08-27 16:05:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2026-08-27 16:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-16 10:57:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-16 11:17:30.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-16 13:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-02 16:27:17.000000000","updated_by":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"reviewer":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"state":"REVIEWER"},{"updated":"2026-05-27 18:44:49.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-15 21:14:14.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-05 17:06:16.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-08-12 10:27:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-25 17:52:05.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2026-08-26 16:34:26.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-27 16:03:17.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"a6d47304dbe985c4f4a49f2f238c02f0f8973125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-16 10:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b79cee7e89726a1408b368834aee684b5cc29d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 10:57:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894bbd785c0444cdb8e23f7629e528eb\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bed2faddf1824cd2a54157679f99bbfe : FAILURE in 15m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9ecd78375144846842701bfda840d7e : FAILURE in 14m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/17f7a02c0921436d8fd5b8bd3821f065 : FAILURE in 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d5dafa82eb3e3cf0bcd0c8124b167d008c8772d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-16 11:17:30.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":1},{"id":"1002f8c5193faae9c1ff6c046316fd95e91921b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 13:08:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8be8226e61494880804324be89f7dc5b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df6353edc2b44d85997958fb5cfea2f7 : SUCCESS in 50m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6f32935d4c494853bf51fb361b904067 : SUCCESS in 1h 23m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6db410651b0b43f38da6bc6038912299 : SUCCESS in 55m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e248c3d3184740ee9d049a89b54ccfbd : FAILURE in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1369c0a72284f44aa611c2bc1e39b67 : SUCCESS in 8m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e3aa1d4fe80437eb8aac930f7c7eb1f : FAILURE in 17m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0ff2874777eb452489c559fb7392dfa8 : FAILURE in 12m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/414f1eebe4594ecf9711f6361993a19e : FAILURE in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc35209853884ab2a9b898efefb42939 : SUCCESS in 12m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ff90749ade9b431992ef0a94071f427a : SUCCESS in 1h 39m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/acdbf2a86d5944ea9504807ed7074958 : SUCCESS in 58m 42s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d7e518b3553d45a28375431392d00ed4 : SUCCESS in 1h 08m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b62933ef57bc4abfb23414a2d9ea53b8 : SUCCESS in 52m 26s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/27375678f0ef4167962d3165733f8a2e : SUCCESS in 2h 13m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/e8659bff5e654d219aa46d0aa5037215 : SUCCESS in 2h 27m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5ef3496c30c64428abfc06962e7346e9 : SUCCESS in 53m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/529f74817dec426cbfed6114db826b7a : SUCCESS in 4m 25s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/585d3db0a1cf4959be78c224b622471c : SUCCESS in 27m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7c802183a446417d8b52aaa32f418b02 : SUCCESS in 29m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3377e37930fd4bc0888da04de154c837 : SUCCESS in 1h 30m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db2f55678a114e1fbf99fbe649812f16 : SUCCESS in 47m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4dcbbe9932be47ae93e82247cf8fcc83 : SUCCESS in 32m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ed9eda8e6dd64ba0a985d0bb7920f9dc : SUCCESS in 30m 14s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/b27b3b57510e4f13a551ebf2109cbd53 : FAILURE in 11m 05s","accounts_in_message":[],"_revision_number":1},{"id":"64fe5c74e19998be3c94876b908bc063d41b7de1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-23 17:04:28.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b127a73b6394c7e739e1012fd32efe6e8bfa3d67","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-23 17:49:02.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e16e141d761a3d37c0ee4002b5dc9c368b01b7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-04-02 16:27:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"09413f10f3a0e6416c645d753aab781accb8eaa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-04-02 16:28:34.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"aa23aab8c2025ecb7713aaf234a2cca0339d1250","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 17:05:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":2},{"id":"7a5b5355ccf7f2a1194e0eed718ac1599198f402","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 17:07:16.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":3},{"id":"8e5101362bfd2c01c79683eb937a315df099842a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 17:16:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20c86bf38a9944ec985143febe092e1e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a9ff0d2abf149f38bb9d22306db1929 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c90390da0e448648634d7e9aac09029 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ecd9a7dd1f794ff497283a5b47446f64 : SUCCESS in 21m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4952f82658e10d5504e7910e5b454b48bed8bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 18:40:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3aaad3193104a089276f5402a932d8a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d6abcd92ae624634b6b5429b2c690492 : SUCCESS in 1h 01m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7ec8929399144fdda31a833af3ead46c : SUCCESS in 2h 03m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3e047f4e3f948fd8d3564c88f24056e : SUCCESS in 1h 04m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c2bba174bbd4fcfb8ec8e63fbb63a30 : SUCCESS in 1h 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac288bfb7bf4234a9682aaf7b0a3d05 : FAILURE in 9m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/407e379ab6bb41828f1fc60e84145d2c : SUCCESS in 16m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1a78610b117a44799aafa57f922a35cb : SUCCESS in 10m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d746c0fed21d423580d4f5d48b478c7a : SUCCESS in 15m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458ff87a4ab34c239e9a73a85453cdc8 : SUCCESS in 13m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/65fcc7d03d4a409695362a7c39876297 : SUCCESS in 33m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/714c61f52aef45909ea0161f2fb17663 : SUCCESS in 1h 49m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fdb053e364bc4af28be1cc28ba4f4e4e : SUCCESS in 55m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5c147bcf434f49f9a2dfb91d9376f310 : SUCCESS in 40m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b9c2b51313144d8c9ef54ec22ee3fc91 : SUCCESS in 54m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/feffbff554104f6988ce4b4a6b5dbbce : SUCCESS in 2h 07m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/8406ffdca5714d54a892b02d0579dbe6 : SUCCESS in 1h 18m 27s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/20f649459de849109bc9b53f9699165e : SUCCESS in 34m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3fe7cc6604ef42f6af941d461ea00e43 : SUCCESS in 4m 55s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7654297add4b46429a95b1f2e1e0bf68 : SUCCESS in 31m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c555141e8af640b39514185437447edc : SUCCESS in 28m 26s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/46f1dd567cef47adbfa448e4d4c8bcf5 : SUCCESS in 1h 04m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2065fe957a3e4f43805409645f02f747 : SUCCESS in 30m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/73ca27cf8bcb4303a2398ca0c9685ab9 : SUCCESS in 14m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/451cdee739e14aa2a5ac769f44613735 : SUCCESS in 32m 17s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/de21f1c33c2a433fa4539791c9beb0a1 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"a99780eb61aeae0bda23ea6ac71460243454668e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-04-02 19:27:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"436a5c0a999da092324731c2c33e944e4966b670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 19:49:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32eaea6c6cbd40c7ab71e2dd0405a99b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c53d4d0efce476b8473a1d4ee72c382 : SUCCESS in 16m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb746f40191c482eac9dd1c0a4206be0 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93d4aa3323f1422680a92889c930ad0e : SUCCESS in 20m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"20954a328c80522a414eb11bd66287a4f1de944c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-02 20:06:13.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":4},{"id":"5f78e0cae4f98503211e558404bbb259b911ce6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 21:59:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d348f35c88ab4eedab5c44ef30733a60\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8c8622237134be48ecbfbd442289151 : SUCCESS in 1h 05m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/158f7ed7b01b4810be2c052c5810f643 : SUCCESS in 1h 36m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04fd5d3b9de141c4808268c39a490160 : SUCCESS in 1h 06m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfb67bb21815414c89b070c511e76e10 : SUCCESS in 59m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f768af67014506b61b8fbdf55158c1 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc559151f492485ebc8c3219b0d32767 : SUCCESS in 10m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/028aaa8f55ff4aa5b6ea73aff977623c : SUCCESS in 9m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/56f0cf14b31f4285aa428ebfc0a9946c : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3ae4f9d84c4cafb83cb366bddf1932 : SUCCESS in 5m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ce23c9946a994eb890747abd03214966 : SUCCESS in 45m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a22a1e3ef9854a8abc3ba91171662935 : SUCCESS in 1h 38m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/da3df84c2071477383c478527fcf1255 : SUCCESS in 57m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0fd47aa1332243088a418831ed655ead : SUCCESS in 52m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/56b584c53bb64dcc95a685531c912565 : SUCCESS in 51m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/02da42cf7cf84a9aa094ddfd707fcf9b : SUCCESS in 2h 15m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/e36c6760464649d997f205ebda74fb2a : FAILURE in 2h 29m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/47d8c03862b5406081210579327deee3 : SUCCESS in 56m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c32ef846f39b417ca99f43a9e7e3e049 : SUCCESS in 8m 12s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5a06dca2111433ea01ef5700c15f163 : SUCCESS in 28m 01s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c61c6262eb4345d9858bcae1584dad7d : SUCCESS in 23m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3e9901cb4a54154a9b7b13d805881ce : SUCCESS in 1h 31m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12b94e1ed7aa4228810cc689d7c9ba2c : SUCCESS in 33m 19s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c16bf0af04184df7aa32210b70d88cc3 : SUCCESS in 31m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1bd891619f42485594393e077a343179 : SUCCESS in 17m 27s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/11108dc328f04815a58e0b46138d0ccb : SUCCESS in 10m 58s","accounts_in_message":[],"_revision_number":4},{"id":"33f484f82864e868e6ccb9b46607885e1026b1bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"real_author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-13 17:09:43.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_14228\u003e\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}],"_revision_number":5},{"id":"ab8eb9c5fd6329103aea28a816253d68281873af","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-13 17:10:11.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"976c1c29aeb15f37a465dd624f5391cf28eb9b1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:38:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f930958ccae748b580e2186b144004ee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/22770ca48961437a960e83094caec77e : SUCCESS in 22m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27540ea5a92b4b5b82bac3e18eca66f5 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/980ac72e1a9d4f9aa5747e2e6f5474bb : SUCCESS in 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"596fe99b80d87934106132fe46e43440216197cd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-13 18:07:19.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":5},{"id":"a19f8c7fc93a7b066c3d2ab7e1655c29fdf3eca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 19:44:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c44c815775d415a8b362669db1fbf51\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8302def9c1149b4840b97db411c57ef : SUCCESS in 1h 11m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9649377ee06a4501bfd0349460cd2761 : SUCCESS in 2h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45a55f611c19486086a792f6f8c13c24 : SUCCESS in 1h 07m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce5791568d1c4b0f90717570b57026ab : SUCCESS in 57m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79d293b115b4be49c030abd88ceb9dd : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c436a3037f0540c9bad78acf4931ef8c : SUCCESS in 14m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d9ca9d6848f74fc4a9bbe05e94f3f129 : SUCCESS in 10m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2267e3a32c624e70914dbd72f1f9f4f8 : SUCCESS in 15m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba46eb9d86148cfb7de67c9d676a928 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c25e6a25babc4dc986cfe31a772b2f3c : SUCCESS in 11m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/526b9902ba5c458c90fd3b3943173bfb : SUCCESS in 1h 05m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/793c14108d4e44c7b652e0f04aafbe8f : SUCCESS in 2h 01m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/78ad5fecd886470b9c3b6a0a108ebe5b : SUCCESS in 32m 37s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/7f981272f84f4dd6acd948bd1c0dae5e : SUCCESS in 31m 33s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4dbde4b3cb4b4cb6ad94391426353833 : SUCCESS in 1h 02m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6661d7b276ca4e4d93eda009e2be7c0a : SUCCESS in 45m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f165a7fdf6f24dfb9a86cfb330cecb10 : SUCCESS in 1h 47m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/e3f26732409f4715bd7df1adc272cdd0 : FAILURE in 2h 13m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/53c227b023b8419ba7333f1bf82751bf : SUCCESS in 54m 36s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7e925ace592d41b49d816a03800f452e : SUCCESS in 3m 31s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/520d4cfaafa644bb91b370e6cca12c10 : SUCCESS in 28m 05s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4812dc614b0645d0980d25672c65caae : SUCCESS in 28m 02s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4b8ff905584b4def80000b22d7591732 : FAILURE in 42m 46s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ae7e4b190f44573989a2d0d3c233102 : SUCCESS in 53m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83e725d091d34cf68ff5b80e0c4e39d6 : SUCCESS in 34m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b8c840fc7a954311b260473d9092d705 : SUCCESS in 31m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0dbebc1e0dc04fc6b41940612f2912b2 : SUCCESS in 23m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f37652e117dd4e63a23373ddbcaf3793 : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":5},{"id":"df4a33cf71759a77c8c31b68f007f3d2e78adcd9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-05-27 16:01:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"de65433b3c0516985f07b1bce7cd75dd90aecb37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 18:42:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9006f02d996f438cb67caa766e616998\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/45be8cf40f7f4a9f936604f1726d16ff : SUCCESS in 47m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/680d56be870349de9f614b80a728431a : SUCCESS in 2h 09m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9b937e7bb774c7e8626b05132d85ce1 : SUCCESS in 1h 03m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cde86ed6266343e69fb57cdc4d75e571 : SUCCESS in 1h 02m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18bf52772a04e65b1f5babe2e39fe5f : SUCCESS in 9m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b7abc806420476ea3fe853b135ab320 : SUCCESS in 14m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dd42a30df544cda8ac7d279469b69b8 : SUCCESS in 15m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81e8955a0f264df6a4a719b6849b0e1d : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad7b704cd0e24163bb4a42804b26a3ef : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be68ed9738b24eea9c15d4e22afb4498 : SUCCESS in 16m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a438268a93184e6a85ef93dc57c53888 : SUCCESS in 1h 04m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b98b3e4efbba4a76a96aa7a4dade5384 : SUCCESS in 1h 49m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc7cadd30a114c50b6c155109f420fa5 : SUCCESS in 32m 06s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/b9cb477af8274744994af420266378bf : SUCCESS in 31m 56s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f0c70150b2794954b4a1996c163d9939 : SUCCESS in 1h 04m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5e0401b7ee3e47e1bfbf5aae31075034 : SUCCESS in 55m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f397ceff03484bb99b46007b3804e82f : SUCCESS in 1h 44m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/f47fa5a8080a4bed8e99a7d265081ef0 : SUCCESS in 2h 19m 21s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/43770f451df1495d8b477a300562d470 : SUCCESS in 57m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cbd02d4a40894a07ba12dd01715548bf : SUCCESS in 5m 31s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a5693d90da545539445cee9cf69a92e : SUCCESS in 28m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3fcb01a7da874d4697712b1035bd20a2 : SUCCESS in 25m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/761270a5fcfb4036803156b27a16a530 : FAILURE in 1h 15m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e4881685deb4f02864ddb18d967e1ae : SUCCESS in 1h 28m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb1fa2a2cf144664b2392ca44195e4a3 : SUCCESS in 51m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6fc6b0eb3cd84a1cb8fc582be6f90476 : SUCCESS in 40m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e2e5e116f5ed4b1f9f0aa4932f117b17 : SUCCESS in 31m 55s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fbef1ecb06c147d8be11aa3374559258 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":5},{"id":"0232fb0737c32555c4f1eaaf18c749e9e113bf7d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-27 18:55:19.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"0bf38e172980389b1a7de42773dbe80e08984759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-27 19:58:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ea27af4bd4d42689ce80d9caae8a60b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d79871ea0e4944049dd4447a57436de5 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/467d7488d2414d30a426fce9615f7296 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/07ad102b741846caaa131bc9ac10babc : SUCCESS in 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e68280f13d5a983347e694412b307b476624432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 09:32:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"7cd52585b074fd576dca7ba8339cc1e394942aa6","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 09:36:25.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9cfb0c7bc3f68e83b2e4e7f83f780dbbd499b315","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 09:43:22.000000000","message":"Topic clif/provider_tree_subtree_ironic_driver_perf_improvement removed","accounts_in_message":[],"_revision_number":6},{"id":"8338f900b8f802eb6a7a4bfadff35c1c95295e85","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 09:43:24.000000000","message":"Topic set to startup-time","accounts_in_message":[],"_revision_number":6},{"id":"1a74e0b5368bc7c82f0662c5edcbe6f2a4ede653","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-06 10:46:45.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":6},{"id":"132b0c771fd55d48d42f0f33a0fb83d55eacc77d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 12:10:09.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905f1001961e4c2d8ab4e7da9a935d5d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/563c64b4dd564ad5bfe000aa2371a555 : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/48493bacdec94b3e9098a2afca9d3aa9 : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/340adb32b0e244799f58c1b59269be75 : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"bec7ad662e1f4cb853b955907154788ac0388b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 15:52:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23318654c6c54b719866331136fe6828\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/da1d0ef42a0b448dbdebd1bc349de8fc : SUCCESS in 52m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8012917518a149aaa9bc88a7d8cc4459 : SUCCESS in 1h 05m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c79f814d380438282736b9daed72ed8 : SUCCESS in 41m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b6968e05f6341fe9b6f245ef923751e : SUCCESS in 1h 28m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e5e67fdc4b2412f92ff11e636262698 : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7af41a2d43434b5c9b5008d0fc97fcec : SUCCESS in 15m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81c0912938d74433989589fb317a0a04 : SUCCESS in 8m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/71df6a15029548f1ae0c6113e8f137cc : SUCCESS in 18m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec164f044cae4436a44903ad3a40358c : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf3022d9ae644f74a2648d58ee45cfc0 : SUCCESS in 4m 30s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3bb9d58b82804611880127036d46a825 : SUCCESS in 28m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6f5c626deec74d37a7e2c9b2daf09c8b : SUCCESS in 1h 43m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c775545c5834496a98c37cf101f46cb8 : SUCCESS in 31m 14s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f5147e30b8e04d3786cb34a2f72b7ec8 : SUCCESS in 41m 03s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6ea406772b834d8380159a8678cc66c9 : SUCCESS in 42m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8943dac40dee4453af6384c645f1ccd6 : SUCCESS in 58m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d3804a97f2b843fcb04885d0cbd27da6 : SUCCESS in 1h 35m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/1eb5d10b87ae4429b0583777912d25e5 : SUCCESS in 2h 24m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5893342029434f05948f0945c4e68fb9 : SUCCESS in 56m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/13ea0e9482a94308af008107adbc0745 : SUCCESS in 5m 46s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/92d3d9c9602c408a896d53b30f5bee41 : SUCCESS in 33m 25s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/73d529153d0f403b833c6ff9103c15d4 : SUCCESS in 26m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5f0dac05686f4c24842e106a74f4edac : TIMED_OUT in 2h 01m 29s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5285f6b7c09846bc9e6bb63be45532a9 : SUCCESS in 1h 14m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9a098902e7445839d2cbd9615b9a924 : SUCCESS in 38m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5a5665204ca44563aefdbc1c04c8efce : SUCCESS in 17m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6993ff3b7e2846928979536e14a4b9ed : SUCCESS in 18m 05s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d7996960ad434658866c9f38f090aa6e : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":6},{"id":"ae73efb87ab7c2380e3db6eb6f2844d9eec45cf9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-06 16:18:17.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"7cc9612ce319e780a6b4e6928f02fe8f1b3f46a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-13 20:07:42.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"53b5f3643a9384a2d787a03a83e7294b019b4199","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 20:58:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a1838f815a4f60a0dc33ad2daf7b11\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71928a3cb9ab4d26ac8260a5a9c69a79 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/476bd8d400264784ad9ae8486b5a70b0 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/920968fcdf4a43adba67843ff830e55b : SUCCESS in 37m 50s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e2aeac4a9e2269d7395994b6364212fa271e8c7e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 21:04:44.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":7},{"id":"2341f76d1ef17b5fc52b8e3f494deb67e4acae1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 21:50:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a22d784903784f538189f4df0c1b95fa\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6738d5e11f5948af9e9a49263118960a : SUCCESS in 1h 02m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b9938906db5c4f868958814720152c7a : SUCCESS in 45m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/606336bab9744fa38bb51afa2b123f66 : SUCCESS in 57m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa0978bb07a9411090d15f8c590321aa : SUCCESS in 1h 05m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/334913b71b0449a7a18c925c8d30626b : SUCCESS in 9m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/01ceb3791d8c4297babddc2afea5ddb4 : SUCCESS in 8m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6e560e2c5a36425bbfe12c09696e4b49 : SUCCESS in 13m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c4b8457d9804e139f172b13a3289131 : SUCCESS in 20m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8ae6324c9d84495afb8238a363d1569 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/036a0a4bec2b41f986fdd34382f749d0 : SUCCESS in 9m 07s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3c7dd9d648324fada45d5e4e0d97a858 : SUCCESS in 57m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0bee99ae924547c4bd3e3b6b388e0306 : SUCCESS in 50m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c64034d2e1ee42faa6b39211b29fec7c : SUCCESS in 54m 12s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f089a94d824a487cb47f0fd6b5ecd089 : SUCCESS in 35m 56s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3878cf37113c48c581a97cb12f4c515a : SUCCESS in 1h 00m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/25db3c865fd4483c943da1eb4ca92081 : SUCCESS in 50m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6494a310ba5e4de79a9589e6b283eb77 : SUCCESS in 1h 31m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/373b1056929743278ff5d9fb6f548da6 : SUCCESS in 59m 02s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/54531e5c7aed43b8a29f93d9d2ccd011 : POST_FAILURE in 32m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a2821b0258b243c180d74841fe00519b : SUCCESS in 5m 30s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f2560d6d09b4b5f9bee1f080718eb83 : SUCCESS in 30m 16s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f8be687bf89f40e48295e7c4acbf21a4 : SUCCESS in 30m 23s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ff280fd33ad54acdb3246bd1c6db5c77 : SUCCESS in 37m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/00493f90ad0f4208a8495b954b7cb7c2 : SUCCESS in 58m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71f3feffdcf84c8ba0e2762767999a85 : SUCCESS in 33m 58s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ff27c682bd9b4473b401953fb137972c : SUCCESS in 40m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c0fed0e00fc04eb6ae0f743c07c3a9d9 : SUCCESS in 19m 50s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9b764345e1ad4d9792452eb1522156b6 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":7},{"id":"1a60a6ba833da9007d35010d7afa11528565f980","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-15 21:14:14.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-ffkh5/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"0e46e1a590b37e2bbe047d388ed745f156894f31","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-20 21:53:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"077759ec22c1f4ec5d6485b96bed745338d77532","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 22:20:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9672ea3c588b431a944bfc0b73dcab6a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/982c8f2a172946aeb8bafa3defedf50e : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4c7d064efaa945e8864cdc280f691cf6 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/27a8f6b192ec4020a1374cfe1d1c6569 : SUCCESS in 24m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6ff08bcf3aa9a1083a253146d6fc0a88e65265f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 23:41:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5894e34cc3984f85870e57a0913bb7e3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0c423cea1f8e4a6ab5ae6a657929932b : SUCCESS in 35m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4a7ecca44dd34c159009f64c568fa436 : SUCCESS in 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0d52e63fe6b4215a5de336aa18de4ac : SUCCESS in 44m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f4e06b4cb474a858b300eddf247fd01 : SUCCESS in 52m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9855bc50c78947edafb97563f8335546 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1993f1492f44ca28943aead2e1f4133 : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c885a542dd2f4635be5bec0591b976e8 : SUCCESS in 9m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/96b47ef33d9743b98c0dcfdf157e7139 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3b2545c14340838674fb59bfb7c040 : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/085f4cf4a66543cda8e41154856cdf68 : SUCCESS in 4m 50s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/983fe4ecdc354f0484b6fa364bfd8832 : SUCCESS in 46m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f14298299ca54b3d8de19a8711837e94 : SUCCESS in 1h 42m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/77cb2bbe37324411a272c8ceafdaec92 : SUCCESS in 27m 00s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/03658096d362431a8fcfe5cf520a4778 : SUCCESS in 34m 19s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5b9a997effc44c63b531639f5a1a8dad : SUCCESS in 36m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/094ca4def7dd4d5ea798e746493e6c82 : SUCCESS in 47m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6a147012712d4f8589569c0ce3260dcf : SUCCESS in 53m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/89f9c984ed9c477db7b4d821bbce7cfc : SUCCESS in 1h 21m 32s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f9d0550d0e47424c9ec300f4486a65a1 : SUCCESS in 28m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f8e480b3f04d470aa6cef9c772401432 : SUCCESS in 3m 01s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9940756d2c274d08845faa4ecff2a82b : SUCCESS in 24m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/28eb7b25ec024ec88888279b77820542 : SUCCESS in 25m 59s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/24a36ba0e3194e81aa7f78a58ed3fd22 : SUCCESS in 33m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9921b3da2a8143cc92a1bb5a8c6033e3 : SUCCESS in 55m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c8221b5b8e64cf7a698650e0d2b8abe : SUCCESS in 1h 00m 27s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8bcb0a39412a44bdad266e4f5d837faf : SUCCESS in 49m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f12d83d7a28b45e1b2fee8c6984fe144 : SUCCESS in 35m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a4a7ede3239c45d28a9e9b8cfbd8bf37 : SUCCESS in 10m 22s","accounts_in_message":[],"_revision_number":7},{"id":"4311849081e683c416605c7a9ae9bacaaac21ad4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-20 23:50:54.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"cc1567dd85b2951e1f701fd6653ea8454651e89a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-05 15:03:56.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"e4a5b73ce9177c598f4f39717c24207b439fce28","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 15:06:53.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":8},{"id":"68b022aa91f197401bccc4d1ad80fe3be954ba78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:28:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1303f0db5c324507bf59935feb34b027\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e1cedb5de254ded80652f5543cf1d02 : SUCCESS in 11m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aac3a7df4d8a449293c7d2eaee468575 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/df484182c7e54c15ae3a4f86de4dbb95 : SUCCESS in 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dd540bd5c807651d2124392858f9d8104f9ec538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 17:01:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea219997846947b49c5a7380295370ed\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7439ac7d7634c7f9c6412f74eab5369 : SUCCESS in 1h 02m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/96fe90b2091a4e84bc8f0f6cf9a56a3c : SUCCESS in 1h 46m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf08d3b798684c0e86700c55e1738f00 : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/071963fd16a94ccbace8797be34d8061 : SUCCESS in 1h 02m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef571be75e124c8c96805bded5df74d5 : SUCCESS in 8m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1237ed0824ed481cbc11d0ce54352af1 : SUCCESS in 14m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02c4fb24565b4d0e812e5c70e6663dd7 : SUCCESS in 13m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2ba9bf1d0bd402bbd5bfec46f0c3f8d : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0b9498c8bc64f7eb5be2bc0d012b170 : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9354545cbd0448b0b05c4ee1f8a1d710 : SUCCESS in 10m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a2d8857a408c4cb5a8fc70ed51c7332c : SUCCESS in 1h 06m 51s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3169adbcd234b05a2c3715ffb1b2f50 : SUCCESS in 1h 26m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/11d33f6a9f4a4b59854ce360f0a7b763 : SUCCESS in 52m 20s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/c95af48905b6466d96d7a9d6da588eaf : SUCCESS in 36m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/eebb9bc2193e429aad896e3d6431ce3b : SUCCESS in 37m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9c0b730911174323bc9ecbcb8772ef32 : SUCCESS in 51m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/aa89d0eabf2943aa832379844e223f82 : SUCCESS in 1h 28m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/1d1ae74dad8544c7b7266bdffbaf476d : SUCCESS in 1h 40m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/771f201dfb44480fa5922d6bc37bddb8 : SUCCESS in 56m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cdbec0dde4cb4807841ec5c93ba09827 : SUCCESS in 4m 45s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e4b8939e9540499aa6edeefce48255c6 : SUCCESS in 23m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a3786cdc3a634fbe951aa03350c22837 : SUCCESS in 25m 26s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/de111c2f5e964ac0a168134d8f0bb7b4 : SUCCESS in 27m 26s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/815934cb1f1e4d38bab8022944c85a72 : SUCCESS in 1h 30m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a3cfcdc1d0c40fcb679d787a7139d3e : SUCCESS in 56m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5a8de4628c2442d9923cd415d52d9ea1 : SUCCESS in 25m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fe17acba66ca4296a7ae8bf94e381c22 : SUCCESS in 15m 33s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5f207c6096b14ad4b6c06a9c0e1d450d : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":8},{"id":"1dda87634aeeceb064d3d0a02e6c9179a41bcd64","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-05 17:06:16.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5e2c137b7393aa4f2b145dd3015669d51259be8a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-05 17:12:09.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"c51c419e9d6848b019ef96499a4ec010f35c75bb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-05 17:34:04.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"407b04dfcbfc23fe20e52cffdfb2152565b0d96c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-05 17:42:57.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a59a4d2fbcc03da9e6f854fd9244856ce4f77b26","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-06 01:06:27.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-6zrf5/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"cebd1f7dfc90dbb757e300c76e52355485311319","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-12 10:27:34.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"107e8519dde50ec0580a6d48a711d8601b7b0181","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-12 10:28:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"95f1288e6a4b66920efc0e308d592fe90b9038ed","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-12 12:31:50.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"70cb1fd6d7708cb41563054a6bd6f245ff2cf94d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-12 13:01:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9b05762a1a071313d759e206abaccefda3f5aad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-17 20:01:42.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c58db0be7fcc49c00f7ba28d88a7e50aa1db3d0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:35:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3e6c508ba34355b55dc24399a3c61e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a14fdafe2ce84a7faaa6d4acd24c8432 : NODE_FAILURE Node(set) request a1b13a71eedd43d69c8345dcdc477408 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6bc46f04b99f49908d1f48e951392756 : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/76a450b23b1941fb984ba1a4e658037e : SUCCESS in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ff6d28d5d67c3ab85c055a4e49b1f1308cbbd303","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 20:49:02.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":9},{"id":"6f51ba80bd0572692a263b18f8f537daa399a69d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-17 21:06:45.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-hmg8z/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"f7778d5b854c122a9dadf99e41e02e09510ed8be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 21:22:15.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f74f737dd43249d9989c8cb536cf2fe4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9023999271764c90ae12b1af682e51ec : FAILURE in 1h 04m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c373c02c050746aa8c188e2a5d86310b : FAILURE in 54m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a40492732ca4f11878dc368e2dcfe18 : FAILURE in 57m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/463528a7a16f4eb9ac97064f26a5d62b : FAILURE in 1h 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b628fc600c0340fdad6d1ae968e1c230 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a86f6fc39ca4a1790d31fc7aa3c5e8e : SUCCESS in 13m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/91f2923bf0b54be3bcfa4ce004d46522 : SUCCESS in 14m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4cb5f1939dc940ffa691509c2de1d148 : SUCCESS in 24m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ba09d97c0646c4a5dfaa5856ad3a82 : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8692e6e7ffbd4041a2b0904bc6702e0d : SUCCESS in 8m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/619d0962ed60444ab83ce925947bdee2 : FAILURE in 35m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5581684b441944278dbe1c483ece20d0 : FAILURE in 49m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/93f01bbd506c424f9911f39cc25f6de2 : POST_FAILURE in 41m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5ff4624ee4fe4736a21ffd04a81e08dd : FAILURE in 30m 02s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a20b5932952c46a9b2ac3c9da4a6b40c : FAILURE in 36m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5a8c1e53674947fbb1aa690261f2a39e : POST_FAILURE in 47m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/56395d9470094daeb3234b6e6de04afe : FAILURE in 44m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/bbdcf256212d4b78abc0a28721d764cf : POST_FAILURE in 45m 14s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9aa6995478cf410fb00055795230f631 : POST_FAILURE in 16m 08s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d67d869ef4784fe38799406f2e30e66d : SUCCESS in 4m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5c3643727ca44088af85b2d4f7594a10 : FAILURE in 34m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d06f0588af60445cb5b6f4a0889a0b6e : FAILURE in 37m 23s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/37b7f4eb3a624e21b0a7d12655929687 : FAILURE in 39m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/74ea8a17cede4271a60504cedd97d8ec : FAILURE in 1h 12m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b32db7c97764b7cb6917a323e8f1c9b : FAILURE in 34m 47s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fc807d7fc4a846abaf988ef4dc21bfdd : FAILURE in 26m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/532dd9eb61b3464b99f01340d851549c : FAILURE in 25m 24s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/eeb9558e3f9c48b686695638e3b50443 : SUCCESS in 14m 16s","accounts_in_message":[],"_revision_number":9},{"id":"481c740e52ddb90da3f497ee0a4f203ae9918797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-18 17:21:36.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"05a38427feb8463c213ac648eb58e25498e98a23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 17:51:31.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6693bb7672e424d9f525c8098182318\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c386ad1091504ee284cd6d601f909b96 : FAILURE in 5m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4dfdd40085b94d4c92ba1311e3499df7 : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/274ca71e9e40425983bac67e6dc03445 : FAILURE in 12m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7aefc125423ad435b192f5daed3020cf3c6531a6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 18:18:28.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":10},{"id":"ad54284d05b0e283d9d087f758e3241e28c7fce1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 18:45:55.000000000","message":"Patch Set 10:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-fmnt2/index.html\u003e","accounts_in_message":[],"_revision_number":10},{"id":"3bd578aad25f07bebe339adf28b2c6f4518b69b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 18:53:10.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc243be312145e3b58fef2d8cb3860d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cd43c7160a3949a1a8ed19511bc4f21b : SUCCESS in 34m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/91ec9c0bc6c74ef3a5cc23df1fec7260 : SUCCESS in 53m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a6b0a9253904526857530c034831f74 : SUCCESS in 1h 01m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bca2d2743de4a26bb29cc6ef88b43ef : SUCCESS in 50m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b32f8c1fc48c460db5b6513665860a0b : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f68065f6c7b44a4aa0adb0c384438b5a : SUCCESS in 9m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e4148b70a4d0401a8c612b44a08669ea : SUCCESS in 13m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0856dcb792444e8ba3d63c435f489065 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bc50e955e946a9b746d7aa7b283ebf : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b2e8305e88f4412ad2714cd505791f4 : SUCCESS in 5m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/75ff0b6999c848449ebf2fe5fcdce8a9 : SUCCESS in 29m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/84982d12993a48429df36d5d6e4ad9f3 : SUCCESS in 1h 12m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b46776be554441c79b1a44a418927888 : SUCCESS in 26m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0f9c26374a6a41eebbf69a322e684313 : SUCCESS in 33m 15s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/eeaed49ef7aa408e831a1d011c49ae91 : SUCCESS in 1h 00m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/aadd6b8fab6648109b66a6e252b62099 : SUCCESS in 50m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fcf3ea7a868e41faafdc7ee0bffec8d1 : SUCCESS in 53m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/a78068f868fd4863b31b23b075e0e49e : SUCCESS in 1h 18m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b24077e5f2664fd18f9c9523d4516ca9 : SUCCESS in 30m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c9d2516cc9c044c595b49434199aac9b : SUCCESS in 3m 28s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cb78d394e63e4fafa5ee7d3e7a132759 : SUCCESS in 27m 40s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5180c1a0e9124a908ba6acf984180888 : SUCCESS in 26m 30s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5b49fc276f724522a52db0f9b6fc8bc8 : SUCCESS in 31m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d2a7c5a0feb4b4eb0895bfccb91456b : SUCCESS in 1h 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/374f3b089a604dcf8170101638abaef6 : SUCCESS in 29m 04s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/31462b2df02d469ca6e4689ff23c6dea : SUCCESS in 55m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a33361d5a12a4dda8f3960147300c641 : SUCCESS in 19m 22s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/349ff6d6c9204d8197f874d1eb955d40 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":10},{"id":"8643dfe742fe754e44fc036763907e521f16a6d7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 19:11:26.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"7e509f0d905bc843d6700a9fa95be8918580f994","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 19:53:42.000000000","message":"Patch Set 10: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e225d1a6e31724ce27657c0e7536eedcf54aaabd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-21 13:51:30.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"a5df6e446f3b7ff0b99620c57d51bf74550bb2d5","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-21 13:52:31.000000000","message":"Patch Set 11:\n\n(9 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2543362cd7086c6651c070822d2ab23db1577e9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:11:12.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47107a7475dc451b8b960e8034b83777\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e27ddbe3f72449c0b98aa0b93536b01c : SUCCESS in 17m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/301d3b6ddc4f4867874411b44b4fb0fd : SUCCESS in 24m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a4fa1467d3fa402183f14fc434653476 : SUCCESS in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"8be134a11afdaf9ba02d5de4aa82f5b66bde6aca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 15:42:05.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":11},{"id":"0ff4283c3d54968b4900767f6508f7834e3e9347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 16:06:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f0beafe0abd40a78d480166291a82cc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95d33c8daf144b77bdc0c5f27d87ba3c : SUCCESS in 1h 07m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/401b1b8694364fd7a1ca30cb6edc1d52 : SUCCESS in 50m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e869d4f0af44c84b58accc8cf730711 : SUCCESS in 31m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0ef20ced3304bf49dde3f6048ec55dd : SUCCESS in 52m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a22d939e25a4cbfb4324983972c7e6b : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79a7f0cdf35e468bb661007d28cb50d7 : SUCCESS in 13m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7bf68adf71614f36a924f6abb557a62d : SUCCESS in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a830c2ee08543bc9229532a788b5101 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe407822f5334923a965bf674484804d : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cb5a2207df746838db78fff6d606610 : SUCCESS in 9m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5b79279f5d7a400ea2e501cb4d6d3e2c : SUCCESS in 58m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7acfc00e9e9f45ce88a6c804a926b5f4 : SUCCESS in 1h 26m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e810449c6f0a460596f92b778b99fc80 : SUCCESS in 27m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/008d094a4b77476c83b2026b72a50ef1 : SUCCESS in 34m 55s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d6b9eb980c4a4f92aa678fd2164539cc : SUCCESS in 1h 00m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7dcb08bb746f4832a3450e16eceaffab : SUCCESS in 54m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6363fd8b75a343cc94bd6aa476eb3ca2 : SUCCESS in 2h 12m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/8acdf606cdf04ed3bfde3906ebf77673 : SUCCESS in 1h 59m 40s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/109d73e793b04d52944e6145e4db05a7 : SUCCESS in 29m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d5521d3ee90b484ca40ddc9536485d72 : SUCCESS in 5m 18s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e39d46c4c4a145a6bd76fb1a389cd3db : SUCCESS in 31m 51s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/42446935946c49c88fca4db16350259e : SUCCESS in 28m 49s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9830b343c9c6455883e04ac364595ed3 : SUCCESS in 33m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7029ac8d3f574411a2a840e93c97d8a8 : SUCCESS in 1h 34m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58168142cde3426a8da576f63a9624ce : SUCCESS in 54m 55s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/611a46d2bcf9438a9a7b3ca96255e06b : SUCCESS in 29m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7a7e7e27e2784ba6bcdfb6df3c58584c : SUCCESS in 34m 48s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c3efc09415bd45fd97eb10dd455b21cf : SUCCESS in 12m 52s","accounts_in_message":[],"_revision_number":11},{"id":"3dd0a100ba8aaf58827ff9cd0775ab78a90d8724","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-21 16:17:38.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"faf62bff4a3748d121705a460926a7e294bbdd39","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 16:49:30.000000000","message":"Patch Set 11:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-f2rzq/index.html\u003e","accounts_in_message":[],"_revision_number":11},{"id":"6fb110087f6e0e8fc9b2ff9c2247ebb0f75c7e47","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-22 20:59:26.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"270d95ef7bd747a67622722a2536b8ecedb5cbf6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 13:39:23.000000000","message":"Patch Set 11: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"618e386a9d770ff9cb7fb40352f01b78835ba48a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 13:44:16.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"83d834e7309ed8ad773e46c48bcce85faf64a496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-24 20:30:26.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"6b010f22072a23d7e77a255eefd5772d04f294dd","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 21:11:12.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-4f2pq/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"26583a63ad1dee2d9edc7999655b6cb1652f8a15","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 21:27:04.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":12},{"id":"05d54c84deba2da7a6d3631fd042618e2ca978a5","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-24 21:59:49.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"04765b6d7ec44b71b3ff7839288bda8678ef29ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 22:52:28.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ddec680b8824bd88ab4b51a2f55c38a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7c836b6aa4741438edd0f686b8e80ac : SUCCESS in 43m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/28ae09a22be8430987dd6c8c876214a9 : SUCCESS in 1h 32m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1ab253ff15b4527a46bb630adc49bdd : SUCCESS in 56m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca6c46d0a3b14072aaaa7a9ce143e7b9 : SUCCESS in 1h 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950597a91f444725b4faa8bf2c552395 : SUCCESS in 9m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eddcf8619420438fb5865e81ac53ced9 : SUCCESS in 13m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/747006b9046c4ccf8104616340ee68f5 : SUCCESS in 14m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb6daee9da2b457bbdd985a5e4aadc03 : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/116393db467946b49753b405e1735cc0 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddc8fd5ce2ad464f9d48568036d18291 : SUCCESS in 8m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b9765f7b69eb44fa99f1b49cfc922b0b : SUCCESS in 33m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c527eda145b8436384804017a35ca579 : SUCCESS in 1h 27m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a91f5b0dd00b42b593f232a88a7f3bb9 : SUCCESS in 52m 32s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/74da96f3791043c9b07822f1a7e07980 : SUCCESS in 37m 26s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d773ff0a01d54df482168f34f559a9fa : SUCCESS in 52m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0b97c6e51ca84ddfa0b231dd643bfb2b : SUCCESS in 53m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9cc649edd1694398b3f0cb6f60597772 : SUCCESS in 2h 05m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/0b48d24cb63848cc941317244da10871 : SUCCESS in 2h 11m 55s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7332c921d2654fbfae214794f02f1406 : SUCCESS in 36m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/48f59ed4cac24ebeb1934f84962b2c8f : SUCCESS in 3m 10s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c26bc877053e4d25a707b960c6916b9b : SUCCESS in 27m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0c1b9ccacc384290aaf8c428d91eafe2 : SUCCESS in 28m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ab16b7ba42944e8ab8d92d391e918eb6 : SUCCESS in 31m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/514abba3b12c40909629b301e8153fc7 : SUCCESS in 52m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a54cd8b02ae4452ab296aff415c1023 : SUCCESS in 34m 22s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/881642df02464568ad46043f273d1c10 : SUCCESS in 25m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d573a9e358a54318a41807defcd2eebe : SUCCESS in 34m 40s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d98d7cecf53d4a30bd679f1a1d00af2f : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":12},{"id":"bfb3dd3537f7f0c6cd73b842cfe97a5112981493","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-24 23:02:49.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"1bfb38f77732b3605e18ddae6200358bb5ac731f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:49:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49758e2dcd6b4e4b98d0a975e59e94d7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ffb396bc236a413f8e8a047dd265a456 : SUCCESS in 13m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18f8ac0260674e3580238d17f5fbe67d : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d73feac5ff94c0d86e0b2595491c366 : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8d159b80900ea2b64c4c5e62f6ea53e98cdf074f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-25 03:13:23.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"cf175805c4e0b6081af2d768d4fb3201739ce574","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 04:11:53.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":13},{"id":"7a8cadef3cd4999d90b2c2ccae290cc573a7ce7d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-25 04:30:01.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-b9x2c/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"2d3bc90e5ede9a0b2dd903a3db4ec1865d41b0a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 05:19:23.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/457490e37f33410592d02a1c3ad4cb5a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9dd3f013106d4ccf98018841c32b903a : SUCCESS in 59m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9c89d73ed8c54ca8a2505d056e7c38d4 : SUCCESS in 1h 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34c46af41776453e87e07f692435c506 : SUCCESS in 1h 04m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7782b2f0d52f49a1a6a114fdd94e3be4 : SUCCESS in 1h 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1e53ce10594d588aaf78e53f9dde9e : SUCCESS in 7m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d238937b36b465da5e696435615992a : SUCCESS in 14m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8075122a28ce4dae8e9d1919c200b1ea : SUCCESS in 12m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d0e3c7f878994c8ab7cd0778fa0ad630 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb0ae72df1e4fc19920fdc00ab02e73 : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61a7ebeea2ab4ef3a6bf54ea563a70eb : SUCCESS in 9m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/583b3ddee00c4325887d6e61bad79ce6 : SUCCESS in 56m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f3a2b06978be44159210fb485b084974 : SUCCESS in 1h 06m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/24520039a2bc47e78f61742edb519054 : SUCCESS in 48m 17s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/594643fcf46e47e1a82d752533203155 : SUCCESS in 35m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/07fe7daf1c724c0089e8ce6af2d2d8c3 : SUCCESS in 59m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/123789db4f6c419188f93edc17cfef11 : SUCCESS in 34m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3eb046fa98004523b91d944c914c0d37 : SUCCESS in 55m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/c7fbb176a3a64612a24a353f887f10b2 : SUCCESS in 2h 02m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2e60633bf5da4cbba647e876799482c0 : SUCCESS in 46m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/31a7000bcc874e578ab4512826075663 : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3699130ed37c4f17906f464c616476b4 : SUCCESS in 27m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3bcbd565d57d48be9ad6c3b9895ec30e : SUCCESS in 26m 44s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/75e1210c5a674fe9ac89f5b80b4134ad : SUCCESS in 29m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/903906fc4a244b62aa129a97357994e4 : SUCCESS in 52m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9051b3cfe845d59d107c015168e54f : SUCCESS in 54m 16s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/006e0c62c27e46749895fc25ec14cffb : SUCCESS in 53m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d6e2d9b0e4c447c5922cf03ad49f0234 : SUCCESS in 34m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5d05d145baba4bcc98467388604f20f1 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":13},{"id":"a04f22fc2775730898e4afe1a28903f360088bdc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-25 05:29:54.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"66f28299c54e60e2108a98e7bee7645544729984","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 05:30:57.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2474b90c88c948588e256c6a98258c5d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b85bd73dc9f4d8ba72bfdceed214b91 : NODE_FAILURE Node(set) request ffac18bad30f4d3fb88d9a2045083dd6 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/51bcf8b057cb4e698a71096ffea96ac9 : SUCCESS in 16m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/570106c7b3494ec8aea456cfbea4cefe : SUCCESS in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4677da3bed2834fee9cd96a5c95704583403b281","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-25 17:52:05.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"28e4f76106d61ffcf1d03d293e9df0a2ef7bb1ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 16:24:32.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"7be8a158c746b7b49633427ae07fabf473ca8d94","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-26 16:28:19.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"880f3ed4906437f0bac650d58b1599adc58be8b0","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 16:34:26.000000000","message":"Patch Set 14: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/568bcb1eeccd47c9b07033793f434a78\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/4ae9baf67170439ab477fe4b03154810 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"2a9c884fced6130aff880bd675d0af76b6cc64fc","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 17:12:47.000000000","message":"Patch Set 14:\n\n(5 comments)","accounts_in_message":[],"_revision_number":14},{"id":"cc8f72b33fa0720065086a2b1142c9f0c193bfa9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 17:19:38.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":14},{"id":"f7f48f25017df8b9fba5687603c7cbe138c7c2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:53:49.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb7f1cc4b0cc4f1eab36fabb81e6e78d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4466ac109564b5a89b3d887a1612866 : SUCCESS in 1h 07m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3f9058e797114ed3a339b8c8abed1f25 : FAILURE in 1h 34m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d360bc0128e4e0baa882980bf3cd272 : SUCCESS in 1h 05m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81ade4b2fb644e31aba5b7285c8cfb8d : SUCCESS in 1h 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db060bd71cf441028cbf6ce889951a86 : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac05e1950f9143fbab74727e29391338 : SUCCESS in 14m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9ade5b94ee754a89b5e67b0321b25145 : SUCCESS in 14m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50b701d62fb445a0b3c538af8ac99b69 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/989ef342c29744a4a480275850f58be6 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76a9d21c5c1f468c9c25f893c615e05b : SUCCESS in 8m 22s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7be094d13c0443909077b49fe5f04b47 : SUCCESS in 1h 15m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e5410432745b40808f7a87a2ec64d707 : SUCCESS in 1h 47m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8d9f007940fc45ce96fff7f499048797 : SUCCESS in 52m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/85e8a1190a074b129ea6668ef9fe2460 : SUCCESS in 41m 27s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d5354d1cbe494ebe81889806a1927cc3 : SUCCESS in 1h 01m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7d73e1cbf59140a08b6ff125f5e4d103 : SUCCESS in 54m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/375bc853d5454ee6a7223ef1c2c59bc3 : SUCCESS in 1h 39m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/bbcd15ee58314f009b9fd270a2608cb7 : SUCCESS in 2h 21m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d3e5760c2b454744b1fe2693d0273749 : SUCCESS in 27m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e2fefce385104b7cb06ecb630f4fbba7 : SUCCESS in 4m 46s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1b4e38e109a340659ba898012b9559da : SUCCESS in 30m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/639c0710155042c0807fc142d50a7bed : SUCCESS in 28m 45s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c302c22254a24fb395b24dd6634d95a4 : FAILURE in 30m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e8ed2eb4c464977ab0b954f1539735f : SUCCESS in 1h 22m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9aa04fe4be8140e1a9f035796ff044cc : SUCCESS in 38m 48s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9f4110197d794fbca530951b590d502a : SUCCESS in 28m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8000a4837c3a43548f995d671e7fcbf6 : SUCCESS in 22m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e0103da406bb4b79900f78e2e85ada7c : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":14},{"id":"4d5e274a967336807345777748c9c69fa0428fde","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 19:24:46.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f930313e9f148f0933811f12f6b349902d1eab44","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 19:32:35.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/f81a4e1d40aa4341bde08a44cb0392e2\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9788617d78474c22a178a1c902de490c : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"9d7557432b4dadebf751c09fc1e35f286ca94e98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 21:30:57.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de6f42ea01a940059e4cc07107481916\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae016c649f29477c88b3b87202a2504d : SUCCESS in 22m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f93e2dd05fcd4101886eca541fadbfce : SUCCESS in 17m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/58bbc1902d8847b2ad81db7201e3ae88 : SUCCESS in 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"310b47f8c4c869e92d19ee61e05448a165a8894a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 21:34:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11b9b7cd9274e8e91d9561abce713ad\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eccb3f072d5a42c8843eb163822d42db : SUCCESS in 55m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/edf5fa2defbc4ed6a527cf58b7a91bf9 : SUCCESS in 1h 20m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aefd77cf4bd6485fbea58ecee59a04b6 : SUCCESS in 49m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d31c477bac34f4e89883b0e832bd256 : SUCCESS in 56m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9575707a5b6c45ba832abd8a57f61e77 : SUCCESS in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/795b51dcf569428faf2d84619429b73b : SUCCESS in 16m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78bad2bf6d174926ae217c5451cedcf1 : SUCCESS in 13m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f4160cb9d9244a68bd6baabe3960591b : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b6b8f97b6134aec9f253585f16eca64 : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e54c01d671b3486db38c84293e0e9921 : SUCCESS in 8m 30s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f82a59aa58a147dd91da5eee5c349f69 : SUCCESS in 1h 04m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b1f7e0e869544102a9e1aa61e772cfbd : SUCCESS in 1h 26m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/abe910b66bd2461f9777e605828bb98b : SUCCESS in 53m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fcc319678f3d464789e8375d613157f7 : SUCCESS in 36m 32s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/421bd62409714fdfae01126d8dd9df12 : SUCCESS in 36m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/36c697bca26f4f8b8e60372276708721 : SUCCESS in 51m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5b2f4b19193d4a7f95795b8b3d555784 : SUCCESS in 1h 39m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/8306c33a9181403aace94faabab31139 : SUCCESS in 1h 59m 33s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ce076041fb3a48e38cb9ad0f6767be14 : SUCCESS in 28m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ca6f25b922904b568295fd878bcffa4f : SUCCESS in 6m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/882587a095de421096776be379bc404e : SUCCESS in 27m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5d8f9555d7ec4767b47013bb671da6d3 : SUCCESS in 20m 53s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d32a9d31f9ba43f185ce564ac9734a85 : SUCCESS in 29m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef29611f14f04650acc741165874640b : SUCCESS in 51m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2057482c7e46453cbee59f772ae38418 : SUCCESS in 26m 49s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fd1d27d65ea341e2b83a0432af5ba25a : SUCCESS in 54m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d0559cb956924b69b0d7981de0cc5183 : SUCCESS in 32m 43s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ac85348231d846278463f3506c29267c : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":14},{"id":"74784217f98368a84230507c44ab258a5c6eed38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 21:44:53.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"5c4e093e5894c2151ed7f356f5995c9d97523bfc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 14:26:36.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"8f0ef24d6ec788737e389cf0c19eb7798574a4f1","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-27 14:51:50.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e26a626c634cbefb69cdc34215075f5bd4f08de4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-27 15:58:38.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"715eacf612f89dcca6af7b10d49b7bc897145a07","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-27 16:03:17.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"70dfd5c2aa5ce8fc2705e83565b2f2af971aee1c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 16:05:14.000000000","message":"Patch Set 15: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/d02cc8c318304ffe933c884a779ba905\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/e543a85b0b4a420fa769af2adde797f6 : FAILURE in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f678ee97ce79618192e262c1db348be0aa29c645","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-27 16:52:04.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I334f80ed76ae869fe241d468c0bf5dd3d0634bbd .","accounts_in_message":[],"_revision_number":15},{"id":"0bd9479a22e0c96dbd1cc16b85bf36cb35c55b94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 18:11:10.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e8adba542ab469e9a433ec1e55c8435\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/99f0d64eb80347f0984705e4bc4d2548 : SUCCESS in 31m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4adf3608881241cbb4d531233eaa2c67 : SUCCESS in 46m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50286869701149b6ab3594a61ab66f73 : SUCCESS in 30m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3826bd7f5115433fae4b0aaaa6a12e43 : SUCCESS in 58m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288e4894816f4ceb95330bfd58e542a4 : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/276ac802f62a4242999a5ae9f40b10f2 : SUCCESS in 9m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35789604bdd34ea38ad2fb0d53faa315 : SUCCESS in 9m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/433c7cdeb74e423196086086dc7cbcbd : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edc65599dfd45b294a63a31dd64d10f : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2487ef150b074f3fb528834ab9a0615c : SUCCESS in 5m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1949fa02619646f7a2ec0117768b725b : SUCCESS in 28m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b79b7a2ced0e4dd3ac0398d05ff5a0d0 : SUCCESS in 43m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bd33781271af4422b01b876b8a806c37 : SUCCESS in 46m 27s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8a9184a2cb3040bea2b0e35127d8c792 : SUCCESS in 36m 52s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/cd13c67b6bda4b3990c521ccffd689f3 : SUCCESS in 39m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dbfda4e709ba4c11a7fcb75da395803b : SUCCESS in 53m 54s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d7a928fb722543e288b982dabe11a552 : SUCCESS in 53m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/d9e2f9c81a6343409d5ba567623e6a9f : SUCCESS in 2h 06m 13s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/fd0ffd5340f2494dad23ceafba520f89 : SUCCESS in 42m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/90e4123f43eb4055b871ee1ab4fec3cf : SUCCESS in 4m 57s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f8d37033f2c446ea16a6fe7744d84d9 : SUCCESS in 28m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2c8ca126104b4686a314bd5067723a8e : SUCCESS in 27m 41s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8bab74772e0b41a589bb2f21261fd178 : SUCCESS in 31m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d2a962ac4dc409d9c00a94da8c20abd : SUCCESS in 1h 32m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99bf4ff9b6d0432db85de2a5ad7cacc6 : SUCCESS in 54m 23s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5c5dd7ff24e14362898f46261defc59d : SUCCESS in 55m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b5884ef183e44e948fcce3a64035c52d : SUCCESS in 32m 14s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/45638694a78e4c35b3a7b201cb380c15 : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":15},{"id":"c3525d8e5f3c6180cdcafb3a92b9fce4fbdce810","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 18:12:22.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"a697d0b2993602522ce7205983e510903639d7d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 18:21:34.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"78dfa40665e3b024c6c2a0c5c4f51f6fcb5a7568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 19:31:28.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851fcb27362746a48be2e8d3875ef83b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/07de9c247aa34a35a590878a5b85c820 : SUCCESS in 22m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f2a47cdf03249008e397028de89f0f2 : SUCCESS in 25m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0abe68ce1bca4d3bb1bbe5fc5d05c3a4 : SUCCESS in 42m 07s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"dbf803d489d381b84fdafc1d5dd703aba632aefd","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 20:28:03.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-w9p8s/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"71ec91a604aa9c48b66508be1054bda5edca5645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 21:33:56.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84aafbe0b8724f9cb26bbbcd13b09593\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea2a2d47aad24a0ea1fb476435f8fcc7 : SUCCESS in 1h 01m 25s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3cf224c84c2d4583bc99b6504d9243c0 : SUCCESS in 1h 30m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e36c08bea16453ba9a611822f1b9c48 : SUCCESS in 1h 00m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943f4f2e24bb49bd9a51463441d8ea3e : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52add9e8b27a4db0bc4d4a5877cfe098 : SUCCESS in 14m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6262b5735609442cabbf5605757f95ff : SUCCESS in 13m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7edac5a58b94383aeeac36777838552 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5df2e3ebc7040acb0cf54efcb6c2a4f : SUCCESS in 4m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/06d2a2a5186a4ad484e74b906752e939 : SUCCESS in 54m 16s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8bf7c2f6e3d54948bfff7fb49777063b : SUCCESS in 27m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e0eb167ab33f40a4ac2be045df70fb03 : SUCCESS in 25m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c293fd08b21a401b9b88ca82dbeaee32 : SUCCESS in 33m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/20aff543bca844ebb982ca8ff1d82c7a : SUCCESS in 1h 36m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/7cc9fc0a8f324df8aeae96552f94a553 : SUCCESS in 1h 00m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2bd3316e3ae34c1d9a7f954950309fe7 : SUCCESS in 5m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/90b612b6b8c842c89e67de15b167a3c8 : SUCCESS in 1h 45m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2a41677559a7456682d9cf8e7da25b66 : SUCCESS in 1h 30m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d11def1fcd864063aa70c9580a0306a3 : SUCCESS in 58m 29s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9addf858b81a4c7fa180a1695daf966e : SUCCESS in 8m 44s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7e9825a3b25c45a1965a5b54cc0fb94c : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":15},{"id":"afac1c72be88f85f4eb0261eb0cf96e9e9d93a1b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 21:33:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"e1c7586ed640362757351305f0381e6fcecec608","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 21:36:05.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4199b34b08ba4812b73fc605cebb585e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102ceab7761145c7957c8207024251a7 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6774ba06bcaf499ab4c656b2f7b8bb5d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":15},{"id":"922d320d47cd2c0e4f78995c9d97fb22a98e0fe0","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 22:04:08.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-7sks8/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"a79ca8e5c899f554133e29f6c098097e7e2c8aed","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 23:09:41.000000000","message":"Patch Set 15:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980676-ntg2x/index.html\u003e","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"318ca9ee790e5c39d9c4b342f2ae85d05c280990","revisions":{"157cff3408e4501c87f2e8d348031d73aa12489d":{"kind":"REWORK","_number":1,"created":"2026-03-16 10:37:08.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/76/980676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/1"}}},"commit":{"parents":[{"commit":"3b0a05d7625710fd13857c2f6eb082ca67e92b60","subject":"[WIP] perf: add ironic startup perf simulator and analysis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3b0a05d7625710fd13857c2f6eb082ca67e92b60"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 10:33:33.000000000","tz":0},"subject":"[WIP] perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","message":"[WIP] perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children), then wire it up via new\nsubtree_only\u003dTrue parameter on get_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() gains a matching scope_rp_uuid parameter\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nExpected saving: ~49 s at N\u003d1000, turning O(N²) into O(N).\n\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/157cff3408e4501c87f2e8d348031d73aa12489d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/157cff3408e4501c87f2e8d348031d73aa12489d"}]},"branch":"refs/heads/master"},"c625d97f849578766fa87ca4b028ab4dc04f7db1":{"kind":"REWORK","_number":2,"created":"2026-04-02 16:27:17.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/2"}}},"commit":{"parents":[{"commit":"653f7a42d0c9db126c0d5eb94af8945732f68d72","subject":"Follow ironic job rename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/653f7a42d0c9db126c0d5eb94af8945732f68d72"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-04-02 16:27:14.000000000","tz":-300},"subject":"[WIP] perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","message":"[WIP] perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children), then wire it up via new\nsubtree_only\u003dTrue parameter on get_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() gains a matching scope_rp_uuid parameter\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nExpected saving: ~49 s at N\u003d1000, turning O(N²) into O(N).\n\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c625d97f849578766fa87ca4b028ab4dc04f7db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c625d97f849578766fa87ca4b028ab4dc04f7db1"}]},"branch":"refs/heads/master"},"80ba4772bb41e5ac1a069b32e9c1b57153a00a59":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-04-02 16:28:34.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/3"}}},"commit":{"parents":[{"commit":"653f7a42d0c9db126c0d5eb94af8945732f68d72","subject":"Follow ironic job rename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/653f7a42d0c9db126c0d5eb94af8945732f68d72"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-04-02 16:27:28.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children), then wire it up via new\nsubtree_only\u003dTrue parameter on get_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() gains a matching scope_rp_uuid parameter\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nExpected saving: ~49 s at N\u003d1000, turning O(N²) into O(N).\n\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/80ba4772bb41e5ac1a069b32e9c1b57153a00a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/80ba4772bb41e5ac1a069b32e9c1b57153a00a59"}]},"branch":"refs/heads/master"},"68cf5a5db8bd602c0db551696a9a0278f39730c4":{"kind":"REWORK","_number":4,"created":"2026-04-02 19:27:53.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/4"}}},"commit":{"parents":[{"commit":"653f7a42d0c9db126c0d5eb94af8945732f68d72","subject":"Follow ironic job rename","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/653f7a42d0c9db126c0d5eb94af8945732f68d72"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-04-02 19:27:06.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children), then wire it up via new\nsubtree_only\u003dTrue parameter on get_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() gains a matching scope_rp_uuid parameter\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nExpected saving: ~49 s at N\u003d1000, turning O(N²) into O(N).\n\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68cf5a5db8bd602c0db551696a9a0278f39730c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68cf5a5db8bd602c0db551696a9a0278f39730c4"}]},"branch":"refs/heads/master"},"33c5e1fba0ac36f3069f3030bc699189fdb150f3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-05-13 17:09:43.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"real_uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/76/980676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/5"}}},"commit":{"parents":[{"commit":"b2d747f7d7248edb632e5982203d2e90e8e57703","subject":"Merge \"Fix functional tests and mypy on macOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2d747f7d7248edb632e5982203d2e90e8e57703"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-05-13 17:09:43.000000000","tz":0},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children), then wire it up via new\nsubtree_only\u003dTrue parameter on get_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() gains a matching scope_rp_uuid parameter\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nExpected saving: ~49 s at N\u003d1000, turning O(N²) into O(N).\n\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33c5e1fba0ac36f3069f3030bc699189fdb150f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33c5e1fba0ac36f3069f3030bc699189fdb150f3"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"68cf5a5db8bd602c0db551696a9a0278f39730c4","theirs":"b2d747f7d7248edb632e5982203d2e90e8e57703","contains_conflicts":false}},"de1fcccd967328064f51bc320e203cec6910f9e4":{"kind":"REWORK","_number":6,"created":"2026-07-06 09:32:43.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/76/980676/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/6"}}},"commit":{"parents":[{"commit":"e325555b184cd36413d348603b3b1abeb13bdcaf","subject":"Limit startup instance processing to libvirt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e325555b184cd36413d348603b3b1abeb13bdcaf"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-07-03 17:23:27.000000000","tz":60},"subject":"Reduce ProviderTree copy cost for Ironic","message":"Reduce ProviderTree copy cost for Ironic\n\nEach resource tracker update asks the report client for a copy of the cached\nPlacement provider tree before syncing inventories, traits, and aggregates.\nThe report client previously returned a deep copy of the full cached provider\nforest. For Ironic, where a single nova-compute service can manage hundreds or\nthousands of independent root providers, copying the full forest once per node\nmakes startup and resource reporting scale poorly.\n\nAdd ProviderTree.subtree() so callers can copy only the provider tree rooted at\nthe node they are updating. The resource tracker now requests that scoped copy\nfor its per-node Placement sync path.\n\nAdd scope_rp_uuid to update_from_provider_tree() so the diff against the cached\nprovider tree is limited to the same root. This prevents providers from sibling\nIronic nodes from being treated as removals while still allowing callers that\nneed the full tree, such as the conductor image-trait validation path, to keep\nthe existing behavior.\n\nAdd unit coverage for subtree extraction and for scoped report-client updates\nthat leave sibling root providers untouched.\n\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de1fcccd967328064f51bc320e203cec6910f9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de1fcccd967328064f51bc320e203cec6910f9e4"}]},"branch":"refs/heads/master"},"88493bec306f09d61a02d61aa7247a13c0e19a11":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-13 20:07:42.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/76/980676/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/7"}}},"commit":{"parents":[{"commit":"870d301254c666b25c6f282fb205ccfa8908a135","subject":"Delegate startup instance processing to drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/870d301254c666b25c6f282fb205ccfa8908a135"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2026-07-13 20:07:42.000000000","tz":0},"subject":"Reduce ProviderTree copy cost for Ironic","message":"Reduce ProviderTree copy cost for Ironic\n\nEach resource tracker update asks the report client for a copy of the cached\nPlacement provider tree before syncing inventories, traits, and aggregates.\nThe report client previously returned a deep copy of the full cached provider\nforest. For Ironic, where a single nova-compute service can manage hundreds or\nthousands of independent root providers, copying the full forest once per node\nmakes startup and resource reporting scale poorly.\n\nAdd ProviderTree.subtree() so callers can copy only the provider tree rooted at\nthe node they are updating. The resource tracker now requests that scoped copy\nfor its per-node Placement sync path.\n\nAdd scope_rp_uuid to update_from_provider_tree() so the diff against the cached\nprovider tree is limited to the same root. This prevents providers from sibling\nIronic nodes from being treated as removals while still allowing callers that\nneed the full tree, such as the conductor image-trait validation path, to keep\nthe existing behavior.\n\nAdd unit coverage for subtree extraction and for scoped report-client updates\nthat leave sibling root providers untouched.\n\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88493bec306f09d61a02d61aa7247a13c0e19a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88493bec306f09d61a02d61aa7247a13c0e19a11"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"e325555b184cd36413d348603b3b1abeb13bdcaf","ours":"de1fcccd967328064f51bc320e203cec6910f9e4","theirs":"870d301254c666b25c6f282fb205ccfa8908a135","merge_strategy":"recursive","contains_conflicts":false}},"706e5823f311805b77d8f358dab3d5bb3d8eb668":{"kind":"REWORK","_number":8,"created":"2026-08-05 15:03:56.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/8"}}},"commit":{"parents":[{"commit":"926ac87a92af987be9bc306a919b137db2cb3b04","subject":"Merge \"doc: Document postcopy incompatibility with vGPU live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/926ac87a92af987be9bc306a919b137db2cb3b04"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-05 15:03:51.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children), then wire it up via new\nsubtree_only\u003dTrue parameter on get_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() gains a matching scope_rp_uuid parameter\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nExpected saving: ~49 s at N\u003d1000, turning O(N²) into O(N).\n\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/706e5823f311805b77d8f358dab3d5bb3d8eb668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/706e5823f311805b77d8f358dab3d5bb3d8eb668"}]},"branch":"refs/heads/master"},"f6265b52a89bfbabfce8177c020a574c325ffc5e":{"kind":"REWORK","_number":9,"created":"2026-08-17 20:01:42.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/9"}}},"commit":{"parents":[{"commit":"926ac87a92af987be9bc306a919b137db2cb3b04","subject":"Merge \"doc: Document postcopy incompatibility with vGPU live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/926ac87a92af987be9bc306a919b137db2cb3b04"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-17 19:58:12.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children) and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6265b52a89bfbabfce8177c020a574c325ffc5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6265b52a89bfbabfce8177c020a574c325ffc5e"}]},"branch":"refs/heads/master"},"375be0325f816220ef591fae04030961d37df86a":{"kind":"REWORK","_number":10,"created":"2026-08-18 17:21:36.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/10"}}},"commit":{"parents":[{"commit":"926ac87a92af987be9bc306a919b137db2cb3b04","subject":"Merge \"doc: Document postcopy incompatibility with vGPU live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/926ac87a92af987be9bc306a919b137db2cb3b04"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-18 17:21:21.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node (K ≈ 1–5 for a\nbaremetal node plus any PCI children) and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/375be0325f816220ef591fae04030961d37df86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/375be0325f816220ef591fae04030961d37df86a"}]},"branch":"refs/heads/master"},"1508188a844eaf1ae4718727d65f5a5e6ee40766":{"kind":"REWORK","_number":11,"created":"2026-08-21 13:51:30.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/11"}}},"commit":{"parents":[{"commit":"926ac87a92af987be9bc306a919b137db2cb3b04","subject":"Merge \"doc: Document postcopy incompatibility with vGPU live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/926ac87a92af987be9bc306a919b137db2cb3b04"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-21 13:51:10.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1508188a844eaf1ae4718727d65f5a5e6ee40766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1508188a844eaf1ae4718727d65f5a5e6ee40766"}]},"branch":"refs/heads/master"},"7b18f5992cf0d019de1c37d972b2cd6dce54e60a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-08-24 20:30:26.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/12"}}},"commit":{"parents":[{"commit":"77d6da9539f08c4b3ac6e347d7e9c04a5194d7b5","subject":"Merge \"Add libvirt driver MDEV ARQ support for Cyborg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/77d6da9539f08c4b3ac6e347d7e9c04a5194d7b5"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-24 18:17:22.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b18f5992cf0d019de1c37d972b2cd6dce54e60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b18f5992cf0d019de1c37d972b2cd6dce54e60a"}]},"branch":"refs/heads/master"},"2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-08-25 03:13:23.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/13"}}},"commit":{"parents":[{"commit":"9aa9a54e04fdfec9a8775bf90488558783179d4b","subject":"Merge \"Filter out deleted instances when looking for build timouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9aa9a54e04fdfec9a8775bf90488558783179d4b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-25 03:13:15.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347"}]},"branch":"refs/heads/master"},"853efa4c96982baf7385ce67ae779050eaecf98b":{"kind":"REWORK","_number":14,"created":"2026-08-26 16:24:32.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/14"}}},"commit":{"parents":[{"commit":"9aa9a54e04fdfec9a8775bf90488558783179d4b","subject":"Merge \"Filter out deleted instances when looking for build timouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9aa9a54e04fdfec9a8775bf90488558783179d4b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-26 16:23:40.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/853efa4c96982baf7385ce67ae779050eaecf98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/853efa4c96982baf7385ce67ae779050eaecf98b"}]},"branch":"refs/heads/master"},"318ca9ee790e5c39d9c4b342f2ae85d05c280990":{"kind":"REWORK","_number":15,"created":"2026-08-27 15:58:38.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/76/980676/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/76/980676/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/76/980676/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/76/980676/15"}}},"commit":{"parents":[{"commit":"9aa9a54e04fdfec9a8775bf90488558783179d4b","subject":"Merge \"Filter out deleted instances when looking for build timouts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9aa9a54e04fdfec9a8775bf90488558783179d4b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-15 23:52:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-27 15:58:21.000000000","tz":-300},"subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","message":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup\n\nget_provider_tree_and_ensure_root() previously returned\ncopy.deepcopy(self._provider_tree) — a full copy of every provider\nin the shared tree.  With N Ironic nodes each _update_to_placement()\ncall copies ~N providers, giving O(N²) total work: ~49 s at N\u003d1000\n(20% of startup).\n\nFix: add ProviderTree.subtree(name_or_uuid) which deep-copies only\nthe O(K) providers rooted at the requested node and returns a ProviderSubTree,\nthen wire it up via new subtree_only\u003dTrue parameter on\nget_provider_tree_and_ensure_root().\n\nupdate_from_provider_tree() now looks for ProviderSubTree\nso that the old/new UUID diff is limited to the requested sub-tree,\npreventing providers from other nodes being incorrectly treated as\nremovals.\n\nThe conductor caller (image-trait validation during rebuild) continues\nto get the full tree as before; only the per-node resource-tracker\npath uses the new subtree copy.\n\nUnit tests added for all subtree() behaviours: correct extraction,\ndeepcopy isolation, child inclusion, sibling exclusion, and error\ncases for non-root / missing providers.\n\nCloses-Bug: 2162924\nGenerated-by: Cursor sonnet 4.6\nChange-Id: I334f80ed76ae869fe241d468c0bf5dd3d0634bbd\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/318ca9ee790e5c39d9c4b342f2ae85d05c280990"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/318ca9ee790e5c39d9c4b342f2ae85d05c280990"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
