)]}'
{"id":"openstack%2Fnova~980679","triplet_id":"openstack%2Fnova~master~I36163947a09b82cfc40ab4b522d82ffb503a0b84","project":"openstack/nova","branch":"master","topic":"startup-time","attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-08 09:08:33.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-09-07 21:05:47.000000000","reason":"Change was submitted"},"14228":{"account":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"last_update":"2026-09-07 21:05:47.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2026-09-07 21:05:47.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-08-27 14:26:56.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-08-26 17:06:01.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I36163947a09b82cfc40ab4b522d82ffb503a0b84","subject":"Parallelize per-node resource updates","status":"MERGED","created":"2026-03-16 10:37:08.000000000","updated":"2026-09-08 09:08:33.000000000","submitted":"2026-09-07 21:05:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"980679-startup-time","meta_rev_id":"7f24300b197df7e2b64a5b2265e179e73e40d9ae","_number":980679,"virtual_id_number":980679,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 21:05:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-08-31 08:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2026-08-31 08:59:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-08-31 09:05:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2026-08-31 09:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-16 11:17:58.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-16 11:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-03-16 13:06:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-06 15:45:29.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-15 21:47:45.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-05 17:08:01.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-08-05 17:58:41.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"state":"CC"},{"updated":"2026-08-12 13:04:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-08-25 21:41:28.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2026-08-25 21:47:17.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-27 16:04:38.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2026-09-08 09:08:33.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"}],"messages":[{"id":"000f6fabd4fe17083613141b86143ec9c0d114a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-16 10:37:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd0bfcd1e4035dbd917a0dd0b622241ad103ef57","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-03-16 11:17:58.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":1},{"id":"87b1138a14d3eec81c47d9e9373fa171b453667e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 11:19:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6911586fbc5243fca6d7f72287ae9ea8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32bc8e96daf24fe29edf96c76f456090 : FAILURE in 14m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/21bdfdfc9ef44689afedc7b700d80676 : FAILURE in 15m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/87f336d45d564529b8af9b60940b0171 : FAILURE in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"964c0fced99960cec81aba5005709db4bdd7a5c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-16 13:06:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baa5184c4a3742419e01bfe303909c37\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8327ce5a195246e3aa2e62e0b755a1cd : SUCCESS in 29m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e555266f44ea49d8b1c64301fc7638f9 : SUCCESS in 1h 59m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc45782816e94ff4abb11fc5f196977c : SUCCESS in 1h 04m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fabc72762724a1aa762fd2405498b10 : FAILURE in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eaf1e889d0e420cabd117e48cea0501 : SUCCESS in 8m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c528d1085ac4ef8a8041627b431ccd3 : FAILURE in 19m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/092a8c1fd27b4ac0a0be76f002820c02 : FAILURE in 13m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7fbc37148efa471c8cf135be8a38959d : FAILURE in 11m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9234f7dc3541f681aa3fe17fc542c4 : SUCCESS in 12m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b36c69619a04b9abb6fc88b03efbdb9 : SUCCESS in 1h 55m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2a6aae1a216543d980fdfb1412e52265 : SUCCESS in 59m 20s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2b7bffcf78914dc4b4fed5f7de7ff0cd : SUCCESS in 1h 04m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1929b1b67a1d40b68757202853c5c1e9 : SUCCESS in 28m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/73384d54081544319bc7044f261ba6df : SUCCESS in 1h 26m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/4270eb171d184e36bdb937f8e939a172 : SUCCESS in 2h 15m 04s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a5caaf54334c405eab779be7ad925d22 : SUCCESS in 54m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c1148faa7b2345af941238c52fa6b422 : SUCCESS in 4m 04s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9551c799a39e4ec3a647f67dd873df2b : SUCCESS in 26m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6e701e7f98674e8eab8946f3eff71f7f : SUCCESS in 29m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/60940e8e808b4d8b89bbed9ad26777eb : SUCCESS in 1h 28m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/694a3ea8faa34afaac225cb1bc618e3f : SUCCESS in 1h 00m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4fdae8b223c449b4be22f8c4aeb3a07e : SUCCESS in 29m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/afa6b7ed845142cc89e4e107699963bd : SUCCESS in 35m 57s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/18a6a112e95c401e8a665f6a31738e89 : FAILURE in 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bc91f18f1d72a729905be31b1bd5deec59fa95db","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-23 17:58:03.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"aee9a5adf81676dfdf2b6a8d0ffc84bc1da61777","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 09:32:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"74bac31fac8b3943da6947eb2361a16548cbff89","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-06 10:44:05.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":2},{"id":"63fcbed3d1b65faa61049c35ebaa83cffac348a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 12:07:48.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f394bee205ec41eeb0ad3a130b18af6d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aa7c658722f04d8192fbb1d4336263a0 : SUCCESS in 33m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2ac764a34d364d339339befee162f189 : SUCCESS in 1h 18m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37cdda8358c04fe19024c433390427ca : SUCCESS in 1h 08m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1bee3ae01f34ccba6b33e45cc00995d : SUCCESS in 1h 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6336c955fd4d64b78971b891c97b33 : FAILURE in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/940ac41cdcaf47958d70b57ceb34a1e0 : SUCCESS in 12m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/67a9833bec1d44daa479b003c37f739e : SUCCESS in 13m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c6b0f89d7fe04eeb81d27c48b2c3260d : SUCCESS in 15m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da4af77f2fa9442fb77ebcd03f18ec1e : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4376a6b580bb406b87b5b4b4a83cd652 : SUCCESS in 8m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/06a8859c87a14cbcac0f3d1c423aeb37 : SUCCESS in 1h 11m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cfa9e175f7dd41d7803142adaaac2c9e : SUCCESS in 1h 50m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/df757c67ad7b4d0992817a4325dd9e70 : SUCCESS in 37m 25s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/23894ae07c1b4518b1752810418833c5 : SUCCESS in 42m 29s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c2d11f13277947ffb01d40b881e51347 : SUCCESS in 1h 02m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/56206362035c48efb2697a30790d4ad6 : SUCCESS in 46m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3498d094518940df81798c0dacd0f9df : SUCCESS in 2h 00m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/34137989c26247a3a907091505bb6627 : SUCCESS in 2h 11m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/902e3b9439bc44838947f31fadda4908 : SUCCESS in 35m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6db8a685208c4b309988829843efafc1 : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f634eb6d5521485f8b4da0a6ae1c4fca : SUCCESS in 23m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7e9b9c20a54b4dbb919853a1d2400337 : SUCCESS in 29m 44s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/032d36563d504869878b2a503623cde1 : FAILURE in 48m 31s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e174c57734de408d873f7bd8c3fde4d9 : SUCCESS in 1h 32m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71ff275a895d45e6820db38f855811dc : SUCCESS in 45m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/02888be1418941918d558c63f571a966 : SUCCESS in 18m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/74137707085e499bb3dbf19b481ae15a : SUCCESS in 40m 24s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4c1cc8874a42432fbc62854a8ee65c11 : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":2},{"id":"c8c8256e43440123630d223bf43422226bb7dd59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 12:29:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad57c04f0201454f9496901abf022cce\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2908ab8de3346edb59832219f285862 : SUCCESS in 20m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/014678be6e8d4021bcaafb5712ff620a : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4f7bec4c85024014ba4372fb72e02305 : SUCCESS in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f3d3ca7bc479274abfe5a512d5287c1823a0dc0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 12:39:26.000000000","message":"Patch Set 2: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d9741060ccc51757db49dd9bf5231a749d5a9eba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-06 12:50:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"13569925351472620c16d3cdb217658b43e7771c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-06 13:52:17.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":3},{"id":"317e1b04ccda468699288736c016f79849b781ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 14:51:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c711cf3e40ea4968a1eff9877f063be3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/805f356a0c534f998d57fce239587802 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ec973575ae14822b6d20b719cb1e013 : SUCCESS in 16m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dc1ef8e3259a4959b7ccf15c533915e0 : SUCCESS in 34m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4b47ac9f3f84857b028dce838b128e787019d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-06 15:45:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19ffd5a9a5bc417ab1b8fae3eb554a8c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b50b8afe0e83473dafa4b17a5d8710c5 : SUCCESS in 1h 02m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6655bbe1edf248f4818efbf3943182d2 : SUCCESS in 1h 47m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df7aa35cebeb4e66994b14743399479b : SUCCESS in 1h 09m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbacfba2ad8e47de95a1ddf497c45feb : SUCCESS in 1h 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2184aa539c704790857251f62fd2bad3 : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a846762e72c04962ad2c4d6ba6914696 : SUCCESS in 15m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9496c35a95024325bbb9cc3368a27b67 : SUCCESS in 12m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7a7a7ad8bcfb4ca18fdc42dd171c2bfb : SUCCESS in 17m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3348646ffd043e78641cbf29f56bb10 : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/921ef69a54da43238493c2dc0ca902a1 : SUCCESS in 8m 57s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c3a4f4546fbe46e6b781ac0438200559 : SUCCESS in 59m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92f5243945fe49c3af691fa4abc11512 : SUCCESS in 1h 51m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/44527218f20b46e09aff05f4db044c36 : SUCCESS in 57m 56s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/3b12d76995ef4adeb649b9229a6a2019 : SUCCESS in 44m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3ca296084e9a4dff98f37c267337a787 : SUCCESS in 1h 00m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6a8463a260d74b31a2ab37cc29a192dd : SUCCESS in 47m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/12a456c0caf847d586febab3957a7f77 : SUCCESS in 2h 11m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/54682d41502840cbbb48e3b02d2f0f82 : SUCCESS in 2h 43m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/002c14fc8e16431ca97031adf64544e3 : SUCCESS in 54m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6b0aa1b9100340ffa12f48724bec90cf : SUCCESS in 6m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ba3aae76851245e38122d89c6a80fd8c : SUCCESS in 27m 00s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4a4fc4fc42534cc3addcc97cca5cdc7c : SUCCESS in 25m 04s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dbc8989fa26c417ebb49608d50ef2d32 : TIMED_OUT in 2h 01m 26s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aee1b28766d84b468581ecda23f7d6d9 : SUCCESS in 1h 37m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/357e6239107a423d9748b15974e45de9 : SUCCESS in 55m 13s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ba79e7dbc54a4f5098c0426f47c10f96 : SUCCESS in 35m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7c2498b4dbbf4fefa0a95e57165225d0 : SUCCESS in 35m 54s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fa59d5bb3d27410eb11e882e7582ce61 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":3},{"id":"823c718c17c0bd2aa7f54bea5929e98cdd190d9d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-06 15:56:38.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"fa7771204b9b03cc93c026758a6de5a8fb2be526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-13 20:08:28.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d71b96683e7ca63dde83a01337533baa6005255b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 20:57:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe215c4ae87b4724a0eeca1e3c3f878a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f3be47f501b4653946485752b31b211 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/38a845753d6846dfaf02ab320d0677d8 : SUCCESS in 27m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/de09dddecc524c53ba31d5cbd35410ea : SUCCESS in 42m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"46b727a3d7dd8b0dff3c383e4a929fb70b423ecf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 21:07:29.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":4},{"id":"57e659109554d7ad0e2f7847d1f50576334d1b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 22:07:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/408c712e980a4eae8585ce282e1e7e34\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50e67b206eef4dc9817984506182185a : SUCCESS in 29m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fddf281e81c9469c9f342f9a702a7f43 : SUCCESS in 1h 34m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79779c6f9c764cd1930b19398a7ad8f0 : SUCCESS in 1h 05m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ede824d4b934359b4441461eeac742d : SUCCESS in 1h 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80cf0de5af1643779161b5af6ee6fe24 : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e13aded4c2c46e0ad198908025b4c09 : SUCCESS in 10m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b421b1d7f70645f6aab2cd7e11550f44 : SUCCESS in 14m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3d250e91c9da4339ab9d68812c567232 : SUCCESS in 18m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d496f387b7b40f59419629d6c1eb668 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c59573ea3bf41a6a40ae1ccaa9334a2 : SUCCESS in 7m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a0f8fd080b7549c98ddfa96439babd26 : SUCCESS in 1h 03m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f6570dab7d47441f899b6c453eff9f0e : FAILURE in 10m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c471c541a517401194478f450de97c9d : SUCCESS in 48m 36s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/be1e3afe02384b4396e40c1e52cec7be : SUCCESS in 35m 25s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1f2badd9b6054345bb9f9259574cde2a : SUCCESS in 59m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/44ccc27c66ae470b8e4ae8ae49fc4aed : SUCCESS in 49m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1a6fe324f5b24fe49911c79c365afd38 : SUCCESS in 1h 52m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/bce363bbc20f42c1832e9e64c5bed3aa : SUCCESS in 1h 52m 36s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f031f62f05de4364a429745880b097f2 : POST_FAILURE in 30m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dd8bed48c34a4c3bbd794d1b8e194939 : SUCCESS in 3m 37s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6585cd11b08a41de9802a13a8a41c170 : SUCCESS in 24m 56s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2c1e11ae36f74b95b4cbaaf7e8dc315d : SUCCESS in 29m 54s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8ca12a58dd40436daf30924c4437c878 : SUCCESS in 27m 18s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/97351e5c3d064e23b73182432e4dc92c : SUCCESS in 1h 24m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dbea5b446cc4e7bafd4a9a326da164e : SUCCESS in 56m 04s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/932e15bc326f4eb6bd91b0a8c86ba200 : SUCCESS in 27m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/efe87f93e7a042c3979c77487ab293b3 : SUCCESS in 33m 41s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ea0a3e15deed4dca9bac5148b0135e69 : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":4},{"id":"d154ab05d1b2daa9cf55be8eaa7e586303402843","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-15 21:47:45.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-vzf28/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"54d13f6b7a1ad5890e14b75556eef4ab2dfce536","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-20 21:55:03.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"687f355cde7978edaff8ad56b2f84cd2e369fbee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 22:27:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fba380c281e43b4ae24f6cfd2e5f5e9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6028baea65ab4c6c86ec3e9a6b653cef : SUCCESS in 20m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/74e366fe74614957b1a6184cc3ed8fca : SUCCESS in 20m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51648e642593445bbd35557c9345467d : SUCCESS in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5da1380934df396aad5e06b7b973c8d37d62df08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 23:52:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2697f6c045bd4f6f88cc4f018b6013bb\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/98277e9a3b4f482aadbb36dac690833f : SUCCESS in 45m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4cddd624f8df41da9d5d1bf6e8dc428e : SUCCESS in 1h 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4ee3f23e5d243e6b7fae6ed8ee74026 : SUCCESS in 35m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1238e433b10742759fa31c8fa8572019 : SUCCESS in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/550a8f76471847b4adcba27ab1d19b4a : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b5de10e99744bcf94be6c93210f0123 : SUCCESS in 10m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f401684e952845a89ec8159c26c5b677 : SUCCESS in 10m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/57e46882e3764f31bbbb71cbc76d343e : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3b692ad8cd432085128660951c1c7f : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/732b29ab4de2418f8713f54c559f527b : SUCCESS in 8m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a2c561879ed64d50a864e26ddc5bc53a : SUCCESS in 45m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e0f6b76c04b64f959ae4908d37515c69 : SUCCESS in 1h 08m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c2a57f9bb0c445909b89116af8900d91 : SUCCESS in 50m 18s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/827372449e6b4c78a1f0ec7bbe67bd71 : SUCCESS in 34m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/43314668fbca489caa4f406dbb825acf : SUCCESS in 45m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2fe8134bf14a4568a1b66b40f68f9aef : SUCCESS in 48m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/01c6718caf564505b47647833dfd2317 : SUCCESS in 1h 49m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/dc6a3be856aa4e54838e7dcfe5ef70b4 : SUCCESS in 1h 10m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c20a316df5654c6c8c9f22e5dde981b5 : SUCCESS in 32m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f4b5b7e553d0427887cb7ef8aa4be68c : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93666009e1484f78907a95f7eb3be14e : SUCCESS in 30m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/32ee57c2a1f74979a026cd101703cc2e : SUCCESS in 25m 58s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7ecfc805ee06495e93337230848873ad : FAILURE in 27m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/44510ef00eb740fa87f783763c3c2924 : SUCCESS in 1h 08m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aa8637cc7494243b7b557884dba5375 : SUCCESS in 51m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a9450550237d4c2d9743c45676a94c24 : SUCCESS in 22m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/913f2a88e59a42129335b4350546313a : SUCCESS in 14m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/90ad98cb78a34b13828c514842fcf97d : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":4},{"id":"7ad2cc2fc700eb318ba45d24791c80f0619c0329","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-21 00:02:36.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"61ef12aa346aafd7633f0968c045e73e7b19f467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-05 15:15:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"9d60b13f1040750d95ba924a4161618e15aada75","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 15:18:13.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":5},{"id":"0a1fc734cbb526762132b972f8e2ab875b2eb880","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 15:45:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18a89e74a9684de28702a99e801cde15\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c276d1f641f7441b8bda03cb13d661a0 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d15d744e70904ed79b128291ed55721e : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/46635bcdb4a548c8b70f21b6293c206a : SUCCESS in 27m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91e666b2d41dbdf79befa58d65234e1b6d5eb36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 17:07:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e7fff1174c4495f9c8d8e33f13c5628\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84628a4a66be47078f6dcdece17dddea : SUCCESS in 46m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d12603d0e3db4bbcb13b308e06c695e9 : SUCCESS in 1h 34m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f69c8ef0e22d42bfa08b3ceb81e78adb : SUCCESS in 30m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39257a9f71dc487fb14b6e9eaaa93e15 : SUCCESS in 1h 03m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d9f038debb4bf4a5ad3ac0a0ba4155 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2523d720650a4515bc490f9cfff79b52 : SUCCESS in 13m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b7d07d59a60e466ba2ae91a5fd4d71d2 : SUCCESS in 19m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a4ca95f676d46b889a2c117caf1f4ef : SUCCESS in 17m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d09d45027aa4d2db6df276984ab8387 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d223bdf90b1a449db368f4c1f28d4486 : SUCCESS in 10m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/fefc1ea43f84452da1a79dd8f6c64f83 : SUCCESS in 1h 09m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1527afd1089a424f9253d6473377705e : SUCCESS in 1h 24m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/de1ea94d9b0449b1bee16b979f1dd3f4 : SUCCESS in 53m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8ecf539be23f4c38ae8c2e973d265d22 : SUCCESS in 36m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/77067bf4d09f483cb1e02155d8cf60ab : SUCCESS in 1h 06m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2cc8a23aaa074886998ad74b38cc9fdd : SUCCESS in 53m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2236a269a9364686bce6f136cf1515dc : SUCCESS in 1h 49m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/cebd1fe8f686437787c2fb7b510e6567 : SUCCESS in 58m 31s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2695885e2d39407a9099235baa423a15 : SUCCESS in 44m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/887fb502a94d4b6ebe02978852d55987 : SUCCESS in 7m 26s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/02e8c56081894ed2aecc635b3ad770e0 : SUCCESS in 28m 28s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c984ff5f75fe4211a1839fd7d5c48bb1 : SUCCESS in 25m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c4f8555fee894694b6e2b90362c11955 : SUCCESS in 33m 39s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef9166b5387e4e93ac9ee0afc542b7f1 : FAILURE in 53m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b756ab3c9a7741ceae9177cc0af9dde7 : SUCCESS in 1h 04m 00s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/52c0935d21c1450f9ebc43f4ca6838b6 : SUCCESS in 58m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/60d9ef0a16bb42748f1feb307c9e2116 : SUCCESS in 18m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2917b74c9a0e4b18a9cf6cbab9d29f4f : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":5},{"id":"1eccfed4e447c5313a02e3cc8e2e052043188114","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-05 17:08:01.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4565ece69fded6e3694f6c64821f46fddf1450dc","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-05 17:58:41.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"296864180316de64274ab87d796a7fc4238687fe","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-06 01:36:37.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-p5t67/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"cafcedd3666f743fd57702f76bf82562291954fd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-12 13:04:51.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b5a28cf66c97bd620d6e4e4c342c4c79b24e2aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-17 20:01:42.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased. Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"9c9b6a96aca6dda6ebd18aba313bd787644975e4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-17 20:33:49.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-2mjdm/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"e117527977a10cf279601ec5d7ea512aa6417b72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 20:36:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e916702cb37d476ebdcb36a30733dd70\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c97b159416a496a811666f29faeebc6 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/15f978f02f2c4dc599ba840336e01e54 : SUCCESS in 25m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a2b89d13a29a4397acdb9b77e6137ac0 : SUCCESS in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e1795b38f6fc81756ded09df3264f1997c81f79f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 20:50:44.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":6},{"id":"d3d891369062425f03b4bad7712461534ab182db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 21:22:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/035e9468b6ce4b3c9e8bffe0c4bd152a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/244a937038f44d91a0fc5ad15ae51674 : FAILURE in 58m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1a1c06c56d2a42f799c07462b2852ece : FAILURE in 53m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f0e3835412e4a03b51644f0d5901175 : FAILURE in 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea867f9a8c284cdfae305dcefe06d39a : FAILURE in 1h 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b1801b15244fa1b9d937f719aa9050 : SUCCESS in 7m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f4b7f9e789844b82858825169ed6afb0 : SUCCESS in 13m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b862b2ecc13e402dbaa2379b6172921c : SUCCESS in 14m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/07ac4c4fe74f49b4976cfb94a43b1804 : SUCCESS in 16m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4827064e6944426f9ea0f55bd100b808 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed68489d59fe4bf2b52487fcce94ac6c : SUCCESS in 8m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e7acf43d994447c9bf42e7468b88b39c : FAILURE in 24m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3463d5d6a63d46658d0459076a61c920 : FAILURE in 36m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7800221bab984091ba05c6be9d1de896 : POST_FAILURE in 28m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/16af8fd8b6ca451a97a80cf4317c1f30 : FAILURE in 30m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/12f3da3458374165ac507a57621ddc77 : FAILURE in 34m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/743789fd5a114b5f9412aca293aa65c1 : POST_FAILURE in 46m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b39c2df3f6494f74abc43b767783ca1b : FAILURE in 57m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/153b7572e9cb4c7e868d2d50db75a05e : POST_FAILURE in 59m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c62865653ce54814be76e596231873c4 : POST_FAILURE in 34m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d154d4db03fc49db973fc8b10424a88c : SUCCESS in 2m 56s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/775f7d1de0d241a0b1b87ec6f73242d6 : FAILURE in 37m 21s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1ce3b80aa83c4958ba6532e27b437417 : FAILURE in 37m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e829072ae4284242b5637121a9fa7377 : FAILURE in 54m 54s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/534a342308554926961e5fad3ed18779 : FAILURE in 1h 07m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af7f1eae3ed4406fad7eac104b87dab5 : FAILURE in 35m 17s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/39677418cfe144c69e5e8f898a6a7113 : FAILURE in 45m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6e3e99ab0cd0447a87f02d8d0c4dda49 : FAILURE in 33m 43s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/457c1cc373474f4d8d6c2c61e7f7458c : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":6},{"id":"dbfb03897e9d3be6024b93c3c8ee3b4adb781fb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-18 17:21:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"1842c1cf21a31e7e54c9b82c4fb6f122510482d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 17:37:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfd0fb32ae384e779736ce5f0c4e6830\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9cd8377f7624d8b82a1f530f10ade94 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a139db3007a457f995c1b2525f8a7ac : NODE_FAILURE Node(set) request ea4e0f6fefc3430696e108c89d81807e failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/aaa2b26c7a504649a09b937a1105d8e6 : FAILURE in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"342fd5ad5de6c699a8d49d3ff8d295e683c512fe","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 18:13:29.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-zsms9/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"e7a1c81fb50e659ca185eceedfc3bf55638332fe","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 18:18:32.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":7},{"id":"7975a9adbf62bad6dbf7cceb05a0ed19de48da5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 19:31:26.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/89815e96c1994d3296270d056e791a3c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9d83c5ccb7664a7898175755e04f6c16 : SUCCESS in 1h 03m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/367534ce624b415fb9716974a9cb78c9 : SUCCESS in 1h 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21271300e14040d4a2ae786ccf496243 : SUCCESS in 56m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864ee290da534eba88c777ecfe08a7bf : FAILURE in 1h 40m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90d0fa87f40b422da1709d834c8f2454 : SUCCESS in 7m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07bf09be94fb4669bdfd6391447f63f8 : SUCCESS in 13m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b59db240609c4c89bca6d5e93a8f5ad1 : SUCCESS in 13m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe2152cab7424346bc5495ba3e742d06 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3c4e12940db45dfaf9d0437d418d777 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7db2e57c9a1340bd82181aa2eff0529b : SUCCESS in 8m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b7bf9f8fa1364c558c7102b863fbdfd1 : SUCCESS in 31m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0942549f87e48c6a0c1948dfec6a4c4 : SUCCESS in 47m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c9013a8e531749a584e4d9ca22c7e63c : SUCCESS in 42m 02s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e31513766d9f42b3aea2f4d1c8e54d87 : SUCCESS in 35m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/bff48c7b1d964ed487fbd24eed51e6dc : SUCCESS in 40m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0e1f68e5f88f4ad7bbffdf329c1c1e60 : SUCCESS in 27m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f15be6a53d7e43b389f2072bb0e09836 : SUCCESS in 1h 24m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/c95cc927d463445aa28947409f7ca673 : FAILURE in 2h 03m 08s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9f40eaf392f44bafaca660f0ed91bd5c : SUCCESS in 47m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2b8dd8b9a0654238b42a10ded8bc4f1f : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d55be9942dc149c086d2991927962c31 : SUCCESS in 21m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9eea74eed80d49f2bd18ed16a2007704 : SUCCESS in 27m 19s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1560795590ed4601a397f6abf0729390 : SUCCESS in 30m 59s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4983006e2ff4b28ac1d8278a3ef9f29 : SUCCESS in 1h 29m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/280bd7ce5c4d4e00b0cbc17e150a5ae5 : SUCCESS in 36m 17s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bad910e805ff4021a63c699ea4e6cec4 : SUCCESS in 53m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1f18e642f3f841a788342ff2890a884d : SUCCESS in 34m 03s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7faebf0e607a44369a45e663bba0d46e : SUCCESS in 18m 28s","accounts_in_message":[],"_revision_number":7},{"id":"303b7d8b8bff649babdc2d76aa2207518292f097","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 20:03:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bc6f8829dfb7f6009d86fa579695c4418c38b6c8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 20:06:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a78ad78093b6d3adb8b2b569e4d7f3980789a625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-21 13:51:30.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"0086d43749794b20b01e5f2ea8e692a869761b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:02:49.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62f9433ab93b4c109fcfb52d2c8fb6c8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a96c11e88924eb8ad1f85ef6a30e066 : FAILURE in 1h 04m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/560b1e942d4b401ca75fa07e4cf46333 : FAILURE in 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d95abef1bdb74040814eca9b3985d9a6 : POST_FAILURE in 15m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405f5640ffde484d840c12a6f2391a42 : FAILURE in 37m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b0f0281604d4a83b82aee5076a83831 : FAILURE in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5a0e52e17eb433d8902366c65c4122c : FAILURE in 12m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ec6bb4c98bf4f05990cbc7f8078ff17 : FAILURE in 12m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/69a0e2bd3e264843bf3d3d0922c0072c : FAILURE in 15m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8323da3f2db4700911ce03b87eb7684 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f7b9da602e54788b55a9789577532ad : SUCCESS in 7m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/910b09c2e5a14f0db96d7c14a8d40885 : FAILURE in 32m 30s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fad6c2b85f8f44debd88b80353620e4c : FAILURE in 18m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/01be66a12399413f8ec934df58bb9aa4 : POST_FAILURE in 23m 59s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/587dcee4606e4648927d4cbbab4be2a7 : FAILURE in 22m 11s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a0844051d6e8484ba37e59a5cee8c4be : FAILURE in 13m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2848f6656dd745909444158c693979ab : POST_FAILURE in 20m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/48179bc0f93e49ebab4bebebbad9a3e8 : FAILURE in 15m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/6f951fe8b08749c2b0fe32dadcd58b92 : POST_FAILURE in 32m 19s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/162d01ff9ae141d5bb053647352d099a : POST_FAILURE in 16m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bf0c8d2a60a94aad926a8b2f2bd8a056 : SUCCESS in 5m 30s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4c8799ca297a4d8b9632530cf73f97ba : FAILURE in 12m 48s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dadd589e79624739aef8df4a14a2117d : FAILURE in 10m 30s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d0295ebd7bcd446db9575165221800c5 : FAILURE in 11m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c72fc13c7214cc0aecda8e1e7d29a2d : FAILURE in 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d85153188a214d62b562439072843367 : FAILURE in 21m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b1d67cb3c36e44798705eb6bad72a7a5 : FAILURE in 30m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c86c5725f5fd4d94aba107cbdca079d5 : FAILURE in 28m 59s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ab798fe3322946a7865ec5ea8851a86a : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":8},{"id":"9db50c3804dfb6d6572452f634860ccea29d6691","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:03:43.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/551d93ce77914fd6bb983e6140cf1e0d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/70943dbbacc74ee9aa0ee6b24582bcf1 : FAILURE in 23m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d8cfd422cb89410eb844a567d8a71427 : FAILURE in 11m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b11a159b0b114dda862be0b7206fcd0c : FAILURE in 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"41c002d38b5137d6469154303ad3c775624b9c99","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 15:42:33.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":8},{"id":"51e586c8c5a8dd8d2bdd5065fd7d5c9be1dc38f8","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 16:17:56.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-zk2qs/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"8770bf1934cc068a4bd7c11e5485f0692bedddd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-21 16:19:44.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"3ce5e4f9643e9638e6c715c72f1df9785ab53c93","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-21 16:20:40.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c3094b92e0764e7819c5c1d6d1a5bc2d078a3496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:08:02.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbeb5bd4053a4c01ac2944224c5d4d3d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39bee175afa246b8844456cd34c035c0 : FAILURE in 28m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05d8b6f61f96442881ff1c6a6269a5a8 : FAILURE in 25m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2cb240dba2d44b5182af789f101df8a4 : FAILURE in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c791244bbb305ea627e9329f658e3f672e16fa44","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 17:14:48.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":9},{"id":"c2e13072e2c1b9fd1f9e1febc17cfda8447d621b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:15:00.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/39730af6d1b84139a4a222986b56e5d0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79ad93e65a3c43aa95573c919f10437c : FAILURE in 46m 59s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7d8a5a2853e048a581eb94286bb56a81 : FAILURE in 30m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1b28eea511c4fba94e8ee9a503295d7 : POST_FAILURE in 16m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f6412620a1d40a08c12e0bb0d6b7e9c : FAILURE in 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84af3f7b4bcc4fa8beacadf98fed2682 : FAILURE in 8m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76bc28ae42674936b163313a93fadef2 : FAILURE in 11m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b662e3d11e34e8480243d8d9fcd4194 : FAILURE in 10m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/77ac3325f52c4c72b1ad0c838af54dfa : FAILURE in 14m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2afe129ad446b29a86c8528de0d74a : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/152e0a5d7367461ea04bd32ef5afbcfb : SUCCESS in 4m 58s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4ef05c3cadd54a689b22c9c0ff9d3a70 : FAILURE in 28m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ab9281be59f6440e995682aedd1e9002 : FAILURE in 22m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6cd68bc97b75402292b09df68e6d68db : POST_FAILURE in 21m 43s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f9f1554fbcf04cdbab24480ed275a1ea : FAILURE in 22m 42s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/774507e04cf34f3ebffa2dc11682e9ff : FAILURE in 20m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/61f894144b9a42969c0f3eab2eab5992 : POST_FAILURE in 26m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/15b681e0d1be4f5fbb65a99f19baf7da : FAILURE in 14m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/b06b3d75457f47fcaa1f7a78909ab4f8 : POST_FAILURE in 21m 36s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/de70b02d41bb40daa7fb19c61d2c8c34 : POST_FAILURE in 18m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/121d352259c247b3823da45047548203 : SUCCESS in 5m 30s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d8502df00ac147a697c9429a349b1471 : FAILURE in 10m 04s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f768a8a8b79040628528c2753056f424 : FAILURE in 9m 39s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0946770aedd64a4694af9c4c57fed544 : FAILURE in 10m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/65024cb1adfe4d3780c23ed0a775abd1 : FAILURE in 39m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a72e3fc18554c008db91e98148e171d : FAILURE in 20m 03s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/26ddc5a3bc44442f92e178d4d68174bd : FAILURE in 19m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/017ddc42ae4b4cbea134d1c0c623e150 : FAILURE in 13m 27s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/048ba3f9303e46ef85f5911ae32ff728 : FAILURE in 10m 54s","accounts_in_message":[],"_revision_number":9},{"id":"6f93f2fca9c5d871e7c23e18a7620d49c6924e16","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 20:24:45.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-xtlvb/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"84a1e2db741aa6c233f5638f537f385cc60f9ea2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 13:52:54.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"031acdd1f0cb0e4b5773af32f02cabaf7f096580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-24 20:30:26.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"3ad55937ddc9c0dd6922712278ae66e5c6cf14f4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 21:29:59.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":10},{"id":"eec44e7b306f844ad82ddd717ed6b36dd61f3b6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-24 21:37:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"27494a299fb4d37d29651907a374412d86d747fc","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-24 21:37:59.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"847078be632a111726a91140ce076441fcf42e07","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 21:42:14.000000000","message":"Patch Set 10:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-l9fq9/index.html\u003e","accounts_in_message":[],"_revision_number":10},{"id":"3cca58a864df4c1d41ff4508fa45730daa1e40cd","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-24 22:01:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"574c5f6c127f12be2ac1a5e1805c5817aa4c1a71","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 22:13:25.000000000","message":"Patch Set 11:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-jkfwg/index.html\u003e","accounts_in_message":[],"_revision_number":11},{"id":"254611811d59c4292f4d3c2a7cda3ef3e8621bde","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-24 22:33:47.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":11},{"id":"0cbee9dd5d5155c553218625772d508c39056d37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:32:40.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddfdcde63a5c417dbd4f5bef2f4e8276\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4344609e27f40719452eb8c911aa8f2 : SUCCESS in 1h 05m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e00cf7678f624da68b24f2662dbd443a : SUCCESS in 1h 38m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00579d71d6374e3b82cc9cab21aba923 : SUCCESS in 46m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bc877d4c2b74d74a9b9d7bd76f41913 : FAILURE in 38m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67dae368458a424fa71143f8b6e203e1 : FAILURE in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0495acac8a20490a89d390a02d7893c8 : FAILURE in 12m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb83fa3b00224643b7e2e4ebfa80a46b : FAILURE in 10m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ffac1b8ab6524e73bd4a5917f9d30cba : SUCCESS in 18m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4f0b6c3f124a96b7a290af22f71d8d : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9305a51c28945dcba2ee2f9de62679b : SUCCESS in 8m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/62758ad976884f2091aec3344448bf30 : SUCCESS in 35m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39d9db18fc0f4b65ab7a81cefd4478bc : SUCCESS in 1h 49m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/03385e415cc84528a51a7494e9edb4fc : SUCCESS in 52m 56s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1f75d363a71d474ab1f132c425883200 : SUCCESS in 36m 25s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3ff4aa9eec5748d3b455cb7407471cc0 : SUCCESS in 1h 02m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4f9f8d1b449e43c09c1f987bdf87efbe : SUCCESS in 57m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/997462b1a2c944cb97977a0fe8567230 : SUCCESS in 1h 24m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/2b67a4adbbf341a9bf521871ac13978a : SUCCESS in 1h 51m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/fda64944d8cc439ab3ce6dc88a146b56 : SUCCESS in 33m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/31ea7e5883ff4c5aa7f0b38b9567b8cb : SUCCESS in 3m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b66ef2cc82e44f36a528c3616d604fce : FAILURE in 25m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a32b827f75444946b8b2cc96748d1037 : FAILURE in 26m 48s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/03d1037753e74c62957b426f7ad32e9c : FAILURE in 27m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe71c39ace844505b002ea239afac596 : SUCCESS in 1h 34m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b09d0310ac6f48a6b52c2228069ee91d : SUCCESS in 1h 07m 02s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1222432553a045979e316c512493df1e : SUCCESS in 51m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7ca1afe7cc9f4bdab4b61757930ea07f : SUCCESS in 22m 58s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/315441abdcce4aa2a8d5d27e514c40c2 : FAILURE in 12m 59s","accounts_in_message":[],"_revision_number":11},{"id":"a01ffd3d0ab6a8d7669f701aaf37c0c56bb3b815","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 00:22:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e32cf91cc2b4744aef40e56ed46250e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f38f525ca533440fb2c5417d5fae5d4d : FAILURE in 21m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4cdd8a1238dd44d2a77bf90765dd6ffb : FAILURE in 21m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5daef3d951f746599d2af9fa7eccd80c : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"4c0d7f7f8b5a4e0057f1a514b60317d95ea2feec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-25 03:13:23.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"988e23771f7c064bc7f35024a92c8754d8771965","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-25 03:57:59.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-tw4dl/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"eae488d8b7c74c9861a3307d52d1daea87f0d25e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 04:12:20.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":12},{"id":"74bdf096b1e5c8037cca68c20df37542e712af64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 05:21:16.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/06ddad2783e74ddd9a66130c1d636c3d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ea79bce84114411cb8a5f5b4ee79f0ca : SUCCESS in 31m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7c165a5276b5456bbe0e423607b1806f : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cca41f477294a43a1ccb46b7eb697ee : SUCCESS in 59m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0bf14ccf66c493bb4b83bd44f8c0f38 : TIMED_OUT in 1h 41m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c232bb250894136b1f9e77230486970 : FAILURE in 7m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62e20726b079419699a42c05886465ca : SUCCESS in 9m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30d701125e244d5a8cd040c634bf8bb6 : SUCCESS in 14m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5e71dd480c842158958a0d13a00b7bb : SUCCESS in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2a9d640395d44c5a70a15cc8f5e29e9 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/571c01a0441c49b48568548f82e9d57a : SUCCESS in 8m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e0942f14734f4a7a9d6c5d70b09fbc2f : SUCCESS in 1h 11m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dffb4eaa794f49bab80e086df5ef11b8 : SUCCESS in 1h 27m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3840ad6a50844ff191a0a8a83952c7c5 : SUCCESS in 51m 42s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/b66e71cac46e41528b341fcc2e8e20a4 : SUCCESS in 34m 12s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a1350135fb6b48c4913271b123cacad8 : SUCCESS in 42m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3557db434ae74df49c9f3e6f434ef652 : SUCCESS in 51m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d80e7fb31ded42ccb8633e07813ab335 : SUCCESS in 1h 48m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5f05e732267426d831c74b88360038a : SUCCESS in 1h 49m 17s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8a97260e5dac4d16b00afe2b94f2153f : SUCCESS in 26m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fd1e5eb07adc4135b8128925b7fa401d : SUCCESS in 3m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f97044a1cb564ec89de675edfbeff2c1 : FAILURE in 26m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/caf9f758db514fe098c2c185a4b3e06d : FAILURE in 27m 00s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/fc5de1a0d198478a9591d75c134be37c : TIMED_OUT in 2h 01m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/13ca8ca96c454d4aa871f684bbaa6ce7 : SUCCESS in 1h 00m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f23984d7d1f84bd99a421dcbfef99b21 : SUCCESS in 52m 05s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/28bd1d77d7d44a4ea56550fef13f5957 : SUCCESS in 52m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/adcd16fff5d64fe084ec77127e229551 : SUCCESS in 31m 45s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8cffccd72f3841818477db674432bab1 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":12},{"id":"a9bc15cae7c236053e152a64280e500778f694bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 05:58:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b2b0d330c7f4b2daa1695a26973ac2d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4bd84ac44a3a4deab72865c11306b551 : SUCCESS in 24m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6cbd4d9934d48ecbd84d05869598fd8 : SUCCESS in 21m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2eb8dc0669984752a3c6af88d168b9ca : SUCCESS in 21m 18s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"ffd9cd21f91795a8000206991a1b44d0aea7ccb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-25 14:38:52.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"9b9d84f342ad78c9623a58d5e3394e93f7c6999e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 15:36:43.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":13},{"id":"8990fdf4a0e04032cca87f3f0f84a84f626cf44f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-25 16:51:47.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3fc8e5b895ed092d23ded677f6602386f1f57930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:01:11.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d7fa871d96d4b94827ab9b3ceeab32d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bc9f0087372c4f759b958356e9130d92 : SUCCESS in 1h 11m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/96b224b70d754718bea40c9bafa5cc3f : SUCCESS in 53m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d114cc0aefa4bca9004f68526eed6ee : SUCCESS in 1h 02m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/543d3e11a4c143c5807924c1d3a45d90 : TIMED_OUT in 1h 41m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acfbaf01f2134cb5a358d5915b624221 : SUCCESS in 9m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b514b85a0a3452eb8bde6698ca35236 : SUCCESS in 14m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1cece3b432184b62b1c8005e42ea5220 : SUCCESS in 12m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b05258d60a7b4ee8a7437b6b7c340e4c : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b813a8f410b489498cf370e7ba85b78 : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e4eebaf2454c64b14b0a79cfd34fcc : SUCCESS in 7m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/142ff5bcb2ca4ba0b57d953003a64bc5 : SUCCESS in 32m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d202af3c5a7d4799baab711875a2b98d : SUCCESS in 1h 03m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/31fabf5217b8414da3b89e6aee82f153 : SUCCESS in 51m 18s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/077f47e056a54f2889d73ce2c272039b : SUCCESS in 38m 15s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c47c3e64d9f34c2ca43cbbdff7c93519 : SUCCESS in 1h 00m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6b8cbfa322a6493bb6891e04e063d067 : SUCCESS in 51m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/28fb22e4e9d2449a8f7db1022295f356 : SUCCESS in 2h 07m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/5cf7e47167544856a471cc020bf6480f : SUCCESS in 1h 53m 35s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/eb485849cf194bcf87841500233916ce : SUCCESS in 45m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f06d1e4ead8b4457b76b309b4bea18d4 : SUCCESS in 4m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/60444700a41b4b9087379673e0094ae7 : TIMED_OUT in 1h 03m 16s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dbabc5b20dee418a8ada7aecd9f014ab : SUCCESS in 28m 38s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/db79ecff1fb14a718a9267d4f3dd7674 : SUCCESS in 33m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f000ed7857b4385acc98377034af11e : SUCCESS in 1h 16m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e19903b053b45538aba5a399d257f77 : SUCCESS in 58m 05s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/3f3cac0e643142b4a79436c177ccec6b : SUCCESS in 58m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/49b7a251c07748fa956fbc4fc8d25651 : SUCCESS in 37m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9f4481f8e4f2448488a2def417fc9147 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":13},{"id":"c64547ee99472cc270ff01970b9d530afae71fac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 19:23:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97fee7f107944f7ba7dd0bb429cdffd3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ab84383304c43b5ae030bde507ef7cb : SUCCESS in 26m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05b51286500d4a15bac4b1c83b8d032e : SUCCESS in 29m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b6e68063f34245358a7cafa264479ed1 : SUCCESS in 32m 02s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"99c93cf73dc5662218bd1842b264523b415cf4e6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-25 21:41:28.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ed68eb6413e1afe03d5105ef858bb4a7208918d4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 21:47:17.000000000","message":"Patch Set 13: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/bf2d2794b09b4130a257dd707b07b3d8\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6d70af92f151429e9eb42a2fb514dd9b : RETRY_LIMIT in 1m 14s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"74e23a5c1794f2f97e614cb45278868d6ba1fd01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 22:23:48.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5f37f97bbb4d8b83f7d7b3140e7ed5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99558878621c4c2eb5c60410a7e092eb : NODE_FAILURE Node(set) request ef9a4a2abaed4910b68bf2b5a70e62c3 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ce6dae79eb5429aad1e637b555da878 : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5efc51794ae04474aeb7c055eabd00d3 : SUCCESS in 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"23d6c6af1de0e11a9bc28e91810b6dcfb5ac6440","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 23:44:21.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2714867c6192417aa73440eec94c85f0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4215b7c9c06455284b99eeab4634df7 : SUCCESS in 1h 03m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d26960b7d9c04af38ed5dbb654cfbee7 : SUCCESS in 1h 30m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/069767b73ce54b5388bf8d153a4dcfd0 : SUCCESS in 45m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/373168f99c44407cb06417b6ae1958c2 : SUCCESS in 1h 08m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279c5357ea054f58ba4d76086a0866dc : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40596e7bf5334b458b35d226f3120d34 : SUCCESS in 14m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71c254a2564e40fbae805c05d69355a6 : SUCCESS in 10m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/adff3a854ed641b580b8072fcbd9c5a0 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb46bfb96cae49b98a16250a3f4f2580 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f5323aa1fec4834b290ef57db95d106 : SUCCESS in 8m 53s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a28bd1434e314163b87e02d6fa7423bd : SUCCESS in 31m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2fd2a36af1744eb194a5a9365c71a85f : SUCCESS in 1h 46m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f27fc59449a54eaf86496b80e89b836e : SUCCESS in 52m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1ae165b9fee445f49ebb5354e2003503 : SUCCESS in 34m 53s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/fac7960042394b9f98be3881f1e9ac78 : SUCCESS in 48m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6d791283a637449bbacb3af174cd1281 : SUCCESS in 55m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/97ca51be0bf0476aa92246acfa2b0628 : SUCCESS in 1h 50m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/dc3fadb7553a49ab8814532276131eae : SUCCESS in 1h 45m 41s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b2799d1fcef34086b21c9414541b0bae : SUCCESS in 49m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b930cc246bf9462b9c14125ab0258e6c : SUCCESS in 3m 47s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/127b680d6e134097830c291fe7899f76 : SUCCESS in 28m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/27e250d701de4a2b9381012c9777be77 : TIMED_OUT in 1h 01m 20s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/40cd540e2dbd4b749e1ebe3ed5e12af2 : SUCCESS in 31m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f72f89bffd0b4907805d25c7bf7c2c26 : SUCCESS in 56m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81503f0a754c42c6806a473341276c12 : SUCCESS in 33m 32s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f0d1dc3d834f4c639e8026e0a4c0b6e3 : SUCCESS in 29m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/22d27dd2a4164c2cbacc1b205b00906b : SUCCESS in 25m 53s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/bd63b0402c244e6e8aab74a10fa4ea5a : SUCCESS in 10m 50s","accounts_in_message":[],"_revision_number":13},{"id":"fc48a505b16ad6b24d88c7da6ad81ec728d97865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 16:24:32.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"c7bb5f7b27f489a456d296911c38bdcb1c8ccdb2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-08-26 16:28:35.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"2355f3b5b3ff8a63d0dfceead2a73d504b137a32","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 16:32:43.000000000","message":"Patch Set 14: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/69ff357fe25349b4a84a65361e9aee3b\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/5675c0020d5c4f8ea03a7670afbd51ab : SUCCESS in 7m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b0005cc74dcbb33fd4283c596e9177bb4793a455","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 16:57:22.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":15},{"id":"1e97ae7636a6f3488e3e063ac3f0c48916a53730","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 16:57:42.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e9fc89b00abde9d729e34e1b480e616a6d7d77bb","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 17:06:01.000000000","message":"Patch Set 15: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/bff712cc0799478f818c770f36efe80e\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/6ae8bb2ccd7a4fcd88c0a9619de5d6c6 : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9243900f2b0eebe0370fa5e5ebdb92c4a6e4107b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 17:22:00.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":14},{"id":"5e5f8a812af0b4e21887a5dc6a694c4274406a39","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 17:53:25.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":15},{"id":"4b2fc6c2e183e78603fc1a53a96edf76c5ca183e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 19:00:01.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0c66a00a42f4679b21c7aad71d626b2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/672f584ed4654be1a4145ef471d5b882 : SUCCESS in 1h 20m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fb1f962f097f4f5ab7317f58a15996d3 : SUCCESS in 1h 38m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67c7b4506a654d6685a41e63b6acbe00 : SUCCESS in 1h 02m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c39e0737f8af439aad5486ad323cc4cd : SUCCESS in 1h 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29cc62f7c1547c99c78dd39e9f1c88c : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a094330b878542baa60c85bbdb192ea3 : SUCCESS in 14m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27e87ad1589b4bfa9018e9e5fb63d4f6 : SUCCESS in 16m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f63bf3437f346e9be60f05590c9fdce : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a51b001ed43a4ec8bd4f3e1f873ab369 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c627751bd6a0490d8548e97fd2c2081f : SUCCESS in 8m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/53ab64e6dded448f84df7e2555177aff : SUCCESS in 39m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1659b71561044eeb757f28ac1519605 : SUCCESS in 1h 01m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7e0032e95d634c6b80cb3ec30d240029 : SUCCESS in 54m 13s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/76f6be09cb704531bf59e19dc472def1 : SUCCESS in 34m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/906e294f55dd442da4ed6aa6db136ca1 : FAILURE in 32m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/809da46278cb4b3e99350453e59229f7 : SUCCESS in 53m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3028c93392c6492ebe105b0c100c604e : SUCCESS in 1h 50m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/b6a669aac5434fa6ba19db05aec5b51e : SUCCESS in 57m 13s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6e64fa3ee1a741589fde66ce60c0a66a : SUCCESS in 44m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e13006f80927454d84fb41b40540abf6 : SUCCESS in 3m 10s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0ec629ff80154ba383724c9d976133ea : SUCCESS in 25m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/58899f210a1f4958836e41fe5a6d4e58 : SUCCESS in 27m 27s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4f9180d56cb242d2b2f2cdd7349c0e88 : SUCCESS in 34m 38s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e06435aa1e7a4da68783e96cc0ec84fd : SUCCESS in 1h 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f699426cb4154f2797e1064acc835f5c : SUCCESS in 37m 55s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5152b9ba1f4047fba340888dd78da817 : SUCCESS in 58m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3b2ef8c62e5d46b1ad52c0e36e914d4c : SUCCESS in 27m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/42407aead8e545d998b8132880cdf86e : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":15},{"id":"8f7061ce77f92dad229fa9c325a6d0574b712945","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-26 19:23:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"15df366895373e8acc091c5b78171c3b57c3dc34","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 19:32:03.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/77792347d088417e91a27ab23708c660\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ceb9a9d308134296a9076bde9f26e96f : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c61dd870cce00a680133a15656547c0722994940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 21:21:27.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718ab82438184f9d861f2bb43d785edd\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5bada4f5d88d4f4ba8de1fa4e616155a : SUCCESS in 1h 15m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1db0df22e564481086a280f0d48afb62 : SUCCESS in 1h 34m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1bacd95f8b2420099ac336baaba8172 : SUCCESS in 1h 04m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/944054e3fa2c46e7950880b3072be386 : SUCCESS in 57m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6e0a437e194095bd4aec1637c1001b : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50da66c9f7424e94a95d0a6babf21795 : SUCCESS in 14m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa3220bf01fb401694b988b4b3ccb04b : SUCCESS in 14m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/06b0af835b9f4e2cade39b62ef172bfd : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5428d8f6bc21407cb951f32523b2a6ff : SUCCESS in 14m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a92398eae0d442809eb7105f714e1599 : SUCCESS in 7m 17s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/193e960a6aa441eda0a9483884f8a18c : SUCCESS in 1h 05m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85eb4cefbc4f46a38d12f9d57af97fd5 : SUCCESS in 1h 26m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/534765650d26408088bc98b0603027f5 : SUCCESS in 26m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/bf14b99c98bc40a7a92f201de8fbe0ba : SUCCESS in 36m 42s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2a3c108d6648442daa4c7776220e654c : SUCCESS in 1h 05m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/29fd1843f6a24551a57883c6b74c33a5 : SUCCESS in 28m 18s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/afd5b0a1b1ec40edb9d368da75ea02cc : SUCCESS in 1h 51m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/561e482ae37a4b3694cc0bcdde24d26c : SUCCESS in 1h 45m 22s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/70a99bd1a8df42d8836be089b91f65c0 : SUCCESS in 26m 36s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ddb56d78871040f4a09f3c385c4f814f : SUCCESS in 5m 41s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fee1096ad1b04243a54c7f497e5a76da : SUCCESS in 23m 28s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/30202c51ff2d4832a69ba97856125009 : SUCCESS in 26m 43s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/88976fae16434d4db9c1db38bfc56c6d : SUCCESS in 30m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/425a78d005f94976978f6c6b94a56a62 : SUCCESS in 1h 35m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46c809cc1864462da72b27f6fa9bf393 : SUCCESS in 29m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9228c31df6d748dfad36a37976684521 : SUCCESS in 27m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c48e7514debd40908db56273634c962e : SUCCESS in 27m 37s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/78bc4bb6de1a45e29018c229c1434492 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":15},{"id":"7ee936ee89778aea1515e8049d64a071f04ef3e1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 21:32:07.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"f511606cde32ba925cd86a42f03674b02142a824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 22:17:59.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6235dafd7f474d90842059045c434883\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9096d7e4ac5454980f768dcdd981d4d : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03dd3f728dce4316aa8a236b7cebc247 : SUCCESS in 26m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/25e8e9a863ad4ea3a1608a8af07eeb18 : SUCCESS in 45m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a33e7b5c7e6e9dba8c2dfd6ac77b4bf734337b14","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-27 14:26:56.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"5ff0782a0ead94c54a4ff56ea2c8152aec183383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-08-27 15:58:38.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"1ef71952f5b01c058844788c94d281da7d39d31a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 16:04:16.000000000","message":"Patch Set 16: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6b518dc30f654957b85fe4b29b98104c\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c3a001d0d089412ea1197420d7e21687 : FAILURE in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4505a1239ed1cfeb99d05766d492e3a1196065eb","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-27 16:04:38.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f3a1c939879c3ce2c38c8d9c17d06e10ac1bda97","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-27 16:54:54.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":16},{"id":"ba7f0c48e8513bb2990495b85405e640594cc436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 18:02:13.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b7382460a6e4d108a466cecf54031bf\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ba44c6ae3c146c7a268ced53281388c : SUCCESS in 59m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7785691999a14798be33e11a7677a4c8 : SUCCESS in 47m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dbb933f558943458c8374697f8780ca : SUCCESS in 1h 10m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b2e7d3c135647c4b5b77f530da72d15 : TIMED_OUT in 1h 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258093668c764097b047de0792d2a9a4 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41f428b5a10744a397189d9ec85be6eb : SUCCESS in 8m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/19ec66e76e524209b4c3605f42b3a515 : SUCCESS in 14m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5235208191a24106a6f7a08cbcb7fa70 : SUCCESS in 16m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e2377cf17e4490bd6e7061790a0819 : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fcb01485df94c05849f37640e7fd966 : SUCCESS in 8m 43s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/dd50e3d3c10a4bfdaba7880ed8cd168a : SUCCESS in 28m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/858f39d46b624c93924eb877021dccf3 : SUCCESS in 43m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/85abca3f8a07421eb75a2e5592d0d9df : SUCCESS in 54m 20s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/149cc21f664941388309faffc0d1f923 : SUCCESS in 38m 55s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/fecd43dfb070484e973170656097165b : SUCCESS in 46m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9debdf0a62e44e2895a610a7e82c414c : SUCCESS in 53m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3dd366b9aef8484ba8192da16ccd72b9 : SUCCESS in 1h 52m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/fa3e52f92d4944079984e973b703810f : SUCCESS in 1h 57m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7ffa7e68968644caa76cf636545b0858 : SUCCESS in 52m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2cb57897e9cb4204b66873c41f6ef2a2 : SUCCESS in 5m 24s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7ba84bde4f7f425f8e38fd7c90616af1 : SUCCESS in 30m 47s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8370f6250e0e4bf9875cbe6e1ccdad1f : SUCCESS in 27m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/57955a9ecc324a5aa7379379b0415fa6 : SUCCESS in 29m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d27bcf498a06401db161606d2465d55f : SUCCESS in 1h 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bfc7234d05d402a8855e95a9a216231 : SUCCESS in 58m 55s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bfbc6e694ed74f8e83ce508377e4ec6e : SUCCESS in 25m 29s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4d68e40789e9462ba0dccfc5ecf92a78 : SUCCESS in 32m 42s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/211f7d716b9a4e84969216659db91173 : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":16},{"id":"34f541dcd0e6f40b0794bd12eef096b53d85f096","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-27 18:09:55.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"41de3fe7c5301b3678b9d911e7bc0ac254055213","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 18:16:49.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/11c7ef12390c4008a410401921c0e62d\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/75df6e60eeb8485ca75e2609b11c27b1 : FAILURE in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"81838211bbc9f2e3b8b7a466888919afc858bef1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 19:34:10.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1746a485560542f29cc8d35e444aa81e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4b33d591a2de48c3b80e04b119039eb0 : SUCCESS in 21m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bb63ee67b8f34eb0bfb461360ed20009 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/45e81bc1386d4e4488d4d035bbd5d184 : SUCCESS in 42m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bd5573559abaed654e1c4c31f98c59480463672b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:10:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fdd7e7af42444e4b40e5dc82a4b93ad\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/58d479deed7349a6b33b5eade0266a30 : SUCCESS in 1h 06m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c3699f7c14a242bfb5a296c37d6cbd79 : SUCCESS in 1h 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/efa4ccdfe0ca45079ea470b9eea743a6 : SUCCESS in 45m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e18d14cd05254975b4d23082405e4467 : SUCCESS in 52m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df53bcffec73445798526d49a894108d : SUCCESS in 6m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e63daa2a5524096b071544d3294ee22 : SUCCESS in 15m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b3e01d796624a1cad41466a0a5b36ea : SUCCESS in 12m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ed4ac31f6f914af0a5a19b17c2ef6728 : SUCCESS in 16m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affa6552eccf41b6bfd4bff0cb4b270d : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33871739bb31446fb8c61a80e8593de8 : SUCCESS in 6m 32s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7e6e69526da04908b5cd834768788592 : SUCCESS in 27m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f9b63e0c1d35498dae667203a516eb18 : SUCCESS in 1h 17m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/52f43a10361841eb81247e827ea9b898 : SUCCESS in 47m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5969e653cba9451ab6a5698689a97f9e : SUCCESS in 37m 44s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f62044a1ecde4972be70e565df38e0d9 : SUCCESS in 1h 00m 35s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f3ff919a1d1e42fcaacbd5b4ce7ba963 : SUCCESS in 23m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/aa51b4d7f25340f29ef8a23b47a44036 : SUCCESS in 1h 53m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/5f3506d21aaf4af9877d945868de9cb9 : SUCCESS in 1h 01m 44s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/113e32603ab0428b8e6a8e577506777c : SUCCESS in 39m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c77dc350f09d4a0b905aaf7d6d687af8 : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5605f17b208c46f2a10135305977810c : SUCCESS in 24m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e4d3aff7e1ee42e8a5a8f603becf2e3b : SUCCESS in 25m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9ad6a6b06d564a4c895439c3151eae0f : SUCCESS in 29m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ec106fe05f641b19aa89a1e7c0a669e : SUCCESS in 1h 13m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc553b6808034888961bbd6bdc01d803 : SUCCESS in 52m 49s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8a6e1f0859434d919cdf80b11f5fb9fb : SUCCESS in 24m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4455605e4ba54dc893197c02795165be : SUCCESS in 35m 54s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/46925c143d12478ebba9bcaf653d95c4 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":16},{"id":"071dca740769d9d387d52560a757b2df2a3e87e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 20:10:55.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"01a45b7784f7c34b1a305a9fa1fb06ae498279c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 20:21:00.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"c675b3989f4bd2bbdf10e743c45901726a11a7c4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 22:37:24.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-8cnxz/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"4d7f64a431e930caaf6087794d59a8ebb7ceb9bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 22:40:32.000000000","message":"Patch Set 16: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4950c962c4b47d0ae99d5a83807ac91\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3be69c012cfa40afb0742b23f0a4cd40 : SUCCESS in 1h 02m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/610999d55d684f3098e1ba70fdb7d43e : SUCCESS in 1h 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ef3aa093dc44096a28cf8e1e6768bb9 : SUCCESS in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6c9112ac804aef850d3915952811a6 : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b0e5d56e91a460cace779b504378396 : SUCCESS in 8m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0e3a3cd9e79c4a81b3b03aac30160042 : SUCCESS in 13m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aad516d77c6e40cf941078315d08bb01 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/153f78330c954ec98bddca9ebe5cc632 : SUCCESS in 8m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/efeb8655a30d44c7b9889301e0b0d886 : SUCCESS in 47m 32s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4fba3d6cfd7f4cc2b02df07e50bc151d : SUCCESS in 21m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ad9c5fd9c0754941b7b452bcea0c7d98 : SUCCESS in 27m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/06fe4660d9234106ad6afc3dfcddf7e2 : TIMED_OUT in 2h 01m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dfd61086e03240df972fe165159a83a1 : SUCCESS in 1h 49m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/9703d7838c344672aaf5648eb895eb4b : SUCCESS in 2h 21m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d176b45243cc466d97c531b520b27078 : SUCCESS in 3m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8c1c1763b80472ba212f1469ca7dcbc : SUCCESS in 1h 01m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/552f7e8798da46a596ab5d27d2ad726f : SUCCESS in 1h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a520a819e78547dba4a3184f841438ee : SUCCESS in 1h 07m 10s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/763faa9bc80148239f81f80a1f36f854 : SUCCESS in 12m 52s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4a9eb3d847f3443fa3acdb92adb5fae4 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":16},{"id":"bcbb898b397ebf1c905a5e93ca19fa6926105914","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 23:25:55.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-kbkpf/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"d4bfe786fa07c02954ca501ddfb53e25e0b536cc","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-27 23:58:32.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-rqmts/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"f26eee0418c6d1eeef93b8e926ab5d5b18b2a3be","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 00:29:37.000000000","message":"Patch Set 15:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-gh8z5/index.html\u003e","accounts_in_message":[],"_revision_number":15},{"id":"13b8bbcef5e345719f6c6d62b1541b5ee5ff9722","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 01:01:51.000000000","message":"Patch Set 15:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-p7wnk/index.html\u003e","accounts_in_message":[],"_revision_number":15},{"id":"3f04ef7042c1becc641866cb9430d1c75a5fd5d3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 01:33:41.000000000","message":"Patch Set 16:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-b7sts/index.html\u003e","accounts_in_message":[],"_revision_number":16},{"id":"dee9d6ce87bd393e6c2e4d9a7a31ccc87c376e28","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-28 06:50:19.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"b1bdb1b3158c1eb405a4829b12d94ada0e7596f6","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-28 06:56:22.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/53ed375fd147486093b60dfbc0f2e16f\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/284e6d6a21d44a258040ee3aa28f7a9b : FAILURE in 4m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"68b7991999a5eb4a83786f053193855d929a9fb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 07:30:21.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1edbfc2b5a584764a392e2d0919abb10\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f0cfb005a3747e1bb9d2f850ea8ba34 : NODE_FAILURE Node(set) request aed10cf522fe4ac09dedfadeee68f70c failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f7ecc4957fcd4021bb3f014a0d01b977 : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a7ba176b71e642a6a8a5533ef52576be : SUCCESS in 32m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"76e3022c2f95485873e5eeb68363764d3674f7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 09:08:25.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eda7b2f6c3dd49e197bc196ebdc86a72\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8567473cb3442dcb2c4d52fee01fab3 : SUCCESS in 35m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d9569d74c5584beda0d1ae620e031bbc : SUCCESS in 1h 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44fefba3fb9c4fbf80ae3724978e58ec : SUCCESS in 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd7e4fd1dc5c424580cdede2bd350f7f : SUCCESS in 1h 02m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c228edf1b4e440b2af62ad53ae262d2d : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36c83e7ab76247ef852a5933aad58aa8 : SUCCESS in 10m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fee80802ddb748ee82d7014ea139394d : SUCCESS in 10m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e37b1abfc47d4a41866ad61280a64865 : SUCCESS in 19m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04b8fe5f7224a41a19e548297b2ad15 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bacef97fe2bb4690a31f8bf40053e4e3 : SUCCESS in 8m 01s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/29d881e0ac884d868c55bb0c6b068f25 : SUCCESS in 48m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec5b31e6d5db4e8985a55a6172d2d336 : SUCCESS in 1h 39m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/29bb29006c764be294eeaa4030dddbcb : SUCCESS in 28m 42s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1b96933441934a59a549dc0665ccce47 : SUCCESS in 37m 03s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/50100b95c4b2423697329c183e3f15e0 : SUCCESS in 40m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/821f063e24e046d2986f83829b0cbf5e : SUCCESS in 29m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2ac4703454d2446bb0ca8041321ca436 : SUCCESS in 55m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/613962db3bfc4c9b8afc679b5289c9ea : SUCCESS in 2h 11m 43s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c406d1263a2144ed8bf1d35c68ef7ae4 : SUCCESS in 26m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3226a44ca9774320819c25017ae173ac : SUCCESS in 5m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/35abcef7608d465c8be5adea3c4ea41d : SUCCESS in 29m 35s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d1d11c6e0e95457b8e7973a308c86ca3 : SUCCESS in 30m 19s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5093aebe592a4637ac02cfd951233fe0 : TIMED_OUT in 2h 00m 59s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cbb776ffcb345ea87a41a57f5e30ab6 : SUCCESS in 1h 13m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61912e1876904c048cfddaf9d4ed952a : SUCCESS in 31m 39s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b2d3d595d4fb466aa59b3a1315a61acb : SUCCESS in 54m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2d49010a655d428dabbc8bc302aa7a39 : SUCCESS in 34m 33s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8256e4bae45a44f78bbb795896f2df86 : SUCCESS in 9m 05s","accounts_in_message":[],"_revision_number":16},{"id":"0f6b844ca9c89abb3aab3c6ddc7789746ca1a83a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 20:08:46.000000000","message":"Patch Set 16:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-4jzbn/index.html\u003e","accounts_in_message":[],"_revision_number":16},{"id":"01d78180e1a4a7713a7f8a7d46733950ee337cfa","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-29 00:58:36.000000000","message":"Patch Set 16:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-c8fff/index.html\u003e","accounts_in_message":[],"_revision_number":16},{"id":"5891bddc0f9f048064ad99ae267575ed2d5fb44f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-29 12:58:49.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"aa4893719c58c1347ea90d5c45db1e63cd0d70b4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-29 13:04:51.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7db540470a1847e9b047532ca6073e01\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/cb4516be5cdb454582291f2e4d322b3f : FAILURE in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f58fed4ca052d24ce8b2391ab13ab6deb5e0cd88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 13:21:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c955a86a18b14177a59db4cf88ebcf2f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dda28737008643889560f747e2d76121 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/381b42d5f57543dc9aa762d2648e5fbc : NODE_FAILURE Node(set) request 57807cbd963b416ab124dc21980e062a failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b67a197da1f945ad94ff2f2d3046710e : SUCCESS in 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"cf36de3509fbddb65a721d80b66c9967e3912349","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-29 13:28:44.000000000","message":"Patch Set 16:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-k29th/index.html\u003e","accounts_in_message":[],"_revision_number":16},{"id":"5946bedb79a80d60284ff67773c65e274b7e9758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-29 15:13:42.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b5a2530f3094d20ab20d2bc95bd5f7e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e60157ce53384457a36439fe982d0186 : SUCCESS in 33m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/299188259df4407e9f2863c64753191c : SUCCESS in 1h 34m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f55285b77b634a47a91a078e8c233e65 : SUCCESS in 56m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/207c5578d64a43f48783896efdab5388 : SUCCESS in 48m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bcfc805c1649a4ad9ce19c4b263963 : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32dd3115e69240e6ae8e052529dda88b : SUCCESS in 14m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cdce237174234a50803bd2cbffde77ff : SUCCESS in 9m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ad662c5706bb476994483ddf7a6c3bf8 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96f780937cf4f4e978e74dbba355dc7 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1de7fe6278fb42d3b3c63650d633bd97 : SUCCESS in 4m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f66f5e0f48a843bf8217354c7cf4ad6d : SUCCESS in 29m 35s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3f5f2088a964dfc853d3ee97be26497 : SUCCESS in 1h 21m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/909477ef36524650b33cbf819c97de62 : SUCCESS in 27m 00s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8a1d0fda40e5439f9a9bc173fa1761d8 : SUCCESS in 35m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f3e58398ba014edc8273cc185ce37c28 : SUCCESS in 47m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3fc99bef4c074732a5c1c89403bd7de5 : SUCCESS in 28m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2e754c2e0575432e8b67e8fdcb089fcc : SUCCESS in 2h 08m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/022a82e6c5fb47f88ef6e8cb0250accc : SUCCESS in 58m 26s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/cbbb183db29a4e589a8e0b58d7cb0662 : SUCCESS in 57m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/faea409f895f473ba071adbb100b3e51 : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a0a57d91fd774874be1af48c5a65a88e : SUCCESS in 29m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/371fe7eb58be4a64824d57f1ad8862fd : SUCCESS in 28m 47s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ad9fe21c677643b4b7d441134f71ddd4 : FAILURE in 31m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8efa08a6a76040c38977ee3af5194ef3 : SUCCESS in 50m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdc33a3664e54e60a9acf1f10902f746 : SUCCESS in 41m 12s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/28911a2d0923421cb66056afeefbdc05 : SUCCESS in 56m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9aee8002bb914350a5398f7d54b9f6ce : SUCCESS in 32m 30s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c8aa8d26b14e4cd9a76e33ee8eb81913 : SUCCESS in 9m 00s","accounts_in_message":[],"_revision_number":16},{"id":"f89bf122924f2344a53edd00578aa7b2ddc61078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"real_author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-31 08:59:44.000000000","message":"Patch Set 17: Patch Set 16 was rebased on behalf of \u003cGERRIT_ACCOUNT_14228\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"}],"_revision_number":17},{"id":"91038718203c428908a4e62766578ccb8ef67c00","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-31 09:01:02.000000000","message":"Patch Set 17: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7ee716eba8781edd46a5b6754219727a1e5f5d3b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-31 09:05:47.000000000","message":"Patch Set 17: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/650998a581e640ce8d4269807b4c2e61\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/9bc02dc12dfe44229bf156584af07a69 : FAILURE in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"40ee3cb04b2fa2fde6d15225a5ffb291b456ce09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 09:52:35.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32118f3113c443b482471f503c16cf5f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8509efdcdf1040f5835f0b7db64ba91d : SUCCESS in 24m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eeb12d4f6f4945f18f0728190eb5d799 : NODE_FAILURE Node(set) request 7fb0a009a505469790cb3d93d1464357 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2f53989f63974ca0b8d489e5d5592c60 : SUCCESS in 39m 07s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"454a70a3563e82821604a313e700362061acf42a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-31 09:52:55.000000000","message":"Patch Set 17:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I36163947a09b82cfc40ab4b522d82ffb503a0b84 .","accounts_in_message":[],"_revision_number":17},{"id":"dd2b4f9f7da527079db7320352b5c21a3c0d0d2e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-31 09:55:51.000000000","message":"Patch Set 17:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-v6tbh/index.html\u003e","accounts_in_message":[],"_revision_number":17},{"id":"128342f54d020140db1fa3bd71cbb10ff753b981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-31 11:12:54.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/593b35bcf79a4008b312e07876bd3f47\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e310e628414448908236d04da8d2b5e7 : SUCCESS in 1h 09m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b82d28fe3ee14ed6907883bee2937b34 : SUCCESS in 1h 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b503b40630942a9bac790814716cf41 : SUCCESS in 1h 03m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8613b57948874baa84e77176bc5a61ba : TIMED_OUT in 1h 41m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e784a2c8630c420bb11a04dd2945815a : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3313989710ff4262b3ed3abc86b781da : SUCCESS in 11m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2abe342afba437e8a8cb0376e7e208a : SUCCESS in 17m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e83c6327cc54457d9d1ad8974a5f2343 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bcc34c001cf497298cff61453ce044a : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/572774b814b24e459bc9f2a0c0efeb5e : SUCCESS in 8m 49s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e2c3427afab547db9f374b1de4d71106 : SUCCESS in 1h 13m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc28b4cf64964d4598ee62eb66268fa2 : SUCCESS in 1h 51m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/acf7438caacc4985a374fa8d709e00d5 : SUCCESS in 24m 16s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ac98dacfc29040ce86af46a71255ea55 : SUCCESS in 36m 53s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/116f6a1e1f59417b944add116ffd3268 : SUCCESS in 59m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/289c8b8183c84dcd8c318b9978d8b42f : SUCCESS in 1h 00m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6199466745364f7bbc4ef83b2abf8144 : SUCCESS in 2h 00m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/f20ad94e9f194ddd98d747535b2e5d08 : SUCCESS in 2h 07m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/72b7fec81e944c24a6f005b75e00f3f8 : SUCCESS in 49m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d97dd9465bd843178b6bce95b80fc2e0 : SUCCESS in 5m 20s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f9a9de5f170e4eb181b01800b25c498f : SUCCESS in 28m 54s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9f10d8e3ac02418bb301badc02fad663 : SUCCESS in 30m 57s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/349c0e7c0db24867b153ae4c45ed9dfe : SUCCESS in 29m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdc7039602cc49219216cfe0f67bb602 : SUCCESS in 1h 24m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e44894631b0942e1bd66cb597037a2c0 : SUCCESS in 28m 59s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/27632053257a4479aa6e3f38c6eaf191 : SUCCESS in 56m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7e0ec0de95c742c9bb9eccdde415c92b : SUCCESS in 22m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/909a53b5f10b4f2c981089318acc739a : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":17},{"id":"fbf63c1978ae8e9e3b6bc5364f5acc4b50448e72","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-09-01 14:36:41.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"f4470270902ca2142a7833da8877ebed2b4b40d4","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 14:42:53.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/cd7d8554c67740f6b9f6a11d78ac4e06\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b102b7b90dc5430d83cea09b23bd8077 : FAILURE in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1081e9d90c86bc559eee7326c3a3d7b558b4c36f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 15:12:58.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa04cb2b5eb24c7ab9f5b895b97fbef8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ff3af84c7da4494ae16b2f9ce425cca : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8113c917b9354ec785d6572568b5b3ad : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2fada8d56bb6443f806ae432e067a912 : SUCCESS in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9b533940dcceebb4127eecb51d31b11094a4b3c4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 15:17:15.000000000","message":"Patch Set 17:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-gthdq/index.html\u003e","accounts_in_message":[],"_revision_number":17},{"id":"18e642b43147b6b2726871d5fafaf61c78ef762e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 16:57:01.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b5f1d4b911040d49376c10256db984d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ef5a0036ff440d7b1e78954d6f75b72 : SUCCESS in 1h 04m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8ffeadac26d84e01a7b39600b0215ef7 : SUCCESS in 1h 07m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/217f73ab35f342988061fae3885332ba : SUCCESS in 1h 07m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a858f3eef6e44319dab6680f07ef52c : TIMED_OUT in 1h 41m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a3283057214fcc856f3c425024242c : SUCCESS in 9m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c8d33ffe2e34891bace34d65cf59e26 : SUCCESS in 13m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e07eb35a149140dd8e0158c8396dae63 : SUCCESS in 13m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c53e9431ec294b7c96a528a0d45c9c9b : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ce4c36f62484f3e813f125424072083 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e13b68ebd8c04e388e81b6835fbcfc2e : SUCCESS in 3m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9c2dc346551d4ee588834ee6b3b46bf6 : SUCCESS in 33m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78fa40def5be4e2cb081f40da6d11109 : SUCCESS in 51m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ca254e68c1624f3abfe3e7ea25857571 : SUCCESS in 26m 43s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e34c6731c7a7484e83b4e2d54ced2b5a : SUCCESS in 39m 22s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/530fdc012b3b4c9ea3022881a82ebf83 : SUCCESS in 46m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/46e594e27fd34986bdc3895ef84e2934 : SUCCESS in 55m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1bef452efd30440ca51a28b9b27c1b12 : SUCCESS in 1h 55m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/725b056c647c408fa34f24e72c42abc5 : SUCCESS in 2h 14m 06s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5eea7aeeb5a64a30b98ab39b94634e0e : SUCCESS in 51m 59s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/500b25708e8a4ad7b796ae93b855b5ec : SUCCESS in 2m 55s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bce3d9754fee4b4abc4ab258b8cf1a4d : SUCCESS in 27m 47s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/639237df1f374901aab62c8e73b11d3a : SUCCESS in 28m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b27de3301c6244b195eacb781f5d83de : TIMED_OUT in 2h 00m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/afabf5b778694eec80634589a66cae4b : SUCCESS in 51m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d98d9af8d5f343b182e0c2d34b54b05a : SUCCESS in 1h 02m 48s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/617b2dd134ca4784a52d07d27adae5b1 : SUCCESS in 35m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f2e0fef9c2ac46bc94c760dcefa832e3 : SUCCESS in 32m 58s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/edcf00a532a341c6a7962fe0ddd06c2b : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":17},{"id":"405ff25ebcbf13645c798f1a2354be09b73a9fc3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-03 10:58:32.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3baa894cba7514e9b102d29122764a7a106adc7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 12:58:57.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fba79833a4cc45adb440fe0dfed51169\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6903e4ca91dd4e2a83cd8956a29feb1d : SUCCESS in 28m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/595e3e9611b340fd9761aa2823f6aa37 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0da86c031daf440a83ba52a25cbc399f : SUCCESS in 19m 53s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"be2ceb7854af68478417e0fecd0db490768e7171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:23:39.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2a8a247f90f4c34ac8d254342c1c98f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a9a2b81b9484e8384b74b08bb4a884e : SUCCESS in 1h 09m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0964b9f63b584158bf726e066b0e03c9 : SUCCESS in 1h 32m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fc8b0c21bde40bfa01691f430ad6726 : SUCCESS in 58m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e683324c1ffc42f2b864620d2d67f094 : TIMED_OUT in 1h 42m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92e48414b53142ed900cde37df8b6f4e : SUCCESS in 9m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b693655231b74f2aa00416ad9703069d : SUCCESS in 15m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbc0d537b05943e691fae6d5baffb09b : SUCCESS in 12m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8da857972dcc4576bdd01ff101074f6a : SUCCESS in 18m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b42c91e7bc4eeaa381fdd3d3646f46 : SUCCESS in 13m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/250b2d78387b42c9bf665a5b03b2d0b8 : SUCCESS in 9m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/fe87a53ec94048eb8df781d85c19173f : SUCCESS in 50m 13s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ef7ed5e0d844e548b4a144658db804c : SUCCESS in 1h 38m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1fb446a5c7fe4be4b2a9799eae47d797 : SUCCESS in 57m 32s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/06253fb71452449082a68ddf0d9b93f3 : SUCCESS in 38m 40s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a3ea5eb859154d80aaacf06584af823a : SUCCESS in 51m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/252dfd65342b4090b106e52f9ec569a6 : SUCCESS in 51m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/78b851fec67d4abd97bbe0d7657fc955 : SUCCESS in 1h 49m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/9b134ba4b8ca4e42a4924be50a1ba486 : SUCCESS in 55m 14s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/340b97054f8345c3acfeb0ecb62940b7 : SUCCESS in 51m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a7c77c6fad874c2f964072493a2d8706 : SUCCESS in 5m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f5f33236e1d54fe5902a37f398d533f1 : SUCCESS in 28m 29s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cb1866d3194e46a09a0af252c50bbccb : SUCCESS in 28m 26s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b35f9ce3b76c4d608617028b3dcc1256 : SUCCESS in 29m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb8b810937574217b06799202dbeb09b : SUCCESS in 1h 32m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddecc47e6593434abebfc49700474a19 : SUCCESS in 54m 25s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec3f3ea35c1e4ff89af7320803c4214e : SUCCESS in 1h 00m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0375c75d14be4bf7ad72cdc373661152 : SUCCESS in 27m 13s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f591e92976264ae0bee4fc91a29d999a : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":17},{"id":"f773d06bbb8e11948a50a6d85c3cfc7c83144bf9","author":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"date":"2026-09-04 21:48:26.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"8c5f1d42293b0521cd9579fffe09d90b709dbca5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-04 22:19:11.000000000","message":"Patch Set 17:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-980679-7hqvr/index.html\u003e","accounts_in_message":[],"_revision_number":17},{"id":"2da2eed3490eb6bf5ca2cb5ccd86fe086cae2a06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 22:23:55.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ba67b1c4f8460f8ad63a550ed5c4d3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99badda1d1d64c33a03972c42eafdc4e : SUCCESS in 14m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2a61f272a2af42b6a0f42958dce58603 : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/34c7cb0e2a7e4e94a3a921b87cd4facb : SUCCESS in 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"afddd109725242371db606f0c0c493ceaebf5b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 23:56:20.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6f46e428a424b8499b5e53caa257f3a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/332daac498584d43a8891720033a8c60 : SUCCESS in 54m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/08b6767498a244339ca8a3c8841c810e : SUCCESS in 1h 36m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1431bd715184fb59ca453347553ddd7 : SUCCESS in 50m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3374a074ae842539b87687c40798f95 : SUCCESS in 1h 02m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1b844cd9cd4a7fa97f43399a70704c : SUCCESS in 9m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4bfb8b7581f4e369698c05fb665b865 : SUCCESS in 13m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d1c5ad6d6524f8aa313879d8ecc7f68 : SUCCESS in 12m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d37cd1d336194abaa9fbb67385344a72 : SUCCESS in 16m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60952abbb0f485d8f22333f51b2b508 : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82743cc79dae44bbaeaed79ec87e3650 : SUCCESS in 9m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/798a583e21874fa29662fd1d7bb8ab41 : SUCCESS in 41m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/498da632aff8415fb912d2affd1b34ad : SUCCESS in 1h 36m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/051250887cd44b0a9082fff604b9e2f2 : SUCCESS in 51m 02s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/58cecd1665fc49ae9bb77d93b4ccae61 : SUCCESS in 39m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/680e7c076451440ba110c2bcfec6fa9e : SUCCESS in 1h 08m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b74a262ecc394418b847808656415341 : SUCCESS in 49m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a622a6417afb48e7a6f41111c1d14eda : SUCCESS in 1h 29m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/b0cefd10724641f6a6ef69b4757b175b : SUCCESS in 2h 00m 35s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b52fba3f9f3443b6b8608cc9540b432e : SUCCESS in 40m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f298fb8cca824dc98c182d0c29c0dd60 : SUCCESS in 2m 55s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/09f2fe6a982548aebe8d025ca51ebd2f : SUCCESS in 36m 28s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7605fa7cd84f45dfb943c1562acfaa93 : SUCCESS in 30m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/12df5da7f3f54f81a1798ece5e96b060 : TIMED_OUT in 2h 01m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ed095b614034642847b4e220a73f65e : FAILURE in 1h 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68e77e6890c645558e90c6f84517dcba : SUCCESS in 37m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a7f0aa144a0c474ba4a1a3be9fdf1f21 : SUCCESS in 24m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/792e847247124134929e758c12811f26 : SUCCESS in 34m 21s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fce36dce35d9485986c41ba6377817f4 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":17},{"id":"91d89f5989eeda32f07391f80301f3c72b813742","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-09-07 17:15:22.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"c665b778f77e73e289dd4286e4dcff52985fd750","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 17:49:36.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/006fcbcbedfa4ee880c6512a7afca415\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d57357989e8944af8847c7a38ce39f17 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ad6fea7eb27463bb335cd59e4300ba7 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/101a437cfa5745a1bf3bd48e5af00594 : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"2c226da35b9f7f2ae388ab345319dca02eb110fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:18:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/914dd8a366804fefa7975036cbb425d6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d167d7c25a849878de9eab861429b25 : SUCCESS in 1h 02m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a97fc51e9c9d4316b127e3f8e001154e : SUCCESS in 1h 56m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a61926a2f604b2892602fee2b81c157 : SUCCESS in 33m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/340a3092dec44c888fe97253a99cf71c : SUCCESS in 50m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056660c724be466b8eaf67b1b123df7b : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d409bbeebe442848eb0d576b3478454 : SUCCESS in 9m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/938de4e258d74685b0acc5e8d9dfd717 : SUCCESS in 10m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5d95878958254893a8f7adf4e75d64c8 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810e5cffb89b46a7b13ae8bc0651e364 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96cba891ce4140d4b538447954599bdb : SUCCESS in 8m 05s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a2b729c9234c41e18b3f29036b846bc3 : SUCCESS in 29m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9e53b0a392b4a7ebb36e27c2f05b3b5 : SUCCESS in 1h 16m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d32012926f494b85b8e9ba0927387475 : SUCCESS in 33m 27s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ac4a71055ed04d25988970dc3e7d1352 : SUCCESS in 38m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/92cd68bd1dfb4ea1bfe58cfe4c1c9d90 : SUCCESS in 50m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5c40326a9ebc49cda0c65b3a7bbc8eb5 : SUCCESS in 30m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f0bbd182f30b424c82554393747f4b14 : SUCCESS in 1h 52m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/bfd1efeec546478ca7b523c414001dfd : SUCCESS in 1h 52m 38s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2791ff16deb94ace9ed89f8821bbc174 : SUCCESS in 34m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a5dad581a7f94e7fa58c64f19e3d7362 : SUCCESS in 4m 52s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d87ab2698a7148d7a38c0b819ae5230d : SUCCESS in 27m 48s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/01d67a9e767b48558bdc1f7cc0602738 : SUCCESS in 26m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c3297e88e06d4dca89688e8cd757593c : SUCCESS in 31m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb0ad08a28634573963b7d8225d6ce62 : SUCCESS in 58m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d44d66db953c4c488019b3c208dfb210 : SUCCESS in 49m 32s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/edc5fe39e79042c08070385c10ed4f11 : SUCCESS in 55m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a38dbd89851845e78d0fb3366250f630 : FAILURE in 34m 37s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0fb201f1e89b45abb812316d1201238c : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":17},{"id":"7186e9f167181c4724eb76ba7e94e211b83f0a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 19:19:06.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"a9903d2005493c085e03c486f3253b3742cbd22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:05:46.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b115c2aed8d41ebad862dd27e643e27\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adf5e9ce1a794c15bb5cdaff814b7612 : SUCCESS in 1h 03m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d35a93adc3e74298acea7f92af4f7dc5 : SUCCESS in 1h 46m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85cfeaadd63045868189e7c469a3836e : SUCCESS in 1h 04m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0773a9416ea47028536de54d960bb15 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92df03fd8df54e71a3ab1f9285369b94 : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af2441a151bb4108ace1f9a2a7c82a90 : SUCCESS in 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0e21be261847e4b020606bfe42fe67 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d809d5466024a879b53ad1d1cbf745a : SUCCESS in 4m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b8e0211889664330a633fa3ec79d11c2 : SUCCESS in 28m 54s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b7db5e3d2c6e42e4bc41aaad1ae0e7a8 : SUCCESS in 22m 05s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/aa54b805cc6b480b9430b61536d87282 : SUCCESS in 28m 48s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d9b59762be6349cdaf74999dcfaff1f7 : SUCCESS in 27m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2a7c1252898e423f94c16c1c14a6edf3 : SUCCESS in 52m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/567c050f7a8b49aaa9e16a0109b61079 : SUCCESS in 1h 20m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3d903363fc81400e9054d1fb2bc562b1 : SUCCESS in 3m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4f1022763bc849ecbfcca91abeafade8 : SUCCESS in 47m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/651592595e7d4ea28f05ed72747da271 : SUCCESS in 1h 33m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41cc2d035209493c813a78f41a18a8f5 : SUCCESS in 58m 13s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/09ca255e5cb144d6a692176807b1f439 : SUCCESS in 10m 00s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8298bcaf134644a59b80a96c3b53f677 : SUCCESS in 50m 41s","accounts_in_message":[],"_revision_number":17},{"id":"1e1e3d0da2f855552a2bd4ea9132e815c2edb1a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:05:47.000000000","message":"Change has been successfully merged\n\n16 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":17},{"id":"3523878b08ef80b1b0d73d76ec85d18b639cd59e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 21:08:07.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac352cca6fd410d9bd10461df224e64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d05ab02f0d4bc2ae2bd945d8fdad84 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be878751f0d5440f888ac59592da9159 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":17},{"id":"7f24300b197df7e2b64a5b2265e179e73e40d9ae","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-09-08 09:08:33.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"65394c320127470660b933425990ccabfed487e5","revisions":{"82a836f74430d217ca05df41795bceee86a8517a":{"kind":"REWORK","_number":1,"created":"2026-03-16 10:37:08.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/980679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/1"}}},"commit":{"parents":[{"commit":"9aa6d98e2a480b82d8ea74a3d161bfb7c71b22b2","subject":"[WIP] perf(ironic): background _error_out_interrupted during startup (B4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9aa6d98e2a480b82d8ea74a3d161bfb7c71b22b2"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 10:35:52.000000000","tz":0},"subject":"[WIP] perf(compute): parallelise per-node update_available_resource loop (B2)","message":"[WIP] perf(compute): parallelise per-node update_available_resource loop (B2)\n\nThe update_available_resource periodic task iterates all managed nodes\nserially.  For the Ironic driver with N\u003d1000 nodes and ~2 s per node\nthis takes ~33 minutes — nearly the entire periodic interval.\n\nAdd [compute] update_resources_max_workers (default 1) which creates a\nfuturist thread-pool executor and fans out per-node updates through\nnova.utils.spawn_on().  When the option is 1 (default) the original\nsequential loop runs unchanged, preserving existing behaviour for all\nsingle-node drivers.\n\nThread-safety analysis:\n  - Each node\u0027s update reads/writes only its own ResourceTracker key\n    in the shared compute_nodes dict — safe under the GIL.\n  - SchedulerReportClient._provider_tree is protected by a per-tree\n    lock; the subtree() / update_from_provider_tree(scope_rp_uuid\u003d)\n    pattern from commit B6 ensures per-node isolation.\n  - All DB writes in _update_available_resource_for_node operate on\n    per-node ComputeNode objects so concurrent writes to different\n    nodes are independent.\n\nFor an Ironic deployment with max_workers\u003d32 and N\u003d1000 nodes the\nexpected wall-clock time drops from ~33 min to ~1 min.\n\nUnit test: test_update_available_resource_parallel asserts all nodes\nare processed when max_workers \u003e 1.\n\nRelated-Bug: startup performance bottleneck B2 (startup.md)\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82a836f74430d217ca05df41795bceee86a8517a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82a836f74430d217ca05df41795bceee86a8517a"}]},"branch":"refs/heads/master"},"82427231e66fbb9c78d343486386cc0a1c0eaa4b":{"kind":"REWORK","_number":2,"created":"2026-07-06 09:32:43.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/980679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/2"}}},"commit":{"parents":[{"commit":"de1fcccd967328064f51bc320e203cec6910f9e4","subject":"Reduce ProviderTree copy cost for Ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de1fcccd967328064f51bc320e203cec6910f9e4"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-07-03 17:48:50.000000000","tz":60},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82427231e66fbb9c78d343486386cc0a1c0eaa4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/82427231e66fbb9c78d343486386cc0a1c0eaa4b"}]},"branch":"refs/heads/master"},"c2806dd895846150edbd2d38363800be89d3f7a6":{"kind":"REWORK","_number":3,"created":"2026-07-06 12:50:55.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/980679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/3"}}},"commit":{"parents":[{"commit":"de1fcccd967328064f51bc320e203cec6910f9e4","subject":"Reduce ProviderTree copy cost for Ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de1fcccd967328064f51bc320e203cec6910f9e4"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-07-06 12:50:21.000000000","tz":60},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2806dd895846150edbd2d38363800be89d3f7a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2806dd895846150edbd2d38363800be89d3f7a6"}]},"branch":"refs/heads/master"},"98f74e88a213ff7f0110b253b2286d3e2c740748":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-07-13 20:08:28.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/980679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/4"}}},"commit":{"parents":[{"commit":"88493bec306f09d61a02d61aa7247a13c0e19a11","subject":"Reduce ProviderTree copy cost for Ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/88493bec306f09d61a02d61aa7247a13c0e19a11"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2026-07-13 20:08:28.000000000","tz":0},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/98f74e88a213ff7f0110b253b2286d3e2c740748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/98f74e88a213ff7f0110b253b2286d3e2c740748"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"de1fcccd967328064f51bc320e203cec6910f9e4","ours":"c2806dd895846150edbd2d38363800be89d3f7a6","theirs":"88493bec306f09d61a02d61aa7247a13c0e19a11","merge_strategy":"recursive","contains_conflicts":false}},"b3ef7801533c9872a2b3b95a02830b2f88f8f1bb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-05 15:15:16.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/5"}}},"commit":{"parents":[{"commit":"706e5823f311805b77d8f358dab3d5bb3d8eb668","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup (B6)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/706e5823f311805b77d8f358dab3d5bb3d8eb668"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-05 15:12:21.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3ef7801533c9872a2b3b95a02830b2f88f8f1bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3ef7801533c9872a2b3b95a02830b2f88f8f1bb"}]},"branch":"refs/heads/master"},"aff964584e20f3ab4d9296a745aff759ee96ded1":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2026-08-17 20:01:42.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/6"}}},"commit":{"parents":[{"commit":"f6265b52a89bfbabfce8177c020a574c325ffc5e","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6265b52a89bfbabfce8177c020a574c325ffc5e"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-17 19:58:18.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aff964584e20f3ab4d9296a745aff759ee96ded1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aff964584e20f3ab4d9296a745aff759ee96ded1"}]},"branch":"refs/heads/master"},"0340d6ef9cca2b2e0412e525a41f753ec6085bbd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-18 17:21:36.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/7"}}},"commit":{"parents":[{"commit":"375be0325f816220ef591fae04030961d37df86a","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/375be0325f816220ef591fae04030961d37df86a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-18 17:21:26.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0340d6ef9cca2b2e0412e525a41f753ec6085bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0340d6ef9cca2b2e0412e525a41f753ec6085bbd"}]},"branch":"refs/heads/master"},"18d30ac601f2b99b5b6b4c37bd6b8d2413f2b73b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-08-21 13:51:30.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/8"}}},"commit":{"parents":[{"commit":"1508188a844eaf1ae4718727d65f5a5e6ee40766","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1508188a844eaf1ae4718727d65f5a5e6ee40766"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-21 13:51:16.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18d30ac601f2b99b5b6b4c37bd6b8d2413f2b73b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18d30ac601f2b99b5b6b4c37bd6b8d2413f2b73b"}]},"branch":"refs/heads/master"},"4499febfb9e980177e1acb98dd9a03a0b0a5efd3":{"kind":"REWORK","_number":9,"created":"2026-08-21 16:19:44.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/9"}}},"commit":{"parents":[{"commit":"1508188a844eaf1ae4718727d65f5a5e6ee40766","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1508188a844eaf1ae4718727d65f5a5e6ee40766"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-21 16:07:40.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4499febfb9e980177e1acb98dd9a03a0b0a5efd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4499febfb9e980177e1acb98dd9a03a0b0a5efd3"}]},"branch":"refs/heads/master"},"87ddabf71e1740a24da8bde087473327e79c7e88":{"kind":"REWORK","_number":10,"created":"2026-08-24 20:30:26.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/10"}}},"commit":{"parents":[{"commit":"7b18f5992cf0d019de1c37d972b2cd6dce54e60a","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b18f5992cf0d019de1c37d972b2cd6dce54e60a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-24 20:30:10.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87ddabf71e1740a24da8bde087473327e79c7e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87ddabf71e1740a24da8bde087473327e79c7e88"}]},"branch":"refs/heads/master"},"3cdd8e096b8c75cf1cf71d5431904466f06d790c":{"kind":"REWORK","_number":11,"created":"2026-08-24 21:37:15.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/11"}}},"commit":{"parents":[{"commit":"7b18f5992cf0d019de1c37d972b2cd6dce54e60a","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b18f5992cf0d019de1c37d972b2cd6dce54e60a"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-24 21:36:56.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cdd8e096b8c75cf1cf71d5431904466f06d790c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3cdd8e096b8c75cf1cf71d5431904466f06d790c"}]},"branch":"refs/heads/master"},"d5d97ccfdd8e0abac92fddb5138e8fb400522e18":{"kind":"REWORK","_number":12,"created":"2026-08-25 03:13:23.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/12"}}},"commit":{"parents":[{"commit":"2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-25 03:13:15.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5d97ccfdd8e0abac92fddb5138e8fb400522e18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5d97ccfdd8e0abac92fddb5138e8fb400522e18"}]},"branch":"refs/heads/master"},"2f243f4e44ae7787ecf553a19014d8c028b89c64":{"kind":"REWORK","_number":13,"created":"2026-08-25 14:38:52.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/13"}}},"commit":{"parents":[{"commit":"2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2dd126318e72bcb6e4f1d2c2e142d9ea0ccb9347"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-25 14:35:23.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f243f4e44ae7787ecf553a19014d8c028b89c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2f243f4e44ae7787ecf553a19014d8c028b89c64"}]},"branch":"refs/heads/master"},"1fff5155a96697d6c04e6846e76b8435941d6095":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-08-26 16:24:32.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/14"}}},"commit":{"parents":[{"commit":"853efa4c96982baf7385ce67ae779050eaecf98b","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/853efa4c96982baf7385ce67ae779050eaecf98b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-26 16:23:45.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed by a\ncompute service serially. This is acceptable for drivers that manage one or a\nsmall number of nodes, but it can make nova-compute startup and periodic\nresource reporting take many minutes for large Ironic deployments where a\nsingle compute service can manage hundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let operators\nfan out per-node updates through the compute service executor. The default\nkeeps the previous sequential behavior. Larger values allow Ironic deployments\nto trade additional nova-compute CPU usage and Placement API concurrency for\nlower wall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If one or\nmore per-node updates fail, the first exception is re-raised after all futures\nare drained so startup-fatal errors keep the same semantics as the sequential\npath.\n\nDocument the Ironic tuning guidance and add a release note that frames the\nstartup and resource-reporting improvements in this series as performance bug\nfixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fff5155a96697d6c04e6846e76b8435941d6095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fff5155a96697d6c04e6846e76b8435941d6095"}]},"branch":"refs/heads/master"},"818c264450048f4aebeb08966443e1d033e3ce73":{"kind":"REWORK","_number":15,"created":"2026-08-26 16:57:22.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/15"}}},"commit":{"parents":[{"commit":"853efa4c96982baf7385ce67ae779050eaecf98b","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/853efa4c96982baf7385ce67ae779050eaecf98b"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-26 16:57:12.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed\nby a compute service serially. This is acceptable for drivers that\nmanage one or a small number of nodes, but it can make nova-compute\nstartup and periodic resource reporting take many minutes for large\nIronic deployments where a single compute service can manage\nhundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let\noperators fan out per-node updates through the compute service\nexecutor. The default keeps the previous sequential behavior.\nLarger values allow Ironic deployments to trade additional\nnova-compute CPU usage and Placement API concurrency for lower\nwall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If\none or more per-node updates fail, the first exception is re-raised\nafter all futures are drained so startup-fatal errors keep the same\nsemantics as the sequential path.\n\nDocument the Ironic tuning guidance and add a release note that\nframes the startup and resource-reporting improvements in this\nseries as performance bug fixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/818c264450048f4aebeb08966443e1d033e3ce73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/818c264450048f4aebeb08966443e1d033e3ce73"}]},"branch":"refs/heads/master"},"6ec722f4a4fe2104dd521840930d8cb03d361f0a":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-08-27 15:58:38.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"ref":"refs/changes/79/980679/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/16"}}},"commit":{"parents":[{"commit":"318ca9ee790e5c39d9c4b342f2ae85d05c280990","subject":"perf(ironic): eliminate O(N²) ProviderTree deepcopy at startup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/318ca9ee790e5c39d9c4b342f2ae85d05c280990"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-27 15:58:27.000000000","tz":-300},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed\nby a compute service serially. This is acceptable for drivers that\nmanage one or a small number of nodes, but it can make nova-compute\nstartup and periodic resource reporting take many minutes for large\nIronic deployments where a single compute service can manage\nhundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let\noperators fan out per-node updates through the compute service\nexecutor. The default keeps the previous sequential behavior.\nLarger values allow Ironic deployments to trade additional\nnova-compute CPU usage and Placement API concurrency for lower\nwall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If\none or more per-node updates fail, the first exception is re-raised\nafter all futures are drained so startup-fatal errors keep the same\nsemantics as the sequential path.\n\nDocument the Ironic tuning guidance and add a release note that\nframes the startup and resource-reporting improvements in this\nseries as performance bug fixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ec722f4a4fe2104dd521840930d8cb03d361f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ec722f4a4fe2104dd521840930d8cb03d361f0a"}]},"branch":"refs/heads/master"},"65394c320127470660b933425990ccabfed487e5":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-08-31 08:59:44.000000000","uploader":{"_account_id":14228,"name":"Clif Houck","email":"me@clifhouck.com","username":"clif_h"},"real_uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/79/980679/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/79/980679/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/79/980679/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/79/980679/17"}}},"commit":{"parents":[{"commit":"445fc2db215e79810ee8ddec00939e88eb41fd03","subject":"Merge \"libvirt: Add functional tests for TDX capability report\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/445fc2db215e79810ee8ddec00939e88eb41fd03"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2026-03-16 00:19:10.000000000","tz":0},"committer":{"name":"Clif Houck","email":"me@clifhouck.com","date":"2026-08-31 08:59:44.000000000","tz":0},"subject":"Parallelize per-node resource updates","message":"Parallelize per-node resource updates\n\nThe update_available_resource periodic task updates every node managed\nby a compute service serially. This is acceptable for drivers that\nmanage one or a small number of nodes, but it can make nova-compute\nstartup and periodic resource reporting take many minutes for large\nIronic deployments where a single compute service can manage\nhundreds or thousands of bare metal nodes.\n\nAdd [compute] update_resources_max_workers, defaulting to 1, to let\noperators fan out per-node updates through the compute service\nexecutor. The default keeps the previous sequential behavior.\nLarger values allow Ironic deployments to trade additional\nnova-compute CPU usage and Placement API concurrency for lower\nwall-clock update latency.\n\nThe parallel path drains every scheduled future before returning. If\none or more per-node updates fail, the first exception is re-raised\nafter all futures are drained so startup-fatal errors keep the same\nsemantics as the sequential path.\n\nDocument the Ironic tuning guidance and add a release note that\nframes the startup and resource-reporting improvements in this\nseries as performance bug fixes for large Ironic deployments.\n\nRelated-Bug: 2162924\nGenerated-By: Cursor sonnet 4.6\nChange-Id: I36163947a09b82cfc40ab4b522d82ffb503a0b84\nSigned-off-by: Sean Mooney \u003cwork@seanmooney.info\u003e\nSigned-off-by: Clif Houck \u003cme@clifhouck.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65394c320127470660b933425990ccabfed487e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65394c320127470660b933425990ccabfed487e5"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"318ca9ee790e5c39d9c4b342f2ae85d05c280990","ours":"6ec722f4a4fe2104dd521840930d8cb03d361f0a","theirs":"445fc2db215e79810ee8ddec00939e88eb41fd03","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
