)]}'
{"id":"openstack%2Fnova~985718","triplet_id":"openstack%2Fnova~master~I108bf14e79deae0c656821e2d7c22bb5b7ed24b5","project":"openstack/nova","branch":"master","topic":"improve_attachment_resiliance_against_timeout","attention_set":{"37309":{"account":{"_account_id":37309,"name":"Bryan Fraschetti","display_name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","username":"bryanfraschetti"},"last_update":"2026-04-21 20:41:53.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I108bf14e79deae0c656821e2d7c22bb5b7ed24b5","subject":"add task state for attach volume and detach volume","status":"NEW","created":"2026-04-21 18:23:05.000000000","updated":"2026-08-05 22:05:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e1b0e23060e06d95da8194c7e934a564230a110a","_number":985718,"virtual_id_number":985718,"owner":{"_account_id":37309,"name":"Bryan Fraschetti","display_name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","username":"bryanfraschetti"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-05 21:22:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-21 19:04:07.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-21 19:45:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-21 20:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-05 22:05:25.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d2f9520f099eaad5f42055ca02393d2df0366f2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37309,"name":"Bryan Fraschetti","display_name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","username":"bryanfraschetti"},"date":"2026-04-21 18:23:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"536856b4427320e7589723f22c202f15366ad9b9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-04-21 19:04:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I108bf14e79deae0c656821e2d7c22bb5b7ed24b5 .","accounts_in_message":[],"_revision_number":1},{"id":"55cae61a7afb300083ca894b0acac8bd54c54ef4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 19:45:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e31cecf574f4e629211a57ff85f4afd\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94b8b52fdeed4c35986ad0f8c7f75c39 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bff22949340249bca34acc309a7d4d24 : SUCCESS in 13m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e08ad90e7ffa42a9b81fef5c2dd425f9 : SUCCESS in 20m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824e4113c2b9ae63ef6707ffda408923c30fe3fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 20:41:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b399826d5caf4b0cb6f9f552f0e66469\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8139612d2b1d403491c6a9c209e663c2 : SUCCESS in 36m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/72c252fff9644d849c36b8a000628e7b : SUCCESS in 1h 54m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf7ef0351c8f4ca2aff2a66fa87ce240 : SUCCESS in 34m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e01619ecc56349489cdc80b094fac150 : SUCCESS in 1h 00m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc6831bdc594fbf8935b520720b0097 : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f8d384feb964f4484007bb61775b07c : SUCCESS in 16m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11f1b7ece45d4fe687c2c0b06bd03466 : SUCCESS in 14m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ddba2dd556924964bc6deeaa94e131fd : SUCCESS in 25m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3b8db43416e43f1b9bb2a76ca027d81 : SUCCESS in 12m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/59296abc0478427999294c5bef816e2a : SUCCESS in 1h 07m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/32ddf3bcf8f34de5a5799e3fd312ab7e : SUCCESS in 1h 43m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/eb42bbaeefd74d9082132534f0561e14 : SUCCESS in 58m 24s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/8a5feb1e797848aa8460b94cc4e84dff : SUCCESS in 31m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1e0df9b493cf4efc84dec5f692185368 : SUCCESS in 1h 04m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e852fab0454b4e0ab70100a15393faaf : SUCCESS in 53m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7a7159c66815407a96c56b6fcb64513b : FAILURE in 2h 04m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/956eaccf0929481180669917a8310d1c : FAILURE in 2h 12m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ccbe169dc80b4a2fa3bbc7edd3fedc02 : SUCCESS in 35m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ca1ca974e1384ac59bff3ab4adde8776 : SUCCESS in 5m 30s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bb263760d2d2402f92b6693704e85370 : SUCCESS in 29m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/956197f174cd4901aa7e70eba0790a0a : SUCCESS in 25m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/47a4e328dd9741779f14866c2d521f00 : FAILURE in 1h 11m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff6d046c575c462aa104b0381d08e910 : SUCCESS in 57m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3deb41a1c2ae4f9ebd32a82bdbbec398 : SUCCESS in 14m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ec040e3f80154a359fef97120ebf19d7 : SUCCESS in 33m 02s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9fa730764e12436bb5abc2e01cc4f6d3 : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"09345c747158a3cf569efbb249632fcdc7cb74a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37309,"name":"Bryan Fraschetti","display_name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","username":"bryanfraschetti"},"date":"2026-08-05 19:20:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"62fae9f078f62135df2d318b406ced6378accea6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-05 19:24:48.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I108bf14e79deae0c656821e2d7c22bb5b7ed24b5 .","accounts_in_message":[],"_revision_number":2},{"id":"de3f32bf9e92a7d72c841882fba9ec691ffc18b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 19:43:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3801d8c5fe2a4890950317bda988cac6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4a4229108d84f8ea38d76c33ebd841e : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c0c3a7d3b77454b907ea67ed1c00415 : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ef9df611360443acbdac7fb7c56dc7b3 : SUCCESS in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e780e6d02a35c683cb3eda6fe2ae853aa79d76b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-05 21:22:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d3ef5b69ef6431b8a57294948250bb1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c2bfa129c494c80aeef8ccad8e2e64c : SUCCESS in 30m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7936496d39c34a1d8c57d707e70c3517 : SUCCESS in 1h 48m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9bfe979a62a74e26820829e421c2ad00 : FAILURE in 56m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32d0dce6a16549cfb1113ecbc07a0045 : SUCCESS in 1h 23m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0563db8b8634659ae32619e2b8bce5d : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fafcf8064c274e819e53aa92b98b256c : SUCCESS in 13m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e1998a7754d44cc973109c5790ffa44 : SUCCESS in 13m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b62d6c631b6b457db12e5126021364ad : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efac4fd7d5c9443c8f5b72d8eac07fa9 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f54e5f0ce24a04813bdab1bbd276fe : SUCCESS in 4m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4337c2d1dd8a40cd960ebca1538aa23e : SUCCESS in 1h 01m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/422faf2f552d4902af99760b26317f31 : SUCCESS in 1h 09m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/efdf9125ced146b38a1c85b0c389112a : SUCCESS in 49m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fdee7e741b45419cb6cc7960d2a2cba9 : SUCCESS in 45m 01s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/899ceba87ef34cba8cc800ced5e6d6f5 : SUCCESS in 1h 04m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a3aaaf10f2bb4442bac8cd8e3c01da19 : SUCCESS in 50m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b6540e73e06f4fe2ad14c30f87847b38 : SUCCESS in 1h 55m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/4680c8e33b6c438eb7de5f557d9b774f : SUCCESS in 1h 40m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b1d38e07fd78459c93943526f5b0d61a : SUCCESS in 37m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fea8e1b523884e2f9da03f1e232a3d0b : SUCCESS in 6m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93ccdce89b8e4e68a9bd6e9215664923 : SUCCESS in 28m 16s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0c2cacab16c943e0a534e6c8225279ac : SUCCESS in 29m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a878ace446bd4313be6d144700e0e792 : SUCCESS in 28m 43s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/95411084e69e4a15bab6caffc44d7f72 : SUCCESS in 1h 27m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7567d66173314f12883fbc8cf073add3 : SUCCESS in 50m 34s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/448bac0046ee434397ac7c148be751b2 : SUCCESS in 57m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c4c8c260f2c1494786fd1d29cb394c33 : SUCCESS in 19m 20s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4250f98cdbca401487efe80de57f68a0 : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e1b0e23060e06d95da8194c7e934a564230a110a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-05 22:05:25.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-985718-jvzdh/index.html\u003e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8947c25b7e63649a63ccd2dbc1c2ae46f8679cbe","revisions":{"b8a8e154f2f04df66542aa10cb1c90079704c184":{"kind":"REWORK","_number":1,"created":"2026-04-21 18:23:05.000000000","uploader":{"_account_id":37309,"name":"Bryan Fraschetti","display_name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","username":"bryanfraschetti"},"ref":"refs/changes/18/985718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/985718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/985718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/985718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/985718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/985718/1"}}},"commit":{"parents":[{"commit":"9ada1f6121cd510980bb3d7237cc21b86a2e4589","subject":"Feat: Add Command for Forceful BDM Deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ada1f6121cd510980bb3d7237cc21b86a2e4589"}]}],"author":{"name":"dengzhaosen","email":"dengzhaosen@inspur.com","date":"2020-09-25 03:29:28.000000000","tz":480},"committer":{"name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","date":"2026-04-21 18:18:36.000000000","tz":-240},"subject":"add task state for attach volume and detach volume","message":"add task state for attach volume and detach volume\n\nChange-Id: I5a18a9ca700d18d08f9f857c4e081cac59a132a7\n\nImplements: blueprint ensure-volume-attachment-successful\n\nRelated-Bug: #1897424\nChange-Id: I5b9b5d4842379f29ad5b57e8e5e938827c7bc57a\n\nAlways Clear Volume Attachment Task State\n\nThe instance\u0027s attachment state should be cleared (set to None) upon\ncompletion of the attempt to attach/detach, regardless of whether or not\nthe attempt was successful. Update unit and functional tests to match\nnew task_states and notifications\n\nChange-Id: I108bf14e79deae0c656821e2d7c22bb5b7ed24b5\nSigned-off-by: Bryan Fraschetti \u003cbryan.fraschetti@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8a8e154f2f04df66542aa10cb1c90079704c184"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8a8e154f2f04df66542aa10cb1c90079704c184"}]},"branch":"refs/heads/master"},"8947c25b7e63649a63ccd2dbc1c2ae46f8679cbe":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-08-05 19:20:51.000000000","uploader":{"_account_id":37309,"name":"Bryan Fraschetti","display_name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","username":"bryanfraschetti"},"ref":"refs/changes/18/985718/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/18/985718/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/985718/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/985718/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/18/985718/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/18/985718/2"}}},"commit":{"parents":[{"commit":"814a8ab1523f4037ffc9ba788719ee17bd1e5028","subject":"Feat: Add Command for Forceful BDM Deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/814a8ab1523f4037ffc9ba788719ee17bd1e5028"}]}],"author":{"name":"dengzhaosen","email":"dengzhaosen@inspur.com","date":"2020-09-25 03:29:28.000000000","tz":480},"committer":{"name":"Bryan Fraschetti","email":"bryan.fraschetti@canonical.com","date":"2026-08-05 19:11:23.000000000","tz":-240},"subject":"add task state for attach volume and detach volume","message":"add task state for attach volume and detach volume\n\nChange-Id: I5a18a9ca700d18d08f9f857c4e081cac59a132a7\n\nImplements: blueprint ensure-volume-attachment-successful\n\nRelated-Bug: #1897424\nChange-Id: I5b9b5d4842379f29ad5b57e8e5e938827c7bc57a\n\nAlways Clear Volume Attachment Task State\n\nThe instance\u0027s attachment state should be cleared (set to None) upon\ncompletion of the attempt to attach/detach, regardless of whether or not\nthe attempt was successful. Update unit and functional tests to match\nnew task_states and notifications\n\nChange-Id: I108bf14e79deae0c656821e2d7c22bb5b7ed24b5\nSigned-off-by: Bryan Fraschetti \u003cbryan.fraschetti@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8947c25b7e63649a63ccd2dbc1c2ae46f8679cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8947c25b7e63649a63ccd2dbc1c2ae46f8679cbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
