)]}'
{"id":"openstack%2Fnova~988777","triplet_id":"openstack%2Fnova~master~Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b","project":"openstack/nova","branch":"master","topic":"bug/2153425","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-06-23 13:46:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"38918":{"account":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"last_update":"2026-06-28 13:26:53.000000000","reason":"Change was submitted"},"38924":{"account":{"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},"last_update":"2026-06-28 13:26:53.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-06-28 13:26:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b","subject":"Add regression test for bug #2153425","status":"MERGED","created":"2026-05-15 14:33:00.000000000","updated":"2026-06-28 13:28:26.000000000","submitted":"2026-06-28 13:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988777-bug/2153425","meta_rev_id":"ca446a80579c88306f0ac54784a6ee4e10b1d4ed","_number":988777,"virtual_id_number":988777,"owner":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-28 13:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-06-23 13:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2026-06-28 11:01:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-06-28 11:01:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-15 15:35:11.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-15 15:44:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-15 15:44:55.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-05-18 02:26:31.000000000","updated_by":{"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},"reviewer":{"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},"state":"REVIEWER"},{"updated":"2026-06-04 08:13:36.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"CC"},{"updated":"2026-06-11 10:14:30.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-06-15 13:01:38.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-06-28 11:01:32.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"4b4ed1780d9f4adc12ac0dc0e5ad1410dcfef2f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-05-15 14:33:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3316b1334a922f1061b4a53eeaab7fff3814c9fa","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-15 15:35:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b .","accounts_in_message":[],"_revision_number":1},{"id":"b6a6418df8bef77e34cbecd542906d2324597483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 15:44:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/204aad6f091e49aa89a3ad15520f33b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c7728668d1b45a2beac6dcfc05690f8 : SUCCESS in 59m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af4d57225da458fa317da2fcf53f04c : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e8cd6751d3a4d6980f47c43d96cffc5 : SUCCESS in 14m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11bf720e52294a99ab5eab177acb00c7 : SUCCESS in 13m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe016560460540c7b75a82400c450dd2 : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4fca48e25444dda9dffce2d27ee8f0 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/271291a76c194dccbf9359b734b57f84 : SUCCESS in 16m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d3b7fc7509484a6bb1ae52ad5f5bd468 : SUCCESS in 1h 07m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8ed7038308a446e489d2a0c4bdd774d7 : SUCCESS in 6m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4c3fbc3eada3460db4563e3d52b9ca6a : SUCCESS in 26m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f86613eac4414883906b92b5d6832d1c : SUCCESS in 30m 07s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f34a3d13e6eb4df09607a9ab53023daa : FAILURE in 1h 08m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/01b86048d8be4c299a905c02105f7670 : SUCCESS in 14m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c6dc8799d9adde9efa5259f988baff758fd336d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-05-15 15:54:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"52429605824254d9fbc9209623a56da1b9c007b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-15 16:51:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7b8bc82789547a18c55eab1944effeb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a2b55df75eb48cb9f89bea2c54d0a0f : SUCCESS in 26m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/47d9da2cce9149c8bc7645633b245d49 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9ec7c9a9bf9c4978bcd48ee899954de4 : SUCCESS in 33m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd9a97be7e07cf730816d34c6be377191c71f99f","author":{"_account_id":38924,"name":"YushoYamaguchi","display_name":"YushoYamaguchi","email":"ys-yamaguchi@kddi.com","username":"YushoYamaguchi"},"date":"2026-05-18 02:26:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a09601c73092bcd877034b30f06b8b97610f211f","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-06-04 08:13:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6f56bdf02ce8daf6d5337d8a2205b6ba36cbee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-08 14:24:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c9417166d6724cef711dd06d11163fe235f5eeb1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-08 14:40:46.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b .","accounts_in_message":[],"_revision_number":2},{"id":"5fa3fbf9eec70fc0294ac4598be18f4441018d57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 16:31:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3641133ec6b641e0b38754fe2fa29403\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d5a0f642e04198a0b6968d1b8f5efd : SUCCESS in 1h 06m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/724ad6dc13f04ac685ea23c0424a0455 : SUCCESS in 9m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5de32e5cd0684994b76c840fdaa91454 : SUCCESS in 8m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/330ae23252df4882acc27304e8910066 : FAILURE in 15m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f2c5445520a4f5ab1f638575f572257 : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59591afa613e4315b11fca9178c111b7 : SUCCESS in 8m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c87f915c8864929b3a18ad3122dfc39 : SUCCESS in 8m 15s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/eabe19b91e644a64ad20608dc8de5779 : SUCCESS in 42m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/97bfef26147a404dabdef1d59433e680 : SUCCESS in 3m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/75c25aadd5aa439d8660c7ec4183d14a : SUCCESS in 27m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e8d0d47538a14386b9b78dcee045f1e0 : SUCCESS in 27m 21s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/cbcf15877f8e43498b0e130d9de73e3d : TIMED_OUT in 2h 01m 25s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e99876acecc64953b8f6d15718db5681 : SUCCESS in 12m 28s","accounts_in_message":[],"_revision_number":2},{"id":"3621a34a0e589af794ad82d3bfe18222cac164c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-08 19:20:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f09284f1086427ead46158bc03607f0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f66a1d341294dcd921c21100cfd55cc : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0feae4cb62e4427a9efc952dad207f1e : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7523525ce80c4326809c425008d2fcf8 : SUCCESS in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b01d5e6fb36ef063821011c1d8888a35736778e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-08 23:55:02.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7fe36bfff06509c8ca9aa06b62b9f2f8e5138741","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-09 00:11:59.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b .","accounts_in_message":[],"_revision_number":3},{"id":"f480a2211b0354c891ab8747fedbd06c21692176","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 01:15:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0006209044f7415d9e6ef25474dff403\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bae102586c7441ca273963726c23a49 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1403bc071c6a4b0a82813fe62d135430 : SUCCESS in 25m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c0c51ffad934149821ad4a93c9b978d : SUCCESS in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"31eef0bcfdc9585770a70e44611c1203d45ef0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-09 02:00:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c4871496b834b3d906ce830350d49f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab533b1ec1d442ab9846f8f1185998b9 : SUCCESS in 59m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8f78108ce064e1bba9b0e1b2f4d76e5 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0457ce2535114ff98e1991953bcf6e62 : SUCCESS in 8m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/233af4cb20c64a7b9a19b35ace19b937 : SUCCESS in 8m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6fdd46e287b14148b6fffa397c889a7f : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0113b448bd84f59a6b2e47bf69eab41 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c8755492ff84997beb261fdc2842a28 : SUCCESS in 16m 24s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/db6780dc80dc49b0accdd6fc0bad0c6a : SUCCESS in 53m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/382f563b63f74e2c80f5752920f5e891 : SUCCESS in 5m 07s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1516ef908a14487b9c4420ebccc68127 : SUCCESS in 28m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d77f7718c0e84e93ba9cffd995b20963 : SUCCESS in 25m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/96748a2c4d0b46218f5be91bae394cde : TIMED_OUT in 2h 01m 00s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/52027eb4fecf48c1933a49c2efcb9470 : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":3},{"id":"2940d523eb0bee3944dbbaec319e72182c6d0367","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-09 02:11:37.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"9c9643504ae0b368cf04e18ab7e19ee8a368940a","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-06-11 10:14:30.000000000","message":"Patch Set 3: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"145f42a691c395648968bf6636231cb5751b90ef","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-15 13:01:38.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"8f38cf2f617c50377a9393d5cbf43fcab3a05a26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-18 14:47:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"1126918f7d9afd14db922c599a0559a401303b83","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-18 14:55:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ab4128a8b8095baeea2c48a371807f6b298fba4c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-18 15:47:41.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b .","accounts_in_message":[],"_revision_number":4},{"id":"f76f725c90ad95696e6ceb41969b3a0c48b977ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 16:56:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c97d4e20ca54d79b48dd555b6cf32a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00e4419578ad4cf8b1f20f91e1c3ef21 : SUCCESS in 1h 23m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/484536f520784ca2ab4687bd74310517 : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ed13579a864483ba43eb20a8ba662d9 : SUCCESS in 16m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b6111e8beedd4ca4a7f6cf4105f92f50 : SUCCESS in 15m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16365a74b1934b2797e8ea9323c8a258 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0c3f1fe84a4687921a69ee5f01faf1 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f69fd4c58eb4a5ea916c5a1b86ff55b : SUCCESS in 9m 36s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6147f28a4205420e8ebbe1da3688bf10 : SUCCESS in 3m 40s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/deed59eea4f64f6e935a37a67f76386e : SUCCESS in 27m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5adc1a67082046699c2bfce748256c24 : SUCCESS in 26m 21s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b48aa38b6cc640e2bb30fdf884a3dcce : TIMED_OUT in 2h 01m 36s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b4fbff6dd30742b99a4d1ad33376d0b6 : SUCCESS in 15m 48s","accounts_in_message":[],"_revision_number":4},{"id":"041e8af667b549afcdc589b3e18b9a00e96b6df6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-18 17:07:17.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"8080b1b82980fd60e00817fe8655d3e5d39cca30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 18:24:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6708cd248c5b4859b96a235f65aaba49\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09f9b5c3d1d34ceca1249747f8a1729f : SUCCESS in 27m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9e989668a7846a186c2dad54c22b846 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e8c42ed8b664457b93ef826d55321a12 : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"18763a397c5d020d497ab656bbf28c443ffa2314","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-19 10:35:06.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"de6d2c99445b118afe3aff617a116bd74dcbbe92","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-22 06:57:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"575a828338974e1cc7c3a6ca5f0e287a5e6ad818","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-22 11:09:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dfd96ead59ab54284c2c980dd044116e8ebd8a79","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-22 13:44:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0b49ace679b90c129b2dc83c6ed1d9390c4fcd90","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-23 11:14:58.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e7ea752c98174e196cac8253e3bc809dc241b95a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-23 13:12:19.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2a52ee407174885d41be946033b8f3f4275517b8","author":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"date":"2026-06-23 13:14:49.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0c52773377767a0ba8125fa5f8bcae52f07e4c3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:38:33.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/066f92a510f84bf2b9811521b4e0e0aa\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8aa8c1985794edeab2b14fac432bf67 : SUCCESS in 22m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0a8d6fd7f034766a5fa8400cb131f7a : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c44e4885677643ac8d84a921704a8871 : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1187eee18affd5726ce597017a88a7233b464f04","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-23 13:46:39.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"70948a656e4472e6da6a91950544445b2d164cef","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-23 13:47:35.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"25b31e613c5ba34d463bddfbc3332d5aa84d6239","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-23 14:14:44.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b .","accounts_in_message":[],"_revision_number":5},{"id":"910605c15ad80999a37e73ce3a8c6dbe21dcecc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 14:45:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16d6436e67f14ed4a4cf1c3fdc571a74\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07e2f38acb974e63b172f85677676e67 : SUCCESS in 1h 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e14b465f844797a56f5e540654d6ef : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc442cc543314311a7b8bd7c9c4901ac : SUCCESS in 15m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/83d6393363a34d51a6150936d3efd1ed : SUCCESS in 14m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/026fe74f917d4b2298d5b857b4cbabae : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f1c4d20e264c7fa272c117c12cf29c : SUCCESS in 12m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83fb6778693147b4b5bc9eff7c55e0e7 : SUCCESS in 7m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d339701b02f0427ca01193b1cdf48e69 : SUCCESS in 4m 57s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b6037572d4de45dfb2440d48cd1e2d6b : SUCCESS in 26m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/07f8203e1ae24126af63da8fabee3050 : SUCCESS in 27m 19s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a2ce3be55cd740989b0582cbe14b27bb : FAILURE in 1h 14m 18s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9e3f5022aa1c45398d9a014dc86a8a39 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":5},{"id":"ec358dd80a1eec35c03d3271be35fb5a86f292c9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-06-23 14:56:55.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"6a4729da0a4f7f079028680b985edd66a208d914","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-28 11:01:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f37834e761c8a0d3b79062085232025fa832cafb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 11:02:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"dbf437b633440a7d42016c6cb789e42b8a5438d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 13:26:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c6abaafbec439687b809eb5823811c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138a6b677cc74f2aba33a8aebd92b1ca : SUCCESS in 7m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff0d85440d23410682f47ead693221c5 : SUCCESS in 9m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f31c7d8b233f4f21a615b1e9bad9aea9 : SUCCESS in 13m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4384872834a340189f458be7fd120815 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b49f012487584e53aa60387e06ae35fb : SUCCESS in 9m 27s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/665870170d6e437db473d5a76b183ab3 : SUCCESS in 28m 15s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0750ccca26224204a54cf0a9c42c3d52 : SUCCESS in 23m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b196856433d3459397164ef6baff5dd0 : FAILURE in 51m 56s (non-voting)\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b6522df1f59d4aab95dd449fa7a1d762 : SUCCESS in 3m 32s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5e5da02931334201ad89d9263ead8fa6 : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":5},{"id":"b5c282fa6cfb290c6b49d4827d8b6949a22d8711","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 13:26:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ca446a80579c88306f0ac54784a6ee4e10b1d4ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-28 13:28:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15c50d49f2e64eb0bda4cd5ea65a1e59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c46cc375045c42289ab62eda86d06ed3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0039911123c493af432d1cca31588cf901161698","revisions":{"1ed848f0468b9337e3973cf01115be6b57765c5a":{"kind":"REWORK","_number":1,"created":"2026-05-15 14:33:00.000000000","uploader":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"ref":"refs/changes/77/988777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/77/988777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/77/988777/1"}}},"commit":{"parents":[{"commit":"7184b3bae701f0c69f49a393c591fb27c027e3e8","subject":"Merge \"Extend the tempest coverage in nova-alt-config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7184b3bae701f0c69f49a393c591fb27c027e3e8"}]}],"author":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-05-15 14:20:39.000000000","tz":540},"committer":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-05-15 14:32:20.000000000","tz":540},"subject":"DNM: Add functional test for NUMA live migration overcommit bug","message":"DNM: Add functional test for NUMA live migration overcommit bug\n\nThis test simulates a scenario where NUMA live migration causes the\nper-cell CPU usage to exceed the limit configured by\ncpu_allocation_ratio.\n\nChange-Id: Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b\nSigned-off-by: Masanori Ueno \u003cms-ueno@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ed848f0468b9337e3973cf01115be6b57765c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ed848f0468b9337e3973cf01115be6b57765c5a"}]},"branch":"refs/heads/master"},"717a8763e17fd508d074d03300af771fa06e9823":{"kind":"REWORK","_number":2,"created":"2026-06-08 14:24:04.000000000","uploader":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"ref":"refs/changes/77/988777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/77/988777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/77/988777/2"}}},"commit":{"parents":[{"commit":"a80cf75414933d9e51c569375c70abd0f3511ba3","subject":"Merge \"Fix typo in log message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a80cf75414933d9e51c569375c70abd0f3511ba3"}]}],"author":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-05-15 14:20:39.000000000","tz":540},"committer":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-06-08 14:23:19.000000000","tz":540},"subject":"Add regression test for NUMA live migration overcommit bug","message":"Add regression test for NUMA live migration overcommit bug\n\nThis test simulates a scenario where NUMA live migration causes the\nper-cell CPU usage to exceed the limit configured by\ncpu_allocation_ratio.\n\nChange-Id: Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b\nSigned-off-by: Masanori Ueno \u003cms-ueno@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/717a8763e17fd508d074d03300af771fa06e9823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/717a8763e17fd508d074d03300af771fa06e9823"}]},"branch":"refs/heads/master"},"f7760fdb68678643733fe404a737409963d7301c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-08 23:55:02.000000000","uploader":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"ref":"refs/changes/77/988777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/77/988777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/77/988777/3"}}},"commit":{"parents":[{"commit":"a80cf75414933d9e51c569375c70abd0f3511ba3","subject":"Merge \"Fix typo in log message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a80cf75414933d9e51c569375c70abd0f3511ba3"}]}],"author":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-05-15 14:20:39.000000000","tz":540},"committer":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-06-08 23:53:45.000000000","tz":540},"subject":"Add regression test for bug #2153425","message":"Add regression test for bug #2153425\n\nThis test simulates a scenario where NUMA live migration causes the\nper-cell CPU usage to exceed the limit configured by\ncpu_allocation_ratio.\n\nChange-Id: Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b\nSigned-off-by: Masanori Ueno \u003cms-ueno@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7760fdb68678643733fe404a737409963d7301c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7760fdb68678643733fe404a737409963d7301c"}]},"branch":"refs/heads/master"},"162d57a0cc81d60a57622b696d3b76fbad3ee7b7":{"kind":"REWORK","_number":4,"created":"2026-06-18 14:47:03.000000000","uploader":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"ref":"refs/changes/77/988777/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/77/988777/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/77/988777/4"}}},"commit":{"parents":[{"commit":"a80cf75414933d9e51c569375c70abd0f3511ba3","subject":"Merge \"Fix typo in log message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a80cf75414933d9e51c569375c70abd0f3511ba3"}]}],"author":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-05-15 14:20:39.000000000","tz":540},"committer":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-06-18 14:41:37.000000000","tz":540},"subject":"Add regression test for bug #2153425","message":"Add regression test for bug #2153425\n\nThis test simulates a scenario where NUMA live migration causes the\nper-cell CPU usage to exceed the limit configured by\ncpu_allocation_ratio.\n\nChange-Id: Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b\nSigned-off-by: Masanori Ueno \u003cms-ueno@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/162d57a0cc81d60a57622b696d3b76fbad3ee7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/162d57a0cc81d60a57622b696d3b76fbad3ee7b7"}]},"branch":"refs/heads/master"},"0039911123c493af432d1cca31588cf901161698":{"kind":"REWORK","_number":5,"created":"2026-06-23 13:12:19.000000000","uploader":{"_account_id":38918,"name":"Masanori Ueno","display_name":"Masanori Ueno","email":"ms-ueno@kddi.com","username":"masan4444"},"ref":"refs/changes/77/988777/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/77/988777/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/77/988777/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/77/988777/5"}}},"commit":{"parents":[{"commit":"a80cf75414933d9e51c569375c70abd0f3511ba3","subject":"Merge \"Fix typo in log message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a80cf75414933d9e51c569375c70abd0f3511ba3"}]}],"author":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-05-15 14:20:39.000000000","tz":540},"committer":{"name":"Masanori Ueno","email":"ms-ueno@kddi.com","date":"2026-06-23 13:11:14.000000000","tz":540},"subject":"Add regression test for bug #2153425","message":"Add regression test for bug #2153425\n\nThis test simulates a scenario where NUMA live migration causes the\nper-cell CPU usage to exceed the limit configured by\ncpu_allocation_ratio.\n\nRelated-Bug: #2153425\nChange-Id: Ie6bcf827cadc67d0937f1bc3e100b2ce09a6e26b\nSigned-off-by: Masanori Ueno \u003cms-ueno@kddi.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0039911123c493af432d1cca31588cf901161698"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0039911123c493af432d1cca31588cf901161698"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
