)]}'
{"id":"openstack%2Fnova~989633","triplet_id":"openstack%2Fnova~master~I2cda35fa0c2e080a67418d557cb9030b11f31394","project":"openstack/nova","branch":"master","topic":"bp/virtiofs-cold-migration","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-09-04 22:26:25.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-06 02:29:27.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-09-06 02:29:27.000000000","reason":"Change was submitted"},"26471":{"account":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"last_update":"2026-09-06 02:29:27.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-09-06 02:29:27.000000000","reason":"Change was submitted"},"28329":{"account":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"last_update":"2026-09-06 02:29:27.000000000","reason":"Change was submitted"},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2026-09-06 02:29:27.000000000","reason":"Change was submitted"}},"hashtags":["virtiofs"],"change_id":"I2cda35fa0c2e080a67418d557cb9030b11f31394","subject":"Support cold migration with virtiofs shares","status":"MERGED","created":"2026-05-22 04:34:46.000000000","updated":"2026-09-06 02:40:57.000000000","submitted":"2026-09-06 02:29:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":110,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"989633-bp/virtiofs-cold-migration","meta_rev_id":"2df9bc38370ef1b0e4f9231d8b7222d31efdc203","_number":989633,"virtual_id_number":989633,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-06 02:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-09-04 22:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":2,"date":"2026-09-04 16:09:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2026-09-04 22:26:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"reviewers":{"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-22 05:30:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-22 07:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-05-22 09:02:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 17:15:17.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-20 20:49:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2026-07-20 20:49:27.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2026-07-30 20:30:48.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-17 23:34:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2026-08-18 01:37:31.000000000","updated_by":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"reviewer":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"state":"CC"},{"updated":"2026-08-26 04:40:59.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"},{"updated":"2026-08-28 09:22:25.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"87ad817e8b3f812dcd8725cd97153ed4798c2725","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 04:34:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db5dea3579a59fd82b98b8572313c6b412d1eb9b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-22 05:30:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":1},{"id":"e4b62bc92324b58ed7bd141b688824db28ffba17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 06:17:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab7963a8bfb827c2e8c292f175e1f71cc54886b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 06:19:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c87783fa53d22477ceb60390fe6ac31b05d35ddb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 07:14:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff3f3071a9c4ad39c85f6adf28f11c3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c772c256fd454b5b94d6d628d11b4533 : FAILURE in 30m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/23f8825d87244d348386a716bcbf1bba : FAILURE in 29m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/71c831caf3414e39bf20d94d2f1d73e2 : FAILURE in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6877c350b4dd7782d66e0e0e5cbbe31db39b0f35","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-22 07:15:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":2},{"id":"b90c83d19bc9ef4b9cfb69a93b0adc2c8297192c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-22 07:16:18.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":3},{"id":"b95dcb2b7769f2e11beae01aa64333ca26b8c48f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 09:02:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/387c87796b6d4c8fa2ea76f891ca60a4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d10ae933521a4d8aa19d3853455513d4 : SUCCESS in 1h 07m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3e2b83d861404cc490a61985c5062ca2 : TIMED_OUT in 2h 37m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/654bb52414bd483898f9f44308896bf5 : SUCCESS in 1h 01m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4ff97861d284bd28dabd3a2ccd641aa : FAILURE in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09ab1403c814599bddbd6145d40a883 : SUCCESS in 7m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2d3219b724e4bb6887a30ae8dbd591a : FAILURE in 15m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0d90bf3ba6542c29b071b6d350b759a : FAILURE in 13m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b7fd7b63889404d9725b08ec4bfc8cb : FAILURE in 10m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c86cd49d43a43d7bc7681dd7597ae4d : SUCCESS in 14m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67262a4e56714461bf4dbc1ea3394853 : SUCCESS in 16m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0cbef22eafd2425fbdfdc2c2bacba358 : SUCCESS in 1h 04m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a678bd057b44c50820cd8aa61fbc81e : SUCCESS in 1h 29m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b604bcda346247fc87e923d40511e02f : SUCCESS in 58m 34s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/ea2a89d9e1054918a3718bf3d1777819 : FAILURE in 1h 12m 09s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c3519d6cc84b4b41a573ff07c0d0a7f5 : SUCCESS in 1h 04m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8b5bf5805292416284984c0631e74364 : SUCCESS in 54m 57s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/576fd7ed5cb347e196897b62ea79004b : SUCCESS in 1h 01m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0a90a76762e3473ab24f1c4cbf304572 : FAILURE in 2h 25m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/4de8fb30875245068ee4cb2232633591 : SUCCESS in 1h 58m 58s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9fac1670b87b47e3939d4442882e9408 : SUCCESS in 1h 29m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f2d9f95162bf4c809e399ed69334efcc : SUCCESS in 4m 48s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/51a1c35a59984404ad6334a348be8397 : FAILURE in 29m 52s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d9978e8f9134469e98638b1107a6e5ca : FAILURE in 26m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f1ac95838f5a425d94098a58fc1cb376 : FAILURE in 47m 59s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9ec1e0bb00c44aeb955c99c2afd9496 : SUCCESS in 1h 50m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/220fc8c14c784bddb170e749b2591a7e : SUCCESS in 1h 04m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5feaf1eaec724576b411336bc59a00f9 : SUCCESS in 30m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e7da3e4de04247e283e3ab24ed7fa97c : SUCCESS in 32m 42s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/33c983f144f64c30817fdde7ca5b87b6 : FAILURE in 15m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3a158b36bf115a7ec0ecf2756494d54cbf04e207","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 19:52:15.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"cb08f843178115dcf96a178f2b66ddac8b847fc2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 19:52:43.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"9df753351db2abe169dcc719c3634611bee30d1b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 19:52:49.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"818f4628263ab726c7fd9978fa2741d1636cfd1f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-22 20:58:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"0b115a085cf0481a4bdf37fb92a94819b0d2a932","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-22 21:55:18.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":4},{"id":"2e8695df9dee8a1c3f3ae1d31b6e39224a5bc207","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 22:21:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2388d0405c4b4ad091d309e7ee167cc9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4346edfe4906435891462b109c731290 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a27b90cc449f438d8c7d88013dd586ec : SUCCESS in 26m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/117599166cfc46438914ea8acd93ee32 : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af6796747ef2594ccbc6892e570617c090fa62b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-22 22:54:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/912367d9234c40ccbeb9759079acf294\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f567297c816d4dc88b622d4076bc5649 : SUCCESS in 1h 09m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a0489e29fba24cd796ca7432cab71283 : SUCCESS in 1h 51m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7d21b16026e41bf89fe667bdd7106d2 : SUCCESS in 1h 24m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62e94c70f0254a40852138181005ec5a : FAILURE in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf22b5462d4840ea86c97976c9b61639 : SUCCESS in 8m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b19199830baa44a8b9295ff092d61629 : SUCCESS in 13m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d034ac1b9e0548fab8c1ef3dbe3345d0 : SUCCESS in 13m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d01060e0c3334795a5e768a078b19999 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00cd46344b874ad3aff2064b73371344 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d086dfe3ca4063a38ed577217cb4bd : SUCCESS in 14m 22s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6095aded97aa4eaabdc1e0546dc5a0c3 : SUCCESS in 1h 02m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09062aaf9abb46849ad2b7ef2b59ad34 : SUCCESS in 1h 50m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6f60710410aa48a0b7aa6dbf94dddc48 : SUCCESS in 56m 08s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/aa34252ac5c240469bb690296f5809e8 : SUCCESS in 33m 18s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f4682495f58b4f27b529df148fc8025f : SUCCESS in 1h 02m 27s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0f0ae5e72c0848a287c72333eefcb176 : SUCCESS in 52m 07s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/52eae082a2024ce687347b378d59e5e0 : SUCCESS in 1h 09m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/09fe1c0a28f94372ad07a7354fd5f695 : SUCCESS in 1h 08m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/8fc9590ceab34005b96ff631c51345c7 : SUCCESS in 1h 11m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6c1866307cf6442e96179f1e936f0355 : SUCCESS in 1h 09m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b0c1d32141a14baf8b7cd37ac16e49bb : SUCCESS in 6m 37s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/65e3008928de4dabb1baeed7a3118f98 : FAILURE in 20m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bc5242e43aaa4d28a57bd9fbe94645cf : FAILURE in 22m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/eed893b451f246afbd2a316ce6444c08 : FAILURE in 48m 59s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/38ae838e16554124ad0824600d20843c : SUCCESS in 1h 05m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a54ae06043b4db0beb9347444946df8 : SUCCESS in 42m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3932f14d0f924c4c9c4c3163d5e8308a : SUCCESS in 28m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e70bd46eb7a541ffbcbfec4f2738311b : SUCCESS in 32m 17s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b667dcb5ea8245bf8afe8f852ca9a4bd : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":4},{"id":"db3fe24224b6b96384af5263ba5baccfc2e0370f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-25 06:22:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"094d5b2682d9ecc6c1c24015923ca9ea7898eaf3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-25 06:22:43.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"397d40badd9fe4a04ec62a1d11a3b752625d5029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 07:06:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86dbfa4875634bee8f4ba8b020e469ac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d553ed01ef34169a7192e33743e7251 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8626bf4764f3473b91147b3b70e36bba : SUCCESS in 16m 39s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b224d29b0fef418c87e333d92e4488c7 : SUCCESS in 26m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f61e573c82cf9d94ac23b962b7f4deb0b386e70b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-25 07:19:40.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":5},{"id":"5add515bdd308cffe6f91c06d3c6e475795e3d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 08:56:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/baf50ede7fc5482ca35d49c97486628e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/29820c38908c4a68b067e86c162c5e44 : SUCCESS in 3m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ea97c2fce1744d6a44b14980600d531 : SUCCESS in 45m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/43c1781aa73a418b8dc8a742e7914065 : SUCCESS in 1h 02m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7085c86ccc234db794111abd418e8e22 : SUCCESS in 33m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7265cad09218462d8fc16bb32996d0c4 : FAILURE in 57m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a32bae24a24c6990f8023aa3bde821 : SUCCESS in 8m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/030f62fc28ac4167b326805952582a57 : SUCCESS in 10m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00feb324ae184971b192b049a41fefec : SUCCESS in 13m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b50c143b374b4006ae862dca4e91bf8e : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314103b45f59495da57f28872d02cccf : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22719e31944c41cb883500ca91368fe4 : SUCCESS in 12m 02s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/20e2c37823754e28bc0b33e578c33cc4 : SUCCESS in 1h 10m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/453225024f6b40e096f0665b93e30e80 : SUCCESS in 1h 18m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9995043d00d64e20b8bba9e1b3cc5224 : SUCCESS in 54m 45s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/02964ab326b94b95bbb3c59ca8768d39 : FAILURE in 32m 05s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/50e6fb75fe5446efbd817b4c4cacb00b : SUCCESS in 44m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ea26f0f7ec93440ba9fca8b616351002 : SUCCESS in 53m 56s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/31d660c6d1614650a9af5bceb40942f6 : SUCCESS in 1h 02m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ca67403299bf4a43bb65cd64d505919c : SUCCESS in 2h 26m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/a53ed464d05c4494a9134af0ba188deb : SUCCESS in 1h 14m 06s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0e8967732e1e4f7faff0dca4a31ed172 : SUCCESS in 1h 43m 45s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2d7169c03569495ab22ef65fd0a7a6cc : SUCCESS in 3m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ffacf92e191b49bba8134ff42f726181 : FAILURE in 26m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/da0f73b8212e4c9caccaee7cbc5654dc : FAILURE in 27m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b5d65c39c1a14ed985c2ba69f0f12520 : FAILURE in 54m 31s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/90f68676ec69422d9db7cbbc004dd845 : SUCCESS in 55m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c65cd9b2ef9643eb84fc609f07a0a8a2 : SUCCESS in 58m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b493a6412fd7482a92f24ad7da92e4b8 : SUCCESS in 17m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7747a0241519403eb47b2b835a9847e4 : SUCCESS in 32m 54s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a094f19170bf4ebdbdfbbebd5860dc45 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":5},{"id":"4fb279931fb9f04c216233735ca85a737610e2a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-25 20:04:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"e367969540894243e4372b7be5e6bbdfec2d5a7a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-25 20:04:26.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"6a2f5ee60c118cc4ca59aac0f6cb035112e3b953","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-25 21:02:20.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":6},{"id":"1b535ab67f95ad9025f45a2dca9fbc99b3b946cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 21:24:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db09bbe688d4e59a07c2e643e11a226\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/de10534b54654477b11728878e8851c1 : SUCCESS in 15m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3b1036d716554aa289bfbc375af97538 : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ecd0c23cd7cd43b689e08af3a265cf08 : SUCCESS in 32m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fce3965039e3be55f9d4a3364e56c5e2da0a153a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 22:36:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9078464f1e894e6a9b99d69d75f4f05f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ffb8da41a0b3434ab5e4a5cd47d87325 : SUCCESS in 6m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d726f389b98a47c782e45d5cecdfc24f : SUCCESS in 1h 06m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/575d8a5ad5284790a6a3c89c0ba33444 : SUCCESS in 1h 37m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/110b1d9304304402ab24aca869dfccd8 : SUCCESS in 52m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3a79e38b7dd46c0849168e09872482c : FAILURE in 55m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b702ec869f634553a6816d001455d456 : SUCCESS in 7m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2b93cecddd3466cba3ef4a68fecc2cb : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af5bb3f102bd41c2afa0b7b0f9d2472c : SUCCESS in 12m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3e5db09c36ea442ab2533eaa08190a76 : SUCCESS in 21m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf8224d619e940288bde7c151866e645 : SUCCESS in 16m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b256bb06a0f7417fb34ba136ec2c9baa : SUCCESS in 12m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3508ed8d232a4669a23796aafb60aa0c : SUCCESS in 46m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5872f729bc1c4b7eb98e81d3d70e3fb6 : SUCCESS in 1h 39m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7373047e7e814ed3b1884f215bcc6f9b : SUCCESS in 55m 43s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/9e99e224a3d349c0aca5855c8ca3c7ae : SUCCESS in 45m 41s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b48b145a09104fb78c0d50aeff3314e9 : SUCCESS in 52m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f261e3dd295d41cab2224b6625f2de64 : SUCCESS in 47m 43s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/0b24b3f695144e319da8e37943fbca96 : SUCCESS in 1h 03m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1fb8434110a3479c874a808a8bc54e61 : FAILURE in 1h 49m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/486ab1b91a084ed3ad959f707824babc : SUCCESS in 2h 25m 58s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d0008301c9d7499381c3fa770b62fe39 : SUCCESS in 1h 42m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3f7a5fc50a7e47ef8ee51692ce42bab2 : SUCCESS in 6m 21s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/14dfeb928093428cbb2d15f6f28b2e24 : FAILURE in 25m 34s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7e45140082a2482bb114eedf19bc37a0 : FAILURE in 28m 57s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1544ab56e651480d87412bfaa7c8bb6f : FAILURE in 1h 00m 27s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/325e2594c08d457aaadd49b250cc124f : SUCCESS in 1h 28m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e7f5d21d32d40b8ab2512cad8e4e55c : SUCCESS in 59m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1c6d46268c12402fb03f8dba516a213e : SUCCESS in 35m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/49e471567b4c4c7f9865c1394fb49abe : SUCCESS in 35m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8b8af2ef31dd493484c222b661c7b0db : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":6},{"id":"a0b2f6d6f032e7fa5650c36473c1d03dc6b22368","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-26 04:34:38.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"77bf7740490af21d0609178d6744c0cb498ddc65","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-26 04:34:50.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"2ef7583cf440dc8018a6725c7b51c842aa4c801f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-26 05:30:52.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":7},{"id":"2182ff62ece04d216d7cee0ba92a7fe14f74483d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 06:16:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d4f5fade9b9486688becb1a59382107\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb3bcbf77a3440f0b805399a4072a125 : SUCCESS in 13m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/980e4112ee884e33b9cece797ea12c2d : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/011477a4507847e6b2ce104f3f57f6a2 : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dd0e7b96be105999f997c30d5206f760773a430e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-26 06:29:00.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"3009434d41d1ad9d02500a772898c18f3b6c1096","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-05-26 07:25:48.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":8},{"id":"b4e7c4884179422a018afa2c636b97fc1ee774c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 08:29:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72cac471102d4f5097e114603a7ca12e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b7426360a3ab45ed88d9df3069ff7203 : SUCCESS in 6m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a61125910c844fc89af0027d0b1fbf3 : SUCCESS in 32m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a2cdf56697034b99871254d1bd6d7e47 : SUCCESS in 1h 57m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30a4c169a8c64ca98d6dc491b11b9195 : SUCCESS in 49m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80cbeec2f0fc4717a4a403b352fea1c2 : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3901f8513ea4b6ea6aade4901b78dd8 : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0072996b73e849858d3b03411796b6ee : SUCCESS in 14m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af5a4e46685f44b2aaa81b4f3a4779ab : SUCCESS in 13m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5998aace6bb64b299c96509a7ef00c4b : SUCCESS in 18m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb480777bc7c4a079a393dd3577ad5a7 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/641d8cc868854c4cbf0de10d7d7d58ce : SUCCESS in 9m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a83f6f60c2d3409b9bf79ed603720aa5 : SUCCESS in 1h 03m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8932e9128da34f8bbac86890bea5df7e : SUCCESS in 1h 07m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/881367d1b7d64abbb297af6653f0916a : SUCCESS in 56m 22s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/c347b6bc05204203b605ac2bbeffc28d : FAILURE in 30m 25s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5cc2511e0aeb480d8c2aaad6960079a7 : SUCCESS in 1h 04m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/80faf82d8f674615b9801887c661c640 : SUCCESS in 34m 21s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b37e80d1bd834a5185d0e6d923ea8467 : SUCCESS in 59m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/257336ecd41d4dfb833579eea6a732d0 : SUCCESS in 1h 35m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5bcbc1f308045ce8ece9cb8627bbf15 : FAILURE in 1h 26m 02s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/c7088e44a004486184b9ba69debc5b33 : SUCCESS in 1h 24m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b9114dcc06b34568be593d3199a1a082 : SUCCESS in 6m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/45619197b29a43fc9fc6c803d064c183 : SUCCESS in 26m 05s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9b9ba9329ea74101aca6dac19fe4c952 : SUCCESS in 26m 17s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f63c5e8ce07043fcb7e3b97f1e6def8b : FAILURE in 1h 06m 45s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fbb0112d5e8c497182a2bea69f2c27be : SUCCESS in 1h 00m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80a9074242e849039f6ad3a711821c3a : SUCCESS in 37m 12s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d031fb0712534bd3a26a36ec32721e77 : SUCCESS in 28m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cf12fbc1207340299d1b33e69372fc5e : SUCCESS in 32m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fcff12cfdd624cdbbc9bdb519e86fab6 : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":8},{"id":"24bf74594a9ef3954a3f11842cfc081986c8105f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 08:44:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574292b9361a4065a7b40be909bc0fed\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db92e14e940b4f02b8261b04b2b6b384 : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cecef59ac8a84be0bfa67033fe898cc3 : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b3bc9c03feec4b45ae4f31d6e11900cb : SUCCESS in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fa4bb55e31a7a323402c15197a2764091fdab883","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-05-26 14:47:10.000000000","message":"Patch Set 8:\n\nrechecknnthe vtpm/next job failures seem unrelated","accounts_in_message":[],"_revision_number":8},{"id":"fd2bc5ca0de0a2b514f9a46f0944e38f8e728e29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:15:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbad886cbc6e4dee9b31da235a04fec7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fc09b7b91de844d591a4cc5bd59dbffc : SUCCESS in 4m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8767906740d4442f8b984130de3c228b : SUCCESS in 58m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/763c72ff7fd34e83bb4a720109a5a6ef : SUCCESS in 2h 17m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/554325b2dddd40e3a0cc167ed7765790 : SUCCESS in 39m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fca3886d888e44ae8c3a1de51f9eae93 : SUCCESS in 53m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68feee78be464d25a55137e75c736370 : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9941b4ce1cf4ecdb115fc4400bc94b7 : SUCCESS in 11m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3b3f08030ed40148cdb1f4e85c2e093 : SUCCESS in 15m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d10d601edfd4b159bd920d4bb12e89a : SUCCESS in 17m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2454cf00f445c5bf8f0600629471b7 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b29e4303ba9045c582de64e8715e2d9e : SUCCESS in 13m 54s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/054aeb877f2a4b41896fc8221dc79ff2 : SUCCESS in 1h 08m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bddcef8687bf48edb577fd51eee88bde : SUCCESS in 1h 12m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/38f223b493b64e909a659ddd8a163060 : SUCCESS in 54m 35s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/9e58b435addd4c3ab88da3701dd25cbb : SUCCESS in 38m 53s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/7bd72815cecf496cb658a1a1a6511539 : SUCCESS in 1h 03m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e5b5e734259d4d01af255734f476f6c1 : SUCCESS in 58m 17s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9ee6674df6c84188bdb760199cf28718 : SUCCESS in 1h 00m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8b5d460c817f48c089e7982355551434 : SUCCESS in 1h 15m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/04f11cf9199e42778bfbec70536d062e : SUCCESS in 2h 12m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d64c7f12c24c477ba6041dea7afc73be : SUCCESS in 1h 34m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/69bb41abdee7479b9af4d54a1e86bd5f : SUCCESS in 5m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/338e726aad6c4bb7b3e021ddeec3982a : SUCCESS in 25m 38s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/043ca80d7de743a4927dd033a5ef6cc1 : SUCCESS in 32m 12s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d6fd85f2147644edbfaf6609ed4590a7 : FAILURE in 52m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bf3720ea5c84719b4e8ad2cb6686cc5 : SUCCESS in 1h 39m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef95eddcac8b4a85bb2ea3d065b8f318 : SUCCESS in 1h 08m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/11eea8011d854a969e191d437b98a133 : SUCCESS in 26m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ad3c329aa496421289d4e1a0d8a22bfd : SUCCESS in 28m 58s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a671d5057eeb441eaf31fff11bdb91cb : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":8},{"id":"00946325b4c55e461e40f728802fafb61e6152e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 17:34:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/722b9d4a75f441beb9ace2cfbaac8c6b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a19a12a208114fafb1883d7b4260d1fa : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81b95246bda94f3d8eaad73714128458 : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4307911375f744bfb322df8cc529e6b3 : SUCCESS in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"57645aef60b9f7157cf3f82a524f6a4227a7ea47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-19 18:53:50.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"52cc522bde6521998751f19b8fcae8f69124862a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 19:14:23.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693173a66f34412daac10d60912fdefb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60c2890231a74870a50c7f7b01bb4654 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd7608adb6f74c6f85b0bb93c2dfffd5 : SUCCESS in 12m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/40ec6bbc687a4079a3615425f87dbccf : SUCCESS in 17m 23s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"517f5a0e270bdcf199e3fb1f36cbb58a758d859a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-19 19:53:54.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":9},{"id":"efc6fc4b7fa45797a2ff8ec2187347549c027278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 21:09:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/247974f3416a4fc590e1410367b525d1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6e85f9cfc3cb4ed0a8511304f01dcbbf : SUCCESS in 3m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/467a962970d342e785d09e90cd691f8f : SUCCESS in 33m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/84e0d9baffb64a4080c344ebd71f1f4e : SUCCESS in 2h 07m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a2d68bab507408d93c78de778249d83 : SUCCESS in 58m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5630ca4799cb4b7981f61a7736dc3922 : SUCCESS in 52m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa06d65501ca4341afa4cef572e5ae9e : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dce17f1065e84ea5b9e86e7bf1adbcfe : SUCCESS in 8m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/592644403a8741fda8f04dfe7c08885d : SUCCESS in 13m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3cb3daf850a143bdab42f0920ff4ff5c : SUCCESS in 17m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d057cae9194482af3e8fe50ae7d099 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b53423448bc340d8b847dcb5b05d331f : SUCCESS in 8m 20s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e9ef976ee5a94ccdb5883b34e286df46 : SUCCESS in 1h 05m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c5c74a9be8974d73bea38a60f87b624c : SUCCESS in 52m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/310a34267d1c4599b0a0dcbbe8eb17a6 : SUCCESS in 28m 16s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/974bc06221cf4437bebcbf072c588a03 : SUCCESS in 39m 48s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/19231d36832d4e59827cc4cd7c055bc0 : SUCCESS in 1h 04m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/77c28d0d643947899b76afa038296523 : FAILURE in 49m 22s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/cd5ba19a12594e86a9b086e43d2aa9e4 : SUCCESS in 36m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a784551271f14e93aa3a1f2c7422e72e : SUCCESS in 1h 44m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/f8a4697c6dd34f138088016a149c4ba4 : SUCCESS in 1h 47m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b5230d5c1ed140fe9047e6b7df526e64 : SUCCESS in 49m 13s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c6a73b4fec604fd3bd1afedd17301705 : SUCCESS in 4m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/76fe8d0033ff47d29bbfa4c028e82432 : SUCCESS in 25m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1b4ed91ed1634fe2bbb5b1491a01a656 : SUCCESS in 26m 20s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0e9cf03471db481d81468f2e2cfd0987 : SUCCESS in 28m 03s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ad04e13b46c3455389f83d7c4a680dc8 : FAILURE in 53m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f018cddda07410f83c1b74da3fc1972 : SUCCESS in 31m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/51ca59323c88451da189d77cc67652d9 : SUCCESS in 48m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c147b1769c304fea892c3dd37616da05 : SUCCESS in 34m 52s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/81b3456957b7470796b1af7b550596da : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":9},{"id":"1808dca6647d2c8db6d5b8cce078bd25f7e482aa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-20 07:05:17.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0b4c9be20506314da78cb26e5716429d34cb0ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 07:37:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b45b437d80584b628275c617de325104\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8d85285d14a74821aff3a1d78901f509 : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94949c13cff043db8f932919767c7faa : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/19a47904f9c944fbb70db4eaf40d550b : SUCCESS in 20m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"072b8c15dddc577bfe43ecd1d9eb7113e056f98c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-20 08:58:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3521b2a35c2e43fabe9253cc317f47bc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aef24807b8554856a71d80eeef438e56 : SUCCESS in 5m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84a653e22cbb41f690dbf2317307659c : SUCCESS in 1h 09m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6bf18236091746358ef68e4513559d51 : SUCCESS in 1h 37m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42c9055903b940179a50c092c5a4adbc : SUCCESS in 1h 01m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b156c3b13dd54b15a3455e2a7a6d8983 : SUCCESS in 1h 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907a82e65a934b3da39036da0ce01dcd : SUCCESS in 7m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ae6a5a164294d48919664c055fdf725 : SUCCESS in 13m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1715c6e13ea34af688e79c072a6e600f : SUCCESS in 13m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6d6969c042a54276bb7adc323d6fd762 : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1e25c6f4c34441860162fb2c2a6f8c : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b373c8bd378b4ded90f295394a2c140e : SUCCESS in 9m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/18cfcef9e81e420a965fff2bb21a4178 : SUCCESS in 1h 01m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/321d4becaad843d18d434f1d1eaf6c1b : SUCCESS in 1h 29m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/13ef11b29b4745d6adf3823c44dfca98 : SUCCESS in 53m 13s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/912e5f9a50a34775ba78f4c2cc43c405 : SUCCESS in 40m 47s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f93e66a4f81e40039098425e42db4d58 : SUCCESS in 1h 03m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/19c266c215df4d6885ca4119914da4bf : SUCCESS in 49m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/692d199b0edc4cd6a9a07d6b2a087f5c : SUCCESS in 1h 02m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fcc5f61f5fbc488c91ba7dae51ef0cda : SUCCESS in 1h 38m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/7f2a2afa2261464586df71bbcdef71ab : SUCCESS in 1h 46m 33s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/106a1967ec684fd7b0cb6f9dc1747c9d : SUCCESS in 48m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/050671bed177461fb6edcce32ee94f4f : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f15480b807c04a5f9aab15fa8e0304ad : SUCCESS in 27m 16s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5ab66d6f20e3440f84a2fcb7b0b755f1 : SUCCESS in 24m 24s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/97dad3ed870c45f3957630d65a1fd263 : FAILURE in 36m 45s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd462ad71d924a31ac390279ced9c91d : SUCCESS in 1h 11m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d246461f5ba24262800c970837d18144 : SUCCESS in 56m 48s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2564ef9cabe345cdaec1fdb4cc5d2fbd : SUCCESS in 53m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1936f44c709249b2bcf5a8b42dd6d8f1 : SUCCESS in 35m 18s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/517d47a4ef3143d4915fbe0a1e4fb964 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":9},{"id":"94952ce2d60a8ed1a7e398205041c86b85d94538","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-20 09:08:52.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"0ad4a012943c2b7c1d7a932614cedcf1b8596cc5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-20 22:55:06.000000000","message":"Patch Set 9:\n\nRemoving WIP, ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"ab3fb5fde1ac10d26749ac6f8fee3979c17f013b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-20 23:14:20.000000000","message":"Patch Set 9:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"61ae69a8f3909a4a2a152536c48aea426dae103e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-26 07:40:53.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"c170d20305a6e1128dd368a0730ea57c5b79cd5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-26 08:10:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5574f649c9dc4c388fed82cba6b7c8ff\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae68d64fd1c947cebc65a8546ef214ef : SUCCESS in 21m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e7f168ca7a6540bbbf9613b8b32e40a5 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a3551a8e974b489b918cc94b83530431 : SUCCESS in 26m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"825624721d5af9597cbfebe903ae630f7b55e72f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-26 08:34:52.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":10},{"id":"84fccdc93dc15a0be19f6b62b89c2eada8d3909b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-26 09:45:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd95d59d43b043818b368732d4765a5d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e65778902ee64870814880d55429ce41 : SUCCESS in 5m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7ccaa417c6847f5ba991416f66a97b0 : SUCCESS in 1h 04m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/63ebde3ee2bb494394ba481acd0ce6ee : SUCCESS in 1h 34m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b34e5d6be96414fb58d5d9c784809e9 : SUCCESS in 1h 04m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12f1fddcd50f4bd787dcc715c22b96e1 : SUCCESS in 1h 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4211025bce184e5abb845456f4318e1e : SUCCESS in 8m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e25229cfeee400a9922bc7309a58ce8 : SUCCESS in 9m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cd13143cbd949d2b5192297c14e00f8 : SUCCESS in 13m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/02dadc0453b249ef88b9fbe7b139eea5 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716fda28129c480da4192d8cc9df5fdd : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/897ad63b8a1f497d8e4e37563feff93f : SUCCESS in 7m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a1ac5364a21946adb0ae4b58d415c495 : SUCCESS in 31m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/26381a88d82743099e3850172181a2d4 : SUCCESS in 1h 06m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/166a1a693a664649a3e6f027248017c6 : SUCCESS in 49m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/2da9cd4358fd489b8259067a4af9fd36 : SUCCESS in 35m 40s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/90e528956bbb4fcb872cacd00a984f08 : SUCCESS in 59m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/74b26716b1a547d08a13a3cd087bac63 : SUCCESS in 24m 33s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d72ecb91fdae470ea36da5e4b845051c : SUCCESS in 59m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a00fc90adf3748499d54ed695bdbf2b3 : SUCCESS in 1h 17m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/26b885a3761d4bebb86c9f9ee9be0782 : SUCCESS in 1h 57m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/07d818c4e9d04927851736da03d98165 : SUCCESS in 27m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/dcb6f23ae328446babd8c632cd9547f0 : SUCCESS in 4m 46s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/770adb5da2f24c9a85e0bb5ce34f58b1 : SUCCESS in 21m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/975dde0542f64c92bc74ca98be3c39cd : SUCCESS in 27m 09s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9e45fd3eaa3a4ec8aef9ff53a211b75c : SUCCESS in 27m 10s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5cb94be814cf4fa3af2af9368e99e604 : SUCCESS in 50m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c21d686d93e54ba39967a891e3f58054 : SUCCESS in 56m 12s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/06cd53839c534330a979232c963ca132 : FAILURE in 51m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f8773840612e470eb67f31ec6d519349 : SUCCESS in 33m 03s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/220c3c501de6442cb53f218c35cf7029 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":10},{"id":"a53b47988525764539e65beac024a7e431a5aa2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-26 09:55:23.000000000","message":"Patch Set 10:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":10},{"id":"18afd29f2d94b73893d4465e5c6446d67269eefe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-27 16:35:50.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"e8cc870882e11d2559c2c4150e3efe7db8ec8ba8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-27 16:38:43.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":11},{"id":"b9308af2c44813d92b8185f71dd1443a5950ef7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 19:08:04.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b92e7675a7843bebf31fcc5953ec290\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d5dd980726524b229d25b052945c02d0 : SUCCESS in 6m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78e8fa6f10604bdebe209195f9c02b2c : SUCCESS in 1h 05m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c173164e8640462cbcdc7f0df4061ef9 : SUCCESS in 1h 41m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12dfc180696d4c86a7ed1e1d1805e119 : SUCCESS in 58m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82f1fc17f22f40d3817897f97870ea25 : SUCCESS in 1h 40m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03af16856ee7474799dbe98146f34189 : SUCCESS in 8m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/970178075e874d2896b539beea0f9e6a : SUCCESS in 16m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34fc6a2c21a44be6bcb28444d0b0ed4e : SUCCESS in 15m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b93e782decb848c1862cbcc4625a0120 : SUCCESS in 19m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df53185e7d044edc842c533cdee980cd : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/069c3a51ce014111a1a6b0f851b72c73 : SUCCESS in 11m 29s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1fe84c47d27a46b780734ba41d6ba5ff : SUCCESS in 49m 51s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2d10f1e99ac04bfab2bc0966f89ef88e : SUCCESS in 1h 31m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a0b5a81bf1ac4c91b0ac7b0be230f0a7 : SUCCESS in 55m 18s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e1e0f43cc688464f8380a3a02ed33f1d : SUCCESS in 38m 13s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6ef96fbbefe048f3a1e37c9dced14090 : SUCCESS in 1h 08m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/409e699825f64f91a5b933bec13786e1 : SUCCESS in 1h 06m 58s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7c8f4586ad6e45cbbdd0d5788455387d : SUCCESS in 1h 12m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ff09d0f73e7d40c497e8bf0f18637a7f : SUCCESS in 1h 59m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/e6085887a70d46cebc5ddbd0578e1158 : SUCCESS in 2h 21m 36s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b2249188c59f41c9bde1b287333f5e36 : SUCCESS in 57m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5cff3b16492a42ebabf2dc4251ff35d0 : SUCCESS in 8m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ad77a6c155e74307a0fd65649f7d4bad : SUCCESS in 29m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b2a8c7bea9ab432e86187b984c660b69 : SUCCESS in 27m 42s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/02a815dae05542b0834d587724594611 : SUCCESS in 27m 48s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d198a46d4acf49c4ac7a2109e94ce479 : SUCCESS in 1h 39m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9948e02160d40f6bbb4aebf59b5aeeb : SUCCESS in 1h 06m 57s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fa54f0c03d614e7bb9af06d00c636564 : FAILURE in 52m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/25d75cea1fdb4d139c26fd86b8668b6f : SUCCESS in 36m 00s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c7bb8e40611d424db35a8c7e02dfc72d : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":11},{"id":"b06aa766c354bb1bdbe032aff1b1e3d1ce2ce195","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-27 19:18:16.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"901051d3670f6e9d613131fff20db9f3e98b6649","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 21:27:41.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2192c8cfb3a4839bb1e392f26901eab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7e8caa57f18d461fb1b49d148faae9c8 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f18168f1110e4763a22cae147c5bff2e : SUCCESS in 21m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/612997d7141042699235cf52bffd1412 : SUCCESS in 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"692605c21f85368246b11eb739c74132207032a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-27 23:07:28.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"5df9713120aefc8f105e4907a6d10bb5296ff8b0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-27 23:10:23.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":12},{"id":"545f5eba2f6de75d62dbc296144b3101c9c1ab04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 23:43:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6868885ddfad45c0b9e792d28366ad55\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/172b300d023f4829b946ac40fe2902f1 : SUCCESS in 20m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ac1426fc589648ccbcff0f4ee5139f27 : NODE_FAILURE Node(set) request cd3e92a4da704492807ac3b855de8087 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e40a8243357040989d20efe34627553d : POST_FAILURE in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c710a5ad7c2e7abf468b29bbacd97a70c5325eb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 01:10:31.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3f07cb4fa424cdd8e51881dfb36d15b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f02e3c25360e4a558b9bae0df75b46ed : SUCCESS in 5m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/83160e3f59d340adbf7d226b172248b1 : SUCCESS in 35m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9711144a3d8c4e1ebdbe88a4b71aafce : SUCCESS in 1h 35m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cc6f4dd4c024fd9ba78cb343a07b44c : SUCCESS in 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15112ec34c7b47daa6e797da6695856c : SUCCESS in 47m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/240a738e4d314bf783320bbbff01a0a1 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e47f008a0c8d4bc5bafbc382b1bcd2e0 : SUCCESS in 8m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cab9237caa9d4e82a3c0d03bdc72792c : SUCCESS in 7m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7694a28255f3419d9cb695cb8c521426 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/846f51c80c694553a9ea54fda62e67bb : SUCCESS in 9m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da03c9382faf4277bea09cb2c01eb7d1 : SUCCESS in 8m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9c07700575af4f788614343b89e5a57d : SUCCESS in 56m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/433e4dba2ab54295ba9143a9d70856a5 : SUCCESS in 47m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0293452745d440ac81d36d24b98201d2 : SUCCESS in 27m 40s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/db06f5604d6f41038f5eb4388719cf10 : SUCCESS in 37m 27s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9d486e704b6047e8bb30a0c070866cb0 : SUCCESS in 47m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/993d8504a5dd41de85d99523c3517dc8 : SUCCESS in 50m 16s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a076e7131a5f4b358566e12e5e4b35b0 : SUCCESS in 41m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/546a316800c04c61bd08570e6f716d0e : SUCCESS in 52m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/c347699fe0704204bc69029dcc67c7a5 : SUCCESS in 1h 56m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8f803ab61b374953983d6bc4b14b7fd5 : SUCCESS in 25m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a5da8d6c583f43dd9b2917f3dee634c2 : SUCCESS in 4m 47s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/441be956bb9b40ee89017bd0b5e3d16e : FAILURE in 29m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0118be6e6408406a854c37f963d549d0 : FAILURE in 25m 02s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/cf6ba6fdbb95403481b7dc94433ff00f : POST_FAILURE in 31m 33s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fcdd692bb2448edb258100148ee3426 : SUCCESS in 1h 28m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e20191e19cd24b6ea9b574b005226455 : SUCCESS in 54m 52s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6cb76fdd985042e5aebec2042319e73e : SUCCESS in 52m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0a3762e1b6b34468985b09b6f54667b4 : SUCCESS in 34m 04s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ac2ca001d0f2492fa80a9b2784454b85 : SUCCESS in 15m 25s","accounts_in_message":[],"_revision_number":12},{"id":"da4eb16924f5dfac3618397de2d31d8497d79216","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-28 10:08:50.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":12},{"id":"ecba6af87de957cbdac725d76cb4d7ae9ad1de07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-29 06:16:01.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"baf3fb09a44d7d3ef38bc308057a45e05ea33f07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 06:50:28.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0415a2fefe6048579464e9ab6994ae69\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aeeae4c065674a1e80d254b790c65900 : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83f0a47e5ab048f6b9beb30bcc38c6eb : SUCCESS in 21m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/54850fd4a67e4f4789c152aee74b2e42 : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"109928124ba5d312877a2d3bcdc942dd1c7a34b7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-29 07:10:27.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":13},{"id":"2be431875b61c7aa8d47e564209672e1f19ca8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 08:15:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d37edb8462c469a900b67fb4f5689e0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f7a412af869d4c739699f7ec524de3e0 : SUCCESS in 3m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a2e89ad2fb0e482fa811ebc052b72948 : SUCCESS in 1h 06m 59s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3792aa8c849a476d9f4ce28e1eecc4da : SUCCESS in 1h 53m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edcea3d12b9d4c8c86d19d5e554bd44c : SUCCESS in 57m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97fb2fe79b764ef0a0bff403c60a855f : SUCCESS in 1h 12m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15305dc620ce46769334ba5b201a5176 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dcb8358460844c8d88506e4a8a75024b : SUCCESS in 13m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b817b9343a66410a9e9f27d7a27e758d : SUCCESS in 10m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34c6445536644c13aa0bee7d3a83b9ef : SUCCESS in 20m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd79b45fbfca4a51bdf161cd53e2be73 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d98ce376fa094b3bb3917bd335f9a084 : SUCCESS in 7m 29s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0b4e3fb66f084971897c4f7af2347594 : SUCCESS in 1h 02m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd93285ca05848e9a23ca72d1cc2a5aa : SUCCESS in 1h 02m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/049f5740d15c4110b30afb282296427e : SUCCESS in 24m 24s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d408d50ac8e3410da95b0c1cabd994b4 : SUCCESS in 54m 59s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/fa2c7335d2484ec0a5fd560f525d9ae3 : SUCCESS in 58m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cfa1edc8e1c74e14a1fb59013132eb7e : SUCCESS in 52m 26s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/636232c46e6c4a76bc79b812be794d2a : SUCCESS in 37m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9b12babec5f446e384bb9e5d454cc182 : SUCCESS in 52m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/e8a51f74a3bc412caf83168d10df7928 : SUCCESS in 1h 04m 07s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/47938a721a1e48708e9d0df0ed9c9b1f : SUCCESS in 48m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ef340106e5ef45d69f9839f0d73bc852 : SUCCESS in 3m 03s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/691b77d137e24f8e81e9397f3165901d : SUCCESS in 31m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1098989b798e49bea95c91aec4c5cd89 : SUCCESS in 25m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c6673621999a471fa26baf16dc2efa5e : SUCCESS in 29m 24s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2fd7aa495314306acbe8dbba25d1d9e : SUCCESS in 1h 27m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9520eb062388481fa7b3599b014a44c2 : SUCCESS in 57m 42s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec9d2493cc4c4473a31e6976a32909e8 : SUCCESS in 56m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/de78e5c160c8408dba6f364f1ae56666 : SUCCESS in 33m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/de33f4613f744119b02d3de2d85271f8 : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":13},{"id":"383183868c1ee79e976c276f123d938bbe469994","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-29 08:26:24.000000000","message":"Patch Set 13:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":13},{"id":"c9d223af85e056cc1ffe1e9450a0ce779f0929f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-30 19:22:56.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":14},{"id":"154a3f2fdfa7956d80484329520231e9683ca0ef","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 20:19:16.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":14},{"id":"7446db9b712ab0ca21599c9ccfb7e5ab343cb0a0","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-30 20:30:48.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-gvz6d/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"cc3c59d06da3eb2a32335b25d3a0477c5f9b24e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:19:48.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f72df1cd7c6402db92947b872b7326e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63588f7eb02e4e34bc109140c5f80ea8 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/286b68ae1d804cb8b4cf5c7ab9770374 : SUCCESS in 25m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4828ac0cfebf460fa36c173702a9c4a0 : SUCCESS in 38m 03s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b9b562edb7a376acdb80ed7161bad59bcbfb7fa4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 21:22:52.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9373cbe69f94784b226eaa3293370c3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c0b2e0aef52f4c36a363a3dc51a53e79 : SUCCESS in 5m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6c4ad877149480aa172cc0feaf7e530 : SUCCESS in 29m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c6d59c2e7cbc426eac5ac65f432e9ea3 : SUCCESS in 55m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/119bfdb3698148e88c4d827864f75fbb : SUCCESS in 58m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24c81416e4434489a30978d46d05d7e6 : SUCCESS in 1h 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63578ee54c4a49de8a2c85c35bc92a45 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac3661c615884dc9b3a0e6cd3eb599e2 : SUCCESS in 9m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3e8866ba11643e79bb008b0eddf5c1b : SUCCESS in 14m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16cb43d6cc8249909607ac44e6f92327 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a792149ee2b74394b5d0eb7db996d796 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6908b59e3bd49329f143a97e710e80c : SUCCESS in 8m 12s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/41ccf33dc08640d09ed75ac9bafbabbb : SUCCESS in 1h 03m 15s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56b3b234852b40e68d341ac6a5302cb5 : SUCCESS in 1h 29m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/62d80747b44a40ecaf8af2fe50477635 : SUCCESS in 50m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8e5169fb85964733a1d28c9605f55c68 : SUCCESS in 33m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/30fabe4b53f148968fcc648f42e3083c : SUCCESS in 59m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/041a76bf89874778acef1d3f59dbccda : SUCCESS in 28m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9e2b529d1f5642f8bdbccc0790e9ef4a : SUCCESS in 1h 15m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/75552ed08d01449cb4c3158d7358dcd9 : SUCCESS in 1h 49m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/c02a18601baa469fb7f3e2d58e586eeb : SUCCESS in 1h 52m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/425ce3b2c80d45e7b47e37cace2ec783 : SUCCESS in 48m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/498fb7264eae4a37a7b1ef41e79ff247 : SUCCESS in 5m 34s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/751c461830aa4dc3855e44fdb3504b61 : SUCCESS in 31m 18s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/007d0127229f4139b3ef95dc7ab92a92 : SUCCESS in 30m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0b71b8efc54e418587742e2a2d788863 : FAILURE in 39m 05s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cedfa65cb5ea48a79ab2be03b838ba1c : SUCCESS in 53m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7469d4d787be4405a6638abc1187c426 : SUCCESS in 48m 26s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/625161be6ba34ac299df67ff13a53777 : SUCCESS in 54m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2f86594e9c3144e6a987280d61e728a6 : SUCCESS in 35m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/93484952e9a3472dbf667a97a652f414 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":14},{"id":"22385ae2e5508b294823b69d32217cab7b950e55","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 21:33:24.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"d9626f6f8da2eb4b17217791a15d1d1abec5dd67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-31 01:45:37.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"44b6c64970c2f3a30f1e3241ba9bc1068bb19908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 01:45:50.000000000","message":"Patch Set 15: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 998752,5","accounts_in_message":[],"_revision_number":15},{"id":"e30bba5d921e0d850f0c8238a64e29fb1b1f34ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 01:45:50.000000000","message":"Patch Set 15:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 998752,5","accounts_in_message":[],"_revision_number":15},{"id":"8c69fc300c05b128641447d993f6dbed5f20e8fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-31 02:02:21.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"6b2a48f1f409a79cab1c369fffa75791a4e3f566","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 02:39:26.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":15},{"id":"298511a97ea49243986c394ed70c23ad62f3f225","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-31 02:49:14.000000000","message":"Patch Set 15:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-f8r7j/index.html\u003e","accounts_in_message":[],"_revision_number":15},{"id":"d7cdc95d0bf20d7043feb1edbe6da7bd1c1f20ca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 02:59:19.000000000","message":"Patch Set 16:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":16},{"id":"78d12d7d967bb20b3085031e792d4e1f33476071","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-31 04:41:57.000000000","message":"Patch Set 16:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-hb7f9/index.html\u003e","accounts_in_message":[],"_revision_number":16},{"id":"85da84ecddc459b1a54dfd1cb26720b3fa240267","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 05:03:15.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32195f6b2e7f41939c9bd33d000bf72b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b816a84b43ed47d592453708ae0bfee6 : SUCCESS in 39m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb2555635d0347e099ad5ff1d83ab5f0 : SUCCESS in 34m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5a0559bfb97e42469224c92dffc23ad2 : SUCCESS in 52m 31s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"4e868190b0db3f7199514881c4c89a49b8e1b90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 05:17:19.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5acda9c7286b40b880611f5fc691e4d3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d02ff53bef8b455d93c318bbda90f480 : SUCCESS in 6m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ef3d36c78a146f98ab57c9a7dca808e : SUCCESS in 1h 10m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bf277faa01794f9d8911b06f49015425 : SUCCESS in 1h 40m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0295c31f446c4574897c83776884143f : SUCCESS in 44m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c92e14185e5476c8aa211094fb6e79e : SUCCESS in 1h 23m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b5b5fc358044e8b47e209e1f68321a : SUCCESS in 8m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea6f2a0fefad4a20a6c750b77fda5c5f : SUCCESS in 15m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/822723de0f184d2da56a4b701d700dd9 : SUCCESS in 10m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/00c57268e63b45b695265dec5fa25a4c : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b17b79b0294e7898911592df1fbf13 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a92f220708954281acf06cc92d7fec62 : SUCCESS in 9m 38s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/694dd31b795e40959fcd6b5482866869 : SUCCESS in 1h 04m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5ec79c8fa12c4da091369e89ec8e58b7 : SUCCESS in 1h 29m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/92c2dca0ed774c8ea05c22f03800b944 : SUCCESS in 24m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/278348524d6a4eebb54b30624169e313 : SUCCESS in 37m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/09e9d794299f4b44b43424a936c972c7 : SUCCESS in 1h 02m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c96fd8c881c94622b89aa08c8ec9f86d : SUCCESS in 42m 45s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3e6bd437ee244e8c8595aefb151d8533 : SUCCESS in 1h 04m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ebdd03753dbb4d0dbd0256c6c6938ad7 : SUCCESS in 2h 03m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/22845853b58d4bdca4c5fc4625111a5c : SUCCESS in 2h 07m 29s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/870972dcb3ff48caa803e140854a6485 : SUCCESS in 50m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/340baf0d7f1e4a14af63e1f9b26ec4a0 : SUCCESS in 5m 51s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2917377071f941deb597043914a03fe1 : SUCCESS in 28m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/07abe2079cd74d149768166e4d553bf4 : SUCCESS in 22m 03s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9527df93c51a45f38d6d6a74c3fbac55 : SUCCESS in 27m 45s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae4940f109fc4045a193ed72865723a6 : SUCCESS in 1h 28m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4766389501a34b1097d3806d701774bd : SUCCESS in 54m 08s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ee7dc3df81314f659bfb0a7977a19966 : SUCCESS in 56m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d9557169f51343dcb92272d0c14cb812 : SUCCESS in 32m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5e58a0a98150404ea4ab2a9bde36142b : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":16},{"id":"410b81e0b882a66be7c06b5a8b36ea5d21a1de57","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-31 05:27:43.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"9883c3049a1c0696399ff9f70ef0b4b489f00be4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-07-31 17:09:53.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"90611e0cd1be2eec4e35299dff289bef316da6c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-31 21:44:50.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"7f97c44869545727e54741b6c131a405cc54ffec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-31 22:01:46.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"db77a99fdfd631c476936646814cc42e4e05b75a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-31 22:21:40.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"d4bf6e4466ad2519a7dc5914e98d84141a2ca41e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-07-31 22:26:06.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"9afa9c0e0a59422d23a7958b2c005c1bdafedd71","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-31 22:49:44.000000000","message":"Patch Set 17:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-9ff8w/index.html\u003e","accounts_in_message":[],"_revision_number":17},{"id":"de75344bcb6dd344697fde7db6072994bd97d56d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 22:53:59.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":17},{"id":"03b9202b351ba8665d0a6461bdf476d6755bcada","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 23:09:44.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a4fb1edd444d5d84aaf462fa28bdc1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c25e0d2ee0b6494db29a7fc665871820 : SUCCESS in 33m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aea0d57f7c6a4b0fb40edb0466026c04 : SUCCESS in 31m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/49af2aeeaf8146319322fd19ed6d4213 : SUCCESS in 45m 47s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"cba2d6c70d4edb5f4e846b610a3eb26b9d389ec9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 23:14:06.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":18},{"id":"7dcd73c058aa0bf58b7d20579b63209288a7bb0c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 23:31:09.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":19},{"id":"feeeb1316514257b9b2771017da0072028f1407c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-01 00:24:21.000000000","message":"Patch Set 19:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-m892x/index.html\u003e","accounts_in_message":[],"_revision_number":19},{"id":"48d64b9242916e4c4982314703b6432a584c2183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-01 00:49:41.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5d5f43ffb34314980612b6f7e4f71f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/736d35d7a418484da766cc5a6665d124 : SUCCESS in 6m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d91177c687df43f4aff31191aac06207 : SUCCESS in 1h 05m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6f48fad94e8d47808c4ec5176c3449a4 : SUCCESS in 1h 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f062e27015df42388f4a757c28067e5c : SUCCESS in 53m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2b269ca6b2c4f4eaf5a7effa2e4539f : SUCCESS in 1h 20m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8073a07528824af8b5be47a6398df76a : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2216b6f3dd3744e1a351ac006ea413c0 : SUCCESS in 13m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd66fe32dafb437a9f5b25a7664f098c : SUCCESS in 14m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d68d517c0f5c47ecb73eabbf27dc8a82 : SUCCESS in 18m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e12ef6e9ca45df9f22453c08cb77ad : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed7c9e368b274855bc5ba201569208ca : SUCCESS in 8m 15s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6eefd42126014524a14a10065329c900 : SUCCESS in 1h 07m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78a9b6efa5e7403db6c7dd563091e5dc : SUCCESS in 1h 28m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5763fd8f40f94f9c9f614f1353127c29 : SUCCESS in 53m 41s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/455b9718b8a241968c687b73d1962e32 : SUCCESS in 38m 29s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/655fc17a11e1469db26d0c96db9b85a9 : SUCCESS in 1h 01m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/722a86e98053457b87dc36d1e5308d47 : SUCCESS in 51m 54s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9871fb66528a46a7b82f08c569725966 : SUCCESS in 1h 00m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f27cfda1115c40d6bc9624de36a35176 : SUCCESS in 1h 08m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/885e28cc95c241c2af66f6eeafa57f9e : SUCCESS in 2h 20m 37s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d749fbe5c2fe465a8723bdc9a3ad43a0 : SUCCESS in 44m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/82f99b1bf10745d9a487d687298fc190 : SUCCESS in 5m 35s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/545d28a381104d32aed5c14f69bd6d62 : SUCCESS in 28m 51s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9fc20b4cb71441caa7ea74d0141e5e62 : SUCCESS in 30m 18s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/eb5ac75df13c429693c6fc245096d72c : SUCCESS in 34m 57s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f8d19df38524f7da8f901f6ef75ce18 : SUCCESS in 1h 25m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17c62e10b2894985931cf575bebdfb44 : SUCCESS in 52m 30s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fdc9a08a53fc46fa937fdecd83f72296 : SUCCESS in 27m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9935055bc8d54226af61ab921f948841 : SUCCESS in 26m 26s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/47a834982e8f47ea9ab31c545ef2d16c : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":19},{"id":"ab3c030a4d4f453aec86bf9bd60d41d53b5f6a93","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-01 00:55:14.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-j98vh/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"9c74e4cd52f7bd29685801f8a4c23b4c4519b238","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-01 01:00:04.000000000","message":"Patch Set 19:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":19},{"id":"733aa69eb01f34aa76fb76ca0fa8e7a598cb4a3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-03 18:11:09.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"9b4ea5ac04ab2a495d579f975cf0cec00b45cf57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 18:38:26.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ae55957507435283766c9474d06a05\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45294ca3d44c468694f8ffc87573add0 : SUCCESS in 17m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6fe235b1b5c34f4db37832d44b9933c8 : SUCCESS in 16m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c5f0bc4de0b043978775c4f9a06d3dd4 : SUCCESS in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"de8ad0ed1f27c564992c62a307150f4ec756027a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-03 18:49:04.000000000","message":"Patch Set 20:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-sj4f8/index.html\u003e","accounts_in_message":[],"_revision_number":20},{"id":"3924938c69f6c7b1c94ca6679867fbe056e171ad","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-03 19:08:39.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":20},{"id":"b7c49204bd8e560d350853375695a73919c8ff55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 20:26:04.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b66f4e7b03154598a9e11488a0e86a97\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d52bfe32e7784ca6893ad01ddc641e29 : SUCCESS in 3m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7f7d2e22e744695ab9281b54552e564 : SUCCESS in 1h 08m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c861d54a9d3c4af0b134d9fe5ec11579 : SUCCESS in 1h 42m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8195957bee52413b83fdef6c3be3e6e2 : SUCCESS in 56m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c1cab7ba0074530ae2278cfbc0256fb : SUCCESS in 1h 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db8e003266f445e8b04071f33ec197b0 : SUCCESS in 8m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/314d52eb28d3442b972992233b194a48 : SUCCESS in 13m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c46281c41b845c785c62e76d92c4e32 : SUCCESS in 11m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dbad8f06193b4fc882ae85f8a0402b10 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7e41f6e80c4b96bf24fa2ff173a60d : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf80a0d6d34d4586a32fe76442e36701 : SUCCESS in 8m 13s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/648b20d8c6054879bcbadaa85f3ac930 : SUCCESS in 56m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ae0255da7054d71be37e83590ce960b : SUCCESS in 1h 24m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e84b5d3458d943ccaeec581d1a2d53d4 : SUCCESS in 52m 00s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8bd1efbe237e41e890e3114edfc6373c : SUCCESS in 35m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/da73c9280e4348a78e040cfa8a30e1b8 : SUCCESS in 59m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e9429563d567402ab2032593a8fc8278 : SUCCESS in 39m 26s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d5979d2b3a114d86a3d8ab514d0d56a4 : SUCCESS in 39m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/badf307538964d3f944bca7cdf0c92d7 : SUCCESS in 1h 37m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/4b9996b4e3064457a33a4c6a7cfb5ed2 : SUCCESS in 2h 08m 29s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/958769e3d2bd4ecd921f07fa12454e2b : POST_FAILURE in 36m 45s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6ea9d3dc59034fbe8f9b909eea8fe8cb : SUCCESS in 6m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/84164d5449b7410ebc6f10258600039d : SUCCESS in 25m 05s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/46ac78d85011477dbc614be54acb0e43 : SUCCESS in 26m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/710600c132ff4c87a3b247dfa5a9fa58 : SUCCESS in 27m 55s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/195305de23194f9eb5ce94e4bd55d563 : SUCCESS in 1h 27m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0004c623e1e74bddba76535c8aadd18e : SUCCESS in 54m 46s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9339ded4faf747b4a07f6d08b612e23a : SUCCESS in 40m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e7f2e523369045f9bf6aa0f199f8a51c : SUCCESS in 25m 21s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7d97506c05fe4683980548bd4d6f24a8 : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":20},{"id":"e9cb3ce0066081634ad66b3c16235efec7324d59","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-04 16:51:19.000000000","message":"Patch Set 20:\n\nrecheck","accounts_in_message":[],"_revision_number":20},{"id":"442e1c84d6cf140c82a2fb3f596c6753c77ebbfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 17:37:09.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62fc7108d2724cecb1d9e06c228e4310\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00efd8dd63a040daab18d2e46746e3f7 : SUCCESS in 16m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7d76b51de88d4b5b8029c1c0cdff9117 : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b1a084b823314b8181da055fbc12c881 : SUCCESS in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e7f1d78396f467c7f93dccb0ba61d10c28747345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 18:56:04.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f4a72b02dd0425eb342dd45fb7dfd18\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a08dac0826bf43799b4d58855bb3e511 : SUCCESS in 6m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e828315e67a24170992938c0852ecd05 : SUCCESS in 1h 05m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f0b0ff4653b240fd821e83366048e5b0 : SUCCESS in 1h 44m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4cc09e2444be44f2a2d9c59555a847ea : SUCCESS in 1h 03m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2d05ef647474555a05a968e2941feb7 : SUCCESS in 1h 28m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b4c6dd6d4240bba63208b8fd449408 : SUCCESS in 7m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa480ba3968b4a858e55828018533d5a : SUCCESS in 14m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08507145cb624efb9d5d37d829347fb0 : SUCCESS in 15m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f8ce1c1a7ab546a0bf5ce4950c26e865 : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3106e809134c4ac6a809de6d6ae318d5 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/633c3e2658384e198bf5fc6d6da940ec : SUCCESS in 10m 50s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/34fdf394ae1e43dfbc8162be147dc763 : FAILURE in 1h 27m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/050558d92aa7439b8f4fd5193b369d88 : SUCCESS in 1h 48m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/14d02f130dbd434e82ceb9aac7c1b7d2 : SUCCESS in 27m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/b536f03f06804cc0ba3572e05aff59f1 : SUCCESS in 38m 28s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/cb8ea806a7df4eaf8523f54e8e2c6d61 : SUCCESS in 1h 01m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/45bf580a1cdf4f7caadfe3ef39977175 : SUCCESS in 29m 04s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/990c0bd33e784fea96bb5b3fcd30d47b : SUCCESS in 57m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b0f0a893d8404bf7b81f8e728f678330 : SUCCESS in 1h 58m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/69c89b06686d42c6a67797f0916ad4d2 : SUCCESS in 1h 50m 07s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b387af7a29df490fb736a3d305282e76 : SUCCESS in 49m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6f24266327c74ed5acde2599960b3d84 : SUCCESS in 4m 49s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/df856992b29a422fae4e6d68c1a10c3a : SUCCESS in 28m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/55d1b6887c19407795307f31fdaa8f01 : SUCCESS in 34m 49s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/280f23e2761248859c7f9a1ee92bbb0e : SUCCESS in 38m 06s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7a8a21960cd4704afade5e294804bfa : SUCCESS in 1h 26m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8173bfcbef1146269c38220ce7688830 : SUCCESS in 51m 43s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b234aef705eb4d14a36e916df1416b95 : SUCCESS in 1h 02m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5b0c7645046343d9adbd2c927b54c280 : SUCCESS in 33m 01s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fa26fb6eb9de49ab83ebb0bd1530b6aa : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":20},{"id":"23ce761743235b8d4fecf2549e0aa45495f59867","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-04 19:05:59.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":20},{"id":"dbd454a93cd53ddd709a62b6f5b46f02a5c72cb9","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-05 00:53:00.000000000","message":"Patch Set 20:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-tg8gv/index.html\u003e","accounts_in_message":[],"_revision_number":20},{"id":"3a4bb9650312c1ad2bd7bcd55c92d71d3e2b1b5a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-05 06:24:22.000000000","message":"Patch Set 20:\n\n@danms @melwitt @nova-cores - I\u0027ve put up a DNM patch at https://review.opendev.org/c/openstack/nova/+/999772 that runs virtiofs share attachment tests against Nova with two jobs:\n\n- nova-share-attachment-cephadm: Single-node CephFS native backend, runs mount and multi-share tests.\n- nova-share-attachment-lvm-multinode: Multinode LVM/NFS backend, runs mount, multi-share, cold migration, and resize tests.\n\nBoth jobs pass. The test code is at https://review.opendev.org/c/openstack/manila-tempest-plugin/+/989634 (cold migration/resize tests) and https://review.opendev.org/c/openstack/manila-tempest-plugin/+/984884 (base share attachment tests, already merged).\n\nThe tempest tests will be merged by the Manila team once the Nova changes land. Reviews and suggestions for improvement are welcome.","accounts_in_message":[],"_revision_number":20},{"id":"3404f7e028f0cf2c4669a46f6630306ab53b9bf6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-06 14:46:43.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"a7839669171309ca945eafc1accd821491f31925","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-06 15:22:37.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"d8680db202d4ff3fcb961248194405aee274b319","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-06 20:49:45.000000000","message":"Topic share-extraction removed","accounts_in_message":[],"_revision_number":20},{"id":"be829b4fe922959828de512baea87042ab812376","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-06 20:49:50.000000000","message":"Topic set to bp/virtiofs-cold-migration","accounts_in_message":[],"_revision_number":20},{"id":"0e40defd338da9a357164d4bf9d4239dcb43f781","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 06:20:27.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"c19d473c358e7cf66c8a9f73bdb57215c3c0f4d2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 06:23:03.000000000","message":"Patch Set 21:\n\n(5 comments)","accounts_in_message":[],"_revision_number":21},{"id":"3a0dd8a4950d9aceb2c79a25935d63018c1db9dc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-07 06:23:50.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":21},{"id":"dfb7cd9b487723fc7b29351aa2d3d4b75eaba73e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 06:30:54.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"880620463863fc4a41115e53f16e744c603feb81","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 06:39:00.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/571c56335b74478fafd389dd12dfa1d6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8d514da2172944d4aa82c470ec83c255 : FAILURE in 8m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/81a970c50d644056b159658626b5a6f6 : FAILURE in 8m 03s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/25a2ab1c516b46df86f76ff9669065e2 : FAILURE in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"2c2425c0f7ad4c7b60e0c06f1d6ed1ee922a38c6","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-07 07:07:29.000000000","message":"Patch Set 21:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-b5fkt/index.html\u003e","accounts_in_message":[],"_revision_number":21},{"id":"9324f460bcd9bb10542dd6539d29270e3c9975c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 08:37:34.000000000","message":"Patch Set 21: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61c8d270f2cf4da8aa31a63fd02528ab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45e082094dd4457e846395501018cf98 : SUCCESS in 7m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3c8bbe149f8a4b75ad2b06d957134bca : SUCCESS in 1h 02m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/da9ff691bec6497aac5a13051ba16fee : SUCCESS in 1h 51m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e8e3325c7bc4721a9cd16b83782f268 : SUCCESS in 1h 05m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc5495ebcd10417a800af59dcfdfa6e9 : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df4c704a6da450f9b6a404db49c6d2b : FAILURE in 8m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba76b4f70c6d4bbdb714d72619d90882 : FAILURE in 7m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0942074de7d649c1a1c9d9773fdee289 : FAILURE in 4m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a0a22c85dc794535b013b2a30d2241cf : FAILURE in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5752f300eb3c40379a59f5bff2fd79b9 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dabafcff6c44b7e8b53f035e61d0ca7 : SUCCESS in 8m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/dd24d655b51b42048e182dda4a4f1760 : SUCCESS in 56m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ddec7531704a41f1a57f7c5f0729ae4f : SUCCESS in 1h 31m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0fef93dfd5f54bb3a06947e063c6f666 : SUCCESS in 56m 47s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/dac16a59edc34cb2a933802271f1e38f : SUCCESS in 36m 22s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/506b6bd76f7b4c39baf22e2587965c09 : SUCCESS in 51m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cf5157459db04f5d868867a6cceeea0e : SUCCESS in 50m 39s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/546804572a4d44df802fa628cf8d30f6 : SUCCESS in 56m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f2edd6c7563046d9b2b07b4dd93d3303 : SUCCESS in 1h 35m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/eaa9920176fc49798adc06c620c992fa : SUCCESS in 2h 09m 51s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/1141cbc05a5f429f9435885437d5ae1d : SUCCESS in 51m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ccf38e1802f64c8f8ab3b5c0997c5f77 : SUCCESS in 7m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/773fadd54dc44912a1725c239d0f1417 : SUCCESS in 25m 54s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b842a5e047644595855796e97667a0a8 : SUCCESS in 21m 56s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/86a35955c8e6471db2a6db22ae4eee47 : SUCCESS in 28m 00s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb9e658ef1c94a5f82c30d6f4053d87d : SUCCESS in 1h 15m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d032b73712e4834bd01ffe79f3593d8 : SUCCESS in 56m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8972c68497034892aee21a94e20fcf8f : SUCCESS in 55m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/87f819aac9b547c9be51e00a007082f2 : SUCCESS in 33m 37s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3062b6ad5b9a4567bfbce73685b1d606 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":21},{"id":"0a83d691dbf8596d8061e9f12338a44b3cedae66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 19:02:43.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"edf14a2e1e6c5fa91924e147f842adfa0ed883b6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-07 19:35:07.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":22},{"id":"0b22f0cb9065e6a249eb7db93ff122355069bc53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 19:42:36.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf14b2d02e41497da4a0a6add432a4d1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/270454f6cd7f4e499036e303b352dd2b : SUCCESS in 25m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7f3bdf96b7c44bd0ade62dc5f2c1acdf : SUCCESS in 25m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/118cdebc4b044215b72858164dcdfe7f : SUCCESS in 37m 10s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"351421d84caa985094ca6e763ed95280527238ad","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-07 20:13:14.000000000","message":"Patch Set 22:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-6dwlr/index.html\u003e","accounts_in_message":[],"_revision_number":22},{"id":"abf37e0126f9b7c49d46188a96741b65d3c73abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 21:00:51.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4022af2da45c43208e52eb6c12ca38eb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/32ea337a21494f10893e29c10e1186d1 : SUCCESS in 6m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/644d12771f7d4684b7417d67fb36f67c : SUCCESS in 1h 17m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4a457eb30a5e4cb796878f71b070fcdc : SUCCESS in 1h 43m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f56d0e94e5e49abba39e1c7a548a451 : FAILURE in 52m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c54b39e61e945c0a8de3cdb3e8fe56c : SUCCESS in 1h 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8539e8a1eea344748db2e595f085288c : SUCCESS in 8m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/206bbb8c46b44964b7aca7c9023f0f69 : SUCCESS in 13m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ded62cbb6834f4f88914184faa14b00 : SUCCESS in 13m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b7d3628a891423d92f1997217a78869 : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fe0399501242d6bc8c333a386490d2 : SUCCESS in 12m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fffd545b85ff4d6b9e02aa745080a92a : SUCCESS in 9m 24s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b81c90c4f1834688bbe9e4e71df1fc94 : SUCCESS in 1h 03m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56d6c75aafba4ca3af6b5d15215bf2d1 : SUCCESS in 1h 37m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fb5d10a72ff7424cb694ea357fc1a53c : SUCCESS in 53m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/64f987fcaa244e4c957fba50c8cc760a : SUCCESS in 34m 45s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4ffdc8077a954a03b0bfe67031760ecd : SUCCESS in 1h 00m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/67901062e2ab4f7eaeb3d3fc11b1de17 : SUCCESS in 49m 36s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9e72d7ea1842496e80aed73dc93c17d0 : SUCCESS in 1h 05m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d0d99e9c4e6a4ca497dd8b556b6031a1 : SUCCESS in 1h 44m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/df205580d648499490856a82f372b606 : SUCCESS in 1h 49m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9f7e9cf39ed0439fb7c4bc8aaab399d1 : SUCCESS in 48m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cd538456f5fc4a309bc5650b17aa8043 : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c01642eaa978423a9a7629d06bbd891c : SUCCESS in 27m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8e9ba7a214c843a4a9cf9fb7f94db488 : SUCCESS in 35m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9867ede6046f4b388ad8031162d3b156 : SUCCESS in 33m 38s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/854fb9d8f8ea43e5b0dbb28b2270e89a : SUCCESS in 1h 35m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fadd0bed36df4fe3a24c9d05cb1be5d1 : SUCCESS in 56m 21s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/faa20e84de4a48b6b65018dc01a0c6db : SUCCESS in 59m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/61fed490eff544298d539eea84a2e61c : SUCCESS in 40m 25s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/30906123542e43eabcbabff28f2f4eef : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":22},{"id":"3020c80cca70fb27342fa338fcee7ec546a3f30f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-07 21:13:46.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"760ab52a182ab09429b7b7bec6b3b9c215440f3d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-08 20:13:15.000000000","message":"Patch Set 22:\n\nrecheck\n\nopenstacksdk-functional-devstack flaky, unrelated to nova change","accounts_in_message":[],"_revision_number":22},{"id":"aaf15db4d9fb9138ef94d1b3c4138f04bab20d10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 20:51:19.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8857788304704c1dad52dff71be5bafb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea7f535d51504817a125bd0efd6538f6 : SUCCESS in 21m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4ef51d92b57b4d6a9c58ad59a2017a62 : SUCCESS in 25m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f21b811dffb94c5898b3c65c5d981b09 : SUCCESS in 35m 40s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"122a6f0bf1798952544eed4373b1a11b4881b998","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-08 21:10:37.000000000","message":"Patch Set 22:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-j2jtn/index.html\u003e","accounts_in_message":[],"_revision_number":22},{"id":"112aa069f562b87d58b438b818a3a5bd40e8ec97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 22:09:10.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f5be9777b64d3cb20ed12373911cf0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/54eab3e38aa848d1b046256050102867 : SUCCESS in 5m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94b8de270a554647a54d1b11c177cdac : SUCCESS in 59m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/115326031c954bffba3967283d08b42a : SUCCESS in 1h 37m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a54d1c9f04764a1fbf7314fad009ba12 : SUCCESS in 47m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c995583363e4499b9463fb9dfb4cfe2 : SUCCESS in 1h 21m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f94477d374648d981e0f243d3025ed0 : SUCCESS in 9m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6940dc90e07546ea92134068c5db814a : SUCCESS in 14m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5a8f28e837e4b3e95bd0f467437870f : SUCCESS in 12m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff7a0e98a7ff46159a80e3c3e01345a5 : SUCCESS in 14m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93467ec5502e43518e12116484a696e3 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06c5caa543c5426697306c0eb287b3ad : SUCCESS in 8m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7315277600e445f794a057e4607911b4 : SUCCESS in 28m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0fdab713677b4fdebe68589fb0d0bd88 : SUCCESS in 1h 37m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f2f8ad9a99cb45e08fcd330b87e1fb47 : SUCCESS in 47m 53s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0729c9c95c174e22b6aab822c8a2ced0 : SUCCESS in 33m 08s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/109d18d9230c4012959b59faae26d133 : SUCCESS in 36m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/431bede876074c38bfe8250c91b39557 : SUCCESS in 48m 02s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7d55513a26a548e98c2b7a85a813aa51 : SUCCESS in 1h 04m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f41370ce7a9e4a13b4ff768a5c52f73c : SUCCESS in 1h 40m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/1fb26f5d13e245c29823d17ab696bbcc : SUCCESS in 1h 49m 38s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/471f6cfe604e45d28db0979fa6dab9c0 : SUCCESS in 48m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f09d6089c0f543639132a30dfdede19b : SUCCESS in 6m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/738bb0f559d84cdcb36c3d8ce6af6278 : SUCCESS in 25m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b2b78f3782384b58bd09f0b5ab725006 : SUCCESS in 23m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9dbfebc192a24d4ea582d6276d681a20 : SUCCESS in 37m 30s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cbdffd55bc05485e80e70a61d3bf8a3b : SUCCESS in 1h 24m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ee65966c5674c3789b8226cb6e4727c : SUCCESS in 36m 42s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/aa57fda3f8dc4743ae41a80d782ee189 : SUCCESS in 41m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a6101d1d4d934ec1bdb6c4d902ebb8d1 : SUCCESS in 32m 59s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/918117f7d1034eceb914b2fa0da54c54 : SUCCESS in 11m 45s","accounts_in_message":[],"_revision_number":22},{"id":"66ee750d8b14a40b426e42dc8e009056035ccefa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-08 22:19:38.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":22},{"id":"9f41c0724897eedd7e27b1d38317fc191010add8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-10 13:46:44.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"94b25be98494ae5bce9e86471ea73839795ab2de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-11 07:40:16.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"3f5b1d518c6619a2c25a57ecb3051b58e1cb20a6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-11 15:32:26.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4953bf54a2c91ef44b87186b727ae04bad907804","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-12 06:35:36.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"abe6861fe80d6b88a8176b5cd9a04f55e5fce6d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-12 06:36:02.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"f7706de8724ced382bcccf333ae4a05cef7a613d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 07:16:03.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5e4e9a6177f4e7da24ef6b51a9610ab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/580834dacbd54b18a9ef890fb5245d3b : SUCCESS in 37m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9f6b3f9a420a40c6ac34c4eae1574f62 : SUCCESS in 36m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/72ef1962d83c4d0d8c5142871aaf9d29 : SUCCESS in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"3692df41578526baaf2630dffac4484a5b3f6c04","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-12 07:31:36.000000000","message":"Patch Set 23:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":23},{"id":"33f3dd1214b213eac2ce4933e301ce93e84a8432","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-12 07:53:24.000000000","message":"Patch Set 23:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-gmrss/index.html\u003e","accounts_in_message":[],"_revision_number":23},{"id":"6828992954342815522942adf51c2baa81a1bbac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 08:36:54.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2ba985e82ce42feb262c25ba43ad093\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35d381de761a40cbb44899bc4ba7eba0 : SUCCESS in 5m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abd25d792cd24ec09b22d477340b4502 : SUCCESS in 1h 02m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/88def2dc0f32468c9f47994f08e73960 : SUCCESS in 1h 30m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13525291f1314c6ba0bbded8afbcdd12 : SUCCESS in 56m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e591a20495fa4b4cb9b11282e68b4db1 : SUCCESS in 1h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2205abd69e28498f85b85351eb14941a : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dad111eff9ee46e884bbd115e2c44e2d : SUCCESS in 10m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3faa9e75de04b5cb5f65ff350ebd05c : SUCCESS in 13m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03d30e3b242d4b6f83084ea9881329f5 : SUCCESS in 16m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef64e6ce1054997b59ed55afa5225a4 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/578d404bea014d399c477bf9bacabb13 : SUCCESS in 6m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/076dce53ea804566952394f6423da9bb : SUCCESS in 27m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/24b1940cdbd94f27b47f9a58d4b4d4e7 : SUCCESS in 1h 34m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b7271b9b2d55469588c399d36cd616eb : SUCCESS in 23m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/7da37a2822af4c0d9961c15d102e5de4 : SUCCESS in 35m 28s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3280a07310df4d33a79926e18898a720 : SUCCESS in 50m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/487a19f1fac1449a9a114b548d95cc78 : SUCCESS in 33m 26s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/92fb7e1bd36c45c1902d1b754a23a924 : SUCCESS in 1h 06m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9abb9522b3404e088d5acf5979d32e96 : SUCCESS in 1h 48m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/175b898284634177a86e07fba09aebf0 : SUCCESS in 1h 53m 57s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/da315367de5d43bbb1243a4d71d68c79 : SUCCESS in 48m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/86dce20d34ba4a929bbac3025a06faa4 : SUCCESS in 6m 34s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c335151e0b5245b1b656aa8678b67166 : SUCCESS in 25m 58s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6c2a7b3de882402192e0195a5e3b953d : SUCCESS in 20m 57s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c14db07d166a489e9918320a9c39cef6 : SUCCESS in 27m 19s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c725a06e73984ce4aa905770070f1826 : SUCCESS in 1h 31m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ed1669fbec249e1a87c07ac64e6d6c8 : SUCCESS in 38m 28s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f7013744e1a0479bb6d05d1d3ea8f35b : SUCCESS in 54m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/eaeffb04bc5b4b3e8ef284de27a24d33 : SUCCESS in 31m 59s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4869468f02934a01b359d2b99352495a : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":23},{"id":"75c595e890ba98ad5013a0fff6acc750df85b950","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-12 08:47:34.000000000","message":"Patch Set 23:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":23},{"id":"79b63f7d67344b5f62388b3f48d4002e474eeb31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-12 17:54:37.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"d172eaf0fe0f42149c282e9c1ac7c39bff25cb77","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-12 18:12:18.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"53b3263195f541b1eb47dcb6d12378314d35da72","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-12 18:25:47.000000000","message":"Patch Set 24:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-ghgfw/index.html\u003e","accounts_in_message":[],"_revision_number":24},{"id":"e79af647a40fd18398ee968c97d640d0831ae9c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-12 18:49:06.000000000","message":"Patch Set 24:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":24},{"id":"fb269492364133583dc89c452b2dbb7d85008ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 19:50:35.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f542c898bb814aa6af09ef5afea6f666\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/510170beb1bc4cfeba5d12f98568262c : SUCCESS in 5m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5d7644a91e624b62bd043099b22fd1e7 : FAILURE in 15m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a6cf9640c749410eaa97424bfee8c4b2 : FAILURE in 15m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d505a13d21649658e8db5ed7ff56f0e : SUCCESS in 50m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f68b1dbea3a4190ac0056dca83d7ae3 : SUCCESS in 1h 09m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a19b054eff4c9e8f3b846adeafd262 : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/90ee3b5af57347979a21e80f5eb155eb : SUCCESS in 13m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/acca2c98d2af4e5ba160d0034ce59231 : SUCCESS in 11m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc56a424c6354e978ab9a246bd0e6793 : SUCCESS in 19m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c0d2157e36463fa6f4a3b771b28989 : SUCCESS in 13m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/133ba9dc4afc499abb25400b8ca7f4b2 : SUCCESS in 7m 54s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/182085b45cd94aeeaa8d13773f0c6f16 : SUCCESS in 27m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9509f1a38a164f1d8aaded1b3f66613f : FAILURE in 1h 24m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c0b059c34a834e42bef19bf768dd10f3 : SUCCESS in 54m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/3330130459534c02b42c4c1f0310467f : SUCCESS in 37m 22s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/211d48c9d0734e23931edcb678b598f4 : SUCCESS in 59m 47s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c9faea3878cf43509f48ddf038123d86 : SUCCESS in 51m 14s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/850f5ef117e848efaeaeb154813589af : SUCCESS in 44m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cb3610a03ebd47b682d00ac2ea34da01 : SUCCESS in 1h 50m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/d37d2dc98e694a86b1cd4b9e08bb6053 : SUCCESS in 1h 47m 09s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/eff788d228514dd0a035a6c85dfa295f : SUCCESS in 48m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/74ee56dd555e4c3694f8cac34da861ff : SUCCESS in 6m 31s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/27995a66c16543aab12ac807944996c3 : SUCCESS in 21m 41s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/beacf501492b4103aa44c31390c56192 : SUCCESS in 28m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7b7c6ba9c7e640f5b8954ff9f6ed04e9 : SUCCESS in 32m 31s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e26ac55bc3f247b787a3ccbfc045e61c : FAILURE in 1h 16m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d545a0938a84378aafce8d41bc59e45 : SUCCESS in 50m 51s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0e3381a4be3e40f48c583ed3d0f0b896 : SUCCESS in 44m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c77e1f079b264d1ebacca609e7078694 : SUCCESS in 16m 01s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dca78fc5224348229191c092cd1a7576 : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":24},{"id":"1303b4eff8b2e120c39c7c17f28d636040c99e9e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-12 20:27:15.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"4fdfb4c24882da9a49dee347d3a2c1613dd7adf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 20:27:52.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cf5b87be6a8477d83e6cb816b66fae5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dd550022a1cd4ed89a7160c1f338e8d8 : SUCCESS in 15m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8735da67ad394b309452247fc086ea38 : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/227ab1f5a24149ef93ab7c951dba8277 : SUCCESS in 37m 33s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"dcf89059ea61cedce85494776d2b7863f3ae63b1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-12 20:58:16.000000000","message":"Patch Set 24:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-qqgv5/index.html\u003e","accounts_in_message":[],"_revision_number":24},{"id":"549f48ffaa700c25c5a154eb0a559046229f4f25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 22:21:48.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54462d7540764fc980e9015bddc1cca0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7cb50acc3c65486da85a08a8d0fd7e3c : SUCCESS in 5m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcb160447c76476fb75e2886b2088122 : SUCCESS in 1h 07m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/666f586ce51f4f4aaf9cd3a6ca6bf391 : SUCCESS in 46m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/886a3059cd1c480593da2606d695fffa : POST_FAILURE in 17m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/247d2ad56e2a4348a173c4f570e59d47 : SUCCESS in 51m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82cffd1b461a410ebcb8e4fd46019cbb : SUCCESS in 7m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40fe434d46684d9588e8bb7828fed7eb : SUCCESS in 13m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44e330cc4e2541a7b739ccbf4ba9af6e : SUCCESS in 13m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4fb5b3f3815f44769444b09a8032c1b2 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b0df70a9374ea0977f5afacc19bde1 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f53dd046b29643928ec285818ab029a8 : SUCCESS in 8m 37s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/6e21b97c8c3e453a86ba0bb966dbf249 : POST_FAILURE in 16m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aec842a43a76465f89d6490e26f85a3c : SUCCESS in 1h 34m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc6ef557c00d409797e955698d7b0e9f : SUCCESS in 49m 00s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e20857081c5b416fa337747a321cb33c : SUCCESS in 35m 27s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/7a01d7a96d1c4e9a82622464ac2ec884 : FAILURE in 17m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7e932eaba3b7433494d99bd89d7c4164 : SUCCESS in 24m 32s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/34634c8c26cc43a1864906188589ad99 : SUCCESS in 48m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3f6d26c9ffd847e7abd490687261d961 : SUCCESS in 1h 13m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/fac2bf7c87f14ad8b0893af78c885187 : SUCCESS in 1h 47m 20s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9b390b7310ac4f92b035b4a8ab9ba11f : POST_FAILURE in 13m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ec924ce347954533beb119e9747abf1e : SUCCESS in 5m 37s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0afba37890354d74beb2dea5276df4b6 : SUCCESS in 31m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/39b02d6cf40b4b2b8af08be06a752069 : SUCCESS in 26m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3c2d7b184c554c399ea5c29275841571 : SUCCESS in 27m 20s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/188f5da228974269acaba3b1da318fa6 : SUCCESS in 48m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/367772a0e8e5442d93519f66f83ff2dc : FAILURE in 7m 40s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4b389219fa954291884988a8e72eb1b7 : FAILURE in 11m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9819f62ace374a40bb0abdf1dd20b6f8 : SUCCESS in 21m 35s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/76561538d9714740827e9187260b9c31 : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":24},{"id":"0dcb2be36d54ae9511aa8fc972895941385d0448","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-12 22:26:10.000000000","message":"Patch Set 24:\n\nrecheck\n\nVoting failures are infra flakiness:\n- nova-graceful-shutdown: test_revert_resize_dest_compute_graceful_shutdown timing flake (passes locally)\n- tempest-ipv6-only: etcd tarball download from GitHub failed","accounts_in_message":[],"_revision_number":24},{"id":"9f8f829ff883507c05a6a9b05d95be1ea62d5653","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-12 22:56:59.000000000","message":"Patch Set 24:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-8lm5z/index.html\u003e","accounts_in_message":[],"_revision_number":24},{"id":"e583636d80a304649b3018eb511fe494bc1c5afc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 23:04:36.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ddf7d1f1944dedb3f214d4eaada8e7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/deeba300ba0b4d2b9b26092c21e37771 : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c88e8f1d9234c0c8d27ff964a762e3a : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f15509bfeaba4b0792ca7f3dbbe46553 : SUCCESS in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"b4390246d5749f4ed1975b8709d15e84a568c9e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 00:27:39.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947a4c3610634e01b25d553a6e63c8dc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2bb296e87f9d4e86a5aff5a7056ece2d : SUCCESS in 7m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7a818c4a2f94bfc8abcb276e402b379 : SUCCESS in 1h 01m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8a514be67f484a839a0e98e7abd9a94f : SUCCESS in 1h 39m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6db6228b0f84b669d05c70373d080ae : SUCCESS in 58m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c1faeb42f6248918e33882351ae9b83 : SUCCESS in 1h 14m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6716a246a84d4935a9d0171d247ad4c3 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4da274301f14e849966c762e9e1ffee : SUCCESS in 15m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/988a0af2fbab4f5e86c53510d998bb1f : SUCCESS in 11m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/db299f21fc834fcc9992a826526934e5 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1168da26c20240a8b1e5e42a874c9110 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4883be66bf35468daf9386a0ea562630 : SUCCESS in 7m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/316efaa234364d159666dd64518a6560 : SUCCESS in 46m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a7279effffe54f7899da0db6caaf4332 : SUCCESS in 1h 51m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0711a068a4204b4d84f3a858900d7f2a : SUCCESS in 35m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/b4bbf3035750499780e9efea1556110a : SUCCESS in 35m 58s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/8805d841c25f4b75a7dd1bb3aed8d432 : SUCCESS in 47m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c88e293792c249179827d1c9d82f5136 : SUCCESS in 51m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/0fee66869f8c48d9a820127824527e35 : SUCCESS in 46m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/77b6dd00c89444b39926a6c1d4b7040c : SUCCESS in 1h 46m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/82e5d78ed2dc4e679abd00def06bb757 : SUCCESS in 1h 54m 19s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e4694ef3b38545aaaf8cabf0a520a9af : SUCCESS in 56m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7b1d660c16b8451d838dbdf5c2715afb : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8fcad463fc284a5a81a0ca0df14c7dd2 : SUCCESS in 29m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/30ba957346704218a9b31f4c84009305 : SUCCESS in 32m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5dc29e403c444303a7c4fec7c7eb7f8a : SUCCESS in 40m 35s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2d304dbcfb8485ca8b7af025b8d7ffa : SUCCESS in 1h 34m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc17da2ba04045e1a3fecec6c5b42f1a : SUCCESS in 1h 02m 41s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/989819617faa4d79b1819db86d6dcbe0 : SUCCESS in 54m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b9269d90d6b94bbd8bb39f17068024d0 : SUCCESS in 25m 00s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a5ffa9955cf343f8a858bf6db1ed6745 : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":24},{"id":"e3f5f46f53258974bf937efc316955de58fa3e41","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 00:38:02.000000000","message":"Patch Set 24:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":24},{"id":"b0cba20ac3f1ad7ff6181d7563e946a97e794e4d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-14 13:54:27.000000000","message":"Patch Set 24:\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"26687b5b69c09a428dff1be7ee4fb189e84b76d2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-14 19:02:58.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"8f05466bf1e07f64389c5b9e18c9dd37cc8bb74a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-17 15:17:33.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"f1692b0758e2fb8a9c38bdbc057b24bdf9e7fa3a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-17 22:44:27.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"5f26d541e16476e10b3b69b7d4a546681d4c5266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-17 22:45:10.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"903bdf13d22df3d86037dc08462c4fe23ac86367","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-17 23:16:12.000000000","message":"Patch Set 25:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-dsxs7/index.html\u003e","accounts_in_message":[],"_revision_number":25},{"id":"04ff4a0ec25417c0cce3c5d625b107ecd9c5e1a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 23:21:41.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908d980ada434207a14e8b62f2a559ee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7fe524195f74508a5147b4342a34ef5 : FAILURE in 13m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec34c9488eea4a6c8a177e1e30ca5d2e : FAILURE in 13m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8e2f43a1e6d64e29bd766b9722d4111a : FAILURE in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"e4f814235b54f238fcb8066ae952c48f3a983dc0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-17 23:34:35.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"881675570a0f4481282c49aed0448d68a90728ff","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-17 23:35:22.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"e89b27ed46613dd1b57e2be0f44ecbec67c55467","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 23:40:09.000000000","message":"Patch Set 25:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":25},{"id":"b69594cd59a92806a7e375213390cdbdd4f71455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 00:49:54.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c38ee3d8e8445028ab66c35c75ca48a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f1ac499dc3dd4b4281696c932486ee21 : SUCCESS in 5m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8896b8cfa6ba486b965e685da92862dc : SUCCESS in 1h 00m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/86db83e4f1cf4ee89d4557ec31722a31 : SUCCESS in 1h 57m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01ac394816bd462fb20f36e84f745dfc : SUCCESS in 40m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfcbc4970baf4a13a204f72fbbfdb5a9 : FAILURE in 24m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009332b1009345f999ebccc74569d4f5 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/abe7156c3b7140cd9d3740cb1a851e47 : FAILURE in 13m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/201e906fa9db42e38b8939e9a9729cbd : FAILURE in 14m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/477961f36f994094a6041dc3fb2049a6 : FAILURE in 10m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5fbc525d8a4eba8fde21e56a532259 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fefb954d5274f49be53a7af4ca103e5 : SUCCESS in 8m 53s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c23f25d58776437f8340693f4704814c : SUCCESS in 1h 03m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f92ffdfcc11c4474b14a46af058ef0ff : SUCCESS in 1h 25m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/077845c2ada94f7cac20e13289cd9a7b : SUCCESS in 52m 51s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f3a0411037104f20b11758e906a69d35 : SUCCESS in 35m 58s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4883be306c1c41558f4c0210028df5f8 : SUCCESS in 1h 04m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/65c3f92552d441458a6730b135714515 : SUCCESS in 51m 39s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c607ef96960b4b689e5b497ac5acd981 : SUCCESS in 1h 08m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/173033f26b97430cbd711040d8943ad1 : SUCCESS in 1h 45m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/7e12f90e9714412887bdc9a799319b02 : SUCCESS in 1h 07m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/be0a805d0f2646e5a20f232056f31320 : SUCCESS in 30m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/48ad1c638f9143938e7f0e37f1907e9d : SUCCESS in 3m 26s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2f82789dbd92465585373b4f3e69ab34 : SUCCESS in 30m 47s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2cc02cdad4b343cc921b9790ccde860b : SUCCESS in 20m 44s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1ec82b45d1ce4b1d940a5a72854f0797 : SUCCESS in 32m 20s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a54938a2d334cb494209c439246b77c : FAILURE in 1h 30m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6637dd7308254cdf9e7d7a9689fd5d5a : SUCCESS in 55m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1daa28d58a4a44bd9fb8b82f066d572e : SUCCESS in 29m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d65a7984b43e4f2d8960d37d7643e38f : SUCCESS in 35m 09s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fb990b7c9aa544e4992d2fc573e90db4 : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":25},{"id":"be63ffa0044fc54c72812f290967802ebfc82cd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-18 01:02:17.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"76e18ce3c44d43bf9bd990bbc371168834394228","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 01:32:40.000000000","message":"Patch Set 26:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-g4b2x/index.html\u003e","accounts_in_message":[],"_revision_number":26},{"id":"158d03bed326c80a0db9750d1760076fce98b829","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2026-08-18 01:37:31.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"eacb7e87ff5004476e0bc5671fe031b0be901cc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 01:59:20.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da0f1b21e4e74b9a894c1cc9e0bde3bb\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3597bc34bd4f491dbe19813a75c9974f : FAILURE in 5m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/abc3a722f485400d9a28ae8f24c89c2a : SUCCESS in 14m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d91b0a097db14f3284e1fee20d80645c : SUCCESS in 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c0b73ea2cd1c8dd4b86c3aac2a3fee9fe3acf655","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 02:01:51.000000000","message":"Patch Set 26:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":26},{"id":"5268d793ac0e673550f8d4143c269d2f82273449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 03:03:58.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/942e91d6ef5b4d8e901ec512571ed725\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7bc949e878774defad1fd920fa208356 : SUCCESS in 3m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/19f516de94e44bf9a025e36f28877fa5 : SUCCESS in 1h 07m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/81873ca2c8e946d1812bd02ab5c323d8 : SUCCESS in 54m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7869dac1b3f5475c8d61ba00c4782015 : SUCCESS in 33m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69de9792d17041ec894599ff5379602d : SUCCESS in 52m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3481a1ee0c064055af4de76b4495e491 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/426537ad0bf4457d951dbd1a272e8bfb : SUCCESS in 10m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d723ae1d729f46a2836136ad7ab70614 : SUCCESS in 13m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33e164bad22249acb080ae280a0ed134 : SUCCESS in 17m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55addf724a004b53b3c51a0c04fd40f2 : SUCCESS in 14m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e3c68d36aad43349f281ace2aacec9a : SUCCESS in 7m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/128410eb30524766a1b57f34c19701db : SUCCESS in 46m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/989b6b9f91f24adb9c3efff0c77c470e : SUCCESS in 42m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0a5bb9e4a963476a8e273b4a6e061fad : SUCCESS in 50m 29s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/3451d55e1c3047f198836e33c463ddd4 : FAILURE in 11m 09s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ac5dbe249fa24aa6899d9a0ef19dc8d3 : SUCCESS in 59m 38s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1daa44e56a2b4a859a64ff4b7ab67115 : SUCCESS in 26m 39s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/af732fdad50c44cd905ee474c278a749 : SUCCESS in 42m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9f81339065754998955bfb3d833fec40 : FAILURE in 1h 51m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/e5458c95c1854e6584bd9a5d973e449d : SUCCESS in 1h 56m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/128343efa10f4c709187c05e021aac48 : POST_FAILURE in 18m 19s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/beee7f12cc4448a498404554ce7f8496 : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7460de953dac4ea181b83feba9250746 : SUCCESS in 22m 33s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4c09cbdefbc4454ebcb5abca2849d988 : SUCCESS in 28m 55s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/302fbc9595604f839190e0a1f116e6e8 : SUCCESS in 33m 50s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/65afbf16a1a04ddea5e50dba6d56c34c : FAILURE in 56m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cb1dd8e412a48a4af11b7012495427e : SUCCESS in 1h 04m 05s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f45c3d46c3c64a7f80457ba52acd0775 : SUCCESS in 27m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/67826c87257a4aebaf07ed4ffdf4c4a4 : SUCCESS in 32m 32s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/152c5e964ee2467c82623c099416690c : SUCCESS in 14m 23s","accounts_in_message":[],"_revision_number":26},{"id":"34600d6cbcf2b97235a72759c3c041d63c2fb1cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-18 06:15:52.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"3c07fb8dbf88b4390f8575211f1a24269ee1f494","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 06:46:29.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99498a4a00744b28cad3acee9a86ec5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5edd6ac6cce048e2ae488f90295ed7b7 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/145007efa4bd4afe89b04592e0daf4f6 : SUCCESS in 17m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0577b3047f304e8199dcc1bd71f3a4d3 : SUCCESS in 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"17b7d392a743fd62f75c6b977bb7cc0ae6e789e5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 06:49:39.000000000","message":"Patch Set 27:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-bjw72/index.html\u003e","accounts_in_message":[],"_revision_number":27},{"id":"4c9ac41a69634d2d70ec0f96095d2aadd6d088c2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 07:11:17.000000000","message":"Patch Set 27:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":27},{"id":"405b52713ac2d8e5d264e3501488cd980dbba7c4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-18 08:22:29.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"a0636cf58ab474f80a86474ebc0f7dc71498a4cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 08:30:23.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a25565d417be475a92b1e8d950aca21f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d345c80909094a9eacb621b9ac3083b4 : SUCCESS in 6m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ad80bafeb054d38a06cd52fbeaf6227 : SUCCESS in 35m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4b46630121614bee8271378d51722ef2 : SUCCESS in 1h 47m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42ab24af562f4729955bce1d02b767b1 : SUCCESS in 1h 03m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf54b796f1f4c5f8292bf791c9a19fd : SUCCESS in 58m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b3c52cbda04bdf942e23905239d099 : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4450b0103034e9cbc14dc67aa98cd9b : SUCCESS in 13m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c19201743354ccfa1ae9e11805e2f7d : SUCCESS in 9m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/624e0b995fb0431c98d8380309347782 : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f15d07819a84585bfc4cdcb3037577a : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f87f2d9c765846599fb811b4608e7d39 : SUCCESS in 3m 49s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/acae8719c3b04cd9a88d55a3db376b24 : SUCCESS in 38m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0a6521effe6b49c1a391b3b440c6d5be : SUCCESS in 1h 10m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f16fc23d081641d2899567ccb2e157ea : SUCCESS in 51m 40s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d6192f587a0441ec8a940bfb3c41fbe2 : SUCCESS in 33m 58s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5c24cdf2dba9461785c61f910098572d : SUCCESS in 59m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9ddf37d19a47454294887e682337db3a : SUCCESS in 51m 38s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3641830cb3734baca0fe66a57814da1a : SUCCESS in 1h 03m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e1ba0b6cb0649b6baaad97d4a5385b0 : SUCCESS in 1h 44m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/957f3396d9194f2387c9a678d1965ee4 : SUCCESS in 2h 07m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/357b160973304d329b88cc0518607cc0 : SUCCESS in 48m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/52f0aaf65b67435f918a09a31de4fa53 : SUCCESS in 5m 04s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a1d05e1eccab4c148131c2470c91fecb : SUCCESS in 26m 13s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9ae8c830f7cc4383902342953e171128 : SUCCESS in 25m 58s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b7001cb34ace4459b8b2d39670c5ba38 : SUCCESS in 29m 10s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a2c49fc7fc04a23869bbe28f810ee3c : SUCCESS in 1h 31m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35f7aac9a22d436d88a0b4b20a4f0c2e : SUCCESS in 1h 01m 07s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/17edb4813181446ab57d90cb4bc4ddf6 : SUCCESS in 39m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/63c22d5e922045e5acc83fa1109dd318 : SUCCESS in 18m 43s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/da637dcb8b854a21bc80b4eba12d8888 : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":27},{"id":"7d76f74b3e1ef99b83fa2142bfa8e4ef078bd68f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 08:49:15.000000000","message":"Patch Set 27:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":27},{"id":"9655e3a5fe0ecb4fd02120f140709b2781f6cbd2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-18 16:41:19.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"23a9ec32ee20c7f3a460c5923673aee19674e364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 06:06:20.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":28},{"id":"333277d864680a7f7454ea5b594025d2264fed22","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 06:07:34.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"4adc44052e5f125425e9c5a19923b547c3c3133b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 07:01:17.000000000","message":"Patch Set 28:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":28},{"id":"1f80f7358c3d55efab261f58ffea08defd6e3a13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 07:02:59.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0f85225fc904d738cc2fbbee6acd1d0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30f629ca45d644429f712e8de133b84e : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c7b0079c7e3047d1b4ac602828284427 : SUCCESS in 15m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/322ff61db5734c70a73eebc138fd3209 : SUCCESS in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"bc5550170e17716d9ff152f24e8cb3908ab1ade8","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-19 07:23:02.000000000","message":"Patch Set 28:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-w78x4/index.html\u003e","accounts_in_message":[],"_revision_number":28},{"id":"23abe9b988da275c4284d6061f6dd09977a9f914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 07:59:32.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09e808c185234f07af557ae107b5172c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e58b9880e2c74aa99ef658b6be163b76 : SUCCESS in 5m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ba4f211411534793bcbf099f464205ae : SUCCESS in 52m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c2b0698090874d7d8184a06331000c06 : SUCCESS in 1h 23m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55676eafb73f4124bf0f0b406cdc32f9 : SUCCESS in 1h 01m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4210805f8fe54e2786b4a98c2f397dc7 : SUCCESS in 1h 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87eef4426b2b4b1cbcc2f47dc3af82f9 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/360c415ae8ea427c8269bdb5c8fbc351 : SUCCESS in 8m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ab322c85f5c843059bef99aeeb477d0f : SUCCESS in 14m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6e35f37bf7a945d9900bbe7b3c0b22e6 : SUCCESS in 15m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4db325a3e6422fa800d2dcf55cd868 : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33c6f74904f24feb8a397d454f7b8538 : SUCCESS in 5m 07s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/76ab332c70f3432aabe9a6a60b2662c7 : SUCCESS in 1h 06m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85de17c60fa6416fb7d165ee7026dfd0 : SUCCESS in 1h 05m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0de8422ec5aa42f5a551ea7fd81f547f : SUCCESS in 43m 12s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1e942f25b059485bb7efb6945a1be23d : SUCCESS in 37m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b1305570a49e464b8b7349ed254f79e8 : SUCCESS in 36m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1be73168a6e3478faa32bb81c534f1fe : SUCCESS in 50m 23s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/ed6e73910b9c4c5ea9c4a388cbbcc905 : SUCCESS in 1h 09m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/674e15069d7748998d0270fefd1e5ea7 : SUCCESS in 1h 44m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/b2ca7cf6f9be4bc5a4b958052156923c : SUCCESS in 1h 48m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/94c6d716118a4d22b7a1d01865d414f5 : SUCCESS in 29m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fdba9c39a0494b64b3be186aea569b0f : SUCCESS in 3m 43s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/147bbfa256384f6aac90ff065230b84c : SUCCESS in 24m 36s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0eec774a9e0e44b4b52c5f19d2aecce2 : SUCCESS in 26m 23s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/00c87a513c0b417096acfb53e79ed966 : FAILURE in 34m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9040a6d7ff8d4297a0553516efa7f066 : SUCCESS in 1h 34m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6112c3ef8e584ff8b0570acaa8406256 : SUCCESS in 45m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/15c8ae5696c145b4a5c049a620670a17 : SUCCESS in 55m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4223b091786244868d9084912d5104e6 : SUCCESS in 20m 07s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b969a1a87a334a7992af3bf406c0c020 : SUCCESS in 11m 18s","accounts_in_message":[],"_revision_number":28},{"id":"e4b7f803ec6d5954e6608f84b87cbae632690375","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-19 18:08:16.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"3a0ab47ea062f6c38470b1d85ee26e2e3970c5fd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-19 18:30:41.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"a2a64f3b528e68e9d0912de095c4551d5abbf79d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 21:14:54.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"f4608e68774afcb726cb8769bb42d0966b32a4d1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-19 21:21:38.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"0d286d4ec394e9a15a46b8154674ddf37d4d00aa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 22:45:25.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"ec424669d80008d4bd6807a3318730a1233e6963","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 23:21:50.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"411bf0f80032730f94227b822efb9dfcdfed2108","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-19 23:29:15.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"65d9afe1ce24c3cc83dfee353c7bf872ab1e711a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-20 16:06:27.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"af57215634231c1b2577d6ff91849aa3ddcec69f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-20 16:40:19.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"d49c55c44979756af80d520ca842b33e24f4ec98","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-20 16:52:50.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"edea296d8b460d89368f45fab1dd800918e46d1a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 05:35:10.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"6162eb4aa3272db7d3147246466c12882fbc1473","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 05:35:56.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"48fe92886c03ec9806705fe9cd95f060c5d2c8d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 05:37:46.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"c55af2696cdbad39d3d842d434f8fe0d5cfcb26b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 05:39:31.000000000","message":"Patch Set 29:\n\n(4 comments)","accounts_in_message":[],"_revision_number":29},{"id":"962ddd3d41e825df1035980be4eef13cfd1a941c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 06:04:31.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"fa3d7810980e75564c28d97637fd4fe984e90c2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 06:28:26.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbd665b21f6944c68fc3c83cbda5911e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e72dbdf3c0a148a5a4e07997122867fb : NODE_FAILURE Node(set) request 1031634473544008a5b705e49a8de513 failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/89a0da26a1fd444ab3509dec7e0a17b2 : SUCCESS in 15m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/cb0b6cb5d4b14eb4a7b3663c15ecb70e : SUCCESS in 38m 18s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"b4c7f01b7d3599cd6025755cf0d8e5394b413073","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 06:33:50.000000000","message":"Patch Set 29:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":29},{"id":"1b12ee0983f2b82ac1a5f76ee052c34caccc231f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 06:51:31.000000000","message":"Patch Set 29:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-6v87c/index.html\u003e","accounts_in_message":[],"_revision_number":29},{"id":"17eb2bc405066dd2eab4ea0795566b49dc83c388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 07:46:30.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa142594a01f4eb68516ee2f1e23d3da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cb9453d8894347e7b3928b2fe0c6bd7f : SUCCESS in 7m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e855a9769c341059344ebfcd220591c : SUCCESS in 1h 05m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/352060b8f9604ca1a912cbe1aaa7d46f : SUCCESS in 1h 53m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e891e95789434651b834f2c560c4013f : SUCCESS in 1h 01m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb0dd6bcf7354b449fd1878590c5d37f : SUCCESS in 52m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70b03fa5219844b9ac9b7f561c9c7f01 : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a650f2c143544c6a15ad706af2b2254 : SUCCESS in 14m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4890f39e8ef440c0a8f92d522483c467 : SUCCESS in 12m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/16b2f8964f8b4d1a8b23af3891647d25 : SUCCESS in 19m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cfbffb6019b48acb73535a2c3d0f8cf : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f9fedd3510243d892a997dba43510d8 : SUCCESS in 3m 43s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4ae4cf2186204ba7a9964d81452c24a4 : SUCCESS in 1h 08m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4c35e52418d432a9bc15b482c9e5632 : SUCCESS in 54m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5c435b661cd44a11b2d10ea34897707d : SUCCESS in 41m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/88666344b8474b48a3c3e6ad6442dfe0 : SUCCESS in 35m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/970b457d8d4f4b2e8326a69be978ccc2 : FAILURE in 32m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/65fa2dc226ba4317b99b3961cc509440 : SUCCESS in 52m 59s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8933e2872f7249da89a573f0ab226776 : SUCCESS in 1h 01m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/81445c712d87428b92503ec3dce802e7 : SUCCESS in 59m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/265c6dd661e84ff89f6739f8c2d336a5 : SUCCESS in 1h 48m 39s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/25ddaf3f20e44ab18d5f12682266e36e : SUCCESS in 49m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4d7c406512d74e9d862d95b86ea27dad : SUCCESS in 4m 53s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/004d817f26274ff2b8e07e5687e9a57c : SUCCESS in 31m 17s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc7fe9f63b0c4c868cea5125a2a16b23 : SUCCESS in 26m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b9bb1149b6bb4da9a3ae1cd7cba50af8 : SUCCESS in 25m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/940ad9dcdba2428a9ce3425112d46116 : SUCCESS in 59m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/946ca787c37a4967945ba8758c0223ff : SUCCESS in 44m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9b714b5154494de4b92c197d847ba63c : SUCCESS in 33m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/46de83e1a5874a1f9bf4d4ca35081427 : SUCCESS in 36m 13s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1cd86b957d6d4ed4a2e878e27e130819 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":29},{"id":"fe46643fcb1725c6e84785bc3dde021a8de8971b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-21 15:20:30.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"a27abd5f08f2144803195901233976aa14def68b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 21:46:48.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"da59ead06dfce0228468f341da4af68955eade67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:47:09.000000000","message":"Patch Set 30:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 989633,30","accounts_in_message":[],"_revision_number":30},{"id":"228aac1e3cf6e744201ebec62862c5a7e28adc7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 21:47:16.000000000","message":"Patch Set 30: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/nova for 989633,30","accounts_in_message":[],"_revision_number":30},{"id":"aa73c8c61efbc85b205cda56877d82fa61926c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-21 22:28:10.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"bd68b74bbf25d1b3da020d50ba9069b40b16038d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 22:42:50.000000000","message":"Patch Set 30:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":30},{"id":"4f374a085ebd7133ab4507980718840fc0f5a4ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 23:05:39.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11e468b0e99746e9a347204ee97844d7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9682cae5b54416f86270f05fce35dd7 : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55e5f5d6caa94e1f9b788554da046d2f : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7e3556a4ca69461a88ed10aeaaa89072 : SUCCESS in 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"82b91ec80202bf10a60c7faa391599de3ce6f1d9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 23:26:44.000000000","message":"Patch Set 31:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":31},{"id":"61ab5cb3e32a0bb3d3560480478e0b770da9ef97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 00:26:43.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2abe685992954165946436c70a3e8247\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/87fc3f6328c643e0bc558045a6379126 : SUCCESS in 5m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eb8cc8001bc4459bbfd3fda6f2ce07eb : SUCCESS in 59m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a7fcbea222974410b3d85bf495cf59a8 : SUCCESS in 1h 19m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c75878fd7bd64efb882f3a22cb6499cb : SUCCESS in 28m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0d9732a39164a21be486fde5cbba6e8 : SUCCESS in 1h 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d89170924b464dbe227e8a16c77edd : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee78c3681f46491f9ff520465f538385 : SUCCESS in 10m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2d85b7e6377480893e8594f45b9e1b9 : SUCCESS in 9m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/33743760affc4b6e8f82914a6ae929a5 : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/290f707c4e4e40ee975be4f9a75374d0 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ab24f9315f54c6a9bbde203a1ea60d5 : SUCCESS in 9m 19s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d0735f91e99c4c81a8e4485f5e737614 : SUCCESS in 44m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/53cabac5ebb94ad683d3531e41af4a37 : SUCCESS in 1h 29m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7801c5fb92644033b564a87ffb688324 : SUCCESS in 49m 32s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/4edc5a034838458b931602d766c58f74 : SUCCESS in 35m 41s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0b293cfabf8f44fdafc992f8310e39aa : SUCCESS in 48m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c1327a450aa846e68248f5ae03115645 : SUCCESS in 53m 17s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c64e07796f584e35b3b1a944f8adaa9d : SUCCESS in 1h 03m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8260e5a9c74a45ea8de2a13557c2d62a : SUCCESS in 1h 50m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/d6aa061280bc4516a67feb6e47870679 : SUCCESS in 53m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/726f8ee356c14605be512d5d3e5d8a1a : SUCCESS in 51m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5a261478d1b542f0a93868dfa18e7230 : SUCCESS in 3m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/79bc80d262c84d379729414750a76fad : SUCCESS in 29m 17s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1e60a6743e67415490299c796428ca6d : SUCCESS in 26m 31s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/41c3893678f14864a65e96f3eb7e347a : SUCCESS in 33m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/173e3b62916a4fc7885164054ae18cdd : SUCCESS in 1h 29m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf596e6c51aa41c8b80fd348cf0a7c91 : SUCCESS in 53m 13s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8f40261164a3480487d1e724f58dc64a : SUCCESS in 51m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/145077f5eec0470fbc362d14c94ec736 : SUCCESS in 26m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8c793143c1c6482587f1038e2d767081 : SUCCESS in 8m 54s","accounts_in_message":[],"_revision_number":31},{"id":"48bb964e8595afc03d1fc3c612e0094f065a737f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-22 00:37:26.000000000","message":"Patch Set 31:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":31},{"id":"e75467a4ab4f60246338ee702d91517e69b20051","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-24 13:44:24.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"b03b435f6b053b041f35b990eaaa692da70b2b08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 00:45:08.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"9d372a0718c628f92670c8edd36de366dbfef847","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 01:37:37.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d573f49cf9034fcf9c5a1364ae3b59ae\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6223261a9fc64383b4831143eb4a38a9 : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/249b5963872c46aa810095321ea1569a : SUCCESS in 16m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1ba8b2ba3eb647a5a6286ae4e9f5f348 : SUCCESS in 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"0595c8cb769e0c7780650baf809f1c230aac3aae","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 01:43:32.000000000","message":"Patch Set 32:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":32},{"id":"7c767e0c90caefc3470b2a2bb3b8fabe14646c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 02:39:54.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f27fecda28540aebf647cd996dff6de\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/aa0e532e0e1a4002a65aa0033d8f738c : SUCCESS in 3m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/712d37741f4d419fac3c66c4a16f9097 : SUCCESS in 33m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0bfd8678bb87411dba2811346d5d8dd8 : SUCCESS in 1h 45m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c39ec0173be94c6287a1813c924a35ca : SUCCESS in 1h 04m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/604514c238f243918fdcf4441f296ac2 : SUCCESS in 1h 15m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f860cbf06a469c97dbecec59a8160a : SUCCESS in 7m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4311e0e8b734e3dbbf4fff3c161bfff : SUCCESS in 11m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3620c2de948648c0be4d612010c0983b : SUCCESS in 9m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a5b2b0489d5244deadcf74c09ccec4a2 : SUCCESS in 17m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6fc42c7fb4483cb360ec56fac2caee : SUCCESS in 12m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a747100a58847f2b16c0dbd83ce46fe : SUCCESS in 8m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/37567e3e8c7a49eab317439ec0e5f7fd : SUCCESS in 58m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bbdc02f29948406bbd6833fb2fa72835 : SUCCESS in 1h 19m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a9b575ff38d0450aa0430e1c1363ac65 : SUCCESS in 27m 16s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/08489a87d8614eb6b9cfd29aff16721f : SUCCESS in 39m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1bcf198fe0ee4ad985251b0f83c9f041 : SUCCESS in 1h 00m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/371e976969754289848c2fd61a3e863a : SUCCESS in 52m 18s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1c5d434fe5b249cfb9aa8568777e4df8 : SUCCESS in 53m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/292a7531de624bb9bd41fa7827619f94 : SUCCESS in 1h 48m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/a113942cb5ec4feca5022f587c3a833f : SUCCESS in 1h 37m 22s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0e6ff80559a547c9b7b141752908191a : SUCCESS in 46m 48s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/fe92d9bb70ac44bf9a57dee444814f03 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f03e223211f74c4e957db24cf6922e5b : SUCCESS in 28m 31s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/629443c9019a4ab7bbe974915a416cbd : FAILURE in 29m 11s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f1228fdb8bbd4e2985484cf84947bd2f : SUCCESS in 25m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1edce364b06d4797aa20d7f557dd0720 : SUCCESS in 1h 16m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47b7a340155742e6bf0ad76216ea8435 : SUCCESS in 1h 04m 15s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5668e380a7904801b2901847df95b133 : SUCCESS in 50m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b6238495998e43ffacabc2bd091d13a0 : SUCCESS in 33m 30s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c93f64bc0c104a9fb380b854deb67427 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":32},{"id":"499af91dd5f498a8aabdf7ed86ca85988c3d371b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 04:42:47.000000000","message":"Patch Set 32:\n\n(3 comments)","accounts_in_message":[],"_revision_number":32},{"id":"3ea2253b1b3ca74116b701fd09c794350ba3f1a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 05:47:31.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":33},{"id":"665e1f253197edfb6ce1404f051269581cc83271","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 05:54:30.000000000","message":"Topic reconcile-stale-share-access-patch1-patch3 removed","accounts_in_message":[],"_revision_number":33},{"id":"32a69b60621a5e5a218b068f1781955f94acec13","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-25 05:54:35.000000000","message":"Topic set to bp/virtiofs-cold-migration","accounts_in_message":[],"_revision_number":33},{"id":"22a336005525cca5a34fb02afa3ceaad903be6c9","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-25 06:47:22.000000000","message":"Patch Set 33:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":33},{"id":"895a1eb491cc315158c4d62c42552ddbbd8db0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 07:42:33.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b22f8940f18a4057b2441cc311b9845d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6c7001e4f75040e288c4fe72d64fdfeb : SUCCESS in 5m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3d721b109d0b4d949d1685e1547ed9f9 : SUCCESS in 1h 02m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fb32ac432988425294d7e47347a4397d : SUCCESS in 1h 46m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eba63e6127e04e7f91120dd28734484c : SUCCESS in 1h 03m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae9f7dafdfd4c17ba5c4c1c53b15536 : SUCCESS in 1h 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f0edecfd4104b46911cd7b5c08125c6 : SUCCESS in 8m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de9caeb87d694b3486e7ef7f652091dd : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2f14caaac41b45aba404250a16ab376e : SUCCESS in 11m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3861c4b6cc8e4e4b955345c9f3a00d6e : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff13ba57aab4e64b152cb7703474071 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c426ce1da8814ba7b92d44f74722ed6a : SUCCESS in 9m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/247ce14b10e94ae583ba223eeddb0587 : SUCCESS in 51m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a89582a0a7794a85a3046ee442b5943e : SUCCESS in 1h 39m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73bc4664b9ab4b91a47c9c5faab3cce7 : SUCCESS in 52m 53s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5b2228d254ba43e4aaaa0157bbd325b3 : SUCCESS in 39m 11s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c027692451094ecbb6b34a4b300ff713 : SUCCESS in 58m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ea61aa48c5994149b3ff6e86e4aab16c : SUCCESS in 59m 18s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8783cb2b199240abad42a0e4aaf37a34 : SUCCESS in 56m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1275a7cb6df3481bba4b341aa1bfb925 : SUCCESS in 1h 47m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/1799b4d3fc784b158818c0b9a8ed15f9 : SUCCESS in 57m 32s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8868e43b4c9c493c92f62814eabdb7d2 : SUCCESS in 40m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/52e41a3228514340a89e01badfddd76c : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/72150ff52cf34c008349b77883074a7f : SUCCESS in 29m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/323a8c6be0e1413699f0d5ceeb59e5bb : SUCCESS in 26m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/635d443982b04a91b82e9729d89b20dd : FAILURE in 38m 54s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/baa093be9e5f4de9bc3ef9d823f0b014 : SUCCESS in 1h 29m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96e5fcac9d1e44bf825421f7923ec455 : SUCCESS in 1h 00m 31s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8f318f31ae95498691ddec8bd27be587 : SUCCESS in 52m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3fd66ca8a57e47c9acf2af186d1f7931 : SUCCESS in 17m 17s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4bdc6b4f05c54700b3b7fdbb716a1318 : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":33},{"id":"1fdffd5a8e0a5513eeaaf13d4f6f4ba92dd0aa6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 09:49:03.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ba2775c399a42499ae5d5bc59b56bda\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ccb17d28a0214d89aad69959d21c6c5f : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/955362975dd545f0b3dd4ef0e683b7b3 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3728dc19b0744bcfa08fde467b751618 : SUCCESS in 45m 20s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"9d0217484dfc5146bc52d8b6ffccc5e71c14c753","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 04:33:57.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":34},{"id":"4ae036b4af1c3877f94b5f70d8cdb6d1f87c6208","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 04:40:59.000000000","message":"Patch Set 34: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/066c8ebf3409461f81fc2b65b4f507c6\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/130ad1c096934f569f03c1815b82e483 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"d6d2592252c41aee59975c749cd3f6875c646dd1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 05:29:21.000000000","message":"Patch Set 34:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":34},{"id":"0098f9185127260c166d5b0230b5b33466ed63c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 06:41:18.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0c7de1e79bd4daeb85e2eaa4e09a819\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/98cdfdee4f2e462385f5069842af2108 : SUCCESS in 5m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0fca9e1999dc4d0ab72ecbae2b9e9d2c : SUCCESS in 30m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bb9a339533a94c89b461af301c08d306 : SUCCESS in 1h 43m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2832c02d5dad48719f49738a7466460f : SUCCESS in 1h 02m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cd7eb4e1b6d4acc92dd5e22a19d1aab : SUCCESS in 53m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e575f8192a134b71b83115e6302536ea : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69ca5606c6754f688b96f9d7d74f5332 : SUCCESS in 12m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b9c8faf8ab547fb83f1e9634f3c9307 : SUCCESS in 10m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d07ca43e6d1490eacbcdb6e83b75721 : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/628f5411b0c6409994d1fe45f4c64e13 : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44494c7167d44195a5a5c8bda46b1678 : SUCCESS in 8m 26s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3d3c3e81bfd14b48b0e4940efb4e9db8 : SUCCESS in 57m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/915249231a6640f890eb1cc8fa207365 : SUCCESS in 1h 40m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c1c8a5f5ccb440e084605b0a7bb27a0b : SUCCESS in 54m 11s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9a60279966274641b03d94c39c2fbff8 : SUCCESS in 36m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0b5b02a5ce274ce6a1bf1cab554ef8b2 : SUCCESS in 1h 00m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6bb461b33b534f4fb0605f46add793f1 : SUCCESS in 24m 43s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/ca5ad40cd6cd4b8abf884f6ee6776ddc : SUCCESS in 58m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b5aaf65efd0b44ed93de50b149ac495a : SUCCESS in 50m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/c928f4dc0a594dafa4f8d75dccf90c00 : SUCCESS in 1h 53m 44s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4a1c8017ca0b48c2acf210f0161594a4 : SUCCESS in 44m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/92147c2a85e4471ab057a50712b36267 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/831a1015e18443ec9198b2c08000201f : SUCCESS in 26m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/104ca44c018747baae905db918c4cf5a : FAILURE in 21m 46s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9167a26427d14330b546aca96471c13c : FAILURE in 35m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/72784212e3b1404a83f08e6799852563 : SUCCESS in 1h 07m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2972c8c9c8944f6f8a775024507ed384 : SUCCESS in 1h 02m 46s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/00c06ffdfbfc4895b99b6919443e2836 : SUCCESS in 25m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/20fb6149bfa34f9fbe016b4cc46f8e4a : SUCCESS in 35m 38s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/84da7d49486c4b4e8096f3f7ecdd45c5 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":34},{"id":"3dc799ae3dbeef5ffa5ca0267f9e58539ab953ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:03:08.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b403dca716b4864867335dc9d89112a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc3d22e01839491b9f3979936bf72393 : SUCCESS in 25m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dd54cda49a9a47008175ac4a0d59b580 : SUCCESS in 21m 32s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8bda967cf9704f0c9b38bf60634ea01a : SUCCESS in 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"26f1917044967550ddb69d710410bb0c4f5ea8e2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 14:00:53.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"a121749b700f345ba456139d493973c58c65a764","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 14:01:02.000000000","message":"Topic reconcile-stale-share-access-patch1-patch3 removed","accounts_in_message":[],"_revision_number":34},{"id":"e6f75828cce9f93e15f150397e1cc81132c372a8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 14:01:29.000000000","message":"Topic set to bp/virtiofs-cold-migration","accounts_in_message":[],"_revision_number":34},{"id":"146971c2950422dc6e500f0cbc79b51929550604","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 14:08:20.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/2dcf13e2dd8a47178be7dbb67d01284f\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bf4148bbbff942c2a19f96d9cd40efed : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"82606842b3b2dabe02d4cca64873ca1a8db75e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 17:08:48.000000000","message":"Patch Set 34:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ac16997881e44bc944a0443ddbb196b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/afe4755ab57e4f81b17740d37e84524e : SUCCESS in 4m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a9798e720ba4da796e0a3e7c5bfaa1d : SUCCESS in 1h 13m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/876f763066c840dcab559be73e3677b6 : SUCCESS in 1h 16m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/826136d54627413f8346bcc3aa2e5f3e : SUCCESS in 59m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f7656235dcc47938e6f58d2a24e148e : SUCCESS in 1h 01m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9275c0fd5e35491480b83cf72ec4da7d : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/332352d73d56485988c25176032ec10e : SUCCESS in 14m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/631c1532ab0b43da8200a053b360057b : SUCCESS in 14m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73555e549cde4b148e25a506b5fc463f : SUCCESS in 19m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a070cfcc89be4d7c8ac8785ab90658bb : SUCCESS in 12m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7021f339876e4a73b36e53aba4af9eff : SUCCESS in 8m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3a28082a1a05402198254f3a8258aaaf : SUCCESS in 1h 06m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a84bfa78e9cd413c8cedc045febfc368 : SUCCESS in 1h 28m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f10bc672c89d452383f32a32a27bf354 : SUCCESS in 51m 51s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/40cd579ebe6640f7b5743960c9dfb72b : TIMED_OUT in 3h 03m 11s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f89529de68e746468390ab24c9d3379c : SUCCESS in 1h 02m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/55830d3338bf4857aa78033cc7c787d9 : SUCCESS in 44m 25s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/821bb6e4d94647b4b0630b5d7b3423c7 : SUCCESS in 1h 11m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a1eca92322654165ad298b84ff92a5fe : SUCCESS in 2h 04m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/c82b49af2e5044edb1a93a61f28aae59 : SUCCESS in 2h 02m 20s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ac430d1a3cf44b73be600f1a90f09848 : SUCCESS in 57m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ad56be1cd74148bfa4aa045a431200c7 : SUCCESS in 6m 08s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/96e87796ca9743449e08f5c40cbcdceb : SUCCESS in 27m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8559e61766c0497c80f4d7dce073b74a : SUCCESS in 25m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ec267bffc23d4da9a435c409415c0cde : SUCCESS in 32m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/79419491d2f548ff9255c3abbeaaff93 : SUCCESS in 1h 33m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07b9f5b3f99848fda83bf2e06facf634 : SUCCESS in 56m 59s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d5e4ce61037549508c62dc2c09ce099c : SUCCESS in 52m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/810d513001c44d97b7ed475844bb9e15 : SUCCESS in 33m 33s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a4aa23bdd55a4cdbaa792d692f3ab890 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":34},{"id":"9f6fe09241cb6ee0c8ac2e7eb1ba9938297b7e06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 18:47:25.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f859462547645999e21ec0c259865bc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5f0e6f75610342d191ab792b44aed1c1 : SUCCESS in 27m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/804be82b8db544cbbef3d53b64506014 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/539eecee3dca424fbe83788c42888f4b : SUCCESS in 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"253acc14aa43b6b6fabd285639d726601d502407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 22:38:28.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":35},{"id":"4bdcc4ad4544fa6e627ff7216b612f676f02e81d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 22:38:57.000000000","message":"Topic reconcile-stale-share-access-patch1-patch3 removed","accounts_in_message":[],"_revision_number":35},{"id":"6e5b2cc7f6531ecf97e9629e8280481811d1d6e7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 22:39:43.000000000","message":"Topic set to bp/virtiofs-cold-migration","accounts_in_message":[],"_revision_number":35},{"id":"9587c0ebf55ca874578be5eb689218ecbe3317b5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-26 22:42:32.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"752136570109969dcd4ac12d1dc5de50a5a6afe3","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 22:48:29.000000000","message":"Patch Set 35: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/9144e1c645a6499d83d76567141e572f\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fc780fb0aaf04e70b5177c27090d1c6d : SUCCESS in 8m 09s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"3c78bb6a0f9d0512025dd9c6298ea28b8f928e8e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-26 23:34:32.000000000","message":"Patch Set 35:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":35},{"id":"29ab50bba424f213c8247a649d9d19e1a11bd611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 00:44:11.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8ec59fcdc54419abbe5a62a3c746dc8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6745e109cf754263b9fc8290aaf25aa9 : SUCCESS in 5m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7386e3da48254638b711a9a26cd61550 : SUCCESS in 33m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a5bb329ca185464d823f687430f2d110 : SUCCESS in 1h 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ba3b31396f9471b86e64967f860c90f : SUCCESS in 59m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e0b52f4368f4053af664e2082d44f80 : SUCCESS in 52m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34aadf2ea0c486abba44fd1bee9e7da : SUCCESS in 8m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e44f993614e45c8b5b283ee38ffc2bb : SUCCESS in 8m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e9c5f6aa18e4588907c463710f4fad6 : SUCCESS in 8m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c74e652e9e2749c4a2ea44cc1e33cd10 : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341bfd942f8e4bc0b04772a2a6ee25fe : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36618d6e028241758cb52d43627b322f : SUCCESS in 10m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a6bbf999a50d4998abf7c6c172585fdb : SUCCESS in 41m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8945652b28df4cec8e6b1913aea01daa : SUCCESS in 1h 22m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/43829f1d3d8e448a99cb318e19886e47 : SUCCESS in 34m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/cb29269c839341cfbf7e380a1e6546c3 : SUCCESS in 36m 24s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5b3a7d11efe04ace9f11d09aeec2c992 : SUCCESS in 1h 02m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/af9859f9d7c44e97849a97c467657029 : SUCCESS in 58m 28s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9944bbc1de1246299494b8edeb7f26be : SUCCESS in 1h 00m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/72ae1c0e4da0419ca6e8a27f91dd423f : SUCCESS in 1h 43m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/4ad91cfaf0744341b3f4b781754fc93c : FAILURE in 2h 00m 26s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/62d17fb1cbea4bfb967a4520445fcc75 : SUCCESS in 48m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f89af5efac094466b0b59bda3dff896b : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0f2f95a3142244eda1c8d81b5b64e10d : SUCCESS in 28m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e6c5d0340ad048bf88e54e891c3671f6 : SUCCESS in 23m 57s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9093b1a8a2374e7f9b30328302873a59 : SUCCESS in 29m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4325f5a97df4ed4b1a4709b91030aed : SUCCESS in 1h 25m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b540f84737b04c31b334bc94b954fa10 : SUCCESS in 37m 04s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d05e075298144eeeaf37dec1d6c500ca : SUCCESS in 49m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/403c23a0442e45ada9e7dfe0829c78f6 : SUCCESS in 33m 02s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0797256d06114ffbadb2027b9cf4e429 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":35},{"id":"22e6e39f3cd6e662a5941d2df86c469f4b65283d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 02:03:12.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce2e2c38be4f4057bbff603b7643f40b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db3d1de3f82e453b8db658cdcc0ee9e1 : SUCCESS in 30m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b08e330cfe08497c868c6cad8ae12d46 : SUCCESS in 33m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b69cc9c46c9646d1b3ea8025802102cf : SUCCESS in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"c2c241ac48066a10c5337b570aca028c9c0fb42d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-08-27 06:23:50.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"c048a18bf96fe652a4c53fb3ad93a5397e3a7a37","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-27 06:31:51.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7d2d789d16f74e74879e2e1255201b22\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/1b5c1509a760428a9c12aabedee1a952 : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"464fe506a309b350a9f3abc127c021b57aec9936","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 07:02:40.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b2345e1c7164a05ae98f5954fa70051\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d692fb97740a425a957f2cfb6139acca : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6b6896784bb745efaf7b536581e74f02 : SUCCESS in 17m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2bc6fa7460864089aa3135f49bbd2fea : SUCCESS in 33m 17s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"455d34398cd8f66640b11a2a25939b5092e0de36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-27 08:33:24.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/995b741a05a34b6792fca69fc185c78d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d30ae9fa73714a49bb34ec2722b781de : SUCCESS in 5m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/915f907ee0d943a7b0c12a744957914e : SUCCESS in 31m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/adf5cffae1d34217beb1d1f648133393 : SUCCESS in 1h 55m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5c7bd5e310d4bcab4f5d85b0d813bd1 : SUCCESS in 55m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7dfcee8a37b48db9112f551837f292c : SUCCESS in 1h 33m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b917dcd43b41358a7fe1ca04ee0f7f : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55cf161d9e9a44eebb6542fef3445b1b : SUCCESS in 16m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdb7e11c4c284fde811f9eb267680bef : SUCCESS in 13m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c7d4c4f06204e369f6890491d7beb7d : SUCCESS in 20m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05f87f45002548fe8904041a76e24581 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a1df9720a2c4eccb50fa719b4f48469 : SUCCESS in 4m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/26591dce2f4440c6a10238d332c117d4 : SUCCESS in 31m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd1de902905647e990be85dfbec97c63 : SUCCESS in 49m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a8bdaab4615540e4b25c7a7ac442a99d : SUCCESS in 47m 38s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/991ac3fa970046cfab05dcdd65841a25 : SUCCESS in 37m 33s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e4a5b6fac7a642ad8c66bf44e7ddd743 : SUCCESS in 59m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/de3a4eb54d9d437b9b011b3bed078c17 : SUCCESS in 50m 55s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/19f97e3c27ef496ba12b2ce238bdaca2 : SUCCESS in 48m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f593d9dfb0be4129b0b238fc6b0bc53e : SUCCESS in 52m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/6ff4b5e8684d4ea4b84fc08cda1ec825 : SUCCESS in 2h 01m 39s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f097f83255aa4b5f85fe5eb449155d1d : SUCCESS in 29m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/64f7cdd59a914c14b3afe1158de0c603 : SUCCESS in 5m 25s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/15cb663eadc741d79e9106669996dc0b : SUCCESS in 28m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/dc3b36e0ff7747e88b492a914dfdc3ea : SUCCESS in 26m 24s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f0ecf79226ea4cde9fc67795bbb97d3b : SUCCESS in 31m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9e748291dfb4becb92b6bf96b6248cb : SUCCESS in 52m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55b6b5b9dfad461f9f75c0449730157c : SUCCESS in 1h 03m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1b9db7dffd8b49d2ac9828f1bb3c4a9e : SUCCESS in 59m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3b289d80a9894d28b0b40efd1a6f4960 : SUCCESS in 34m 36s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dd03c129d7b94d3c9b64cfea2b6c1b09 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":35},{"id":"bb6153c5dce4bd30f70fd4fb43bb040d20b1e859","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-27 08:43:56.000000000","message":"Patch Set 35:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":35},{"id":"b12ad45ac0738c05d99c4b0176f5bcdb8697c157","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 03:48:17.000000000","message":"Patch Set 34:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-69xpm/index.html\u003e","accounts_in_message":[],"_revision_number":34},{"id":"6ebce28555724f069459407a5b2fb26f795a6b06","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 04:52:14.000000000","message":"Patch Set 35:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-xhn8q/index.html\u003e","accounts_in_message":[],"_revision_number":35},{"id":"989e14c00dc4b4ff70a29b79c298060e4e166215","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 05:24:21.000000000","message":"Patch Set 34:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-jl7kj/index.html\u003e","accounts_in_message":[],"_revision_number":34},{"id":"7b121f4718c91f6e554768626effbac0adef4eef","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 05:56:29.000000000","message":"Patch Set 35:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-rgd2z/index.html\u003e","accounts_in_message":[],"_revision_number":35},{"id":"896b9f03c42a643cbc39270d60cdb147e4b03fcb","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-28 09:22:25.000000000","message":"Patch Set 35: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"57e9466dab459ab6d271232725717babd82caf28","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-31 15:11:05.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"c5ebbfdacec47ea99d165ce39c9a65b3ab674595","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 06:30:56.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"8dd0e238fbfd5faea8997c1c7f35abde2cf90036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 07:11:05.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":36},{"id":"b816eaab42a98b4a17f82e44047e9863a1fb353b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 07:18:19.000000000","message":"Patch Set 36: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/70968bd34d5048b9b777d4de35ed1091\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/d937bf210e624381939ddbbe1be9ad1e : FAILURE in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"a4aebce56cf0f9ff09fcaf2bab23c5ebf82a2377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 07:38:08.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":37},{"id":"986dd2401de76a67cfad41cc2ab34dd8f17c7aad","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 07:46:45.000000000","message":"Patch Set 37: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/221d953e3d60490b879f48652de35e21\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/210a213f04cf4b679addb5360478fe4d : FAILURE in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"1b1817338ee8be7afb83893fc81047e05a0a0bc4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 07:49:36.000000000","message":"Patch Set 36:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-8zj5z/index.html\u003e","accounts_in_message":[],"_revision_number":36},{"id":"4635296b8dcccc533062bec96e2c91335cadebb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 08:07:01.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":38},{"id":"9b85364600226de55c5aa77653b518571fb2de95","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 08:07:21.000000000","message":"Patch Set 36:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":36},{"id":"57f30911d4f04e91cd3c3ad826bd04761807ff7f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 08:07:26.000000000","message":"Patch Set 38:\n\n(3 comments)","accounts_in_message":[],"_revision_number":38},{"id":"bc76706e1e9c5873b94a6a7639b780d3756c89de","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 08:14:10.000000000","message":"Patch Set 38: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/dc785a5218fe4411afc3b158b5d245d6\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/2efe39bfbd7d4073bdc81630c38f1a2e : FAILURE in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"fc88522890e1638c3aed008827d04121943fba0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 08:28:56.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":39},{"id":"2f5f6d65498f6aeebd41aa789137cc59ea864f47","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 08:33:06.000000000","message":"Patch Set 37:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":37},{"id":"46f4aac10d3a2a1c6f3fbed8713871c0d074747a","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 08:35:31.000000000","message":"Patch Set 39: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/c62bde2934624e3fb55820a5b121c751\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/8a852070d42d490eafc9938bf59a1502 : FAILURE in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"457a28840cad6193d60855bf75d45bbc752d1433","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 08:40:02.000000000","message":"Patch Set 37:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-ftxdh/index.html\u003e","accounts_in_message":[],"_revision_number":37},{"id":"f64d007fa66c2b90b2c55c73e7af69ad61279d9b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 09:02:36.000000000","message":"Patch Set 38:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":38},{"id":"072d3aa997c55e0f8afe1a59043be59bc798ac55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 09:04:23.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f5d16c95a449a78cc56a16cf2bd093\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4dc0e5eeb0b845aabd6b513f88f32c5a : FAILURE in 19m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/39e354cdf5464c0e8c9dcf83a3319cca : FAILURE in 14m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2b8ab22849db4fd193f7fd4b752338f9 : FAILURE in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"43066a69962a58a8898251beba7d2319a1534ae3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 09:21:58.000000000","message":"Patch Set 39:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":39},{"id":"452c8cf6fa9d3065963ef34fe2b584e00a74faca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 10:07:41.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4000a8edc99e424db435780816069c2c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5da59b19f8cb4b7cb8196369ea3d9aef : SUCCESS in 32m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e40c432d60a84f639ee2208b3a33b0b7 : SUCCESS in 1h 02m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd4da5d8270f46c99a94664b51cbcac8 : SUCCESS in 30m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fc7934e26ac4de69cbc5f598266f367 : FAILURE in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db56be90e3944319b27d3f7c7faaab96 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71dc634fed394b3498e649fab9e7a6f2 : FAILURE in 10m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/712aa3b3d19d48d4b605f872648fa5df : FAILURE in 8m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c9402d70bff94998a3bd019a40a70689 : FAILURE in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b3b43bd2aa4e3b9d430fef30996cdd : SUCCESS in 12m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80ac67620a4442f1868d0fa9a38b529d : SUCCESS in 5m 19s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0bcbbdeb129d422a8fceaf086295f955 : SUCCESS in 1h 14m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/199b807c544f4f08ba5afbbdb0930dc1 : SUCCESS in 59m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d9d30c1ddc2c49de95fd4bd7e8367ac8 : SUCCESS in 46m 25s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/4dc57c3a6e9b45bf821f253de29ffa8d : SUCCESS in 37m 12s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a16251951f0444f3b183bccc78c658dc : FAILURE in 52m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8d11a3794233448ea59695e14d20c6bf : SUCCESS in 52m 58s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4b0c115b59c54df6a751231edd3d3035 : SUCCESS in 1h 04m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b925026f9f31467b9fbcb47f145b6e8a : SUCCESS in 1h 33m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/6510113c1e8140bebd66e5022b06da58 : SUCCESS in 59m 52s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/dc0f030064ee45439eea89a318518028 : SUCCESS in 30m 59s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ce1958e4b83c4276af7dc1a7c1877de0 : SUCCESS in 5m 45s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b51b42429c5943f8b3f3286c6efaa4ac : FAILURE in 29m 18s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8d2363f1b44d4b6b92245b624ef3a1c4 : FAILURE in 26m 52s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c4675e540f8143519cde373582f76222 : FAILURE in 38m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6024c3e947214fb7a76728db9f7d2113 : SUCCESS in 1h 03m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e05cf23775147de992a76913710e35f : SUCCESS in 59m 44s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4ec60a7154a343c6a0ce59d70ee5a598 : SUCCESS in 27m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/53638b60eefd47299dddca0d72a8a34b : SUCCESS in 36m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/59dc08c4c7c04f46a71638db67a252e4 : FAILURE in 14m 06s","accounts_in_message":[],"_revision_number":39},{"id":"d6f2d7f0789e81a68133dfd385155c8d817aeb9b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 10:36:23.000000000","message":"Patch Set 38:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-fz2t9/index.html\u003e","accounts_in_message":[],"_revision_number":38},{"id":"c8cbb094ab78d28044e8a6810e6536e22cdd7140","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 11:09:21.000000000","message":"Patch Set 39:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-ml7bh/index.html\u003e","accounts_in_message":[],"_revision_number":39},{"id":"491395afbce16dc8e10fd92a504ffa83852b76f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 16:42:03.000000000","message":"Uploaded patch set 40.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":40},{"id":"090e643a5e41bfcf6c439f66849a37603885328b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 16:49:24.000000000","message":"Patch Set 40: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/6571079b8b1f40a8917eedb6a62e2127\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/a4ec54d85362428cbe1b8618e88ba560 : FAILURE in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"878890883d162031a2b7c5fa93fd94b21857a798","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 17:17:19.000000000","message":"Patch Set 40:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-pk46t/index.html\u003e","accounts_in_message":[],"_revision_number":40},{"id":"79823d0a5a7a79b4302e6c939de77031fc95a201","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 17:28:02.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76e583b9cef44d68ab2af4e7e614ac17\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0353c89270024a4baf8b18dedda5e4ec : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ff2aa1dc8bbe4406a5f4a836f63529ef : SUCCESS in 20m 29s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9d836954f40426fbe808f621935ecae : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"4a3a7ba322bebd490daa4ffdc2a66e8860fbe527","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 17:36:46.000000000","message":"Patch Set 40:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":40},{"id":"923ec4a62e39e565f9184784862b080c459bdb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 19:07:37.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18e429b19d9c4b619d5e99a8f10f935b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3942e699f2ec41499c1110decbe8caae : SUCCESS in 1h 11m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8bd514a9f35b4925ac453ccfc7e2e402 : SUCCESS in 1h 05m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac2498b6cbd84e5c95c111d1866f04cd : SUCCESS in 34m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58760792d56946a396e58654c797a0fc : SUCCESS in 48m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a06c112fb4e4175b633212d67fe72f2 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7b3a98449b04420a5422cc05f0b2c6b : SUCCESS in 8m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d6b2ed0e57448b09c6d846af6a886a9 : SUCCESS in 12m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1c8b5fcd6dec4d01ba57cae6968ab759 : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0492c008a6cd4e45a7c0053d62edf6c2 : SUCCESS in 15m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f9dbfc5fbea47f9af99d73d8f25511a : SUCCESS in 9m 12s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/cfd1d23e603e46ffb592b369f6cb317e : SUCCESS in 1h 11m 20s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b331d9d8b7b345f8a7ac940a071b5050 : SUCCESS in 1h 32m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d859658647e64310a9ad8c7e9123e88f : SUCCESS in 52m 30s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a3d818a39b824cfbade4099e483afab3 : SUCCESS in 37m 38s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ba9b25e7a9cc44a0af36afb7a3ef35eb : SUCCESS in 1h 01m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/778328ecd693413a986c6bde5db4f127 : SUCCESS in 51m 13s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9dbbca56bd85448e80d02107be5bc11d : SUCCESS in 40m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a885ee92fdbf463e9dd22ff4abd39179 : SUCCESS in 1h 32m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/61f8c8b7cfb74c3a91e96f8f3faba229 : SUCCESS in 2h 16m 10s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8a7fe53c41ec40ff85402321ee164eca : SUCCESS in 51m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8cbef9f6c2b54264a96f5bc622baa4d6 : SUCCESS in 6m 11s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9c97ae365de0450e9b925dffda6fba7f : FAILURE in 32m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4e6b867173ea430eb416b71eb4daf5e8 : FAILURE in 30m 43s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d0023b53a87a4487bb576821dd243f06 : FAILURE in 36m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cf49bbf5589411da4dae0fa5ca62d58 : SUCCESS in 1h 35m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35697a268efc4c5a8888c417a9baa022 : SUCCESS in 1h 04m 50s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f14a644202ce4a3aa6096b6c54640fe5 : SUCCESS in 53m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/715008318ff34aa6b0648d70520a4264 : SUCCESS in 34m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0af1c2ccffd84108a0544e65a2978330 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":40},{"id":"ed274771b4e0a5f24ad916c7d6c59c54d0fb3f33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 21:14:51.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":41},{"id":"cc75b5392cd410beac040686055c2058945dc529","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-01 21:16:30.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"c7727002c4dc6335c1314197723e6adbc6b878b1","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-01 21:22:33.000000000","message":"Patch Set 41: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/32f3a438df474a29bfe725df24f17d86\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/de1976aa9d88483ebf3265fe4dd82256 : FAILURE in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"4179b3e888e5f8c9469b482b644ec5aa8766b208","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-01 21:45:00.000000000","message":"Patch Set 41:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-cq62j/index.html\u003e","accounts_in_message":[],"_revision_number":41},{"id":"bffc316ae55b8933f113e0ad0c18ff8138f1af42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:46:41.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dded0ed971f491ca7469bfe86f69ccf\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f7f210be72464b86a1320a1be89df4ab : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8949987df2984d77b343bb83aae7fe9f : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4fb6a0f6a3744fc28162d515c6c00300 : SUCCESS in 29m 54s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"2310dc5339e07f0f53c2eda599b5620bf4121806","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-01 22:08:59.000000000","message":"Patch Set 41:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":41},{"id":"13ebfc85dd2e1f8f5e1e475bc4be0cc3e738b4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 23:21:02.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4286e75bcbc34325b247aa5203b679ca\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/565fc795c1084617be6539769e599c39 : SUCCESS in 29m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/30202ac7f84349fd8074d20daeded253 : SUCCESS in 1h 34m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6116cf1bad784eb0b1e94f79fedb53b4 : SUCCESS in 1h 03m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b5358264ca547c0a597412b9bc32739 : SUCCESS in 55m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02eb4173205d46ee8595677514fce868 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9299ced195564a1b85c7b02df138b413 : SUCCESS in 13m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02508425996a4da7b7493ea505693852 : SUCCESS in 13m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d9fb175cda91498489ec818d3a4153e0 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea991b4e1ea4478907fbc3c9e787014 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e65cb0916ac54198a402348f0d1cf0d9 : SUCCESS in 7m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ea6533ceadb04d5db8269a40d07e2e0e : SUCCESS in 1h 02m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bbceed55d8384abbbdce08b59e6c2928 : SUCCESS in 1h 31m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/03305c5b885d4a2bab370578916669c9 : SUCCESS in 51m 14s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6fc4deea158f417f9c619ff1b9b124ea : SUCCESS in 38m 48s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/569b22df33304f0ba909820a518c1911 : SUCCESS in 1h 00m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/622d3791f887429cbb46d91f06137d7c : SUCCESS in 1h 00m 48s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/23c00894917344da87c8d6ca66fc9c82 : SUCCESS in 1h 01m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/015a2a6146e64f8598f0aab908373ec9 : SUCCESS in 1h 58m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/12699489b8f345d582e09dfa67ebc5e2 : SUCCESS in 1h 56m 48s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/91799a97520e49f3958b63a2a8094efc : SUCCESS in 28m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c63ef0f09ce3407d86a85ff99516b755 : SUCCESS in 6m 43s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4e0ac2546e3741a78298a82999d966e3 : SUCCESS in 35m 06s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/814e2a9d407c4b96b67801eaed4c78b1 : SUCCESS in 26m 34s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/de18208401084b5797d7c97bfa1eddef : SUCCESS in 37m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a0261c837ec4c70aed442d1debf3ac9 : SUCCESS in 1h 29m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e12ae82b204c42f78fc685f5250da06c : SUCCESS in 58m 44s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/93d31d551767444aa311c974bcffcfd2 : SUCCESS in 29m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/54d927c0af0146c0912aea9e91103941 : SUCCESS in 34m 22s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e6bff47c88814c0eb00790cb41fd0c1a : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":41},{"id":"45855f0aa245135d558da029bccc6a6e6d931b0f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-09-01 23:34:27.000000000","message":"Patch Set 41:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":41},{"id":"6943a21b04333f6350556fa5e9cf17927bc59ba5","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-02 03:04:49.000000000","message":"Patch Set 41:\n\n(7 comments)","accounts_in_message":[],"_revision_number":41},{"id":"59bded361a277bc6923589799a1fc9818ecae349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 05:19:55.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":42},{"id":"7ed016d0fa0deac8383041a1cf5ea62d8773bc6b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 05:23:45.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"e142b75d6f8eb8f3efb6bf5622320905db3178e7","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-09-02 05:26:35.000000000","message":"Patch Set 42: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/d90dc730b75a47af9b0ae14a7366bdba\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/c870fd02666845b08c5fb10c16f9413a : FAILURE in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"73557ad2b4b4a7dc2bbc0403d8a9df9e183c4742","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 05:27:48.000000000","message":"Patch Set 42:\n\n(7 comments)","accounts_in_message":[],"_revision_number":42},{"id":"bdb36d2560febbbc4c86575d7d657609e8f81573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 05:45:18.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d279e6e5406a48368be615c8d807ea52\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bb4f530b1155462c9ae8735a78b6a748 : SUCCESS in 23m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a4ee378defc145489e463b5bfc1b4498 : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/420897d240e049c59a6f4398cadbdf9a : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"eeabab90fe3abf745b0f693ff3e4020301511590","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-02 06:08:46.000000000","message":"Patch Set 42:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-dbvbb/index.html\u003e","accounts_in_message":[],"_revision_number":42},{"id":"ec76437d5b7408a4f457f1fa2177ea7834b2eb4b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-02 06:12:56.000000000","message":"Patch Set 42:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":42},{"id":"77e97e7d9e0346cc7cc727269bd6cfec725f9203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 06:51:54.000000000","message":"Patch Set 42: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8301c623394044898050dec22e8cbb68\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d286961e0d944ddc9f34a0c7ce4fee15 : SUCCESS in 42m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2cfd167469b7431195076d549f52926c : SUCCESS in 59m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2292656e53540a6ae31c35218c4d387 : SUCCESS in 36m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55e9d8f4f4314f8488e8d047bc8a1e03 : SUCCESS in 1h 02m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f20c8267497848b0afd25928f55b1b06 : SUCCESS in 8m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88c58d2bfe1242e4a763ad85bb439caf : SUCCESS in 15m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/860debbee6f7457099f50001f5d260bd : SUCCESS in 13m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f2ade0b5d0b345a7b1fd395b4bb0c4d0 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484972c809274d748866df7414cd9218 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f66e33dda8f4164842e924ed4583a12 : SUCCESS in 9m 30s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3e4f7d1995c544a681866faf8f580f9a : SUCCESS in 37m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8bd36c57223347778c3adbdb6e0e3c41 : SUCCESS in 1h 14m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5f3133469f9847d7a77194dc5fe8962a : SUCCESS in 48m 54s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ce9e29ad8faf4639827f31a82b8ef0ca : SUCCESS in 36m 01s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3c62f1d0745246f5b825afd253df3238 : SUCCESS in 37m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e95a8f5928c64709a7b6dae2faeaf534 : POST_FAILURE in 18m 49s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2defde88be2744e2a6afe0fec9a46c02 : SUCCESS in 52m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ad11540e2d484828b2a8368861cf46d3 : SUCCESS in 1h 29m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/eec6018e398448ce97afc5e65a5e4b37 : SUCCESS in 1h 11m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bd843db9da1e4c2aab5cbc70f2e38b32 : SUCCESS in 32m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ae1dd12a77f24cf2b5d167cdecc1f611 : SUCCESS in 6m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2a95640e8bfa451e82733d202dbdc437 : SUCCESS in 33m 38s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/241a16be42394d02b0b17b790f7f9b17 : SUCCESS in 27m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a6dd9c3a323d4665ae1fca3657ae3ffb : SUCCESS in 31m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/76ed1a2f722d412b8cc7155429840f63 : SUCCESS in 1h 01m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a54e956db047465eb5ed1d669af680d3 : SUCCESS in 27m 57s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/235fbf5e3db549868ff611a572a16b61 : SUCCESS in 1h 01m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/241a82e2d78442ec97d7e5186539cfe3 : SUCCESS in 27m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/50b1ef428a774e208c93be9f4afc8263 : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":42},{"id":"55eddfeee70b94c344f558347331886cfe24e7be","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 16:24:23.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"b8fc8b37f1d66bffd3e86dfaeef50c41e10d685e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-02 16:55:02.000000000","message":"Patch Set 42:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-j4dgv/index.html\u003e","accounts_in_message":[],"_revision_number":42},{"id":"58dbc146ada879701d03e796d28ce3f8bb1d8ab5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 17:57:30.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"22ecd46227d5be0a4e307b5325fbf058adcc4a4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 18:46:36.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d7c7eff48f4b1aa859c841acdbe326\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4e2c13401f2f433e84f62c556907e30e : SUCCESS in 29m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c844d326d2b4198a581466cdd1a1727 : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/51ccc61895b0409bb4621627bd0e1009 : SUCCESS in 23m 08s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"742b3ec48934084a760b17c6bb5570544d27360d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 19:21:43.000000000","message":"Patch Set 42:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/997cad4270ca45b7a3ba30d6e88bfd93\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d04bc8b8b23a42a3b2a4d9600e401f25 : SUCCESS in 1h 12m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/22835237b2124d438d44efa0f65ba3df : SUCCESS in 1h 01m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61191b3276874b9cad321840cbe70c23 : SUCCESS in 37m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3271722653740bcb0d4266fa24a7959 : SUCCESS in 1h 08m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da010103b514d3085ccfad34bb80e4b : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f555b1c410b4980bd474fe1295cfa20 : SUCCESS in 14m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e725b84c771544d6835d3a652d094a24 : SUCCESS in 14m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/13dc395950e54e41965d9ec8ca31e8f2 : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb8eab1b6284c8999b82bade0b2d426 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1239d9c617ba450ebaa736693bc552c8 : SUCCESS in 9m 16s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bac4951dfcb44766a965891219f7ea80 : ERROR Failed to update project openstack/glance in 4s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/757dbf46326c46f9a717b3b205bcd7bd : SUCCESS in 1h 37m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9ce4dded3edb4ada81aab660f4b2be7a : ERROR Failed to update project openstack/glance in 4s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8365707ba60042c2a998dba4df5d198f : SUCCESS in 40m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e9b96401ecd24488a3e13fd0e8afb79d : SUCCESS in 1h 04m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a49af2b8af3e4e72af2b8fdb8ea47f57 : SUCCESS in 28m 26s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5348552aab5e43edb041ec1cccd0930b : SUCCESS in 1h 22m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cf713aebcb5144f5ac2fdfde36c273d0 : SUCCESS in 1h 54m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/432afe1ca94f49c08863dfe221bd1181 : SUCCESS in 1h 51m 44s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0a27f274b05249fb9c36f99e18f2e890 : FAILURE in 50m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d84c8bf2b04c418ba756ff99ef35bca2 : SUCCESS in 5m 48s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ad96ac816ec6441ba54beaf1c576a71e : SUCCESS in 28m 25s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/de5e1f62442d44c8b660854f0bafc885 : SUCCESS in 29m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/fbd591c2b62b492788da1afbc61db437 : SUCCESS in 38m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9630968d1ecc47f8b7e4aaf42a3ad4ce : SUCCESS in 1h 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd91f44abf644ab28114be3eec24c8e3 : SUCCESS in 1h 02m 03s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/01b48629e87c4f3497458e2fe1393e5a : SUCCESS in 56m 01s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b8f4764f52334f32945508e7811e427f : ERROR Failed to update project openstack/glance in 9s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2dfff11bddfe4c04b9a6dfed47466d92 : SUCCESS in 13m 20s","accounts_in_message":[],"_revision_number":42},{"id":"0c97af079c90b47a312ccb5a098171cea92eb34f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-02 19:41:02.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"f7286220948accca8e5ef93e3a68eadf0a637ef9","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-02 20:14:58.000000000","message":"Patch Set 42:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-tcsbs/index.html\u003e","accounts_in_message":[],"_revision_number":42},{"id":"118d9ff2287f3f2f3cf9d0e0b32db93088ed9b69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:16:52.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2cf775bd68d4fbeb10dc302b25cc4b1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1fa16257b6c47c0a71f394b5ec88001 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f602f2eb4cc4a19aaa5f67ba1c2a232 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/dd3cc42949484b0f967af86a05fc32ff : SUCCESS in 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"309b21ca5b65230180e0b4a89fdda7e3717de915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 21:27:12.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c20e9e58c94c568af8fcb69d36409f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec3830fa943f4ff5ac7e017bb6e612cc : SUCCESS in 1h 04m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/05c8ae537920420da599d498c12ab615 : SUCCESS in 1h 09m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d428bbc9ab6e460dbc88ef5a115f7fcb : SUCCESS in 29m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d5b6250991d4b928c2cb0ff9a66046b : SUCCESS in 1h 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814fb40156644d02ac528b48aa9665eb : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6cbafceb9094792a5af172c9e869052 : SUCCESS in 12m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17ce5058f680449fa512ca1a36e53a03 : SUCCESS in 13m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/367be8dddac646318f1e25ab1d24ea43 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a38d8a829e4a859456bf59e3168b4d : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2632b48982904f81a138373e3f862009 : SUCCESS in 7m 52s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d7d4b6d968d145909ee0f4410adcb64f : SUCCESS in 34m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f07c6e2b66f4ea69bd4b8588053230b : SUCCESS in 45m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cb8da8f734b745098704fdb256bb4e5d : SUCCESS in 51m 09s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5533ea384d244de197094f244fee01b3 : SUCCESS in 37m 54s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1d00a2ef12fb4b36aaa391fae2e7a69d : SUCCESS in 1h 01m 18s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2212ed040316469f89596f0c74e6f435 : SUCCESS in 30m 04s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5a3de7bb4082402fb6a0b987b2025c7e : SUCCESS in 38m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6eec02331d604d0d8111f6cddbc6d234 : SUCCESS in 1h 38m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/784f4389872842c3823609bfc4369ef4 : SUCCESS in 1h 41m 52s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6f24ddc8cbf14cbba4de0de4ba9e64af : SUCCESS in 27m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b49c10e1fce645fea81523c44542b1fe : SUCCESS in 6m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8be23a5c311b4faeb48197e988c32ce1 : SUCCESS in 25m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c7c4bf6130de4e11a1d0bffa3228b826 : SUCCESS in 27m 41s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/29fb7ffbe8fb45f7b418dd312020d232 : SUCCESS in 35m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ae9eb63ba734cd4b5a68622aaf27374 : SUCCESS in 1h 15m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b00f0cd4dc414c93b42e58e2da0ceac8 : SUCCESS in 28m 12s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/196110d400934f4e8c539a74c798c615 : SUCCESS in 44m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a0880f9a3e3543528f124bad5422c578 : SUCCESS in 34m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/881a852f733f42c9984dd8769f064d8f : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":42},{"id":"e5707070b7a060cf4ee12fd6d17db2ddc1b376cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-03 02:12:31.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":43},{"id":"0000da264e51c8d4855cdb7c5a447e28d3cedb55","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-03 02:25:23.000000000","message":"Patch Set 43:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":43},{"id":"6b30c0aec067f98667dd6599e1c5f1d6149cd7a4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-03 03:15:12.000000000","message":"Patch Set 43:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-86dpn/index.html\u003e","accounts_in_message":[],"_revision_number":43},{"id":"ddcb9fc25e9c469a6239fb39e109b80c8aadd8eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 03:39:21.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797477a690064d64b44aa8fdcc8f3458\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ea620385e174b18959ceb514234eded : SUCCESS in 22m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8fb3241413d04f10852dbe75cfdb4d9f : SUCCESS in 21m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a8ec3179ae0640af8a72e299a5b4b7cc : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"4531b504fc1884f1ee51dc01faf9815bcbbf9e55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 05:09:16.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e30a17b56542118be404da088c18a7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c148de5981e34eeca17325e5b52f5644 : SUCCESS in 1h 11m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c8d58ab4f5844f4ba5031ab7c4ca1c1b : SUCCESS in 50m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d6c20ba1e3640de9ea574226f6a1530 : SUCCESS in 34m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/852111ccda9d48c79dd6b3c8eca70ed8 : SUCCESS in 1h 18m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f82cb8bcca2d4f1785b032a858264175 : SUCCESS in 7m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa33fbfddcda443495f9181b36007e6f : SUCCESS in 13m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1b37dc6548154f76abf49af5a5218880 : SUCCESS in 8m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b1858b9acdcf4419b85be849446c2545 : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a18039668bf4fc2ac000dd0593622ea : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0413706f1d3f49b3899ca2e1997093ad : SUCCESS in 9m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ed647f0ce566496a8f30b90a8833f901 : SUCCESS in 30m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad6e2fb34bb6416cb3b69bb3029a6582 : SUCCESS in 1h 25m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e9057b6d7fb849a287729bbd62862374 : SUCCESS in 55m 09s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1072d59b2c574c86ae4374ef49303047 : SUCCESS in 40m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f37f74a722444df784f9da85e0264436 : SUCCESS in 1h 01m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a6e7756c60af479c9890b69142dd910a : SUCCESS in 26m 24s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4b401315a4e84762a209286d4516dfd2 : SUCCESS in 49m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c314391cff6e44c7a91ec2f3ca59cd19 : SUCCESS in 1h 58m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/1d7900cc5df7475aaf45438e4f6b1d12 : SUCCESS in 1h 55m 19s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8bed7f1738c64646a7f0961b0d937b84 : SUCCESS in 47m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ba3b958f1d6649449cbca535756edcf1 : SUCCESS in 4m 57s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/941ce1375931426b9a0188f875ebe7e9 : SUCCESS in 30m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f0536812400b4b1c991cb4e6425c4b9c : SUCCESS in 27m 00s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3091fc7d646f490ab6174f3b3df686a6 : SUCCESS in 36m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa8cbc54e9004f86ac453d31be1c2ba2 : SUCCESS in 1h 28m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53b1155ec2644dba87be371805672468 : SUCCESS in 1h 03m 15s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c0b5d1f2c1c74d019a8220bc2e8045be : SUCCESS in 43m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7f03c0fb96a84f6f9d6541ece1dbcc29 : SUCCESS in 33m 33s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d742ecd6883145cf812d20be4707f853 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":43},{"id":"ca55fa4a518b31c9717998e5116f75333f8647e6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-03 17:52:48.000000000","message":"Patch Set 43: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"be07053414fdec7c2746a235e53726b4009d8ea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-04 03:21:57.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":44},{"id":"0dde5c68fc7b90f7a15ee1630cd7fb8bbb630172","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 04:19:16.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c26d94cdbcc74174b39a0262bc359e50\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/035ffeedd2d544d3bf6e60d0a131e1f4 : FAILURE in 53m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da7fa8f1a4894ae6a86e8e763d72b78e : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e099e5ec3c4f4d27941304b394286d3c : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"415e54e63e0a67cc67356c0f9cdb5e13365caa90","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-09-04 04:19:32.000000000","message":"Patch Set 44:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I2cda35fa0c2e080a67418d557cb9030b11f31394 .","accounts_in_message":[],"_revision_number":44},{"id":"8d951db43be22f65c698af8f3c7ee9c9035a8d68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 05:12:50.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38eefa6fdd3a47d3b31f9bfdb726c321\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b158235bdda94226a32d7319bc6d90f9 : SUCCESS in 31m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3524d581f493484998c9d13b3b616254 : SUCCESS in 1h 38m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe72c70755b84d04a8dcbd63bdad9c93 : SUCCESS in 47m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d889813656fe490586eae5fa4ff43fd9 : SUCCESS in 57m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014e738ebfa84a4999f81c3055fd0249 : SUCCESS in 7m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e297376ffbd4a5e8bee6f55e691fab5 : SUCCESS in 14m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4436998da454ad8b3c4dc8666f3a251 : SUCCESS in 8m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a0b7757bebc84711af31e50de3f8be0d : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a6f527da5b34838b186bf33f37947a6 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ac026dceedc4918aeecf68bfae6d152 : SUCCESS in 9m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1eb08b39392d46eaa0c87101345ac748 : SUCCESS in 1h 11m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d5a7e7393704f2b998a21faad55b1a9 : SUCCESS in 1h 25m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/66555394bf2d4fa7b8c6c5edb54e20fe : SUCCESS in 53m 14s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1d485eda4b844c7996d9d24bd045e136 : SUCCESS in 36m 57s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/83a2c44a86e04001829dc69cd7657775 : SUCCESS in 53m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5d23c9247e5141fb91041f36ce0db102 : SUCCESS in 27m 40s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/30e4d7e3aa6942ac9b028dc10ae734c9 : SUCCESS in 1h 04m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/40c33b80dd3a40a1bbd5d9b34f6cc823 : SUCCESS in 1h 18m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/66f7f2da800e48a6935a26a9b5b2e509 : SUCCESS in 1h 42m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b12853ef40f0416a949d32b7f6f6345b : SUCCESS in 49m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6b9ba048e58544e9a0727a46cb12c833 : SUCCESS in 5m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/eca44a1c51f24be8a00e0100de0c9bc6 : SUCCESS in 27m 50s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ccfb23ea6b464064a0b0d32784d0486a : SUCCESS in 28m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/02c94214196541f3bcbf9f99b3eb777e : SUCCESS in 30m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c6e36772b9741ed8057e4f5f05abc6c : FAILURE in 1h 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f273c14a41ed4dc29689aff42de84565 : SUCCESS in 45m 52s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9deadc6670e44eccb042064d342ac998 : SUCCESS in 46m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3f9c1ef1abe242edb75a45f8f107d381 : SUCCESS in 34m 44s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/978cf400af1848df9784594a547829cf : SUCCESS in 9m 15s","accounts_in_message":[],"_revision_number":44},{"id":"f6762bfe1d157f83d086a990323b8ec5a453072f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-04 05:39:58.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"90e8672c1e98c0e76eccabd0cada91fa044c9bfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 06:02:52.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88ddec00d8394897bd27ce75bcc9d561\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c75cd6cf94c44f68a87e4aba85d95d07 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/554dfbe1ca1c440f8c0f8d39e3de78a8 : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f9613222e2f54dfa88a2e82e5ea70d70 : SUCCESS in 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"0035bf058060114db283a3552f81035875578f18","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-04 06:54:31.000000000","message":"Patch Set 44:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-59k9g/index.html\u003e","accounts_in_message":[],"_revision_number":44},{"id":"552d59513a7a3609be3e3f384fb1fe4f7afdf6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 08:08:00.000000000","message":"Patch Set 44:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab8728e00149460aba55a574b55c343e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7dffb6d84be542ee9d35d7c836a0d2e1 : SUCCESS in 30m 25s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/881b2c61e77c4d5fa09bc8894cb7ac68 : SUCCESS in 1h 19m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a7a958c3c1b4f759bff47b0cc305e66 : SUCCESS in 39m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba65aab3ae394190bcc57f1a5749b8e6 : SUCCESS in 56m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778d420c9e374f5ca9a0a421a9417e9a : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ddf7b0f200c4ab2bdf7197426e6b254 : SUCCESS in 11m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b4434239a4e4180a3c519e47ff825f9 : SUCCESS in 13m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/da97f906447043f9887eba64a78bf85a : SUCCESS in 11m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7913bd53e9874a718a409685f5e38852 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41c3e10e50b4407594efe17a4a22489d : SUCCESS in 8m 27s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/85c990dc511a4eeaac6fcf4e80e30c69 : SUCCESS in 54m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02559d12a52040ad9479ad1184e02490 : SUCCESS in 1h 34m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/46f66ba9bcd74d008b9e7b70cbb73c2d : SUCCESS in 27m 19s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/4058cb475b634bae9e36ce876b25dbe6 : FAILURE in 38m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c97b466f24cd4c4388af5f3775267077 : SUCCESS in 1h 01m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1de4472f935d40e0ad450881da49c957 : SUCCESS in 45m 00s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9cbb909761354193a1e113e2bdfbd807 : SUCCESS in 1h 04m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9992d6aa53ce4c5ebeb791cfe983b452 : SUCCESS in 1h 48m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/c136aad2cced49c49fb95c4198c43cfd : SUCCESS in 2h 20m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a3ffbd843cca4dd48aa9d26bf5cb58e2 : SUCCESS in 49m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7699b1440a2f43e8b2384ab8442522ee : SUCCESS in 5m 18s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b06777ccbdc04e71a20002b5e0543f7b : SUCCESS in 32m 48s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/253b175126eb41df98b691ee2160d234 : SUCCESS in 26m 59s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/25a00a629177414ab5184774a840ea94 : SUCCESS in 30m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5341af33b424bc2a280b5bdf87a2b21 : SUCCESS in 1h 34m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ccf264306404141be9f7b82f7b4c7f9 : SUCCESS in 1h 03m 06s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1e3e203fc43b4ec6bd51484e9bbbb40e : SUCCESS in 30m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7a7a2fa1dc1547f2b9f45816783cd33c : SUCCESS in 33m 39s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b22df8b6156a4b05b5745ee3cf79705b : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":44},{"id":"b3b0744b695f356efc2656f9cc95b35d130ba543","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-04 13:52:21.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"24b4c0c45061e3b8fb73098f1048b6ba73ab95ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 14:34:38.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36bf32a43d2d4a31a3c38818279b3dd6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bac0ccb894284fa08fe7eb91eeb78aaa : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25a4e1adf5db4519bbd66e512bbc5955 : SUCCESS in 29m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/66d1add951df408ca2bda883e756c99c : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"3385b28913598c96e920dfaf430c7733595a88a0","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-04 15:29:02.000000000","message":"Patch Set 44:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-tgf59/index.html\u003e","accounts_in_message":[],"_revision_number":44},{"id":"8bfc68b745cdeb51dc8c36d802f6658f9671bfc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-04 15:36:39.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bb222e8000a4645a889aad2fc3ff104\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fd7d836ac1f842de857c642c7e97b808 : SUCCESS in 1h 03m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/76b79352ec3a4ba885a757945644e367 : SUCCESS in 1h 30m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4665d4785cd04bcb887e1074391f9887 : SUCCESS in 59m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/613776072d42434db1147e419c429db8 : SUCCESS in 1h 01m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34b5051a43014e7382c5b891aff6c03b : SUCCESS in 8m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/601be85ef92e49aab965f4bc4f7e3de8 : SUCCESS in 12m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9a8316ed9c3423c9a710927e3823f83 : SUCCESS in 14m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/73c3dcf7510d45cc86096b8be9789212 : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fec3db88b39c4ba093052c24e7400ab7 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58a8f23af0694cb1b0ee9b69b83a38ce : SUCCESS in 10m 10s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e88606dc8975487fa7f914e1f7637c5c : SUCCESS in 1h 10m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/98d8989311b64ecf9755111faf08cfc4 : SUCCESS in 1h 38m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/60e9e4c715d849908a8ec7109862220d : SUCCESS in 48m 07s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/065bdce35d024e639c07912acbef1ae4 : SUCCESS in 38m 23s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a5c599af2a0a4a208b96331d00bcdf83 : SUCCESS in 1h 02m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f1a8ba8764a34284bb008c445d7d21b3 : SUCCESS in 26m 54s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/301f18fb3b2c49168164a9fc0651a270 : SUCCESS in 43m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/51e029c1ce43409e86922c214b443f1e : SUCCESS in 1h 28m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/8260805f64224132b89cf73679c83edb : SUCCESS in 1h 08m 18s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ea5adf1561a94b0982502260c54826a7 : SUCCESS in 47m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0c3c48504ea147a8806858d858c768dd : SUCCESS in 4m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/48b458acd2e2442ea0826b39901ea1d0 : SUCCESS in 29m 59s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8e7aca86294843859fbd68a7218f309b : SUCCESS in 27m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7d7be54b81ff4cf0bc32d9a9d650cbbe : SUCCESS in 32m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d6f9c4cd7ac4d24ba94d005f816564f : SUCCESS in 1h 31m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/725840f0f93a478d987730ec974e3266 : SUCCESS in 52m 23s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2b7661a0b46047adb8cb9e046764216f : SUCCESS in 53m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0868a7cefda24216a14f66c0f51acd84 : SUCCESS in 35m 25s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0fa68b446e3640d381ede6e5deb247f7 : SUCCESS in 13m 06s","accounts_in_message":[],"_revision_number":44},{"id":"d09e61888c8526e5c036731956786edd8550c65a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-04 16:07:08.000000000","message":"Topic stack-fix removed","accounts_in_message":[],"_revision_number":44},{"id":"1bd6676ff65a62783b7f3bee1a27d2f401113e60","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-04 16:07:59.000000000","message":"Topic set to bp/virtiofs-cold-migration","accounts_in_message":[],"_revision_number":44},{"id":"c6d4772ef5ea54d55b4f5c2a66e8647ef3c9b846","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-09-04 16:09:07.000000000","message":"Patch Set 44: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"69694e72ba34f56efc62722c62b48f4c27249cc6","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-04 17:17:48.000000000","message":"Patch Set 44: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"86f0def553cab03817e66f54deab63a1a882599b","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-04 17:23:36.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"0f4b11020e7e438447855c8289f8a2c13b2d68de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-04 17:59:16.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"d99208917ca8b6b8b5a63e85a6f9391e237fb3d7","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-09-04 22:26:25.000000000","message":"Patch Set 44: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"128e096291da82885bb948f289f1687b9d77e880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 04:54:56.000000000","message":"Patch Set 44: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":44},{"id":"880fd611e4ffac50ec2a688f0e2c3051e6bd2d51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 06:40:39.000000000","message":"Patch Set 44: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/999349 is needed.","accounts_in_message":[],"_revision_number":44},{"id":"13b1e3a0b77ff79280da27bcb7f763934a040c2a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-05 13:54:58.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"eb82f1bb060c400aa4c47247543446fc3ba263d6","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-05 14:25:45.000000000","message":"Patch Set 44:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-rd48t/index.html\u003e","accounts_in_message":[],"_revision_number":44},{"id":"04bd8635ce306f625e06f5d4483f4e9e67927f52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 14:28:48.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efcdf6dada664be5a8ed4b3e658617b8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2066deb57648470293db917df8f0af79 : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e71432e4df674044ae29f9864d097fb0 : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/73c8f216423f4692917a50eca961ee17 : SUCCESS in 31m 51s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"8456fda3b73d8333fd2aa9230525fc60496f7eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 15:47:05.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e392af47121149769a0a0e5add360153\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a214dfeca8e5461fb5fabf9301bcf241 : SUCCESS in 29m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/eaaf6007f85341bf8738b1d3f2c473a0 : SUCCESS in 1h 26m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11e7a60650b94f78b0c8e5380139e30b : SUCCESS in 41m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/608201d0a5d94c83a43bbaacbde149fb : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e40a4c3b88114645999f2bbf8ad3bff2 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b43f2498fc049a8847a0a8cdfdd1a03 : SUCCESS in 14m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de43cf1d8ca748be8c14923eb5e95025 : SUCCESS in 8m 02s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3b1c553b5d53431a9ac9a73915db2abe : SUCCESS in 12m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38979a46c020474bb2c6b1866a8a3961 : SUCCESS in 15m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb43bc4ead24cef971b3b21d7ff0532 : SUCCESS in 7m 40s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a6e14a63049b46209369eb91b8abaf89 : SUCCESS in 44m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/16039a432b164ddfaed158767350a2f3 : SUCCESS in 1h 23m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c95d3014ff6148d491373d42cf8ba022 : SUCCESS in 41m 55s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/60800af0e22a48a695021a7cfa8aa14c : SUCCESS in 38m 49s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/38da2d24776c439d9962d0abe13fe0e1 : SUCCESS in 1h 00m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/742c7e91382d4cff9a38cf7f5071e3ab : SUCCESS in 33m 07s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/27804e13f95c47e8aaa516db6371280c : SUCCESS in 1h 05m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/47ddf56a39f441b9ae09db8ac024f94c : SUCCESS in 1h 46m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/e6fd33d9f9b04c3693b00571a8bbe4ab : SUCCESS in 57m 50s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e7f21c3b748441d998705d82f58aae54 : SUCCESS in 50m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b293c5e11c3a4183a8ae166381155194 : SUCCESS in 5m 42s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cbd5bf51cde641699910cdbfc678cb92 : SUCCESS in 26m 22s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/49256e6502874a2a910a01f652eae363 : SUCCESS in 27m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8657ba8366b749529aba590830ba0c98 : SUCCESS in 33m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8741426918d9433ab6dfae82ed72af03 : SUCCESS in 1h 20m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2553666d234745fcb78baf7c2e91c2d1 : SUCCESS in 25m 28s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/dd9803f8e17c4a3caf892572a249fa54 : SUCCESS in 39m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1b8b4c7fe4d34f7ba813fa00bc53a78b : SUCCESS in 19m 20s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a47047f172984bdd9a9ba6639d7e35e8 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":44},{"id":"50ba089c44dfaed2d8efb782c24d3df8c62bf449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 20:42:11.000000000","message":"Patch Set 44: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":44},{"id":"0d329e7e8b2b335785e8c41368c45f4ce9548f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 22:43:11.000000000","message":"Patch Set 44: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d830ac814b3d437ba5a23eab8bb2e187\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f365506c7f8140298f6bcfc49853a698 : SUCCESS in 48m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fa13fdf9101e4e8ca3f539c52df3f084 : SUCCESS in 52m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b07911e4e8a54aba801659688d1b5a1a : SUCCESS in 58m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ca6ae61d6b4dbfa4ef41785b98d90a : SUCCESS in 8m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/546fdb22c5ce4c40b230b55a75d7e7b0 : SUCCESS in 9m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04cae4e13c574b3ab099f3d5196c5aa6 : SUCCESS in 14m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f195b35ed054f39bd7aa709e906f4d9 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6b13fbd8cae4dfa994dcc7c1234a3e6 : SUCCESS in 7m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/47933fa3962d44c084028af5574b19c9 : SUCCESS in 45m 59s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/464a15531b2b41c7b34dbe6203544ff3 : SUCCESS in 27m 40s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ad648d8a4a4140178162ff9519be6e59 : SUCCESS in 27m 25s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/78265e8800de45a3adb3b0832c6902ba : SUCCESS in 25m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6e10876ec5bc47979f5034364af02339 : FAILURE in 53m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/4a4e1ca852624752842124692a38b0ab : SUCCESS in 1h 57m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e177a843f0a94c048b472a23fcf0847b : SUCCESS in 5m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e8bbc6659b0454fb201e2eca5ddd259 : SUCCESS in 1h 19m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/07798eae612142dfb41c7d4ac7544450 : SUCCESS in 56m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d7553a968cf4adca6b4fc87ed2de7b6 : SUCCESS in 42m 49s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/949e8ad27e9b41349eabb42a6b5735d0 : SUCCESS in 13m 36s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/3eea60a5a7d541cfa0bcca8d779947a2 : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":44},{"id":"0a9b043f8272a67d0b75a9acc732d5611371e579","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-05 23:02:49.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"0d6c577e5a53ee0926c739edc4d8d5b8b71cb7b2","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-09-05 23:33:29.000000000","message":"Patch Set 44:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-989633-ng54d/index.html\u003e","accounts_in_message":[],"_revision_number":44},{"id":"bdcbe843075c52a5d6ae5ead6eab1f98a55628c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-05 23:45:44.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d92164718c504846ac399cda37d51b13\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/258cf6755adc4defb174acb19302f59a : SUCCESS in 29m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2c225876638b41eebff0374de91eab77 : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f61ac737978042f78d5b3211338895c9 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"837c4cf6a97e7f92f59b108c1d152a50d2ae362d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 00:27:08.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfffa99fb45f4bdab832f99bf0113f81\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91c66371ad534d26a759fc7ccec1627b : SUCCESS in 1h 02m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d80adfaf24894c24a40823cc064bdc81 : SUCCESS in 47m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2dcc274707c4ada8b4fb47ec66a3bc7 : SUCCESS in 28m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f77b5e19784c422d98b0618cdd1f6160 : SUCCESS in 49m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647f1e619c8348da9e0f6923be7d889a : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5486d94212e24f579172a40dd253ae06 : SUCCESS in 14m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/86525ae3207d4eadb25b6c9c8e75ff05 : SUCCESS in 11m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/921d36a82a614b6ebb33e70b7efcf8fe : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5e06e8a14d42d7993bdc3067821a26 : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39c2b43291da4aceb5782a2a6dd0a9e4 : SUCCESS in 8m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/75e1203c40454b4f8109f498efe3f91d : SUCCESS in 1h 07m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c2d795d5e4a4b47bcb0e391a8d44341 : SUCCESS in 1h 16m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6d16ccc315824dd499d9ffca543c2b8f : SUCCESS in 53m 17s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5a672fee6d774626a2edd3f1fe6ed61e : SUCCESS in 38m 16s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/807f60a5f9b94838b37daa0bef02951c : SUCCESS in 47m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a6e90843332842349cb6e0f1684bc046 : SUCCESS in 28m 15s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a5ff3b6ee5e74d9b82e9dfc6b3e440c6 : SUCCESS in 1h 00m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/72b09dbac5f74e408f1e49d041dcee14 : SUCCESS in 52m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/35ee4fe68c7c45139f2c21fdbf305e20 : SUCCESS in 1h 04m 25s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2395e16d8e7647e0ba466af76930d3da : SUCCESS in 47m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/402804ae053d4758abbb9706a875ae07 : SUCCESS in 5m 45s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bfbe3dde7c284f0ab432f8fc6e767453 : SUCCESS in 31m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6fdf8b4c6d1045caa8aac312c764f14a : SUCCESS in 30m 53s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b56c3ed6d5654dc4bae160d50287c23c : SUCCESS in 34m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ec5df954e7e4e2e8d82dd362f5061a9 : SUCCESS in 47m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/222dc7d2e8f24dd786310dce1967caee : SUCCESS in 30m 39s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec76b7f6fc8d4c81962a8587361a7b1a : SUCCESS in 50m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/07e4c149389342cdb873b14a77bd18cd : SUCCESS in 22m 28s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/94a86fc7b47b45359aed332dbd58f8f7 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":44},{"id":"31d2ba47447dd58313a0d1cd5456efd41d4772f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 00:27:50.000000000","message":"Patch Set 44: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":44},{"id":"91ba6990f72ca7797357e8a9db1cf854b59ff654","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 02:29:26.000000000","message":"Patch Set 44: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef3fcd62fdc436987f3acbd8190729e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1bb78e539c14012afd69773b04094ba : SUCCESS in 1h 02m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b2f814382b0e4e7da2fc208f0bc4d31a : SUCCESS in 46m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f711adfc3d14c858dca9757f01970b4 : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bf73b59b47b44479d1e5dd6289ec37a : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fa3a2daff054cf3bd6940be87a7bf30 : SUCCESS in 12m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1deb7cc140ce41fea8f0889d1f0d2cd5 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efbb888d32d04cff9bc8e5448a691588 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63a9ce5f0b294c7c92f56085fb064a6c : SUCCESS in 9m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dd0bc2dcd3c74636b9a7584611e080e6 : SUCCESS in 42m 02s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ecd576e54cff4da692e578912a9660ad : SUCCESS in 28m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1d36de81648e493ca7487793462a42fa : SUCCESS in 26m 21s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9ebcb4668a0c4044accdaad6e88f604d : SUCCESS in 31m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c0e1672713d1404d8a5320e5c39eff4b : SUCCESS in 1h 36m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/eefd40e12e5f4cafbe2f06fff933f64e : SUCCESS in 1h 56m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d49c1f7bdc7b466c8029b42fd2c02ddb : SUCCESS in 5m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ac88a9d8b824dfbb2d2895367c17ce3 : SUCCESS in 1h 05m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3ad2401ade5498183fdf8597b9499c3 : SUCCESS in 1h 19m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/893669d57867442b991e7c1b7ed39f63 : SUCCESS in 43m 28s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0a490be2bdab4c27b6226cf721da1ec6 : SUCCESS in 10m 51s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b8cb4c9cbfd94593bcd9aabfe5eabdd6 : SUCCESS in 27m 44s","accounts_in_message":[],"_revision_number":44},{"id":"94e8269ecc3f444effd16d5fb94160824d8a6060","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 02:29:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":44},{"id":"2df9bc38370ef1b0e4f9231d8b7222d31efdc203","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-06 02:40:57.000000000","message":"Patch Set 44:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d61079944c642bf81f344ccf5a9da5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85352069284349eba6b4028602d8eb49 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e3295386b6b488b9de3f248a5eb14ac : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"fa1d7431d51749963d39c917f05116da9164e0d6","revisions":{"40bb3084b13dab621695306335475679ed193274":{"kind":"REWORK","_number":1,"created":"2026-05-22 04:34:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/1"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 21:47:44.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nCold migration was blocked for instances with virtiofs\nshare attachments because Nova never granted Manila\naccess or mounted shares on the destination host.\n\nThis change lifts that restriction with a lazy cleanup\nstrategy: shares stay mounted on the source until confirm.\nOn the destination, _finish_resize() grants Manila access\nand mounts shares before calling the driver. Revert cleans\nup the destination; the source retains its mounts\nthroughout.\n\nNew helpers _grant_share_access() and _revoke_share_access()\nencapsulate Manila access operations without side effects\n(no status transitions, notifications, or DB changes).\nThe existing allow_share()/deny_share() are refactored to\ncall these helpers.\n\nThe @block_shares_not_supported() decorator on resize() is\nreplaced with an inline SERVICE_VERSION 73 check, gating\ncold migration on all computes supporting it.\n\nThe share_info parameter is added to finish_migration() and\nfinish_revert_migration() in the base driver, libvirt,\nVMware, and fake drivers so the destination can include\nvirtiofs filesystems in the domain XML.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40bb3084b13dab621695306335475679ed193274"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/40bb3084b13dab621695306335475679ed193274"}]},"branch":"refs/heads/master"},"b1e5b01af8af88804eda1726459be7992032b062":{"kind":"REWORK","_number":2,"created":"2026-05-22 06:17:14.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/2"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-22 06:16:59.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1e5b01af8af88804eda1726459be7992032b062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1e5b01af8af88804eda1726459be7992032b062"}]},"branch":"refs/heads/master"},"9f3ab8215a2d8a558ddfa515131a52b26e35d356":{"kind":"REWORK","_number":3,"created":"2026-05-22 06:19:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/3"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-22 06:19:14.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f3ab8215a2d8a558ddfa515131a52b26e35d356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f3ab8215a2d8a558ddfa515131a52b26e35d356"}]},"branch":"refs/heads/master"},"59de9aca53c7c8edeee00964b5541135540d46dc":{"kind":"REWORK","_number":4,"created":"2026-05-22 20:58:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/4"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-22 20:58:16.000000000","tz":-480},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59de9aca53c7c8edeee00964b5541135540d46dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59de9aca53c7c8edeee00964b5541135540d46dc"}]},"branch":"refs/heads/master"},"68c5e4aa7b48c95cf8a72c67d6fd3c0654742446":{"kind":"REWORK","_number":5,"created":"2026-05-25 06:22:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/5"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-25 06:22:08.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68c5e4aa7b48c95cf8a72c67d6fd3c0654742446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/68c5e4aa7b48c95cf8a72c67d6fd3c0654742446"}]},"branch":"refs/heads/master"},"da42443ade4a4b956b142c0c74f55ad560ee2467":{"kind":"REWORK","_number":6,"created":"2026-05-25 20:04:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/6"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-25 20:04:12.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da42443ade4a4b956b142c0c74f55ad560ee2467"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da42443ade4a4b956b142c0c74f55ad560ee2467"}]},"branch":"refs/heads/master"},"c433cf1f7a761ef133bd25e4da81539b06b002ae":{"kind":"REWORK","_number":7,"created":"2026-05-26 04:34:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/7"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-26 04:34:30.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c433cf1f7a761ef133bd25e4da81539b06b002ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c433cf1f7a761ef133bd25e4da81539b06b002ae"}]},"branch":"refs/heads/master"},"e51a4f42620bee783e4d8347a6229aed691ae065":{"kind":"REWORK","_number":8,"created":"2026-05-26 06:29:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/8"}}},"commit":{"parents":[{"commit":"f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe","subject":"Merge \"Reproduce bug/2152697\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f03e09d0a6eb0e3d4085c2448c302b9a5dda74fe"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-26 06:28:52.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e51a4f42620bee783e4d8347a6229aed691ae065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e51a4f42620bee783e4d8347a6229aed691ae065"}]},"branch":"refs/heads/master"},"491cb0f29a686bd93e18abd1f79effd79dcd77ad":{"kind":"REWORK","_number":9,"created":"2026-07-19 18:53:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/9"}}},"commit":{"parents":[{"commit":"c0c6c4c3e6f5235a2b65e7ffb0f58673f03ca315","subject":"Merge \"Reproducer for https://bugs.launchpad.net/nova/+bug/2160224\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0c6c4c3e6f5235a2b65e7ffb0f58673f03ca315"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-15 08:28:18.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/491cb0f29a686bd93e18abd1f79effd79dcd77ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/491cb0f29a686bd93e18abd1f79effd79dcd77ad"}]},"branch":"refs/heads/master"},"17fdff154e392e17656808398368866844d594b8":{"kind":"REWORK","_number":10,"created":"2026-07-26 07:40:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/10"}}},"commit":{"parents":[{"commit":"c0c6c4c3e6f5235a2b65e7ffb0f58673f03ca315","subject":"Merge \"Reproducer for https://bugs.launchpad.net/nova/+bug/2160224\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0c6c4c3e6f5235a2b65e7ffb0f58673f03ca315"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-26 07:40:36.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude Opus 4.6\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17fdff154e392e17656808398368866844d594b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17fdff154e392e17656808398368866844d594b8"}]},"branch":"refs/heads/master"},"11974a6b8d3e5e9295b8bf7b019237e029e44eea":{"kind":"REWORK","_number":11,"created":"2026-07-27 16:35:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/11"}}},"commit":{"parents":[{"commit":"7921b22c1972810d97a4e1943dac1099fb7bdae5","subject":"Fix deny_share host locality for NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7921b22c1972810d97a4e1943dac1099fb7bdae5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-27 16:35:02.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/11974a6b8d3e5e9295b8bf7b019237e029e44eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/11974a6b8d3e5e9295b8bf7b019237e029e44eea"}]},"branch":"refs/heads/master"},"3d3840eec0b952041ceabe5021131fdd8058cba8":{"kind":"REWORK","_number":12,"created":"2026-07-27 23:07:28.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/12"}}},"commit":{"parents":[{"commit":"7921b22c1972810d97a4e1943dac1099fb7bdae5","subject":"Fix deny_share host locality for NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7921b22c1972810d97a4e1943dac1099fb7bdae5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-27 23:07:09.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d3840eec0b952041ceabe5021131fdd8058cba8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d3840eec0b952041ceabe5021131fdd8058cba8"}]},"branch":"refs/heads/master"},"1f4a729455184df55d9a358f48ddc644acf92f34":{"kind":"REWORK","_number":13,"created":"2026-07-29 06:16:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/13"}}},"commit":{"parents":[{"commit":"7921b22c1972810d97a4e1943dac1099fb7bdae5","subject":"Fix deny_share host locality for NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7921b22c1972810d97a4e1943dac1099fb7bdae5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-29 06:15:46.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f4a729455184df55d9a358f48ddc644acf92f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f4a729455184df55d9a358f48ddc644acf92f34"}]},"branch":"refs/heads/master"},"df2323b0dace86b5be49fac1654f04dfb35a2efa":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-30 19:22:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/14"}}},"commit":{"parents":[{"commit":"5e1a6f55559f6a677842b2c6d5e4229ce12af54d","subject":"Fix deny_share host locality for NFS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5e1a6f55559f6a677842b2c6d5e4229ce12af54d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-30 19:22:35.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df2323b0dace86b5be49fac1654f04dfb35a2efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/df2323b0dace86b5be49fac1654f04dfb35a2efa"}]},"branch":"refs/heads/master"},"d0e200f55679cf83b52bb66f2a7b573e50ba3c29":{"kind":"REWORK","_number":15,"created":"2026-07-31 01:45:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/15"}}},"commit":{"parents":[{"commit":"3484bc31a6ca2378f8de296144fd3e5e455a1f29","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3484bc31a6ca2378f8de296144fd3e5e455a1f29"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 01:45:21.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0e200f55679cf83b52bb66f2a7b573e50ba3c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0e200f55679cf83b52bb66f2a7b573e50ba3c29"}]},"branch":"refs/heads/master"},"ee888cbad865671571e7d76ddebccd7afd4c31f6":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-07-31 02:02:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/16"}}},"commit":{"parents":[{"commit":"a77dca85a44cbc65687bf860abc69abfba3b278f","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a77dca85a44cbc65687bf860abc69abfba3b278f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-05-21 19:22:50.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 01:52:52.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee888cbad865671571e7d76ddebccd7afd4c31f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee888cbad865671571e7d76ddebccd7afd4c31f6"}]},"branch":"refs/heads/master"},"946a18aa6120f5da6f3fa388f578b2c97cb65f8e":{"kind":"REWORK","_number":17,"created":"2026-07-31 21:44:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/17"}}},"commit":{"parents":[{"commit":"803e1e7c46f726c4f2e7cec15444ce2ed448406e","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/803e1e7c46f726c4f2e7cec15444ce2ed448406e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 21:21:03.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 21:21:03.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/946a18aa6120f5da6f3fa388f578b2c97cb65f8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/946a18aa6120f5da6f3fa388f578b2c97cb65f8e"}]},"branch":"refs/heads/master"},"f5bae07c27e33a299c4cbd5d8112372b4bbda798":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-07-31 22:01:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/18"}}},"commit":{"parents":[{"commit":"f3f5722e89498ad620ece7237f61d40e79be2176","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3f5722e89498ad620ece7237f61d40e79be2176"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:01:10.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:01:10.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5bae07c27e33a299c4cbd5d8112372b4bbda798"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f5bae07c27e33a299c4cbd5d8112372b4bbda798"}]},"branch":"refs/heads/master"},"606e9d3e82f2631063a2dac4a23a3a44b11616dc":{"kind":"REWORK","_number":19,"created":"2026-07-31 22:21:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/19"}}},"commit":{"parents":[{"commit":"f44c56e5eae93668a4240c765304fd8b437d713c","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f44c56e5eae93668a4240c765304fd8b437d713c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/606e9d3e82f2631063a2dac4a23a3a44b11616dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/606e9d3e82f2631063a2dac4a23a3a44b11616dc"}]},"branch":"refs/heads/master"},"c6a327497793c31a80859cd5686c73d3a4bf7f66":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-08-03 18:11:09.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/20"}}},"commit":{"parents":[{"commit":"d8c77b2e3ed3bd063fd810ca6d36cc4e2ca33410","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8c77b2e3ed3bd063fd810ca6d36cc4e2ca33410"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-03 17:44:44.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6a327497793c31a80859cd5686c73d3a4bf7f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c6a327497793c31a80859cd5686c73d3a4bf7f66"}]},"branch":"refs/heads/master"},"b50090d160d6b3084275dd39136a2c587d53684f":{"kind":"REWORK","_number":21,"created":"2026-08-07 06:20:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/21"}}},"commit":{"parents":[{"commit":"95f7357442b3395c8d97eb8bfe78ff78a9f6408f","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95f7357442b3395c8d97eb8bfe78ff78a9f6408f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-07 06:13:13.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b50090d160d6b3084275dd39136a2c587d53684f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b50090d160d6b3084275dd39136a2c587d53684f"}]},"branch":"refs/heads/master"},"b4d68ed721a4c2d9b4338e72ca7fe0324b8be5d0":{"kind":"REWORK","_number":22,"created":"2026-08-07 19:02:43.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/22"}}},"commit":{"parents":[{"commit":"e021464bf6dc7da42624fc7e752b4e2d8f75da54","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e021464bf6dc7da42624fc7e752b4e2d8f75da54"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-07 18:56:44.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4d68ed721a4c2d9b4338e72ca7fe0324b8be5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4d68ed721a4c2d9b4338e72ca7fe0324b8be5d0"}]},"branch":"refs/heads/master"},"2bbbcb41110c5657443843084e62178ab8440bc7":{"kind":"REWORK","_number":23,"created":"2026-08-12 06:36:02.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/23"}}},"commit":{"parents":[{"commit":"afdd114fa81645d9fe666a15b78feabe028777ab","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afdd114fa81645d9fe666a15b78feabe028777ab"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-12 06:22:25.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bbbcb41110c5657443843084e62178ab8440bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2bbbcb41110c5657443843084e62178ab8440bc7"}]},"branch":"refs/heads/master"},"9645e79dafe4ad08357c448225d1922ea4055a1d":{"kind":"REWORK","_number":24,"created":"2026-08-12 17:54:37.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/24"}}},"commit":{"parents":[{"commit":"afdd114fa81645d9fe666a15b78feabe028777ab","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afdd114fa81645d9fe666a15b78feabe028777ab"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-12 17:54:24.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nA two-layer gate controls access: API microversion 2.104\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9645e79dafe4ad08357c448225d1922ea4055a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9645e79dafe4ad08357c448225d1922ea4055a1d"}]},"branch":"refs/heads/master"},"62642b07bfff6927c61189ff7b30e079cfeab68a":{"kind":"REWORK","_number":25,"created":"2026-08-17 22:45:10.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/25"}}},"commit":{"parents":[{"commit":"afdd114fa81645d9fe666a15b78feabe028777ab","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afdd114fa81645d9fe666a15b78feabe028777ab"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-17 22:41:15.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62642b07bfff6927c61189ff7b30e079cfeab68a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/62642b07bfff6927c61189ff7b30e079cfeab68a"}]},"branch":"refs/heads/master"},"b4676f0b2edd618cd5a4dd23cd3215ea4add8086":{"kind":"REWORK","_number":26,"created":"2026-08-18 01:02:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/26"}}},"commit":{"parents":[{"commit":"afdd114fa81645d9fe666a15b78feabe028777ab","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afdd114fa81645d9fe666a15b78feabe028777ab"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-18 00:58:31.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4676f0b2edd618cd5a4dd23cd3215ea4add8086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4676f0b2edd618cd5a4dd23cd3215ea4add8086"}]},"branch":"refs/heads/master"},"96b3d2d4580474bd5de14de9405f770f138e7ed4":{"kind":"REWORK","_number":27,"created":"2026-08-18 06:15:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/27"}}},"commit":{"parents":[{"commit":"afdd114fa81645d9fe666a15b78feabe028777ab","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afdd114fa81645d9fe666a15b78feabe028777ab"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-18 06:09:22.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96b3d2d4580474bd5de14de9405f770f138e7ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96b3d2d4580474bd5de14de9405f770f138e7ed4"}]},"branch":"refs/heads/master"},"8cd3142687f646a8ec517af44ba01284863c51a7":{"kind":"REWORK","_number":28,"created":"2026-08-19 06:06:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/28"}}},"commit":{"parents":[{"commit":"afdd114fa81645d9fe666a15b78feabe028777ab","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/afdd114fa81645d9fe666a15b78feabe028777ab"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-19 06:06:09.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8cd3142687f646a8ec517af44ba01284863c51a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8cd3142687f646a8ec517af44ba01284863c51a7"}]},"branch":"refs/heads/master"},"7e55c221503d4d79165b8906420f150a497d959c":{"kind":"REWORK","_number":29,"created":"2026-08-21 05:37:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/29"}}},"commit":{"parents":[{"commit":"0b236c4838891fad82c5f97ec4a36d14d3c142f4","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b236c4838891fad82c5f97ec4a36d14d3c142f4"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-21 05:25:34.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e55c221503d4d79165b8906420f150a497d959c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e55c221503d4d79165b8906420f150a497d959c"}]},"branch":"refs/heads/master"},"fafdb11ef01432b97e207d89c0b8129d63304f87":{"kind":"REWORK","_number":30,"created":"2026-08-21 21:46:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/30"}}},"commit":{"parents":[{"commit":"4de0e21b4a422c105a492cfc237d343841942b7e","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4de0e21b4a422c105a492cfc237d343841942b7e"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-21 21:23:12.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fafdb11ef01432b97e207d89c0b8129d63304f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fafdb11ef01432b97e207d89c0b8129d63304f87"}]},"branch":"refs/heads/master"},"1c34a00eededbbfc9b51141bce5215e615d3e6b9":{"kind":"REWORK","_number":31,"created":"2026-08-21 22:28:10.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/31"}}},"commit":{"parents":[{"commit":"60bd1b7c94679490a188862f481db13ab49cd842","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/60bd1b7c94679490a188862f481db13ab49cd842"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-21 22:09:49.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c34a00eededbbfc9b51141bce5215e615d3e6b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c34a00eededbbfc9b51141bce5215e615d3e6b9"}]},"branch":"refs/heads/master"},"6ccf46616f25f993ce968ab03911f1d60cbfb734":{"kind":"REWORK","_number":32,"created":"2026-08-25 00:45:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/32"}}},"commit":{"parents":[{"commit":"55af6237d741a63ec936d752611ecee60a427228","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55af6237d741a63ec936d752611ecee60a427228"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-24 17:24:20.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ccf46616f25f993ce968ab03911f1d60cbfb734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ccf46616f25f993ce968ab03911f1d60cbfb734"}]},"branch":"refs/heads/master"},"4402d4dfeeaf070b0223dfea9cfa9612d4a62c4d":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2026-08-25 05:47:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/33"}}},"commit":{"parents":[{"commit":"565d899fc01b443ab6f5ecdc3e274b0f0c452d3c","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/565d899fc01b443ab6f5ecdc3e274b0f0c452d3c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-25 05:39:31.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4402d4dfeeaf070b0223dfea9cfa9612d4a62c4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4402d4dfeeaf070b0223dfea9cfa9612d4a62c4d"}]},"branch":"refs/heads/master"},"a31449c4226a6a04adb9ebd386f84b16f9e6b09b":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2026-08-26 04:33:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/34"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-26 04:30:22.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a31449c4226a6a04adb9ebd386f84b16f9e6b09b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a31449c4226a6a04adb9ebd386f84b16f9e6b09b"}]},"branch":"refs/heads/master"},"13fdabe1199aafdecfcf3c0fc9e723742a1e2cc9":{"kind":"REWORK","_number":35,"created":"2026-08-26 22:38:28.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/35"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-08-26 21:43:46.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13fdabe1199aafdecfcf3c0fc9e723742a1e2cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13fdabe1199aafdecfcf3c0fc9e723742a1e2cc9"}]},"branch":"refs/heads/master"},"c85fdc9a5dbb371284f80c0901d4a94e57c7b8fc":{"kind":"REWORK","_number":36,"created":"2026-09-01 07:11:05.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/36"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 07:04:36.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 73, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c85fdc9a5dbb371284f80c0901d4a94e57c7b8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c85fdc9a5dbb371284f80c0901d4a94e57c7b8fc"}]},"branch":"refs/heads/master"},"6d6e8957cd9283b795dbef85177fadf18888134e":{"kind":"REWORK","_number":37,"created":"2026-09-01 07:38:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/37"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 07:25:18.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 73, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d6e8957cd9283b795dbef85177fadf18888134e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d6e8957cd9283b795dbef85177fadf18888134e"}]},"branch":"refs/heads/master"},"4118c4f35e49cd56faa2f401aea3d920e82998d6":{"kind":"REWORK","_number":38,"created":"2026-09-01 08:07:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/38"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 07:50:54.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nA two-layer gate controls access: API microversion 2.105\nsignals client support, and SERVICE_VERSION 73 ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4118c4f35e49cd56faa2f401aea3d920e82998d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4118c4f35e49cd56faa2f401aea3d920e82998d6"}]},"branch":"refs/heads/master"},"ca7eb6a49602938dc389e981e5bf13836ec065f3":{"kind":"REWORK","_number":39,"created":"2026-09-01 08:28:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/39"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 08:28:48.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 73, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca7eb6a49602938dc389e981e5bf13836ec065f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca7eb6a49602938dc389e981e5bf13836ec065f3"}]},"branch":"refs/heads/master"},"6779530b5cce371caf812f782a9501222572f0f6":{"kind":"REWORK","_number":40,"created":"2026-09-01 16:42:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/40"}}},"commit":{"parents":[{"commit":"18660e15eb39287a01a3892c972b2fb1c12a5dbf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18660e15eb39287a01a3892c972b2fb1c12a5dbf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 16:41:53.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 73, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nDepends-On: I35738e2280e18b62564a90fdbda32e86fe5f179f\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6779530b5cce371caf812f782a9501222572f0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6779530b5cce371caf812f782a9501222572f0f6"}]},"branch":"refs/heads/master"},"3420b4d6c9cd50714eba75b36a5f2fd47555cda0":{"kind":"REWORK","_number":41,"created":"2026-09-01 21:14:51.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/41"}}},"commit":{"parents":[{"commit":"6cca020d22325d64139141939e13dbc3316cc3fd","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6cca020d22325d64139141939e13dbc3316cc3fd"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-01 20:21:16.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 74, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3420b4d6c9cd50714eba75b36a5f2fd47555cda0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3420b4d6c9cd50714eba75b36a5f2fd47555cda0"}]},"branch":"refs/heads/master"},"b1e37ba34d98063cf9889652ede4621132200b0e":{"kind":"REWORK","_number":42,"created":"2026-09-02 05:19:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/42"}}},"commit":{"parents":[{"commit":"6cca020d22325d64139141939e13dbc3316cc3fd","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6cca020d22325d64139141939e13dbc3316cc3fd"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-02 05:18:55.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 74, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude claude-opus-4-8 claude-code-2.1.258\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1e37ba34d98063cf9889652ede4621132200b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1e37ba34d98063cf9889652ede4621132200b0e"}]},"branch":"refs/heads/master"},"c2f163c0f4cca6542e4484fcdfe1b67d6e2c7c9a":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2026-09-03 02:12:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/43"}}},"commit":{"parents":[{"commit":"ed094b1f7f03a49fbdc3d3702b13791e9c87a89c","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ed094b1f7f03a49fbdc3d3702b13791e9c87a89c"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-03 01:59:32.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 74, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude claude-opus-4-8 claude-code-2.1.258\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2f163c0f4cca6542e4484fcdfe1b67d6e2c7c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2f163c0f4cca6542e4484fcdfe1b67d6e2c7c9a"}]},"branch":"refs/heads/master"},"fa1d7431d51749963d39c917f05116da9164e0d6":{"kind":"REWORK","_number":44,"created":"2026-09-04 03:21:57.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/989633/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/989633/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/989633/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/989633/44"}}},"commit":{"parents":[{"commit":"b7ebc1e615826fcf16f1def7422fe765dd48adcf","subject":"Extract share management from ComputeManager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b7ebc1e615826fcf16f1def7422fe765dd48adcf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-07-31 22:14:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-04 03:16:44.000000000","tz":-420},"subject":"Support cold migration with virtiofs shares","message":"Support cold migration with virtiofs shares\n\nLift the restriction on cold migration and resize for\ninstances with virtiofs share attachments. On the\ndestination, _finish_resize() grants Manila access and\nmounts shares. Revert cleans up the destination; the\nsource retains its mounts until confirm. Same-host resize\nskips redundant grant/mount/unmount/revoke operations.\n\nmount_all() now calls grant_access() before each mount,\nso recovery paths (hard reboot, stop/start) re-establish\nManila access without operator intervention.\n\nFeature gating relies on SERVICE_VERSION 74, which ensures\nall computes can handle cold migration with shares.\n\nPartially-implements: blueprint virtiofs-cold-migration\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nAssisted-by: Claude claude-opus-4-8 claude-code-2.1.258\nChange-Id: I2cda35fa0c2e080a67418d557cb9030b11f31394\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa1d7431d51749963d39c917f05116da9164e0d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fa1d7431d51749963d39c917f05116da9164e0d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
