)]}'
{"id":"openstack%2Fnova~994421","triplet_id":"openstack%2Fnova~master~I67315422eccc113390fedc905d3217a2d73367a1","project":"openstack/nova","branch":"master","topic":"bp/intel-tdx-libvirt-support","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-24 12:47:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"38744":{"account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"last_update":"2026-08-24 13:46:09.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-24 13:46:09.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-22 09:19:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-21 19:26:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-08-24 13:46:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67315422eccc113390fedc905d3217a2d73367a1","subject":"libvirt: Add Intel TDX host capability detection","status":"MERGED","created":"2026-06-23 10:55:34.000000000","updated":"2026-08-24 13:47:39.000000000","submitted":"2026-08-24 13:46:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":121,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"994421-bp/intel-tdx-libvirt-support","meta_rev_id":"42c7a5c302b3bf0bd5e03ee4c27c017c96004b4d","_number":994421,"virtual_id_number":994421,"owner":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 13:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-08-24 12:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-08-21 19:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2026-08-22 09:19:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2026-08-22 09:19:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-23 12:00:29.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-23 12:18:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-06-23 14:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-24 16:23:59.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-07-20 13:42:54.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-07-27 12:17:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-08-07 12:58:41.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-10 10:07:54.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-08-18 20:31:48.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-21 01:36:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-08-21 17:08:48.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"}],"messages":[{"id":"d8d7efc9420c86de08902126dba1ea502fc9903c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-06-23 10:55:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8619521f5d1cb93770ac00b8e57e524fea855b01","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-06-23 10:59:34.000000000","message":"Topic set to bp/intel-tdx-libvirt-support","accounts_in_message":[],"_revision_number":1},{"id":"19ddf2457d23bb9cf33911a74322cd97c5ee9988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-06-23 11:29:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"24dfd35fe3d485c558343a8900603b60365148bc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-23 12:00:29.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":1},{"id":"560e94de73ca622f6e40580ee4382cfd18533e5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 12:18:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98365e0d121d47b497f8d5f0397880e7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eddb143f6fbc40939f5fc3e1b6edfd48 : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd43165ae1e640f18d0428ac856bcdc8 : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/da65123ff83a478893c6422ab1f4860c : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11638dbee2eb2b0fae17ba8943caaafd50e992a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-06-23 12:31:31.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":2},{"id":"fcfe449afd9ad6bfd502ede57dcf0039258e9bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 14:24:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad32b673ab5440e69c9feaf4854e734f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8b94b8505c7436397029175d6515e05 : SUCCESS in 2m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81c4290d4ccf4e41bd9799a072e51265 : SUCCESS in 1h 05m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3d80e462e5d741edb793cd5c113058dd : SUCCESS in 2h 02m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa356df3ea1e48d0bb2349310c5d10be : SUCCESS in 39m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ab8d53b642f4e12985dc0fc1ce91b0e : SUCCESS in 53m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b181498323264d71a8637c6aed92cd21 : SUCCESS in 7m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d27817593d0c4ae7baae10a8f42286a2 : SUCCESS in 8m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa2dd8ab17ae473a9be28099b0792ded : SUCCESS in 14m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f5f6082168147ae977009d0eb265cd2 : SUCCESS in 20m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2263ca451e493ea9eddf8bdd3cedee : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52f22879d54a46279fc2b56f93093f91 : SUCCESS in 9m 22s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/60eb4b2c552b4c62b7ce062e12682437 : SUCCESS in 28m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5401b62341274f88a431c5d430a8e4a6 : SUCCESS in 1h 03m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d343aad486ca4257a1a3a2c7272e7e5e : SUCCESS in 44m 29s\n- nova-vtpm https://zuul.opendev.org/t/openstack/build/4fa42c3af6294208ac331d90ff6049de : SUCCESS in 37m 37s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4c172d410b7441239a439ad519385024 : SUCCESS in 1h 03m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b02bc25ed86647628e3904a19d5fae3a : SUCCESS in 56m 36s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e9b683d3d6524da985802a2ef10e20c5 : SUCCESS in 53m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/64907f739b7e48bf8273fdcdd0f44aa0 : SUCCESS in 2h 18m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/17d56b26d7274fefb8dc053efeb4d4a9 : SUCCESS in 2h 45m 37s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/3b2aa1bd752a4441aa9517eb04621369 : SUCCESS in 31m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9b6ee579b9a348bfa11c490528bdb02f : SUCCESS in 5m 00s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ed62aec1fb9d49f48f25cd1638750395 : SUCCESS in 25m 27s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ee58acf0490c431c93986773509a319c : SUCCESS in 26m 53s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0eca93c20fcb4897bfdc6464251a5835 : TIMED_OUT in 2h 01m 26s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff8968228315497dbaab075f0dc04281 : FAILURE in 17m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a784adfb20842cf8248b68c03aa3a29 : SUCCESS in 1h 01m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/94c4ff061a8e483e9e9383ad4b7ce90c : SUCCESS in 17m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/799804f3725c47baad1f1c3dcbe484a1 : SUCCESS in 36m 43s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4fce499bad6343ea85e6064cc8bad218 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":2},{"id":"076e4ec79a0a1c1328d8abed4f2e1388c78cd597","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-06-24 16:23:59.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e9f4ed455414ae83f80ce4125d4131686f0e744f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-06-26 09:36:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"70f5ddc18628b63921353c6a606b35ba9a1fb9e3","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-06-26 09:46:24.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"21846c337e6a354f0c775e1799dceefaafdfb094","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-06-30 16:05:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"263f9e4abce8ff0deb9c2697c4d35cbf05a852ca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 13:42:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0c77092633394ebebeb5b0571a51fd8791f18065","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-20 14:21:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"396785b10ab55829738fb55657150c5ec7189c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-21 12:12:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"28cc245a851ab305665099dbd01ca18c43503cea","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-21 12:15:48.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"427dfae14008cefb0bdf9c8d76a42678618c92f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-21 12:50:14.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"35ae3c95a5d9c63773e4ccda0f9446c213ae3965","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 12:52:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb4a675e7c443a181022208b81898e9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66fd50ea8e964d999ee5ea602fc4ce46 : FAILURE in 20m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d65145f551274c0f92f1b5e9e4930dbb : FAILURE in 12m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb8314b86c3c46bf99e93941d2f1d436 : FAILURE in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b8ddd5d25bcbbc1db555886ced8088e93538d0cd","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-21 12:57:50.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b15429b1a8638abe33d1f25791fc0ab2172abdfc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-21 13:08:29.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":3},{"id":"3cd6719183487d3f575a6122837108a0f3b93daa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-21 13:42:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b23e45ff2b6ef98e2ff254f3b0ef509518856677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:14:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3be075fdb394ce593270dc766b5e886\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/90171f23167c4910994c7c6a6078e6ea : SUCCESS in 2m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6950735258e49ef81418c1b6583f2f0 : SUCCESS in 38m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f71a1912adab4e499bff3377967feaac : SUCCESS in 1h 39m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb40caef26574b1ea531a6bd422e7a78 : SUCCESS in 34m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c4f0a43f94e42619ed4e4aa865c346a : FAILURE in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c335d92708ae401593d526e3fbe26956 : SUCCESS in 8m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08c9795994f84f648731d259c2fc047d : FAILURE in 12m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fe881157b4d846afb2eba61819ea8279 : FAILURE in 12m 46s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6b32035392414acbab6808bdb9627491 : FAILURE in 17m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99c43be2aa34892addde0ce40509d3d : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35afae79753e40229e96d96b906420f6 : SUCCESS in 6m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e88212fb713a428087a506c212a11cb9 : SUCCESS in 1h 01m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b8bdae9716064e21b9040e23294fa2f7 : SUCCESS in 1h 25m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/16dade0ee03246d080ca4e160b27124c : SUCCESS in 48m 13s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/0564aef4bbab498180d9fcc9e1ea5198 : SUCCESS in 53m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/09ec486a8ede44f9bb9191eba856a250 : SUCCESS in 1h 00m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0a9bf893624f4b1f8e0d404da2aec2c8 : SUCCESS in 50m 22s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9d07d6bcd38441639a37c528e6afbf77 : SUCCESS in 1h 07m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/62cea27ee2aa456abed89519279fc2e8 : SUCCESS in 1h 52m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/9797440833c143e28ce6f9567affc46e : SUCCESS in 1h 48m 50s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6aee94c1e31148c2904698f64d53d1e7 : SUCCESS in 52m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/31cd0a0201074996aeb435b559e58d7d : SUCCESS in 5m 08s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f73b44299f90495db8721a86b61acbc9 : SUCCESS in 29m 13s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c3fb29793f4846f6b03366b3cf6c80e0 : SUCCESS in 28m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4187042f34d7492cbb5de8ed07745883 : FAILURE in 30m 23s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/67e53ffb94c3425d8e919097ed2ca1a8 : SUCCESS in 59m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/355a7d3b4880476599b59cc062c5d7aa : SUCCESS in 52m 57s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/af233b09ce2545909fe6cbfe895d936f : FAILURE in 34m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e499eda7336a468b913d8a717710cf2f : SUCCESS in 19m 43s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c8a12cb3499f4468b02baba0c62a7e0e : FAILURE in 13m 00s","accounts_in_message":[],"_revision_number":3},{"id":"13f88f9d2925936a3b61aa36e117f25935c0f900","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-22 08:53:05.000000000","message":"Patch Set 3:\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"742ef59150bd5271e373161784a790ea06ed8f0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-23 12:50:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"a41191f94a198c68b6cf97551187fbd1ca45c387","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-23 12:53:14.000000000","message":"Patch Set 4:\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"40f306fc0d6dc77b200c44dc56b0aece99d69252","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-23 13:44:42.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":4},{"id":"367c45d013649a6e02525b38dae0bed03e733cb2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-23 14:36:18.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4c447f14e4b74a1564c39def24d5b1b5c0b96de6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 14:37:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7cbaf0e28194bc1a1861b50815f15dc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5244f0f85c7e4858bef1f014de240cb7 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/112f1a0c637c4d999850575631eb2eca : SUCCESS in 33m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7a43d524d359439d95ae495558adc7bd : SUCCESS in 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f1829f4685db2deb4a9326e5ffea27d09739480d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 15:12:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16d5cf1974e14c1caabd58890b5970da\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8ce75600fc874b7bac02c7d5301e7397 : SUCCESS in 2m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/febeb67d56d14a62b5c86fc46105d5f1 : SUCCESS in 1h 03m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/909fe7c7245242f5986d4493403d34a1 : SUCCESS in 1h 45m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56b4135d99c440668afca25f4325d5ca : SUCCESS in 1h 02m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5ce13cbb60d456694b381add16472c8 : SUCCESS in 1h 22m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de880b2897364e069a3249110f2088db : SUCCESS in 7m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79044b3c1b2344d8bc754d1ffbff7eec : SUCCESS in 14m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e32b2bde3461492394000dab1708e805 : SUCCESS in 16m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d653ccf749f45ad95df40a5eb73272d : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cea0fee1b394e34a5bfc9886122067e : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1373deed8d9940558c00ab8327bd083c : SUCCESS in 4m 43s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c08341fd72654ab1bd97e6bb0693daaa : SUCCESS in 33m 23s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d85e50dbdf694f38aa619828777a6ba0 : SUCCESS in 1h 15m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/717e57e7bf4947d39437d00f823e83c5 : SUCCESS in 29m 53s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a08a4c539a90447eb18b8c9ea5c46294 : SUCCESS in 38m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e6f17488565a49e38b1681c74aa069ea : SUCCESS in 39m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/262ad3f9c39e4174b16f4deec80c6830 : SUCCESS in 50m 29s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/e66cbb0c2b564e6f97e727b84fb780b5 : SUCCESS in 59m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a469534fbd824afbb048d9ac0eb23d06 : SUCCESS in 1h 53m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/6de470dc755541ef9062b0495abc6946 : SUCCESS in 2h 14m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/30562ae2859f420ab5b2d0bc78c0fc7b : SUCCESS in 50m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/33c27b7557f14feab1e09d64d4acc40c : SUCCESS in 5m 23s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/01b300d8af0a46e29ca1d2b87b1159e3 : SUCCESS in 30m 14s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e5eb11a779104322be0239ee5f19ec98 : SUCCESS in 25m 52s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/48564e66912f423eb271ec32f4794ef3 : FAILURE in 37m 33s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b36c2866ac0f4c88bf8f67241f82a587 : FAILURE in 1h 31m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a94c5c6d4249416d88ac3c27413fbb1d : SUCCESS in 1h 03m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/a874abd0ca314c1ea5c5c0aab2a88103 : SUCCESS in 1h 00m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/db805c44161a45f490337e856173c690 : SUCCESS in 38m 39s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3585a080714b4b798a9088deafb755e4 : SUCCESS in 16m 43s","accounts_in_message":[],"_revision_number":4},{"id":"f0f873a563951b3302d7ccde350805aec20a1b89","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-23 16:33:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6a159135f9ca348318b96de4d159c33688781a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-27 12:17:33.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b6758a4e9eff3b2dc851284c5c9cc84f2768b325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-07 10:51:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"786b51c91c04de9d636a0370317f0d9ea3c70dde","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-07 11:24:30.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":5},{"id":"0b5e826ac0cf06824d13e27be99f3b262203e24d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:42:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cee2ed052b04bbeb56d45ffd7bd4022\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9935ef8daa04ec0b5b8a74513b22fef : SUCCESS in 27m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/00a1b5dcabfe49e989a41432581e0709 : SUCCESS in 27m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/936233021ded40e89ca089278df0fef9 : SUCCESS in 48m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5586a3bce0e51af8c3ba3dd96c1036b20baf086e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:45:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07d8f075b8ea4ef7a40462d741899ebd\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6197fa686c524a97b92a89bfc62987c8 : SUCCESS in 3m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7fc9a36d5e54f7da6d533fefe98d499 : FAILURE in 1h 09m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cee97d8fb537435286afdbe83fcb6156 : SUCCESS in 1h 49m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d6e8f82ce0d4e3896446bdc9105742b : FAILURE in 36m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40ca832ce43f4388b8fcdccfa58b6cbd : SUCCESS in 48m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768a5bd39d5e4c319da209067859b525 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f41e47ed2d748b0b43fb478b3948d37 : SUCCESS in 17m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c80015d55ff34150a663a59e271c0f2a : SUCCESS in 11m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d27acaae8c834e3f913bc15c07aa9c72 : SUCCESS in 19m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0fdbbb3c51472badfd9174cf2843d4 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98dd03471fc649c3a7f3d80a40ae5154 : SUCCESS in 9m 45s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0f117415941840b4bac143f5997cc80c : SUCCESS in 48m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b158acdbd1974af28e01f1e7eb8160e3 : SUCCESS in 1h 21m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/14b65b14f5f4408da5f12cbb8ee92bef : SUCCESS in 53m 56s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/85e819c05245466ab813d2c9dc55edf4 : SUCCESS in 36m 07s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/43b684065fb3408d9ab17c92b1e93e32 : SUCCESS in 1h 01m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b71fb01a92ac43b68d4b7407262bd324 : SUCCESS in 30m 33s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b1a0b3c47e314cc1baa91c4dc3fce4fa : SUCCESS in 1h 04m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d1e8eeebb09f4d7d8107bc9e2593d732 : SUCCESS in 1h 41m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/0ad21f3693bb461aa9c2cca6f7af1dca : SUCCESS in 1h 24m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bdf6633897e044b3888713463a00c3d1 : SUCCESS in 31m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a60a11dd974a4a989ad2bb875a551a15 : SUCCESS in 5m 09s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d9623349d5cd4537a0f5e270f1f2e91d : SUCCESS in 32m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/26d242056c874aa5aeb981d88c0e2bf5 : SUCCESS in 27m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b0e73dd198e940369328f07444f7e678 : SUCCESS in 31m 39s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/988052fcc8a14891b6bd49d7a073b45f : FAILURE in 1h 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30121e4fcad847bb9e506a26bc4b0440 : SUCCESS in 1h 04m 32s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/71c4598cfb5243608cb05f39f742d153 : SUCCESS in 56m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a27a6d8ee28641748865eeb796c4b76c : SUCCESS in 34m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/21c92a6876be42c88ea5841b19d1995b : SUCCESS in 14m 18s","accounts_in_message":[],"_revision_number":5},{"id":"90a19a10bf4ae97ab7602f8facf3a814c170e84b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-07 12:48:32.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5b8ac4d661990e8a95c558583c5b3509621c087f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-07 12:58:41.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-gn7cf/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"68997c8c750bd59f10578044e7e407cc85c92310","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-10 10:07:54.000000000","message":"Patch Set 5: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4245b54fb73e29285817131e36c5859a55ca9fca","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-10 12:04:26.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"77f625d85f030559e84f28a73ddd31aefdaab05e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-10 12:55:12.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ba1627addc936bcf6b5d5089e4a6d5372bd8f30f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-10 13:31:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"19a1a05c52192abc849b41480ff5e459d8ccbe45","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-10 13:32:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e5e261fea73c164b20bd6d67d85bf92ff9d0f13c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-10 16:10:58.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dd7261ffc9c0b38312137aa6cb67efaa6837240d","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-10 16:13:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5a3a47edc528edfe2bf3a93ee9c465232f2143dc","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-13 13:14:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"820b30eafca6665c9430419ed6bd236acff8400a","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-17 09:14:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9b415b65178b97a0a28984e572b4a78b3a49c213","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-17 09:49:38.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-tgztt/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"b7a7d68bb7374a01e24ff072ee7afa9a14b20117","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 09:51:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a2a65e949a40779f75a5ccc83734f1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e478cb585b1149f99e1d80220e305beb : SUCCESS in 21m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e26465ba3094b499179f126d1fc10da : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/836df0bdf9234896be85a966d40241d4 : SUCCESS in 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d6ee8051f83e3f836849bb0319dfc8535312ce76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 11:24:01.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aebba686988049a28c1f880bc75ae65d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef5e2371bfbd4ed48a2efa42d6835ec8 : SUCCESS in 3m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/264ee4799f744275bb2a40f2679dd50f : SUCCESS in 1h 04m 47s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5be5e6a60d7946a5b1afb331ff8e6af0 : SUCCESS in 1h 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7b9860d59ca4852b7183d319def7215 : SUCCESS in 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0269ea31996c494ab89f660085f60720 : SUCCESS in 1h 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981a67b566ef4cb89e2f32358156e08f : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f45ffb88e8ec46b09b8df836fffb8fd4 : SUCCESS in 14m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bea69a3a9281488585a235461d35e858 : SUCCESS in 13m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ff579b37eba54f80a9944c2108214eff : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55dfa5ce78dd48a4beeedc6e25090846 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0417c737a9f34b189a1e5e289fcd7b39 : SUCCESS in 5m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1070d8ab03224984b7c0960490abdf97 : SUCCESS in 58m 24s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0410610be3564c62ab034bbfe65c3f4c : SUCCESS in 1h 28m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/50b0fb03b38e4252a63856586e34ce60 : SUCCESS in 53m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fb84663ac5b9477b8a6e71802bdeb333 : SUCCESS in 33m 55s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5c79e9e709a24dcca918371428b54ec1 : SUCCESS in 1h 00m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d1824ce079aa4c0d94e56c8daa1e3ce1 : POST_FAILURE in 31m 49s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9c7c4d2922cc4743a6f73e77a612b967 : SUCCESS in 37m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0baaaad4a8e64d558053f558a1fee27d : SUCCESS in 47m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/4ebc77c0213046ddb66355f6b760941a : SUCCESS in 2h 01m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/caff0c2fcb644b7e9acb6978bd4b1edb : SUCCESS in 41m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d1813e1b2b784945a6b6488c718cb62c : SUCCESS in 6m 34s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/32ad9746de7142ebb9d9393b9015f773 : SUCCESS in 29m 36s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/460eb47321184f34859ce6f42d6c3d59 : SUCCESS in 26m 44s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/69e2d0ad47a241b1bb8591bca42336b9 : FAILURE in 36m 56s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8b0bbf8482c40f994b5473ac76acd5a : SUCCESS in 1h 38m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b43fd4719e14307921f896092b0006d : SUCCESS in 55m 28s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1c5587bfb0c3427dac6ed15410e65c38 : SUCCESS in 57m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fd6e6711596446eeb836fed9684b8deb : SUCCESS in 21m 58s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/81043d0d534e45599356d863bdd46c76 : SUCCESS in 10m 37s","accounts_in_message":[],"_revision_number":5},{"id":"6f71261a4645c4b5f88944e88a312eb30e8abbac","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-18 07:52:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff945d2ba85886e43c3784add5785012455b9e84","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 08:26:45.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-wmf9p/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"ea4c396b5ead1404a7f118bed813fb03c0829aba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 08:31:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/511200922fb4469d9a1658318f3aed91\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5eca611fb2cd40a690f18bf97d5308ae : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c4c68c27e514788b1de368b81b14d1c : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9443c71893544f99a83b6530a7df2158 : NODE_FAILURE Node(set) request c2cfa24034e14ed3bb3f8f17bb08fd7a failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d9686897b977f325571701dbe50a4d385140ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 09:49:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab885a0326454c8288e9eddd6261ecc8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f66a652f2f944f79a5dd8b502da60c40 : SUCCESS in 3m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f5a3e96037504d4b9f558cf0b71c604a : SUCCESS in 34m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a755b35dbb3649f9bf6ed64f8c2c544c : SUCCESS in 1h 44m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fd9299d8566414c99ebfdf16f6805b0 : SUCCESS in 58m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e87362ba67d4440081d691de40f0f093 : SUCCESS in 1h 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b42dc4b54a694527a45c5f9849fb27ff : SUCCESS in 7m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a78d2dc97b64fe1ac7fae582ba4131d : SUCCESS in 14m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b90bd81fe43a48b4bb3f5808e16de7e5 : SUCCESS in 8m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8f8d3f66fcc9461a8d7ddc9531293127 : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9b48ba1e0e451593e1a76da1564303 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86c278ad98cb47b3bf6239fa4abff907 : SUCCESS in 5m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/fc9a9c096a6849f18ef88fcd88b6d1f4 : SUCCESS in 1h 04m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b9d7ba3566534ab8a87cdfe781b8c901 : SUCCESS in 1h 27m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8b79341547974ff694ee1dd2b3c818c2 : SUCCESS in 28m 31s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a612dd2c253447e9b210f5cad4bd3afe : SUCCESS in 36m 01s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c11302994e2545b8ad2b861e1f27e48d : SUCCESS in 1h 00m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/db6b637d1ae4428da13c8ca2d7550d5e : SUCCESS in 47m 06s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b011f559414c443f8b06d7da40d0defb : SUCCESS in 59m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/db0cecd54a7d45cf96b25719882eb072 : SUCCESS in 1h 45m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/451872fa8c114fa48dc839f7565d542d : SUCCESS in 1h 54m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/48a1f5578c5e4dddbfbbe53800d4fca6 : SUCCESS in 26m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ffbc48c8dc094693b10addee7a9a2e08 : SUCCESS in 5m 33s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/18244a6b22df46e5a966c8b4097eef9c : SUCCESS in 27m 34s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0173ab7ee85b4e2ca8d39cf03ad474dc : SUCCESS in 26m 20s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/756f76d4945341378abe36a9343e3e9a : SUCCESS in 29m 01s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/41b9490984ee43d1badbeb047ec9cf37 : FAILURE in 1h 28m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/784fa4281e5c4bca83816b4bbedc3c9d : SUCCESS in 56m 59s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f82334e6aa85458a8fb34ca78e3291e7 : SUCCESS in 55m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e7d8fd6451bf4242a557c8b267476620 : SUCCESS in 18m 38s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b427f7941b4343d2bc0153cee68e1dd5 : SUCCESS in 13m 43s","accounts_in_message":[],"_revision_number":5},{"id":"1ad93ae45f81083f1052de5a251aa4a1814d7be5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-18 11:46:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3854df22b90d5f12cda3ff8b0756bb6a55ed1713","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 12:11:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/537f91f5df9344a5942427afe98ff9e6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79be797ba53e445ca6eddc7a8ab6676a : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebbe36f46879467db76b3bc783e9bbc2 : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5c605fc88d6b43ffa748e43a204e02ac : SUCCESS in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ad9dfe76c783042594cf9e37bcdfde5d9e1bcdbb","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 12:48:05.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-zr464/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"24de1a5e0965a0c62a3c1ef25af13988a455a73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 13:25:22.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9756251462cf4d72b402c1602efd2307\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/489dad4f7a0a4410b7d27c9c4dce7593 : SUCCESS in 3m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/361f603413f948e0a54b4dc580c45b97 : SUCCESS in 33m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/741ce75674ff4c6f94bf78cf33a4ff8f : SUCCESS in 55m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f19ab58db3684aa9988948eb30ade434 : SUCCESS in 28m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c241fa8fbd894bf8896e6bde8b05a157 : SUCCESS in 1h 02m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c16490b3a6774414b1265b78d9f149d0 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31da36b906be497cb2a1dbbb89ce3b6b : SUCCESS in 8m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/15b4dc5bbae4464aa40f1d959323a3f8 : SUCCESS in 10m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2985270a63364de58d1052056e617680 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23a4eac5fb54093860ff525e04f248e : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21d587463fc64a47aa566820fcc6f736 : SUCCESS in 7m 05s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/023b7cbc9edb43598a351083f3d4697f : SUCCESS in 27m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8201ad7221146f9bc719a76e6fd1622 : SUCCESS in 47m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/522542c16d6d45a89eef48a244b24165 : SUCCESS in 43m 04s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a7bd8b9111d34e35833379b2e03e5da1 : SUCCESS in 34m 02s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/55e74002a48c4b8fbe844e4f20b55135 : SUCCESS in 37m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d9886353812e48a2945755c8de2ce64d : SUCCESS in 29m 00s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3dccae6777024eec9bb1d55902dd7edc : SUCCESS in 47m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ab124fd00e9c43b8b9a535f95d13686f : SUCCESS in 57m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/721d8305345248a99ffb96a894c0676e : SUCCESS in 1h 11m 41s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/fc2fef6017574596bcb1c1b6ac876924 : POST_FAILURE in 15m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5b7e4318ee704e7c9cac5f635753edf0 : SUCCESS in 3m 23s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7e2551c51629454a9f9cf6dccaec0347 : SUCCESS in 29m 40s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/db0886846a0549fb8f14e3d72207f965 : SUCCESS in 23m 45s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7b36872a5e4e4c17b1430c136e86c531 : SUCCESS in 36m 18s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2d814c063db4570abc8672d7f595212 : SUCCESS in 1h 32m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/005d68b14fd8412c84939a450f1a0535 : SUCCESS in 48m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/dc248d8aba25483097ba02cf08a37a24 : SUCCESS in 37m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9eac8359ba3a475898587ce531d4fac1 : SUCCESS in 32m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6fd201956e74468495d136e63b1e925a : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":5},{"id":"c194d9f50110de9910600cb276e80b975a9d8f78","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 20:31:48.000000000","message":"Patch Set 5: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"bc09fe14bc35b79a6e380eebd3425c83a032af48","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-19 08:20:59.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e631e58c4b4eb9e2e98279ece15d170a5795a640","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-19 08:38:33.000000000","message":"Patch Set 5: -Code-Review\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"307ec143e13b4fbb9381b2e953e51fa243b8ff20","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 01:36:52.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5a0eaca406f3fb1d8b357ce1d861746c51370fb8","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 07:29:49.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6044829da20bfa2bce19ab0af2f73b713248462f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 13:41:14.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"34120baec93a35607f5cb8afa12b7f57ec3e06d6","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 13:46:13.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1d375768e5a3a9cbf939854fbe1fcb5b11feb7c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 14:23:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9413486110754da2bab322d57e9a75a7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3289a8162835475ca80b6293505faa24 : FAILURE in 23m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a079ba9ec3a540a9b2a49c49ef65dff3 : FAILURE in 23m 20s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/19b6be6af1424b2eaea19eb292ae598f : FAILURE in 37m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0275473eab56af1bde1817628bf3b3c60d2b3a8e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 14:41:55.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":6},{"id":"e31da0e68ac300043f2a95209aea98cc3cd6e52f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 14:56:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7c16cb63f9def5dc3608ee83b1744b19d0ecd242","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 15:17:12.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-wbz47/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"6a5b974b64ad110e99294028d6f680c9d0905782","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:40:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faa305f971bd4826a8483b3e2f2a8112\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c84e0aff14174b57b7cee813ea667249 : SUCCESS in 16m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/42e3639c3033464db8d935b7161e6140 : SUCCESS in 25m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b552dac4d1254a1b8138789065ddca89 : SUCCESS in 40m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2090c9067b09a335686f6f772c0fff61281a1276","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 15:58:35.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I67315422eccc113390fedc905d3217a2d73367a1 .","accounts_in_message":[],"_revision_number":7},{"id":"1434be156ea61f3b199d3f77a8b0ca2dd6cf6630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:08:41.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa0108ff2924e8dbd5bbc5dcc96346b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8af2953fc3f44615a0d354692cca5271 : SUCCESS in 4m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/04810a8f9ef8401e81d40d7a28536937 : SUCCESS in 51m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4508614dc6b0456c840fbc0c09e5db05 : SUCCESS in 1h 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da1874be2e194ef48ea4b9032b00e1be : SUCCESS in 53m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a5296e70ed14ab18e6d336ceeefba3a : SUCCESS in 1h 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04ff3308fd7a41efa5871d7140d5f5af : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d52aa2594004cff8caaf3da764147c9 : SUCCESS in 11m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcc414768fad411e8d366bb104166a9a : SUCCESS in 10m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/466123c1a974444a8e51496692b56bee : SUCCESS in 21m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b313fdef0ce495b8c32bbcbb998f95c : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9fbcf227d14546a0ac68a03d4c73cecf : SUCCESS in 9m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d4e1c476a4f940b280e200ef72265828 : SUCCESS in 45m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c5bfa65911f0404abde6c6fa3f43c4a4 : SUCCESS in 1h 31m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8b8efa580aaf4c3990dc1afdb09e7248 : SUCCESS in 1h 07m 38s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/d326165f076c4c969dc76e19d1032137 : SUCCESS in 52m 54s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9dc3713884364aba9d684ca5fe6f453a : SUCCESS in 1h 06m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3bdbd61411f640a49dcd254b1acf6460 : SUCCESS in 50m 07s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/36183441f1e349b6aaa9f3beac630916 : SUCCESS in 51m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b840d1db207141a1b0126a04027ee3d1 : SUCCESS in 1h 44m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/8b66795e8a804ca1a18b746cc317c841 : SUCCESS in 2h 03m 05s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/681330903cb946a9869ba4ed9f92671e : SUCCESS in 48m 25s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4932ca556bca4715aa8fda24bd1d1c7b : SUCCESS in 5m 20s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e74c3a52e56043a18fc90b9505558d5c : SUCCESS in 32m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d11cbeeb559e4a849a1f16bc618b11eb : SUCCESS in 26m 43s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8e35a80273f24ee1899b8e580c359ea4 : SUCCESS in 38m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/01dfa516ddb642549dffddffe9465766 : SUCCESS in 1h 29m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f2558ba2970464f93ad60b3e0edf43d : SUCCESS in 57m 46s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4f5e2faaba454baba816b3eea1c1c451 : SUCCESS in 56m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1d3d67bda8084e31aaddeb746fd8a7d6 : SUCCESS in 25m 42s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/16c7a01f6e70470f8c785b5257889c27 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":7},{"id":"8c84fd4d1015aa079b9081b2fc4991c4a380b2e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-21 17:18:58.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"581f7be7a5afdce87622cfc93768bf19a199315a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-21 18:21:30.000000000","message":"Patch Set 7: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":7},{"id":"86c7e3e32124bf8f2d0653cd452f6d0a3abd263d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 19:00:33.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-5h4k4/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"a1436bb02c3590cb450fd0c9a080737482574aeb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 19:26:40.000000000","message":"Patch Set 7: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"36c7cd73ccbc75c635d8edc6d15f33e360cfc21d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-22 09:19:22.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"33e3d5211413ab02578e2bdb142a42da337b813e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 09:19:56.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ea79386523ade2de19a8f800d8f5c33706d1d282","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 11:32:06.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46dd1202be3b4dcc9ac613c3236f9f40\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/366bb59c11de461d939043c927510c7b : SUCCESS in 2m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de714707bf6c4ce0b017b9e1f5ef07a7 : SUCCESS in 32m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/88e8c1532c844ec0b1aed459e6bab1e9 : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/195d93fc26d24b86929c7237d0403491 : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584e6c2751f24e5794a6aeb45c2ca22c : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3cacaf9a19a46ce8981c6fb4d7ee0f8 : SUCCESS in 13m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75f0a99b769044d6bf034ef81c09b805 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b3329eab5c401ba4295de2c4978de1 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096890d787014a938de10d9e3acc4d36 : SUCCESS in 5m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/09a491604174440eac0f6dc71cf0bab8 : SUCCESS in 53m 14s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3eeaf9fc406041a8a657a53299792a63 : SUCCESS in 27m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6ad695c3b3494418bbe1c045bcd99c84 : SUCCESS in 26m 26s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4ffd49c3114d4ce08428c3cfda2a55fa : SUCCESS in 35m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/650b95d525a148d48709ae0c79ecf9d7 : SUCCESS in 1h 08m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/0b51e5dfa8964428b00d286981e22a76 : SUCCESS in 2h 07m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/060e72587b7943198498f22235119bd3 : SUCCESS in 4m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bdcd8b9e73eb47648df672c035282986 : SUCCESS in 1h 37m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7bca35ec04b341359787894aa24a9931 : FAILURE in 1h 26m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8efc6467f0f44ef1838032d565c7580a : SUCCESS in 52m 37s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/02914824066d4fa4b4b0279776786e21 : SUCCESS in 12m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d019a321451944aea67e83373a148340 : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":7},{"id":"ebea6df1cbb6e64f1cabe869f1b14fad43ba5e3d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-22 18:04:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bfb17b9538ecc74afa4e894e6a6338ca376b9ab0","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-22 18:35:56.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-cj8r8/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"43fafac420c52ac006ad48b9437507def95dea4f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 18:51:22.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccf794ddb3d645e288761672ea0d3b47\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad979957fb174489a53649295e7c8500 : SUCCESS in 19m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18b7fb08c02e4f6f80f516172991c524 : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5e063a21fd3e4014bebe0937a8c54478 : SUCCESS in 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c35523fb3da0709fd793abca49e82b22bc18452a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 20:09:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2789f57acaed41d38a2971c1789e9809\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/71dcc3f425ea4e22a3d336e2d1eac4e1 : SUCCESS in 3m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e01e4f780c0641b6878dde840184b4dd : SUCCESS in 47m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fcc888865370466d88888fc72740cc68 : SUCCESS in 1h 05m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/761a9add0a954dcbb57409399e4c7009 : SUCCESS in 44m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/483a31cb02624bfc88e34effde2a8b41 : SUCCESS in 52m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538cfa760a284d5e94f0e63ccdcc8778 : SUCCESS in 7m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b0130e90b544a00b014b54170727f2c : SUCCESS in 14m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a7becf479a824317aaea0a31a72f14ee : SUCCESS in 9m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/afc5a1bc77cf40e9a92e3e76b4a7794c : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2ac494792f49248001db404c0b157c : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2670f91a8a604c70a287ab73ec51db3a : SUCCESS in 4m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a8c4853fbb1e4fb5a9ae7daef12e0218 : SUCCESS in 59m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea4792ef323549b083079d6639316675 : SUCCESS in 46m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/db453c59b0ae432ab33cff69aed77725 : SUCCESS in 51m 29s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/725e3ca465444372b0880df767e51042 : SUCCESS in 39m 00s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f77c31504f6a4b7db101580270037788 : SUCCESS in 47m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3948195fe36c4fa697318204c1693040 : SUCCESS in 50m 00s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/42bd1f6ff2e84709a824f0158aa1f278 : SUCCESS in 41m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4dbb591934084a4ea05326ad4ad6aa75 : SUCCESS in 1h 43m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/6120acf717a740009d60813d0d2f3bdc : SUCCESS in 1h 50m 09s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/3970d4ad8d1d4d4992757d6ec8a35032 : SUCCESS in 50m 36s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/a8a2871474e64cab97d06da944a479e4 : SUCCESS in 5m 45s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5fac87b2059941dd8391f55e16f9f0e6 : SUCCESS in 28m 53s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/56b0fba821344f89ba2f97e8f2b22f70 : FAILURE in 23m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/102aa83568cc44b1a5ea8df32ccc970a : FAILURE in 36m 23s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f8ea54aff5c442c9b34930783cd4646 : SUCCESS in 1h 27m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da076790226548a6a2c0d0b63fe05c20 : SUCCESS in 54m 46s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/135041e6b91b431caf0839c45481e14e : SUCCESS in 53m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f50280df5a0f41a7818b8ba50c312642 : SUCCESS in 32m 55s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a805e5dd425841808e5c2abc80291b2a : SUCCESS in 13m 12s","accounts_in_message":[],"_revision_number":7},{"id":"55172b3e21e9a0136abeb8fdbe2aa7d1b58c81c3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-24 09:33:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c89c63eba2429a63b722f6b4c0bc169f02083365","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 10:17:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cdb7468c74e4e0c9ef1b987d4bb8727\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/685e67dff91441ae82e300bf6c6ee321 : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/331aa7fa722f498b8fdfb858468f3187 : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8ec82200b35e4e5f8f5dd5d874821732 : SUCCESS in 38m 17s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c62dc7985cb1850aab34fd2c9d20124a723b19fe","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 10:20:14.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-994421-n82zr/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"5ec114222dc2823bca0c5b67a47518889fc8fb4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 11:25:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3bf9d1f731f4132ad1f64aa2ed4f5ed\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b0a465b43b8141698c712544886b926b : SUCCESS in 2m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f572e8b18b77482dacc9bfdcdef3645e : SUCCESS in 1h 05m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2f357af8e6bb4cc1923223c12b7ec64c : SUCCESS in 50m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e674aad6f4f6405fb7419c7586d7e856 : SUCCESS in 1h 13m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02e976824d274ab780cc00be43b233cd : SUCCESS in 50m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b0b31ac21a4638b4089ca745f9b243 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8931d36da3ab4856a70fd6d284caa8d9 : SUCCESS in 9m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c449dc0db3db4331bcc214e6c610f8c6 : SUCCESS in 9m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c1dfcb776aaa45eea8d6acef3d95cb0e : SUCCESS in 12m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e2fdda46b945bb91e6cf035fd6342c : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2315e548715347bcbfae4c02b289da97 : SUCCESS in 4m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/deaf1dec165a41848e41a5c53c090bd3 : SUCCESS in 30m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/500189d29434448996c49d115bb9f055 : SUCCESS in 48m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/36670f6db669426fbcdc575da627dfac : SUCCESS in 27m 56s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/2ab8a8f29c91478d8123edc2794ffc57 : SUCCESS in 36m 04s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c05546e235c34e6b902d8d73fff1e0a5 : SUCCESS in 48m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5b98f7eb30f041f4902c0412905d0ed8 : SUCCESS in 52m 48s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5b592b4a80b649a6bcf0af0a3e261bbf : SUCCESS in 1h 02m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/809e66c8e6eb419b820d4cca4cd3b384 : SUCCESS in 53m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/81669117e60b41ceafef818c0bbc0230 : SUCCESS in 1h 46m 35s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2f11fde20ec5459dbbf614e6586d6121 : SUCCESS in 48m 35s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/28eee4d0c2054d76b1650d609056a344 : SUCCESS in 3m 34s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/527bf20c0ca2423bb014e0dcdef5f708 : SUCCESS in 28m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/214951ab431a43538a6596bcb5e52538 : SUCCESS in 25m 52s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b701dbe8be9e439883271accc76df0ba : SUCCESS in 29m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/91de4b911e484adcaed4860c37ceffe8 : SUCCESS in 1h 28m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01cbba9a265e48fb9425af9bec6fdaf8 : SUCCESS in 54m 00s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fde1a926af7540cbbac3837fbb42890c : SUCCESS in 30m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b68ed979186a4e378c7f57380c53f211 : SUCCESS in 23m 41s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4faa0d5274fa4f3c885213e2c9eec564 : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":7},{"id":"f75211dd71b9a1faf4d4cd3f61176e3a8d478ad8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 11:26:34.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1de3438588d9db0a81b8dc8ea6d263513f43b781","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-24 11:36:30.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"abfe7e03dd3d20ffd617f3b74f755ecca29f13a8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-24 12:47:47.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"558148502a5f74283809810994f1b68d44986f9f","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-24 12:47:51.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"d54ab7bc619a21f423e8623a40ffb8771f38c503","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:46:09.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63af5c6a17b84a829a3237d3061e9ea8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d17e2c1761aa4477857e3a3fa89f3f60 : SUCCESS in 3m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/736322acec2040419ff331bc5f399cdb : SUCCESS in 56m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f8c756358e7740c9b2cc7f7a73899ceb : SUCCESS in 1h 50m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65c9776a09c14853a86410532f0353fa : SUCCESS in 1h 01m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3566771ea57a4718b86fb8259cb7117e : SUCCESS in 7m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2f47d7d54d94450b58b350e3f6c0299 : SUCCESS in 14m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b1ffbafede9479b97e43ea510f3fda0 : SUCCESS in 14m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0034b6a42ed84276a8b32a7d293bc9de : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a675c0669e204c1b9de20e380e310329 : SUCCESS in 8m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a5503b5a70df4dccba25264589a189eb : SUCCESS in 43m 33s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2ec12e3a463e41b48e42c47c85c78b8b : SUCCESS in 28m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/408e73a8ebde4ff9918de25aba0a7e4d : SUCCESS in 25m 04s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4bf0ccc3b8c949e3acf84cca010b94db : SUCCESS in 30m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ff65b23bb1b14623bd6e5dc7c41baeb1 : SUCCESS in 1h 30m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/656f4a84277248169051073dc30cf0af : SUCCESS in 1h 37m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/63b11a2cccfe4bd58c39a7e6bbec73e5 : SUCCESS in 3m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d921df9494b4e6a855e95830b71df4f : SUCCESS in 57m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7bb0609a54f4dc09f4fdaba82e48d95 : SUCCESS in 57m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/862d43d499424bc29ec25b1f31b9eb8f : SUCCESS in 1h 02m 42s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/38a3fd8121394554aaa9e52a59a20a07 : SUCCESS in 14m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d7e392df3413442c920ca7ed2e76740c : SUCCESS in 33m 33s","accounts_in_message":[],"_revision_number":7},{"id":"6c49f2e7a7463c1f6a34deb31de928a1a931125d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:46:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"42c7a5c302b3bf0bd5e03ee4c27c017c96004b4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:47:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c28281fdf9746758e8cca067f87fdd4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206b5ffb0da54dddba86065f8df15831 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a69768e478d7a08355b11fba5a814b08e495ce09","revisions":{"695da54189d97c589fda1bfd9fed6ffa0777a7e4":{"kind":"REWORK","_number":1,"created":"2026-06-23 10:55:34.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/1"}}},"commit":{"parents":[{"commit":"3ac36dbfd1f3ac00c17ce8dbb4ae97ab120e153e","subject":"Merge \"docs: Speed up release notes builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3ac36dbfd1f3ac00c17ce8dbb4ae97ab120e153e"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-23 09:10:42.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using three mechanisms: the kernel TDX module\nparameter, libvirt domain capabilities, and the capacity exposed by the\ncgroup misc controller (/sys/fs/cgroup/misc.capacity). The latter is\nused to report the number of available TDX keys.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.43.\nBumps ImageMetaPropsPayload version to 1.19.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/695da54189d97c589fda1bfd9fed6ffa0777a7e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/695da54189d97c589fda1bfd9fed6ffa0777a7e4"}]},"branch":"refs/heads/master"},"025ee8a335a952101a7f350edd1683bac711341d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-23 11:29:04.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/2"}}},"commit":{"parents":[{"commit":"3880295b520794a4e173a9bfce7815036746f187","subject":"Merge \"Fix cover env HTML testr results missing unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3880295b520794a4e173a9bfce7815036746f187"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-23 11:22:43.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using three mechanisms: the kernel TDX module\nparameter, libvirt domain capabilities, and the capacity exposed by the\ncgroup misc controller (/sys/fs/cgroup/misc.capacity). The latter is\nused to report the number of available TDX keys.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.43.\nBumps ImageMetaPropsPayload version to 1.19.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/025ee8a335a952101a7f350edd1683bac711341d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/025ee8a335a952101a7f350edd1683bac711341d"}]},"branch":"refs/heads/master"},"7ed58c05098e80366d860788d65589d268e36d05":{"kind":"REWORK","_number":3,"created":"2026-07-21 12:12:44.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/3"}}},"commit":{"parents":[{"commit":"9d1dd3b8acd2df19d83692e0cade3911c17032b0","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d1dd3b8acd2df19d83692e0cade3911c17032b0"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-21 12:02:13.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using three mechanisms: the kernel TDX module\nparameter, libvirt domain capabilities, and the capacity exposed by the\ncgroup misc controller (/sys/fs/cgroup/misc.capacity). The latter is\nused to report the number of available TDX keys.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.44.\nBumps ImageMetaPropsPayload version to 1.20.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ed58c05098e80366d860788d65589d268e36d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7ed58c05098e80366d860788d65589d268e36d05"}]},"branch":"refs/heads/master"},"462e89e40c76139169c8f3e553ec84207d55cbcd":{"kind":"REWORK","_number":4,"created":"2026-07-23 12:50:20.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/4"}}},"commit":{"parents":[{"commit":"9d1dd3b8acd2df19d83692e0cade3911c17032b0","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d1dd3b8acd2df19d83692e0cade3911c17032b0"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-23 11:51:23.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using three mechanisms: the kernel TDX module\nparameter, libvirt domain capabilities, and the capacity exposed by the\ncgroup misc controller (/sys/fs/cgroup/misc.capacity). The latter is\nused to report the number of available TDX keys.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.44.\nBumps ImageMetaPropsPayload version to 1.20.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462e89e40c76139169c8f3e553ec84207d55cbcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462e89e40c76139169c8f3e553ec84207d55cbcd"}]},"branch":"refs/heads/master"},"8dfb7d5943563a998f91d1a4a9a15f391bd72a93":{"kind":"REWORK","_number":5,"created":"2026-08-07 10:51:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/5"}}},"commit":{"parents":[{"commit":"9d1dd3b8acd2df19d83692e0cade3911c17032b0","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d1dd3b8acd2df19d83692e0cade3911c17032b0"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-05 09:40:35.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using 2 mechanisms: libvirt domain capabilities\nand the capacity exposed by the cgroup misc controller\n(/sys/fs/cgroup/misc.capacity). The latter is used to report the number\nof available TDX keys. Intel TDX also requires CPU mode to be\n\u0027host-passthrough\u0027.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.44.\nBumps ImageMetaPropsPayload version to 1.20.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dfb7d5943563a998f91d1a4a9a15f391bd72a93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dfb7d5943563a998f91d1a4a9a15f391bd72a93"}]},"branch":"refs/heads/master"},"f87b89ac900bb58f64a4b1a98adb3b4ae1f84562":{"kind":"REWORK","_number":6,"created":"2026-08-21 13:41:14.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/6"}}},"commit":{"parents":[{"commit":"9d1dd3b8acd2df19d83692e0cade3911c17032b0","subject":"Merge \"Remove disabling the scope enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d1dd3b8acd2df19d83692e0cade3911c17032b0"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 13:29:03.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using 2 mechanisms: libvirt domain capabilities\nand the capacity exposed by the cgroup misc controller\n(/sys/fs/cgroup/misc.capacity). The latter is used to report the number\nof available TDX keys. Intel TDX also requires CPU mode to be\n\u0027host-passthrough\u0027.\n\nThe config option [libvirt] \u0027num_intel_tdx_guests\u0027 can also be used to\nconfigure the maximum number of available TDX keys. This is to\naccomondate deployments where cgroup misc controller does not reflect\nthe information by the host.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.44.\nBumps ImageMetaPropsPayload version to 1.20.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f87b89ac900bb58f64a4b1a98adb3b4ae1f84562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f87b89ac900bb58f64a4b1a98adb3b4ae1f84562"}]},"branch":"refs/heads/master"},"a69768e478d7a08355b11fba5a814b08e495ce09":{"kind":"REWORK","_number":7,"created":"2026-08-21 14:56:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/21/994421/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/21/994421/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/21/994421/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/21/994421/7"}}},"commit":{"parents":[{"commit":"07714e5030cdbda8d92dd0e2e8540fe0ee058cf0","subject":"Merge \"Unify the thread pool executors into a central factory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07714e5030cdbda8d92dd0e2e8540fe0ee058cf0"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-16 13:34:51.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 14:48:20.000000000","tz":-420},"subject":"libvirt: Add Intel TDX host capability detection","message":"libvirt: Add Intel TDX host capability detection\n\nAdds host capability detection for Intel TDX to the libvirt driver.\nDetection is performed using 2 mechanisms: libvirt domain capabilities\nand the capacity exposed by the cgroup misc controller\n(/sys/fs/cgroup/misc.capacity). The latter is used to report the number\nof available TDX keys. Intel TDX also requires CPU mode to be\n\u0027host-passthrough\u0027.\n\nThe config option [libvirt] \u0027num_intel_tdx_guests\u0027 can also be used to\nconfigure the maximum number of available TDX keys. This is to\naccomondate deployments where cgroup misc controller does not reflect\nthe information by the host.\n\nBumps os-traits dependency to version 3.8.0 to use Intel TDX trait.\nBumps ImageMetaProps version to 1.44.\nBumps ImageMetaPropsPayload version to 1.20.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: I67315422eccc113390fedc905d3217a2d73367a1\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a69768e478d7a08355b11fba5a814b08e495ce09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a69768e478d7a08355b11fba5a814b08e495ce09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
