)]}'
{"id":"openstack%2Fnova~995840","triplet_id":"openstack%2Fnova~master~Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa","project":"openstack/nova","branch":"master","topic":"cyborg-vgpu-support","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2026-08-20 18:53:46.000000000","reason":"\u003cGERRIT_ACCOUNT_4690\u003e replied on the change","reason_account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-08-24 16:55:07.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-04 17:04:07.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2026-08-24 16:55:07.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-08-24 16:55:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa","subject":"Add libvirt driver MDEV ARQ support for Cyborg","status":"MERGED","created":"2026-07-02 15:35:25.000000000","updated":"2026-08-24 16:59:09.000000000","submitted":"2026-08-24 16:55:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"995840-cyborg-vgpu-support","meta_rev_id":"00cf507f3a6c5b23a57f0c5c8d4f306fb16609fa","_number":995840,"virtual_id_number":995840,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 16:55:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2026-08-20 18:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2026-08-21 18:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2026-08-24 14:21:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-02 16:48:48.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-02 18:31:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-07 17:30:44.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-16 00:35:43.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-30 22:21:02.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2026-08-04 13:52:48.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-07 14:08:45.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-08-17 11:56:25.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"CC"},{"updated":"2026-08-18 18:13:08.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"adfdb00f54fb74cd01de8dc5f015e62610912819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-02 15:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86f27c0046d5e634e2ef798b7e85f16538fb71d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-02 15:59:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"103663e5f351f063f30aac68c8caf557fb248557","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-02 16:48:48.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":1},{"id":"4c046a57b006de25010424ffa1e78e8abed5719d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 18:31:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d764695cb9384dbab4b933e91336657e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c773d4a3df424b70a0f2872e5d3ee74a : SUCCESS in 34m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9303991caf4748fe994e2ea695269456 : SUCCESS in 1h 58m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e27bbda5e919436fa983d0c92ec70b84 : SUCCESS in 36m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/924c1b1691114f6c92bb7b75a376e805 : FAILURE in 1h 00m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07716824d10f4e10bb9b950a2470371b : SUCCESS in 9m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/655c49cdfe294686a8a0333b4b55836c : SUCCESS in 15m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9da2925f55fc49f392c20ac7c9c04f7e : SUCCESS in 14m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/97d752a10b73491183b12aea2139e096 : SUCCESS in 14m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5f41b1dd0e7410e83f5aa6d8e8eaa95 : SUCCESS in 12m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/732b985779e3412db037ebe728aa5f6b : SUCCESS in 8m 28s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d28baeda0b6044c69be15e6cd49240d7 : SUCCESS in 52m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe8c165416d54200aa1cec08370117a9 : FAILURE in 1h 28m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ed7dc997fa474f99982e330843514204 : SUCCESS in 59m 20s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/7b46776008e240599a3fa612909087a3 : SUCCESS in 48m 38s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f3db2f1548df4a7798bc8048dccaab12 : SUCCESS in 45m 33s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e5f62d89cf95474da3563c67c4b6adb2 : SUCCESS in 50m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/77355ff532084addb7892311dd510516 : SUCCESS in 54m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/09a31814479c4f0eaf9f02327ee531b6 : SUCCESS in 1h 38m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/7c37a5b24baa431895fa214995e1a1fd : POST_FAILURE in 2h 25m 34s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ab937825ba7241d79f35348956f9ca2f : SUCCESS in 54m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4cf1f1c53ab745e39b0f48599f41c3d3 : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/74a768d0355e4f5bbde50f7153594e5e : FAILURE in 26m 31s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/28c1082ea0794d00b2eeeb9c1849ae6f : FAILURE in 29m 53s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0781af2909cb4ad98720f4405003e691 : TIMED_OUT in 2h 01m 17s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/94009fe980fc47efa1852d89d8318fc0 : SUCCESS in 56m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/262d6d844de842e0ab0fffe88275b13d : SUCCESS in 1h 02m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/34aed8f60c454b6e99f2ac7f932f9310 : FAILURE in 30m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/101b21e5c4954abd9968247233227a04 : SUCCESS in 31m 09s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b2890950bf00430682e8b4b040872775 : SUCCESS in 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"cda864d9239edf0ccbcf8a46c8eff7a7d632efee","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-02 18:49:04.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":2},{"id":"b9bbf5388dcca34df5cba39a7fbebb02dcd63213","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-02 19:04:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2bfe3adab149ee96b2a669a1de2931\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/88935bab9109476785051542254b8d4b : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/86c4a0309a2d45538eb564d341cebd35 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/21a0b60c03b34a34bf46532ef5fc3e51 : SUCCESS in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0aa0ab57de06950fe84470d6881ff3eefaeec623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-03 11:33:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ef1bf7c1a6eabae03222a8edf31821277253abe5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-03 11:46:36.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":3},{"id":"cb300da67b964df9394952293a51a6903cfc34c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-07 07:34:01.000000000","message":"Topic set to cyborg-vgpu-support","accounts_in_message":[],"_revision_number":3},{"id":"fdfd0f07566a27c67419499340e7affd347e4698","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-07 07:34:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f6034d65f368990aac720df17df2c3277492aef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-07 09:54:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4020f531a459d92aba8e548ca0987b25c83dc520","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-07 10:56:31.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":4},{"id":"1825cb157cc45404b033b1ad6870ca199dcb4ec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 12:35:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b02eafa455c447828b4e9fbf7b860c0b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e62991b9c93e40ac9c0119b0ba881c31 : SUCCESS in 1h 07m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9e3fe6767fc044bbb4861e83eebc0acb : SUCCESS in 2h 03m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29f3ce4b67cf4a0a89a1645ff60d0475 : SUCCESS in 1h 09m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7205a8478cef4b0db9b96735c48f0ae8 : FAILURE in 37m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7c79d36a21478dbbfccd5b337dd5db : FAILURE in 9m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e689014c95844a45bbd2d5e9fce397ca : FAILURE in 15m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0913ca7f89ce49ecbd2933fd1477719e : FAILURE in 13m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b484e5e892ce4b1c9257eb94a5ef33c5 : FAILURE in 17m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dab76b606f64e27baa9ef7d1d06569c : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502e9a81ad5f4137940a668565c4acf9 : SUCCESS in 8m 07s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f68fa82152064b2ba422a85f4bbb61f9 : FAILURE in 1h 27m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c6a0571c20984d268fb15229ee3bb1b0 : SUCCESS in 2h 01m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/04fe2229fe2342ee9362c110934194cd : SUCCESS in 59m 27s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/324459240f514bfb87426af13781b9be : SUCCESS in 43m 26s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a5f8d51315ed41ba976ac95680a2deff : SUCCESS in 1h 05m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e433cab353e54ffa9ab390f66c430f0f : SUCCESS in 54m 06s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b3e162ab1d43421b8bdf5a94553c3929 : SUCCESS in 1h 17m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fbcea6b5b4bf4ae2b13e5a03d2d92350 : SUCCESS in 1h 54m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/45dd688e26f04b6aba8b0c855fd41c0b : SUCCESS in 2h 29m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0536f12ce22946d7be643d4c7b5816c9 : SUCCESS in 47m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0a1f946bef52495fa4064f1aba9e6ca8 : SUCCESS in 6m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a5aca573a7f6454aa8fa9b77ad23ab2c : FAILURE in 33m 13s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ae2837a4ae0c4a7bbc6535a7cef9e7bf : FAILURE in 23m 42s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/94d7e6885d144b88be0489d9e8769d2d : TIMED_OUT in 2h 01m 56s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9e9ab506f10424d94c926d8725a5493 : SUCCESS in 1h 33m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c01a55329b541308f8fbb30b8868cc7 : SUCCESS in 44m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3549fa7b73f1416cae3097f84248dc61 : SUCCESS in 33m 10s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5acf947c678d4016aca3866be223ef9a : SUCCESS in 40m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5f8fd25795984178945c298e0270045d : FAILURE in 14m 01s","accounts_in_message":[],"_revision_number":4},{"id":"7fa3d3cdfb4ab39ec647595c70cb18b58932227d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 13:54:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7958ce5a51094d0493a2d6046bb9cb4d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d7293699b0d49f9aa55b9788273081b : FAILURE in 34m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5435780ca8684b73aac66f9dbade8425 : FAILURE in 22m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8c64cc68b85a44579ffe8e953415d3bc : FAILURE in 17m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e9b2cac0222602c965b271db0b33ac93e9c94a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-07 14:43:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"f4b446b8e12594de535c92a3eb10714444954dc6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-07 15:46:00.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":5},{"id":"6803e2213da9b91dcf9e0d126fa10ce279a14613","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 17:30:36.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca003e06cdc84cf0ad6b1f7411469ec1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/005f0de399f64ef99dc45d1751db85c0 : SUCCESS in 1h 10m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5734daae694442dd8bc38d0bd66a70f1 : SUCCESS in 1h 56m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82269ee7d9d3424aae5986f60a50a31a : SUCCESS in 1h 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ea3b462cc8048e5982e1a04621ee47b : SUCCESS in 1h 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5dfc3ef11f148679614ef1c927b6f6b : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/737fcd2c589642a0983e6f59cdf0ba17 : SUCCESS in 14m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50356220fc654ebbb3e0a4949aef5e07 : SUCCESS in 16m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3f5810159c649a7819e4d19671d5a10 : SUCCESS in 19m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e12d629fd3dd42c59297be494a0120d1 : SUCCESS in 15m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/054b0339ca47491e87b42bd4f261cee0 : SUCCESS in 8m 32s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/611bafd2ba9942c6be45af0c6459d1fc : SUCCESS in 50m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70e431ec87df4556ac7674ae07bda715 : SUCCESS in 2h 03m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c2a115398875414cb64108871391ccd9 : SUCCESS in 1h 01m 09s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/dc20d7a1000d4dcbaacbf05e54205e95 : SUCCESS in 43m 38s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b20ac67507ea4bd9854b1878258dcd99 : SUCCESS in 1h 11m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5eaafe3dfc824819ac5303b2a48124fe : SUCCESS in 59m 39s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6494bd609c0245f58796a8fe306af344 : SUCCESS in 1h 18m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fccedef70a34476f8aa790f5f729ad83 : SUCCESS in 2h 08m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/8968fd6408e44afc9d92cab028d48b80 : SUCCESS in 2h 33m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9d8ff66dff28442683399d59b598fa42 : SUCCESS in 55m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/905d9a84c4a7415ba540ad6cb2484e04 : SUCCESS in 5m 10s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/46bb7841c4df4c869ef0ddaa2f7d28a5 : SUCCESS in 29m 15s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cfb664d4a42549ff9027d8b1be61f7b2 : SUCCESS in 28m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5564f9a9b1ff4e7681ec98ee7b04bb10 : FAILURE in 54m 19s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f612d92ea004ddc9166140fb944e098 : SUCCESS in 1h 35m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e08ab17eb3b4ac39686e5c1141095a0 : SUCCESS in 47m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9fc680f537b74f6481a0ceb54169f1cd : SUCCESS in 30m 57s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/cf8a5d30dc7249818acc35e97ffd1349 : SUCCESS in 34m 33s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/56df92b4c4414b87852b88b46ec141d7 : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":5},{"id":"f63c40a8489daf46643bc33d0fd4825a566f638c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-07 17:42:04.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"6798f693e1b90f682619fec4cbd986979b3a4bd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 23:12:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/719da85d5a194154951f729c345894af\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8077091212f7435f8529e77a310d1b2b : SUCCESS in 24m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7c5ab09b1d244e40bf6871f070deea3f : SUCCESS in 25m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/55b8664376d94210b6c08b4392f6f294 : SUCCESS in 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7aa1561214501e7df4da012b70e0c05a8f4d39d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-13 13:12:49.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"4196cf8e3d3df92a670534d40b907cda36a03bc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 13:52:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5756e444b6d24322a9a7cded92bde181\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee830fcce18c4b6a975f7018bdfa902a : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1036adab63ef44148d0dd626cc435a9a : SUCCESS in 22m 08s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/44cf14f9920c430b98476dceb751b344 : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e6300f6a6234e4e9c4eaaf28bee56fbaa63fa424","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 14:06:38.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":6},{"id":"496febfd1796c9c6ab57cff9c5d0c2657f9734e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 15:05:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a8e114b16ad491daea6f4c20c0a84bf\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/977855966f4c47c692a289c10b61c2d0 : SUCCESS in 51m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a78009f935b249be8075133841ccf2b5 : SUCCESS in 1h 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c2488e8c62242cc892fb918d7cee83f : SUCCESS in 46m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7c94d0ba6be485b93a9aa0a2dc2801f : SUCCESS in 1h 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/074f4fe16e5a4a22acbfccb432fd0884 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3329172832c546549f860eea95799a84 : SUCCESS in 10m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f38fd6473c14724ba4867990cc8cc0f : SUCCESS in 10m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c5d6e2f5ed4846ab8b9e4a0668f39501 : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31452c3a5fe497a89f0c9f574363dca : SUCCESS in 13m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1baea66b1e69447bbc50b2c86ee76dea : SUCCESS in 7m 37s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4da2d6242290492fba2c59ad923b39b7 : SUCCESS in 1h 05m 34s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ce9d0a3d47646ecbc9e93e3d726a68f : SUCCESS in 1h 04m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c3c165ac6264433d8094cc25cb33ad88 : SUCCESS in 39m 43s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/039695869daa4e14a56aeebe47284c8c : SUCCESS in 36m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a743114743e748618bd6fa37a9d0c89d : SUCCESS in 1h 05m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4c13ce9004d84690ab53c0fc08bc4fa0 : SUCCESS in 29m 04s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/38104352aa3743f6bde62add815d3f1a : SUCCESS in 37m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6f016e60ab8a4efc8593fc4bd0062fb9 : SUCCESS in 1h 34m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/0af3c781d08d4ff98bce110c7a82ecd3 : SUCCESS in 1h 43m 39s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2d1e828e681240b9b0c65b558df27dca : POST_FAILURE in 41m 47s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/16536954e9574c09851666f37be073bf : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0647a2fa375445839199352438c30722 : SUCCESS in 28m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/14f89c604f664942ba2ee857fd0a20ed : SUCCESS in 21m 10s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f0dec9178d0f4b86add3064aea1c7369 : SUCCESS in 1h 09m 53s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/33971705aa034b0fb7b1d115d10e4035 : FAILURE in 1h 02m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbda0189b1dd4f92ab2ef212002a5775 : SUCCESS in 26m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/51d3f238e11f41b2ab1158f43fe376e5 : SUCCESS in 38m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e11e1491b6614aa18cff950103c174d5 : SUCCESS in 19m 27s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/59abc84ff578405382338661949b9b7c : SUCCESS in 14m 32s","accounts_in_message":[],"_revision_number":6},{"id":"e4ecc9fc1d7969c01a0925ebfa526d49852a596c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-13 15:35:35.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"169ce4e3bed61f7bc0beb2355c1eb60dab14fc97","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-13 16:30:27.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":7},{"id":"de22ff24eb3b2364cf709ea3085d01bef5b62890","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 17:07:16.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff6966023714a998c68024520abbad1\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4941e91da8e484eaf5e3f7dcc11d26e : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a56963d82b4944aaa1c3c56c5d33d1e6 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/513529450e264cea8581ea711ae43a38 : SUCCESS in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f94d69e5b6886094c1b88a947839f861c472ef52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-13 17:33:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a696b4f1b66b42bca97676a426448023\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/44e2dc341db749a5b6b4d30502e787fe : SUCCESS in 1h 06m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/02a659c484fe471d82c1e4f1edcc34d4 : SUCCESS in 1h 42m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f87a744f8f54b7aaf0040868c95fd50 : SUCCESS in 1h 05m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f536bece3d408eaf95097fb75478f7 : FAILURE in 1h 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421c4347d3424ebaab4a1d11c3d987cf : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/658e2942802a4fac8517c3e2887c2470 : SUCCESS in 14m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5a85970aa3e14ec3b0129e493985a734 : SUCCESS in 17m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/29dcee023f924d09bd09856d5cb4bc45 : SUCCESS in 17m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ea1c9f75164523a930021ebebba052 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62177b35bb314ea3a447285d90705c5e : SUCCESS in 9m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d6d90ee348c6450da3dc2dd8b1d73d7d : SUCCESS in 1h 04m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44bb7aa22ea14ba18ecd9a22dafc75af : SUCCESS in 1h 34m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/40705212d7104cc58874dd37fd4e2da5 : SUCCESS in 35m 42s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/03ddc7c5466544c4a624278ff80ca26a : SUCCESS in 41m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/02fa3b93166a4bbb992317b0f1f49060 : SUCCESS in 1h 09m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2749a24506f143ee91dab786d7137552 : POST_FAILURE in 37m 19s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f4e236dfd61d4ca8be4150c591faf054 : SUCCESS in 1h 17m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/45d67cf648b649d1978d7cb005ff10d6 : SUCCESS in 1h 50m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/1de3419e68454bf98327b3b2ae745ad7 : SUCCESS in 1h 20m 26s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/79bac704dcf24fce8a9f02d7b5341d16 : POST_FAILURE in 21m 01s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ef77f900402c43aca823ac2baea31746 : SUCCESS in 5m 35s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c9fe069fe7d64a2f83fa7186cc8bc5ec : FAILURE in 21m 35s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/84b010297eee4281aaaa99802f565aa5 : FAILURE in 20m 48s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/16b89985c1b04cc68356b37bac0a4ecb : FAILURE in 45m 18s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ba2ecbe43054413a5a2ebef2c6210dd : FAILURE in 1h 44m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2202a073bfa4b6094373cd876c8ccae : SUCCESS in 28m 51s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b614846db65842a5b8c44f1e2b374403 : SUCCESS in 39m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a9265b0f24a34399aa91e97be5057c58 : SUCCESS in 25m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/74d33c5e3282462d92999e23a5841e52 : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":7},{"id":"9a540b56d3dd5ded7375e4575af7409b31084688","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-14 12:59:37.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"54f8a9884f838a27e15344d25e067f3d9919c96b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-14 13:54:35.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":8},{"id":"82e0e09727b3ca8179d82e6f7b60092c41eae4d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:05:25.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b88364256e24170a6e3a0262e5f1d65\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7abacf77e81640daab74910e09082dc4 : SUCCESS in 50m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6a630f477870458a92310dd0eaac449d : SUCCESS in 1h 46m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d28dd65e63b748b0925c494bdedef397 : SUCCESS in 44m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a567ca3ba5f34e38b1b658fc8f5e6262 : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f1c5dc6f54412bb15906ad642a8d91 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d652c63ac90b4cbe835cedcf92b954ce : SUCCESS in 13m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1d91daadae847ffb3fa9957270a0d16 : SUCCESS in 14m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3c074de6a6664506b500ed774ea1937c : SUCCESS in 14m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c698d7b4b24d0fac82d8239127308e : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74ab9c08b548444a9e0174d65ca41b75 : SUCCESS in 8m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/99d58f0b346943c9a08e6ca81e2062b7 : SUCCESS in 58m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8dfb986be62e4c5fa15701420a5092d7 : SUCCESS in 1h 23m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d2eef2b9a9c44879b20fb326fb6b91df : SUCCESS in 57m 31s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/833931a7f2ad49ebb483561c3ecdc31a : SUCCESS in 37m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2799b50a78244d95bd7f97c005643684 : SUCCESS in 1h 00m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a83a0073740c47f7b94fcfb2a16619da : SUCCESS in 51m 24s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fa327d1b76064e92a87a7572c69bc10c : SUCCESS in 1h 06m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1423b95f0f03480db4951987104110d8 : SUCCESS in 1h 44m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/bce73a19288a47d3b694bf27789cd243 : SUCCESS in 1h 56m 07s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7275e156045a4cffb10c99c4b0a08e56 : POST_FAILURE in 39m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/baf1f62cf3504c51ac7fcd954cae4999 : SUCCESS in 6m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c065141b36fa4edf8b5fe48f28cbc74a : SUCCESS in 27m 21s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0fc2a73fc60b4a468dfbd96464edf3e1 : SUCCESS in 27m 04s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7dd360c2d9a046c2884daf5db615b8da : SUCCESS in 25m 44s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/35f3fecc76c24b78b2a37930ab230b49 : SUCCESS in 1h 27m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59adba15598343baa6b08bed8abecec8 : SUCCESS in 58m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ba446904c295403793a9a1c487ace8bf : SUCCESS in 38m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6f98ef9f846b455f99d29e83270260d9 : SUCCESS in 31m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/683cdf19cfad4021852aeb30414707c5 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":8},{"id":"270f35a4e4e8ef3e1120e6a48cf86555906f9a84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-14 15:13:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/005f8775877a41d895cc60873914fa09\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a763ac3ad5e48498bfebb2d2680903d : SUCCESS in 23m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2d16a0115aab46488a6128375adcbd01 : SUCCESS in 25m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/661e6fb569284a09906d7a5e56c2f378 : SUCCESS in 40m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"da6f59f1f45dc5b18aacc10dc53972dcaae916ec","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-16 00:35:43.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-wx9ng/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"a0ee6b5dfdf5418105d0d173c9f4750bc6eacde4","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-16 05:03:42.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-bk68q/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"edfd3a3f1e325c9676997db8629a82e68f604ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-27 14:57:22.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c5e83da898a10ec56409bcdf4893a3ce77d1e0a6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-27 15:00:48.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":9},{"id":"b50859c95913c94f28fc746fdddb9529529a2819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 17:17:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c168580a41ea43f1ba22c87f245dfba8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/504ce86af97d48928814c106d48e780e : SUCCESS in 1h 10m 51s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2d0bffec3ead4e72acd3478412273442 : SUCCESS in 1h 42m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2a8d1eb84734c3d93f2ee36b6abd239 : SUCCESS in 1h 02m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a938da46d6b46e58a10bce64397ef9b : SUCCESS in 1h 24m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6cae83ef4c46a88311606bffc9f8ae : SUCCESS in 7m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6745ea3bc0694e1c94b22dc795181474 : SUCCESS in 14m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4ab3519056b4027be10570048c0670a : SUCCESS in 13m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe712b40ff4c4da18a991c5f9b0aa45b : SUCCESS in 17m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a11935b82586474481d06391fd3423e2 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c65b81655fe4fbdb1f3877a78a7e3dd : SUCCESS in 9m 30s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/04c1b90a0cef4bf3bdc24f7ec70be15f : SUCCESS in 59m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f73813d418ce47c8bd2c7b73863ed64a : SUCCESS in 1h 33m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3644ff8f13b547e4ae093dd422f039ee : SUCCESS in 44m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6e97785865d444ddb6268cafde84317e : SUCCESS in 41m 16s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0f693592dabe4860bbc0808ddd79d0b3 : SUCCESS in 1h 05m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9b67cfb5a4404f64a293248538315fc6 : SUCCESS in 55m 34s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/897a4237b8df4d88a7ec18904aca4d44 : SUCCESS in 1h 02m 57s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/da2c4eba9edb42ddafd808ad975921a8 : SUCCESS in 1h 21m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/63883b07434740d1b90324a18a41f688 : FAILURE in 2h 10m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d2962ee413d8424b91d018d4612a5f8f : SUCCESS in 47m 26s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/46a134b26d50426a9a59b724ef75d8ea : SUCCESS in 5m 35s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e4cc8e1d4e2e457bad1dd368171a9bf1 : SUCCESS in 32m 16s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/c3d0ad207d074097ba2db747439926e2 : SUCCESS in 26m 52s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/53d38946b81745458b9dbbb3f39c1334 : FAILURE in 38m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f338ef0ef6274b05a2148de31dc5866a : SUCCESS in 1h 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9dd0047dba14002ab8e8adb6661f7e2 : SUCCESS in 1h 10m 03s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/01c7d4634807463bbe27a48059c09bf0 : SUCCESS in 57m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/27ccad008ef14f6a9b10bd875da67a8c : SUCCESS in 34m 58s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d6091c70ef0d48bdb6a99a2ffd83f49a : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":9},{"id":"463b71759beac2f83ad901816491c8cef274ff44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-27 20:00:41.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d63754490e45efa74c1b41d3a7b667\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f5148010c7ce47738a704448a067574d : SUCCESS in 23m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c0545602e79541769838d9e4802d5f04 : SUCCESS in 25m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/99895747a511496ca2517858edf900e4 : SUCCESS in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"769f374e7520fe4a5dee31133d5954e6e51d6a81","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-28 09:16:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bb27d5785d143953dc7fa9a90f7b2fa5c5e75ee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 09:40:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5528078ef6894171b8ae32fca93345ba\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/207fc92441a04731af34fca9124302f3 : NODE_FAILURE Node(set) request 001370c9ca734a72ac427006b1adb4ce failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6ba7dc4af4e142f0805e02aa789324a6 : NODE_FAILURE Node(set) request 29a728ebeee4435a96502976a1e565cd failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/62d32b75f04244d1a4584d7f7cae3c5f : NODE_FAILURE Node(set) request 5353294ac7bc4d12b074f50027cd0472 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2e1602846e023f58c741234bafc55df6bbbe3c59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 11:44:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de7f2fe406404df586618e2a5e6bc34f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8b018a4732149749680bca4ad03f832 : SUCCESS in 1h 03m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d50b8dcba4cc4f00bf4b460f04f0d5ad : SUCCESS in 47m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb8b6fe2c5ac458f9500b9f384ea8472 : SUCCESS in 58m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d44943c69044b8eb2f450aa5fb6e40b : SUCCESS in 57m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa32e5c883b4991ae84904bfe73e9eb : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b3abe07f8b14902adae0de060bb332d : SUCCESS in 15m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0347089172147069221d37428eade23 : SUCCESS in 13m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fecbd57c401b47b28b4dc855f3d806b6 : SUCCESS in 17m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d52b994b3a9e46a2a0420db76ff24e57 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e2612abef1f4ffaa1348337fed1ebc9 : SUCCESS in 4m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/474a8411121f4e72a6b7660094a0a3df : SUCCESS in 32m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e85c035415444afb0a81d04eac9209f : SUCCESS in 1h 29m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/216c3692ec1c47efbd140f666b52f57a : SUCCESS in 24m 17s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e6d2ccee36794559b706939208eb6114 : SUCCESS in 37m 26s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5f440c32d61d49a4bad6a12af15179bb : SUCCESS in 1h 01m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/71cbbed7cef04d23b378e9ec53ddbf69 : SUCCESS in 48m 29s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/18f3247f97c54249a95e34d87175478a : SUCCESS in 1h 10m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/21e336ed8a7a41aca6e8a74065bda4a9 : SUCCESS in 1h 32m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/22207c938d8748269a7b57fd74572ee4 : SUCCESS in 2h 20m 29s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/79d27822b03646bdaf70a377b0f5ceea : SUCCESS in 29m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ef039eb87cb14475a7bee42b43031992 : SUCCESS in 5m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/74bdd70e7a0347ab811aae26bfd32ab0 : SUCCESS in 29m 41s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8e2987c2a1194937b5ba384d7eba2271 : SUCCESS in 25m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c022dff94e354ea78b831417657cbdb8 : FAILURE in 30m 08s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4b6977383284f8fa584d726344c79bb : SUCCESS in 1h 39m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afc3de76b1954ee0bc80e1a493ad3056 : SUCCESS in 57m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/47db634fdba8420daf6ce03202d0a94a : SUCCESS in 49m 15s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1f99764879294b6390224aa4602b8755 : SUCCESS in 37m 21s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7f52df3aaf6c402e988266e257340d57 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":9},{"id":"64d000db1149712cf25ead860f944a4e6a8699e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-28 11:55:03.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"75f57c22e31f57c85dbe932750f333ab20019b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-29 09:48:26.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"632adde7460f6e419fc372bb300669c5b154e4a1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-29 10:44:47.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":10},{"id":"b95dc352c36b7869da5f0ed7f48c5e5a7c3cb5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 11:56:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f65a71384d4e4380ad08ad2c277698aa\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7383f444bd0045368aec40b69231359b : SUCCESS in 1h 05m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cbde1100f75e44e9bdb07b3d25037b78 : SUCCESS in 1h 28m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d32bbb97e4f44c7a9566b319ac071d8a : SUCCESS in 1h 00m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e30c84bec52d41969d0241fa5405b327 : FAILURE in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e08647216def4daca23a1ed6aa8fb843 : SUCCESS in 9m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5973ad07730345de8df3929274fb4392 : FAILURE in 13m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e922b5af28254c739e85f8bfc267996c : FAILURE in 14m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/64f5975175ff4b6b842d54a36ee042bb : FAILURE in 15m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949b19d6801f4ba0908831abc811bed4 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3476433346c45689f35cef1f9b55dff : SUCCESS in 7m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3726b7fdba4a4c089a663a33be3ebf01 : SUCCESS in 46m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02fd7b56cead467eaf98840573174550 : SUCCESS in 1h 36m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b42b79219e3c41b9aebce571ee45d27f : SUCCESS in 33m 50s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8580a9f78a6240f2aa04afac025996af : SUCCESS in 37m 16s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a15312044f574b71b21b6bd86fe8f7c7 : SUCCESS in 1h 03m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ba752fef5cfb491a998e0adf1fcd7808 : SUCCESS in 26m 10s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a5c857f36c674e6ab1c2981ffcc1c0a1 : SUCCESS in 42m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4167dd9f70f74b7aa9d7accc88d8500a : FAILURE in 1h 57m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/f15dbc65173a4ca7ad6e07562f03388a : SUCCESS in 1h 38m 59s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/83074f9c4fdd4af3915db510d62fd78d : SUCCESS in 52m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7e31df290d2a468bb72f1e45129d35cc : SUCCESS in 5m 27s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a0476564c5b4757ab43e0058eb5dddc : FAILURE in 28m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e1b6feabe7df422f8e4cb12226dc4fd2 : FAILURE in 29m 39s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c56499d4ff20430f97f1b598d42e60ef : FAILURE in 39m 43s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec349c6d2abf412d9f6934eec08ba2e9 : SUCCESS in 1h 29m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3660ce2ac0ff45d5870fc7b9acb8fad7 : SUCCESS in 58m 22s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/28ac2d2ff5ad4a8d99dd678bec3865aa : SUCCESS in 58m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/eb2dc949b91142869b183be1ccd59745 : SUCCESS in 38m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/81484518cd90473c834ca2deb1397bc2 : FAILURE in 16m 37s","accounts_in_message":[],"_revision_number":10},{"id":"3d4d093e12db7ecaba0b7b9e6e80805fe2cae7d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 13:20:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7c6c080302b40c3af4d7a88fc574db7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e299c9db9f474661aeabd6e1f9bcaed8 : FAILURE in 21m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56870c73a4844d708b16a38211180cf9 : FAILURE in 43m 10s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/429471d5b04e4a879e9aaa5e85829ca5 : FAILURE in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"695f13f69e4ec042e34d0056852fa1a0d2a5d2b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-29 13:50:57.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"0c6b1737a800f6ccee99ca1b6e14c28579ec6a79","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-29 14:46:51.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":11},{"id":"9b077e183baa28e05e795e283ca6456716876bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 16:14:09.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/960946ae5a104876a4b96e62d9ad969d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7999e6c1ab2444eea77834c0c0ad053e : SUCCESS in 1h 01m 59s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a0bb12dce90443ec8dbbadc9e85ce825 : SUCCESS in 1h 35m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3250d0dea34b456f84b24c102a760281 : SUCCESS in 56m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ad5f234b23244029e71fd0a944689f5 : TIMED_OUT in 2h 01m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56468f8fe4644336b4985cfb5a8c9d38 : SUCCESS in 9m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fae6a31575444781bc87b3054d745f3c : SUCCESS in 14m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0deb897283f04f47968b429268a4fad4 : SUCCESS in 14m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb3acbc8911844728b4df6d994ec3055 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0035642de33b458fb98a508d0b3e4ddc : SUCCESS in 14m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1eb5d846d63c47a49978fd87c1dec4a7 : SUCCESS in 9m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bc655285225b44bb96b4bb916a660d59 : SUCCESS in 55m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d864503b014045dcb79e6caefe5f08f4 : SUCCESS in 1h 49m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d08fed15269d4edda9421c403504d30e : SUCCESS in 49m 19s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/5b0a8d14186749d381258ff0dcf6718a : SUCCESS in 45m 21s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/edaacd0994854669a9fbe03b718a7931 : SUCCESS in 39m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7a9d50d03dc94693bdd4cb1eb24e15b7 : SUCCESS in 28m 10s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f952530ca84041cea51036d6c162c2c0 : SUCCESS in 1h 10m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/56d8e9213ebb47d4b48641f0b3241505 : SUCCESS in 1h 57m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/61d1c5f47d3846088670e2fa4a0c10d6 : SUCCESS in 2h 09m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f313b0b764a8452380ff1bac532524f1 : SUCCESS in 40m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e1c15f8e30954434acefa21ab9308c72 : SUCCESS in 5m 21s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2184cdc8223644259fa0438abb19ceb8 : FAILURE in 58m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a252fbe2c7d5414aa688925ea4b2cd69 : FAILURE in 30m 48s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/98f49894b15d4fbf89e723f42816fd0a : FAILURE in 35m 57s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b45c7a1906fa4b05901cd924ee0d0298 : SUCCESS in 1h 29m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/018ee319f8aa4953a724829bd532b090 : SUCCESS in 40m 59s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0cbefc40ad364912ba5b8ab3e8513c7c : SUCCESS in 57m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/71297ef04890475581a28ef8815b9a17 : SUCCESS in 39m 05s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d3a29750d5534218a374e45e2c4262ae : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":11},{"id":"d3363c8ffdf790cd36665ddc30ce150431a6f1ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 20:53:58.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c27a26b967444e298ed74eb62625e3b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dcaa0058bbe64503bc4e6b3a81112bbe : SUCCESS in 22m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6071d9c1e3b54b8d81d688c44268bfd2 : SUCCESS in 36m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/20a51dca21e342bcbea630ecbd5cf3a9 : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"86060f18402321f037efd5ec054ed7869258437c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-30 07:40:11.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"5e213be16225c0e204cb290cbbbed776afcd02a1","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-30 08:27:00.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-n4t9v/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"41a1438d36c2a50dc997c54c2c1dd4fe3d66e672","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 08:36:36.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":12},{"id":"b9a32bf7d4a50f1dd0680ced1e75fc458818aebc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 09:16:23.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a70d74bf3f84cc9b0dcd2603b1f5245\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be4b785ec8b84e13b3654bdd41dee01b : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1965df90ff544ac29907c10d1ea27c97 : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9c5ea2af3ab44ae2b1db61de21d5c931 : SUCCESS in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3e6585533aabcd498091f11fb044afa14fe08f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 09:51:58.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1f2d25a888f4b49a3e0017833fc94ad\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56156192f38a441488af9a329d94ba35 : SUCCESS in 1h 02m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/faf2284985ab4a5d943e2b5870a73917 : SUCCESS in 1h 49m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/634610e7bca1443e974e171c2d1caf6a : SUCCESS in 58m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64d391d5b3e9457abd846b0f44b4b4f1 : TIMED_OUT in 1h 41m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d771a8731c0f40c3a4ff638f676706da : FAILURE in 7m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f4a8f96bb1642dfb0e4b24db96e510c : SUCCESS in 15m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24bd199280014726949a36287212339c : SUCCESS in 13m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5618fa18e3174c2da76df41c09554aab : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35686386b1943cb926eb7680bbfbba8 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4b7d92fd35a421fb289aaeb566e84d5 : SUCCESS in 8m 26s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3933df6d15df4edba263581bcc4022e5 : SUCCESS in 1h 02m 32s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb2b619c795341a292a2f52e69bc8ddf : SUCCESS in 1h 22m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0c784198fd0d41ccb5197bfd85b89885 : SUCCESS in 45m 08s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/164337b1ca02496d8ce796ca763bc41f : FAILURE in 20m 23s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/411abd09512549bab10592e2f18f4592 : SUCCESS in 1h 01m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5e8591a78802497095fb0134a7f3f611 : SUCCESS in 48m 54s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/935145fda6bf4999ba27b0f38610376b : SUCCESS in 58m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b600c43ab5ee4828a7ddc310676e4d78 : SUCCESS in 1h 27m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/cbe407862064425bb3a25ca355d19ac1 : SUCCESS in 2h 05m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5def497d6a3f4ed7a26d0335547afff4 : SUCCESS in 53m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b5631e1c955e43bd9d66a9b8fac07492 : SUCCESS in 5m 21s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/43ce6da38ab54916a7600ed93e184354 : FAILURE in 26m 45s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f0c9e632ceb34200940b5dc94826b699 : FAILURE in 26m 28s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6f81364b0f6d4948bca335d1cc7acebf : FAILURE in 41m 31s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea371cd936dd44c08aedf58fa846d049 : SUCCESS in 1h 31m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0391e55c8b284b7f80fb1292449e4fbe : SUCCESS in 54m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c6b4ad2e39334cfd89809f42fd686b99 : SUCCESS in 45m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/29c1341279da456d97cc12de23517b9d : SUCCESS in 36m 39s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fed804b244944bf5a3f235f09374aab2 : SUCCESS in 14m 09s\n\nWarning:\n  Comments left for invalid file nova/tests/functional/integrated_helpers.py","accounts_in_message":[],"_revision_number":12},{"id":"5adc13090be6bfc9e61ee7cc3d320e36fbd6f2dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-30 15:22:45.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":13},{"id":"0ebd3643843579ea5e56aa06108ae372adc07f13","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 16:19:40.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":13},{"id":"95c5546c00d288dfdb2d8b26bed27bc51f5df648","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-30 16:36:56.000000000","message":"Patch Set 13:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-nl5r2/index.html\u003e","accounts_in_message":[],"_revision_number":13},{"id":"88712424197d1d900dcb0c35a53b4c4447054643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-30 16:44:06.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"6b4c9486931310ce7ea022f73c758353eb7549ce","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 17:39:07.000000000","message":"Patch Set 14:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":14},{"id":"d2bde53db2b93f82511cf5b0b528526912890b90","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-30 18:00:45.000000000","message":"Patch Set 14:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-bgn2h/index.html\u003e","accounts_in_message":[],"_revision_number":14},{"id":"12e8c3a2afc7876399bfeebc9e6d521c111e969a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 19:03:56.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a7d3edc4ae84ba7b33487e037f1acc5\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05850139e0f749f8b166ee8de1f525c1 : SUCCESS in 1h 03m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d17296729c594253a4f92bf4dddd3be5 : SUCCESS in 1h 41m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c83ae9d196e4d239adfcb294f04a0a0 : SUCCESS in 31m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fc4dd00a49b45719d82ac7089d702cd : SUCCESS in 1h 02m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2474aea3b871496cb58d08bd44df97ca : SUCCESS in 7m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db5b8a1f59144b8f904c44459560f2b4 : SUCCESS in 8m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bde5529e873a493eac3ce4e4aace22df : SUCCESS in 15m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c0410274ab2544b7aceadc3260171027 : SUCCESS in 17m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7d581a5d5b54862b264bb9fc8ebfc54 : SUCCESS in 12m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df559a00af0e4c6788ad4325e2cd0027 : SUCCESS in 4m 05s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/19ed7e681d1c45569e2ee20a3c181a28 : SUCCESS in 56m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8eb8405745454729801abfb5f441c288 : SUCCESS in 47m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ce37e7b13de6497486c527d3b5702936 : SUCCESS in 46m 48s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fd6f26f7b9ff4333a1d79a6c950152dc : SUCCESS in 34m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f22a337cb6ac4f83bde1d0efeab17cfc : SUCCESS in 40m 23s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/23f74ba5ba0b442f9769cfa4cd435aef : SUCCESS in 29m 17s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d0d01a90c9394ea8b8230e788f83435e : SUCCESS in 49m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/02be5eb917224263b328a5b891b1b1fb : SUCCESS in 2h 12m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/a8a2fbb5d7a74fbfaa81aadb71ed006d : SUCCESS in 1h 00m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4b4ff488d62a4ac6ba640cbe5933c52f : SUCCESS in 49m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3ff685ed1e414798b53716003257b71a : SUCCESS in 3m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a47f8af371964800bb05511669ff3838 : SUCCESS in 27m 18s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/78613085a9774199881f78dd16f16236 : SUCCESS in 27m 22s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6ebe42e48b254d8aacd741dd933cecb9 : FAILURE in 36m 29s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4bdc2a97f48347ec896a7da947ddf950 : SUCCESS in 1h 30m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f3d63db48d4439eb5e61985309fa039 : SUCCESS in 28m 33s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/187a998e8bf846b792da6d1c598cc0a8 : SUCCESS in 53m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c6e89b2f15484cdfa896873aadfbfd49 : SUCCESS in 38m 18s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/49403ff172504249946ae680568a2214 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":14},{"id":"be66ffa55f3bebed82fe33fa662f9b766af79db5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 19:23:38.000000000","message":"Patch Set 14:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":14},{"id":"381ae94e1c9a04cda4275803b286b69787c684e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 20:42:53.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3ce8fc517314822a133c4941d78cbdd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c8c2a9fdd2c9446aa4f5810e81f39b1d : SUCCESS in 38m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3c084645aa164fab96cd6f249be63679 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7bbbc163e1fd4affb2f149f753d7c864 : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"dbe5bd5dfcba98415f82373ac3716547e22e4a3e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-07-30 22:21:02.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"3accfff35da04d4169deb52ae1d9ee582590c50c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-31 09:30:51.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":15},{"id":"1605958db30f4722a8de09e9c016a1c5c903fcff","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-31 09:31:23.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"dc9bf9ad9b4b49f5a7087bdf705578c22685171d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-31 10:24:34.000000000","message":"Patch Set 15:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":15},{"id":"aa791da66ff29be8a6e8e7bd8fda7b1e3ed98671","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:15:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afa3ecc94f24fcb8ade3d39981b3eff\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7af092e3baa14640aa8b82eea7769ef5 : SUCCESS in 21m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/320a60a870ce49c49934afbde6d8f73d : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/83849ca6c9cc46fdaf0d0ec515b23999 : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"da526e68789796c4f1d64da67c58f24354cb3039","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 11:43:13.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dac34c515b6942c187ee7f7cd1539b5c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71f601f307c34d3ba996ee9131d0261e : SUCCESS in 1h 12m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a5033e9ed79a488aa9a3cbc8d8073acd : SUCCESS in 46m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcee303b51994d4e8d58fc39c8a45420 : SUCCESS in 55m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12ca49335c5c44d59eca843865812c8c : SUCCESS in 1h 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd57f3b9f9f24b21b21e32b6b775ce98 : SUCCESS in 8m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f438ba0916e345f89639c0b656dd4b68 : SUCCESS in 10m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bee269fb1204e10962a5351107ef8c5 : SUCCESS in 8m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1b5c39ff4bd64a96830245618bd684a2 : SUCCESS in 18m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837da72a278e4d6398315d936814cbc3 : SUCCESS in 11m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2771ce36dcc1433ebf7c2ccda2d2221f : SUCCESS in 9m 16s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d7abfd1ba4fe42a78e87f261e95a5941 : SUCCESS in 58m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ddf016d3a20c491491521bda85734afc : SUCCESS in 1h 41m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b406e440bf45408abee0dcb112e3803d : SUCCESS in 1h 01m 48s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/40c08690c9d84db38baeb8ddbbdb9597 : SUCCESS in 44m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/569397934c9843128ba1c8168f9783b3 : SUCCESS in 1h 06m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/679f6768384a46f2873c9067b6f2256a : SUCCESS in 24m 27s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/5c13d307dd244c35879264b5b9c93c08 : SUCCESS in 49m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d0f7b5a6ee9541fea226f84b44145fe0 : SUCCESS in 59m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5e0eada08374bf7aaf9f052075bc2d6 : SUCCESS in 2h 04m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/488f9413ef6749febf48c558699b0b3e : SUCCESS in 46m 59s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/af48d3cd683d47bca4e2e2df8e6d84d3 : SUCCESS in 4m 56s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/7bddd1193d2f48efa7e60602058a30ef : SUCCESS in 27m 40s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/602e0dcd4bbb425b8e110915f7bb026d : SUCCESS in 24m 36s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1dc8c31d2d0e4f799c781fbc343afe75 : FAILURE in 24m 49s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e035292f8c8e4df1937051580f58354f : FAILURE in 1h 05m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9a61b0a44384ad0b573d216c6052e90 : SUCCESS in 52m 31s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/029bef53fda94217bbacbe0a5dd9ac92 : SUCCESS in 54m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/83a4ef7f92574be1aec44f594181376c : SUCCESS in 33m 50s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/449dea5b57e746e78d47cbec4ca4e64c : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":15},{"id":"d592b2ee39ffc686b6e37a487b65c0deb24cc395","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-31 11:56:36.000000000","message":"Patch Set 15:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-pvvl8/index.html\u003e","accounts_in_message":[],"_revision_number":15},{"id":"956b4432868bdc3335fc2d5babcdd2f9ac720025","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-07-31 12:21:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"73ca6f3b78a8c42f26eb72ace9a70643868e683b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 12:57:23.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6d92ee5d35d4f728037010ef031c176\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98f8a4fee1d44bde81c1e51459fc394d : SUCCESS in 23m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6a1cff1f66c7486e831c9252f5a6ce11 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e779245b055d41e8900c6c525d69ef14 : SUCCESS in 34m 04s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"195751c2cb434bedc880ca4f6c7cde0d3339d189","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-31 14:04:36.000000000","message":"Patch Set 15:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-cll7c/index.html\u003e","accounts_in_message":[],"_revision_number":15},{"id":"db0beef57a3127bbe505f32cbd5353d0c483a14a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-31 14:31:30.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c2796603da944ec9242754684753e08\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6918e8ef83074e499c8bb11c8173416a : SUCCESS in 1h 07m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/53193e41e0ca428da2aa6ad902d9ac84 : SUCCESS in 44m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72a5d8fab2c843b8aa4cafc7973cff65 : SUCCESS in 47m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bcb5ed11f9e4ee5855272757966a50b : SUCCESS in 48m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac59c948df2423e9360bf18ea552e5b : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21f2acb752af4fdd9bd7157daf5a4a31 : SUCCESS in 15m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d492f1f4374f4a108bf686f971815c00 : SUCCESS in 8m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5083e67639ac470fabbc5dabd4587914 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7c0605ba6774ef6a43a59d45c8af80f : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b8aa51177404e7aaa0be05c84e62979 : SUCCESS in 8m 41s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/80507fb70db545feb8703e6303dff07e : SUCCESS in 32m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36ec4dfb04074289906cb7ac30719268 : SUCCESS in 1h 09m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7063291c7e0e427792773cad3772acdd : SUCCESS in 55m 58s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a25f9dd9e3364cd8a472d94cc2c7849c : SUCCESS in 34m 45s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/65d92778a39e4c63baa25044560d9c15 : SUCCESS in 59m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f529119fbb524b31a6f71c475422d496 : SUCCESS in 55m 04s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b4ee830ca895425687e162c814da1f5e : SUCCESS in 1h 02m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f097c7b183c74cd082da74522a957683 : SUCCESS in 2h 03m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/a8c3966f0bb542f58ce5a0b1ae4069e0 : SUCCESS in 1h 33m 26s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/94abc293839b4a7d8e3cd29cb196109a : SUCCESS in 48m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3ed3c483b1e34e31a0a70e80706f3bf0 : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/39e5f7c4269f4f97adf8ecd3a2f5ea68 : SUCCESS in 29m 31s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4161d8f35ed04f7ab61674397168cf4f : SUCCESS in 21m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/20f2148613dd4e0e919a585f714a8d27 : SUCCESS in 28m 14s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a53e4735065429f8481a9cfa33a4f6b : SUCCESS in 1h 27m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764b1bfa5d3f4ad1a0653d73898ae07d : SUCCESS in 58m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/6f3a53f4a4664bd68eb68e6821c9a19f : SUCCESS in 38m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/15ae6a8f56414cea851cede87d8517c0 : SUCCESS in 34m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7a2eddfa75c24b9087e8367a3f33e82e : SUCCESS in 15m 37s","accounts_in_message":[],"_revision_number":15},{"id":"85b2e22ac08539b3f6bc359f64a048d1e3ce21bb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-31 14:42:02.000000000","message":"Patch Set 15:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":15},{"id":"56c170f38046eaa69369032bcb96cc697351f9dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-03 06:36:18.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":16},{"id":"efa90344350a552d42c9d2b2f17671cfae8d335b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 07:16:52.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6f828a61ded403b922b349990d01bab\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/941197a2930d4cf6896f6457b77f31bb : SUCCESS in 32m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9176076c70de4d79965ba1778fc43369 : SUCCESS in 30m 26s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/05159dd05f1b473f9ada691f436a9ea3 : SUCCESS in 38m 20s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"df82d550fa6aa00f7696e6f1291c33ce4588712f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-03 07:44:36.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":16},{"id":"f94baad06cdd07849b24d8a74cc88117299cff4a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-03 08:09:56.000000000","message":"Patch Set 16:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-kt827/index.html\u003e","accounts_in_message":[],"_revision_number":16},{"id":"d5b9f8054243a0229096cfc48b351935a4f7c603","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-03 08:29:32.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466de51b181e40629eb9885f5f4c03d9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74154df09d764b3eaba755c1dc878496 : SUCCESS in 1h 04m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0ef96ab802b74291a0e616cd5fe8c3a3 : SUCCESS in 1h 46m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5bd4228faaf4b6d8d077fa322c4d4c8 : SUCCESS in 59m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e517c46ca1cb471682e1b03ef067a599 : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f08b55fd7cb446ba9411cd2c94e0bea : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b637b8014b734722a4d55aaf1bb70ed6 : SUCCESS in 14m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5f5ea6aaa8049e5802dd014fed99aba : SUCCESS in 13m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/95c55d21395449f68e2dbd83b88fb155 : SUCCESS in 18m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a485104038b4f88a94d0e6cde7957f2 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c388c3987f104ecdb5a1e338631813a2 : SUCCESS in 7m 59s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e2dbd06e13cd467f87a0f1b4e4d1abc6 : SUCCESS in 1h 02m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d69dbb4f840d44bfb240efc4d4e4d29d : SUCCESS in 1h 30m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/54eb6af369024b86822481a75869ead4 : SUCCESS in 49m 25s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9e86ab016941421a8ff61f77d630a28e : SUCCESS in 35m 57s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/86d5a2236e8c4fd1b7ff17837a37083c : SUCCESS in 59m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1bb1acbbd7c24e6fbb9b4ada1d0ee6ca : SUCCESS in 51m 10s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/84afaa34e9224a66aee4aa3de67f7ed4 : SUCCESS in 1h 07m 08s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/59bcc729a37c43e495b4bcab04c68759 : SUCCESS in 56m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/06fe4ac8932d4f59b51260f355ac94db : SUCCESS in 58m 30s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/145a3b168085412395d052caf1aa4c6f : SUCCESS in 49m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e0fbea7dea0843c4b3c1c7873f464335 : SUCCESS in 6m 28s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ade4aec3588f45d994a8d47457434d8e : SUCCESS in 28m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e607e0efc89649e0b2c52b9e8e7a3173 : SUCCESS in 24m 28s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/044722a85d8c42e9a7ac62b55fb37618 : SUCCESS in 31m 41s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f42edda13fd9424f9fb9bb6a1bcdf960 : SUCCESS in 1h 25m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3f4e82d577645cfad7df59022a8e48a : SUCCESS in 56m 18s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/4991fdbe91e14ef58f690f329a2fbef5 : SUCCESS in 50m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/440d0db79656498abe0c7db05df736ab : SUCCESS in 19m 25s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/01a06a409a6b4e6b88aa4622be833708 : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":16},{"id":"c66074e83eeff6feb1f848e8b7d0c3708ee851cb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-03 08:40:22.000000000","message":"Patch Set 16:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":16},{"id":"bffbcfd6429f7ed21669be72718f4f2a16c2e201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-04 07:48:25.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":17},{"id":"33e37d18ba23c5eaf998d7db4360da2e9c73da1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-04 08:07:26.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"f20a0ad3a0f3846e021c23e1858a3fdbcf5f12cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 08:29:40.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4fb911f81a7458787726019fae0b834\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fcac2ce68eab4a24873a95b59ba582e1 : SUCCESS in 14m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e76d23a991334dba8a388f097bd7926d : SUCCESS in 12m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/99c25037853944569645e6740e60ec9a : SUCCESS in 19m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0e1d39a771c164677eea72971bee8a22d3c663ca","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 08:48:24.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":17},{"id":"9b6ee8b6f10f4495f7b102733d9817af1a0ddb7e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 09:03:12.000000000","message":"Patch Set 18:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":18},{"id":"48007e73234463c596702fe845ea3d9e695cb6a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 10:11:24.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e3a2d89a2549cab14b2517b034a031\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9c69073a15743f5bc3dc67bf4aa9775 : SUCCESS in 1h 03m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b6bc78082bc94d68bb929c49deb8b821 : SUCCESS in 1h 34m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1113f325a5c469093f9f45b29535698 : SUCCESS in 58m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fa668da040c4ed0a2ed602bf54cf1db : SUCCESS in 58m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f17b6561d3940889fcb61a8e83c896b : SUCCESS in 8m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ecce1f2e58547b993647ca357eef785 : SUCCESS in 14m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a6e07752dec4d55b330997161164581 : SUCCESS in 11m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/298d61ab41a14bd4baf58e6cc3beee94 : SUCCESS in 18m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b95783eafa80456baa9b183edb0c2118 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a7bb06b61004c40bc3211c3b8c942cd : SUCCESS in 8m 31s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b80157a856b74c119905b33206b3781e : SUCCESS in 1h 01m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/87ec9a4c71e843238eb42c8f6a0d7b99 : SUCCESS in 1h 58m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1cc2b802b580405ebfb8bcec82e51a43 : SUCCESS in 55m 30s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1c2f35eaffd04df88ef5476ac5794c3d : SUCCESS in 32m 44s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0c0053079d14453bb213473af2c7e3e0 : SUCCESS in 58m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/cf6ac1024ef844ae8a0349ec2aa66cc6 : SUCCESS in 40m 55s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7a579b24b8c8412eb98e5b4001c8f762 : SUCCESS in 43m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f38893d1b971468a9e23d7664d5f6377 : SUCCESS in 49m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/6569adbc84d24af1a992705963146f69 : SUCCESS in 1h 56m 02s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8415c129e0b74d378e75f71b008aeb7a : POST_FAILURE in 34m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/af9540933ea849a3956967ead9eb0720 : SUCCESS in 5m 13s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cde2aa0b676343a783cfca4bbade4148 : SUCCESS in 24m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fba0e623b21b40198f316ab2d3757e73 : SUCCESS in 25m 17s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/06e59411de3d4b3b9486838c80c0a545 : SUCCESS in 26m 50s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5975a2770144cd5867e54fd75f4883f : SUCCESS in 1h 36m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1b3f845433b4d69b71549255854b949 : SUCCESS in 1h 02m 20s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/cb964e429221437285235d935b4a070e : SUCCESS in 51m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/366888ef460d4e108c28cf5673b97f33 : SUCCESS in 36m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6b01670c889940ba8faf601538463bd0 : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":18},{"id":"8ceae5622d1327f1dc10c2f248eed2a1c19d0420","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 11:07:43.000000000","message":"Patch Set 17:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-8d7ph/index.html\u003e","accounts_in_message":[],"_revision_number":17},{"id":"2f9d6c9051f09ec8b17c1429c795a7222ae4c156","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 12:01:59.000000000","message":"Patch Set 18:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-b595c/index.html\u003e","accounts_in_message":[],"_revision_number":18},{"id":"6e01edb085c375db1fab1c37fa832ca68b19dfcd","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-04 13:52:48.000000000","message":"Patch Set 18: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"f99ac83c99112fa852196a137e5a5466605ea7d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-04 16:10:08.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"e19bd25a5cecc4e18e59c373657c072baf1c45a3","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-04 16:13:14.000000000","message":"Patch Set 19:\n\n(6 comments)","accounts_in_message":[],"_revision_number":19},{"id":"6dbe03298108b34455f78e2f3a2723ec3e6f05a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 16:45:08.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8480b54cf6d1445f9229907eca669d8d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d5d36369453445cb6ec8c22dd3a1027 : SUCCESS in 21m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0638f4f8d092466c814c9089c2ccf3f0 : SUCCESS in 21m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ba1ff3bf3dd845b19f7a2bf1a84199ad : SUCCESS in 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"2f439dfa93e2a5fe3e6a7e518743980a6399f969","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-04 17:04:07.000000000","message":"Patch Set 19: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"06b397c194108ccd64d0e170b002e0e55afe6966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 17:59:23.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/174875cff44f465480b17d1afead555b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/017210aab8df4c33856dd762d9780fba : SUCCESS in 1h 02m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0c763c44b12147d1b67c90accb30cbf8 : SUCCESS in 1h 31m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc61b9b605414b50a5d68f6c34935a4b : SUCCESS in 1h 05m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78523fa828fa45f1a803eb3da9702216 : SUCCESS in 1h 04m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12c273ac2e6459483fcf241b8e8edc1 : SUCCESS in 7m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/820d862599044aedadc145c5b104148e : SUCCESS in 12m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c8c9e72fee1e42d18f91b30699ff9f0d : SUCCESS in 13m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ad6fea93d5d4422b88f7d39e4b3d6db4 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b7ec2168cb4be28801c0b4b833c6d0 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ccb5081465745e1b845d365dd709c5e : SUCCESS in 7m 58s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e6b4c203907b4c1d99e15fdf7c3b2f31 : SUCCESS in 51m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf0d84829e734062b825004bdce7e3d9 : SUCCESS in 1h 09m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1b22a028cf8f49fd8609a982fe80bdfd : SUCCESS in 52m 49s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f03eb8a8356941d292db359d3d43f5ad : SUCCESS in 32m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/cf8fca3782cf47b181e6f6ce71088935 : SUCCESS in 57m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dbcf02e537a84203a662fb32e4ce391f : SUCCESS in 31m 19s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/bac1f16a45aa4878bc29d3cb36562bd6 : SUCCESS in 57m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fc277f16f5eb4388a4ad993626a66f38 : SUCCESS in 1h 41m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/afb9291e48e243d48647518c191ba6d2 : SUCCESS in 1h 42m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8bc335c101e845d3b65e29707c9862b5 : POST_FAILURE in 22m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8b658d56026a4575a4a40b4b2f7e1d1b : SUCCESS in 3m 32s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28bee83cd83c45f3b8e31a135e6f24ad : SUCCESS in 25m 45s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ca6bdf4d062344c09f45a8223dafb64c : SUCCESS in 25m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f340172a2d254c6dafa30147f09cd5a3 : SUCCESS in 28m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a524654933164942b89301524223ae29 : SUCCESS in 50m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bf4724ab074467f848fe25c002290af : SUCCESS in 1h 03m 22s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0360d06c47094e408a3b31fcb74a2a8c : SUCCESS in 55m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/01b9ed292e794745a0da7a95cc5ff4b1 : SUCCESS in 16m 08s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3a01e782055b4fa68d7277b3292fd806 : SUCCESS in 13m 13s","accounts_in_message":[],"_revision_number":19},{"id":"54347db68ba287f448319e3c71999b8be3e04890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-08-04 18:44:12.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"3b335762fd6e0d1960fefbdc9cce071a7baa968c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 19:42:09.000000000","message":"Patch Set 20:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":20},{"id":"415945ec2e293511d4473a1387d6a7460d69d9bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 20:25:12.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a16ecc03ab0f4e95a81963cabe61447e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f965da4e5b2049fcacdac1b77cc96787 : SUCCESS in 28m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f01dad2bd3844a71ba610100c9986d9a : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/364a08cafb4847c7a4f47d0f5740a682 : SUCCESS in 49m 23s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"56812ff11e12d00513039fee31c436e182671da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 21:20:11.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31c257c0eef2472cb4733e608939f6f2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb82ebe83c3d40a58996de940befcf73 : SUCCESS in 37m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4ca65fb311994650b34fcf9e5f898a64 : SUCCESS in 58m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7682014019384af0a6974abf88304a09 : SUCCESS in 1h 03m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c01d56c8dad4cebbdb8fd4e32b16b88 : SUCCESS in 1h 07m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c306fac955f248a09da5162712703583 : SUCCESS in 7m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9ede4866656405080cb02275f6ce287 : SUCCESS in 14m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca7309856913422aa00a7ef931fb5a30 : SUCCESS in 12m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/34c47249cf0e443ca1647e70e5ba35f7 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba12060301434c569a3eb4a637b442ca : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c0508e336d4cf2a786c732a0935242 : SUCCESS in 9m 21s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/764a61f767494c9bb4c38e13808a228e : SUCCESS in 1h 04m 46s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c20395c1e9354752b5b1db246c595e01 : SUCCESS in 1h 29m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e6ab0cc929c24986b023edcc0ca82237 : SUCCESS in 55m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/616580ed2883445283f6ee532c1e72e0 : SUCCESS in 41m 26s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e61db6efd88d4e72ace4f7538ba769ee : SUCCESS in 59m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/12674a1d3e0a4872837d87f76e6fc472 : SUCCESS in 46m 56s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/109c1dea423e4c179e7b8d865f1dd522 : SUCCESS in 1h 14m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/db0745863be84e33bd63ac22f5e57f21 : SUCCESS in 2h 08m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/1fa37dd865c14f9ea46bf3c303708b9a : SUCCESS in 2h 26m 41s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9430e841b198447aa95812cc916a11ba : SUCCESS in 54m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d620c6a0346b45f1ac37bfd3a645cc5c : SUCCESS in 5m 29s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f4c472b6f8df4d61946dff35bcc7365f : SUCCESS in 21m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0e5fef37f62d45feb6ac8470c28315ef : SUCCESS in 28m 17s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8fed953d6ad14d508609b19e9931f69f : SUCCESS in 33m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fad0d120e53e4310b2697d41f1b113e2 : SUCCESS in 53m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a75f32d96c1144d8813b7a12889442cb : SUCCESS in 33m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec20c6c740304849afddf36f8a9d7f53 : SUCCESS in 52m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c63b0926c5ff4e02ac539120ed55fa67 : SUCCESS in 16m 51s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b08c7e6b07704736b2d79504081b4f03 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":20},{"id":"fe5934a3e096fb0245ca32529284799d6d188499","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-04 21:30:47.000000000","message":"Patch Set 20:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":20},{"id":"8edcceb4c3b9ad750c6545d39bd763342ef26eaa","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 22:34:27.000000000","message":"Patch Set 19:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-zrpcr/index.html\u003e","accounts_in_message":[],"_revision_number":19},{"id":"c5e59fdb19223b72e96ef1c6b0e583fbd2d5cacc","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-05 01:55:37.000000000","message":"Patch Set 20:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-wgr8s/index.html\u003e","accounts_in_message":[],"_revision_number":20},{"id":"4853fd8dd0d18b484fc126f4a94eba479df514d4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-07 14:08:45.000000000","message":"Patch Set 20: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":20},{"id":"3dec0b79e03abbd6703764e731066cc6f63dc8da","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-07 16:32:38.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"660c07e7facbd26a25a7c41577fdbdfa668cc32e","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-10 09:34:46.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"3a53b5cf74e879af191fff96918f9ee57290a190","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-10 12:55:49.000000000","message":"Patch Set 20: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"b12c13a49d02d80e698f97f9412b4ec2994e6a90","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-08-17 11:56:25.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"d0cfbef7984e64cf641deae5364efe9fd95ff7d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"real_author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-08-17 11:56:33.000000000","message":"Patch Set 21: Patch Set 20 was rebased on behalf of \u003cGERRIT_ACCOUNT_34452\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"}],"_revision_number":21},{"id":"bfb599dcc40e91de55085a0bbba859c78e49e410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 12:36:55.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5244eec181194cc8be50e02f6aaaf09f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0903a35c2bd1427b97741dc56b14957e : SUCCESS in 25m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/27d6fcb430f04d3cb5467ef96dfe556d : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/08eb8c8d2c7f486aa59eded08e3ff457 : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"4ee512f050f0700ef8cee18206bdcafa21f2f9fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-17 12:55:17.000000000","message":"Patch Set 21:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":21},{"id":"9f1e44340295a5b860ad8a2bcf305d496d53699a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-17 13:12:36.000000000","message":"Patch Set 21:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-tstx9/index.html\u003e","accounts_in_message":[],"_revision_number":21},{"id":"fe4d57f51b69f26c380b2a53ee6ea821baef7fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-17 14:10:47.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613783100f2546308653f6e09bd250ea\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6647e08ce0214e5081887b3bd879e635 : SUCCESS in 1h 01m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/40c5441c14664d52a00d0f277b9b6706 : SUCCESS in 1h 57m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/677eef53f7514dccacd6bb6fc3ffab06 : SUCCESS in 1h 06m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec4bca883c5e4220af3cf8d6ae4d6abd : SUCCESS in 1h 25m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffca5ebdff974a749d8da768a846f617 : SUCCESS in 7m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f81d332753b4ee184a9a9686a87df08 : SUCCESS in 8m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1f5031f76a143f8a7dbb563b499e5fe : SUCCESS in 12m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10d8e830ae414729bd06ccd1eecf4a2e : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca6911c88094664b93256e95997f2e0 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ca3d23ca11a4d02a7fb90f7537e6f2a : SUCCESS in 9m 24s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7bbd89de8cc04f9088531248f66b4c96 : SUCCESS in 1h 08m 10s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64faf43bf51c4035a76b9b8da5fb8754 : SUCCESS in 1h 41m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1e38e2bc03434b80bd0d427a034bed6f : SUCCESS in 53m 21s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/edafe9ba2e154126bf0a655a9181b442 : SUCCESS in 36m 57s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2069e4d0bec74cdeb1747298f2b740b2 : SUCCESS in 1h 02m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/d604c98ee6754e3081d00e63343dbfcf : SUCCESS in 55m 48s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/187f9fa4697643628eef85b3a2cfe716 : SUCCESS in 1h 01m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/66484bf0a50540759f78374f1ee10cf0 : SUCCESS in 1h 47m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/abafdc5fc077445481a21e2b9e8cbec3 : SUCCESS in 1h 54m 51s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a36f382aa9ae4062b43b713b1085ceaa : SUCCESS in 48m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cfe9f683250049a389ffe456975351e0 : SUCCESS in 3m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68845cf330254d40bfe3ee8bba1b7f41 : SUCCESS in 27m 49s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/574f0a03e05a4e5e9516abcfd33b0e9b : SUCCESS in 27m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/671f1856080342a48fe0c84683f6586e : FAILURE in 42m 09s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb2676e6aab345c9bbdeee3daa99d4a1 : SUCCESS in 1h 08m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16a4ecb6850940d49447a34d1c0abf44 : SUCCESS in 27m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d45059eb54f145d7a2db3c96ba2793d4 : SUCCESS in 50m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8dce425991db41e9b7c393fb3b7fe7b9 : SUCCESS in 32m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f44cd7b5c76e43d0961578a5c7575717 : SUCCESS in 16m 56s","accounts_in_message":[],"_revision_number":21},{"id":"2c71b583c16f94300753da819a97be9a56fdc0fe","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-17 14:30:45.000000000","message":"Patch Set 21:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":21},{"id":"d03b9e3690f8561526fff6b12e05df07ea9f531a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-18 18:13:08.000000000","message":"Patch Set 21: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"bf162c6a6da5c09e5dd9c630eb7d9fcda34967fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-08-19 05:54:30.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"fd75b26a833b7a6dfb12bc404aaf086387ea6239","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-08-19 06:03:46.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"76ccef71291d51eafd093a19a2b133c304885fa2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 06:23:27.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e37e2dbc12604c71a6e5b2d201243200\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4831ad7629a54c39808b311b3021d50c : SUCCESS in 26m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94fb5d51858d4a5c9327da5da1730c16 : SUCCESS in 26m 11s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c68d68759d944f34aa45adddba098124 : NODE_FAILURE Node(set) request 6db73df2f91e43479c8809d956f93312 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"8e44c41dea0903d76feb97e9d81de97a2dac4934","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-19 06:29:04.000000000","message":"Patch Set 22:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-4bqqs/index.html\u003e","accounts_in_message":[],"_revision_number":22},{"id":"618e3fb39db36d1d06aceb504ac91c348812461e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-19 06:50:25.000000000","message":"Patch Set 22:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa .","accounts_in_message":[],"_revision_number":22},{"id":"c98650ec832e624a6dfead69396bcd2e5a3ac729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 08:03:57.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf516e07827c4f45a32b732aba2df0f1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/109cbe587ea44786b2c4962ca7e7c593 : SUCCESS in 52m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7d3d83bca9a34083acd4f6c006bd5b0d : SUCCESS in 1h 26m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b39dd96604d4558a41013b22135e31d : SUCCESS in 53m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd66e5f1df3d42f6be891af6f5002589 : FAILURE in 1h 26m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cc0b057a20d4d4ab9ce558b40d9c52e : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e10cd5e7d3b4fd38617823be663497c : SUCCESS in 13m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/58b5cf2bd2314b72b4df539267930bdd : SUCCESS in 12m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fb26251501fc4999b1443c1d05dd7503 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca2e1b45a2a4d0cac1d359e18d7d4fe : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca609f9612b14b08a50249843246c7b3 : SUCCESS in 9m 43s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/642ee6762bf448b49666c735eebac87b : SUCCESS in 1h 02m 55s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fbfdd096852d4f32a1eda94510cdabce : SUCCESS in 1h 36m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b2143f5a4a75480fbbcfc5aa2643e4d2 : SUCCESS in 51m 54s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e4d1b75a2e7f4deab948911b25b3d489 : SUCCESS in 34m 12s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/aee80c8eda29435aaffa1ad04ae06c31 : SUCCESS in 1h 00m 20s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2d051ce2a0184772974f1259bfe7b3c4 : SUCCESS in 34m 35s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/de939fe5e1e1456b8e61d49bf3fe2505 : SUCCESS in 1h 02m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8942589b5deb4f27a7e89176a74dce02 : SUCCESS in 1h 26m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/61799206d7074b19879365b52811fd75 : SUCCESS in 2h 03m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b4cd3fa4507d4d4a9f75e1e84479a15d : SUCCESS in 48m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/69e814f462e8486d92df2952e5b1062e : SUCCESS in 5m 33s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/abba106af78240429d96a2f45159e8d6 : SUCCESS in 26m 43s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/55361f088d254d9387ce6aee6327f76c : SUCCESS in 27m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c932412f6ca344fb81e6570a72c767ff : FAILURE in 27m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ddf25ee53144af8bc5dacb3d6685bde : SUCCESS in 1h 24m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/619603d78013454eb1500dc27fa7fca5 : SUCCESS in 45m 19s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/91d17a8caeaa40b0a1c5adb6a4cb2fb4 : SUCCESS in 24m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/980d193321b8416cbc27fb847dfd62cb : SUCCESS in 19m 04s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5f6d96d065974820a68bba90d651a074 : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":22},{"id":"ae2376d4dde334679e8ae96ef46c17a618a506fa","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2026-08-19 08:59:18.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"c65473d666f83526d6b78890839762bf6ac839c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 09:55:44.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aec08674789f412d90a450f80f854e66\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5d7f4f14f8a40459d3d32062eacd212 : SUCCESS in 23m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a04eadaa13da4fcbb70e9948f2b317b4 : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/98bd4bbfa95f439faa732705a102edc5 : SUCCESS in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"eba0e715507e07eafa83e9dc447d25f7ae602016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 11:01:19.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e526cd1cfe941718e6fd1930e59f15a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6a0ca48f0bd4571ad42622e53547856 : SUCCESS in 54m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5655e08900e14bf7b4625175c6848e0e : SUCCESS in 1h 45m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76a432aaf25c421fbaa0bf7200f679ca : SUCCESS in 48m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d7ae7dc921468d88fc2e26d194b556 : SUCCESS in 53m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f033c5e579ef408e91538431606e32d1 : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/459dbc1a1b2c4d34bb9062b61900a9f9 : SUCCESS in 14m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4f6bc2b64dd45b6bf51aae387d79c28 : SUCCESS in 13m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/118b3f385e3c42069e05e1ae2b6b1ff8 : SUCCESS in 10m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181f4670a24045d89272829161b06dd1 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b31adfed70f4c8bab6fed100140cd02 : SUCCESS in 8m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d773b4d30692480faee8685cc855265b : SUCCESS in 1h 05m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/22ffee7b8e784518ba29b675347f65e4 : SUCCESS in 1h 03m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/044a83febc1842059a990eefa49c11c9 : SUCCESS in 44m 58s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a5693099df8b4a16a4090c6bfdded8a0 : SUCCESS in 39m 11s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/907fcaf6cfa7496cba081cded0ffbe8d : SUCCESS in 53m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/54f9c8c909ad4a56bb2340db511f30da : SUCCESS in 47m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/958791e9ffe8424a8aea57e9500c7979 : SUCCESS in 38m 44s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1d187b55b0bf4da79ca2a6498b86e319 : SUCCESS in 56m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/3e2dacff9a72494e982667c490b88ab4 : SUCCESS in 1h 59m 37s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/07a0bf44582a42f9b052d71c2af002fb : SUCCESS in 35m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0d5aa886971f4bea93987474c68aa188 : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/20437272196e42338b8a4de038765f82 : SUCCESS in 27m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9a34d4e4715d46f2a4752a0c542a64f7 : SUCCESS in 33m 45s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/944dda97df114e59b64b3c9778cfe99e : SUCCESS in 33m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cd41f2245494ccfbc23c13947ea69f6 : SUCCESS in 1h 32m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa5ee2b845bd4597a846af0f8ba46572 : SUCCESS in 46m 11s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/122caf4b45a3420e86f009d48b9ff06a : SUCCESS in 52m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6dbf2c51d0a14377923d05e26db97b43 : SUCCESS in 15m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/2cc1b092a8674eadb6112728800307d6 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":22},{"id":"84c1ce1cb357cbcd8336d1bfaae2ed4828e05ff4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-19 11:18:59.000000000","message":"Patch Set 22:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":22},{"id":"72a5aa47fcbb8a84dc3dd724ebb96253a111e68f","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-19 11:34:38.000000000","message":"Patch Set 22:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-995840-lj5z6/index.html\u003e","accounts_in_message":[],"_revision_number":22},{"id":"67c13ec190b19a1fab296d5fe19d8bc9e4ffc9a9","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2026-08-20 18:53:46.000000000","message":"Patch Set 22: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"d61e190123df408fc371c024e79a4b172fddc132","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-21 18:52:45.000000000","message":"Patch Set 22: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"6dbd65de3874c89449af7ba32ae11696930b5469","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-24 14:21:18.000000000","message":"Patch Set 22: Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"9a359b2b30a9917abd862d84396b63beb22c7a8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 14:21:52.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"2e494b0ef2d199fd7cb8f6a5e8ceca34ccf551a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:55:06.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd9a9243a8849269555d4e30af0f910\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63520d05dc5d47acb74aae6db89789b3 : SUCCESS in 1h 02m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/578b092a242e4555aeee8b1613c7c83d : SUCCESS in 1h 34m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53a86f42684049f79aeb51e534070b63 : SUCCESS in 1h 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e64fbae1d734d6c9fb082fb0dc0d61a : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3188fec19b4f485cbf7bad51c16356e7 : SUCCESS in 15m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48dfde8d267e4e819e601828eb97d96c : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3aca3631a44549b8fd66cfb46969af : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a71e4b0bdc8f4f7cb9f0450cb3af2d65 : SUCCESS in 10m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6bab1e002efd4666be8d311971d5ec70 : SUCCESS in 35m 17s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/81c8124d5c2d46e9a985f26eea928ba1 : SUCCESS in 28m 53s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f24bb492564e45609073113ac47681b0 : SUCCESS in 28m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ceffb65253594e01b288ed546ea3b9ff : SUCCESS in 31m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e9e4f957ad3a4e9a8e955da2d98d2469 : SUCCESS in 1h 21m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/5d4923e279584633a7a29f2ef71bd5e9 : SUCCESS in 1h 52m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0ca9b081a7ea4274a4ff19d07d3d0fcf : SUCCESS in 5m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8cedd2b6681142e098010968fac13e7d : SUCCESS in 1h 43m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/16c8355e90c649ffb9667f6b827ff170 : SUCCESS in 1h 31m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3adcbb2b4e2f47078272e1e57f35a51d : SUCCESS in 57m 49s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b847b99bdb1e4c6d8eaca17b788ac942 : SUCCESS in 15m 19s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f70a79d28de349babd92a7ed0463d0aa : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":22},{"id":"d808edb2377f2a2e038cf12fde7befbc5a37d195","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:55:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"00cf507f3a6c5b23a57f0c5c8d4f306fb16609fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:59:09.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3592a36989b4dc58165a6ccfa864a12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0731aaad7e14509a3aa2014c794f5fd : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2701aad557eb426aaff0c049f4a3588d : SUCCESS in 47s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"323734dae0d9d4757b498ee66e5d7474ee128b5d","revisions":{"e41615a481d53b4193bdb9968a1b5c62e838ce2a":{"kind":"REWORK","_number":1,"created":"2026-07-02 15:35:25.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/1"}}},"commit":{"parents":[{"commit":"be7f7848bd8178be8269e1505ad9401f12c8de64","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/be7f7848bd8178be8269e1505ad9401f12c8de64"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. A _warn_if_nova_managed helper\ndetects when a GPU is configured in both Cyborg and Nova\u0027s\n[mdev_*] sections, logging a warning about potential resource\ncontention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e41615a481d53b4193bdb9968a1b5c62e838ce2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e41615a481d53b4193bdb9968a1b5c62e838ce2a"}]},"branch":"refs/heads/master"},"cdffba069c9397ff2e511f4a89d393988daacdb9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-07-02 15:59:07.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/2"}}},"commit":{"parents":[{"commit":"f3ad2b4b0c29c8d76c06c72eafa959683921f9d9","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3ad2b4b0c29c8d76c06c72eafa959683921f9d9"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:58:47.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. A _warn_if_nova_managed helper\ndetects when a GPU is configured in both Cyborg and Nova\u0027s\n[mdev_*] sections, logging a warning about potential resource\ncontention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdffba069c9397ff2e511f4a89d393988daacdb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdffba069c9397ff2e511f4a89d393988daacdb9"}]},"branch":"refs/heads/master"},"770fd4dacc62a943ceb0b6ab28ff868a9027a58a":{"kind":"REWORK","_number":3,"created":"2026-07-03 11:33:15.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/3"}}},"commit":{"parents":[{"commit":"09ef4d292b38cd458ca4948d466d6f95b65026a2","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09ef4d292b38cd458ca4948d466d6f95b65026a2"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-03 11:31:59.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/770fd4dacc62a943ceb0b6ab28ff868a9027a58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/770fd4dacc62a943ceb0b6ab28ff868a9027a58a"}]},"branch":"refs/heads/master"},"3457c0a687006fb15f0ab80c36d372f46c9b1a66":{"kind":"REWORK","_number":4,"created":"2026-07-07 09:54:52.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/4"}}},"commit":{"parents":[{"commit":"09ef4d292b38cd458ca4948d466d6f95b65026a2","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/09ef4d292b38cd458ca4948d466d6f95b65026a2"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-07 09:54:39.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3457c0a687006fb15f0ab80c36d372f46c9b1a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3457c0a687006fb15f0ab80c36d372f46c9b1a66"}]},"branch":"refs/heads/master"},"7d4fa62cb8276ab1b8bc298e759dead3c432fd13":{"kind":"REWORK","_number":5,"created":"2026-07-07 14:43:31.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/5"}}},"commit":{"parents":[{"commit":"59413991c47574910fcd3595f4f432d7510f81fc","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59413991c47574910fcd3595f4f432d7510f81fc"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-07 14:43:21.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d4fa62cb8276ab1b8bc298e759dead3c432fd13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7d4fa62cb8276ab1b8bc298e759dead3c432fd13"}]},"branch":"refs/heads/master"},"ff3648078fd991ed4d4ebe17befbce93d25d3e56":{"kind":"REWORK","_number":6,"created":"2026-07-13 13:12:49.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/6"}}},"commit":{"parents":[{"commit":"59413991c47574910fcd3595f4f432d7510f81fc","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59413991c47574910fcd3595f4f432d7510f81fc"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-13 10:37:25.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ff3648078fd991ed4d4ebe17befbce93d25d3e56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ff3648078fd991ed4d4ebe17befbce93d25d3e56"}]},"branch":"refs/heads/master"},"807874c206946bf6c067bb2719b9da46d23916e9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-07-13 15:35:35.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/7"}}},"commit":{"parents":[{"commit":"71f785765f0822c006f1a019bfe9b3741a00ae50","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/71f785765f0822c006f1a019bfe9b3741a00ae50"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-13 15:34:45.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/807874c206946bf6c067bb2719b9da46d23916e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/807874c206946bf6c067bb2719b9da46d23916e9"}]},"branch":"refs/heads/master"},"ae905a7790828a7a0fbfd3fa4bdef1430559427a":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-07-14 12:59:37.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/8"}}},"commit":{"parents":[{"commit":"8d872c1553be6b5726629bba017d98f339145c92","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d872c1553be6b5726629bba017d98f339145c92"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-14 12:59:22.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae905a7790828a7a0fbfd3fa4bdef1430559427a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ae905a7790828a7a0fbfd3fa4bdef1430559427a"}]},"branch":"refs/heads/master"},"a7b6e9b3ebea11fcdf79ed2c71321128993d45fd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-07-27 14:57:22.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/9"}}},"commit":{"parents":[{"commit":"2a7260c7a58b4eab18a490716f35b0631e1ef1fe","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a7260c7a58b4eab18a490716f35b0631e1ef1fe"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-27 14:57:11.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7b6e9b3ebea11fcdf79ed2c71321128993d45fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7b6e9b3ebea11fcdf79ed2c71321128993d45fd"}]},"branch":"refs/heads/master"},"abc66f948929620d9851bdf1afc792e3e64e47d2":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-07-29 09:48:26.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/10"}}},"commit":{"parents":[{"commit":"ab1737fd9e12abddc99a42814eb9915dda0a67a9","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab1737fd9e12abddc99a42814eb9915dda0a67a9"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-29 09:48:15.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abc66f948929620d9851bdf1afc792e3e64e47d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/abc66f948929620d9851bdf1afc792e3e64e47d2"}]},"branch":"refs/heads/master"},"2904eb2c5fbecb2cd3d2d363666b56ad52348d9c":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-07-29 13:50:57.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/11"}}},"commit":{"parents":[{"commit":"2882e61adbbe1757901b1561e744a3fa4b67228b","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2882e61adbbe1757901b1561e744a3fa4b67228b"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-29 13:50:44.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2904eb2c5fbecb2cd3d2d363666b56ad52348d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2904eb2c5fbecb2cd3d2d363666b56ad52348d9c"}]},"branch":"refs/heads/master"},"ea9f9e0f2c0d512a00c6fe14aef9fcf224df1e7f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-07-30 07:40:11.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/12"}}},"commit":{"parents":[{"commit":"a3f059802376689ffe5395c22e4f9b0879ca551f","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3f059802376689ffe5395c22e4f9b0879ca551f"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-29 16:26:34.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea9f9e0f2c0d512a00c6fe14aef9fcf224df1e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea9f9e0f2c0d512a00c6fe14aef9fcf224df1e7f"}]},"branch":"refs/heads/master"},"ad47b81155c8b7e5efc3a8270bf346450d6c3152":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-07-30 15:22:45.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/13"}}},"commit":{"parents":[{"commit":"b3d5a300bef636cbddcce5a1872fbf9c62990621","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b3d5a300bef636cbddcce5a1872fbf9c62990621"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-30 15:22:30.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad47b81155c8b7e5efc3a8270bf346450d6c3152"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ad47b81155c8b7e5efc3a8270bf346450d6c3152"}]},"branch":"refs/heads/master"},"c12b44011349d2e8656e630ea23be63f63e22ec3":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-07-30 16:44:06.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/14"}}},"commit":{"parents":[{"commit":"7dd6cde72f80555368bb31211ad08fe34c71c04d","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7dd6cde72f80555368bb31211ad08fe34c71c04d"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-30 16:43:17.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c12b44011349d2e8656e630ea23be63f63e22ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c12b44011349d2e8656e630ea23be63f63e22ec3"}]},"branch":"refs/heads/master"},"dec3aaebfb25157f3e6c951081916cf031965198":{"kind":"REWORK","_number":15,"created":"2026-07-31 09:30:51.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/15"}}},"commit":{"parents":[{"commit":"184e10fef0e0b99b9b0e483b03894bc3d3cb858c","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/184e10fef0e0b99b9b0e483b03894bc3d3cb858c"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-31 09:30:37.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dec3aaebfb25157f3e6c951081916cf031965198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dec3aaebfb25157f3e6c951081916cf031965198"}]},"branch":"refs/heads/master"},"83a54e3a335c1d8a6a2b7092d064ca28601e01b1":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-08-03 06:36:18.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/16"}}},"commit":{"parents":[{"commit":"c99e6e25d55f8c07402ab522845b1a7008a56054","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c99e6e25d55f8c07402ab522845b1a7008a56054"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-03 06:35:59.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83a54e3a335c1d8a6a2b7092d064ca28601e01b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/83a54e3a335c1d8a6a2b7092d064ca28601e01b1"}]},"branch":"refs/heads/master"},"1d75e1971efa5426feff595d3e0b17a1071138ef":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-08-04 07:48:25.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/17"}}},"commit":{"parents":[{"commit":"9b55679a92e2edf4ff59253ac1fe8da5c97ccf0c","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9b55679a92e2edf4ff59253ac1fe8da5c97ccf0c"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-04 07:48:16.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d75e1971efa5426feff595d3e0b17a1071138ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d75e1971efa5426feff595d3e0b17a1071138ef"}]},"branch":"refs/heads/master"},"bf82b5fb8b367d58f477bc4a2bb452aa3bb45934":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-08-04 08:07:26.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/18"}}},"commit":{"parents":[{"commit":"30ab45bef69943a45cb39c9cfec42b43d475302a","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30ab45bef69943a45cb39c9cfec42b43d475302a"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-04 08:07:09.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _guest_add_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf82b5fb8b367d58f477bc4a2bb452aa3bb45934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf82b5fb8b367d58f477bc4a2bb452aa3bb45934"}]},"branch":"refs/heads/master"},"51887fcd8e849fe628f9d0db8beffbb40f3fac5e":{"kind":"REWORK","_number":19,"created":"2026-08-04 16:10:08.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/19"}}},"commit":{"parents":[{"commit":"e9365e229eb1ae46faaa36f7801ac4b17f463c03","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e9365e229eb1ae46faaa36f7801ac4b17f463c03"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-04 16:06:40.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _allocate_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51887fcd8e849fe628f9d0db8beffbb40f3fac5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/51887fcd8e849fe628f9d0db8beffbb40f3fac5e"}]},"branch":"refs/heads/master"},"c4923e83079c99a32bf0e15a0171ee467d508ab6":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-08-04 18:44:12.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/40/995840/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/20"}}},"commit":{"parents":[{"commit":"8c3639cbf02184a1b9a93c0eec2ffa85f55a00e3","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c3639cbf02184a1b9a93c0eec2ffa85f55a00e3"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-08-04 18:18:48.000000000","tz":120},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _allocate_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4923e83079c99a32bf0e15a0171ee467d508ab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4923e83079c99a32bf0e15a0171ee467d508ab6"}]},"branch":"refs/heads/master"},"d0cd8bbb3ec1c6c1cbfaac75c12dbe6b1484a8a9":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-08-17 11:56:33.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"real_uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/40/995840/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/21"}}},"commit":{"parents":[{"commit":"dc5667116458804b34e46708c7fc1d711f50cfac","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc5667116458804b34e46708c7fc1d711f50cfac"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"Joan Gilabert","email":"jgilaber@redhat.com","date":"2026-08-17 11:56:33.000000000","tz":0},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _allocate_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0cd8bbb3ec1c6c1cbfaac75c12dbe6b1484a8a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0cd8bbb3ec1c6c1cbfaac75c12dbe6b1484a8a9"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"8c3639cbf02184a1b9a93c0eec2ffa85f55a00e3","ours":"c4923e83079c99a32bf0e15a0171ee467d508ab6","theirs":"dc5667116458804b34e46708c7fc1d711f50cfac","merge_strategy":"recursive","contains_conflicts":false}},"323734dae0d9d4757b498ee66e5d7474ee128b5d":{"kind":"REWORK","_number":22,"created":"2026-08-19 05:54:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/40/995840/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/995840/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/995840/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/995840/22"}}},"commit":{"parents":[{"commit":"dc5667116458804b34e46708c7fc1d711f50cfac","subject":"Report OWNER_NOVA trait on resource providers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc5667116458804b34e46708c7fc1d711f50cfac"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-07-02 15:34:41.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2026-08-19 05:53:24.000000000","tz":330},"subject":"Add libvirt driver MDEV ARQ support for Cyborg","message":"Add libvirt driver MDEV ARQ support for Cyborg\n\nModify the libvirt driver to handle MDEV-type accelerator\nrequests (ARQs) from Cyborg, enabling Cyborg-managed vGPU\npassthrough to instances.\n\nThe accel_info handling in _get_guest_config is changed from\nan all-or-nothing approach (only PCI or ignore everything) to\nper-ARQ type dispatching that classifies each ARQ individually\ninto PCI, MDEV, or unsupported categories. This allows mixed\nattach handle types in a single instance.\n\nA new _allocate_cyborg_mdevs method creates mdev devices from\nCyborg ARQs using the existing _create_mdev infrastructure and\nadds them to the guest XML. It checks if the mdev device exists\nin libvirt before creating it, to handle the case of a hard reboot.\n\nA _warn_if_nova_managed helper detects when a GPU is configured in\nboth Cyborg and Nova\u0027s [mdev_*] sections, logging a warning about\npotential resource contention.\n\nImplements: blueprint cyborg-vgpu-support\n\nGenerated-By: claude-code (claude-opus-4-6)\nChange-Id: Ibc60f3fb78ff7ae9294928a4ff55d0a8738200fa\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/323734dae0d9d4757b498ee66e5d7474ee128b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/323734dae0d9d4757b498ee66e5d7474ee128b5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
