)]}'
{"id":"openstack%2Fnova~997878","triplet_id":"openstack%2Fnova~master~Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-10-10 12:27:40.000000000","reason":"Change was submitted"}},"hashtags":["libvirt-cleanups"],"change_id":"Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb","subject":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES","status":"MERGED","created":"2026-07-19 15:53:08.000000000","updated":"2026-10-10 12:39:42.000000000","submitted":"2026-10-10 12:27:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"997878","meta_rev_id":"a13bbd905f75d50e7df6a4eac8e264be13d3daed","_number":997878,"virtual_id_number":997878,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-10-10 12:27:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-08-24 22:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-10-10 10:29:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-19 16:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-19 16:47:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-19 19:05:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-19 19:05:42.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-24 22:16:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"713a8aa1402e5d930cf6827bc16ca293acb6f805","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 15:53:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fc740c7c24c8eed4108728e88213e09a188951f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 15:53:22.000000000","message":"Hashtag added: libvirt-cleanups","accounts_in_message":[],"_revision_number":1},{"id":"d1584114e1700395e2d1c7fb626980623f923070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 16:08:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d48ff997e469a7457bb83a564144452fb97b6a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 16:33:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4e55140998b40e383a9f95487d4dc91\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b173eeb20e5f411ea5f94aa8c6cf2713 : SUCCESS in 22m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/28d58b29c2c84416b35d426f7b7e2060 : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/93ef72d2b2264b26b1babb12915e948d : SUCCESS in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b3df417049d8444689c22353f8198daab02bed03","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-19 16:47:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb .","accounts_in_message":[],"_revision_number":1},{"id":"1cce515840bf27ddfbdc20704f407c1bb1d15b68","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-19 17:04:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb .","accounts_in_message":[],"_revision_number":2},{"id":"e1ba65455ed34229d0d417b3969b5fd66ba16b8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-19 17:12:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1321b88f0a235215408ef2ebf5ed8e52c7c2dff5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 17:39:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415b0aedd985413fbb858dc8365e1fac\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3c80e2365fd4c08bea8103de122aaad : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dfe3e272c43a4de9b828987a81b241d7 : SUCCESS in 15m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/29678fdbf5994c5c9cf65d65ab3584a9 : SUCCESS in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1552e3dc5d5f2b88d44e5c3071220dea48d47d8c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-19 18:06:50.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb .","accounts_in_message":[],"_revision_number":3},{"id":"7c620f6d4a3ccb4bba408238e3eaad03a26d93f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-19 19:05:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0d6a080840472d9ee639891c804010\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f52ed6959434948828b291032f4c8da : SUCCESS in 1h 02m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cb07c75eea1e4a1aab417b5aebb9b176 : SUCCESS in 1h 33m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03fdb93d63b14d5fbf91ad6f26457646 : SUCCESS in 1h 04m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20473ab9536741268c366cd1d0cf754f : SUCCESS in 53m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb5aab7f85734ec69c271a64df6226a1 : SUCCESS in 7m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8942690dbebf4abdac5a9df5e3d392f4 : SUCCESS in 9m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d390b0eac5ab4e99959c614ae92cf50a : SUCCESS in 13m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f737b37444dc4e44b271ee87d81542c8 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360fe46f5915445bb5ec829c3287abf5 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abe2fb9d2c44470894d97ca4d6090ffa : SUCCESS in 9m 30s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/3ba5b8e273544308b701de1522b44c4f : SUCCESS in 55m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de91c9c34dbe416b867da96d280faace : SUCCESS in 1h 40m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc4398b069f14381b37020f0b676d47c : SUCCESS in 48m 26s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fcb60844f48849078aaf8cdc844a8205 : SUCCESS in 36m 55s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1ebee91da2d84d8ca7cec5724e314e17 : SUCCESS in 39m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/aee5a730abea4f07b10435efba3ebc2f : SUCCESS in 34m 43s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/41c9f9031ea3468891eb3cfd364788de : SUCCESS in 58m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0e3837cc8ebf4c56b28e64bddfba6b6c : SUCCESS in 43m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/a9f770e866444cef8139580caf112750 : SUCCESS in 52m 43s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/e23662ee40284505920efd14e3eee4a3 : SUCCESS in 29m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/aa702de98e82465e99c8475d42874cb9 : SUCCESS in 6m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fee6b82e78684afcbf6b5613be04db71 : SUCCESS in 26m 41s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d49971cb182d47ee986a740f4abdff0b : SUCCESS in 25m 21s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6bf30b91663d47c1810f3c094a787607 : SUCCESS in 24m 28s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f47a9fe773e54812bce2548b36805bb3 : SUCCESS in 53m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52f0b988028741608293ad8cb4278cde : SUCCESS in 51m 21s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e0d86a1110dc480bafee31c4418f69e8 : SUCCESS in 49m 06s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/180cf5fcc98640d88885658715faeb8f : SUCCESS in 35m 19s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/299862c7cb2d4e1fbf295f8cd23a502c : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":3},{"id":"a138f0e735878053270b3b7e261aeb5171d87b32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-19 19:15:21.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"5d9f62310f1234921083121c85aeb3733120fe20","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-24 22:16:11.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f30ea4e5f80221df6df17bfb4f37eb8e59e35483","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-09 21:52:15.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"46de724ac396a7f925cfb6b61e15bd826498f11e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-09 21:52:28.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a28181b5be78dd9eb47e870142db70defee3fd85","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-10 10:29:26.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b283239fac1d5fa089e4c3e0822f09a6bc095763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-10 10:29:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"36d82d67f1c05889808bb89b7cbe3a85cc2cbe9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-10 12:27:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339b96d0c72c4ab2a5eb3f81ecd8dd4e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27e47442d5f64aecba9874e65ea55cf0 : SUCCESS in 1h 02m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/84562c0b5eeb47209d9d5b4b9fdfd92c : SUCCESS in 1h 40m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/278a2b32838d4ce88a01ff0315c56cc3 : SUCCESS in 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15db0d8930694ca7ae16129b50740126 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b1f02fda2cb493ab9712ddb1be57b8d : SUCCESS in 14m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/65c52329553c45c9afb0c4a58774e4fd : SUCCESS in 18m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c8203c29df4ef8a17062fdb3474201 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93bae4a602be4900a4f7691410d03917 : SUCCESS in 5m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1d9a41d2acbc4d47917c7d83e2e15713 : SUCCESS in 54m 12s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/db984491c9a0407e8ad934f9a0f60736 : SUCCESS in 27m 58s\n- nova-tox-functional-py314 https://zuul.opendev.org/t/openstack/build/be7a5833af86445aa21ac61d3f85c702 : SUCCESS in 35m 52s\n- nova-tox-functional-py314-threading https://zuul.opendev.org/t/openstack/build/5db9b0c04f6c4bccb655f67d332a5d72 : SUCCESS in 34m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/42dcbe1f8fed4a0d8307e99539d1ac2f : SUCCESS in 49m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/dce381a259584c3e93795d3024043915 : SUCCESS in 1h 48m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6d833339956641708a085640a04237b8 : SUCCESS in 4m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e9c29ad3264b4a72a1b374b626823c06 : SUCCESS in 1h 22m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d2176e7edf2408c91c1248b30c4d3da : SUCCESS in 1h 07m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e15af6e304434092a765ce0fdfa9ae59 : SUCCESS in 46m 49s\n- nova-tox-py314-threading https://zuul.opendev.org/t/openstack/build/f87bff9ed9944d1ab61cf08822b2a5fd : SUCCESS in 19m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a272e482adec4656aa874318eb95ddf8 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a2324ec501a18a910dbc45e8837c98e3a327fdb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-10 12:27:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a13bbd905f75d50e7df6a4eac8e264be13d3daed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-10 12:39:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbd96225215f4aa8aae683c95ea81b79\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afc5b88cfb14d7d8ce9dcc9df592742 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2bc17c2ae05123a187eccc7d9b00a7951c9b3b2","revisions":{"fd41ac6f3991b9a91162385924cc2f4a6211907e":{"kind":"REWORK","_number":1,"created":"2026-07-19 15:53:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/997878/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/997878/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/997878/1"}}},"commit":{"parents":[{"commit":"9fbcf558d43b2a61ad31723aa8e6d06f186226ec","subject":"Merge \"Remove [libvirt] num_memory_encrypted_guests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9fbcf558d43b2a61ad31723aa8e6d06f186226ec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:52:50.000000000","tz":540},"subject":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES","message":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES\n\nThis flag has been available since libvirt 1.1.2 [1]\n\n[1] https://gitlab.com/libvirt/libvirt/-/commit/d4952d36d0799ce4ce85a6ccfe143a72ed778bab\n\nChange-Id: Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd41ac6f3991b9a91162385924cc2f4a6211907e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fd41ac6f3991b9a91162385924cc2f4a6211907e"}]},"branch":"refs/heads/master"},"01331a88dd8d6a611b3f22777ae9d1f9ca44595d":{"kind":"REWORK","_number":2,"created":"2026-07-19 16:08:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/997878/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/997878/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/997878/2"}}},"commit":{"parents":[{"commit":"9fbcf558d43b2a61ad31723aa8e6d06f186226ec","subject":"Merge \"Remove [libvirt] num_memory_encrypted_guests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9fbcf558d43b2a61ad31723aa8e6d06f186226ec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 16:08:28.000000000","tz":540},"subject":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES","message":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES\n\nThis flag has been available since libvirt 1.1.2 [1]\n\n[1] https://gitlab.com/libvirt/libvirt/-/commit/d4952d36d0799ce4ce85a6ccfe143a72ed778bab\n\nChange-Id: Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01331a88dd8d6a611b3f22777ae9d1f9ca44595d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01331a88dd8d6a611b3f22777ae9d1f9ca44595d"}]},"branch":"refs/heads/master"},"f2bc17c2ae05123a187eccc7d9b00a7951c9b3b2":{"kind":"REWORK","_number":3,"created":"2026-07-19 17:12:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/997878/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/78/997878/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/78/997878/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/78/997878/3"}}},"commit":{"parents":[{"commit":"9fbcf558d43b2a61ad31723aa8e6d06f186226ec","subject":"Merge \"Remove [libvirt] num_memory_encrypted_guests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9fbcf558d43b2a61ad31723aa8e6d06f186226ec"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 15:50:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-07-19 17:12:06.000000000","tz":540},"subject":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES","message":"Drop redundant check of VIR_CONNECT_BASELINE_CPU_EXPAND_FEATURES\n\nThis flag has been available since libvirt 1.1.2 [1]\n\n[1] https://gitlab.com/libvirt/libvirt/-/commit/d4952d36d0799ce4ce85a6ccfe143a72ed778bab\n\nChange-Id: Ia0be9d48d5a4464a47bb8d50b2e4036737e888fb\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2bc17c2ae05123a187eccc7d9b00a7951c9b3b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f2bc17c2ae05123a187eccc7d9b00a7951c9b3b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
