)]}'
{"id":"openstack%2Fnova~998466","triplet_id":"openstack%2Fnova~master~Id89462b3652aee0f695dea60c2fae1b9430f8b8d","project":"openstack/nova","branch":"master","topic":"bp/intel-tdx-libvirt-support","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-26 12:51:14.000000000","reason":"Change was submitted"},"38744":{"account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"last_update":"2026-08-26 12:51:14.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-26 12:51:14.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-22 09:20:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-21 19:27:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-08-26 12:51:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id89462b3652aee0f695dea60c2fae1b9430f8b8d","subject":"libvirt: Generalize memory encryption config","status":"MERGED","created":"2026-07-23 12:50:20.000000000","updated":"2026-08-28 13:04:55.000000000","submitted":"2026-08-26 12:51:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":67,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"998466-bp/intel-tdx-libvirt-support","meta_rev_id":"7077750faa964de5285774007eb7326c24509715","_number":998466,"virtual_id_number":998466,"owner":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 12:51:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"recommended":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-08-25 15:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-08-22 09:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-21 19:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-08-22 09:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-23 13:44:28.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-23 14:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-23 15:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-23 15:15:53.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-23 16:45:32.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"CC"},{"updated":"2026-07-23 16:45:36.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-07-23 16:45:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-07-27 12:15:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-07-28 05:46:58.000000000","updated_by":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"reviewer":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"state":"CC"},{"updated":"2026-07-30 11:32:33.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-10 12:55:50.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-08-18 20:44:09.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-08-21 02:02:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-08-21 12:36:05.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-25 15:33:00.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"57a6be66ff255247ab035493ba7643fc5ed0aac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-23 12:50:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3c34bc7193af8dc5a8787a274acf8231c996000","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-23 12:51:16.000000000","message":"Topic set to bp/intel-tdx-libvirt-support","accounts_in_message":[],"_revision_number":1},{"id":"be69fedf4d7a23ea6c023debc2b86715b62d9e5d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-23 13:44:28.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":1},{"id":"770df00b25f6690be8d0016c9e25970b424b6e8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 14:56:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b7d29b3d1114acaae6f98bfb23231c6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/660a8793c7fe4f0db0f4abc89ddc2fc9 : SUCCESS in 23m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8abaa27d37cd4026bf64f57c78454a68 : SUCCESS in 23m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/50c395074bbd44d1b835c964d6ae221d : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de8b4390b3583f8e2aaa008b5712720742e7e3ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 15:15:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190ecd6d306e4ef0934a4148351f40d6\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4732ced3a76d407889c6e9eb5846a11e : SUCCESS in 1h 08m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3d1e1e99c0c945f28025cf1747158cc7 : SUCCESS in 1h 26m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7bb5701563e4aad98b5d04056569565 : SUCCESS in 58m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b6039209f5e46e1930150be57a5037a : SUCCESS in 1h 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d9a64b2d9945deae6481f3d1f62e07 : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fd43a326852474f8fea2f8a94f8f147 : SUCCESS in 14m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/70bb67f58858475089ad37ed8aa0188f : SUCCESS in 16m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aaf9c02f1c8e453baa8323981bedff21 : SUCCESS in 26m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3314a337460642c3b6378acd9708bf25 : SUCCESS in 13m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77f8a243918a4ebd9e1528b80f9bd503 : SUCCESS in 8m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/89bb6dca6f834559aa9630d3df9fd620 : SUCCESS in 34m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/36fd33caa5e74e69bba0116062b94fc6 : SUCCESS in 1h 30m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/16b64c402eda480a9805570144630f68 : SUCCESS in 50m 16s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fc7a93b91b9f4c90b43dbf92d0eb6cc4 : SUCCESS in 39m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/0ecc7d3c8ead442e90be17b45867b10d : SUCCESS in 46m 16s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3421e77d24de466dacebab331301d1e3 : SUCCESS in 26m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7f03b922f2cb4d82a6f4b63575c64dd1 : SUCCESS in 2h 10m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/9b00891ceaae432088924f190f37eb36 : SUCCESS in 1h 55m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6b05f674e538460787502a2043ed3ed5 : SUCCESS in 29m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6c79c2aace8d4623bd0cac4c952751a4 : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c703d502dfba43f793c98a2d02d03cbc : SUCCESS in 31m 00s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a64463a2e61a4dd6995ac18a0312b467 : SUCCESS in 30m 15s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/7711f99dfb134450af983a598bb7fe58 : SUCCESS in 24m 59s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/807a239a3b6047b692431b9d0389d533 : SUCCESS in 1h 28m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cddd3d3a8744b398b8102be7cbcf84c : SUCCESS in 29m 42s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/58fe028cf1254dba8765b7f18258b2dd : SUCCESS in 44m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1ac37a269f6c45eba950352907d4cb17 : SUCCESS in 34m 34s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/54f63afe188d4a22aae40771aeabc39c : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"eef39b7a0f81b820b966996a04e2a889d3771e84","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-23 15:26:14.000000000","message":"Patch Set 1:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":1},{"id":"da668f1eb3acdd3a039a0b971dc89fe3f672886f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-23 16:45:32.000000000","message":"Patch Set 1:\n\n(9 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1d141881d1f0413923930c5a9bac321563ef13d3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-23 16:45:36.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"89d7377ccefaf51b56376b7df2e65f5b1bcc6f9d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-23 16:45:50.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1b241f9fc89b862efba1e902b292c4ac0d5f986d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-24 11:57:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a3826761c7d9faeb2d5fdd04ab0d5b45bb23fe63","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-24 12:00:09.000000000","message":"Patch Set 1:\n\n(12 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0b57006d6c7938b84479dbe1f5b874033edcc594","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 12:52:24.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":2},{"id":"b94f25c90df2732bed199df67b35c4e4e20b36ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 13:07:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b53566055c42c4818511545aefa73d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23950db972254b65bbe2d3affa3c652f : FAILURE in 20m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c755390e6a4d4e9a8d43691908b998d9 : FAILURE in 19m 16s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8f98a5cd8877417386d84ca71f21c5b2 : FAILURE in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b28fd0d0c104d8cb9dacdd4e2b52921e8225567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 14:29:33.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab3f497a74ed4bc89f471ef87d8c6aa0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f73732f018cb4dd389422107b427a12e : SUCCESS in 1h 01m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9df0ac30dcd340cebfe3d4586c7c3272 : SUCCESS in 57m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afc6f84489954c3bb6075f4b6cc88747 : SUCCESS in 51m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf365a9ebbc041dd8889858e6814d76c : FAILURE in 35m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36386ee55feb45cf90b73d2dfd196138 : FAILURE in 8m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e7b54c196f049c3968b657353e24d0a : FAILURE in 17m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdbbb953a3634cbfaf8da9c5b18b9274 : FAILURE in 10m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ab48e05adac6409388c6fdf378e2e6d8 : FAILURE in 15m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8972b215384d4866b8fb028f9ab21f21 : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/222c9bbc040340b2a807fa70ceb16a10 : SUCCESS in 11m 26s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d3fd62324f4d475187aeb904f9005898 : SUCCESS in 1h 07m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/723bd9c8124d40138e3816ce3fbae712 : SUCCESS in 1h 14m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/36a0e3a09edf423db17d1d36f424422f : SUCCESS in 56m 08s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/79eb281a750a43d0952bb4b207bc0a7b : SUCCESS in 39m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/471d8f80ae0e4b6098f6ed6276f7113e : SUCCESS in 50m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ac4894c6449045f5b7bea32317fcba33 : FAILURE in 54m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/55149ff829834054b46b1e2995c16700 : SUCCESS in 2h 14m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/9b85fd2036174894a333b69ee822983c : SUCCESS in 2h 24m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8ecb38e42f0e49579b8aa2cd4486b65e : SUCCESS in 46m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5de086f89b6e4b2281d493d9ffb6183c : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/58db976807b0410098a7c454adedc21a : FAILURE in 21m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a1dc690894d944f69ec1e5f03bbd9a24 : FAILURE in 30m 31s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/39756df0fb064b2690d45e6a8707760b : FAILURE in 28m 08s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e79c936acf184652a49029ab9e694a33 : SUCCESS in 1h 26m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c673874f2dda4cc08f22895ac3917cc6 : SUCCESS in 1h 04m 08s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2e86e7659d7243e6831c7737f7a66856 : SUCCESS in 57m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5f572b391800467f9982a13566b49d43 : SUCCESS in 43m 18s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9b539b0764dc41599051e9f79ce017fe : FAILURE in 16m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9a90c8ac4e45a18643b8de0174bc92cc4c44f5b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-24 15:14:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f465a211e4120151c3f477660985fd1ff91749a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 16:04:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fefa5dbdc1164d50a51897ec0ff3f6b9\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4de66d276f144f549b2e2d686a16694e : SUCCESS in 22m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d984e615be64d1eb84c226530e412b2 : SUCCESS in 36m 04s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1c623155935449ceac305409c086b4fd : SUCCESS in 48m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e141a7048d8476de597c5decf508d87329f60b32","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 16:11:24.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":3},{"id":"6476d355f7e5ceb4a31616480e57b93f801a2227","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-24 16:49:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"289faecd596e15c71d5e866d35d8d47e7f530d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 17:29:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac7db4ef6f744ee69fa6036fb1443093\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9d2d6149a5e49918b558ae092157c9b : SUCCESS in 1h 07m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d5da595d1b9f447fb5c0577edb2c578b : SUCCESS in 1h 47m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e11ed8fd4ad4a96bcfb8fde2ccdb9e7 : SUCCESS in 58m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf81e644a6dc4b6087ab4c3300ab3eb0 : SUCCESS in 1h 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07253957b0234fe5849830a003cb6f2b : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81390536c6d647d484f518819d836d25 : SUCCESS in 14m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf162cb066d349a2a5b69b57cd7a9bc6 : SUCCESS in 15m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/767c8e5804b34fb6bba55c7f25372109 : SUCCESS in 19m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d463c06174414fc29891a3e2cfdc9a2e : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbdae22daea741a5b2214eaba2611e44 : SUCCESS in 12m 11s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7c1b31b0c95c4d94859e0c898785bf12 : SUCCESS in 1h 02m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d76e51976d9249ff859978e6a5b43294 : SUCCESS in 1h 42m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aa38d08332d642bdbcfb28611af656d7 : SUCCESS in 54m 01s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6fffbc3f9dc7425e8d39350dcfc94d62 : SUCCESS in 53m 32s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/3cd4f6515c3249e2aee5cd01eb710fdc : SUCCESS in 57m 49s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/fffd0e8cd15344469c98f1318213bfe0 : SUCCESS in 50m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/22872cfb843c4bf4b8319dc9b5b742f1 : SUCCESS in 1h 43m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/f1546adad8fe4a7487043084bf4e4489 : SUCCESS in 2h 08m 34s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6470ecc2403e458b93a079fe38c77c29 : SUCCESS in 54m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/be942bc530f04d5fab9e3cbef202bf9c : SUCCESS in 5m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f7c18412c94740388108ce11947a86a8 : SUCCESS in 24m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/44fcf027c22546c49db6774301d39f34 : SUCCESS in 26m 54s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ea6caf40b13c49ea9faec59d25ba49b5 : FAILURE in 36m 19s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a30172aef6d54055bc7c28ce1342a983 : SUCCESS in 1h 36m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/336e86def69d425f9b97f82327b22b6d : SUCCESS in 56m 17s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8a0e42f9bc964252adcf5f70295daf2a : SUCCESS in 55m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e7b1839696ff4d8ab5593a2f5a694e5f : SUCCESS in 29m 32s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4019ca259a4e45af8482c289a0adad1b : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":3},{"id":"b3b456cf2455cc7610ebf09f120e4c6ed40e8721","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-24 17:40:10.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"e6f07ddc21055bcb0fda58888dba98d73aae723c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-27 12:15:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e596d645d007394ebd3f184bb5edc9957f887f41","author":{"_account_id":35307,"name":"Taketani Ryo","email":"taketani.ryo@fujitsu.com","username":"r-taketn0517"},"date":"2026-07-28 05:46:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cb4eaa39ba9832629a40a2e166d0237446a63ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-30 09:25:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"5cf2b2aca94a0f6db6ce5e035ed785033ae086bf","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-30 09:26:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"810ecfe066bd8dc36f536539592bfa052aab4b54","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 10:20:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":4},{"id":"ea25b2229060c9d7a802e6c19230ec35478eae6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 11:27:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdfa06f243f94373a5df3c3b129ae6ab\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11e0edbd92fc48a2b471d8f56aef04ce : SUCCESS in 1h 03m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8ec1ae9d779449db99b8f23c11b613e8 : SUCCESS in 1h 53m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0dde88677024ccda9652bc3e0ae1228 : SUCCESS in 49m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/414aa0a38f8b462db3e82f920e97a2d1 : SUCCESS in 56m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7c50fecf3e461d841f14f951186061 : SUCCESS in 7m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edfa4dd35df44d4eac15257388595ea5 : SUCCESS in 14m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4fcb2f5e66f34117b8ffaa19e2d6ed66 : SUCCESS in 13m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fd56e3d2cb5e4d27ae51ae42f518efcd : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b872ebe640eb4f9e88d2fa4316406d6f : SUCCESS in 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39c3f8a0db3f47ab9ad9a819b7961dbb : SUCCESS in 10m 01s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/e4d0d66fe2c24e47a7f4f7d0e7d1b60c : SUCCESS in 1h 00m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a735929120454c55920610170eeb5e8f : SUCCESS in 1h 16m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/71581a1b5cdd4d87a4479dec90b84a53 : SUCCESS in 51m 31s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/965b5794906c4d6e8aa6f4dd0095eb07 : SUCCESS in 37m 17s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9767f471ffbc43dda0d019782da77d15 : SUCCESS in 57m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f37ff2259c664cf39422bca6446c3fa3 : SUCCESS in 28m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f7cc9f9aeb4840f0bd1355595436d5a6 : SUCCESS in 49m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/1072370e48bb4f0694c4c50cd2ea9f96 : SUCCESS in 1h 52m 08s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/becacbc1f2ce425ab026d194e98f70b7 : SUCCESS in 45m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c3201178b84a4d619ecce473d533849c : SUCCESS in 5m 49s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5c05916c7d824291bbb6537402d24a3d : SUCCESS in 30m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cba7e07beb20456d93a5b240cb5c5c9d : SUCCESS in 25m 03s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ffcfc3f4b3364112a9ddfb644b1d31f0 : SUCCESS in 25m 09s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2dc14831d8fa4af382b82413aed4e654 : SUCCESS in 1h 27m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e136dceefce54c7ab774bb9ddea01ee1 : SUCCESS in 58m 41s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/439a18f33cfd4272b902c7658403aae5 : SUCCESS in 25m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/52de944e27f04f138f0d914d7d736546 : SUCCESS in 33m 29s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/0ddcc3dc8c7144dda55ccf471dfcb83a : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":4},{"id":"21ca9f472723f3dce3f535aeb8c740e53e921462","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-30 11:32:33.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-gbtrc/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"2872796958c508000ee7037979edf4db15051b58","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 11:38:09.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"39456e86fa1d7ef8ba29c542d3a1e4b09dcd102d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 13:30:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1a95ceb1b04204b251712484bba968\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/625cfe0abba2460681cb2095ba4a3ccd : SUCCESS in 40m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/598ca22dca8c442da9ad5c464e3ef4ac : SUCCESS in 42m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a621ebca3a82451f871741cc2329d6a7 : FAILURE in 58m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3d0f62b67d1363a16853e0d9485cfa353609532","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-03 16:19:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f591ec3be1efef0da91bf3f55068850f98a5b34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-03 16:45:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de32545e94540e782ef8416f59394de9cf04e98e","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-04 09:13:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"569500cc315cc2c9ede62317facbc61cec463f57","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-04 10:09:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eea4d60e85198c981f023510f553a7059005e49b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-07 10:51:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ddd30f58a87face2a3091dcee6beeffab40c307c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-07 11:25:42.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":5},{"id":"70e54e457a2a61c877188661c772f488aa7ac596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:52:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a0fe4c869e44b21a674bac3294b5b41\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/177081ee928b4024a4ce7d2d4a849439 : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/afc08d6ada1d4018a221fbcf22c9c453 : SUCCESS in 38m 15s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2d05fa55c35e452283234e486f14c172 : SUCCESS in 49m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"50aa9b40e3437239197ac40a0f9338bdbabd9370","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-07 12:28:09.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-vl8hw/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"c6b35fcb3f418380f5d10487dafc42e812576b63","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-07 12:50:36.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"625a992d6598080f1555279fffeb8a12672e89fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 13:11:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ae6fc2d5464644a88d1decf165343e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/21945acecf8e46b7931cbe4982be8243 : SUCCESS in 41m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6462d7aeb53f4c86ba87b07aa1f01c9f : SUCCESS in 1h 26m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eac6a805690149118e03e03d9c28743b : SUCCESS in 37m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5e9f762289d4519bdf4756f5d16fb6a : SUCCESS in 1h 02m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6f43c83a254015b99e7289631f81dd : SUCCESS in 7m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f0444ddf84964b8c9bcb0f6428d3fa91 : SUCCESS in 12m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/755dc3684bf249c6856c6b6a2890271b : SUCCESS in 14m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f13fa695a63049cbb55579b86008721d : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b58eead40c40cbb0d51a6fa66e1db1 : SUCCESS in 13m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/773e7b85f5a04737903d5cf45c629c95 : SUCCESS in 5m 38s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ed338f8cd1f549d4a8f2ab0006b145de : SUCCESS in 47m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65246113f04c4f18acf74611d3cc9e69 : SUCCESS in 59m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/77307592e49947b9b2244f6253ad883f : SUCCESS in 50m 39s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e315c6a4191a4ac794c6cf95b88f7307 : SUCCESS in 36m 08s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a24bcb1f273d4843bad8eea43170c3c7 : SUCCESS in 1h 04m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3ab07353cc6d4fb3a1de448f97c2704d : SUCCESS in 50m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cc17948c29354b58903287ebbe1e47cf : SUCCESS in 2h 12m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/3171ecc2f13844feb93f51a5892afdf2 : SUCCESS in 1h 53m 18s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/14c4d42282e24d60a43ba853f3ee7564 : SUCCESS in 41m 48s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3886d76db5434129a7948a7b09f32670 : SUCCESS in 7m 33s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/394edbdfce8a4b40b03d7d6a580a2f72 : SUCCESS in 25m 16s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3b1e8c8a2a42495fbee4a5cbb2bf9180 : SUCCESS in 25m 32s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e29cb37593da41f684e328847389190c : SUCCESS in 33m 24s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/697d792d6a684a81aafcf12ff6104242 : SUCCESS in 1h 20m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a9c9904a464401b8387450dc737126 : SUCCESS in 1h 00m 53s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b4d7bbfa35db4e9196d3470bbec9e56a : SUCCESS in 53m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/019c4812e9994fb9a82cd4aa6a5a4dc7 : SUCCESS in 39m 37s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e2a6472e617247f38f3f9f7d63aa8bca : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":5},{"id":"f97986bd33ba51b85a7c463b20c8d33e5fed7620","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-07 13:21:34.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"dd3133e1af1c7c4ecde833b7e5bc91d81a5f0485","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-10 12:55:50.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3000827daacf9d7b6eceec798806137f55c6b114","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-10 16:13:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9539bdb0a3a84592d3aab9aacdfc334caf7097d3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-10 16:14:16.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a5db0caafa51cbce223ee32ab344e4d8c72ef45","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-10 16:19:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e9ccc728fb5186fa773b1bb5f5e96320bf82c03","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-11 14:10:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fba137f5d77b7bcb6f87a62f2e885f06c5554139","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-11 18:49:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"da9f14684b69eb526447b59a932b2e1db3ca81c3","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-12 08:14:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c600ba4665e11bd1732bcde069de080e80631eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-13 07:25:43.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"93b01e734677b52a74485e2f48252494fce53b3e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 08:28:02.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":6},{"id":"839b55a9643355d6c16d88e943e86e3fc7b549b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 08:29:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7332fad744ce4e0cab6ed556859ed08c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ef0e0dcd3864ac7bac8a080d870e776 : SUCCESS in 33m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/069bcae054bd456b876c1d3918a6f3ea : SUCCESS in 32m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f309186ecfef4367add5cba8fcfb8260 : SUCCESS in 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d1e82ff5e6a223437229d2364a8a8eb45ba968ae","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-13 08:32:11.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-6nh74/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"8c2cf14bd15c9cdd0318a4bf50e3ade62979bb58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 09:23:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c919ee32c5f54fefadba4de97830751c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/978bf17a1e564ef4a64dbb8222689235 : SUCCESS in 58m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/58546b1113f442eab9cb433346ed477f : SUCCESS in 1h 28m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc2a7778105c42e99f021f4941a68613 : SUCCESS in 59m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afe5d2d8f6a64e63a1b309410acd937c : SUCCESS in 48m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afb66654e6b94fa6b93128d2567f14a0 : SUCCESS in 7m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b8a5c2c0ee64cc59bdab1fbd4b65f15 : SUCCESS in 13m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e75aaf9fdd9148319df7b6b6a9ef941b : SUCCESS in 12m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1ce1332aa66d4429ba5529da5a9284cd : SUCCESS in 18m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6181a44b274364bedc391102bbf293 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b8b50595363407797bca82e6e1b7d47 : SUCCESS in 8m 03s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/475740d37ae142f7be616e6e3545bb7b : SUCCESS in 45m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67851a0139874256a968349ba2f3eeea : SUCCESS in 1h 38m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9bbf4bcfd23244cea1922c7b26906159 : SUCCESS in 55m 59s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a0ac9b046672424bada269c5203dae22 : SUCCESS in 37m 41s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/911d0b99d11746e199db49bd66494997 : SUCCESS in 1h 01m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0624b2a818d74fbdbcd7fcadb8a8fd6b : SUCCESS in 52m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/815bdc5049a34d8ab818b305d61f45be : SUCCESS in 1h 46m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/d6baaf1e42a84eba8ca11675e75315af : SUCCESS in 1h 52m 15s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6df4fd6d28ed404ca7cf3bbbf3bbe5ee : SUCCESS in 45m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/356f7ee6f210432f9e9594a5cb39c113 : SUCCESS in 7m 06s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1831543946af40a2801a1d2d3461439e : SUCCESS in 28m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/cb7c6102c42d46eeac829f17164962d9 : SUCCESS in 28m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/317862f41013463f8a21cd203476dd04 : SUCCESS in 34m 05s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c4a331221e4455196bd54c1e6c60586 : SUCCESS in 1h 36m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40faac6189a547dab6f2f8de6b7b15bf : SUCCESS in 1h 18m 17s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c9e8bbfd9e084834a84f2227d23bddb9 : SUCCESS in 55m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/21bfda9de34a428eb5335ddd0521f2e1 : SUCCESS in 34m 19s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/81f0e674fa1c447aa30837f18cfe07ae : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":6},{"id":"a082afaab87650a3567dce58fcbdf19c995b5554","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-13 09:34:27.000000000","message":"Patch Set 6:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":6},{"id":"d7b088ab43f066e10d6ef9478ee1318e04c667a3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-13 13:04:11.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7ecc886340b01fe357f70f1cc22720dc35caabd9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-13 14:57:08.000000000","message":"Patch Set 6: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d92770ae6f42c77cef47447e65b201cf465dedfc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-13 14:57:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1f152748200f55aa49731cddaadbcc54bf9ede97","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 20:44:09.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"3e32e7e7fe9cfd572edd65c7b9682b77a9c59de1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 02:02:52.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e92e5c49c58ef7e7fcd360e533991b0462096ba8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-21 12:36:05.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"04bd4ee63d356646a04944c234a678d205028634","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 13:41:14.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"4f680fae23d600cf1320f8ba7674825c1ef80b78","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 14:14:18.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-4p6hd/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"e4daf4f1e1660ab04aed520a8381b79521418a2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 14:17:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60b20da892c34003acae76e568fba9e3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e16923db82b44a1187afe5d3997ce37d : FAILURE in 14m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5f2779058484d70b9951b6a552009d2 : FAILURE in 23m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bb62ee7b8d5b4a90bf53a02dcf6a547b : FAILURE in 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"bfac1dc2335bfe7d4275d366703c1fba81b565c7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 14:41:16.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":7},{"id":"665857293a8f88a3d39fbd96f26262cf744384a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 14:56:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"765eea59b2ab3344ae928e5e7e10d7ad9b7fdad1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:44:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67d2c6c0891474ca6629791f13dbbc8\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dc0796e083534a2583b6423b0bd58936 : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cd9ccc3ba8a2480a8dd02abc95f93f76 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/921e994aa6334412bf63c8817416700e : SUCCESS in 24m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2faf230685d108c5d4712576c0340dea12f1e623","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 15:56:33.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Id89462b3652aee0f695dea60c2fae1b9430f8b8d .","accounts_in_message":[],"_revision_number":8},{"id":"1fa858a72bb40b5187fdbc2fafdcce9ba0261461","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 17:35:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7ad15349e7a4ee188f0009c1fb703f8\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/940a468f9f4f4df1858bc51317b9e1af : SUCCESS in 56m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/682c31e824d64a0a897d4e8b47d9f687 : SUCCESS in 1h 38m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2678712d611a4b56be104362af7c0225 : SUCCESS in 52m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b87dd40f324943f493673e413127bfcc : SUCCESS in 54m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5b83a3b7a44f3bb8c68eaee3020f92 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cf798b7800644018d3aac85f6559e05 : SUCCESS in 14m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/05cebdd809274588847cf3dc0aa0c2ef : SUCCESS in 13m 38s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fbef4ebe23e1454bb04e98c865cae970 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6749f3aa17384f1d9ea66e20ea4c3348 : SUCCESS in 13m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5a3617cad654217b928e0b7b4c9568e : SUCCESS in 9m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/73ffb229b37c4470903d8f6cb2b182c8 : SUCCESS in 1h 14m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dbbf6a9308a24a2ea04cf43090275fa3 : SUCCESS in 1h 53m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b7e46458220e4425a6727b1d8d69f681 : SUCCESS in 44m 33s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/226b33165aed4892ba9af5472076f7ff : SUCCESS in 40m 31s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/420791ecfe9942b5859386a55aea7572 : SUCCESS in 57m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/65fde1fe6cf04b4a8af6584f8388b9e1 : SUCCESS in 55m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/07b824b843a64348ad2a56edbdadf9f7 : SUCCESS in 2h 29m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/9a41635285354f69a41c48220fa204bf : SUCCESS in 1h 55m 17s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b258cdbf2e9b413fb69254a71dfc21a1 : SUCCESS in 54m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/01868e851f7f49c59e68830f8453b134 : SUCCESS in 5m 31s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8abd6ec6fe0241c49387a5080d443c98 : SUCCESS in 28m 01s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b32bfaf2b91f4b868bfa98a226f67cb0 : SUCCESS in 27m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/048c11531e764dd29ab072fbfcde271a : SUCCESS in 30m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb46b82378a24c1e9def2d831af7904b : SUCCESS in 1h 33m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50c7ecfce8fd420fb4657581beb013bc : SUCCESS in 56m 38s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/42ebe4d8277a418b96a59823d783d904 : SUCCESS in 48m 33s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9e3a755661d44867afcfe00ae268980a : SUCCESS in 39m 06s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4c2e13b88dec40b8a95bfaac1c8d7578 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":8},{"id":"257e1b2c1a19e682ee02f372063dd2e275e71da8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-21 17:45:59.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"90b9860a67aa39e8a94385d7592e9e09f224a771","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 17:56:25.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-wsdvv/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"849d4f132f93c39377043c1e6cd592e5fc3455e4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 19:27:38.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e8cc6f6e9fd6efb1ca5023fdbfdab127b0038180","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-22 09:20:02.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"480148a5ef247d5b13049552c6f57128fa03f7b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 09:21:39.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"12d941bf17b7922e2f4ac5ffc8b2c28728d59655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 11:32:08.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/994421 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"dfbca29605598f9075808c4e5f5fdeac2c7acdc3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-22 18:04:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6ef79fda5fc1cb03630ea1cf22fc40271b19fc07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 18:43:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2e059b7b0314e9b8af7ccf4bd3f8800\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6a978ec2a9b4bc19eea17209660470b : SUCCESS in 25m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c15edc70dc46419094b463e7f5d8dd59 : SUCCESS in 23m 48s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f15792bb533649ceb9423d8b60f5d728 : SUCCESS in 35m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8407cff3a87d1d884f6fb2c5617db9a4436b7324","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-22 19:08:30.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-b98jb/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"1322feefcdd517240e5f51032ba895b19cffd354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 19:59:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f38e978f400642eba7de9a16b390246b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c69964bb6714a98a0cf973dfc5fb7a2 : SUCCESS in 1h 03m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/906b5d3a0fde41cebf895612563562a7 : SUCCESS in 1h 33m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/979b96d04ee648a39e620320b6e33b14 : SUCCESS in 56m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fbec9ff795a491888cb1497cd6df225 : SUCCESS in 53m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79c9aea81b04d919f5acd21e65da7ab : SUCCESS in 7m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98fab2971520429f8ceb93805311a092 : SUCCESS in 13m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eda0260c95c04c4fb04a0a411862d4fb : SUCCESS in 9m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b669d0b668aa4fce9bba717ab1c78ceb : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd773b267cd4397bef8da8b7f18e1ed : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c235d6a772d42fbbca26d6aa86bc63c : SUCCESS in 7m 36s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5169c1facc3c4414bcd885ce074fbebd : SUCCESS in 1h 01m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48562a596d7e45189413b0b47b628ef6 : SUCCESS in 1h 28m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/57764c3056124815b315e5ba4e820deb : SUCCESS in 49m 51s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/10c5a3a89c9f463e9c8d04dddbfb53ff : SUCCESS in 33m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/bc8d31b04a37489c9bcf89fe9a04a00e : SUCCESS in 58m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/de7fff55a00f48939af62445c874c55f : SUCCESS in 37m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5bff15abcfa2404b91920b4d8e721418 : SUCCESS in 1h 21m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/dcd8349960b0485e8a8ba3930746269c : SUCCESS in 1h 49m 59s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2282d3126c664ecb808f4e030173365e : SUCCESS in 26m 02s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f0a91f28701341baa96472bdac401771 : SUCCESS in 6m 26s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d2d8bfba17b94650be2458818a8d4547 : SUCCESS in 21m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/77766f5c2fb94ca2a632851ffeed6ed5 : SUCCESS in 25m 01s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/33b57e6615a8452ebf8991e3134891f5 : SUCCESS in 30m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/49017e6f1cef4aeaa45432d5051c9735 : SUCCESS in 1h 27m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3bbeeed061949c5b5fa438e82e037f8 : SUCCESS in 35m 53s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d0a3905825084682b8fb02a868c46859 : SUCCESS in 55m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7ca87ad43f3a4cf1b6c4955e52d8504b : SUCCESS in 17m 31s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/5088d49654804e9584e010e1382abdf1 : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":8},{"id":"8085c73864b4d54e6f735f792c6d117dbf6dac18","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-22 20:10:43.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"d2959b930658d567a207c57254da6b2bae2aaee2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 11:26:38.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0ad9df9989c7b3f2e134f8a461c5ea4b085ad4b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:57:18.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fc4ba31e07540e59623bd8dcf90c02c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/094081a7dbc24d4a8ab906820ab749da : SUCCESS in 1h 00m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cbf95321903f4afea784d7b5274208d5 : SUCCESS in 1h 56m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c60bb9dbd38e40afa6eb632267f8e8cb : SUCCESS in 1h 02m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c343da2f3d4b5d862ba8e8d27c0185 : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ca07aaa2155e4bcd86c0da8e3ba43b0f : SUCCESS in 13m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de98e8ff195a41faac10a44f1bf60124 : SUCCESS in 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ad93dc7a7c45dc82cb70aebfe60133 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc3707b998f84e979ef5fc45d46047e5 : SUCCESS in 8m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d8a0894f8a814960b7dcb25dd7dc5fd2 : SUCCESS in 53m 15s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/658f60270a2c400386773731d5aa7d10 : SUCCESS in 28m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/96804448653b46769dffcd3f895a5362 : SUCCESS in 21m 38s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/12b4fee3dda24c9b97339fb54065e943 : FAILURE in 28m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6a11e3042c0e4ba3b3c73b6948d03d5e : SUCCESS in 1h 24m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/24a4e83894a543b0be4bbb39a8e92249 : SUCCESS in 1h 50m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b80e07d8a9ed4e3e968ccb35a9f4d95d : SUCCESS in 6m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/100f30f9bac246c9a0214db70b4a13bc : SUCCESS in 1h 13m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8429b5888448455da789073bd65b73ff : SUCCESS in 1h 38m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd29669a2477490a8cf9e9e4c12508b6 : SUCCESS in 53m 03s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8b8493b8e45449bcbf2852657e02f9d6 : SUCCESS in 15m 39s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/92cb1e390eae4f8cbb835b38c049790c : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":8},{"id":"9de7dfaf395e216855597e327b9dcd02cebe3851","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-24 14:56:05.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0b93362e89d3e06be4798ffe55d432608e394004","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 16:16:46.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-2dn8f/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"743ef3c81f4a3b1d344f59fda5fb3baa2bc2575a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 16:58:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbb54213d91e44d483adb4d6f75dc5ba\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/06029135be2948b0b141f2cc9e30c3d8 : SUCCESS in 1h 01m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/60f11fb455eb497a934800575e0cb94d : SUCCESS in 1h 35m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0888e9a98734de0b9d081b6745bdcad : SUCCESS in 40m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f35bbdaa6084422384ee0f2216f1ee76 : SUCCESS in 58m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9fd324010948ba92ca83b771c5b68a : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec86cb16db63444cb5787ce5fe39bf1b : SUCCESS in 15m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16b22c81bad547a4bce7d938f1110873 : SUCCESS in 15m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14242515f0834c5da328b5bc0629fb22 : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc705db38b949259977dde2f3cebdae : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be76b2b4ba3f4065b5a425a933e32682 : SUCCESS in 8m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0cee4887cefe45009c848314349d086e : SUCCESS in 31m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b1c0b2b8a2b94ca9b6f6c73ccaac72fa : SUCCESS in 1h 45m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2e09cccc190c45ee85434537e8a461dc : SUCCESS in 29m 10s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/20aa6a4638c547a79783bd1d0eb28aa2 : SUCCESS in 36m 41s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a41bbf4313c44339b54a45cd07c1a4e0 : SUCCESS in 41m 29s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0764ca3a6f684145acac1f441776d17d : SUCCESS in 44m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/79cbed6702fd4b06bebb4649c766573c : SUCCESS in 1h 46m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/27bbe97b55704fb0838cf8b7e943a76e : SUCCESS in 1h 55m 53s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f71bc9e4558d400cb9a3f0738ff18d09 : SUCCESS in 49m 06s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ecabf267faf541149512e616a4ed1584 : SUCCESS in 5m 16s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f70c5811a4644cc28283512031e54eab : SUCCESS in 25m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ea235490582c4a2387ec2d63ed225422 : SUCCESS in 26m 08s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5c4ed850c2a543cabb5326dadbdc149e : FAILURE in 29m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/cf839d17c65e41d8a1863c4b2e84c729 : SUCCESS in 1h 38m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcbb968384914e7bb1f21158f175eb5c : SUCCESS in 1h 07m 13s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/720d12857ad84f1eb7ef797ade18711c : SUCCESS in 1h 02m 48s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/27822db1868643f3b66366f52ebd5761 : SUCCESS in 26m 41s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3d6f93e8d4f74ad9b2f7da001dfc2b91 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":8},{"id":"7a7509cdce36a053d28fb059594fd7d7b6258760","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:00:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423e1f4d00ac4586aa71550e17aec9d6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c4ddb0f7cbf46f5b8aa9205266f9476 : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25373a46b74940619afa478f5b10a67d : SUCCESS in 27m 00s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/672593d8a7334f66ba1b8d736655cdc5 : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9cb1d6f5fb4b66ac53683c5b6d94f071426502a3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-24 17:55:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a4e39dcd77861b725efbdedec30e389417e93a23","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-24 17:55:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"36d2a27409f867eea3d663afbd936ac206bead31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:08:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7a04df6bf0b4e9c8f15db9b3d8787b9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/065ba1ecda8647e4a6397be40339f975 : SUCCESS in 1h 03m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/02e569a869844f13b141577ef30661c8 : SUCCESS in 1h 45m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4efbdd0849c44704ac1bd3da51f01e9e : SUCCESS in 33m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da9f14e85e76417d9a711a201555a04d : SUCCESS in 58m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f40484eeadd47dd9dbe5d34061cb3e4 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45735b33d93a4af187e169c4cdb6c7fd : SUCCESS in 11m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95e508d5e761444e869e2d00839ac11e : SUCCESS in 13m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bd3a58b7ee044d258c5b7002e7ac4c9e : SUCCESS in 16m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286fc5d258e44b87b521cb1ebf50ac2a : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f32a6968eb7848e0b62b9100a641b140 : SUCCESS in 8m 36s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/806b57ffd51245dab1a432513cc03244 : SUCCESS in 31m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bdf913711bf843f4b0149092a7268ec7 : SUCCESS in 1h 30m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4ab0646086e444a69ca7d15941fa071c : SUCCESS in 27m 31s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/09bf9a41f5c84c10b1847293c31bbbcd : SUCCESS in 35m 26s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6fc957769a2747b9a2913730e490fe4f : SUCCESS in 50m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1e51a0b79d4a4e19844df5c37f9a9894 : SUCCESS in 50m 57s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a2d8d04d4fb948fab914f38a0b013a1c : SUCCESS in 2h 05m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/3174bb888b8a4e959aa77005d49986fa : SUCCESS in 1h 59m 46s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9e01d59d84e042b280c946d04b7b2fec : SUCCESS in 52m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/95af11dc287f4df9b6503dac9d39d978 : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ec5608edade2423a84d3feb4847cddf2 : SUCCESS in 29m 07s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/1a656e03fbac4d23b22a09c3d940f5db : SUCCESS in 26m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ffe95282d6ee4df4a4fb0dc44037b3d2 : SUCCESS in 38m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/60f8733723dd4140a2b23eb45d8c296a : SUCCESS in 55m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed66ddf562ca44f2afc19c315d59caaa : SUCCESS in 1h 00m 06s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/d577466a1f57445394f5f333d7d90ec4 : SUCCESS in 1h 02m 19s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a89b7f00accc4a99ace3d7636c8b9f68 : SUCCESS in 32m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6be924c3683e45f9ad40c7b90ef06154 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":8},{"id":"9f3fbbbe8d94103b69f4e3306b9ef6d0aed61c16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:08:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"0d1a59998e4c16f49f2cb7bd936ba13340e47844","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-24 20:34:20.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"7e0b4e1bf395f12c1c8b2c5a3005a941b24b3b94","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 20:38:26.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-hxzdc/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"7ab74ae608486cc43866298035ab7f9475d418e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:45:33.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd644c0017c24e2f837134e63ca4d0fe\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e98fab626ba42939c98a3c8f346d21b : SUCCESS in 1h 00m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/27020511d94c441e8fe67eeb0d95cc72 : SUCCESS in 1h 17m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e484b93ed80543e49baf6ccbc898f8e7 : SUCCESS in 1h 01m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7b0b17eb5643a4b0627d61467326d6 : SUCCESS in 7m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32be0bd9c48f4940b35fc47709af41c2 : SUCCESS in 11m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5fedc73ae5364a039217fe3d4a34f0e0 : SUCCESS in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36da14e6c34b4a73820926e7d7b16d52 : SUCCESS in 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b23f4a98ae164a4298d71e09a44cc219 : SUCCESS in 4m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e841b794769d4749bf4e2a6b77a43b49 : SUCCESS in 31m 45s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/df497ba2f3c044f3bd6a58a5388e6ba3 : SUCCESS in 24m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/75b6ffcc1b1d40b8a62d553d456c879e : SUCCESS in 26m 45s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1cbfd29a4c54499985475263599cb510 : FAILURE in 25m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/af6d91450dc34d4a95886e0f33d51363 : SUCCESS in 1h 00m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/1e31f54f694148f1a8827e02afdc7010 : SUCCESS in 1h 09m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/88b8a3d5747c4eb3a3199101e82fcc5b : SUCCESS in 5m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf5a82c07d2540119cd3acda6af8b43d : SUCCESS in 1h 24m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/231983d43b244ae98bb15e223c57152e : SUCCESS in 1h 31m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ba9b42f845d4193b907337fe8d18121 : SUCCESS in 1h 00m 13s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d3d2bbb4af344504b7bacc219c7d5ed0 : SUCCESS in 11m 31s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ec0e27c4413b4f6fb8debff129563130 : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":8},{"id":"0894455ece1fd7843929a1cd75bb082fec3585e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 23:12:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a8d9a85637044afb9ef479593237aed\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7372e867c07c47f29d3844645ed66548 : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1a0dbabf635c40329aff8d94098634b2 : SUCCESS in 27m 19s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4a3ce5c9a2314e6cb2605d92e95a30ac : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c55ff91fca0d8ffd60d0caefbc5956f91c0d55e3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-25 02:04:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f280eec9a274f504ac680b62f3e3957ee8225f6e","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-25 02:48:35.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-p6rhw/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"b80abc5ab15dde475e274f8216e6f9515fb3a6ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 04:48:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28e18717309b4d519029f9652b83deee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fcce6268a6a24399a793edfea7448940 : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3f792d4d451b4973be3ee11f946817a3 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5d9a414750bc403585acdbadefae1e63 : SUCCESS in 32m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8e1fe160ada672ad507767eba4aceb6a7c9034d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 05:02:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e34da6998dcb46e29e6477ab20bf8321\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/14a639e5e53e4ac58a4b3d150965243a : SUCCESS in 1h 01m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/58995a30925f4aa4bf663146fbf0ae2a : SUCCESS in 1h 34m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cf5e0a393c84a05b3edd6d4a388f4e2 : SUCCESS in 1h 04m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/618842badc804b28897d83d4fe343b38 : TIMED_OUT in 1h 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a611c160b04e5b9d63d793bbc16941 : SUCCESS in 8m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b21fd54e151b407ab4391d8ec3d2998e : SUCCESS in 12m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9d8f33842614d388f9f5def1b75c4a2 : SUCCESS in 11m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0baa47fc7fee4426896d51f2be2a9e72 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4e3bd3b98545de876e2dc58c84e420 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6a3a165c8bc4a56bb8ba7c3d2a35c99 : SUCCESS in 4m 33s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/192b8ac2e6fe426ea8b2659dd323bf86 : SUCCESS in 43m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dcea3f50ce594592852798e9596b1974 : SUCCESS in 1h 33m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/573212f16db946eea23ec44294e5595c : SUCCESS in 49m 41s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/c0615bcfa89141d7a1ba2ef4479637e3 : SUCCESS in 35m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9f35664a8aad4414af42398b7f1244b8 : SUCCESS in 58m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9544a92891cb40a5982a3a0dc84aecc1 : SUCCESS in 53m 17s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/747f47c50acd405cb74be5ec0f700b70 : SUCCESS in 1h 30m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/b38f906ea6bc4f5db9b5fba687624e4a : SUCCESS in 1h 58m 14s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/95832303c0fa4872800d0c90cb411d99 : SUCCESS in 45m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5b85bcb5b2c7470d9ffc02043d34e4a9 : SUCCESS in 6m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4e25be7720a94149a9c17c013906ce81 : SUCCESS in 28m 31s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/387410376cf04a1c8c8c0dcaf7e662f6 : SUCCESS in 32m 28s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b605dfed876945a6a286f66889f31a85 : SUCCESS in 32m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1db70929e5c24fb0849d6d21eaa1cb53 : SUCCESS in 1h 24m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52a0734d279e48b09fe18134bc21050e : SUCCESS in 38m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0f0c49d8ad7a4b8b9fd23702914da51f : SUCCESS in 52m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ce4d2adb4f2442358e939a72e86da389 : SUCCESS in 32m 23s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9e1180bd50e34f92b5e93778ffdd826f : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":8},{"id":"82d90d42498bee420c50ce6b0e2698d1243c957e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-25 08:14:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dce55651415822e8541aff194b6632c5bf87db78","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-25 08:51:12.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-mqflc/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"b18ebc4ab6abe80c42a1e0a28ab1725f5212d355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 10:08:25.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5318445f214b4e289d5d48163c747777\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb4a955b4bd644f6af689a2a650b42a4 : SUCCESS in 30m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1f4b91b5199d4076bf73650b5fc9fcdd : SUCCESS in 1h 41m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3207deab2f04370a6550b0fc7021ad7 : SUCCESS in 42m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13dfe62fc4b147c88e1d0e520c9ca3fe : SUCCESS in 1h 24m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3853f0861184db18b9e2ac0dda0537c : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/568c732ec0334a888a7986552a68ef41 : SUCCESS in 8m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc4a576fb9c546f092b1e631160613aa : SUCCESS in 10m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/03323a64956c483cbcb6eea3496fd256 : SUCCESS in 15m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac74d51723ce43d39af2e902420c5d88 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5944f82f50be4d8296e3f2f7f5c0d8df : SUCCESS in 8m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/29ce194136864a53bf090d5e71dabeaf : SUCCESS in 1h 08m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/860482ffd0c3438d8169cceb5b9d8a2f : SUCCESS in 1h 10m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/441e945d95ae4a59a2c07741a62ba574 : SUCCESS in 35m 49s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/15d9b460c1354a1ab1e42a2b2d29200f : SUCCESS in 36m 32s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a84e1da6ab8a4955a6abc3189ceb71e9 : SUCCESS in 58m 45s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/764913f8e069466cb58299936e5d4914 : SUCCESS in 36m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b986341c07c54c63863acda52d87811f : SUCCESS in 54m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/bcf2e0a2a87d4e81a83815cac5bb8864 : SUCCESS in 1h 48m 42s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/0efbfdfa882f46c784eb91d010710af7 : SUCCESS in 50m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2fe6ed76981c469da60b821e4c08b633 : SUCCESS in 5m 52s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e845e3a29afe433a929fbc9550f1dabe : SUCCESS in 26m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/12d668b72b794564a908ee7e38375c8a : FAILURE in 25m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/5c872d36138049bbb6de8e6a12c62bb0 : SUCCESS in 28m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ead7fd29c025452d900d46cee870548c : SUCCESS in 1h 36m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7db3471873c43d39f57b61a63663e77 : SUCCESS in 1h 02m 21s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/002d48f8d5a347e18a96b3e6495d54e2 : SUCCESS in 54m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2a31985b568c42b18718f5b8d425840f : SUCCESS in 33m 01s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/3fa1cfd7fd234bfc83e126d842dcdfd7 : SUCCESS in 14m 37s","accounts_in_message":[],"_revision_number":8},{"id":"fe07da3db171e3402b1f1fe390b5112d0f570ccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 11:43:44.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab758b94e8ff4992b010698cdb7b7636\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e915c5901e46402bb8d9d9bb21470287 : SUCCESS in 13m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/05b55cc316884c37a9ae92203e0d25c4 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/71dc4ed72fe44991850d86825fb45222 : SUCCESS in 36m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"eb02542666054e41ba6e57ac7630fbd1c8eab1e2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-25 12:36:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a5006ab7a00cfeaaca1e31ac5d79b0ab63b7b5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 14:33:28.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f428015c78304c64b32c21c5d51beeb9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f1097a52e74d4667b104a85705b09780 : SUCCESS in 47m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/43f2ac7f369b4d6e94d6ad3b7b3fd91f : SUCCESS in 1h 07m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/719e7b8d42e849a79d806f2c396523c0 : SUCCESS in 42m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/775fc4650c334c048ef5bf7e653bc295 : SUCCESS in 53m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4b482e67ef2450885cf16b2e2c04611 : SUCCESS in 6m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5dd1a997fed946eda2bad48aa11f2400 : SUCCESS in 11m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1dc42840d5a94219a1ae8e9ecbe71cff : SUCCESS in 14m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2f89877480ab43988a2a09d9fa3e8f96 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0925268c6ad4c36a0db8ff78b9bd077 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e3f05c1e4714661a05be541c57416f3 : SUCCESS in 6m 50s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/1215fb8ffb93461bba2d62b4c72078b8 : SUCCESS in 1h 03m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ebd42a828264f19b6f1a036adfc8a29 : SUCCESS in 1h 29m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5fe08588bc4b40c1bec42bf6acebebe2 : SUCCESS in 57m 28s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/cddf6fbf40b14326aac48e531f7c6e65 : SUCCESS in 36m 27s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c32da66b55cd43ca8a50a8fa14e6136f : SUCCESS in 1h 01m 15s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/46d40da0af8d4e289a5aeaeb3dc6b635 : SUCCESS in 29m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e25479097177468aad0880db38ce633d : SUCCESS in 1h 51m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/08c56736bdaf4e6482914fd8e02afc79 : SUCCESS in 1h 52m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/4456ab7c6fe246c49bc03ee046d2eeba : SUCCESS in 49m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/09542d47c22243aa838719e416c397c6 : SUCCESS in 3m 14s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/99528506b9e54c51b09fdedcca884391 : SUCCESS in 29m 10s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/0cbb9bb0c5d44c43b0eda088d144fcdc : SUCCESS in 28m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/c6be6d054c0d41ac913d1957b5c2c31a : FAILURE in 40m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/786af13f4eb54b4788cc3936e2f49806 : SUCCESS in 1h 32m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ac9295683dc47ceb736baddd2598094 : SUCCESS in 31m 49s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/f4ae64bbd120434aa276740191cda576 : SUCCESS in 54m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/342f70651a8949fc857fb076853554c0 : SUCCESS in 19m 53s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/e51edcf01684404ca5549eccab7ed154 : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":8},{"id":"693a6e6f067822e75e1ebd4d39a1322212c55a6c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-25 15:22:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4cfceec44ddaa1313ce770d3bc38b9925d928479","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 15:33:00.000000000","message":"Patch Set 8: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/7efa104a95984044adffc36a6ec22a8a\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/b6a83cb1845c43d4bd9428dc86d103ca : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d08599afc8829badce59b7c4a2aefecd6cc8781a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 17:40:29.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0813835395406ab59e8557bc72915e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2e05a2059c6e4691a397f77bd63caa9f : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/821b9fc9a50d48308f837ef1e5a398c3 : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fe401babcc204f0b8df600f1442aee09 : SUCCESS in 42m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fc598e66b4a3d612fc74d78ad31801588e9a7e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:20:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21a6ff19d598488f9ae11ba82b3e80f1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b77bc66149174f6793323fd30fbd69f8 : SUCCESS in 1h 07m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/faccd6653bed4a56acddf374ed959f3f : SUCCESS in 1h 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b68cab5b6a8f463ead4194916fa96c48 : SUCCESS in 1h 04m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0772a18e88f24507b49bda2de64c5f45 : SUCCESS in 1h 25m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b490a9da98742d2a28a65ed4c7a3e64 : SUCCESS in 8m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20c64c1d5d6c45aeb5be9737b0854edd : SUCCESS in 13m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c2d5310a96884fcb923dc770785ebd27 : SUCCESS in 15m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/346ee10883f24e9f8a5cb8a05c549dcd : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f13d3ad2e740c4a37f346c0e52bd27 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1444059c0a44b2ea8e8659e6ffc396c : SUCCESS in 9m 06s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bba8f319bd8e480496a2f55ab4dfe1f8 : SUCCESS in 32m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1eeb1b4a778f46d1898b9a342884c550 : SUCCESS in 1h 45m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cc2a56b59de443b6b5a3e9bbc6234985 : SUCCESS in 43m 34s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/83ce5dd7c7c34981aaa453473e00ad69 : SUCCESS in 38m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/249650a951d244788f7b9deda6698c64 : SUCCESS in 41m 16s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/97b7a2038dac45be9588b7d493c9492f : SUCCESS in 39m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/eadff01b7c4d42a5a490e55df11f42ce : SUCCESS in 2h 12m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/3362d799d13d4340a283abe8410d028c : SUCCESS in 1h 37m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/af6678f6f93c404fbda35b231a87bcd5 : SUCCESS in 53m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/16acd9c1751742188021bb74b402a58b : SUCCESS in 5m 46s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9232713d99614e25aec64a5f40a14b6e : SUCCESS in 29m 55s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc136daf94fd4e9c8b659cd1229bdc33 : SUCCESS in 27m 18s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3c5ce08f52ff45da9b3d82494219240d : SUCCESS in 37m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9795cf09c68d4ff4b66af7d8013e67a8 : SUCCESS in 1h 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d2eeea6395540f085b6ad48c2d26c08 : SUCCESS in 53m 04s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ff16d71fe6c04070bd1521682f7f8621 : SUCCESS in 44m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3e36ab0a35114b1aa019c8ce74ac3cd6 : SUCCESS in 37m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/56b7b31cbfbb4c008efd254d89db3a7d : SUCCESS in 14m 01s","accounts_in_message":[],"_revision_number":8},{"id":"7cda79d817c61d60d86d3ceca6ee606fe87d0489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:20:45.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d02d471f247566865332c925be1b99391a2faf35","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-25 18:30:25.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"4347a4610118b1bf22ee95e1e689685849896d14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 20:33:34.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d21a2e623c8f4ef5b6c022f04ed88bee\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/581824bdae3c462e8aef8b2a6ddd4287 : SUCCESS in 46m 00s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/57265ee6d861410283be843909bf50d5 : SUCCESS in 1h 57m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e77ec6dac7184d0089c7502f8d56703b : SUCCESS in 1h 06m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2da2d49a61e47e894b6dc35e7ff6f23 : SUCCESS in 7m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b685be2b572242dfb88be824e66e1a74 : SUCCESS in 14m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7dc11f283f22438b8d48cf4bf5269b4e : SUCCESS in 16m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6e0160acb184cd1952c7cf51ecb65ea : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89c72ec18ad0432488593fea4d9e530a : SUCCESS in 5m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/66ac59a676734e26b220551d4966ee68 : SUCCESS in 54m 04s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8ef791e0675148eead7eb262db8cc203 : SUCCESS in 27m 44s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/be28412cad8d423c9bde9b6f45913536 : SUCCESS in 25m 53s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ae835b642cab4b269d1576f384707353 : FAILURE in 28m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/616d87d442bf4f05b9e6c5498debd046 : SUCCESS in 1h 29m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/393bfac928a840b4a832dd43679de8d8 : SUCCESS in 2h 06m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/969de273c9654162a0ff5d3a8aaf7d0f : SUCCESS in 5m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4593dbbe2e0f41d0a5367f5435c93b56 : SUCCESS in 1h 24m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8de7c75c0d3447dea87f6ed56848fdff : SUCCESS in 1h 30m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fed2576aef0547f4b41a89c0b4222b69 : SUCCESS in 40m 42s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/21d97cdd65cc4412b6a616ff8ee9b415 : SUCCESS in 14m 08s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bb91ca1455574842a65f2ee22ed1878b : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":8},{"id":"82ebc470b3947ce49f834d1d41675eb29809fbe4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-25 20:47:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d095dde6017067fa400a7464fe771019eef84d4c","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-25 20:54:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/8e7f0ddab71140bc93ab82811df83fe6\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/193e3d04c1fe41d5a62ed2bcd96957db : SUCCESS in 6m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"dc93030d6425c21d84b50c96f2cf86eb8fb3b9dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 21:51:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85295fe23f344d348d602031a090aa34\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8ec10c447604a2fb435fba05f2ae683 : SUCCESS in 18m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/46afa4297cc3493fa089c5817cac4b0a : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/1b1c2ff7a422477fbc194e093288084d : SUCCESS in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2b28976af581fe055bdaf87822f91528ff5805cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 22:57:35.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da86f7268bb547e8b99d09bfe237da3b\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4dc5ff6447934d0c8b8dc86165f60f8a : SUCCESS in 42m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a89fc8d31b3642c98b00041d71010806 : SUCCESS in 1h 42m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/137755a8dc1f4c069d3439fcc42f4afa : SUCCESS in 1h 05m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3241556060654ce18e8c6c3b13e735df : FAILURE in 55m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1e4f2a096e4e85846c870d91f024f5 : SUCCESS in 6m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23bd7889c6ff4d8193dacbf015a62d26 : SUCCESS in 13m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60c3599abdd94a5c83f9f3fd34327dbe : SUCCESS in 14m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fa93167a3be34070a9089fff702e0ed2 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80c981f87ae64bddbc91377e6d77f4fc : SUCCESS in 12m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4c2a2fde72a4672872ff2668b43b7cd : SUCCESS in 7m 28s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/88ec3644ebcf4dff8a82c9add121aa91 : SUCCESS in 1h 07m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/33e358ebe0d842958623b06dde4983c6 : SUCCESS in 1h 28m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/33feb4f7473a4513968dc75ce64430b5 : SUCCESS in 53m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/213aa59dfe0240e89176dc310976dfa1 : SUCCESS in 42m 59s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5553c64cb5484e91b97f7c65dbdded75 : SUCCESS in 44m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/325ded304f354105b0ac011bbde3b985 : SUCCESS in 51m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5e98a309aa5645a79241bb2d36566a64 : SUCCESS in 2h 04m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/e2a510d3cc38494890acdbffb5898cda : SUCCESS in 1h 29m 34s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/fd71805d2ebc45f78a77fd099fab15c2 : SUCCESS in 47m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/04f90c33da7c45e8aeb901a14c8a97d6 : SUCCESS in 4m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/022462c805bf4819855fbc350044cbbc : SUCCESS in 36m 02s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8e1790e810dc45638f043468f0c34f51 : SUCCESS in 28m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d3607b3df4904dec9e0ee178ee9d590f : SUCCESS in 29m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/82488908e1464143be9402f6f9d1af9e : SUCCESS in 1h 12m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59bb52b94f28439d8936d7b4963454c0 : SUCCESS in 37m 14s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/001d84bf1a4141b5aab5147f2fcd804e : SUCCESS in 43m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0210d625364341d99997198d562da577 : SUCCESS in 35m 06s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/73d479adaf7d4f9290930f5891ded182 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":8},{"id":"aa8c0a2dc58f949ad15e20655567accda857250b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-26 00:22:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dc470c61290e021737cee4668f8d8b08f1157bfe","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 00:27:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1e2ed5439d494f36a3087b4439c3f8ac\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/bc07b8ecd06948e1ac33733087e449ac : FAILURE in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b05e7aeecdb544325000393b4c5ca495cdba41ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:03:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55838662722047b892e40839af69d6de\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee2ba8042d6f4209a335ae8486c03663 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/67d8ff3e2a974fa69fa2259af29a6a9f : NODE_FAILURE Node(set) request f411c01329e24c5facff9bad6b52be68 failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6107f7e3d472446db48c7fd3d1b6415c : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"11a3519aed090f41a8d163f1b444b1aa92717861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 02:37:32.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55523a59f75342679231d88c9bd3529f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ccd8387e7954e8caf849bf27f7363e0 : SUCCESS in 1h 00m 47s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b45b56b1927b4eaf9bfba227e605f5e2 : SUCCESS in 1h 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc815cfd97184510820fb2e9a2206578 : SUCCESS in 1h 04m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe8d3d2ac30343ac931bf82a8bf87bab : SUCCESS in 1h 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/933d72f22f0948218c226c0070469049 : SUCCESS in 7m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2be20998f0024a09b26dfb9093cae7af : SUCCESS in 11m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29eafa55362b466b94b09a41b0b39ac5 : POST_FAILURE in 9m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/868f5dfe95124261ae89cdf06563d5f1 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28d28f37cbf4935824996d34fb7e7af : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8e9e6ac5bb54fe991ebb3c7c0dbabb3 : SUCCESS in 8m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4ef5a9f2a2ee415bb9bb7e539c2cbfe9 : SUCCESS in 1h 06m 29s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/abf574b8ef034ead856d71bf44eccf82 : SUCCESS in 1h 14m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/086b65b320404663aaed1259f253ad5a : SUCCESS in 41m 41s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/2545cf99525846ee963f39d18bd1b873 : SUCCESS in 35m 24s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/d92577ebd89e44a7a2f4f29a0e665b1c : SUCCESS in 50m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1100d54892de4901842c5f9d5a21db3b : SUCCESS in 53m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/742fff124f994b69aaef8fd06d0448cc : SUCCESS in 1h 47m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/0a01b9b432dd41a981dccf92d03dbad2 : SUCCESS in 2h 07m 10s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/912dbacb1fa445a28e7df8c9d01ff8f7 : SUCCESS in 33m 04s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6ed3df55fce5473f90f1c38d8db2e09c : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/53fd6eb3be7a4794b774654ee9283666 : SUCCESS in 32m 58s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9f4c36ebf965404e860707e7c5e43c9e : SUCCESS in 26m 28s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9a488c390a69426db051a0e2f50b1010 : FAILURE in 36m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/da63126c50de49a0b50c6df012768d19 : SUCCESS in 1h 28m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e65bb8b2b8a642c082a27a2492a20834 : SUCCESS in 53m 11s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/276140657a36413aa47fa3734f790cda : SUCCESS in 53m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7bc1e7dc70b0424380bbc45fd8405485 : SUCCESS in 26m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/9855bdc677be410aa85f8886c0649ed8 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":8},{"id":"5d08a7dcfe5c91ba5419ded093e735d7aed4deab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-26 03:16:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"366f4e94a1138e1923a0dc07c43e182e4e9a7b5b","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 03:23:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/052231e32e0141358bb5d7254980e5a5\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/fb8a489d4bbc46d684f1e8e03416fb22 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"261d3d762f39fa08f65417cce221428c56e531b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 05:26:58.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b0760b609284256834dfb9384034a36\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/becace6bf05c4fdfbdba0d6345994dd4 : SUCCESS in 43m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/78e6b60b71ae42d08322981726ed560c : SUCCESS in 1h 02m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3505c2a82a414d749ad4ecd105f7ebc9 : SUCCESS in 52m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daa628db0db442fb97c0400666a9a7a0 : SUCCESS in 57m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6874da6bd44c68aae335009bc01462 : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d69faae19d9c4561bbd6a84b93d19e5e : SUCCESS in 16m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/126ad397e8bd4f65ad52ebcc6b2e9c65 : SUCCESS in 15m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2c0f5871a9614759854370f7b7e3bbad : SUCCESS in 20m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e46c3b6000f4fa391794891f556c30a : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/560930b0e84d4dfc86344d6f1b12479b : SUCCESS in 8m 57s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/47050179343946da8b72406f0a19fa3c : SUCCESS in 1h 05m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44209e71e22a40dd8525276399a025ad : SUCCESS in 1h 34m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/96c2eb88e33843ba8560b283ec7c5097 : POST_FAILURE in 34m 59s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/140c00d887f747f3bb42b594e5d53c10 : SUCCESS in 35m 48s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/a2ee93329dde485d9a1e012faed16c14 : SUCCESS in 48m 54s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b9d859d1ae744bd09f585f70db234a5f : SUCCESS in 41m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5c384b31307f4e0fa8119421dd818159 : SUCCESS in 1h 49m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/7a4f04aa62b74fc6810ee425538f6177 : FAILURE in 1h 57m 56s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/789fb3fa5f5d43868fdd027b20cdfbda : SUCCESS in 47m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/e9bde9540cf6447c973a911de6d2b247 : SUCCESS in 4m 31s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/736ab4eb35a84a73adf43cea85b364c5 : SUCCESS in 26m 41s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/fc0b6afcb4bc475183f70d61971723b0 : SUCCESS in 27m 38s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0a407c4ac7284923a36f42cbe81f3abf : SUCCESS in 39m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3aaa47ab00046bdac504415816d494d : FAILURE in 1h 29m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a008d38f3274a5f919bc7630024d51f : SUCCESS in 40m 11s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/c86f1ba3167b472fbb7a4530cad37a42 : SUCCESS in 53m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/12482dcba86d49d386c54360e113a66a : SUCCESS in 32m 37s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/35244036c51d410d9f766e5cbece506a : SUCCESS in 8m 46s","accounts_in_message":[],"_revision_number":8},{"id":"d1607c82f7d3c9d45d22c8b70333f6bfd4d576fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 07:04:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c43389fb587c46a5a3cc3c3c25297943\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af994235621342739d8b47274fe11f82 : SUCCESS in 17m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f078e62b7ca64d8a9901038ab711a897 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/315e033cc3ba41858f80a774d3645eca : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"04f10fa01a957b0ee0b28d57e74b732a18d177be","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 08:18:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5ddb8733d46e83ca155e49138216c00a60d84d18","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 08:27:07.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/1c18f2f9f6ed4d628725650f1642fba5\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/ea6c37c81f2742939f20b2c20093a859 : SUCCESS in 7m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"51eb943a308b475cbe9b519b72b2439ee48a52f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 08:48:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1528916da874386adde1d44f416e002\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35c1ca7113514465af077946934dc710 : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5c3a0a9296db469689009edfe58ce087 : SUCCESS in 22m 06s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a46217076ec0474c8c5f765fb3e6174c : SUCCESS in 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5af239115f865f88e42d66d1edf605ed335ca0d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:45:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d822b8941bc4a8d92374e13a9af91dc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05d40beeb05d4dbdaf495b8928a02d5e : SUCCESS in 31m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cf11f320c49c47e5b2af11ab56b28d7a : SUCCESS in 1h 22m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ed6f023abde4cb7adad65dcac26b00c : SUCCESS in 1h 01m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dad936430f0e483999a50b9b8dd66c8a : SUCCESS in 1h 04m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4dbf2edd2ac41e8a69ff8678074b5ee : SUCCESS in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05426433d8be4f9f9e8e05dbf9974768 : SUCCESS in 13m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/682d471f0875455d89582285a5750305 : SUCCESS in 11m 35s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c042e0d9469f4e4085c56babe8a7e610 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76eeb4cd87f645b8b9d6015d5e1f9758 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd33a8a4ed8746ccaf1e887ef1f782b9 : SUCCESS in 6m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/62bd93d49501479099e697ec4646b4c1 : SUCCESS in 47m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cc37f946ee34f98b8ab38a32483bf12 : SUCCESS in 58m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fd3e29fcedc94abcb32d6482d49f1602 : SUCCESS in 27m 29s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/34c0093ec38447debfd2b28a5aa682b2 : SUCCESS in 38m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/08b5bf0c95a24d5997fa889e3a2051b1 : SUCCESS in 1h 04m 58s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/533bf009c3b448ee9c307c0d5b7fdaa4 : SUCCESS in 32m 59s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/aee1ba02284142b1b0b2e88208ec6b96 : SUCCESS in 1h 57m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/e6b96bb1f8524f5e80c6e1ed317b4cf4 : SUCCESS in 2h 08m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/19412ede8d7e4d328b7de60117666eb4 : SUCCESS in 37m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/61bb5645ef4f484194b93f46796bacf9 : SUCCESS in 3m 39s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1053d2c398bc4b95bb483b67296ad42e : SUCCESS in 33m 06s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/5d5d511c8ec24e9592c95449bd17bb5c : SUCCESS in 28m 18s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/96a579b618ca4cfba3927e7d393ded05 : SUCCESS in 37m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d08d4d4d8d64434e960ad7e303ae0a1e : SUCCESS in 1h 29m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/579dc6fb77fd4930b693abc4bcc47d57 : SUCCESS in 54m 52s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/651c0cce685549068fbd41828372e4cc : SUCCESS in 55m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0ae45a13187f43b59149e6303f258c63 : SUCCESS in 23m 02s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/254c83a5abfc45e9a1a1c14a117795f8 : SUCCESS in 14m 28s","accounts_in_message":[],"_revision_number":8},{"id":"80fdf0d68a74214bb8033a4d6ec71c8b389395df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:46:20.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"1220689734085b87714abc165b35fb6925c14be5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 11:05:47.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"0ce191f2bd652ed8759ce79bfa05ba32ad311b32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:51:14.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46326fc081db4ebf90b4f2afb1558ecc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6239273d92114ed094bb49238c0c43ab : SUCCESS in 1h 01m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/69ae539852b04ab7b36a42ee545fa323 : SUCCESS in 49m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96d09ef0e77b41db91788f13d3242e82 : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/539f9016adb840a28848347e463ef9b0 : SUCCESS in 7m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f09b898d2e947bda807caccec7c0f39 : SUCCESS in 13m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/52c47553855c40569d27a1ea2f628880 : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5742c0c0fcc4cfda1fdc0477a272ec2 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7009fc6302c4a80b28e33b4203fe782 : SUCCESS in 4m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a055629fe1f4277b5c4c47da8ab0296 : SUCCESS in 52m 49s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fcd4f17c46644aa186f7e8d19a05fc01 : SUCCESS in 27m 32s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/49e2cce2950447a59a8048a5343a07fc : SUCCESS in 29m 48s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/0359869b5c5b4b30abbfce3cbd019915 : SUCCESS in 36m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d92ee8f082b8465a9fd10237eebcf94e : SUCCESS in 1h 49m 17s\n- nova-next https://zuul.opendev.org/t/openstack/build/0f405921db5742d8843342e1896658e5 : SUCCESS in 1h 33m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0978c15dd84c41fc9d18493aa80c627f : SUCCESS in 5m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e488692e2cc544d0b0121b0d2a9adb14 : SUCCESS in 1h 30m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/32f0bf60d4fe40a3866dbae35a89a949 : SUCCESS in 1h 31m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d37194992a44507858f5265f95e993c : SUCCESS in 59m 45s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/96a4a7c060b84e6ea9d32b4ab4184152 : SUCCESS in 13m 55s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/79a4a9f7dcdf49efa85cd30d0d1b09d5 : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":8},{"id":"d7a2e425da08e0fa2557a941bd067ac31aa6330b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:51:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"3df65f46bece65dc937c41cf29e7d39fb13c3d0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:53:27.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdeb12ab817a4a12ba993f2457462788\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5add2470a0a4c48b36fd06057c04973 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":8},{"id":"762db3de34175cfef4ec2404a1313f7abbf56c75","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 08:43:28.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-qsktz/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"84e0e201008e91e192f773f6c8ebd1ab4a4fa910","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 09:16:09.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-2grtt/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"ffe1bd437133fa6b0a6daea3d550b07cacb6f0fa","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 11:25:02.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-z62v2/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"8f4ae19acc22cfde9004377e76db2f4b77b459b5","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 11:57:38.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-xtvjl/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"706d472a8115e170587d0179a954cd75801595ae","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 12:30:40.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-pms2b/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"7077750faa964de5285774007eb7326c24509715","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 13:04:55.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998466-kjwpf/index.html\u003e","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6ce06745ed511a0a2c9f7b473c923c1655726dcb","revisions":{"f3ccffacbb1b17efb988a99fce45bab92a27148b":{"kind":"REWORK","_number":1,"created":"2026-07-23 12:50:20.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/1"}}},"commit":{"parents":[{"commit":"462e89e40c76139169c8f3e553ec84207d55cbcd","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462e89e40c76139169c8f3e553ec84207d55cbcd"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-06-25 13:15:31.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-23 12:49:16.000000000","tz":120},"subject":"libvirt: add memory encryption config for Intel TDX","message":"libvirt: add memory encryption config for Intel TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config. This\nshares some requirements with AMD SEV and thus machine_type and\nfirmware_type is extracted to the common class MemEncryptionConfig.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3ccffacbb1b17efb988a99fce45bab92a27148b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3ccffacbb1b17efb988a99fce45bab92a27148b"}]},"branch":"refs/heads/master"},"bc6824e90d2ab2ba9f19425e46432d476ee41db3":{"kind":"REWORK","_number":2,"created":"2026-07-24 11:57:47.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/2"}}},"commit":{"parents":[{"commit":"462e89e40c76139169c8f3e553ec84207d55cbcd","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462e89e40c76139169c8f3e553ec84207d55cbcd"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:55:46.000000000","tz":120},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc6824e90d2ab2ba9f19425e46432d476ee41db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc6824e90d2ab2ba9f19425e46432d476ee41db3"}]},"branch":"refs/heads/master"},"cb37f8e1223af83e16309100487b16993ac99147":{"kind":"REWORK","_number":3,"created":"2026-07-24 15:14:03.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/3"}}},"commit":{"parents":[{"commit":"462e89e40c76139169c8f3e553ec84207d55cbcd","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462e89e40c76139169c8f3e553ec84207d55cbcd"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 15:01:39.000000000","tz":-420},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb37f8e1223af83e16309100487b16993ac99147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb37f8e1223af83e16309100487b16993ac99147"}]},"branch":"refs/heads/master"},"910436857da955ba31996c7cd116b1c724bc2dc7":{"kind":"REWORK","_number":4,"created":"2026-07-30 09:25:16.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/4"}}},"commit":{"parents":[{"commit":"462e89e40c76139169c8f3e553ec84207d55cbcd","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462e89e40c76139169c8f3e553ec84207d55cbcd"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-29 09:49:53.000000000","tz":-420},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/910436857da955ba31996c7cd116b1c724bc2dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/910436857da955ba31996c7cd116b1c724bc2dc7"}]},"branch":"refs/heads/master"},"291eb39ba15bcf428c3eeb2ee7ae2a2e9ba92d4e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-07 10:51:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/5"}}},"commit":{"parents":[{"commit":"8dfb7d5943563a998f91d1a4a9a15f391bd72a93","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dfb7d5943563a998f91d1a4a9a15f391bd72a93"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-05 09:41:54.000000000","tz":-420},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/291eb39ba15bcf428c3eeb2ee7ae2a2e9ba92d4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/291eb39ba15bcf428c3eeb2ee7ae2a2e9ba92d4e"}]},"branch":"refs/heads/master"},"8457ef6437c83961643c69ac6f702ebf52d5531d":{"kind":"REWORK","_number":6,"created":"2026-08-13 07:25:43.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/6"}}},"commit":{"parents":[{"commit":"8dfb7d5943563a998f91d1a4a9a15f391bd72a93","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8dfb7d5943563a998f91d1a4a9a15f391bd72a93"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-12 15:39:17.000000000","tz":-420},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8457ef6437c83961643c69ac6f702ebf52d5531d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8457ef6437c83961643c69ac6f702ebf52d5531d"}]},"branch":"refs/heads/master"},"3834b49da0e44c18727b701e8fbceeea2b34dc98":{"kind":"REWORK","_number":7,"created":"2026-08-21 13:41:14.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/7"}}},"commit":{"parents":[{"commit":"f87b89ac900bb58f64a4b1a98adb3b4ae1f84562","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f87b89ac900bb58f64a4b1a98adb3b4ae1f84562"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 13:29:11.000000000","tz":-420},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3834b49da0e44c18727b701e8fbceeea2b34dc98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3834b49da0e44c18727b701e8fbceeea2b34dc98"}]},"branch":"refs/heads/master"},"6ce06745ed511a0a2c9f7b473c923c1655726dcb":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-08-21 14:56:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/66/998466/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/998466/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/998466/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/998466/8"}}},"commit":{"parents":[{"commit":"a69768e478d7a08355b11fba5a814b08e495ce09","subject":"libvirt: Add Intel TDX host capability detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a69768e478d7a08355b11fba5a814b08e495ce09"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:40:41.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 14:48:25.000000000","tz":-420},"subject":"libvirt: Generalize memory encryption config","message":"libvirt: Generalize memory encryption config\n\nExtract common constraits for memory encryption models from\nMemEncryptionConfigSEV to the base class. This is in preparation for\nIntel TDX and possibly ARM CCA.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Id89462b3652aee0f695dea60c2fae1b9430f8b8d\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ce06745ed511a0a2c9f7b473c923c1655726dcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ce06745ed511a0a2c9f7b473c923c1655726dcb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
