)]}'
{"id":"openstack%2Fnova~998571","triplet_id":"openstack%2Fnova~master~Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","project":"openstack/nova","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-08-20 21:11:12.000000000","reason":"Change was submitted"},"11082":{"account":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"last_update":"2026-08-20 21:11:12.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-20 21:11:12.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-20 21:11:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3","subject":"Unify the thread pool executors into a central factory","status":"MERGED","created":"2026-07-24 00:52:38.000000000","updated":"2026-08-20 21:12:21.000000000","submitted":"2026-08-20 21:11:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":128,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"998571-eventlet-removal","meta_rev_id":"14f4ffffb5ac14d7c913886a476db2b1f56eeb6d","_number":998571,"virtual_id_number":998571,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-20 21:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2026-08-19 14:14:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-19 14:25:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-08-19 14:25:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-24 01:14:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-24 01:49:17.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-24 02:53:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-25 06:25:40.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-29 14:14:14.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-08-08 19:35:59.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-10 16:31:25.000000000","updated_by":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"REVIEWER"},{"updated":"2026-08-14 17:33:06.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2026-08-19 14:14:44.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"057287ac9f1e35c072ff48f280ae8353352f1d08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-24 00:52:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bceb95a35b601f5f64b2208cbdc41853b5644709","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 01:14:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8c13c4546634c8b806a42e79b079fee\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7356d8ed5d38468ca13466fba3edceb1 : FAILURE in 20m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7855fb024e3c412aa3cb4861c8a293e6 : FAILURE in 19m 09s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eb46311ad4c342be83bead42944808cd : FAILURE in 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fce3e35384b31c908a687f4756d04bb1cd2d1419","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 01:49:17.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":1},{"id":"69e55977549fe0ae7a9b42db1f101c6f59b9ea4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 02:53:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d2583d086d748b7b8873d6b36d316b4\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/377f24f4a026459bbb5020373285a971 : SUCCESS in 34m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e92db6a1360348e9b0918a375fbb218a : SUCCESS in 51m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2a9a16ea63447918aefa923acbda54e : SUCCESS in 56m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/954831fb763543d5907f3a4a59b999fe : FAILURE in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f677dd43824e32aeb2a0556a805f02 : SUCCESS in 7m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4374c1f422c64966b001f26cde5173f0 : FAILURE in 12m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5189f8c43adc4dd281f4c0daf5bf1681 : FAILURE in 12m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/91cab26a2d2148d0b34b7fac8b34c879 : FAILURE in 10m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18acf4c99c3c4403ae2dfdb772ce4597 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1736791097a945bbb535628299ec9514 : SUCCESS in 7m 54s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/01dfb9055d404b7488a4742e6ca94d6e : SUCCESS in 26m 37s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5cc9fd00f5a949d79326ad8b8f54b95e : SUCCESS in 1h 34m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3ed5a8e457304ca5a39b153ea0e429de : SUCCESS in 22m 27s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/675b001143ed41a2a0d3dbaebfe5e7a4 : SUCCESS in 37m 32s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1d461173d9d8447ab6379a8ede83b7e9 : SUCCESS in 58m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f60bd8ca5f4045c2ad4e376ac9a6ac8a : SUCCESS in 27m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/07cae5f113bb460b917b27209149cc63 : SUCCESS in 1h 14m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/dbccdc53abb9438782d2ef687c75ad3f : SUCCESS in 1h 55m 37s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/23026a87346e45beb9834ec3d3825d55 : SUCCESS in 35m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/472b5daa2b294084b6df9f15fe396952 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e49ed42b37484906b891d4288f9f8bd5 : SUCCESS in 26m 04s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2e031658cffa4090a9ab98da06bbe644 : SUCCESS in 20m 54s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/913ab13ac1bb48568de065a5cca9aeda : SUCCESS in 31m 29s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a8a0d9ecd80d4956bcfbfd61d07bd7f5 : SUCCESS in 54m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce4d441e3a2147ce9089e49a9b01e0a3 : SUCCESS in 41m 51s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe5e9652c15445c4b420d0c921f29bc6 : SUCCESS in 51m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b482633f0c6644e99ded3e49a1ced201 : SUCCESS in 15m 19s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/41faf0cffeda4ed79c29ac73b0cabe66 : FAILURE in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5b1a91ab930f2bcf056b83721329d5016ca7c558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-25 04:29:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"32209a20943fb84fcb91f511ffe01f3c79710db1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-25 04:29:33.000000000","message":"Topic gs4 removed","accounts_in_message":[],"_revision_number":2},{"id":"9ca3ea75a51bec09a0a15988d9dff93684880333","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-25 04:29:38.000000000","message":"Topic set to bp/nova-services-graceful-shutdown-part2-task-tracking","accounts_in_message":[],"_revision_number":2},{"id":"9b771570a1031f28d3d615d9d3f175be8b230c04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 05:11:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5c8da000ce4542a05658542730b580\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/46e853a393734fc09243dafc0ad36309 : SUCCESS in 29m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/523cbfc11d8d4cf7a9c486a4917da9b1 : SUCCESS in 27m 23s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ca77266932e44e5090c78764c3a017a0 : SUCCESS in 39m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c6e307fd7b91982e7536be7e84de0d0b26720fd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-25 05:24:54.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":2},{"id":"467bde93fb5f840622dc0f081ccca9ecd78a2854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-25 06:25:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e8bc344d3d840bd9dccabb095a78e73\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c08dafafc051473bb3924ddba53fac72 : SUCCESS in 42m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3b25110cb4d0437c959805c9aba334ce : SUCCESS in 1h 32m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58c8878ad2a8486e87ca19d4dbadc8ad : SUCCESS in 28m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5331d8f59204ff2888ec73350ff4a03 : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5758f7329ced4998819b0f7cacef1ef5 : SUCCESS in 8m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/822621ff7472491480086fc6e6473459 : SUCCESS in 15m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04e7ea430a054a8296e06dd641ef2eb5 : SUCCESS in 8m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c7b30f23099140a7a1530df6db923897 : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9f98941d9445d7a0dfbf222bf9414b : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2330d3847ee42be8ae75ee0ba43a37c : SUCCESS in 8m 25s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/19856514fb8342d5a717ac7350246a69 : SUCCESS in 56m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1152514414504b6896a7ad25221fcda4 : SUCCESS in 1h 13m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73701e288b0b4145936de77d26d3a53e : SUCCESS in 48m 57s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/2049358913034f6da593f4b914102aec : SUCCESS in 34m 51s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e69f0328abd24d52a000ebb91962524e : SUCCESS in 58m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/88ff2d2a3f38436082a7309dd2684a21 : SUCCESS in 51m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7deb3989bca74bfea6ba854a9b73d67a : SUCCESS in 1h 49m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/47783227ddc747d3a7a4c570929f6fbf : SUCCESS in 1h 04m 34s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/f578ae9dea4f4395b7f1e0e187d28f1d : SUCCESS in 25m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0acbbfc6dd2b41279ee982b1d5c8b816 : SUCCESS in 4m 59s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/92849f254f3a4cce8542a6b0c8dae07b : SUCCESS in 26m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e70b2d8d8f0f41588625d9dea07e8686 : SUCCESS in 26m 11s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9b3740ddd4904ea29247b316167c4475 : SUCCESS in 32m 15s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f427dc003ea043c39dc0911384d3b619 : SUCCESS in 49m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9b3023efcf74534815a8cf23aff5878 : SUCCESS in 24m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/43b86622bb584d2dbb246d8fb9b20460 : SUCCESS in 54m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6f8b3825f9ba475fae258a24a74843af : SUCCESS in 16m 12s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/45fe823728ab4359911702f17054669e : SUCCESS in 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e323985b6360759feb788d476ac3a34500b1cc3b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-25 06:35:45.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"f00d9dec6e94037e68ebb49cebd672d3fad6d290","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-29 14:14:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"268d7419161b169de4ab789bc92d44ae85abea15","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-30 19:27:56.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ced09c2de526b1f661cb651e02925ba0a6de4ccf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-07-30 19:42:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78fc96cd43c81e961be0fd076a681e2ac4ad104f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-06 09:30:06.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ca3146dd7169a63b09f8a193009518fa27ddae9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 19:06:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"49641e0463d8cd98d407a04d4253e73f687f782f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 19:09:36.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3795d81f819b592a907db0eef3c67824dc896eca","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-08 19:35:59.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-j9cr5/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"f97bfee3eac696445717b0148427bc4168f92f71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 19:36:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b0bd76bdf894a7b282f9a89f4debb16d09a62971","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 19:37:04.000000000","message":"Topic gs4 removed","accounts_in_message":[],"_revision_number":4},{"id":"c94bd2ffe467834049f63d437aea538b0c13263d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-08 19:37:09.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":4},{"id":"eabacad9dce1679c730e6b707c0c43d4ec254876","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-08 20:01:53.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":3},{"id":"f89c3139cff91073048e3105956998fb571e5a81","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-08 20:07:38.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-m96j9/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"6c23bf41a53d4ea14083a51e9fc84ac15173007b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 20:18:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4c9da3ead0042ee9a5e82b68e31b9de\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7bf06e03ee0c4b878f838077977a7e4f : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4697acd190d54c8cb1d8dceb8094bcba : SUCCESS in 25m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bef980b8f07545968cae8377bb7b1d88 : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6972ed9b877b02cb59a951a93105ec4ecbc87ee3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-08 20:32:02.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":4},{"id":"b5b42a1e0dcaede6003854f4dfd163059ecf0010","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-08 21:42:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3a22e8fd786472985f43ec2a9bd2805\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f0b95a636284752b3b824eea14273dc : SUCCESS in 44m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/794fcabe3bb2437396ac6fbec8893370 : SUCCESS in 1h 31m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e41318ee95b94d81a900a83a07a4383d : FAILURE in 39m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4a0cf234c194038a36798727accad2a : SUCCESS in 1h 17m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ebb461add2044f18d820ebaba47fb48 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fec9766442d4e6c85305444b8dddfd2 : SUCCESS in 13m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ff1037099ad9475f9c8bbe4365a2bee3 : SUCCESS in 13m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/dff6b943302d4b39b06d829b971f68dd : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00765a3e09ee4d4eb64973950a53f645 : SUCCESS in 15m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d401b45539d04f8bbf69abcde3b771d1 : SUCCESS in 11m 12s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/988b8d8d8e064c60b48187c3aef521d8 : SUCCESS in 26m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1f96b4634054fe9b428d4d505495c90 : SUCCESS in 56m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/52be61dcb49446aeaf4fd1e52f02bf80 : SUCCESS in 30m 36s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e860008c0b36458ea3352bf57ffac211 : SUCCESS in 34m 18s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/2ac279c28378464b88520f313305a574 : SUCCESS in 1h 02m 59s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/92e6b2aa381c4de997cab1bba06161a9 : SUCCESS in 25m 10s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/6ea9dc01478c44c194af2de7901f3e37 : SUCCESS in 38m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7496a20203e1432ba35a6e9f76f68cda : SUCCESS in 1h 56m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/aee4a621044e432f89a3cfb09ddce251 : SUCCESS in 2h 00m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2fec4c9ea06f4a28b6b74c964a620eb6 : SUCCESS in 25m 44s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/265b0e4f7ec8402ebe03edb3d05bc1d1 : SUCCESS in 4m 38s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d23fec4260d443dda25e04d94927dddc : SUCCESS in 28m 03s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7652733abbd441c58abdd3fef6ddea70 : SUCCESS in 27m 22s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6847eb9297a041e3b22fb11d1960712c : SUCCESS in 33m 33s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcc5e5df261a4f00be0ae89fdb83fbb9 : SUCCESS in 51m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abd45ac64f0242839eda727801bc7dc3 : SUCCESS in 1h 00m 52s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/02a5de7f695149ea9e8d6e0eb517e66b : SUCCESS in 51m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e8f87c0844a14d6db9d7c18471e15e64 : SUCCESS in 35m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/6e1ba1834342438396238802f2e9d4cb : SUCCESS in 13m 52s","accounts_in_message":[],"_revision_number":4},{"id":"c31763807e3c9c0cff0dc04e12e42c4b8535ca80","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-10 16:22:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bc9a9dc95111463b6c1e8bc3af6e12f4bfbfa447","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 16:49:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae4dc6e03a6b4f4e9c620fbfbf4523f6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2426af99cdef4250b2f51fb8639220e8 : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d14892b1e4b94fc4a417849404b4233d : SUCCESS in 22m 43s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/e2866f2163c3443c88d5b1c83c7b59a1 : SUCCESS in 24m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ff05e3cd1360b677891bf3cf8512ee05b35a3aaf","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-10 18:06:49.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-wxsbz/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"2f19f8be7901d3dbcb025cecfc6d68be5c31460d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 18:35:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/915b5c82c44e407288d9fe10677502aa\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ed90a467b6b42db82bdbeb8673876d4 : SUCCESS in 59m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4860ef65539e4403b77c6c762a65d090 : SUCCESS in 1h 02m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/546ae4b7b6914bfab31de882bdeb5250 : SUCCESS in 28m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03170d9979c54937977cb288eddc156e : SUCCESS in 1h 24m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a081d0b9947490db02c200c194a80c6 : SUCCESS in 7m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ada19732218b479395b8a275ce60c699 : SUCCESS in 12m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/785f3b7a37ff4a07b1dda01a03473fd8 : SUCCESS in 10m 51s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/191938a3b2d84f3b85831eca9686c6b4 : SUCCESS in 17m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71ba31af8f8f4521837b14abc0ff5043 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90bd39bc58ad49849e440be3e30c1fef : SUCCESS in 9m 08s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/942ed7a0ba7a4f36b048f4348bd7dad7 : SUCCESS in 1h 03m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1bf5c19d0b94b8da706cc7858932e5c : SUCCESS in 1h 23m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6b951358d7154dd9b60bf0898557c8d9 : SUCCESS in 52m 52s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8ec55fd883374affa7acc3b4f4b451cb : SUCCESS in 37m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ce5768d3ba174c1c839a833a8ba961b7 : SUCCESS in 42m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7fe5c5faa4d04ee88e51c4be9488ad02 : SUCCESS in 52m 22s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/090d1bd3e96f4c06b3d9a6b6f9e52a73 : SUCCESS in 1h 06m 15s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/47150304a3184125aadda14deb70690e : SUCCESS in 1h 57m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/839fa6da8e4b469c83734d231b9ae2ac : SUCCESS in 2h 06m 59s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/23f6db05876349fc9104c5431e3f56c0 : SUCCESS in 25m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/24b1e60615f34d91b25633c888a83554 : SUCCESS in 5m 06s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1bc6fb43314a46b18a6068a5181d619c : SUCCESS in 26m 00s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7204da053c9e4621978761376ffb7c13 : SUCCESS in 25m 51s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/758e361ee08a4fe8992db8a45973fb06 : SUCCESS in 33m 54s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/91d2886609ef4093bfe20b2e4ff9a224 : SUCCESS in 1h 26m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2de4f691614f489b8b2bcb5fc1af0d4c : SUCCESS in 34m 34s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e4c29cc674094210a25efbb894616166 : SUCCESS in 31m 38s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3abe0679f56c4d87a5ff2c1b60d1ad62 : SUCCESS in 35m 59s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/943d3964699c4379bdbe9c7c18a2a364 : SUCCESS in 13m 27s","accounts_in_message":[],"_revision_number":4},{"id":"15adfb3b69b4549dcec26fb8f89da13ab23c4ee5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-10 18:46:05.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"d79bf43f39fc1b6fb263015ad8596c7e61a81cab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 04:14:51.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"de43248dc9ec98e22485ffc179132a5fd9423932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 04:27:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"92b553e8ddd7845eed186191dbcd745937c8ff6b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-11 05:14:53.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":5},{"id":"7db63e56ecc07af59c2b5c9d050fddccdb494b23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 05:26:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ca307f1e07d43c5881ede101de8f702\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c7e3c081ba0c4af7bd917bd2387f0497 : SUCCESS in 25m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e42de3cde14f4378afa1af633ec066ae : SUCCESS in 22m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/bf4578e0d4d1454fb14ea2e6d773290a : SUCCESS in 41m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"015784d1a9af5f2c3f7d5dff00635a4d9c268a7e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-11 05:47:54.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":6},{"id":"c2581812bfc0bce8b4dda7207fb0937c3b7d0826","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-11 06:00:19.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-2fgw4/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"fc37f51778f8a163b6e8cfc71309f377930c6a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 06:27:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726ce3b0c0744d5c9c4cc4b3afa8ff0c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1fb0216d5e4b4cadaae71c091b873c4f : SUCCESS in 1h 04m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2a85574ca9a144a5b9cdc9da82fd30b4 : SUCCESS in 1h 08m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c429a67858e44c2af4975afec1276a6 : SUCCESS in 54m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dde15e32336f4e1981ea3d9e84810577 : SUCCESS in 1h 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690a817fd9ce48108b09118d59db2916 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b280d49e2d5b4915ac63e35444386d88 : SUCCESS in 13m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d3b23f2c98264353ad1f9c16e4003674 : SUCCESS in 12m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0d95342acfd04728a409d45410fec79f : SUCCESS in 16m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52446329969c493ca17a8f5246f03514 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562123fff65b4037a64b2b127103f5cf : SUCCESS in 9m 23s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/58cf20ceb3c148af832143da00777e28 : SUCCESS in 26m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd77eb3b394a47aa93fc71dcf8d34667 : SUCCESS in 1h 21m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/47d1ddee02734e64b92ff514509e84df : SUCCESS in 53m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/fff5ef53cb0e44f5b588670cb8b12c56 : SUCCESS in 35m 08s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/b63e721ac5b5443ca6b4ec5678fe3205 : SUCCESS in 42m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2ea644ec33bd46b891082fb0782c2d7a : SUCCESS in 48m 24s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/14f94a48165249ac831eac9a8107ee86 : SUCCESS in 49m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6bad84bf53d9420f9c4fa18b1a1609bf : SUCCESS in 1h 43m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/6819640ba38946de91df5f973940eb99 : SUCCESS in 1h 53m 03s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/2a5ef59e232b489d8ef7c0e9e65efce2 : SUCCESS in 47m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d3ecedb690e5497dbe2ec720cee04403 : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5a74bc4ada7d48de9e53c139919ba4be : SUCCESS in 28m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2e216463ec82496293dd1194103ebe97 : SUCCESS in 31m 19s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/df2c1f504754431d890a08fa78f71554 : SUCCESS in 26m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a45fa39e0a95411e9d4269474c172a63 : SUCCESS in 1h 37m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/031c4c4d60f940f680cd2e5621e6d230 : SUCCESS in 59m 49s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0d27c745dd7a48b1ad674a2e95c06658 : SUCCESS in 52m 58s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2255cdc2a5794c36a37d868c0ae8d296 : SUCCESS in 27m 04s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/c51900159f0646adb3ed188cbde3401c : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":6},{"id":"779e87e63b2820eee5b972c498f461ca58a6dbee","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-11 07:40:42.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-26ppk/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"7f6e81e51d7d99b73cdba0bfd0af75d5834ca2f6","author":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"date":"2026-08-11 14:43:54.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e644d02698bb895efe662165e90ad3b449c43118","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-11 16:58:28.000000000","message":"Patch Set 6: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a1b675c5228d1ae25b30ba2571d055afc4a03515","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 17:53:45.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0345f370a6dae9c9510006becebca8bb28cb1f09","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 18:06:27.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e34efb44b6ca82b30a2fe610462f7637021915db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 19:37:25.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"abbec74e53aeb3ac4617307e623466c6f7aab742","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-11 19:37:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"db9ce38c55177e6d42f1a980e27b364925c3618e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 20:18:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b10de4cb138747bcb17e90b6052d110e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45bb582e8522448bbfe4c2e5ee84ca10 : SUCCESS in 27m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/20e7dd1e64ce4c1dbf8e1fa2d8f62896 : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4b61c0e3de1e4e3fbe739a03854f8b11 : SUCCESS in 38m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c2af9df00e4fcf93c1db461640e55dda434cb670","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-11 20:26:20.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":7},{"id":"26f29806ee58df077eaa1adf4f47b58fd6e04f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 21:06:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f299d78d3f534dc6b1cc993023fbb14c\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f88776859404748bf847f29cf411fbb : FAILURE in 1h 01m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/751d64e73db54f63ae9bcd52ae9872d5 : FAILURE in 57m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2f94961c7704e53ae1d2749cdbffbf3 : FAILURE in 27m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b5e1f38614a47b78e28f5abcd5ce890 : SUCCESS in 1h 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35563efe98a4f7981f50719193c793e : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f887fc953623403893f69d59e090d6bb : SUCCESS in 14m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/013b8d55394a406caf8f65f52c84c91c : SUCCESS in 13m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2282be80cb3c418db7de7c50556d93a3 : SUCCESS in 17m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d197e4ca43fc44a2b3a6105a71f1b295 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5069592e219c4d2e88d10d5e976ccb72 : SUCCESS in 8m 22s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ff5f2de5db874df1ba5f7fd6276564b9 : FAILURE in 34m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4aa9ce1cb5524d14a1f0814dd1e54804 : FAILURE in 54m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a10db8d41ba24c01b933463aa1280e97 : POST_FAILURE in 48m 46s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9e7baaf35b60455fa2fe73f529ac66e6 : FAILURE in 31m 04s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/dd501e1a95f44f2fa60179b5a7445677 : FAILURE in 34m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1948204e99c14d12990b5317f55b5c67 : POST_FAILURE in 47m 24s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f54dd8780b2e4e5b97c860d89a042a25 : FAILURE in 47m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8e57af81702e4169ae178a4ea9e5c967 : FAILURE in 59m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/cd182a8ff4c145ea94080e841c473e86 : POST_FAILURE in 1h 02m 07s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/db7a55bb16cf44c38a22f11bcc0b5641 : POST_FAILURE in 34m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b7f7142d1bc043d68d2c570dc67fb08f : SUCCESS in 6m 20s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf3808378e434e03b9f694c6a17c49ff : SUCCESS in 27m 04s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/04d85c5fa2ea4e1e91c60ef0b183fcfd : SUCCESS in 25m 54s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/ffedbf40c98f4ac39a6b94e1e8776229 : SUCCESS in 30m 14s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a075a470910b4589940e61c020062f53 : FAILURE in 1h 13m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c9e63e2e8e444ea99d17ad08f969d1e : FAILURE in 42m 03s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fb88443b663e40b19097fc5dc09c2de8 : FAILURE in 44m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f7d27400c2474aeb8a6f3d690d89e6ea : FAILURE in 34m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/94f8aee45bd940c18fdb8ac940a23188 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":7},{"id":"0720858923adf91e3cdf7ad9b9d616f56e9a1954","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-11 21:14:33.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-g8fpm/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"79b3dbc98d3a026f8c961dee2905fe4b4543b2da","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-12 04:06:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"95635993dacb66c2f7aa0602319573a32bc9ce56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-12 05:01:24.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"d17f7d01f61cb93fdbd822229824603ef7df56ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 05:41:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8f3d02116b142d9bc88e22338530ee0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ebb4c60838cc40b3929b872eda78a297 : SUCCESS in 23m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3d836ce3bbcb4573a2658b294ae63c46 : SUCCESS in 15m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3bc3443560e2471c85cb92676bdb9c14 : SUCCESS in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"850b67a5be7d52c258e4b83e099f99a7dd223129","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-12 05:46:27.000000000","message":"Patch Set 8:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-wsh66/index.html\u003e","accounts_in_message":[],"_revision_number":8},{"id":"d1a2d422231e31684ea17e59baadef73ee07e4b6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-12 05:58:08.000000000","message":"Patch Set 8:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":8},{"id":"9a0de845127ce44578af6f2b3a1e99396c2664ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 06:51:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd48fcc07ad24213b91042380a54d094\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f3942ecc9965418abb2a5a6acdcb78fe : SUCCESS in 30m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/81a68cc5e8cc4646b9a3e3b5bd1f58ac : SUCCESS in 45m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a610da9e9dc04c7685948ec4c907e7b9 : SUCCESS in 32m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a373511a61744b46a19af87ac6c84d1d : SUCCESS in 1h 24m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d6d60eb88147aba16e97c4e73b8594 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e6c4d9ff35c4575bdd940da99b67904 : SUCCESS in 11m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/783299cfaf08401e826d4fdd859b23a4 : SUCCESS in 13m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ee90d083bc9c48f6888d4927fbbc27ad : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa78e301cec947b8a7b663a3e0b70c7d : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e751e07d7d4494faa19be925f2d783b : SUCCESS in 8m 46s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c8a2a780b4c94295aff5f6984f270495 : SUCCESS in 1h 06m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/425dc657a9f14999849b8d0a30e9eecf : SUCCESS in 1h 32m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ea59fbdbf53343eab07eb059c5624ccb : SUCCESS in 37m 47s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/39d4497b0918431895cd4f9d47b83f52 : SUCCESS in 36m 33s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1f7c4326949c4672b116af9edae62c3a : SUCCESS in 1h 00m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/200283f596d34a2ca53b67c4c4c78d80 : SUCCESS in 39m 55s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a4a78feb4306423f972f2110c7097b19 : SUCCESS in 1h 04m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2c30a3cc211a4f75ba3c61594fe38b61 : SUCCESS in 59m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/732e3cbad6e744cf9aac6e70adb7ca76 : SUCCESS in 1h 47m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bcf7ba18257a41daa9bfdd9d4ace20ff : SUCCESS in 35m 08s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/517df3f32fa04910a0353d66df2d1953 : SUCCESS in 5m 18s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/923b1fe8b34e46a7b1dd12a4e615a446 : SUCCESS in 27m 19s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/d2ec1a14ebfc41009a21b9fe064206a3 : SUCCESS in 26m 40s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e8ca6541fcc94dad939dec24b7acd78f : SUCCESS in 31m 51s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae1d7f06d7cc4a0c950bf0c97f2b3de4 : SUCCESS in 1h 36m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4d25366ebc3429487630c99bf772314 : SUCCESS in 52m 37s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2ec151648f7f42b3a22796c1c8661802 : SUCCESS in 30m 15s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7431f4a121d84d66a2024e7475520768 : SUCCESS in 19m 08s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a3db23a6a12f4127b6591a42a034656c : SUCCESS in 11m 43s","accounts_in_message":[],"_revision_number":8},{"id":"d30e8358d4044d0ad421ef1309eb0102e30ab475","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-12 07:02:47.000000000","message":"Patch Set 8:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":8},{"id":"dd90ba110515765651d129b59f0091b76d4274bd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-12 11:19:08.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a852b447f3ebc3e3ddfedb3b59141efa6fc4c92c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-12 12:48:52.000000000","message":"Patch Set 8: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8deaab69a78d6d2f459ab86b4d598b4ce0f09159","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 04:01:40.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"258da2b0dde12e3c36c686d988cb207aef065b93","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 04:02:10.000000000","message":"Patch Set 9:\n\n(16 comments)","accounts_in_message":[],"_revision_number":9},{"id":"cd3176649b6912dfb166e1d4c820c7d49c01cb26","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 04:56:03.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":9},{"id":"44cfce182c9172674bee78d481b0bde83e0d6468","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-13 04:59:28.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-lwmfk/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"b62fd86ac5c8bee829a1f3e01d2bb1e114055365","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 05:24:06.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b444b24a54df4fc4a6805618bef60d99\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/effb065a97fc4640af34ac76e3af75c7 : SUCCESS in 44m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/70465fdcab3a477e95e5156367411c80 : SUCCESS in 59m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b28c6b48147442c8001d10ebf6fcafb : SUCCESS in 28m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cac6f18ece194a7ca2e77537cb098861 : SUCCESS in 1h 00m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/542d7af2c77c4f1aaf3a7456a1a62df1 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d84c986d64d04ede99cd8fb8756c2132 : SUCCESS in 15m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0c8993937e049738dedfde0cf72d4a5 : SUCCESS in 8m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/555244f0be0d4a72bea5c769792c5993 : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63c2dd09829454095a48d4b457e7759 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74f269b692e4d80a1d10288c7cef442 : SUCCESS in 6m 19s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7273131e77394eabad004ab2f5f28921 : SUCCESS in 27m 39s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0647fe8f3f4548e7a867f76de98c52b2 : SUCCESS in 55m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2353869ad1664d53ab188c8ee5df7ec6 : SUCCESS in 28m 40s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/46de86f6a0c34c829c85db8583271df7 : SUCCESS in 35m 01s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/aacbba0cc8a647b2b29296172106ec39 : SUCCESS in 1h 02m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1fdfe68508f04c718f59b74533a1f5de : SUCCESS in 50m 28s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8b106dde47214068aabdc571af70541e : SUCCESS in 37m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6868b797fd6543b3802a374b69fcbbaa : SUCCESS in 46m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/ef5956aeaa5049c494263eec6acfa16c : SUCCESS in 1h 09m 46s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/bd7fe6c2d04e47a09552e7ac8f540f8e : SUCCESS in 49m 58s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9bb826fa71f54627a14ba036d2088cbf : SUCCESS in 3m 27s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f4de222a44b0433890c1767a5723b41b : SUCCESS in 31m 35s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/2321fd6b5cfe4209b871c30f83189585 : SUCCESS in 30m 52s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3a2e1aca8af74d9ab87678dc654624d4 : SUCCESS in 27m 44s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1074df6bda144348bc20997b5dbd5bf5 : FAILURE in 57m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d90f144866f4909871c350edcdd050a : SUCCESS in 52m 35s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/2ec319bb5ff04bf4b19144c6ce6e2b8c : SUCCESS in 31m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/98e8bc98a5314b649fb1db291ae4fcb2 : SUCCESS in 33m 32s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/8c2b12ba122345448d84156e588b7611 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":9},{"id":"804e5af7d15bba4e451495d91ff6f7fd378023f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 06:16:19.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d451dae0e6db4ca1bbb34b7f8a7fae37\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fae5c407cdcb4409abe12e2bad211a3a : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cad00a78520043eabfa7b7379295f700 : SUCCESS in 30m 21s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9b4285b6696c4786814e4dbe2622e800 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4cf7ee01bb21a3337ac5759cfdea37d6d846e078","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 15:04:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"12a066069840ae8f448deef2f91b6d220188b071","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-13 17:19:35.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-f299t/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"20599fb2dbeb459f1690c71f261d2c921be015ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:19:47.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd0afc967bd2411696568bd04aec8614\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d44ee99171b54b2288c89eae9742bd75 : SUCCESS in 1h 09m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bfbc6416ee27453690583ccbcbe10961 : SUCCESS in 1h 40m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e14a14a52e749e3a1ad6505e292f057 : SUCCESS in 57m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a77e2af20e8548498b44e547e9dc0026 : SUCCESS in 1h 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1115da0c0037487c898cb3ed440535d1 : SUCCESS in 7m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a02488e002645a8a0a99d5bb58b8e1e : SUCCESS in 11m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87609fc20a8b474194c2319aa12514ad : SUCCESS in 14m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6bd4f96d60d0461cb79925964e1c84ab : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f9f917809140eab1f7f719f46291cf : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdce37fdf4a94e9099a90d503c6f1e4d : SUCCESS in 7m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/243984e25a374c639f0c423ef08d7d17 : SUCCESS in 33m 29s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/35d61efdc5d64032876f525526128c76 : SUCCESS in 1h 11m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/02297267c5f54d1fb9d36dc15e89d17c : SUCCESS in 50m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f9c06932088f42a4b6d0027e1ee15cc3 : SUCCESS in 36m 29s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/446dd9f44ff04f958cfd4b71e9d7247e : SUCCESS in 51m 26s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7c3cd7b84f844c7db41c8e7563fd8465 : SUCCESS in 40m 05s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/53980986b958466fbe93e016050a3f50 : SUCCESS in 1h 05m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2cf2d3a8d06449d6ab8c62bea5124655 : SUCCESS in 51m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/86ea6391e5ba4863ba5d5cf088233ba4 : SUCCESS in 2h 08m 17s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a7ffb75385c2468396b61c45df34bf35 : SUCCESS in 48m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/feb3cab012924f9ab5169a099f560e36 : SUCCESS in 4m 53s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f7dd6f5f89b24b70b139169e2dd2bb65 : SUCCESS in 30m 12s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/793627f0ce1d4d12b492065a9dc5b406 : FAILURE in 29m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/9c6c7551b91241e79d6604c3dea282c9 : SUCCESS in 28m 26s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5035d4260e94a2aa0cc2d15ae03faa6 : FAILURE in 54m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/918616c7cd0945f29899c32fd6ee257b : SUCCESS in 51m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/7a6a1bf1ef5840b0a60b78bdd752424e : SUCCESS in 32m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d4623f934ed94ced808c4ab8302e6d63 : SUCCESS in 37m 03s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/f137936b8af34063bb157ac835d43153 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":9},{"id":"e44c95975a74b5a92c90d665c0ec8ab1a2e2374a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:25:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81165f77f85a45408c6b550707b7b598\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6cb7bbb93a34aeca7bb237cda3f1e75 : SUCCESS in 13m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/92c41aa1c6db4bfeb9dbfcde5e53d3de : SUCCESS in 24m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/5575f457f3dc42dea62aa31248211fb9 : SUCCESS in 37m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"31df938c5229c807c726028066a014d8db6e2e1c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 20:01:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6ddf9eddd2bc526c1485b7c32bb5e5c24a711b33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 20:39:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c56b8e4cac3347f99de6ec14e2643587\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b432da909c9044adb3f33ca06f813ced : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c613065acb6649d282f59debf1e07d6e : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/16292b34d5be499cb286bdce328960de : SUCCESS in 35m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"920581f73821fbc125e2b230b924e81e3bf001d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 22:01:12.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bba9958655a4e37ae7856f017b044c7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad7c606a5a1441a4a0218a2f345588f0 : SUCCESS in 1h 09m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d5e31cace8564bccbeae008f1f652008 : SUCCESS in 1h 36m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eff2107ee6404155a5f8b42f1fab447f : SUCCESS in 43m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adf6539d8a4749e3880269edce3e19c1 : FAILURE in 1h 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89233bbc2d954229b5cd3cb1e27f5247 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/886f642790c341bc9a149f5a0ee1f405 : SUCCESS in 11m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf952e91f89b4a3583dc0e95a2d3f5db : SUCCESS in 12m 54s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5c8c3fac008a48b1a1089550267049b5 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a59bfb78a843fd90e4b5fbc3c8f536 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1929c93c2abd433fb91ce4e6d5a597cf : SUCCESS in 8m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a481f8bbbe944fe3bdbdcfa437af2bbd : SUCCESS in 38m 48s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd993b88291a4b9f960f318e377530e9 : SUCCESS in 1h 38m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d53232494e85435cac088b97a9d8a95f : SUCCESS in 49m 41s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6e933d2b2e784b0db5739804c0ccd2d5 : SUCCESS in 35m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9f87058855134340a78d8b27c9c1b458 : SUCCESS in 57m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8329ce3c6cf5461d96979aa1568d6fe7 : SUCCESS in 48m 32s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1acb88d644854018b870866beaeef2c3 : SUCCESS in 41m 50s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b34f2fa283d24dcdb05201c9bfc42aa5 : SUCCESS in 1h 54m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/72493c6957204cf08d70e4d19b253d64 : SUCCESS in 54m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/7840b3b5b4244fdca4e099042c982df2 : SUCCESS in 54m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/381c30b0ef684dd18dfcc30f91f93e47 : SUCCESS in 6m 02s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1826870ab53e45a28af277a5d4d40a8a : SUCCESS in 28m 28s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/3a528b7a28624f2aaa9c020b81b6f77b : SUCCESS in 21m 29s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d31352adf4ac4e609a1cad6b159fbaf4 : SUCCESS in 39m 45s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebc06be9fab84d608a8d29698d51b867 : SUCCESS in 1h 25m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c01d6f6dd44aefbdaf8b069c5d333c : SUCCESS in 26m 56s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0407a3ada3294c63b6819ac6d04b2de9 : SUCCESS in 25m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e8d14936ec2b477fb97d573911fbfc24 : SUCCESS in 32m 32s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4a3c6125c1b5429496c99622d64f6bbe : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":9},{"id":"a5c7dd00e733f96c7463d352e59db908e8ec7632","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-13 22:08:48.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-svxxv/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"3e25ef15e80a0b3d3cd59dd2b8888fe4227255d4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-13 22:42:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b2fd5fe9f92c3c3c03df39551b1afbb6289d6a7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 23:08:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a73e363fba3470f9e5194888bbe0317\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8c1203d69684300a0a28be5bbf9be08 : SUCCESS in 22m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/29a3d0fb14c24095ab453fd290ee823b : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0c48b1b01d8b439da7e0090e8eef20ea : SUCCESS in 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7e4c8e2447ff9821c7c1c7e4fcbf58930e5fccba","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-13 23:12:46.000000000","message":"Patch Set 9:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-dr9sl/index.html\u003e","accounts_in_message":[],"_revision_number":9},{"id":"ccc0bc9b4f3518c8c36b60758a1262f53ea72fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 00:16:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db853e25179b42389d942a6d45b32292\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9096765b832746cfad0866e144d92260 : SUCCESS in 38m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e1d3698f8dc04476b446ce58a765c380 : SUCCESS in 1h 00m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8f649dfdfcf44148136f2fe9f653b03 : SUCCESS in 33m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c8d6146d5ef4e71b9406ef0469aab00 : SUCCESS in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba8aacd96664c8899e175bba515a0ba : SUCCESS in 9m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5403f12c34f45a4af080fdc29c57082 : SUCCESS in 12m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1820bef4f71b425785c3d365d2074ec7 : SUCCESS in 13m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/0a4896e9050344ac83ae86488993cca6 : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a6437d85aea48deb98c4abbd0284764 : SUCCESS in 13m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ecca4f4ed6b4fc1a86ff5db9902c6ad : SUCCESS in 8m 28s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a3d7ecf528054866aa37ae2833a84053 : SUCCESS in 56m 08s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bcae254537a743408a89f16fc12fc58f : SUCCESS in 1h 25m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c4ef0be3fa7446fbba2547e111779f4d : SUCCESS in 22m 23s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/ee87e37f4c4444af82dd627f9465ffa8 : SUCCESS in 36m 14s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/eec53cbb191e416a9c82d0dceecb9b68 : SUCCESS in 44m 42s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5ec838b6c14346e7ac3b01e64f89f97a : SUCCESS in 48m 52s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/94ede5e2468d4cf186ba04dfd307cbf9 : SUCCESS in 44m 57s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4c2c52bc634d4857ac7c8af985f9d189 : SUCCESS in 57m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/8b1c8e8881c34cf29500f235775a16d0 : SUCCESS in 1h 30m 57s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/6c6aab5a21454d8ea8d302549a86ba58 : SUCCESS in 33m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d00338dd1cc94099904e8c748004c4f1 : SUCCESS in 4m 50s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16238e1934d14c9486e972fdd358f18a : SUCCESS in 23m 02s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b845e3c18820496f8a7998383f573f90 : SUCCESS in 25m 07s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e59ca6fd2d0e40bebeaa53c0aa0b467d : SUCCESS in 34m 10s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f93ba2a5fd5495faa9701b20188e12f : SUCCESS in 1h 21m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34bab7f28a1a476ba357e961384929b8 : SUCCESS in 54m 01s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/fb2a106613084318a6640bb52ee05aca : SUCCESS in 53m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6030dcebd2ad4aa2aaa79371e9473c8c : SUCCESS in 32m 47s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4fa4e54729074dff8b9633d367c6608d : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":9},{"id":"f7277897edd6431dfab433e4c68fa86861687458","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-14 00:25:58.000000000","message":"Patch Set 9:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":9},{"id":"14bf20ba6a478d8931da3df92b1928a70b629224","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-14 09:24:17.000000000","message":"Patch Set 9: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b35cff73157656e4c532093be0d00b49861da713","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 13:38:21.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"cd50ea30c68bec40183306a54f18fdef40522b0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 13:38:28.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"65d7515f2cbc42d281d59752cf6e7e1e7f643a43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 13:39:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab08a8bd8ff42b4b34d6ca2950d1375\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e77d4cfcacd446a9244d2a47fe77e92 : NODE_FAILURE Node(set) request eff95a0684fc4775802da937c0ff6a9b failed in 0s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fe70f167ff734a8cb3cfbefe89b7838a : NODE_FAILURE Node(set) request dea4c4dd3cc549118aac6c08ad38c83d failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/be834e77c0b44d96a0da4e4b482c6844 : NODE_FAILURE Node(set) request 466a6425e9ea442081a24a01f7a014dc failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5c7e718094195317e49c33d5ff3dad3d49702646","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-14 14:01:34.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"da09e045da614e5be8dbe951589f0df40f618bbc","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-14 14:10:26.000000000","message":"Patch Set 10:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-zrtmf/index.html\u003e","accounts_in_message":[],"_revision_number":10},{"id":"ef697d7c437947e878b1cd346ed768ecde9c2363","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 14:36:21.000000000","message":"Patch Set 10:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":10},{"id":"de2659a2ee13f67b2a90b68909d376457ddf6b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 15:46:14.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e58e70d1e0f4c1fba391a7b20d9b3f0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/285f86b49e13428f8c07950446c14698 : SUCCESS in 36m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d2ab4bde845048e4bff7c5b3e96843ec : SUCCESS in 1h 52m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ffea7889cad4690825c555b163fcaf5 : SUCCESS in 59m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39c6bdddf6a046eaa2cb92a607b9f7df : FAILURE in 1h 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7fcbf5c6c7e480bb3517b4c1b4355cf : SUCCESS in 7m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc0fef4c21ae41c3b7f64f81f5f9a3f0 : SUCCESS in 13m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d2e81df5771c44448326ee7b89694352 : SUCCESS in 10m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cb5aab69319d4696804d1b0c02edec37 : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6464ccb4639043dcb81b4f8e7dab91b1 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1fbacb1a65b488a94519ce60b798e4f : SUCCESS in 7m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bd0947e900c1413d96ca642e98b65056 : SUCCESS in 1h 00m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d91086fb4e3541619cad7966a8fc0e45 : SUCCESS in 1h 35m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7df6f60d38e44b048e31629e60b35b8b : SUCCESS in 29m 21s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/4d94c284e63440aea48d9d74747d1ce2 : SUCCESS in 38m 28s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/5903c26f9bb1403087a03a2bba2339d9 : SUCCESS in 40m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a374db3517304a71a162f0e5fdd9faa0 : SUCCESS in 28m 42s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/ea68e2cad7194667aaf4468c20b6d669 : SUCCESS in 1h 06m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a369f482fbbd4b6eb87946457a1b6e3e : SUCCESS in 1h 42m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/5be41b5c9044406690388eadf06a7e6f : SUCCESS in 2h 01m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b9d0ee3df5894524a38cf55740ad8589 : SUCCESS in 26m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7674435d8d74476cb4de92f2297e15ac : SUCCESS in 3m 43s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/38fc8f6866344882a11a418b7bf21f44 : SUCCESS in 27m 54s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/42a97bf54b9d489b81eb8e4d5527ef9d : SUCCESS in 26m 51s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/b744af7263484d5aa229da3587ff75f3 : SUCCESS in 31m 52s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ddbe68ae89e54547a88cc98ae568ac1f : SUCCESS in 51m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8ffa61e3dcd49d18f8e2947ec8cdf82 : SUCCESS in 36m 13s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/58d072ce696946ce82387095af32182f : SUCCESS in 53m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c001af95a47e45e1bd02516d7c11e3a1 : SUCCESS in 36m 00s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/68026889f6e146b892729375d7ac329d : SUCCESS in 14m 01s","accounts_in_message":[],"_revision_number":10},{"id":"42d4dadb77f23144fc0983b75218684bc873c720","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-14 17:33:06.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d6aeac1298fea5123622b1223e0a1c9c98280614","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 18:05:34.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"caefffb2f6a2eab6437e58869cf395fdbdbf6e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-14 18:29:28.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"07f98d05e8d3b5e1d1389334cf9d7378dd3125b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 19:12:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eac9399c281d43288c1f425a82533df6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c0afae8c5244e61b86101c751fe46a3 : SUCCESS in 17m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ce972bdef23f40ce90c9d514809249d3 : SUCCESS in 25m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/98062980146c4d1d8651d7c478adb02c : SUCCESS in 40m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9ee5a9603274ccf2340ffdf23cd7861be3826846","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-14 19:25:15.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":11},{"id":"068af962082bec6da34175deac7880e01c2bd111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-14 20:09:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b168911575f4ff09d0ba63be4be3f1e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf246b83c3074f5d94afd97746158823 : SUCCESS in 29m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/66b61a4adaee434c8067f6e68f98af29 : SUCCESS in 1h 05m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3ca3534e3d64483810baad20a9a4c4d : SUCCESS in 1h 01m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c106ccd9fd5a4b69a7ae62b561abb98e : SUCCESS in 53m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c14eb0fd862406096a5f07369cf870d : SUCCESS in 7m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a1c49272c014b95884933446de55988 : SUCCESS in 14m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/253b4ba68e7a4d2fad8957617cdf177b : SUCCESS in 12m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aae30506c31f481b955342685271c403 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63caa8ace92a4e3c99e885e6ab4359c9 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbb27587804b4fa7a8c1d4b8b34a7392 : SUCCESS in 8m 20s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ed549c92a34648b2a3ace67b611474d7 : SUCCESS in 32m 35s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd614204082544f28372738a981d750f : SUCCESS in 1h 04m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c70779d27b324c819693cc8f5d49e6b3 : SUCCESS in 51m 02s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/f02c44479def47d5a78537572c51746c : SUCCESS in 35m 31s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/6ba62a0cc0a9454ba66f66a6955a1126 : SUCCESS in 37m 19s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/eb6ea46cd9174532a4a6913acbcfdea2 : SUCCESS in 43m 42s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/610eab312c4d40508830e11bafcdebb8 : SUCCESS in 54m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/70b3974fb33241859b2964abbb08d2ae : SUCCESS in 1h 28m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/a27fe17f130b43f9b3b6458c4a013f7d : SUCCESS in 1h 36m 40s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/1409ede480a84fefb915a231f105fc51 : SUCCESS in 45m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c694a8617a924b3f97732e4485d2ece0 : SUCCESS in 5m 05s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5cba4894ad0848269a3406b965ed3ad7 : SUCCESS in 27m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ce753c72103047e78fe135c7db3f6bf1 : SUCCESS in 26m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/559a1845bd4147c1a3d1321f20cfee69 : SUCCESS in 34m 47s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e39ea9fcd8ad482c91fdd23d6ba9737a : SUCCESS in 1h 24m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe52f32ddaae42bd834ee6a1b00b23c5 : SUCCESS in 1h 04m 35s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/018f3c6353dc441db24ab13a54d8607e : SUCCESS in 50m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ec7d14598e89486481cedce9efe75c56 : SUCCESS in 19m 08s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7d02eb2b8ec64b40ab9d54128d34fc32 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":11},{"id":"edac014f51e022ffd1fc19b38591a0d20e4fc750","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-14 20:19:19.000000000","message":"Patch Set 11:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":11},{"id":"85e7263a5601b3888686592151ae622873f50d3a","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-14 20:25:08.000000000","message":"Patch Set 11:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-dfqjr/index.html\u003e","accounts_in_message":[],"_revision_number":11},{"id":"a0fb3f7bacf0a405ba065dade260e2afa19daa4f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-17 09:06:01.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4ac86c492797d6b15de89abe195bcad9c59b458a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-18 13:46:50.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d27f609f7dc4277c8adfc509a66f155d1c82efce","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-18 15:04:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3ab0f6356455d5165d4cb1876ba98ef1bf0d32c9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-18 19:07:22.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f113a2120af1203014c2563e315c991cb1399c5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-18 21:34:50.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":12},{"id":"e2ba93413b5a881eff849675b025232a5b2fef9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 22:03:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8674d5b1b1614021b96f352e72bdc770\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a334e598ab4d4ef9bbebb686c975af81 : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f5027721d4c464d82a4be0f7d90f15f : SUCCESS in 21m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/8b3a4a3066ae40b7b0cd54ebe92b9d44 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"aa79a02a0b6cd0c1f323818697031a176002195d","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 22:05:22.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-tsgmg/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"79dc5a7a0672dc465a6c01f1e64b30abef51c01c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-18 22:29:54.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3 .","accounts_in_message":[],"_revision_number":12},{"id":"c95316005eb02481b6d957436cfdc1854a11ed44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 23:28:03.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03bf802ba8644de5b02a1f6dfb200ce0\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d7086b42f4014695afc2a709f60a6663 : SUCCESS in 53m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3a8e728c87d04412a745d102740ae730 : SUCCESS in 49m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04e35240901c49e8971a8c908036c7ad : SUCCESS in 59m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae0a49ad6bf04bffa14f92dbd7d38ef3 : SUCCESS in 58m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd32d0a406ab4b40bb78fb9c602c3bdf : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10f8fcaa70d449cb8c2db55f90b95a5b : SUCCESS in 10m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/198a68b60492475785eea19c66efd25a : SUCCESS in 12m 32s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ea5de951eae64c8cac7f12b76d554c88 : SUCCESS in 18m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bb2ea1e8144fd6ba14112f1f0bde2b : SUCCESS in 14m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e725aa8f8d142aab452b03c00c9e582 : SUCCESS in 7m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/4441980b3204494da3425868ab9d89bf : SUCCESS in 27m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5426703eb0964f2ea41e74b83acc779b : SUCCESS in 43m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1d60f6ad7e284237b80bfcd705e56ba1 : SUCCESS in 52m 08s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8cc3e1dbf597470082e63c05fd27c9cf : SUCCESS in 34m 00s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/63b7164520e1419280084d270ccd9e2e : SUCCESS in 46m 49s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e4c464dd2b904b3880bd76b3fb51d1dd : SUCCESS in 41m 20s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/44c4f92be8734d88813ffa39f1e9e054 : SUCCESS in 43m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ef11d74d60c0448c85e20ff9c9b11304 : SUCCESS in 1h 46m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/00a68c5d1f2d4e588dbbf90f11f734fd : SUCCESS in 1h 49m 54s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/90d84c9d2375449c94e89ee1ca584ee3 : SUCCESS in 30m 05s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4e5270fa8bae43b2b6cd721c13964b65 : SUCCESS in 6m 03s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/85b299c7baa54eb9884dc021ab413c92 : SUCCESS in 24m 26s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/50b853810b1a452dacf1eee90a550ee2 : SUCCESS in 26m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/4db6043bc5fb47fd867f81c9820769f3 : SUCCESS in 32m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff05c7df565c4e29b7d0e6971999f945 : SUCCESS in 1h 19m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f4495cd410f42308d12395de78c8a90 : SUCCESS in 30m 59s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/5c45e7cae30a42bc93df060cf143a5c6 : SUCCESS in 44m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6935a94178c14cdb87464a5fdfeb234b : SUCCESS in 31m 36s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4f7e226b30dc4767afd54855af52df5a : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":12},{"id":"c42f85e7fe7f780e4f407b5536e0e7610027b6a1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 23:38:07.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"aa8aae2a37757b05be586b555bf31b3a0e231782","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-19 07:51:43.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"013e658701ed8c26b2bdc141c5149fca8e452834","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-08-19 14:14:44.000000000","message":"Patch Set 12: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"27ca98e001ce472a8b80c023cdb44ecf3c5906f9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-19 14:25:41.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"09c1ce5ed64c14ef3ba25822c6e5d6e4a03f1f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 14:26:39.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"242b4a1cc04b96e4dcbc036f2539689abdd89667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 16:55:29.000000000","message":"Patch Set 12: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b66f1983dd44442b91c13fd60aa4487d\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2445ab94bb5542d4a40e8e0f3fe929e6 : SUCCESS in 59m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4ecd9037a42c48f4be650a7fcbeb0596 : SUCCESS in 50m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92e3d20d1ade4e879623edb3f539e38d : SUCCESS in 49m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3236b36a014cc09c3620577aa0782e : SUCCESS in 7m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86cc6e652f79478582ced306e6d289d7 : SUCCESS in 14m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/03e82452dbe44496ad14c33f1134d92a : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a14fd8f7b5741759561847a85ef8a4b : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cd7fc9378ee47f48c3be01dac0bc6aa : SUCCESS in 8m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/354faad54a06420aa502bddc61e64344 : SUCCESS in 36m 14s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ba83a1db33b14fb8b71c2bda5856047c : SUCCESS in 24m 11s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/96af7a882c7b4fa08696e51622782eb1 : SUCCESS in 27m 16s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/07c0f0aecebb42fc99ec209ad6a9df3e : FAILURE in 50m 26s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0db74713606a43c6b224c23a4f76ad67 : SUCCESS in 1h 49m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/be2af597bdc34a2c8849c26c773ac4f7 : SUCCESS in 2h 03m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3f371f2478664dbe864a67bdc7632d0d : SUCCESS in 5m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a44221b663c46aaa47b1b617ec1b042 : SUCCESS in 1h 07m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/94fc996cab084c798104e58e72367a72 : SUCCESS in 1h 33m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20abfff59aa5461e96aa0bc71292108e : SUCCESS in 45m 01s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fc846356733d44bba508d4bdd2299bfb : SUCCESS in 12m 35s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b476d0daeee04f7e80c6de9b4a1ca409 : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":12},{"id":"72524c6d20606cafcf150455551a4e900d4bcd9b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 17:51:42.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"5c8a517cc3d69269be394c839e1a2ade1bbe98b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 19:15:47.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c41b2e25968c919b8bbb7db21e555b27cd26fada","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-19 19:47:00.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-56vnl/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"ae157224f2f7bbd5e03fb6b328eeaa4b2c027f6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 19:59:43.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30b1be1971544f68d71a31e759ea5ba\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f5d7cdfad0949b38a5924cab8fa7162 : FAILURE in 5m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0b34a3369e0b4b0eba2c650cb4a6b404 : FAILURE in 4m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b91f465eb9fc4014a240a78a150dce6f : SUCCESS in 30m 58s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"375e9ee0eed01a8ef8e294956e4942cd6e0d9321","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-19 21:00:26.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ed25bf2590d9f7a36a72940f62668f284f546e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 21:34:29.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/405fcd1204644f00bf5f1784a59b8000\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87e8f6457a284f90b296d0fc183d6280 : SUCCESS in 29m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/bc347d1e14274e9495780fa1410f7e62 : SUCCESS in 52m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f1dd07774ee4699b0d27e9ca70b4785 : SUCCESS in 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91929ea7bec34e9c9bafa9b851eb54bb : SUCCESS in 1h 00m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7498cfb6e79b485c907dc0153a928ed8 : SUCCESS in 8m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/304daae651e54202852ba24ee1519a0f : SUCCESS in 13m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/adcfe7d8c5534ec29b4158be6ee2ee4a : SUCCESS in 10m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9e2c14fbe3d44210a14be54f62dafd54 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d24466cd32c46ee873f89827dc4e209 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784d936f826c45f6a7d3599019a9aad5 : SUCCESS in 3m 42s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7c581565da61495a9f87e8570ad3791b : SUCCESS in 26m 33s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4bb363031254850bfd33afbaeec027d : FAILURE in 54m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5abe8542ade04a7eb4150a4f7ff1b668 : SUCCESS in 51m 01s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e7160a46e9d242be944dc687b0fa6141 : SUCCESS in 38m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/75f1967c72f24f3db8245daa7e3beec2 : SUCCESS in 59m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6a419569c112430bbc371dcf42d2d5f5 : POST_FAILURE in 29m 18s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3bb08c4fe5544542aafed2877bac6e0c : SUCCESS in 1h 00m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/67296a0f5afd4561b13889f97d53390e : SUCCESS in 1h 48m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/e1f6cc8e9a9e4778b5a22edeb31d67ae : SUCCESS in 2h 11m 12s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d21c391964fb4fcc93ea63e0fd09ecc4 : SUCCESS in 56m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3abf9ffc0bc14a088b96c62042122432 : SUCCESS in 4m 09s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0807e331a33a41279ce2ecfcc99c8091 : SUCCESS in 21m 52s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/8f5f73e9401043ba904e1442273fa252 : SUCCESS in 27m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/1736d6a661cb4f129a309c9b5bd766fc : SUCCESS in 27m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b45e0a9fffbb4929a5904cef10d6b31d : SUCCESS in 1h 42m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec9217300b864b1abdf6917348c3b96c : SUCCESS in 35m 12s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/bab3463f918d4167b379645c4d8573c4 : SUCCESS in 42m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9f8d4817f18a4e29ae08b0dcbb415d2e : SUCCESS in 14m 47s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/20d1b3891be64253b6255ab6610bd019 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":12},{"id":"f78fa33defbbbabbb0caee442a63478fcf51b641","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 04:11:11.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"672f0610848f645ae64158fb1742d4c4100dfffd","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-20 04:50:57.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-2ftsh/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"fdc7e47e881c2ed2e79830326b3de5025b5b27a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 07:14:02.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23948510abea4f4391cd4726a55ea1c7\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/725c3ca855744774aa9093eae3dc9e46 : SUCCESS in 34m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1430fd411df943ee94ffd85ddcfc72ee : SUCCESS in 54m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ece0aa63a9d2411986c409c0cb92d71d : SUCCESS in 30m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27e84721226747718880a88d256501e8 : SUCCESS in 56m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06be337ae9648a395110b0585982ea9 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/180f24917b02457f918af03135cab3cc : SUCCESS in 9m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2adb748d0574c15b96416140a1813a1 : SUCCESS in 9m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8c40f40e7ee04e6d9d098654c4d90ebf : SUCCESS in 18m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6aec9983114a62b71234bc9bdb791f : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55b810ce8f634decb278ff8c3bafa785 : SUCCESS in 4m 47s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/b6a112b82abb4f74a335bf89667c018a : SUCCESS in 30m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef0f111b4aac472bbad965630f744e57 : FAILURE in 46m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8f03134198a94d8c8356ee1eeef7e743 : SUCCESS in 26m 31s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/e1f6a8493c644977bfb84907892161ca : SUCCESS in 36m 47s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/cc8253af65ef46e7b472cf398c4650ae : SUCCESS in 39m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dfec27eff0c440aca55c3f12d118b1bb : POST_FAILURE in 31m 54s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a0a902a5caf94f96acad40d8ff723074 : SUCCESS in 1h 02m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3af944a3dbe8422c8e8b2d005337bbdd : SUCCESS in 1h 37m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/c5b13a8275ea48508b727a34f90999a5 : SUCCESS in 1h 27m 49s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d0b271d6623b4217a77bff3647ae358d : SUCCESS in 49m 31s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/077de0c9abfa406ca08db5941fe703dc : SUCCESS in 4m 47s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f1167772ef6448cd960928a509ae6c41 : SUCCESS in 29m 23s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7a2cb66506e8493aad9fdac93568150f : SUCCESS in 23m 50s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/989b2d2492f648acaacdd7af20ec2bea : SUCCESS in 25m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/93967fbb444b4751a6687e1db4897e0f : SUCCESS in 1h 35m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac26400506604a30ba7a362b71b80535 : SUCCESS in 49m 09s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1768ec8e058140fbbc0797918e0915c1 : SUCCESS in 54m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e14a624963874794971708573ccb1aae : SUCCESS in 34m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/7f124a7c812e4493a9a7ef02da4eea17 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":12},{"id":"44be2d50db39157412d72e520ce022d7f1e0e29f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 07:28:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5843d8888b214bdaac91224602e0460a\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a8e369c8ec894647870a8c958582cef7 : SUCCESS in 24m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7e85dbaf652e44138b6237e4ec0c5ee1 : NODE_FAILURE Node(set) request 86db8b43077046379a942ad6795c2b1c failed in 0s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/15db77938f844e9d817d8c1d99953aca : SUCCESS in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3fbdbb6bfe4b933573842712fa7c90f74fe481de","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 14:51:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ed03ac1a1d4bd14b751e4e6480e7cbb02ec98bb3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-20 15:24:03.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-h2fhl/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"d679693bce2736e9c88bb9cb94e962421d484637","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 15:41:01.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f6ba6334d92438898ed6016e22434d0\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c5bb90ec164e44ffaf7a0b09724136ea : FAILURE in 17m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/388ac569cd4e4952939777c4ba3315ab : SUCCESS in 34m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3e7f2962926e475c811ec68b32e9cfda : NODE_FAILURE Node(set) request c15cb690edc24d7ab1462236f3c98c18 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"48cdf78805cb70845aa4cf0ad7fa13bf261a59a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 17:04:02.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73e69c84b3ae49ba8e020f26e778132f\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b49331bbd864d9d8bddd91a9928616f : SUCCESS in 1h 02m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/669da2d1fc4b4912af49c101dc60306c : SUCCESS in 53m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e4bf31f41e741368753e47769e14cdc : SUCCESS in 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3aaa04910b5348a0b185f2ba1de2b6af : SUCCESS in 1h 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0426b2eea4f34da89067cb02633851f0 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee7522e47cac404e87ec76c0fec10377 : SUCCESS in 15m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a55c3776c124ed59a40daed9d56df55 : SUCCESS in 14m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/704900dd4bb0436895a93d52e27164be : SUCCESS in 18m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e80f07a2331435291ea251141e01fa1 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8099900db0094ff3989dd28dd7339cbe : SUCCESS in 6m 31s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a5278cfae6db452fac44c5a2cb64f255 : SUCCESS in 31m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e5c600fe0aa347a29d0796f3eaeff4c2 : SUCCESS in 1h 56m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5c86cdced22b4e4dba3f9cbfea0a7453 : SUCCESS in 51m 07s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9841d0dd0fb04c94aa5e5c85fa8de02a : SUCCESS in 40m 56s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/f22e3493959941de8d22f1965b1c3b3e : SUCCESS in 1h 04m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a9db3853a9624a5cbfc331155509b400 : SUCCESS in 36m 24s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/c337af3a1e1b4df2bc0f1d93028592c6 : SUCCESS in 42m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b186ca21b44c4a40927463ccb6c48fae : SUCCESS in 54m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/a51f3acb89c3459dba8dfaafce6501c9 : SUCCESS in 2h 02m 33s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/22013ab570ff41e19368cb9ee7e786b7 : FAILURE in 49m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/0f27b42a704441f9bdc32ee1e94ba28a : SUCCESS in 5m 59s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bc0277b4cae9415aaadd0bf62c3224d1 : SUCCESS in 30m 58s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/56ead18505d941c78aef8327df2f4d46 : SUCCESS in 29m 24s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d5d9d7db396b45038e0eaa8b8a890d50 : SUCCESS in 35m 40s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/664d565eddfd4c09b096fe62f4167842 : SUCCESS in 1h 12m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3ef8f0ee59a40619d77c4159a9c0c2a : SUCCESS in 30m 08s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/de15313817ec43a581518139c8a2e642 : SUCCESS in 52m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6ed43b3e016f44ceb03391708d22b479 : SUCCESS in 33m 00s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/dcc204447cf1462bb3533965f280b61d : SUCCESS in 12m 17s","accounts_in_message":[],"_revision_number":12},{"id":"ec5732c54f7c1c4356d868d0746c03193aa822b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 17:10:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"30cf7fc94b05d3157b90c91ed8f028006ebaac94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 17:52:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feb2b77bccac4fdc98e68734dc847e75\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/452ff44fdbfb45df8cbb7e948eaf461e : SUCCESS in 20m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3152c79f015467aa50741054f0c9c50 : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/13c89e917b694c838d73e1cc3b660164 : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"7bf20e0f37508e31fbac93b79a88ee5f3fd27b1b","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-20 17:55:44.000000000","message":"Patch Set 12:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998571-tjn2p/index.html\u003e","accounts_in_message":[],"_revision_number":12},{"id":"7b84c7d3b88c7d7adc0ae84f2cdf9f2dc14db74a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 18:06:40.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"61a4e91e1c3a2b7dac76eee4694d07589b952206","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-20 18:33:38.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"00ede3867dfb9b01189480576c500256ce4ca517","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 18:59:23.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144a190d179c4a88a09b5b7d75b1fb49\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73368282c44748dc83f2bc087bafda04 : SUCCESS in 1h 01m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4dc0a759c3ad48ae8fc96bcc584e6b17 : SUCCESS in 1h 42m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad6bf2330af44026b3735e7f294c21d2 : SUCCESS in 47m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/734c8ed899df4ffbb14ebe6ff20e7724 : SUCCESS in 1h 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67e7e73de5654cdda68e2c299908e231 : SUCCESS in 8m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8249510f42a54cd8b833a2ef81f75cfb : SUCCESS in 13m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08677d14f65344ea99d98e67d34bfff9 : SUCCESS in 13m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ef8453d1c4ff414b8f4cc8d2a9ac7767 : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8fbcb679be4ceab8315fd0a51079e7 : SUCCESS in 16m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28765046921e4db28a11407e83b558ce : SUCCESS in 4m 35s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/9b2a4847af2b4d0ebb1f32f301ab3296 : SUCCESS in 28m 16s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21f81cb6921245a48e964a2a55d5fa0d : SUCCESS in 50m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fb039fcf2d0e4b8ea00bdbe3c400d44e : SUCCESS in 57m 12s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/8ab45ec273b049308356500fae0f7853 : SUCCESS in 36m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/39080b130c7b411eb65f69fca5d27ea5 : SUCCESS in 1h 01m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/88e32fcec1cd476d9eccaea15097ca80 : SUCCESS in 55m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1707aa19e0c343b285bb05d49ce40e80 : SUCCESS in 1h 10m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/965220c4519249cab889921c248776b2 : SUCCESS in 49m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/a4689339fb614f58892da4b23b4808ba : SUCCESS in 1h 01m 02s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/68eb96417e0f4e4f976bd25eccb9761d : SUCCESS in 50m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/be3b019fb61a41c394bff6003b536864 : SUCCESS in 5m 19s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b664e532ae434fa390be46da1625d858 : SUCCESS in 28m 46s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/65e85b09965f4c09a47f5fa9781e0de2 : SUCCESS in 26m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/dfc5fa6c887649a78d663186e1a75dd7 : SUCCESS in 28m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1bca0fb214e4fdba2584d1b9ddcff4d : SUCCESS in 1h 26m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/099bc4e31f0743a38af8a1420cc465bf : SUCCESS in 56m 17s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0cfbcecb61804911800bce526399b5e4 : SUCCESS in 42m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2ab92eb597fc4acaadb55e55f90a3cbf : SUCCESS in 41m 17s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/841f307cefb74ee9af6cf69011f0d999 : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":12},{"id":"20b60a140d986c9f7aebf481af3ebab548a61491","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 19:00:04.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e08dcd83b997cbcbfa999701101891b8b76f9676","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-20 19:10:16.000000000","message":"Patch Set 12:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":12},{"id":"c1276f99e88b95eb84f474da1af2461564a4d1f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 21:11:11.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c802508ab0a64ea78d81f97765cbdcd1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c1f122ec864a4ff195916bf89b60c422 : SUCCESS in 58m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a245df5d4aa540fbbaad96ec91aeb1f9 : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dde8fa65704467a887b72c07ef10437 : SUCCESS in 57m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22683578af014ec7a46b87a4f4bd9a06 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54b9f42149854fc7ac2c818aab7e318c : SUCCESS in 10m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/66802ba578b641ab9ac3e3b29e663329 : SUCCESS in 13m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d25c7e74e9a41daa92fdd58266f9a7e : SUCCESS in 13m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d52a3e91ad614251a7c1eafcd43b1108 : SUCCESS in 6m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b08db0dc65354e74bbee307a808d5fd3 : SUCCESS in 33m 59s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c04b86fd829047608b4fa6145b5e6ced : SUCCESS in 29m 06s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/40b7a77ed27b432a93f5e8b73084daa2 : SUCCESS in 24m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e83128da8aee4916ac1c2f9483748074 : SUCCESS in 36m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/69e6d0eff57849f3be8d355fa99e2d74 : SUCCESS in 1h 49m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/a1991c039fde4d91908cb19e1cfb255d : SUCCESS in 2h 04m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5b465dcac9694046b2b0db3b00b4167f : SUCCESS in 4m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c690b90e32a43638b70f67804bf58f4 : SUCCESS in 1h 33m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e830fac217ec4701bd42444a6cbabf4a : SUCCESS in 1h 24m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af2ea29a87054715a2eb27a48095dc73 : SUCCESS in 55m 45s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/1bdd7efb675642ebb58253eb9958a84c : SUCCESS in 9m 29s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/8741963dc056420bb53ed6c47ccf6458 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":12},{"id":"2c929ce10c822a6441fc23d491d18df0653809c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 21:11:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"14f4ffffb5ac14d7c913886a476db2b1f56eeb6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 21:12:21.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce33d431352a45ada2f28013636b6b85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64e4038c503a40329979abceae50b4e9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0842edaa998d12e77d622dae5aedfb5e0024c16a","revisions":{"ab9c851304095087f650d66cf59b864d67076f7c":{"kind":"REWORK","_number":1,"created":"2026-07-24 00:52:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/1"}}},"commit":{"parents":[{"commit":"185c59d3452585261c4704ecd774238d96648d60","subject":"Task tracking mechanism for graceful shutdown","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/185c59d3452585261c4704ecd774238d96648d60"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","date":"2026-07-09 02:31:57.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-24 00:49:07.000000000","tz":-420},"subject":"Shutdown thread pool executors during service shutdown","message":"Shutdown thread pool executors during service shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done at the end of\neverything else done during shutdown and with wait\u003dfalse so that it\nwill let in-progress tasks to complete and will not block shutdown\nif any task still running on any executor.\n\nAdding a global list variable so that all executors can be tracked in\nthat and can be used to shutdown all of them.\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab9c851304095087f650d66cf59b864d67076f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab9c851304095087f650d66cf59b864d67076f7c"}]},"branch":"refs/heads/master"},"064d1f5dfac8ec2b5906115262c92c182748fa0f":{"kind":"REWORK","_number":2,"created":"2026-07-25 04:29:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/2"}}},"commit":{"parents":[{"commit":"1a061441dfda2ac9409c4fc6e5b6b907d324229a","subject":"Call manager graceful_shutdown while RPC server waits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a061441dfda2ac9409c4fc6e5b6b907d324229a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:28:51.000000000","tz":-420},"subject":"Shutdown thread pool executors during service shutdown","message":"Shutdown thread pool executors during service shutdown\n\nWhen service is stopping, we need to shutdown all the thread pool\nexecutors created by that service. That will be done at the end of\neverything else done during shutdown and with wait\u003dfalse so that it\nwill let in-progress tasks to complete and will not block shutdown\nif any task still running on any executor.\n\nAdding a global list variable so that all executors can be tracked in\nthat and can be used to shutdown all of them.\n\nPartial implement blueprint nova-services-graceful-shutdown-part2-task-tracking\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/064d1f5dfac8ec2b5906115262c92c182748fa0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/064d1f5dfac8ec2b5906115262c92c182748fa0f"}]},"branch":"refs/heads/master"},"2ac30f76a2b480f2945fd1b8d4f32d6f1e62949d":{"kind":"REWORK","_number":3,"created":"2026-08-08 19:06:07.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/3"}}},"commit":{"parents":[{"commit":"1a061441dfda2ac9409c4fc6e5b6b907d324229a","subject":"Call manager graceful_shutdown while RPC server waits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a061441dfda2ac9409c4fc6e5b6b907d324229a"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 18:50:20.000000000","tz":-420},"subject":"Unify the thread poool executors into a central factory","message":"Unify the thread poool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ac30f76a2b480f2945fd1b8d4f32d6f1e62949d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ac30f76a2b480f2945fd1b8d4f32d6f1e62949d"}]},"branch":"refs/heads/master"},"2002a7198bebea694d18d0db255990090a26227c":{"kind":"REWORK","_number":4,"created":"2026-08-08 19:36:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/4"}}},"commit":{"parents":[{"commit":"d8646c2c9e2cd1a5302918829e2bb160685b63bd","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8646c2c9e2cd1a5302918829e2bb160685b63bd"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-08 19:30:00.000000000","tz":-420},"subject":"Unify the thread poool executors into a central factory","message":"Unify the thread poool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2002a7198bebea694d18d0db255990090a26227c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2002a7198bebea694d18d0db255990090a26227c"}]},"branch":"refs/heads/master"},"f6a4aff20132ed07ab30cdfc5211da76135dc055":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-08-11 04:14:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/5"}}},"commit":{"parents":[{"commit":"949ffd7617dffc57bb9c4bdbaada8d7e64ce0569","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/949ffd7617dffc57bb9c4bdbaada8d7e64ce0569"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-11 04:14:51.000000000","tz":0},"subject":"Unify the thread poool executors into a central factory","message":"Unify the thread poool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6a4aff20132ed07ab30cdfc5211da76135dc055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f6a4aff20132ed07ab30cdfc5211da76135dc055"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"d8646c2c9e2cd1a5302918829e2bb160685b63bd","ours":"2002a7198bebea694d18d0db255990090a26227c","theirs":"949ffd7617dffc57bb9c4bdbaada8d7e64ce0569","merge_strategy":"recursive","contains_conflicts":false}},"b1b8b6062a695339e37802a1b6ef2aa5771d2ec4":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-11 04:27:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/6"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-11 04:27:20.000000000","tz":0},"subject":"Unify the thread poool executors into a central factory","message":"Unify the thread poool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1b8b6062a695339e37802a1b6ef2aa5771d2ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b1b8b6062a695339e37802a1b6ef2aa5771d2ec4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"949ffd7617dffc57bb9c4bdbaada8d7e64ce0569","ours":"f6a4aff20132ed07ab30cdfc5211da76135dc055","theirs":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","merge_strategy":"recursive","contains_conflicts":false}},"ddfbf9b389a01dac0ea5321ce6b2a6a8ea274c8e":{"kind":"REWORK","_number":7,"created":"2026-08-11 19:37:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/7"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-11 19:37:12.000000000","tz":-420},"subject":"Unify the thread poool executors into a central factory","message":"Unify the thread poool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ddfbf9b389a01dac0ea5321ce6b2a6a8ea274c8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ddfbf9b389a01dac0ea5321ce6b2a6a8ea274c8e"}]},"branch":"refs/heads/master"},"c1bfbd327bbc4bec89e030aa3404b54ddd04f235":{"kind":"REWORK","_number":8,"created":"2026-08-12 05:01:24.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/8"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-12 04:58:12.000000000","tz":-420},"subject":"Unify the thread pool executors into a central factory","message":"Unify the thread pool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- A single method to shutdown and reset all the executors created\nin specific process.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1bfbd327bbc4bec89e030aa3404b54ddd04f235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1bfbd327bbc4bec89e030aa3404b54ddd04f235"}]},"branch":"refs/heads/master"},"af7bbc157227ad984c28328bec1853025bef9a20":{"kind":"REWORK","_number":9,"created":"2026-08-13 04:01:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/9"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-13 04:01:29.000000000","tz":-420},"subject":"Unify the thread pool executors into a central factory","message":"Unify the thread pool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- A single method to shutdown and reset all the executors created\nin specific process.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af7bbc157227ad984c28328bec1853025bef9a20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af7bbc157227ad984c28328bec1853025bef9a20"}]},"branch":"refs/heads/master"},"7c7c68136de2f63de76579fbc6c37bb3ea68333a":{"kind":"REWORK","_number":10,"created":"2026-08-14 13:38:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/10"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-14 13:37:56.000000000","tz":-420},"subject":"Unify the thread pool executors into a central factory","message":"Unify the thread pool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- A single method to shutdown and reset all the executors created\nin specific process.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c7c68136de2f63de76579fbc6c37bb3ea68333a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c7c68136de2f63de76579fbc6c37bb3ea68333a"}]},"branch":"refs/heads/master"},"8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c":{"kind":"REWORK","_number":11,"created":"2026-08-14 18:29:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/11"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-14 18:28:26.000000000","tz":-420},"subject":"Unify the thread pool executors into a central factory","message":"Unify the thread pool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change might looks big in size but most of the test file\nchanges are because of moving things from utils.py to a new\nthread pool factory file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- A single method to shutdown and reset all the executors created\nin specific process.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8af9f3bbf9440fdeab2c9b6a31f6e490aeb6087c"}]},"branch":"refs/heads/master"},"0842edaa998d12e77d622dae5aedfb5e0024c16a":{"kind":"REWORK","_number":12,"created":"2026-08-18 21:34:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/998571/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/71/998571/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/71/998571/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/71/998571/12"}}},"commit":{"parents":[{"commit":"2a745c7e15376b8c4b9581e9f149bb02ef596ebc","subject":"Track and log the main RPC server in-progress task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a745c7e15376b8c4b9581e9f149bb02ef596ebc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-07-25 04:24:20.000000000","tz":-420},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2026-08-18 21:33:42.000000000","tz":-420},"subject":"Unify the thread pool executors into a central factory","message":"Unify the thread pool executors into a central factory\n\nCurrently thread pool executors creation, destroy, get are\nin utils and their singleton-ness is achieved via globals\nvariables which makes them less readable as utils.py is\nbecoming large. Also, there is no way to know what all\nexecutors are created for this service/process which makes\nhard perform single operation on all of them. For example,\nshutdown all executors during graceful shutdown.\n\nTo solve the above issues, we are unifying the executors\ninto a single factory who will be responsible for their\ncreation, tracking, shutdown, and get. This will make sure\nthose are singleton per process and can be shutdown all\ntogether.\n\nThis change is big in size as it moves the things from utils\nto new file. A few key things changed here:\n\n- Replace the per-manager executor attributes (e.g.\nComputeManager._live_migration_executor, _long_task_executor,\n_sync_power_executor, and the scatter-gather/cache-images/default\nexecutors scattered across nova.utils) with a single\nThreadPoolFactory that creates and tracks every executor as\na process-wide singleton, keyed by ExecutorType. In production,\nno change as manager is a single instance per process but our\ntests create multiple managers within same process.\n\n- Gives get_executor()/spawn()/spawn_on() one common entry point\nfor every caller (manager etc) and gives shutdown_all_executors()\na single place to shut all of them. This makes usage easy as\ncaller does not need to worry about executors creation/get when\nthey want to spawn the things.\n\n- spawn_on takes executor type as arg and then create/get the\nexecutors by its own. Callers just need to pass what type of\nexecutors they want to spawn things on.\n\n- pool size which are mostly from config options are also\nmaintained in central place.\n\n- A single method to shutdown and reset all the executors created\nin specific process.\n\n- ComputeManager\u0027s _cleanup_live_migrations_in_pool() no longer\nshutdown its own executor directly, since the live migration\nexecutor is now shared across every compute manager instance in\nthe process rather than owned by one. And RPC server stop will\ntake care that no new live migration is heppning during this\ncleanup.\n\nChange-Id: Ifcb4b4ad1113d742e26d91cbb35e5b92a14878e3\nSigned-off-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0842edaa998d12e77d622dae5aedfb5e0024c16a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
