)]}'
{"id":"openstack%2Fnova~998608","triplet_id":"openstack%2Fnova~master~Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0","project":"openstack/nova","branch":"master","topic":"bp/intel-tdx-libvirt-support","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-08-24 12:56:19.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"38744":{"account":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"last_update":"2026-08-26 12:56:13.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-13 14:58:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-19 08:42:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2026-08-26 12:56:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0","subject":"libvirt: Add memory encryption config for TDX","status":"MERGED","created":"2026-07-24 11:57:47.000000000","updated":"2026-08-28 13:36:31.000000000","submitted":"2026-08-26 12:56:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"998608-bp/intel-tdx-libvirt-support","meta_rev_id":"67f7f78f1a02adb1731214a921e5d71b989b3ba2","_number":998608,"virtual_id_number":998608,"owner":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-26 12:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-08-24 12:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-08-26 08:25:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":1,"date":"2026-08-21 14:56:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-08-22 09:21:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-21 19:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2026-08-22 09:21:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"}],"CC":[{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-24 12:53:42.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-24 13:31:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-07-24 14:35:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-24 16:49:44.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-07-24 17:40:12.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-07-27 12:16:31.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2026-07-30 12:38:19.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-10 13:00:05.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2026-08-18 20:50:08.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2026-08-21 02:03:33.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-08-21 12:35:37.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-26 08:25:51.000000000","updated_by":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"reviewer":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"state":"REVIEWER"}],"messages":[{"id":"f722baccef59cfb41c2a9c06a91702f9abc8fabe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-24 11:57:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5e06a84954a1dc57d49e7c1cc327b23c53a7279","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-24 12:00:37.000000000","message":"Topic set to bp/intel-tdx-libvirt-support","accounts_in_message":[],"_revision_number":1},{"id":"8ccbb82a27f62b9adcd626af96b3ae443c50d550","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 12:53:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":1},{"id":"c7af2f1a5920b08f1cc4b3c00104758cfcf1d404","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 13:31:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32d98c0ab84b4baf93794d2d7349f273\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f306601e5a841499538c780f1bcc3fd : SUCCESS in 20m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a7c91b1c55041a4b031e5d7ea59b0d7 : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4d54f23ee841444e8ab4411e87e22909 : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad1d2a100779ef37a78149c9286cfea8c8a410ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 14:35:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f7d48f48b3d4f118f0c2435d68844f3\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22684dab2dc84815b764395881687ebc : SUCCESS in 1h 12m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e843b54fdb024d37bc8f043ca5f698f6 : SUCCESS in 1h 35m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec1eb08d720143dcbf3320453703632f : SUCCESS in 27m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/470e46c477f84cc39924e052240f6e27 : SUCCESS in 54m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e88ea827e644c8af2e514374f44aec : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96da77bb1efe45ceb7914a7e6d7f3fcb : SUCCESS in 15m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7c17448ddf44e778b4f5b81642802d1 : SUCCESS in 17m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b5a17b86f9004f26ad06d5df0bfa6e33 : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37d1a17d5994a89b5ce1ccad68680ea : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6bd8d235e1b4bfa9c2f28a8db2a26f5 : SUCCESS in 10m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/ed555cd21c8d4d4aac52e0834d12c417 : SUCCESS in 1h 08m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be1caf42c7a340e6b39035c2236a53e4 : SUCCESS in 1h 30m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b34ac0a896814e32a2f4701c43ca526e : SUCCESS in 31m 44s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1c165fc2a6454d8483509133cec71a2d : SUCCESS in 39m 34s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4da09a719c184f2f88e01651c2a74956 : SUCCESS in 1h 00m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/48ed92f6b02a427ba4d5c4719e75926c : SUCCESS in 51m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e5ae5abfe04b4c24863112211170edbb : SUCCESS in 2h 28m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/25e6f19ae2be4dcca9de44e9cf0916c9 : SUCCESS in 1h 08m 11s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/52365ebf68e54a4babae4a80f5d1894a : SUCCESS in 47m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/90507b083dd94d3fbd2dbcc5c0548b97 : SUCCESS in 2m 59s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d403d009dbbc4a80a67924352347ea30 : SUCCESS in 28m 24s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/23a6033075e44c0fb6b6351ffd3e6558 : SUCCESS in 26m 13s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/8358d874ab1d4a138697714df3076b3d : FAILURE in 39m 47s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8e427eb46a5c45fdb3be65e7a1bee0c3 : FAILURE in 1h 19m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f69e327e80ce447894cb46eb06f4ad45 : SUCCESS in 27m 22s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/1a3c7dd07db64df8bfa6aad763fddf34 : SUCCESS in 52m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/51b70158c74e439cba3ea3e7cc63fe23 : FAILURE in 34m 59s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/bbecbfa60b98496b807558cd2769a223 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":1},{"id":"13e40927b47c4e46cb7792095bba047ea82cc7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-24 15:14:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"ad175e03124d855c98a720f723b9ee92b65a4b7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 16:00:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39583c6ce2d4771abcf4a0ef16635a3\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e392fbe95b154508922116f236296abe : SUCCESS in 22m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a235d1944e5642f8887afb2a310e494c : SUCCESS in 35m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/61f3049de0a5406a945e913b047752b4 : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed3b9c801a5549786a9f6e0f95d1b193ba1bd596","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-24 16:08:39.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":2},{"id":"4a0823a26b8197efbc7d7b071f2a4050639906ed","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-07-24 16:49:44.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ad2b9088218ca2900a10376775fa5f038938295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-24 17:32:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2337dd989c3c491291c037ff40f07842\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/998fa36ac7b841589996005d605ff623 : SUCCESS in 56m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/85b4de857b4e4e92bc8d5f21dda3c0a5 : SUCCESS in 1h 40m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/510b56c8e4454344be235bedfc3a66ac : SUCCESS in 1h 04m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2589c9dcb1f1499ea73b0f8ec44e7d5c : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13be9c17bda140db92061c50f0376e40 : SUCCESS in 7m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11b2c991ab684e349b818bc86de1b4e8 : SUCCESS in 16m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38e4fe44d0c14aaeb0e18a0927028e48 : SUCCESS in 12m 42s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/82f7b5c45b8a469996cf3ea8c1c4f5ea : SUCCESS in 21m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cf9850dd3f141e78d4a25d6c286d737 : SUCCESS in 15m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af6fb8fe436b46809fc6368deff53901 : SUCCESS in 8m 44s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/5fd15f0470ca466e851b62794917f701 : SUCCESS in 1h 04m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa7048e0f1a04313b4a5bdfb0e691939 : SUCCESS in 1h 35m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/58323385bd7e4d429d01877d52e2d982 : SUCCESS in 1h 00m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9c97a0bd3d9b46579395facc1151d7d0 : SUCCESS in 42m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/964d93247c094880b4d50b60d03bbc97 : SUCCESS in 1h 00m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/151149ed56fd4ef3b2bdfa24ddb8c03e : SUCCESS in 48m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/05bfb605822c4d7ba8ad4bd4fcb99460 : SUCCESS in 1h 41m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/f1e9dbde34164fff978f03a8f436c44d : SUCCESS in 2h 11m 18s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/71729be7685541ab8988d1abce5e7d4b : SUCCESS in 49m 16s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3627485807c34190b7d4691333e12dad : SUCCESS in 4m 43s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f8a6b4b59ec6439fb94474e96891b62f : SUCCESS in 27m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/35b00d8a7e2f41ecbf7ef0fa25e4a8a2 : SUCCESS in 26m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/f31e13cdd453453eb4787229f08268f9 : FAILURE in 37m 36s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2491091ac8fc497fb7dbda55adefd80e : SUCCESS in 1h 32m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78a4b44484344890bff063fec8799d14 : SUCCESS in 1h 02m 47s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/9188ce13d00043b3bd61c63739412c4e : SUCCESS in 59m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a73d824a98684483a45ff1f66651f23a : SUCCESS in 38m 03s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/ffe5574238e2443883c5692d38de5470 : SUCCESS in 14m 24s","accounts_in_message":[],"_revision_number":2},{"id":"338ca705a4b5a53934027c9d74f8b51de3ec239d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-24 17:50:07.000000000","message":"Patch Set 2:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":2},{"id":"786b1c94706d7ba5464c10fcb5f7b5183d4259d8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-07-27 12:16:31.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4232cf7fad1391bc3d413bcc75706da76ec25b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-30 09:25:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d14a89189c1cc603d098e036e09de5a192dbfa5b","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-30 09:27:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"92f3699a3a2799e6e3f2dee8507e51a4c1994b17","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-30 09:27:41.000000000","message":"Topic bp/intel-tdx-libvirt-support removed","accounts_in_message":[],"_revision_number":3},{"id":"7282d6787a68a4159910c4342d674a3cb19df6df","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-07-30 09:27:53.000000000","message":"Topic set to bp/intel-tdx-libvirt-support","accounts_in_message":[],"_revision_number":3},{"id":"a337b723324798ce4cd4710b9296453d132f8a2c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-07-30 10:26:23.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":3},{"id":"030c7b19e79ae953b8be98dd2bd846a79bb2e06b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 11:15:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa40b04e19a48b5ad7d9244c7f32d1e\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7d055db270640bd8d3c9960235b4539 : SUCCESS in 1h 00m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fbbb934eb5b34883a0e88e58e24d021a : SUCCESS in 55m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e06a2e28abc04f6c90bd6a218731b466 : SUCCESS in 52m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1611648120f14fa2bb4efe93eba45116 : SUCCESS in 1h 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99f72a56d8d4268a790b89448588f8a : SUCCESS in 10m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e73f2b3ee1db4236bae7b47bb4653fde : SUCCESS in 8m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c47b3b113c2747489401a26d0a4a7efa : SUCCESS in 13m 36s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b065ab5d3cd41528fdafa36fe4b780c : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63bc74b074544353bd2292297db9af01 : SUCCESS in 12m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e2b3259d20148699ea191da33752cde : SUCCESS in 7m 38s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/809e21ae80ed47c198fabd96de3100ae : SUCCESS in 1h 01m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e92367019a504864abab097a8fd3fada : SUCCESS in 1h 16m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6d1a32f4f50b4a459b53079ee0278615 : SUCCESS in 50m 54s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/3cba09048d144d208a793b3432cf1b3b : SUCCESS in 36m 03s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/19c177d9161f43deb1ddfb1ea050676c : SUCCESS in 1h 00m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/57779b5d311547c5a622dbcca7152c97 : SUCCESS in 37m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c9125e5a2fa9479db7be274e9bca634a : SUCCESS in 1h 44m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/30c5b33f1a44454c98ca5d1e009cccc0 : SUCCESS in 1h 37m 59s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d560c0f482f747cba707ebd098672142 : SUCCESS in 47m 20s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/515563e1a8014c5e9991beb0b5562c96 : SUCCESS in 5m 35s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f6e8e2c1b3d948d6b66608102c3b365c : SUCCESS in 30m 42s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/ddd74b85c3114201b1231e8167d63738 : SUCCESS in 21m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/484538d2cb5a44f7a0c23d05642cd892 : SUCCESS in 34m 25s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b68dc0a982e4f3c949f8b4e51dae08e : SUCCESS in 58m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9873cdd0a2ee443fbf3e51bfc467629a : SUCCESS in 58m 22s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/06d56944b3614860858f1393aec0e641 : SUCCESS in 40m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3dab0d2ab4434f29ae8c091b39818295 : SUCCESS in 20m 06s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b1e70ab022d546ca83eb530bfbabb0f8 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9bee531080f4ad467b17ddb381022098ee66baf4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-07-30 11:25:15.000000000","message":"Patch Set 3:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":3},{"id":"5f1fcfd70619ab6a431b8b16d793e893f7639ef8","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-07-30 12:38:19.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-vhrxg/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"f7f3706fea8700355a919b81c9f8a1cdc6d1e68e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 13:22:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1570b75a8e984986ad78103621d4ab44\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5e5032bac281411aad7f1e51aee1b5b5 : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5de82a8439634244b658b8cfef008469 : SUCCESS in 38m 25s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/67b9ff6f4150446e8bd342b87ebb260c : SUCCESS in 35m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd1dea38d6c8a57b3b424a7dc86fe509ed36b7ef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-03 16:22:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9083e90cabcdbff7c602cc3a149cbfe39083d6c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-07 10:51:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6f4e653a7a81e89e7a47f33136fc4b2edb8c12d6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-07 11:24:48.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* quobyteci:  nova_quobyteci : FAILURE, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":4},{"id":"ca2494d5c5cc7213c1308224211f885990db7845","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-07 11:26:27.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-zjhvb/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"a7a970757f7a269f195eccb1ce724895fabd8a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 11:41:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/975d094897d14e35a221ec14fe8aadc5\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/853eff9bba40417583eb8dc1f2d08ef4 : SUCCESS in 31m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/708c0e36cd914615b15873b34573229f : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d249aa809b7b4040867d2918614740b3 : SUCCESS in 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"020faed810bd921bea4c6eccc3c9e0f1488c22fd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-07 12:52:42.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7a61e1d5060942b78dc23893065c3ca2093e3d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 12:56:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5778b5a9bc0b410e89b7abce9f5a54d9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1489a589ad644b47b3fb79d09ec66654 : SUCCESS in 1h 06m 05s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/231da529a0ba4a1abbeedb4658e170b8 : SUCCESS in 1h 42m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1824f9297bf741d99dd92c5e50c92080 : SUCCESS in 1h 06m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99887bbacd864a9d83de3dababbd0825 : SUCCESS in 1h 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d557efba2ff447a0a63a5d74b78a0036 : SUCCESS in 10m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/228c3213b81543e399151b5c68a2dd3d : SUCCESS in 13m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80b2f055a3144c31aef0ae28568cd188 : SUCCESS in 13m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c90d4128054a4189af2b3b674d300cab : SUCCESS in 15m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec50ca877970477eb1083f07e4001770 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0d0864529f64610842a9fbaade1dd5a : SUCCESS in 6m 04s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/487a1947974d4bbd8b52e3255257ac72 : SUCCESS in 1h 03m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5f6e263475b4b94a5dc094cbe5fb22b : SUCCESS in 1h 04m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7ee430b83be5420ab2ead3e303879e16 : SUCCESS in 54m 22s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6efd923e22104d00852f0d55125d0419 : SUCCESS in 34m 30s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/e8f364f4faf64bb8b9301f16930e164d : SUCCESS in 55m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/dd1269a4ea99499b9efa8434c2682f42 : SUCCESS in 41m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/41cf2466097643838007057baf88b7b6 : SUCCESS in 1h 56m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/14d671eaa3b6467f946170602b09a392 : SUCCESS in 1h 50m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/299be961408c45ca99592a13d4809d92 : SUCCESS in 47m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b032bff184e84fd5b7804af7ad8f8aef : SUCCESS in 5m 10s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0d14f1e47262459495f4fed0a181b0bd : SUCCESS in 27m 08s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/521ab7b6fe2f404daa127117ffbc8625 : SUCCESS in 26m 42s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/eae9f9d46c194100a88e38b9ecb77e5e : SUCCESS in 27m 55s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/803cdb3eae4e4b1e9ef2966d339ba10e : SUCCESS in 1h 28m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7eac193496a413186634ebdf773969c : SUCCESS in 36m 10s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/b3a9622145544f22a74ca97a8a1e2cb0 : SUCCESS in 55m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c6fd5f65aaa347759ad6715641254b2a : SUCCESS in 20m 48s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/cb4153c42e4f4d428a0ce8e516152765 : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":4},{"id":"36146a0e06bc74a53e0a63c3cb9628f2a028bc50","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-07 13:06:50.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"f8b320704df96538190a56b8b7d737b10fabc8d6","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-10 13:00:05.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dd9248e77d4fa4821680117142a6ec5cb13c8299","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-10 13:21:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2ed518f416c8d340662aa9d190042b9f4ef1912","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-10 13:30:14.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45b71ae040196225e21289eaf0370e36e0478bc7","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2026-08-10 13:33:40.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2ba1aaa3ec8e4a9d6aa6c57fae6f9bb2dd511172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-13 07:25:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"df93bc067bf39c8dfd891dcd2bd472cf8abe4ab3","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-13 07:59:26.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-jd48d/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"aac34c282fed41466a046bdcb5ecb522be9bbe0f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-13 08:25:06.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":5},{"id":"9b103df4a9f5a1437dc67e47d85d47feb9c31bb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 08:47:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5422e7a949fc458b9082928eabb54d52\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d232c37eac2b4f95968babba0309574e : SUCCESS in 23m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4920ff7a2dcd4af795219bf5fed6c8a9 : SUCCESS in 22m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2e925dda00754878881c033d886efbe7 : SUCCESS in 50m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ee2d4c5657e7724547fc132f103443c9a6db7745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 10:02:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/309aa20af6ae4a24a8c0bc1863ce1aa9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1a52c4d4afa14ebcaa164951067255b2 : SUCCESS in 1h 12m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fb6074d53d864c9395e2ce19f8dd55a3 : SUCCESS in 1h 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bf82819f2fc431bb2b97d7ce1192491 : SUCCESS in 57m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dd858be8d7447398ada72a0fb011150 : SUCCESS in 1h 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a07d12725f244f1bd099c62be93f924 : SUCCESS in 8m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e6aa52dd79a466ca0523da1095aa071 : SUCCESS in 14m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9931d9bb13b540909608d61cf9e65252 : SUCCESS in 13m 30s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b32fcfa8381143889c6775faf8414b3c : SUCCESS in 18m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e6d7efeb0f49fca8b218f458a702b4 : SUCCESS in 12m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bce00a0df5c14a01a22050a8b309b90b : SUCCESS in 9m 34s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/7c55c3d1b6aa40a2b68eb2a844d1a591 : SUCCESS in 1h 06m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06d9c1072e2d47679a7e8ce0ffc258fc : SUCCESS in 1h 40m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c7de5bff60c403d99a22f0fdbc1c73d : SUCCESS in 53m 55s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/a085c53f772f4e08ad0ce851fd766983 : SUCCESS in 40m 42s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/4f8c41914c0441aba46b2606737d90aa : SUCCESS in 1h 01m 43s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a74941612e924df49f9f77ed3e03a92c : SUCCESS in 56m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/625b9ae6525d4f228ec2912cdf38fcb1 : SUCCESS in 1h 52m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/f3624da5c3024eda92122da395eebbb8 : SUCCESS in 2h 27m 50s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/15c64624e36d4317b685fae43d16162d : SUCCESS in 47m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d1f8ec3e4c3c46f890f4f2ae655567fc : SUCCESS in 5m 28s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a7cb9dee1107432bac59eb70518330e2 : SUCCESS in 26m 21s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f92c98d546374851ad147f18b8561dea : SUCCESS in 24m 14s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/e196c82c20cc48b7bc899da99f96dcdf : SUCCESS in 32m 28s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/97dabd16aca84546a4e5bb47a44a0dbc : FAILURE in 1h 27m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/989fb63803c744e6856c31611bb00552 : SUCCESS in 1h 00m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/50ecbc234e3a4835b6f8bcbe293a8412 : SUCCESS in 52m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2fb42929911d47588d7fd19dbd370d3e : SUCCESS in 33m 21s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/719b99fee4cb445abded790c3c0fc63a : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":5},{"id":"8bb4c211885d561aa656a209f3f4f11daca4ad35","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-13 13:05:08.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7f5e9aed76e7589eb1cc0cc98fe1e4ebdbb02019","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-13 14:58:41.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8625bb2e0fe68bace1211aa6c7c29a792f578545","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-18 07:53:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e12f63a49ca305a10ba6cd4bc0ff96607f7776de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 08:20:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcc4bc16e9e94180bbb850cfd8ec51d4\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a80917174cd47d3ab71ea61ebfd18a2 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/01ba063b1c794b0a9196290d774895f4 : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6ad0c8c839f0449b90a6f9a2a1c84601 : FAILURE in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f7c56bad45024fdbf7df0d4e3b1da45268c238ea","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 08:58:44.000000000","message":"Patch Set 5:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-cvf8t/index.html\u003e","accounts_in_message":[],"_revision_number":5},{"id":"94013b8912448a79c5d8d159a18a97a2d09a5c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 09:57:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35f50e639f84e81ab7dd3b36153de25\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/88a2ad6ed51b42bfb609af957ec4fd30 : SUCCESS in 32m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5f70cc416b7b44a58ccb0148debe6f84 : SUCCESS in 1h 38m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99f42112cc074da5b3052fbec1c8d9b2 : SUCCESS in 1h 00m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/719b0661e05f4c94aafd86d9abfa4445 : SUCCESS in 55m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13deaa31d6424e98bae3e1e194d47019 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a99d82a27624b13982e6e26a9e60510 : SUCCESS in 13m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8f103c418ced432c81d66532db03b1ab : SUCCESS in 13m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fc2c49b803da4966b574c901b74d93b0 : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d04524f55e99410f83aa8d7a8f4ea334 : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a49e47c06e541f1a5e709c0e89f4916 : SUCCESS in 8m 18s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d12098c165974d7d8639652fc3bc1a97 : SUCCESS in 1h 02m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e71076697d64137ba87728dc8a9e5e8 : SUCCESS in 1h 37m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c1a34db8c27b4727b1f069fbef4e0d33 : SUCCESS in 51m 05s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/42569853ced14c378aac8a96a9f1fce6 : SUCCESS in 37m 50s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/1d61c4acda7648c688863f0e42109e8f : SUCCESS in 1h 00m 01s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/8e699cd00b2f430e832c59b0cf85f96b : SUCCESS in 41m 12s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/04bfda9d9e88411faeb62e5d31b02874 : SUCCESS in 1h 30m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/6291f4d61a774a9081997138fa3a0654 : SUCCESS in 2h 00m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/98e95a1811c241aca12b0f23d9ca4c21 : SUCCESS in 49m 24s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8258149b1b4140698394bb0806038479 : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b88da527e41c4116a9f54a4c26bfd71e : SUCCESS in 27m 59s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/37b688c687e44b9c80feb9e8425b9a5a : SUCCESS in 30m 12s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/d9f09c8a6e784e8f98039b201fc154c9 : FAILURE in 36m 29s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/73fcecad3c534965a7657b70025f292d : SUCCESS in 1h 30m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30068769c06d42b886cdf96eba5aa69f : SUCCESS in 54m 58s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/e8fdb69434d442e1bb342f0fa83baee8 : SUCCESS in 55m 23s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/225f2d0e91144d01a601dd8a0f0a0102 : SUCCESS in 33m 11s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/df763dec04d84374a47f5128ab6e3034 : SUCCESS in 11m 57s","accounts_in_message":[],"_revision_number":5},{"id":"d4756e8ebc8c8150ff8593f49eb6f9510855a6b7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 10:12:34.000000000","message":"Patch Set 5:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":5},{"id":"be8e0becad965c8df03a94744987a4e787f28cb0","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-18 20:50:08.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c83972479bc3d7282c5b66200631c9738e5cd8ae","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-19 08:21:50.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e38269427eb573e4071c5afb7207d615db98a8d7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-19 08:42:14.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7d776d6eca65396ae618c8c9ac3f59fe4f10f2be","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-19 09:43:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0777c1f7b26b5a3b5d7a137130f6b3f977fc9c3b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 02:03:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1eb2915dfb1b515130fa3307462ca4737e339b8c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-21 12:35:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"df91978e26c69247526b0fd2089f96457cf8c16b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 13:41:14.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"0f916afa25d6dd8c46eaaa4213f201b85888c23b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 14:37:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0237eecfb948ebbcc5498bd5150174\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f5a98b001624c6d933914870f7f5230 : FAILURE in 25m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4b98f64071104f6999eb9976ca5b8add : FAILURE in 25m 47s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/14f73a4a72814e0690b704f4a69e8d3b : FAILURE in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1f8839bf8d18a46eeb2896e9ca82732d3f1366c5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 14:45:56.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":6},{"id":"27abdbe9d5e111b6531e7b841e02914bc210ef16","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 14:46:13.000000000","message":"Patch Set 6:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-d6ndd/index.html\u003e","accounts_in_message":[],"_revision_number":6},{"id":"3e5af6e91a80c6a9ccbaa9fd9130cbd01030ebab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"date":"2026-08-21 14:56:39.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"1a32b8042eb1a418ccd1bde1468c3d04f4898565","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 15:23:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b7316854c0e464babba11b6e6a77c7c\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b527fd71de4242ca8c7a08e6fa8f9811 : SUCCESS in 15m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/56975b03df7d4afd899a3b7dec5f87da : SUCCESS in 18m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/217003cf662d4de680a375f055e7587f : SUCCESS in 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c6b4f74e244c52e56a8a93951b3169756339b3c0","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-21 15:59:53.000000000","message":"Patch Set 7:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0 .","accounts_in_message":[],"_revision_number":7},{"id":"dd2ebeeae4025a2dd3edff643ee4c71ac93a8487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-21 16:35:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6711349928f8403aae04067702808717\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c4d8aecccad4c27a694f908d5dd9e39 : SUCCESS in 1h 10m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aaa95982f0c64b509a8165cee9ac24ee : SUCCESS in 1h 06m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b0d7089a634424da6b859e105259620 : SUCCESS in 1h 02m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9144a2eedac34641bc9534f2df9d2557 : FAILURE in 1h 04m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb7f830255ae4144bef13927d60b9964 : SUCCESS in 9m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9148ef6f9ac14c0d92720c0f400ccecc : SUCCESS in 15m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c4595c1beba4d508dc1b0ea2b6fdd1d : SUCCESS in 14m 05s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/954c232e72614b368bc3a39a6f27529a : SUCCESS in 19m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d31b63341745dcbb0d1859598b3ef9 : SUCCESS in 13m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0847d376a244e0e858fe92c20e745c4 : SUCCESS in 9m 13s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/f79991bd2d5d46dfbfe12f1f358c4f59 : SUCCESS in 1h 20m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3178c951ad645bf9930acf1e12769f8 : SUCCESS in 1h 32m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/230d6aa9f37a4086b16a50d9a92eee03 : SUCCESS in 33m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/4a35a26e13b242ac906703bb060172d0 : SUCCESS in 54m 25s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/7b27a953b0684055a901b94613642222 : SUCCESS in 1h 00m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/db69e2ee51ca4e7eb18cef07b77f42cc : SUCCESS in 53m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c0ad1cba37c74df39b99738b78609d5d : SUCCESS in 1h 04m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/2be35fff67c44b93839c0c8bdb0af1ca : SUCCESS in 1h 00m 05s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/33cdfa349f8f41c68c4e2048b12976b3 : SUCCESS in 34m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4bce009e27d84db8b4233bfca8a44226 : SUCCESS in 5m 28s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5d105e9a256442e595d36ac17e983d10 : SUCCESS in 24m 40s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/9ab319aad107449d836f021f38e53226 : SUCCESS in 24m 35s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/603bc1e559a14be68f7f70b14cf80725 : SUCCESS in 39m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f443aa5aa8604b998db238f5f1263386 : SUCCESS in 1h 10m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c545eb59a8648ada47d25f130f202e6 : SUCCESS in 1h 01m 56s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/8647e604218342bcbda7f4a092e457d2 : SUCCESS in 52m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/682bd6ab597549aaa2adbf3762741b8c : SUCCESS in 21m 57s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/fadad2255c304bbbac880668a67a82fc : SUCCESS in 10m 49s","accounts_in_message":[],"_revision_number":7},{"id":"f0bc6d049efaa899b99b865eb0258c97eaba9f04","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-21 18:28:01.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-vv7w9/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"f6692e35c9ad2f9555bff3938874e158c84c5d91","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 19:29:02.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eac31c20a80e68795c857c7c81ea516f94417dcd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2026-08-21 19:30:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a1b5dc67a7a02b7f77b396f1b3230c45e63acf31","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-22 09:21:06.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"ea73334875ea39e11596659195caf881ebdb5701","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-22 09:24:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9d8222d780be2df24b2781d80dcac9c02f8f497c","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-22 09:56:01.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-d98kw/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"ecd4d7410004581e20494d974e6b41cefe98a092","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 09:59:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/781fe55aa07c45edbf721a951e0a5789\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2bdf7b384df45dba21ac7f947bba408 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5679fb08e9114a6eb3445310db4439ab : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2455d0f845034271b0deb3829d86fd95 : SUCCESS in 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"961e054fe0b3ef9334ac042e2992d7a962b8643f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 12:25:40.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/212a929984234116a7fb76a838ef34dc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/12f703eb199d407694d55bd7b018032e : SUCCESS in 1h 10m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/38e4346dd10b4cafbecded2021f27991 : SUCCESS in 1h 43m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f3475d570674f5fa302821e1d60fbd4 : SUCCESS in 44m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a23e5768acb8411394568b42c3274719 : FAILURE in 1h 18m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31d29091da84278a515a8057d3474a6 : SUCCESS in 7m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7737aca2d8f14fada4babf0873abd34a : SUCCESS in 12m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/992dfd83d783406087a8aaa48e7eb9d2 : SUCCESS in 10m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/076d7a53ee4e48338b947ff1b334fbc3 : SUCCESS in 14m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a530a24c0f41e9a3e8e641b9f7d841 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1760896d891747048123e44aa65045e5 : SUCCESS in 6m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/c9483d530bff4fa98868b83ecb4390c2 : SUCCESS in 1h 01m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cef00a3df27d4d7daf1ed6b47fe024b2 : SUCCESS in 1h 27m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/62f962983f6c4fdb92b8e376817d3899 : SUCCESS in 39m 48s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/6d11397e96a64281aabb3f3a73d22dee : TIMED_OUT in 3h 03m 15s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9f33b681f02f441dbad9efc4b80d4de7 : SUCCESS in 59m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a7d3fbdc0cff4ae5aaed3e849e539c6f : SUCCESS in 50m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ddc4ce96f81043928ab6e6ef3067ba9e : SUCCESS in 1h 17m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/8e56f8925c9040d5b0936c5172ede318 : SUCCESS in 52m 29s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/b3487c5f6c7f43b282919581fae15fb8 : SUCCESS in 47m 00s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d36b8ab5cd424094a7120e7842bab5c3 : SUCCESS in 4m 58s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0172a9d0e94f4005a30bf349aea233b5 : SUCCESS in 28m 35s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/596c063fe14b469e8a4c168536139eb7 : SUCCESS in 22m 31s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3d37d01b57b44fe998833e40c3fb1254 : FAILURE in 38m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8d914cb2be814f8f9242a0c247aec701 : SUCCESS in 1h 01m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8d7ddd15dbe47e7a01a7a120193bfbc : SUCCESS in 54m 24s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/911663cee1974e20b409e38f362768c7 : SUCCESS in 44m 35s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d768f9ecfc664beca066da7d2676ec25 : SUCCESS in 16m 54s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/390a09d13c5446598d062cab95364b72 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":7},{"id":"179e83be4662a75f2ad6d6199a331ccffb1c142f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-24 11:28:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"31b006a8354fa6c246af03e263b1b92319a8cfa9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-24 11:33:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d6dcb8c59ddb3b1112f83fbfe7fbd65f95d7753f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 11:55:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcfdcd65b70a4841bb3caeb7dccae5e7\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd7f16608c7545758cb93256b1362dae : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0e23e180a8e44513a7deb4e48535c70a : SUCCESS in 13m 56s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/4e8024da02ba4aceaa1e6d63f6ec43ae : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"68ec02b15b8e4eb9c0eeb0e5bbb602008895b384","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 12:02:41.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-x9jcd/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"ffa06687afec293ff2b83b8a2100948ee8d233a1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-24 12:56:19.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a0c5d49ccc8567a3d887a3d62d82b60a9a0133a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 13:14:53.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3f899d4d415435280d363f5579c1d6a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1de8d609d4314cdeb1897fd4e9131fcf : SUCCESS in 1h 14m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/453dc9fc87174d3c92cde9011bd95493 : SUCCESS in 1h 19m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8efedd6c1ce04351b70e70bb0185dcde : SUCCESS in 1h 00m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1b835c955ad4fe18ee06284ac50b160 : SUCCESS in 1h 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3497144dd86a4a1d8bfd92844ba10332 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62e0616f382f448883d9c061cbad2cb7 : SUCCESS in 9m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/213878db864c4bbd8e53b93f08d0eeba : SUCCESS in 13m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f407fd7c1234346b912787543e96460 : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159bd96c15c245ebb9edda11043df90b : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abd821c231774aff90f93ef816716856 : SUCCESS in 7m 51s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/d8229173bf1344e2b841ffaeda7ee469 : SUCCESS in 1h 02m 42s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9cb134a80dd409091404f1a86bd32c7 : SUCCESS in 1h 42m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7e8098906a1d4065acc72b143eb03606 : SUCCESS in 47m 38s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9e8e6adee33a4563a99ca2b0f22d87ac : SUCCESS in 1h 17m 30s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/c2c0576da292425087a790055b1ab445 : SUCCESS in 43m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/57efbafa169b4081a8dc0a98e250d001 : SUCCESS in 56m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4d487a7aea464b9a8dc72cf403e3d8c3 : SUCCESS in 56m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/eb5c040185ce4e34b4f7f4955ce442e8 : SUCCESS in 56m 36s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/52fac13ec8fa4b238ae8ca9fb02f4f1c : SUCCESS in 50m 51s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4fd16b6bd6c8447198e8d867c88a94cb : SUCCESS in 5m 44s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ff158304f8d14f3ab5fadabb7a364727 : SUCCESS in 23m 15s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/b928279f16ed4a168864b2214b1abdb8 : SUCCESS in 26m 33s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/972aebaabc474ad39925769d5214b4f3 : FAILURE in 35m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f40fdca337ad4da38f12fe8ad21c89c6 : SUCCESS in 1h 20m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3c6438c5da8422b9777b2bbc440ad22 : SUCCESS in 31m 25s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/0f6a6e04d3e0459ab235eb2db6520e68 : SUCCESS in 53m 29s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5f1228cd72da46a08ad9301a0c87c736 : SUCCESS in 20m 56s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/a0dd801ea495416c904da2d93f052580 : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":7},{"id":"f394c8cf0cdedc3c3c4ff8f585895ba95cd8b159","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-24 15:03:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fabb64094130b58cb2359e5270135eace697bcc8","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-24 17:04:00.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-pnjll/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"f07afb35e855713cbfaf93ce14d0d991d04bc18d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:30:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ac6a92eff314fdfa2ad04ea268de1dd\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4dc54dfbf62d4917afc417596db37d3f : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0d62fb9cdd6946d7abe586a84f2807eb : SUCCESS in 25m 50s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f6cdf35f6209465687eccf924d1bca24 : SUCCESS in 40m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9f3351acdda34dc67e2e5383e63ad01a402d7a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 17:33:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cf917216f4a468186b50655527fff7a\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0577bf1e30324709aac44d0747f898e1 : SUCCESS in 1h 03m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/055784f984414327b137a108eeaf9a42 : SUCCESS in 1h 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11afbd0d320a43939df9378176565ea2 : SUCCESS in 1h 00m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/310cae3b03ce43be81706be8ebaa8987 : SUCCESS in 1h 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ee47d4ea20487f874e551c87ea733d : SUCCESS in 7m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8322f37fa5b647c387adc36420a23c3e : SUCCESS in 14m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/40bbcae171c54635812b4a44e6679a08 : SUCCESS in 14m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/911288e221e343878bd4031e02221d8d : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c355c557f940908edeb5e10edf189e : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b316e23b96641e59c9db9293a753131 : SUCCESS in 8m 40s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/a0269e84fd734a748a79235ca46b6196 : SUCCESS in 50m 02s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/795c6e118b4f4ee5a05bc7d5a7cd2bbb : SUCCESS in 1h 27m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/201b80d871dc476a909828c42f55543c : SUCCESS in 42m 45s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/1f8ece591e3349938256995a09da999f : SUCCESS in 42m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/ac9afab3fac14658b13fd0988200e9d5 : SUCCESS in 52m 00s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5d621bdd6b3d4f2888a2dec24ec2863e : SUCCESS in 57m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9559091edda343fc8dce600663d48e2d : SUCCESS in 1h 56m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/327c07639e884be085438c913627b7f8 : SUCCESS in 2h 19m 45s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/9cde0879e76649efbf9a3d1ab3a5d9da : SUCCESS in 49m 56s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9b5d65fd742e484baf6d6c95cd20a0cf : SUCCESS in 6m 54s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b053a8bb83f04952b3fd0891539b7d32 : SUCCESS in 31m 06s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e08214fe582841e682f23567ee077add : SUCCESS in 30m 06s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/504c22b7473148e39cd43c45be6f5581 : SUCCESS in 32m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/331e33c58b334b2bb912cfe0a3e70e69 : SUCCESS in 1h 47m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd784c0fa6334f5998cee7e4be24f279 : SUCCESS in 53m 35s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/ea5941237f234b5698141fca01ceaaca : SUCCESS in 42m 08s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7d1d3d3b21844b06a6814012c837974b : SUCCESS in 36m 05s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/4e1900409bac4314b301ef4fbb824d87 : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":7},{"id":"6441a3f39d470af7ef9c51c4bcdc0bdde64549ab","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-24 17:44:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"9ae5e13a075d0e5a0391e82d7752fe5dd1eef1db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 20:08:54.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5b1958fde30193bc0412a8e641344159ada1a717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 21:45:38.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/998466 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"bd2f71d8c614d54930cbb5eb8c19277c2c75c396","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-25 08:14:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"11ab8cf1129cf277c45f1ee7186cdff683c05647","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-25 09:24:05.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-8jtjk/index.html\u003e","accounts_in_message":[],"_revision_number":7},{"id":"62134e78a09b31ce9887536e5e129f703f615fb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 10:14:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b13d48d9c65e4b5aac540fc182177238\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7e4d0c9af134a88b72828456ec071ea : SUCCESS in 30m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dcef00a5813947aca41c590753df4ff1 : SUCCESS in 1h 23m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94b7287c21774270991c9083f0946b75 : SUCCESS in 56m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36fd276b78914d8d810b9d78448f6343 : SUCCESS in 1h 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973fef9d39ee4625a2f4f04b3a4c1f2b : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18252e655a2343eabe495373361590dc : SUCCESS in 14m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cf53daad21a44c2873f9b1b26a7cf38 : SUCCESS in 13m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e04fabf46b3e4fbbb4609d5bd9494313 : SUCCESS in 19m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7d9e3eff1746e8a1fa037330f22566 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ec2cf12975641b28fb01eb76eb6bac9 : SUCCESS in 8m 14s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/92d4217e43894e5ba98c0525e356d3ce : SUCCESS in 41m 13s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bddee5cd7bd64614971db79893e4ae4c : SUCCESS in 1h 36m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ab93a8131292449081ecb195ddd3a6f9 : SUCCESS in 56m 21s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/9625faaa400c45469d4a0062037364be : SUCCESS in 36m 39s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9fdbd1d62c41461fb907df364894474f : SUCCESS in 39m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/437dc3aadc6b4cd8ae35d46e8965216f : SUCCESS in 53m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/19dd7325638342988ce4d6fc97ef758f : SUCCESS in 1h 53m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/d761f9090e6e4929954c6fbb03660fec : SUCCESS in 1h 24m 28s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/d1278e94a8974e6aa626328408fe1b3b : SUCCESS in 38m 22s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/473c6f154d724f4182e7014fcd0e6442 : SUCCESS in 6m 04s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/311a5fa010f6433aaad33f8a603d4354 : SUCCESS in 28m 08s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/64c2360de08d4d7d8ad49f5973463235 : SUCCESS in 34m 20s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/6305d086d02047f383708c08ecfb851c : SUCCESS in 33m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/77ac15f23e4f49aea76f59d9b96df38e : SUCCESS in 1h 32m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ead4475e7504c45946ce03eb9de5209 : SUCCESS in 1h 01m 54s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/34a6841ece7e4c02bff3c4a1b4d10406 : SUCCESS in 56m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7e35980f614647febcefa251ebe471f2 : SUCCESS in 31m 46s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/b011c833f46a43ca8458dfcfc506e798 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":7},{"id":"fd450d26c90836f9f8d213c6cc54ffd7a43511c6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-25 10:25:46.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"9d276df729e4b7b583aba45240225d107c7fd6dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 11:45:42.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d68c6ea038574a0d815ac5e4e80daabe\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa1014929c134fc4bc1864fe94e56249 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/49130ea17b9b4a56a003a8ed545b507e : SUCCESS in 22m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6a2eeb94c80241e790bf72d3cc7a46e5 : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8305ec4d82ebfd76a483f611d6e5935bcb6fd08a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 18:20:48.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9a8975721317dc5d99c8bc013f7bf27e0fa5a533","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-25 20:33:37.000000000","message":"Patch Set 7: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/nova/+/998466 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"cdd44151d16cd8a8847693e4e0b95c2678c7dabf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-26 08:19:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8cc5b434ec4798238e7c1e59e2acd6e48f30d9cc","author":{"_account_id":26471,"name":"Software Factory CI","email":"softwarefactory-operations-team@redhat.com","username":"sf-project-io"},"date":"2026-08-26 08:25:51.000000000","message":"Patch Set 7: Code-Review+1\n\nBuild succeeded (check pipeline).\nhttps://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/buildset/b71caa5326da481886aaefc6c0b6a48e\n\n- s2i-openstack-container-content-provider https://gateway-cloud-softwarefactory.apps.ocp.cloud.ci.centos.org/zuul/t/rdoproject.org/build/869b03d16aae4b0db828c52bf806c2ee : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"514a54a9a05aa71c6d79afee98274e97b8e0ae21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 09:06:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dabcd5e82afb426caf79ceb08aa5734b\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5243177a9d2b4ab0a373741c529e1132 : SUCCESS in 15m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/80416f458a074917a5a9bcd576cf0602 : SUCCESS in 23m 14s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2c81cd032cef48fa902d98828d8e0426 : SUCCESS in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3e9007791953a71f08cd7bc076f70648c3264438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:12:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a717d628e95c417ba0fa8d5ee65e83a1\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e6013eb2ed04b1e9f9d668fb8e421cc : SUCCESS in 1h 06m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a428f932fd2844aaa0e2d5ac243e2eb6 : SUCCESS in 1h 33m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5cefb24214c84a3e8b8408fb563222ca : SUCCESS in 30m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62e6dc13e54c4cc2b47c3636d8846faa : SUCCESS in 58m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3e9febcd29448a94ced4c43d747f60 : SUCCESS in 7m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8adf8c398a654004a9cef5768832f8ab : SUCCESS in 13m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0398a81a56e4495ebd2545972c2d6016 : SUCCESS in 8m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/f88f83135ce249549f4c930c280b2475 : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e037618e62614dcab43c4475c40a9353 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8f9f027ca6d4e5895155c10a87fe470 : SUCCESS in 9m 48s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/bb35cbdab38748e8865091614bc8a6ae : SUCCESS in 55m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b390da019ac46e6ad137a140e1d9881 : SUCCESS in 1h 41m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3b0515a99309482f893c8775ac66e2f1 : SUCCESS in 50m 18s\n- nova-whitebox https://zuul.opendev.org/t/openstack/build/80bc43f69d9b4907abf1dce5acf76bff : SUCCESS in 38m 35s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9b97bb98dd3c4c36a09495328bbdd65f : SUCCESS in 1h 02m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/90f3c6c89e4a477494d8b92ebef45679 : SUCCESS in 52m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3b04cb3ab9e3486599347135542e4676 : SUCCESS in 1h 44m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/3e85d3cee4c445b39c6eada8268a5212 : SUCCESS in 1h 13m 20s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/ce253516e5794bb5861ddfd2ff759fd1 : SUCCESS in 48m 48s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5ff6a750a48547909758a4b3bc54513f : SUCCESS in 5m 15s (non-voting)\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ee4a2763bf4a47a5bfc7451f19dc2fc2 : SUCCESS in 27m 48s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7876b121a8604ff9afab720cf5f62336 : SUCCESS in 29m 37s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/a97fdac4be7646438df95a77e7673f3b : SUCCESS in 30m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/69eea0edf8cf4a8880801a2ad70f9079 : SUCCESS in 1h 19m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d02789259ae64d5497374a1b5bfeb580 : SUCCESS in 51m 38s\n- cyborg-tempest-py3 https://zuul.opendev.org/t/openstack/build/445da10d13b54e979bf898d96dd7fb2a : SUCCESS in 53m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1c507e62ad89474fabbd4975a15639aa : SUCCESS in 22m 09s (non-voting)\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/33968928cb5943279872e027e1aab7dd : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":7},{"id":"2ff5e4ba940d872cbb37445744144fd147da5880","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-26 10:22:56.000000000","message":"Patch Set 7:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":7},{"id":"6abe6d5e2261bc70cdcd9f5c1740245cc8d56924","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 10:46:22.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"564147c88abe64636778ccc99332d86709106d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:56:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13199b394a4648d2999715058edbb511\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84680793ea35480285a045bf867fdd55 : SUCCESS in 1h 06m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/625eb4bbafa0466a98d03aefa2d6c747 : SUCCESS in 1h 36m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58911f4420ac465d925cce16454f7291 : SUCCESS in 34m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acf1db72dcfb4d6e886fcaa64137eb41 : SUCCESS in 7m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab2790ae210d48499ee8c3012042dbe4 : SUCCESS in 13m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1fe72b311c72478fbb5457a94a269e7d : SUCCESS in 15m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500be06aa48944549eb85ef285063779 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e74c5ce58366470bad916f1a2b2885ea : SUCCESS in 8m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6b5eb9defeab4a8f8829beb3ff4fbeaa : SUCCESS in 53m 38s\n- nova-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d52040cf3ca64e21bbb951e580b3ef07 : SUCCESS in 28m 57s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/950551f63a544dd5af0191592470f8e1 : SUCCESS in 25m 12s\n- nova-tox-functional-py313-threading https://zuul.opendev.org/t/openstack/build/3a20de01205949119554716f38231d25 : SUCCESS in 36m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/83913e15ea414ccb8eb0a692836a4d5b : SUCCESS in 1h 55m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/b5172d090c594e6f88648b0e750cfa26 : SUCCESS in 2h 03m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/837b667556a14626854186edda721e7c : SUCCESS in 5m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93101456f7344e46a66714f0934a2649 : SUCCESS in 1h 43m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2cec2a303d94ccb98c3db4edcd783a0 : SUCCESS in 57m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71fc7e4a6aab41988942cf597576e576 : SUCCESS in 1h 00m 15s\n- nova-tox-py313-threading https://zuul.opendev.org/t/openstack/build/d051b6f448c248bd8514542a0e8cdf03 : SUCCESS in 13m 47s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/354b079fe9304231b36cd1c4469047a9 : SUCCESS in 45m 35s","accounts_in_message":[],"_revision_number":7},{"id":"0199c8ad43c4bc099774c6aa44a3c74e86862ff4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:56:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"80a0f3cfffe70092cb5bc8a75c7f45ba00f640ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-26 12:57:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a4a0f7ef9bc4e8fa74c39985833faad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca00adf26c0a4832baf9d5f4e468c2a7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":7},{"id":"67f7f78f1a02adb1731214a921e5d71b989b3ba2","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-28 13:36:31.000000000","message":"Patch Set 7:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-998608-6ftn7/index.html\u003e","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"adf057cafa94a6a3d710b10c030190613304af18","revisions":{"0a61a492623c3a84d3432bf0e34e58ef585d0993":{"kind":"REWORK","_number":1,"created":"2026-07-24 11:57:47.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/1"}}},"commit":{"parents":[{"commit":"bc6824e90d2ab2ba9f19425e46432d476ee41db3","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bc6824e90d2ab2ba9f19425e46432d476ee41db3"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:55:51.000000000","tz":120},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a61a492623c3a84d3432bf0e34e58ef585d0993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0a61a492623c3a84d3432bf0e34e58ef585d0993"}]},"branch":"refs/heads/master"},"6a4e73025d1235753e8f03d5264c4b37212f57c9":{"kind":"REWORK","_number":2,"created":"2026-07-24 15:14:03.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/2"}}},"commit":{"parents":[{"commit":"cb37f8e1223af83e16309100487b16993ac99147","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb37f8e1223af83e16309100487b16993ac99147"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 15:12:16.000000000","tz":-420},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a4e73025d1235753e8f03d5264c4b37212f57c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a4e73025d1235753e8f03d5264c4b37212f57c9"}]},"branch":"refs/heads/master"},"2da09f56fd0c023f11f0255dadeb528f1ceee792":{"kind":"REWORK","_number":3,"created":"2026-07-30 09:25:16.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/3"}}},"commit":{"parents":[{"commit":"910436857da955ba31996c7cd116b1c724bc2dc7","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/910436857da955ba31996c7cd116b1c724bc2dc7"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-29 14:10:18.000000000","tz":-420},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config. TDX\ninstances cannot start with a video device attach and thus only \u0027none\u0027\nis allowed for video model.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2da09f56fd0c023f11f0255dadeb528f1ceee792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2da09f56fd0c023f11f0255dadeb528f1ceee792"}]},"branch":"refs/heads/master"},"1a2509941152c3bae57bd95ae1a893fc7204356d":{"kind":"REWORK","_number":4,"created":"2026-08-07 10:51:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/4"}}},"commit":{"parents":[{"commit":"291eb39ba15bcf428c3eeb2ee7ae2a2e9ba92d4e","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/291eb39ba15bcf428c3eeb2ee7ae2a2e9ba92d4e"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-07 09:36:49.000000000","tz":-420},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config. TDX\ninstances cannot start with a video device attach and thus only \u0027none\u0027\nis allowed for video model.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a2509941152c3bae57bd95ae1a893fc7204356d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a2509941152c3bae57bd95ae1a893fc7204356d"}]},"branch":"refs/heads/master"},"c2b149d1f870d45c797b8a1d8d1cd11c26674100":{"kind":"REWORK","_number":5,"created":"2026-08-13 07:25:43.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/5"}}},"commit":{"parents":[{"commit":"8457ef6437c83961643c69ac6f702ebf52d5531d","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8457ef6437c83961643c69ac6f702ebf52d5531d"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-12 19:47:51.000000000","tz":-420},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config. TDX\ninstances cannot start with a video device attach and thus only \u0027none\u0027\nis allowed for video model.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2b149d1f870d45c797b8a1d8d1cd11c26674100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2b149d1f870d45c797b8a1d8d1cd11c26674100"}]},"branch":"refs/heads/master"},"521a55db00ffa87da15ea2424b3b9fae1ea390cd":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2026-08-21 13:41:14.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/6"}}},"commit":{"parents":[{"commit":"3834b49da0e44c18727b701e8fbceeea2b34dc98","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3834b49da0e44c18727b701e8fbceeea2b34dc98"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 13:29:11.000000000","tz":-420},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config. TDX\ninstances cannot start with a video device attach and thus only \u0027none\u0027\nis allowed for video model.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/521a55db00ffa87da15ea2424b3b9fae1ea390cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/521a55db00ffa87da15ea2424b3b9fae1ea390cd"}]},"branch":"refs/heads/master"},"adf057cafa94a6a3d710b10c030190613304af18":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2026-08-21 14:56:39.000000000","uploader":{"_account_id":38744,"name":"Anton Iacobaeus","display_name":"antia","email":"anton.iacobaeus@canarybit.eu","username":"antia","status":"Canary Bit"},"ref":"refs/changes/08/998608/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/998608/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/998608/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/998608/7"}}},"commit":{"parents":[{"commit":"6ce06745ed511a0a2c9f7b473c923c1655726dcb","subject":"libvirt: Generalize memory encryption config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ce06745ed511a0a2c9f7b473c923c1655726dcb"}]}],"author":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-07-24 11:44:46.000000000","tz":-420},"committer":{"name":"antia","email":"anton.iacobaeus@canarybit.eu","date":"2026-08-21 14:48:25.000000000","tz":-420},"subject":"libvirt: Add memory encryption config for TDX","message":"libvirt: Add memory encryption config for TDX\n\nAdds MemEncryptionConfigTDX to define Intel TDX instance config. TDX\ninstances cannot start with a video device attach and thus only \u0027none\u0027\nis allowed for video model.\n\nPartial-Implements: blueprint intel-tdx-libvirt-support\nChange-Id: Ia6eb110a90e4598d9e3f84dffc6f40291ccceeb0\nSigned-off-by: antia \u003canton.iacobaeus@canarybit.eu\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/adf057cafa94a6a3d710b10c030190613304af18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/adf057cafa94a6a3d710b10c030190613304af18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
