)]}'
{"id":"openstack%2Fnova~999667","triplet_id":"openstack%2Fnova~stable%2F2026.1~I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d","project":"openstack/nova","branch":"stable/2026.1","topic":"eventlet-removal-stable/2026.1","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-08-18 17:43:03.000000000","reason":"Change was submitted"},"5112":{"account":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"last_update":"2026-08-18 17:43:03.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2026-08-18 17:43:03.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2026-08-06 08:53:32.000000000","reason":"\u003cGERRIT_ACCOUNT_9708\u003e replied on the change","reason_account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}}},"hashtags":[],"change_id":"I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d","subject":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14","status":"MERGED","created":"2026-08-03 23:35:39.000000000","updated":"2026-08-18 17:44:02.000000000","submitted":"2026-08-18 17:43:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"999667-eventlet-removal-stable/2026.1","cherry_pick_of_change":987815,"cherry_pick_of_patch_set":7,"meta_rev_id":"99149a30a5345545ce63fae8b0d9b79551b15a28","_number":999667,"virtual_id_number":999667,"owner":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-18 17:43:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-08-04 16:40:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2026-08-06 08:50:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2026-08-06 08:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2026-08-06 08:53:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"CC":[{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-03 23:35:39.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":11082,"name":"Kamil Sambor","email":"ksambor@redhat.com","username":"ksambor"},"state":"CC"},{"updated":"2026-08-04 00:05:59.000000000","updated_by":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"reviewer":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 00:15:12.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2026-08-04 00:15:12.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2026-08-04 00:30:01.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-08-04 03:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-04 03:38:34.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2026-08-04 16:40:09.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-08-06 08:50:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"97af048688b2c8f6d9ebb13b4cfe0d32a126c48b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-03 23:35:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8f505be40f82bdfd6ea46e8a8fcbb0a5bd745546","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 00:05:59.000000000","message":"Patch Set 1:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-999667-lfrqw/index.html\u003e","accounts_in_message":[],"_revision_number":1},{"id":"55fb84ef6d71c89b73472a0695c90c6ea0003845","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-04 00:20:09.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0a0a4a987b04a0e2c354ff12c64ecb823daaad9e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 00:30:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d .","accounts_in_message":[],"_revision_number":1},{"id":"26c4af65dcdf42e9da1c5f7488d04e481ba5b3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-04 00:40:59.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5de196957c13dff410097bd95d262e4ef4e2374b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-04 00:46:49.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"1c3d60a23d614b9e691795f45c21246f365158a2","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 01:13:15.000000000","message":"Patch Set 2:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-999667-gdkgm/index.html\u003e","accounts_in_message":[],"_revision_number":2},{"id":"42209191b3580f862b19b545d1334a1e8b91deb7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 01:16:46.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d .","accounts_in_message":[],"_revision_number":2},{"id":"66e384eefc33e9dc2be1fc4e1a0e115687fcd778","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 01:41:01.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d .","accounts_in_message":[],"_revision_number":3},{"id":"02c5ede8769b3bc20a2f669a9b6d5b848b540ed7","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 02:15:19.000000000","message":"Patch Set 3:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-999667-q6f4f/index.html\u003e","accounts_in_message":[],"_revision_number":3},{"id":"97270567fc4d977af7d31db5f275c1cb28ddcc00","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2026-08-04 02:15:37.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* quobyteci:  nova_quobyteci : SUCCESS, logs at: https://oscilogs.quobyte.com/logs/I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d .","accounts_in_message":[],"_revision_number":4},{"id":"c8d7ec541d20384ffc96b25a9d04861f4f506dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-04 03:38:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/598866e26cec4d7eabf19bd3587759e2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a5bcafe73bd4b6fbce6912934f81c15 : SUCCESS in 1h 07m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7b77d97e39fd42389141b3bf3a97542d : SUCCESS in 1h 57m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/020415a497954fb196db09a5e1d9e4f7 : SUCCESS in 1h 04m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c944dbbc9536401d99972dee3c849c4f : SUCCESS in 56m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e88c924fb2845babb55c9aeeca8c7f7 : SUCCESS in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afdbf53b003344f3867cdadc0f4fd13d : SUCCESS in 16m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/90b430cb45384c4d8803c1f49782f849 : SUCCESS in 10m 01s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/19b10aa7b31f486fbf76f2836b7ffca9 : FAILURE in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612657196bbf42b6bdddbb674f93466f : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72965e608f3b4a51a693d522115fc1aa : SUCCESS in 8m 55s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/db1545287a6d4e4d8374e720a10d9af2 : SUCCESS in 1h 07m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6903d98f3e7141c88046c924957a20c3 : SUCCESS in 1h 44m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/56a1fd077f174e6185d2a11854074982 : SUCCESS in 57m 43s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/9adf4c199bd14cc08db118a6eebdd8d2 : SUCCESS in 1h 03m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3a6d612b192c4c5ca929458548b81e58 : SUCCESS in 53m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f920a2caf4734901a8688a50effb0dca : SUCCESS in 2h 14m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/7e73d6d490774a6992517b3a877b4fa3 : SUCCESS in 2h 48m 23s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/761b1d94368f4f1e9b1d6960be4bfafc : SUCCESS in 53m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f3fe300bb66a4287a77678d41af54b65 : SUCCESS in 5m 17s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a122f239460423f83ce89839ea07319 : SUCCESS in 33m 30s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/62e21b5f28c2452a8d80ae01377709fd : SUCCESS in 23m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3681e369fc74f73bb98aa23b828e1b1 : SUCCESS in 1h 30m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d0519bff97b409a8aa184308e3770fd : SUCCESS in 57m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7efa67a359be45fd937de5e58febc3ec : SUCCESS in 31m 50s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5a2781c7edbc4bf0b3454819ed0dd263 : SUCCESS in 34m 32s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/47820f5010b84e8280c0f26635ea1439 : SUCCESS in 17m 53s","accounts_in_message":[],"_revision_number":4},{"id":"c525385e95812dac37bf439c0833978e55de3189","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-04 03:49:03.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"4837e79990e742b6e4a8761c3a17c3b7dfcae2fa","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-04 05:00:55.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-999667-djqsd/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"9574557561e73bcbedb38df9a1dbdbff3832954d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-04 16:40:09.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"44004a2aa65244d8b7571064bfc72446fc50a141","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-08-06 08:50:30.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d4ba27ad277357479d56474816e8d99f06bef712","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2026-08-06 08:53:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"be731e9bdf3cdb4de757a382e62e9ce0e4d43fd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 08:54:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"43681a80bc82fd76a6bba47fbbfaccd74632db2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 11:09:07.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/633f8bb856a74e459dbaebc92dfa47c9\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/74748093b0eb4f9b99509aa69cdf2cec : SUCCESS in 1h 07m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e055abd5375a4ba786d2f37c2e045e80 : SUCCESS in 2h 14m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d0bfa5f28de4d14ac323af3098b09f2 : SUCCESS in 44m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ca7db5ef6f4a16a57f2506a01db7a1 : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f5ca4b5caf949efa13c70ac053ff53b : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e8586d4ba1544408c1f85beae4aafae : SUCCESS in 15m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623e540e8b674d39ab4f45703a002645 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90912e9583a84feda7bf8635fde8e34a : SUCCESS in 8m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7e77e49325bc4fe78056494248b9db69 : SUCCESS in 55m 36s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84c2f9c3b82941b5a16f2d3978a497ad : SUCCESS in 32m 20s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/07b2643e9eaa46af9a895f02f8151735 : SUCCESS in 21m 57s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7bf55ae4ad854adeaa0df6af50c5e499 : FAILURE in 29m 42s\n- nova-next https://zuul.opendev.org/t/openstack/build/d892132fa1c34c3e8c1488b944f91dc5 : POST_FAILURE in 18m 15s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/924d6e01ff3441309c92131ef55a3d2d : SUCCESS in 6m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c3031220d9148cdbee4a0500df38c91 : SUCCESS in 1h 46m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/53180e5b6d91453cbaf99bf4289af596 : SUCCESS in 1h 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c04fde0ff2ee48c7be4d33f78b232234 : FAILURE in 18m 08s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/fd17373669f84a2b8d2540358a66307e : SUCCESS in 13m 16s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/73a8c8b3f0c74641b60d73ebfd032189 : SUCCESS in 55m 49s","accounts_in_message":[],"_revision_number":4},{"id":"edf8570c1ff9b4ec653de0cc00698ceca7a44042","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-08-06 12:21:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d1b94cb93469e108ee2adc400c7a25c974c2da5f","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-06 12:53:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"92906062c2cc5039ff82648ad3aeed591aa4c3d7","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-06 12:54:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8cdef2c0cb3c745fd19e228481b9cb2df29c9e13","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-06 13:23:06.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-999667-5bnzz/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"29046f57d62ace49f2cf46a6f2492f7340bbd0bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-06 15:07:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/742fb683416f4822b1df88a1ec529a28\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c753ab0b02d4e0980cd99bb9814755b : SUCCESS in 37m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6e8a48ff8afb46aea84366fa2a03b186 : SUCCESS in 1h 23m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f767b18d1454f1bb7910a029ca50380 : SUCCESS in 38m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/031bea75ff9646af82ce1dad7946da02 : SUCCESS in 1h 09m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa6edc80522c48e490524d47b4f53bd2 : SUCCESS in 7m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f1446b4a38a457cbf19f2267b341055 : SUCCESS in 14m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/47372915439d4878bdd61b2840efc160 : SUCCESS in 15m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/14e7313a50cb402ca8f9c12b7afcadd4 : FAILURE in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c3ec2b9f5849079fd14f74a6f7d483 : SUCCESS in 12m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f7f5beba3114582bffb8cee09b637b2 : SUCCESS in 8m 15s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/8c1bcd544aa542bc922333b6ac3f35db : SUCCESS in 31m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1b4ffb02b8b432b886fe49738b4f32c : SUCCESS in 1h 12m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a05eb5141f8d4bbc8ef15f521f0396ce : SUCCESS in 34m 36s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/244b36bc2f804bf2be7ad5a29441d86d : SUCCESS in 54m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ced7ed0c974e4756b438076814d8cdd1 : SUCCESS in 55m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/37c569756e8e44d5828095c28e0d6e96 : SUCCESS in 2h 07m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/89e71d6ce79f4353aa1242c053e670ea : SUCCESS in 2h 12m 24s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/a954416fc7d64c508fff1f633eb5c4d6 : POST_FAILURE in 35m 37s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/5fd4421b426142a38a4917598fc23d59 : SUCCESS in 3m 41s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb670a1417434d8f86dd3b21d3a9080c : SUCCESS in 22m 37s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/413a407fe313405ea45ccaa1a96fc510 : SUCCESS in 27m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/97ddc2a9498748d882680443bd318173 : FAILURE in 1h 04m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edd016ae2c9c4f209d8596d297a01601 : SUCCESS in 59m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/22033252e8e24967bf3130af51db089d : SUCCESS in 41m 43s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8cdf94038c2f477693b844417670df7b : SUCCESS in 26m 23s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/171907d75e5248ffa77cea76706dc469 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":4},{"id":"3b430971dff12e2df92843e8b800adab2d8ab4af","author":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"date":"2026-08-18 12:52:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3ca9f0e31df50dc8e38f068ab70d7abe7dd2132","tag":"autogenerated:sap-openstack-ci~result","author":{"_account_id":36571,"name":"vmwareapi CI run by SAP","display_name":"vmwareapi CI run by SAP","email":"openstack-ci@sap.com","username":"sap-openstack-ci","status":"CI account for integration tests of the Nova vmwareapi driver","tags":["SERVICE_USER"]},"date":"2026-08-18 13:27:42.000000000","message":"Patch Set 4:\n\nIgnore for now: Test failed. Logs on \u003chttp://openstack-ci-logs.global.cloud.sap/openstack-nova-999667-sh2c9/index.html\u003e","accounts_in_message":[],"_revision_number":4},{"id":"b77fd20b7f06c17519d79fea1fc0507a18755c10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 15:06:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3883f244e4fa4bc0ac3d73c61be619bc\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bd5647ff8da48e4b7e823372692c2ef : SUCCESS in 48m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ad1f93b1209c444c8dc4c07e33f087d2 : SUCCESS in 1h 15m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba572a7a3d0b4f0da5d8049dffc52be0 : SUCCESS in 39m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e20376d85274ac2a9107c0b438df806 : SUCCESS in 1h 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6a114f61c74724b8eb3c696614b7c6 : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/494a44a6fe3649bbb5082e9d591a5656 : SUCCESS in 10m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/956db4a3db78433f9c95b450b1796cad : SUCCESS in 13m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4d7fb2bd3ae2470a95341b42f48e6c98 : FAILURE in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8cb2e4394914468b03e5b3ffefdfd45 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6f43af6093b4e83aa8c57a042d502b1 : SUCCESS in 8m 57s\n- ironic-tempest-bios-ipmi-autodetect https://zuul.opendev.org/t/openstack/build/0ec3b2d3b92948bf9a7b46129ff02d9b : SUCCESS in 43m 15s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/24bde02b1c8447ee874e6f713ff94593 : SUCCESS in 1h 07m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dce7efd6d2044571a1352e44a1724f38 : SUCCESS in 55m 52s\n- nova-graceful-shutdown https://zuul.opendev.org/t/openstack/build/567c1a243e574d6585da9dd308e713c1 : SUCCESS in 42m 40s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/88a7834c08c14c0295822468e3d99691 : SUCCESS in 41m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/15485fbd5cfe41f59e63ee1af5fae2a0 : SUCCESS in 2h 09m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/ed46d462f93640d3a208abeb6bb2675c : SUCCESS in 2h 08m 32s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/51e5a4a51782451daa6b56d8fb6858c2 : SUCCESS in 55m 29s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4166af7dd5364ad9a7f213f027fc8d61 : SUCCESS in 5m 02s (non-voting)\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cbc3226249524318a8bccf142e1a948f : SUCCESS in 30m 34s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/21c9500ea19b4258835e827029733451 : SUCCESS in 25m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8264d3017b5941c2a75c5bc9108fd38c : SUCCESS in 1h 29m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29357965fe604b7b80f24beab4df959a : SUCCESS in 1h 10m 48s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fa8f0e80e59246eba9a24bbcc857aef5 : SUCCESS in 35m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0bc3dd8987794629b722e7741dff5c6b : SUCCESS in 24m 07s (non-voting)\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/817d572d990a419aa07d4a2013dfa5d8 : SUCCESS in 16m 39s","accounts_in_message":[],"_revision_number":4},{"id":"4a2365c428d3b654d10458045755af2dd393b6ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 15:07:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"db5d9c28e83607df748c955bd8207b78dbf62ad2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2026-08-18 15:17:31.000000000","message":"Patch Set 4:\n\nBuild succeeded. log path: ${LOG_PATH}\nLeave a comment with","accounts_in_message":[],"_revision_number":4},{"id":"2ca4004aadae4d797cc1e1df5f631b6303919b7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 17:43:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c7527f42522401c9d0fe516c94a9db2\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/35229189f53e48efb21cf14fa338d851 : SUCCESS in 36m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6fad47a147d7465482349b3cd0902020 : SUCCESS in 2h 30m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/044a48abac8b41459e2a235904dbd6cf : SUCCESS in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671cc6c7252d44b38fe875dfd8c3f7b5 : SUCCESS in 8m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4789518dae104961ac76625d34a23b69 : SUCCESS in 17m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3451e5919e9c46b2a3d69216ee8edce7 : SUCCESS in 9m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5366b45d53d143efbe8bb9549b009dcc : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb0fbe377e674bd5be35d521dd427ea5 : SUCCESS in 6m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dd564468f9a54fa7bbf3776b3188c8d6 : SUCCESS in 1h 07m 39s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45009bc1ba1e4263a28b8e62ef33fe63 : SUCCESS in 27m 49s\n- nova-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7ea56ff209564095ba072e1283a22f1d : SUCCESS in 26m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a07f725fa89a4a1b988812a2fd3b535c : SUCCESS in 1h 40m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/945dcebb18dd4079a36bc6383aaf8cef : SUCCESS in 2h 24m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9e3ad212a500484788228323c40877d9 : SUCCESS in 3m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/687441364c1b4678a548b83b3f41e97c : SUCCESS in 1h 53m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a801b894aa19497f9db6eff3c919da32 : SUCCESS in 1h 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf14fb087b164d4da60f05369a536348 : SUCCESS in 35m 24s\n- nova-tox-py312-threading https://zuul.opendev.org/t/openstack/build/7fda7d9055a84e20b4df458ef2b4ff5d : SUCCESS in 13m 09s\n- nova-alt-configurations https://zuul.opendev.org/t/openstack/build/5ffa3ffaede14b769f6cb8f5933011ca : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":4},{"id":"b31f653a5450dfa0cccfcdb4d82bc48747f8738a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 17:43:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"99149a30a5345545ce63fae8b0d9b79551b15a28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-18 17:44:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41f2cc62dc1460ba49027d703b20db1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b3f15905811410895d137221fc9b01e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d9b413f116588a2d799be73a2e0dbadc68910a68","revisions":{"8b5a580ebc208d33f0839723664ada30a61e6a16":{"kind":"REWORK","_number":1,"created":"2026-08-03 23:35:39.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/67/999667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/67/999667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/67/999667/1"}}},"commit":{"parents":[{"commit":"b62e7d97f0b50b3aeb99547c879d39f661621103","subject":"Merge \"[novnc]Only log traffic if debug\u003dTrue\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b62e7d97f0b50b3aeb99547c879d39f661621103"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2026-05-08 08:23:40.000000000","tz":120},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-08-03 23:35:39.000000000","tz":0},"subject":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14","message":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14\n\nChange-Id: I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d\nSigned-off-by: Kamil Sambor \u003ckamil.sambor@gmail.com\u003e\n(cherry picked from commit 3356917d0ce88baf65c9ce332dc40c202240ea79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b5a580ebc208d33f0839723664ada30a61e6a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8b5a580ebc208d33f0839723664ada30a61e6a16"}]},"branch":"refs/heads/stable/2026.1","conflicts":{"base":"1a8e5e0cefd5322d94719e5002f6bd4a4d83c944","ours":"b62e7d97f0b50b3aeb99547c879d39f661621103","theirs":"3356917d0ce88baf65c9ce332dc40c202240ea79","merge_strategy":"recursive","contains_conflicts":false}},"b6fec4a6ed6e5233af141aaefcd17c4a79d2470e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-08-04 00:20:09.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/67/999667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/67/999667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/67/999667/2"}}},"commit":{"parents":[{"commit":"b62e7d97f0b50b3aeb99547c879d39f661621103","subject":"Merge \"[novnc]Only log traffic if debug\u003dTrue\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b62e7d97f0b50b3aeb99547c879d39f661621103"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2026-05-08 08:23:40.000000000","tz":120},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-08-04 00:20:09.000000000","tz":0},"subject":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14","message":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14\n\nCloses-Bug: #2156932\n\nChange-Id: I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d\nSigned-off-by: Kamil Sambor \u003ckamil.sambor@gmail.com\u003e\n(cherry picked from commit 3356917d0ce88baf65c9ce332dc40c202240ea79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6fec4a6ed6e5233af141aaefcd17c4a79d2470e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6fec4a6ed6e5233af141aaefcd17c4a79d2470e"}]},"branch":"refs/heads/stable/2026.1","description":"Edit commit message","conflicts":{"base":"1a8e5e0cefd5322d94719e5002f6bd4a4d83c944","ours":"b62e7d97f0b50b3aeb99547c879d39f661621103","theirs":"3356917d0ce88baf65c9ce332dc40c202240ea79","merge_strategy":"recursive","contains_conflicts":false}},"1fd82d16ab1f3156c9a3463b604a3c9e4bf2043a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-08-04 00:40:59.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/67/999667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/67/999667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/67/999667/3"}}},"commit":{"parents":[{"commit":"b62e7d97f0b50b3aeb99547c879d39f661621103","subject":"Merge \"[novnc]Only log traffic if debug\u003dTrue\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b62e7d97f0b50b3aeb99547c879d39f661621103"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2026-05-08 08:23:40.000000000","tz":120},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-08-04 00:40:59.000000000","tz":0},"subject":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14","message":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14\n\nCloses-Bug: 2156932\n\nChange-Id: I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d\nSigned-off-by: Kamil Sambor \u003ckamil.sambor@gmail.com\u003e\n(cherry picked from commit 3356917d0ce88baf65c9ce332dc40c202240ea79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fd82d16ab1f3156c9a3463b604a3c9e4bf2043a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fd82d16ab1f3156c9a3463b604a3c9e4bf2043a"}]},"branch":"refs/heads/stable/2026.1","description":"Edit commit message","conflicts":{"base":"1a8e5e0cefd5322d94719e5002f6bd4a4d83c944","ours":"b62e7d97f0b50b3aeb99547c879d39f661621103","theirs":"3356917d0ce88baf65c9ce332dc40c202240ea79","merge_strategy":"recursive","contains_conflicts":false}},"d9b413f116588a2d799be73a2e0dbadc68910a68":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-08-04 00:46:49.000000000","uploader":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"ref":"refs/changes/67/999667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/67/999667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/67/999667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/67/999667/4"}}},"commit":{"parents":[{"commit":"b62e7d97f0b50b3aeb99547c879d39f661621103","subject":"Merge \"[novnc]Only log traffic if debug\u003dTrue\" into stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b62e7d97f0b50b3aeb99547c879d39f661621103"}]}],"author":{"name":"Kamil Sambor","email":"kamil.sambor@gmail.com","date":"2026-05-08 08:23:40.000000000","tz":120},"committer":{"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","date":"2026-08-04 00:46:49.000000000","tz":0},"subject":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14","message":"Restore \u0027fork\u0027 start method for daemon mode on Python 3.14\n\nCloses-Bug: #2156932\nChange-Id: I8a2dabf85a4d8bb1cf1f48f9bde891add394b73d\nSigned-off-by: Kamil Sambor \u003ckamil.sambor@gmail.com\u003e\n(cherry picked from commit 3356917d0ce88baf65c9ce332dc40c202240ea79)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9b413f116588a2d799be73a2e0dbadc68910a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d9b413f116588a2d799be73a2e0dbadc68910a68"}]},"branch":"refs/heads/stable/2026.1","description":"Edit commit message","conflicts":{"base":"1a8e5e0cefd5322d94719e5002f6bd4a4d83c944","ours":"b62e7d97f0b50b3aeb99547c879d39f661621103","theirs":"3356917d0ce88baf65c9ce332dc40c202240ea79","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
