)]}'
{"id":"openstack%2Foctavia-dashboard~660813","triplet_id":"openstack%2Foctavia-dashboard~master~I18bc3ce532e163acc1d49daa84cd7c7495dbc777","project":"openstack/octavia-dashboard","branch":"master","hashtags":[],"change_id":"I18bc3ce532e163acc1d49daa84cd7c7495dbc777","subject":"Fix devstack plugin python3 support","status":"MERGED","created":"2019-05-22 19:47:09.000000000","updated":"2019-05-23 11:34:31.000000000","submitted":"2019-05-23 11:34:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660813-1558611271819-2a496124","meta_rev_id":"9f0927000a794c5e3c30e08c5a797a85428b4ff7","_number":660813,"virtual_id_number":660813,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-05-23 11:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":2,"date":"2019-05-23 11:12:28.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-05-23 11:18:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":1,"date":"2019-05-23 11:18:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 11:12:28.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-05-23 11:18:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-05-23 11:34:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d9792818a7920c6c9f77cd2f90e3cde4fa14d8b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-22 19:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcf90e155840c2e6816f3ff2e88670805c580bf3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-05-22 19:48:56.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit fb14e34253b518d879f386d64d6cf2302d91f06d","accounts_in_message":[],"_revision_number":1},{"id":"2a3271716e9448c0400b11bc26da51c3455c53ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-22 22:57:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/13/660813/1/check/nodejs10-npm-run-lint/2fbce1b/ : SUCCESS in 5m 43s\n- nodejs10-npm-run-test http://logs.openstack.org/13/660813/1/check/nodejs10-npm-run-test/f66398f/ : SUCCESS in 6m 55s\n- openstack-tox-pep8 http://logs.openstack.org/13/660813/1/check/openstack-tox-pep8/dce46f6/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/13/660813/1/check/openstack-tox-py27/3ca7468/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/13/660813/1/check/openstack-tox-py36/2999ad1/ : SUCCESS in 5m 48s\n- openstack-tox-py37 http://logs.openstack.org/13/660813/1/check/openstack-tox-py37/2d30ff8/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/13/660813/1/check/openstack-tox-docs/a319755/html/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/660813/1/check/openstack-tox-lower-constraints/bfbc5ce/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"c6f5c696cfd101ac4c007c37da67c3db2f09dfee","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-05-23 11:12:28.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"a2118bbf2458671ff88faea9edd605098725347e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-05-23 11:18:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f19ed02364a23142cb3325084e5861f9472cb954","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 11:18:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"641b1e4e0ec11e75a000d8566d7938bec3b424e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 11:34:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9f0927000a794c5e3c30e08c5a797a85428b4ff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 11:34:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs10-npm-run-lint http://logs.openstack.org/13/660813/1/gate/nodejs10-npm-run-lint/8b522f8/ : SUCCESS in 6m 01s\n- nodejs10-npm-run-test http://logs.openstack.org/13/660813/1/gate/nodejs10-npm-run-test/e589d6c/ : SUCCESS in 6m 20s\n- openstack-tox-pep8 http://logs.openstack.org/13/660813/1/gate/openstack-tox-pep8/2369f5b/ : SUCCESS in 5m 02s\n- openstack-tox-py27 http://logs.openstack.org/13/660813/1/gate/openstack-tox-py27/bd72c12/ : SUCCESS in 5m 51s\n- openstack-tox-py36 http://logs.openstack.org/13/660813/1/gate/openstack-tox-py36/d22b452/ : SUCCESS in 5m 31s\n- openstack-tox-py37 http://logs.openstack.org/13/660813/1/gate/openstack-tox-py37/9e4c8c3/ : SUCCESS in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/13/660813/1/gate/openstack-tox-docs/12de2a3/html/ : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/13/660813/1/gate/openstack-tox-lower-constraints/a5c5233/ : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3b601d5ac243b5a97b522d871cf31d6060d2a8e","revisions":{"c3b601d5ac243b5a97b522d871cf31d6060d2a8e":{"kind":"REWORK","_number":1,"created":"2019-05-22 19:47:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/13/660813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/13/660813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/13/660813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/13/660813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/13/660813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/13/660813/1"}}},"commit":{"parents":[{"commit":"360e41d7c84a3f5d96bbe855ec67cbece7fd10bf","subject":"Merge \"Changes for url transition to OpenDev\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/360e41d7c84a3f5d96bbe855ec67cbece7fd10bf"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-22 19:45:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-05-22 19:47:08.000000000","tz":-420},"subject":"Fix devstack plugin python3 support","message":"Fix devstack plugin python3 support\n\nThe octavia-dashboard devstack plugin.sh script was calling\n\"python\". This causes the plugin to fail on python3 only hosts.\nThis patch corrects this by using the devstack $PYTHON variable.\n\nChange-Id: I18bc3ce532e163acc1d49daa84cd7c7495dbc777\nStory: 2005761\nTask: 33469\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/c3b601d5ac243b5a97b522d871cf31d6060d2a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/c3b601d5ac243b5a97b522d871cf31d6060d2a8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
