)]}'
{"id":"openstack%2Foctavia-dashboard~668261","triplet_id":"openstack%2Foctavia-dashboard~stable%2Fqueens~I8ef8f740b356344213b24f68d911fae50964d3e3","project":"openstack/octavia-dashboard","branch":"stable/queens","topic":"constraints-updates","hashtags":[],"change_id":"I8ef8f740b356344213b24f68d911fae50964d3e3","subject":"Update tox.ini for new upper constraints strategy","status":"MERGED","created":"2019-06-28 22:58:40.000000000","updated":"2019-07-23 07:05:39.000000000","submitted":"2019-07-23 07:05:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668261-1563865539666-97be2c9e","meta_rev_id":"92ca76fd4f39b433b67ac86d810823cfce8071bb","_number":668261,"virtual_id_number":668261,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-07-23 07:05:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-23 06:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-23 06:45:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-07-23 06:45:30.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 06:37:34.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-23 06:45:30.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-23 07:05:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60ffd8bf972c3d1753b27f30ce733cef05112a0d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 22:58:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57f67110883bcbc2b7302bff6b635b688c72b115","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 23:09:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/668261/1/check/openstack-tox-pep8/6377138/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/61/668261/1/check/openstack-tox-py27/b0b41a7/ : SUCCESS in 4m 31s\n- openstack-tox-py35 http://logs.openstack.org/61/668261/1/check/openstack-tox-py35/9c53166/ : SUCCESS in 4m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/668261/1/check/build-openstack-sphinx-docs/f60ab51/html/ : SUCCESS in 3m 56s\n- build-openstack-releasenotes http://logs.openstack.org/61/668261/1/check/build-openstack-releasenotes/98250e4/html/ : SUCCESS in 2m 58s\n- nodejs-npm-run-lint http://logs.openstack.org/61/668261/1/check/nodejs-npm-run-lint/97b10e1/ : SUCCESS in 4m 54s\n- nodejs-npm-run-test http://logs.openstack.org/61/668261/1/check/nodejs-npm-run-test/aa07643/ : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"cbb136b263280ac3deebfbf71f59a75564dd5633","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:54:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e22472f861e30f93d961ca78c3f96fed4f5ccaa","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-22 20:36:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2c792d1417846222009b51f88c38a6cad9921b01","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 06:36:51.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b322ce1164d2e49642d80ffa4f33c632d83d995d","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 06:37:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c35a476bba56a6583ca47a90acb8335bef786f18","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-23 06:37:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e3bc49ab2ade5789facec520d5f0d170547608f6","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-23 06:45:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"adf13d1bd6f46a53a5d9e09fc94683156712049b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:48:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/668261/2/check/openstack-tox-pep8/960e54d/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/61/668261/2/check/openstack-tox-py27/11e8cf9/ : SUCCESS in 6m 01s\n- openstack-tox-py35 http://logs.openstack.org/61/668261/2/check/openstack-tox-py35/37ee75e/ : SUCCESS in 6m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/668261/2/check/build-openstack-sphinx-docs/881d3c6/html/ : SUCCESS in 5m 15s\n- build-openstack-releasenotes http://logs.openstack.org/61/668261/2/check/build-openstack-releasenotes/b6ccc71/html/ : SUCCESS in 4m 26s\n- nodejs-npm-run-lint http://logs.openstack.org/61/668261/2/check/nodejs-npm-run-lint/cb0fd25/ : SUCCESS in 6m 19s\n- nodejs-npm-run-test http://logs.openstack.org/61/668261/2/check/nodejs-npm-run-test/6a3fdd2/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"b7f21e2a3f92f858eb4ccd50d2e37c978fc03be1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 06:48:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d650dad48b19febdd39c9014c6044e6aa4f3eba0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:05:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"35abba5095d18b768486955ee7fb790025ebcbc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:05:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/668261/2/gate/openstack-tox-pep8/ef9fd5f/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/61/668261/2/gate/openstack-tox-py27/62d202a/ : SUCCESS in 5m 23s\n- openstack-tox-py35 http://logs.openstack.org/61/668261/2/gate/openstack-tox-py35/037e9b3/ : SUCCESS in 5m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/61/668261/2/gate/build-openstack-sphinx-docs/1593268/html/ : SUCCESS in 6m 03s\n- build-openstack-releasenotes http://logs.openstack.org/61/668261/2/gate/build-openstack-releasenotes/0db9cda/html/ : SUCCESS in 3m 58s\n- nodejs-npm-run-lint http://logs.openstack.org/61/668261/2/gate/nodejs-npm-run-lint/484dc03/ : SUCCESS in 5m 56s\n- nodejs-npm-run-test http://logs.openstack.org/61/668261/2/gate/nodejs-npm-run-test/ead78ac/ : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51d1915049618edf56d76df9ef19ba4ecddce44b","revisions":{"b6305efebe4b32f96776e559521220e0408f8aff":{"kind":"REWORK","_number":1,"created":"2019-06-28 22:58:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/668261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/61/668261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/1"}}},"commit":{"parents":[{"commit":"8767e127723c95bab432a138907a692f2c1b6669","subject":"Fix auth url for Barbican client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/8767e127723c95bab432a138907a692f2c1b6669"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 22:58:19.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 22:58:39.000000000","tz":-420},"subject":"Update tox.ini for new upper constraints strategy","message":"Update tox.ini for new upper constraints strategy\n\nThe requirements team has defined a new strategy for handling\nupper constraints[1]. This patch applies those strategies to\nOctavia dashboard.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/ \\\n    2019-May/006478.html\n\nChange-Id: I8ef8f740b356344213b24f68d911fae50964d3e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/b6305efebe4b32f96776e559521220e0408f8aff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/b6305efebe4b32f96776e559521220e0408f8aff"}]},"branch":"refs/heads/stable/queens"},"51d1915049618edf56d76df9ef19ba4ecddce44b":{"kind":"REWORK","_number":2,"created":"2019-07-23 06:37:26.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/61/668261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/61/668261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/61/668261/2"}}},"commit":{"parents":[{"commit":"8767e127723c95bab432a138907a692f2c1b6669","subject":"Fix auth url for Barbican client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/8767e127723c95bab432a138907a692f2c1b6669"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 22:58:19.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-07-23 06:37:23.000000000","tz":0},"subject":"Update tox.ini for new upper constraints strategy","message":"Update tox.ini for new upper constraints strategy\n\nThe requirements team has defined a new strategy for handling\nupper constraints[1]. This patch applies those strategies to\nOctavia dashboard.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/ \\\n    2019-May/006478.html\n\nChange-Id: I8ef8f740b356344213b24f68d911fae50964d3e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/51d1915049618edf56d76df9ef19ba4ecddce44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/51d1915049618edf56d76df9ef19ba4ecddce44b"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
