)]}'
{"id":"openstack%2Foctavia-dashboard~763569","triplet_id":"openstack%2Foctavia-dashboard~master~I0eea32565ade95d14ecf00faafc17f3dc0c15ac9","project":"openstack/octavia-dashboard","branch":"master","attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-08-31 10:27:27.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}}},"removed_from_attention_set":{"38562":{"account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"last_update":"2026-08-31 10:27:27.000000000","reason":"\u003cGERRIT_ACCOUNT_38562\u003e replied on the change","reason_account":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2026-05-31 20:06:14.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-05-31 20:06:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-08-31 10:31:11.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-05-31 19:57:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I0eea32565ade95d14ecf00faafc17f3dc0c15ac9","subject":"Add TLS versions support for listeners and pools","status":"NEW","created":"2020-11-20 14:36:47.000000000","updated":"2026-08-31 10:31:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"30d9bb60cc0e5c700a0e5eeb85e215db498865ca","_number":763569,"virtual_id_number":763569,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-05-31 20:16:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"all":[{"value":1,"date":"2026-08-31 10:27:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2026-05-31 20:17:03.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 10:32:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 08:42:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2021-02-19 18:45:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-03-11 21:31:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2026-05-31 19:58:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-05-31 19:58:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"state":"REVIEWER"},{"updated":"2026-05-31 20:06:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"40849bd69535a21216c497603c98571fae1e01b1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-20 14:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5217b99be8b728fb38eb8b8877ce57195098e73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-23 10:00:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3c0f244b2c833670b4cba17e9fb8b34864d1db49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 10:32:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2d8d7332781465083838500d60bace9 : SUCCESS in 4m 13s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/55b19420c60f47b0a2e6969cbf4bf1bb : SUCCESS in 5m 06s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/6cadb72fd1a8453eba70f78459fc54f0 : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/497d1b7d1cb24d5285e2f59058dbd76b : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40e6bbcbf8524f2084870723f55b8887 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e36f00a967d4bd5a3500c5bef4cc1c0 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d10335d5b07345f7bba2ccccc94ab0fd : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70181ced6e8f4738ad18d1b1960834b5 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cfca2af3875484390378020dac978fb : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"89254ab4ea16adc297c6cbe3060ad6d8002cef0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-24 07:25:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cdd8dfa51c36b6fde477138c49e8ea0b1a2d05df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 07:45:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f94f294935034cbf9c08e54d811d90a5 : SUCCESS in 4m 18s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/8044d8aa2eab4abdabc17119eb1eba10 : SUCCESS in 5m 15s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/fc4e2dc0eda24d78969f3e4f4d49e7bd : SUCCESS in 7m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/626a56cde2d64be1bbd494a45639114a : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ac318ec9c345d2a9300028315fb3c9 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d4eee9c44c480aa4a934e751c2a43c : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46934faa9ee14ce7a84f9f26c754d4e6 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300912b5f61341099924956d040e3e29 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/304698459337490e9858d9ae51b9ea09 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"a909086c039bc4607ec4437149f6a946cc9ac608","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2020-11-24 17:00:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8a16d734b9685edc57bc153a9815917c7f8c1aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 17:28:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/958170c9126a4b1a9f44983e15700980 : SUCCESS in 5m 36s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/2c0998ff20a2455aac5e26d180f789e2 : SUCCESS in 6m 44s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/d65a967e6f0c4494bc2239d31a4b3449 : SUCCESS in 14m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e05231f91e404207a1a946d3433ac2c2 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc235271781643ea80de8224edb00ae3 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c646a478b9444fcb01433e76ddeb0bd : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f04c8f7137ed48e7b126971941a9f88b : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3ea7f034e1a4504b67f95cfacb17de6 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/736e6ead24c44f118cabbee3db1e983e : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":4},{"id":"cd43b70a183cb2615adc213cc285995605687933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-27 14:35:46.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"528fa6fa3f4e4a3d5e5b05b6a223a2501d3e87db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:00:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d56accc858fe4701accd0ee64ba0a733 : SUCCESS in 4m 35s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/cfd900eaa8ea428b9af9a25eb5fd426b : SUCCESS in 6m 00s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/360e43cd6b63409ba374b4a3171e9816 : SUCCESS in 9m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9a67ace8a08f45ffb98298f9000046a8 : FAILURE in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f8a476ee7774f09b440ab315b4583ce : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acbf689df8914a82b8a5b594a5977df2 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eee01b79627b406aa8412a128ce4c719 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc3bfd1f79c34fd089fafe7814c0117d : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f72a7b37d0bf4249901f7d0c2cbe50f1 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":5},{"id":"b20c278a5b1e454bb0afc92150956f316a48d538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-01-29 10:02:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"8e169485c94fa372090fca1a1bae5d795cf17740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 10:20:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00b1f37f374f4be38805367ad56cbe34 : SUCCESS in 4m 20s\n- horizon-nodejs10-run-lint https://zuul.opendev.org/t/openstack/build/fc464bf820b142ba9a4655ff31fe0cce : SUCCESS in 4m 57s\n- horizon-nodejs10-run-test https://zuul.opendev.org/t/openstack/build/17a78d2b153b47bdb15df74e7b12f67d : SUCCESS in 8m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e90accb7243e470ab85c3029f027707b : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d394ba999a544f04b579b5546ca1c048 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49fbc950ce834632a0d596421ea015ed : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f403b061355460ca0e5d9ae575d36dd : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4d19210f20489ab5fe2a97adb824ee : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdc436ab8c82410c8b1e03bfa2923690 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"9d3157dcebbff3ebb2f462b8bceb8ebb0813b588","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-02-12 08:42:16.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c1a96ac671404e7ff6755b3df47fd1fb4c2d166d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-02-19 18:45:52.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"a3e1d7059213a6e7499dac0f14e03e5b5d593734","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-03-11 21:31:16.000000000","message":"Patch Set 6: Code-Review-1\n\nI ran into a problem with this patch in testing.\n\n1. There is not help text for the TLS version field. I noticed this when I could not enter \"TLSv1.2,TLSv1.3\" in the field.\n2. The required formatting on the field seems odd. My natural instinct was to put a comma separated list in this field, but it won\u0027t accept it. Can we make this field more flexible or at least document the required formatting? I had to put one version per line for it to take the change.","accounts_in_message":[],"_revision_number":6},{"id":"6961e24e787ff4959604f734c1386ed3e10afb77","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 15:27:22.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6: Code-Review-1\n\u003e \n\u003e I ran into a problem with this patch in testing.\n\u003e \n\u003e 1. There is not help text for the TLS version field. I noticed this when I could not enter \"TLSv1.2,TLSv1.3\" in the field.\n\nAck, I will add the help text for this new field.\n\n\n\u003e 2. The required formatting on the field seems odd. My natural instinct was to put a comma separated list in this field, but it won\u0027t accept it. Can we make this field more flexible or at least document the required formatting? I had to put one version per line for it to take the change.\n\nThe next textarea uses the ng-list\u003d\"\u0026#10;\" attribute, it allows to automatically convert a string with comma-separated items into a multiline textarea (one item per line), and it does the reverse conversion when the form is submitted.\nI agree that without neither help nor placeholder value, it can be confusing for someone who knowns the API.\nI can reverted it to a simple comma-separated string","accounts_in_message":[],"_revision_number":6},{"id":"d1cc6e25c5f60c80fc6ccedb3c2b2a619bdde518","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-03-19 15:27:55.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"7e702211a874ec0332ad834b444e393255f4e38e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:43:42.000000000","message":"Patch Set 7: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"d62169001d20d28b1a2fe6f04101b7446454d9a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:43:52.000000000","message":"Patch Set 8: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":8},{"id":"aa6df2b8db7e46767f0a1cf4d3492a901172e535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 19:44:17.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/octavia-dashboard for 763569,8","accounts_in_message":[],"_revision_number":8},{"id":"1f2bd00b8b5168367e254045a2e10c1a6fdb20d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:44:22.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":9},{"id":"c390e9fc7b56efecc5d5f27cbe091e5e8bf1c0a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:44:29.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"e65a55bd2b14251cc5e5662ac512574446df40ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:57:37.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"701ae2c76e79d64aaacef8b033bbb37014e0fa31","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:58:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f49630903d6c40e9c50758d01e5728ebe9e3c8ac","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 19:59:04.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"70c2a5a2e85ae444377eddb919efd01d123b277b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 20:16:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bd216a2fd1a4e129caf3103a1b67714\n\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a810503f440944bf91428bf59ed9f428 : SUCCESS in 4m 11s\n- horizon-tox-python3-django42 https://zuul.opendev.org/t/openstack/build/d0d669f5f0fc447ebd12914269388b8c : SUCCESS in 4m 04s\n- horizon-tox-python3-django52 https://zuul.opendev.org/t/openstack/build/39ae0c1e6e924258a7d97a7b00f6b21e : SUCCESS in 4m 17s\n- horizon-nodejs20-run-lint https://zuul.opendev.org/t/openstack/build/dac2d70fb27e434e876254b06140dfa1 : SUCCESS in 5m 50s\n- horizon-nodejs20-run-test https://zuul.opendev.org/t/openstack/build/6debb04960c640beb6766056b192d79b : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcbdd537cfa4466299050b4549f8fde0 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bf69173fb1d446a99d6a5c7118e6ee7 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/77184446315546f896a4d77fee1414a9 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567317bfcb9a426bbddaa78bdd29c4d5 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/965ad1013a1b4cd290abe1f7b622dded : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":11},{"id":"a1f067e1803985b8df482e93ca308049a1b4ca5d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-05-31 20:17:03.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"6005a504e3af5183a0cd4a576529e004a5d38fba","author":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"},"date":"2026-08-31 10:27:27.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"a917dd8b548ab28936fe523505ce0f78cd96d5e6","revisions":{"8b3757eebdd1aec5e4b967b02584fc86f4652d59":{"kind":"REWORK","_number":1,"created":"2020-11-20 14:36:47.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/763569/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/1"}}},"commit":{"parents":[{"commit":"6465986d69c7d63f4cb131d1794902c16d37eb7c","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/6465986d69c7d63f4cb131d1794902c16d37eb7c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 14:24:29.000000000","tz":60},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/8b3757eebdd1aec5e4b967b02584fc86f4652d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/8b3757eebdd1aec5e4b967b02584fc86f4652d59"}]},"branch":"refs/heads/master"},"585a6b1f123ff3da1a59eeef8cfebf329273a193":{"kind":"REWORK","_number":2,"created":"2020-11-23 10:00:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/763569/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/2"}}},"commit":{"parents":[{"commit":"f88e32740169007723ceed35ed913a64c0adda56","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/f88e32740169007723ceed35ed913a64c0adda56"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-23 08:41:43.000000000","tz":60},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/585a6b1f123ff3da1a59eeef8cfebf329273a193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/585a6b1f123ff3da1a59eeef8cfebf329273a193"}]},"branch":"refs/heads/master"},"1edd801526000636f4b990eb437c8e0cd2ed7dcc":{"kind":"REWORK","_number":3,"created":"2020-11-24 07:25:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/763569/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/3"}}},"commit":{"parents":[{"commit":"30949aa5d48f32ca2fb30df6c3ce01873bc280f3","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/30949aa5d48f32ca2fb30df6c3ce01873bc280f3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-24 07:22:35.000000000","tz":60},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/1edd801526000636f4b990eb437c8e0cd2ed7dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/1edd801526000636f4b990eb437c8e0cd2ed7dcc"}]},"branch":"refs/heads/master"},"7bf67ec7a1cea2840ae38053682c84300db4e113":{"kind":"REWORK","_number":4,"created":"2020-11-24 17:00:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/763569/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/4"}}},"commit":{"parents":[{"commit":"ec30f434efafb2b029f6f927ed5f54f0335b65d2","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/ec30f434efafb2b029f6f927ed5f54f0335b65d2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-24 16:54:44.000000000","tz":60},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/7bf67ec7a1cea2840ae38053682c84300db4e113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/7bf67ec7a1cea2840ae38053682c84300db4e113"}]},"branch":"refs/heads/master"},"286626dafb6a3d28cd7ad684315783f9b7f27c64":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-01-27 14:35:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/763569/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/5"}}},"commit":{"parents":[{"commit":"0ec4786c1f8633527b4d69fd49efde16786ec35e","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/0ec4786c1f8633527b4d69fd49efde16786ec35e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-27 14:35:15.000000000","tz":60},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/286626dafb6a3d28cd7ad684315783f9b7f27c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/286626dafb6a3d28cd7ad684315783f9b7f27c64"}]},"branch":"refs/heads/master"},"bc043ef1d0f52d60c246c1fb0924793834dc5e69":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-01-29 10:02:36.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/69/763569/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/6"}}},"commit":{"parents":[{"commit":"6ce037518a770486631c132d85caa9006c172195","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/6ce037518a770486631c132d85caa9006c172195"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-01-29 10:02:15.000000000","tz":60},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/bc043ef1d0f52d60c246c1fb0924793834dc5e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/bc043ef1d0f52d60c246c1fb0924793834dc5e69"}]},"branch":"refs/heads/master"},"702a0ea795c940416f1876059b1b927de6169987":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-05-31 19:43:42.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/763569/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/7"}}},"commit":{"parents":[{"commit":"6ce037518a770486631c132d85caa9006c172195","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/6ce037518a770486631c132d85caa9006c172195"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-31 19:43:42.000000000","tz":0},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nStory 2006733\nTask 37206\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/702a0ea795c940416f1876059b1b927de6169987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/702a0ea795c940416f1876059b1b927de6169987"}]},"branch":"refs/heads/master","description":"Edit commit message"},"08b6b333b0305746eee86abf650207fc9b39a789":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-05-31 19:43:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/763569/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/8"}}},"commit":{"parents":[{"commit":"6ce037518a770486631c132d85caa9006c172195","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/6ce037518a770486631c132d85caa9006c172195"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-31 19:43:52.000000000","tz":0},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/08b6b333b0305746eee86abf650207fc9b39a789"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/08b6b333b0305746eee86abf650207fc9b39a789"}]},"branch":"refs/heads/master","description":"Edit commit message"},"26b752393867c85bc1c207d87806df9b4e323481":{"kind":"REWORK","_number":9,"created":"2026-05-31 19:44:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/763569/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/9"}}},"commit":{"parents":[{"commit":"6ce037518a770486631c132d85caa9006c172195","subject":"Allows creation of resources without forcing the creation of children","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/6ce037518a770486631c132d85caa9006c172195"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-31 19:44:20.000000000","tz":0},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/26b752393867c85bc1c207d87806df9b4e323481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/26b752393867c85bc1c207d87806df9b4e323481"}]},"branch":"refs/heads/master"},"9c0d2ca288bb521a3c62b783746f74df96a31796":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2026-05-31 19:44:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/763569/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/10"}}},"commit":{"parents":[{"commit":"b4eaf6c23829476b86507b820711fdf09f13b5dd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/b4eaf6c23829476b86507b820711fdf09f13b5dd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-31 19:44:29.000000000","tz":0},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/9c0d2ca288bb521a3c62b783746f74df96a31796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/9c0d2ca288bb521a3c62b783746f74df96a31796"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"26b752393867c85bc1c207d87806df9b4e323481","theirs":"b4eaf6c23829476b86507b820711fdf09f13b5dd","contains_conflicts":false}},"a917dd8b548ab28936fe523505ce0f78cd96d5e6":{"kind":"REWORK","_number":11,"created":"2026-05-31 19:57:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/763569/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/69/763569/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/69/763569/11"}}},"commit":{"parents":[{"commit":"b4eaf6c23829476b86507b820711fdf09f13b5dd","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/b4eaf6c23829476b86507b820711fdf09f13b5dd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2020-11-20 11:57:47.000000000","tz":60},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-05-31 19:57:36.000000000","tz":0},"subject":"Add TLS versions support for listeners and pools","message":"Add TLS versions support for listeners and pools\n\nChange-Id: I0eea32565ade95d14ecf00faafc17f3dc0c15ac9\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/a917dd8b548ab28936fe523505ce0f78cd96d5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/a917dd8b548ab28936fe523505ce0f78cd96d5e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":38562,"name":"Richard Cruise","email":"rcruise@redhat.com","username":"rcruise"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
