)]}'
{"id":"openstack%2Foctavia-dashboard~876166","triplet_id":"openstack%2Foctavia-dashboard~stable%2F2023.1~I7b504c6374f42655255025154b1476093eda1549","project":"openstack/octavia-dashboard","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I7b504c6374f42655255025154b1476093eda1549","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","status":"MERGED","created":"2023-03-02 15:05:41.000000000","updated":"2023-03-07 10:37:21.000000000","submitted":"2023-03-07 10:35:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876166-create-2023.1","meta_rev_id":"4e27cb16f6aa75fc1f6cdf1c65b437923e67c8b4","_number":876166,"virtual_id_number":876166,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 10:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2023-03-07 10:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-03-07 10:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 15:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-07 10:07:58.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"03d0b686dad4d854cafedb139080f8491a711191","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-03-02 15:05:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81a795f18134d9b445354a6f9f4f7d47ab01aab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:25:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5dce5fe56f406c8c14f22d03082950\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b84b44bcd64c9e8905edbe238a573b : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b050e1ab571645478545f35cfa1185d1 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93ce6ffc785c41ab8045f9053f892923 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8724c7c959654ef1b6e14cc8fc125189 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d7664bd8c85401bb0337228dfe2c86c : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"af058d375c70e54fb2d613b764db886b5a248433","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-03-07 10:07:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b58e841ab8865d2834edd7881e7ce98f775d6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:08:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f1c59f307436f193915e9a7c0995e757e6e2937","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:35:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f46ff797b194dcab88a478e794c2328\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023f89b3cfb54c1baf9982b8ff9b0019 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01127f9044cd48f3893d8b9763ff2459 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e65e4dc23fc41d980d6cb5e3a948292 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97eac76da43b4125975ad23c0195322e : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d2122aa96af479a96221fe8c848f705 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1be959aa4bf0abc7ed75951cc40f59dfaa0ed129","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:35:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e27cb16f6aa75fc1f6cdf1c65b437923e67c8b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 10:37:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0523353c859413d956bbfdd21b61b65\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6704df6b268a4fc68eba63f9e8c39930 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3ee1467ae14fdd91a5cacad390f8d4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1a6dd8995c0fcd7333408eae4f38437f65ef9b8","revisions":{"d1a6dd8995c0fcd7333408eae4f38437f65ef9b8":{"kind":"REWORK","_number":1,"created":"2023-03-02 15:05:41.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/66/876166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-dashboard","ref":"refs/changes/66/876166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/66/876166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/66/876166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-dashboard refs/changes/66/876166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-dashboard refs/changes/66/876166/1"}}},"commit":{"parents":[{"commit":"ce03362aa98e398e2dfa3e3e4f49e22dfbf0c1ea","subject":"Update .gitreview for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/ce03362aa98e398e2dfa3e3e4f49e22dfbf0c1ea"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-02 15:05:38.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-02 15:05:38.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2023.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2023.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I7b504c6374f42655255025154b1476093eda1549\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/d1a6dd8995c0fcd7333408eae4f38437f65ef9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-dashboard/commit/d1a6dd8995c0fcd7333408eae4f38437f65ef9b8"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
