)]}'
{"id":"openstack%2Foctavia-lib~660238","triplet_id":"openstack%2Foctavia-lib~master~Ie1e2532f3aa9dca542ddeee120002e11d0f9f0b1","project":"openstack/octavia-lib","branch":"master","hashtags":[],"change_id":"Ie1e2532f3aa9dca542ddeee120002e11d0f9f0b1","subject":"Add \u0027additional_vips\u0027 field to driver datamodel","status":"MERGED","created":"2019-05-21 01:13:32.000000000","updated":"2019-06-07 22:39:37.000000000","submitted":"2019-06-07 22:39:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660238-1559947177667-b35f9ced","meta_rev_id":"75b4f2639a59e72c51724b2315e90daa00b96845","_number":660238,"virtual_id_number":660238,"owner":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2019-06-07 22:39:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":2,"date":"2019-06-07 11:24:53.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 22:27:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 22:27:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 01:45:33.000000000","updated_by":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"reviewer":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"state":"REVIEWER"},{"updated":"2019-06-06 01:55:30.000000000","updated_by":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"reviewer":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"state":"REVIEWER"},{"updated":"2019-06-07 11:24:53.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-06-07 22:27:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-06-07 22:39:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60f1fbdb99d01293efadc39331bf08db35c772b6","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-05-21 01:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ea0c877acc6a486ce33f96fa01226ddabb81cca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 01:23:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/660238/1/check/openstack-tox-cover/ab5a39e/cover/ : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/660238/1/check/openstack-tox-lower-constraints/e565d22/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/38/660238/1/check/openstack-tox-pep8/36f6005/ : SUCCESS in 4m 09s\n- openstack-tox-py27 http://logs.openstack.org/38/660238/1/check/openstack-tox-py27/6f1a1df/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/38/660238/1/check/openstack-tox-py36/4b8ab3a/ : SUCCESS in 4m 01s\n- openstack-tox-py37 http://logs.openstack.org/38/660238/1/check/openstack-tox-py37/9d6cb44/ : SUCCESS in 4m 23s\n- openstack-tox-docs http://logs.openstack.org/38/660238/1/check/openstack-tox-docs/f3328dc/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e25f3747c4ae32e330e4a4313ca1feaaba785c8f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-05 21:43:39.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"b68a27e6e6202360169392a3c6496b849e581472","author":{"_account_id":25564,"name":"yanpuqing","email":"yanpq@awcloud.com","username":"ycx"},"date":"2019-06-06 01:45:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d95108decd726763cb9aeee2d0d2ac245222b898","author":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"date":"2019-06-06 01:55:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"82f0b8129dc6149abbdbc9ef76cbaf99eef586eb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-06 13:15:00.000000000","message":"Patch Set 1: Code-Review-1\n\nStory 2005608 should be added to commit message and a release note included too.\n\nhttps://storyboard.openstack.org/#!/story/2005608","accounts_in_message":[],"_revision_number":1},{"id":"f74dad123ebd0e2b32bf54a7e9f234d2cf5fd645","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-07 11:07:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bfb1c6a636e354b73b2aed0f8d29dabd307b0aef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:22:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/38/660238/2/check/openstack-tox-cover/9efcf0f/cover/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/660238/2/check/openstack-tox-lower-constraints/c4c3907/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 http://logs.openstack.org/38/660238/2/check/openstack-tox-pep8/f24cca4/ : SUCCESS in 5m 05s\n- openstack-tox-py27 http://logs.openstack.org/38/660238/2/check/openstack-tox-py27/aa4645a/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/38/660238/2/check/openstack-tox-py36/ae53f87/ : SUCCESS in 4m 48s\n- openstack-tox-py37 http://logs.openstack.org/38/660238/2/check/openstack-tox-py37/f02fc61/ : SUCCESS in 4m 30s\n- openstack-tox-docs http://logs.openstack.org/38/660238/2/check/openstack-tox-docs/4e05ebd/html/ : SUCCESS in 5m 04s\n- build-openstack-releasenotes http://logs.openstack.org/38/660238/2/check/build-openstack-releasenotes/6227edc/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"91095f4aab1ea434623f6b05a67f7d848778ae65","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-06-07 11:24:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"327fd0d0a0c0e4216aaabee668f9984b4d0e3a81","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-07 22:27:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"3919f4f7aa363f9c220c77a55deed671d345b1c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 22:27:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d239da6ee3bc15e867817eca9e5cd8fa6ee87a64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 22:39:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"75b4f2639a59e72c51724b2315e90daa00b96845","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 22:39:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/38/660238/2/gate/openstack-tox-lower-constraints/f194af4/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/38/660238/2/gate/openstack-tox-pep8/4490d5f/ : SUCCESS in 4m 06s\n- openstack-tox-py27 http://logs.openstack.org/38/660238/2/gate/openstack-tox-py27/cdba58d/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/38/660238/2/gate/openstack-tox-py36/27a161d/ : SUCCESS in 5m 18s\n- openstack-tox-py37 http://logs.openstack.org/38/660238/2/gate/openstack-tox-py37/ace3657/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/38/660238/2/gate/openstack-tox-docs/ae11d97/html/ : SUCCESS in 4m 00s\n- build-openstack-releasenotes http://logs.openstack.org/38/660238/2/gate/build-openstack-releasenotes/32fce42/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c9cb9dcb9afad092fef9bdcf527a12fdc121695","revisions":{"8c829b48a6e338e918ce652073d73ad5e3add1fd":{"kind":"REWORK","_number":1,"created":"2019-05-21 01:13:32.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/38/660238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/38/660238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/1"}}},"commit":{"parents":[{"commit":"59faa1f718a94fec4563a87145cfa4f9a5ac37e0","subject":"Merge \"Do not install README.rst and LICENSE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/59faa1f718a94fec4563a87145cfa4f9a5ac37e0"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-21 00:56:40.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-21 01:13:20.000000000","tz":-420},"subject":"Add \u0027additional_vips\u0027 field to driver datamodel","message":"Add \u0027additional_vips\u0027 field to driver datamodel\n\nChange-Id: Ie1e2532f3aa9dca542ddeee120002e11d0f9f0b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/8c829b48a6e338e918ce652073d73ad5e3add1fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/8c829b48a6e338e918ce652073d73ad5e3add1fd"}]},"branch":"refs/heads/master"},"7c9cb9dcb9afad092fef9bdcf527a12fdc121695":{"kind":"REWORK","_number":2,"created":"2019-06-07 11:07:37.000000000","uploader":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"ref":"refs/changes/38/660238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/38/660238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/38/660238/2"}}},"commit":{"parents":[{"commit":"59faa1f718a94fec4563a87145cfa4f9a5ac37e0","subject":"Merge \"Do not install README.rst and LICENSE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/59faa1f718a94fec4563a87145cfa4f9a5ac37e0"}]}],"author":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-05-21 00:56:40.000000000","tz":-420},"committer":{"name":"Adam Harwell","email":"flux.adam@gmail.com","date":"2019-06-07 11:05:35.000000000","tz":-420},"subject":"Add \u0027additional_vips\u0027 field to driver datamodel","message":"Add \u0027additional_vips\u0027 field to driver datamodel\n\nStory: 2005608\nTask: 33612\nChange-Id: Ie1e2532f3aa9dca542ddeee120002e11d0f9f0b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/7c9cb9dcb9afad092fef9bdcf527a12fdc121695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/7c9cb9dcb9afad092fef9bdcf527a12fdc121695"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
