)]}'
{"id":"openstack%2Foctavia-lib~665027","triplet_id":"openstack%2Foctavia-lib~master~Ia69d1f61571a1a65dee585037affb317999d7ade","project":"openstack/octavia-lib","branch":"master","hashtags":[],"change_id":"Ia69d1f61571a1a65dee585037affb317999d7ade","subject":"Add get methods to the driver-lib","status":"MERGED","created":"2019-06-12 22:31:29.000000000","updated":"2019-07-30 10:55:11.000000000","submitted":"2019-07-30 10:55:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665027-1564484111171-520616fe","meta_rev_id":"82e2865391993bf726edd3f3938dffe0317f306e","_number":665027,"virtual_id_number":665027,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":2,"date":"2019-07-30 10:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-30 10:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-29 18:56:14.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-30 10:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":-2,"date":"2019-07-29 18:56:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-07 11:28:00.000000000","updated_by":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"reviewer":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"state":"REVIEWER"},{"updated":"2019-07-29 18:56:14.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-30 10:43:33.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-30 10:55:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4df8524815f1a9a3a966c29d97a439daefb128b0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 22:31:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdcf565f3593362117d6c0f89ecbfaba439684bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-12 22:34:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"dd13cdeb85d9d806cb54cf13dcd995590b169d9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 22:57:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/665027/1/check/openstack-tox-cover/d731bbe/ : FAILURE in 6m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/1/check/openstack-tox-lower-constraints/f296f90/ : SUCCESS in 4m 05s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/1/check/openstack-tox-pep8/2c67757/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/1/check/openstack-tox-py27/a846330/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/1/check/openstack-tox-py36/019c3f6/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/1/check/openstack-tox-py37/e3a489c/ : SUCCESS in 4m 39s\n- openstack-tox-docs http://logs.openstack.org/27/665027/1/check/openstack-tox-docs/258761c/html/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1007affa0b1a754ab694c5c053798a1a9921e852","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 00:28:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"46a49564feace4eae4509b16c625b9b30217286d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 00:30:01.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"440afdabf378b3f2ceef929f05165fa324f2517f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 00:41:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/27/665027/2/check/openstack-tox-cover/069064f/ : FAILURE in 4m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/2/check/openstack-tox-lower-constraints/38ddf4f/ : FAILURE in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/2/check/openstack-tox-pep8/1410043/ : FAILURE in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/2/check/openstack-tox-py27/a2ba36e/ : FAILURE in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/2/check/openstack-tox-py36/6e9bacf/ : FAILURE in 4m 39s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/2/check/openstack-tox-py37/f71a131/ : FAILURE in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/27/665027/2/check/openstack-tox-docs/eb5b29b/html/ : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a69f47f3709b191ea2ab913b5cf01d24d38a6c52","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 01:52:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50190b95808b43161b8eeaa7dd50e984019964d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 01:52:51.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"0277a42c9eea324159b767f8682492e81ead32e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 02:09:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/27/665027/3/check/requirements-check/28a48d0/ : FAILURE in 3m 51s\n- openstack-tox-cover http://logs.openstack.org/27/665027/3/check/openstack-tox-cover/db1d040/ : FAILURE in 9m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/3/check/openstack-tox-lower-constraints/8a7b2d8/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/3/check/openstack-tox-pep8/657ac6b/ : SUCCESS in 4m 28s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/3/check/openstack-tox-py27/3b6b7e7/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/3/check/openstack-tox-py36/ff73c08/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/3/check/openstack-tox-py37/6ea84b5/ : SUCCESS in 4m 16s\n- openstack-tox-docs http://logs.openstack.org/27/665027/3/check/openstack-tox-docs/fc56a30/html/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"2fe9341d28946bec2d7281ad4c7bbfa32c9bc825","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 18:41:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"56a6411e6ad035f76296ec71d91b59b422fbc81a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 18:47:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f67c2b99a3d533fa54d433b182758a218ad84bc5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 18:51:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3da22ca1cee85afe7d0c534cbe0894ab0aa83038","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 19:11:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/27/665027/6/check/requirements-check/765c38a/ : FAILURE in 3m 29s\n- openstack-tox-cover http://logs.openstack.org/27/665027/6/check/openstack-tox-cover/8994cf3/cover/ : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/6/check/openstack-tox-lower-constraints/9a8c4eb/ : SUCCESS in 4m 24s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/6/check/openstack-tox-pep8/8b5bbc9/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/6/check/openstack-tox-py27/dc2dc03/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/6/check/openstack-tox-py36/2f7c76d/ : SUCCESS in 4m 14s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/6/check/openstack-tox-py37/721f06d/ : SUCCESS in 10m 33s\n- openstack-tox-docs http://logs.openstack.org/27/665027/6/check/openstack-tox-docs/7e54f70/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes http://logs.openstack.org/27/665027/6/check/build-openstack-releasenotes/cfd2c31/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"135d138624fd3d0de56dee479198d932eed57c99","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 20:18:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"17de30093776af59b4ed3aa187bc2beb7f40ba65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:08:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/27/665027/7/check/requirements-check/1f40b98/ : SUCCESS in 3m 53s\n- openstack-tox-cover http://logs.openstack.org/27/665027/7/check/openstack-tox-cover/9e245fe/cover/ : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/7/check/openstack-tox-lower-constraints/00f375b/ : FAILURE in 10m 57s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/7/check/openstack-tox-pep8/d60a346/ : SUCCESS in 9m 36s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/7/check/openstack-tox-py27/2c3ca95/ : SUCCESS in 9m 54s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/7/check/openstack-tox-py36/595ee67/ : SUCCESS in 4m 03s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/7/check/openstack-tox-py37/6384bbd/ : SUCCESS in 9m 58s\n- openstack-tox-docs http://logs.openstack.org/27/665027/7/check/openstack-tox-docs/1cb29c7/html/ : SUCCESS in 10m 01s\n- build-openstack-releasenotes http://logs.openstack.org/27/665027/7/check/build-openstack-releasenotes/81a3caa/html/ : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":7},{"id":"9c2bb96266edabfa9d9f1a28d262864430d70627","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-20 21:18:44.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a16c97c31a2b3a023d7b818dc71e7fb38e192cba","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-06-20 21:20:22.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"20bc83fdd6fa84d836d31aa83d24f0258557be0d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 21:24:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ccc6ed6b6886a033cc7c6518e3d51db78cdd27b9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-20 21:25:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"169800a0c2fe57ff5b3aced4722c441342f3aa25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:46:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/27/665027/9/check/requirements-check/bdd275c/ : SUCCESS in 4m 48s\n- openstack-tox-cover http://logs.openstack.org/27/665027/9/check/openstack-tox-cover/6677f01/cover/ : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/9/check/openstack-tox-lower-constraints/d1a3f53/ : SUCCESS in 4m 47s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/9/check/openstack-tox-pep8/c609a43/ : SUCCESS in 11m 39s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/9/check/openstack-tox-py27/537f828/ : SUCCESS in 11m 36s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/9/check/openstack-tox-py36/76f1d86/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/9/check/openstack-tox-py37/4091172/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/27/665027/9/check/openstack-tox-docs/9574fcb/html/ : SUCCESS in 8m 01s\n- build-openstack-releasenotes http://logs.openstack.org/27/665027/9/check/build-openstack-releasenotes/7e575c5/html/ : SUCCESS in 11m 31s","accounts_in_message":[],"_revision_number":9},{"id":"171a50f3c9962f600d3eb90d31750f6c0b8ad076","author":{"_account_id":20363,"name":"Adit Sarfaty","email":"asarfaty@vmware.com","username":"asarfaty"},"date":"2019-07-07 11:28:00.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"3d66dc0042c38966e88940d6dd61e5362b40692c","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-26 22:03:21.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"910f4f378bf991da7b02f6a32022b8cb9dbeabce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-26 22:16:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4d4217fbd393ee489cf5d5f28632a55511610890","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-27 09:18:19.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)\n\nDocstring nitpicking, otherwise LGTM.","accounts_in_message":[],"_revision_number":9},{"id":"a3c2016182ae89ced60bb8f26a9646fa1e474302","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 18:30:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9f54068b4f49686766053c3d38c08340ff19b348","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-29 18:31:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"19e68ae61f760784477c7980f9b9c9c83b7069d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:42:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/27/665027/10/check/requirements-check/8e4ec4c/ : SUCCESS in 3m 31s\n- openstack-tox-cover http://logs.openstack.org/27/665027/10/check/openstack-tox-cover/6298742/cover/ : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/10/check/openstack-tox-lower-constraints/8051f02/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/10/check/openstack-tox-pep8/19413ff/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/10/check/openstack-tox-py27/879be22/ : SUCCESS in 3m 10s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/10/check/openstack-tox-py36/724e750/ : SUCCESS in 3m 56s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/10/check/openstack-tox-py37/49d8592/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/27/665027/10/check/openstack-tox-docs/c0436c6/html/ : SUCCESS in 5m 20s\n- build-openstack-releasenotes http://logs.openstack.org/27/665027/10/check/build-openstack-releasenotes/c56f507/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":10},{"id":"7046382f2162781436761291aa7be13cdf2d7beb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-29 18:56:14.000000000","message":"Patch Set 10: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"ee4145b66769585836d99f000fd3a565ee2fbb08","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-30 10:43:33.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f2a2fb586d43569552f6c32057e7d85c5e19becb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 10:43:44.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"5986bcde1f8f265cea931fd60dffbbf9fa346338","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 10:55:10.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/27/665027/10/gate/requirements-check/c067788/ : SUCCESS in 2m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/665027/10/gate/openstack-tox-lower-constraints/c838c1a/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/27/665027/10/gate/openstack-tox-pep8/606cd01/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/27/665027/10/gate/openstack-tox-py27/0a18f7f/ : SUCCESS in 4m 16s\n- openstack-tox-py36 http://logs.openstack.org/27/665027/10/gate/openstack-tox-py36/abd2357/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/27/665027/10/gate/openstack-tox-py37/1446605/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/27/665027/10/gate/openstack-tox-docs/69d8a45/html/ : SUCCESS in 4m 11s\n- build-openstack-releasenotes http://logs.openstack.org/27/665027/10/gate/build-openstack-releasenotes/b15c2b3/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":10},{"id":"82e2865391993bf726edd3f3938dffe0317f306e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 10:55:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"d700c00a90fd62b4f6cb9eb30ebe5f619dd6bfda","revisions":{"cba3027059c4313c7e8e4c75ad3a16713ffefc58":{"kind":"REWORK","_number":1,"created":"2019-06-12 22:31:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/1"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:31:28.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cba3027059c4313c7e8e4c75ad3a16713ffefc58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cba3027059c4313c7e8e4c75ad3a16713ffefc58"}]},"branch":"refs/heads/master"},"dc28cae69c3fd8fb39a732155e99fe7592816042":{"kind":"REWORK","_number":2,"created":"2019-06-20 00:28:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/2"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 00:28:44.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/dc28cae69c3fd8fb39a732155e99fe7592816042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/dc28cae69c3fd8fb39a732155e99fe7592816042"}]},"branch":"refs/heads/master"},"ad08b208af30f5b73e29d0994d2ffc69c221fc00":{"kind":"REWORK","_number":3,"created":"2019-06-20 01:52:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/3"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 01:52:36.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ad08b208af30f5b73e29d0994d2ffc69c221fc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ad08b208af30f5b73e29d0994d2ffc69c221fc00"}]},"branch":"refs/heads/master"},"b4df76c6e29f07175065211e7bb06d6813b93484":{"kind":"REWORK","_number":4,"created":"2019-06-20 18:41:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/4"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 18:40:44.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b4df76c6e29f07175065211e7bb06d6813b93484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b4df76c6e29f07175065211e7bb06d6813b93484"}]},"branch":"refs/heads/master"},"10438eb0719e455e02ffcfb420ffe23c465e2d10":{"kind":"REWORK","_number":5,"created":"2019-06-20 18:47:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/5"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 18:47:30.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/10438eb0719e455e02ffcfb420ffe23c465e2d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/10438eb0719e455e02ffcfb420ffe23c465e2d10"}]},"branch":"refs/heads/master"},"b049fe1e42646d880b54e0e862964ad05edd7318":{"kind":"REWORK","_number":6,"created":"2019-06-20 18:51:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/6"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 18:51:02.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b049fe1e42646d880b54e0e862964ad05edd7318"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b049fe1e42646d880b54e0e862964ad05edd7318"}]},"branch":"refs/heads/master"},"24f0c78fd3787275cb5b953256133232b1abaf30":{"kind":"REWORK","_number":7,"created":"2019-06-20 20:18:24.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/7"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 20:18:19.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/24f0c78fd3787275cb5b953256133232b1abaf30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/24f0c78fd3787275cb5b953256133232b1abaf30"}]},"branch":"refs/heads/master"},"03cd013585f3976dd8b3dbf752b46a9902cb5460":{"kind":"REWORK","_number":8,"created":"2019-06-20 21:24:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/8"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 21:23:57.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/03cd013585f3976dd8b3dbf752b46a9902cb5460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/03cd013585f3976dd8b3dbf752b46a9902cb5460"}]},"branch":"refs/heads/master"},"61c9a832134dcf0e9bac7fa60891a73e5d3e4e72":{"kind":"REWORK","_number":9,"created":"2019-06-20 21:25:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/9"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-20 21:25:32.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/61c9a832134dcf0e9bac7fa60891a73e5d3e4e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/61c9a832134dcf0e9bac7fa60891a73e5d3e4e72"}]},"branch":"refs/heads/master"},"d700c00a90fd62b4f6cb9eb30ebe5f619dd6bfda":{"kind":"REWORK","_number":10,"created":"2019-07-29 18:30:45.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/27/665027/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/27/665027/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/27/665027/10"}}},"commit":{"parents":[{"commit":"307b15e2a3755c70a0e6062bad3bdb79e687c3d7","subject":"Merge \"Add \u0027additional_vips\u0027 field to driver datamodel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/307b15e2a3755c70a0e6062bad3bdb79e687c3d7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-12 22:20:41.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-29 18:29:22.000000000","tz":-420},"subject":"Add get methods to the driver-lib","message":"Add get methods to the driver-lib\n\nThis patch adds get methods to the driver-lib which allows provider\ndrivers to query for objects by ID.\nIn support of the get methods, this patch fixes a bug in the data\nmodel to_dict() where it may not properly recurse the objects.\nIt also improves connecting to the driver agent sockets and adds a\ntimeout while waiting to receive data from the driver-agent.\n\nChange-Id: Ia69d1f61571a1a65dee585037affb317999d7ade\nStory: 2005870\nTask: 33682\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d700c00a90fd62b4f6cb9eb30ebe5f619dd6bfda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d700c00a90fd62b4f6cb9eb30ebe5f619dd6bfda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
