)]}'
{"id":"openstack%2Foctavia-lib~668267","triplet_id":"openstack%2Foctavia-lib~master~I573d10820fdf8e4ce74393b2eaef3cad87667e8e","project":"openstack/octavia-lib","branch":"master","topic":"constraints-updates","hashtags":[],"change_id":"I573d10820fdf8e4ce74393b2eaef3cad87667e8e","subject":"Update tox.ini for new upper constraints strategy","status":"MERGED","created":"2019-06-28 23:19:03.000000000","updated":"2019-07-08 12:41:17.000000000","submitted":"2019-07-08 12:41:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668267-1562589677107-29425bd4","meta_rev_id":"37096f7b3bae933a8f88a122b85af40d4a2e4ee7","_number":668267,"virtual_id_number":668267,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2019-07-08 12:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-05 16:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-08 12:30:11.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":1,"date":"2019-07-08 12:30:11.000000000","_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-05 16:53:50.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-08 12:30:11.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2019-07-08 12:41:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00c4caabe0337a2e14f32533e0b0a1750fedcc72","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-06-28 23:19:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e7ec8aa6362ddd09aa7dc1afc57373b2af3b7b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 23:27:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/67/668267/1/check/openstack-tox-cover/cf2de52/cover/ : SUCCESS in 3m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/67/668267/1/check/openstack-tox-lower-constraints/6aa3be6/ : SUCCESS in 3m 02s\n- openstack-tox-pep8 http://logs.openstack.org/67/668267/1/check/openstack-tox-pep8/813a047/ : SUCCESS in 2m 50s\n- openstack-tox-py27 http://logs.openstack.org/67/668267/1/check/openstack-tox-py27/994b38c/ : SUCCESS in 2m 51s\n- openstack-tox-py36 http://logs.openstack.org/67/668267/1/check/openstack-tox-py36/4b7fe0f/ : SUCCESS in 2m 43s\n- openstack-tox-py37 http://logs.openstack.org/67/668267/1/check/openstack-tox-py37/3099d18/ : SUCCESS in 3m 05s\n- openstack-tox-docs http://logs.openstack.org/67/668267/1/check/openstack-tox-docs/d980f6b/html/ : SUCCESS in 2m 44s\n- build-openstack-releasenotes http://logs.openstack.org/67/668267/1/check/build-openstack-releasenotes/99b215e/html/ : SUCCESS in 2m 32s","accounts_in_message":[],"_revision_number":1},{"id":"94e4d7fbf3d3e465a9fe2e4351a2d9607b5ee1c9","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-05 16:53:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dc8433e9c097c25a0e544388d76afc424cff6b5a","author":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"date":"2019-07-08 12:30:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"2119a705f24fd77b064bcf4a48af710593f2284e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:30:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83857a7b4b53261b7877aa646f9eb8c4f9c5ae40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:41:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/67/668267/1/gate/openstack-tox-lower-constraints/9de8a87/ : SUCCESS in 3m 33s\n- openstack-tox-pep8 http://logs.openstack.org/67/668267/1/gate/openstack-tox-pep8/6d24227/ : SUCCESS in 3m 00s\n- openstack-tox-py27 http://logs.openstack.org/67/668267/1/gate/openstack-tox-py27/6adc1f2/ : SUCCESS in 2m 55s\n- openstack-tox-py36 http://logs.openstack.org/67/668267/1/gate/openstack-tox-py36/b66263f/ : SUCCESS in 2m 48s\n- openstack-tox-py37 http://logs.openstack.org/67/668267/1/gate/openstack-tox-py37/dc84854/ : SUCCESS in 3m 15s\n- openstack-tox-docs http://logs.openstack.org/67/668267/1/gate/openstack-tox-docs/9b56ad8/html/ : SUCCESS in 2m 32s\n- build-openstack-releasenotes http://logs.openstack.org/67/668267/1/gate/build-openstack-releasenotes/14d702c/html/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eabee137f7d237201026b8c42f2fb63c275ab43b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 12:41:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"19fb4f19c90833fadece13a9c00229e00f7f6919","revisions":{"19fb4f19c90833fadece13a9c00229e00f7f6919":{"kind":"REWORK","_number":1,"created":"2019-06-28 23:19:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/67/668267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/67/668267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/67/668267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/67/668267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/67/668267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/67/668267/1"}}},"commit":{"parents":[{"commit":"dd6a93d62c363f95ed5682559e2a776bfe989c47","subject":"Merge \"Add tox \"requirements\" env\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/dd6a93d62c363f95ed5682559e2a776bfe989c47"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 23:18:45.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-06-28 23:19:02.000000000","tz":-420},"subject":"Update tox.ini for new upper constraints strategy","message":"Update tox.ini for new upper constraints strategy\n\nThe requirements team has defined a new strategy for handling\nupper constraints[1]. This patch applies those strategies to\nOctavia-lib.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/ \\\n    2019-May/006478.html\n\nChange-Id: I573d10820fdf8e4ce74393b2eaef3cad87667e8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/19fb4f19c90833fadece13a9c00229e00f7f6919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/19fb4f19c90833fadece13a9c00229e00f7f6919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
