)]}'
{"id":"openstack%2Foctavia-lib~673681","triplet_id":"openstack%2Foctavia-lib~master~I9b601fe2294de2536127a6856f101f4dd0bf4e91","project":"openstack/octavia-lib","branch":"master","hashtags":[],"change_id":"I9b601fe2294de2536127a6856f101f4dd0bf4e91","subject":"Add constants to octavia-lib","status":"MERGED","created":"2019-07-30 20:29:08.000000000","updated":"2019-07-31 23:52:22.000000000","submitted":"2019-07-31 23:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673681-1564617142114-b398d57e","meta_rev_id":"1900985a29ac6c2c50eac9d75de11d407cc65b78","_number":673681,"virtual_id_number":673681,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2019-07-31 23:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-07-31 23:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-07-31 16:32:16.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-07-31 23:41:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 16:32:16.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-07-31 23:41:46.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-07-31 23:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2241332ae673151ca6d39f9f08ab091d1ed6e5dc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 20:29:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"116fd6540e9d54e02d5e1e18ea8adcfe6c130b67","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 20:31:46.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"581cd4806813216d2d39b9083a492f70e7b283c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:42:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/673681/1/check/openstack-tox-cover/86ed06b/cover/ : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/673681/1/check/openstack-tox-lower-constraints/dc5fcaa/ : SUCCESS in 4m 12s\n- openstack-tox-pep8 http://logs.openstack.org/81/673681/1/check/openstack-tox-pep8/18f9b42/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/81/673681/1/check/openstack-tox-py27/44543fe/ : SUCCESS in 3m 54s\n- openstack-tox-py36 http://logs.openstack.org/81/673681/1/check/openstack-tox-py36/9b1f634/ : SUCCESS in 3m 42s\n- openstack-tox-py37 http://logs.openstack.org/81/673681/1/check/openstack-tox-py37/618b95a/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/81/673681/1/check/openstack-tox-docs/cfc9e9a/html/ : SUCCESS in 4m 28s\n- build-openstack-releasenotes http://logs.openstack.org/81/673681/1/check/build-openstack-releasenotes/f91d671/html/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"da03fd9eae9d3e1344ee713c3715c76b5d9496f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 21:42:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37d9471617e50c9e8d49a8c041a2f5b389001dcb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 22:00:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/673681/2/check/openstack-tox-cover/be44af9/cover/ : SUCCESS in 4m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/673681/2/check/openstack-tox-lower-constraints/020679e/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/81/673681/2/check/openstack-tox-pep8/99f5819/ : SUCCESS in 3m 25s\n- openstack-tox-py27 http://logs.openstack.org/81/673681/2/check/openstack-tox-py27/09d8eea/ : SUCCESS in 3m 47s\n- openstack-tox-py36 http://logs.openstack.org/81/673681/2/check/openstack-tox-py36/ee57afd/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/81/673681/2/check/openstack-tox-py37/82457e4/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/81/673681/2/check/openstack-tox-docs/0223bc9/html/ : SUCCESS in 3m 07s\n- build-openstack-releasenotes http://logs.openstack.org/81/673681/2/check/build-openstack-releasenotes/d4d1020/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d144b750d86622e3c57352d86721affed7309c7b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-07-30 23:23:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cec311885a078ee0b5f063dcb12149cfd3b495cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 23:35:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/81/673681/3/check/openstack-tox-cover/7de72d3/cover/ : SUCCESS in 3m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/673681/3/check/openstack-tox-lower-constraints/6725bb9/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/81/673681/3/check/openstack-tox-pep8/5f3a658/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/81/673681/3/check/openstack-tox-py27/880dd04/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/81/673681/3/check/openstack-tox-py36/7559dc2/ : SUCCESS in 2m 51s\n- openstack-tox-py37 http://logs.openstack.org/81/673681/3/check/openstack-tox-py37/56dc95f/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/81/673681/3/check/openstack-tox-docs/b5a9201/html/ : SUCCESS in 4m 26s\n- build-openstack-releasenotes http://logs.openstack.org/81/673681/3/check/build-openstack-releasenotes/bc7c111/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":3},{"id":"daa114cf0357f55f49545a04c79803fc8aae77ea","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-07-31 16:32:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0fafd282ff4dc97274ba9bb2dc571c73636c9776","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-07-31 23:41:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"27d5079de4e0f0364bd2e586d92325bee372ae4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 23:41:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e0217e520547f0661badc31771a4b8dbfa01e589","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 23:52:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://logs.opendev.org/81/673681/3/gate/openstack-tox-lower-constraints/008312b/ : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://logs.opendev.org/81/673681/3/gate/openstack-tox-pep8/4834185/ : SUCCESS in 3m 46s\n- openstack-tox-py27 https://logs.opendev.org/81/673681/3/gate/openstack-tox-py27/3f37eec/ : SUCCESS in 3m 04s\n- openstack-tox-py36 https://logs.opendev.org/81/673681/3/gate/openstack-tox-py36/018aaa2/ : SUCCESS in 4m 03s\n- openstack-tox-py37 https://logs.opendev.org/81/673681/3/gate/openstack-tox-py37/876e202/ : SUCCESS in 4m 18s\n- openstack-tox-docs https://logs.opendev.org/81/673681/3/gate/openstack-tox-docs/a5d3833/html/ : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://logs.opendev.org/81/673681/3/gate/build-openstack-releasenotes/177bbcc/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1900985a29ac6c2c50eac9d75de11d407cc65b78","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 23:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c467bf04858c8e059a9ed8c126083f5980bfbcf5","revisions":{"0b0839d6bb7d7db54b10ba8b64e21ce7bc43ec71":{"kind":"REWORK","_number":1,"created":"2019-07-30 20:29:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/673681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/81/673681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/1"}}},"commit":{"parents":[{"commit":"43aa1c4024c109d4a1d4feff1d6ea04114b729e1","subject":"Merge \"Add get methods to the driver-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/43aa1c4024c109d4a1d4feff1d6ea04114b729e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:26:44.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:29:07.000000000","tz":-420},"subject":"Add constants to octavia-lib","message":"Add constants to octavia-lib\n\nThis patch adds some missing constants to octavia-lib for the\nprovider driver API fields.\n\nChange-Id: I9b601fe2294de2536127a6856f101f4dd0bf4e91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0b0839d6bb7d7db54b10ba8b64e21ce7bc43ec71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0b0839d6bb7d7db54b10ba8b64e21ce7bc43ec71"}]},"branch":"refs/heads/master"},"f62f19c564541e43bc4aaf353d48d5c51f861d4d":{"kind":"REWORK","_number":2,"created":"2019-07-30 21:42:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/673681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/81/673681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/2"}}},"commit":{"parents":[{"commit":"43aa1c4024c109d4a1d4feff1d6ea04114b729e1","subject":"Merge \"Add get methods to the driver-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/43aa1c4024c109d4a1d4feff1d6ea04114b729e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:26:44.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 21:42:35.000000000","tz":-420},"subject":"Add constants to octavia-lib","message":"Add constants to octavia-lib\n\nThis patch adds some missing constants to octavia-lib for the\nprovider driver API fields.\n\nChange-Id: I9b601fe2294de2536127a6856f101f4dd0bf4e91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f62f19c564541e43bc4aaf353d48d5c51f861d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f62f19c564541e43bc4aaf353d48d5c51f861d4d"}]},"branch":"refs/heads/master"},"c467bf04858c8e059a9ed8c126083f5980bfbcf5":{"kind":"REWORK","_number":3,"created":"2019-07-30 23:23:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/673681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/81/673681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/81/673681/3"}}},"commit":{"parents":[{"commit":"43aa1c4024c109d4a1d4feff1d6ea04114b729e1","subject":"Merge \"Add get methods to the driver-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/43aa1c4024c109d4a1d4feff1d6ea04114b729e1"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 20:26:44.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2019-07-30 23:23:25.000000000","tz":-420},"subject":"Add constants to octavia-lib","message":"Add constants to octavia-lib\n\nThis patch adds some missing constants to octavia-lib for the\nprovider driver API fields.\n\nChange-Id: I9b601fe2294de2536127a6856f101f4dd0bf4e91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c467bf04858c8e059a9ed8c126083f5980bfbcf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c467bf04858c8e059a9ed8c126083f5980bfbcf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
