)]}'
{"id":"openstack%2Foctavia-lib~674461","triplet_id":"openstack%2Foctavia-lib~master~I5f42dac1987e0f90860beb12fc842e77cd627820","project":"openstack/octavia-lib","branch":"master","hashtags":[],"change_id":"I5f42dac1987e0f90860beb12fc842e77cd627820","subject":"Update for storyboard","status":"MERGED","created":"2019-08-03 15:32:58.000000000","updated":"2019-08-03 16:08:41.000000000","submitted":"2019-08-03 16:08:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674461-1564848521988-bfcf0b41","meta_rev_id":"333a039c65838b881e286f1b84d8a06d2ad1f0f6","_number":674461,"virtual_id_number":674461,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":2,"date":"2019-08-03 16:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":2,"date":"2019-08-03 15:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-03 15:39:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"all":[{"value":1,"date":"2019-08-03 15:39:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-03 15:39:30.000000000","updated_by":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2019-08-03 15:39:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-08-03 16:08:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad1f21e609f68ecd1d0688e5ea0ef5c81fdf978b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-03 15:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ac96558ba3bcb393137e2cfb3d6e2f01054c78e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 15:37:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674461/1/check/openstack-tox-pep8/2aecbdc/ : SUCCESS in 3m 06s\n- openstack-tox-docs https://logs.opendev.org/61/674461/1/check/openstack-tox-docs/f0d186e/html/ : SUCCESS in 2m 30s\n- build-openstack-releasenotes https://logs.opendev.org/61/674461/1/check/build-openstack-releasenotes/8f76947/html/ : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"86520c0989ef28f1eceeb5606d3b6897b0df5869","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 15:39:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0e6c372e774438e6675fe1a69fbbde1649de4def","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 15:39:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"794c15412e7f1cc28fa467684014ce3a01552e32","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-08-03 15:39:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c16e602de29cb1ebd114dd8e48ddbb6d43d7d244","author":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"date":"2019-08-03 15:40:12.000000000","message":"Patch Set 1:\n\nPushing this now so we can include these trivial fixes in the next release (which is going to be made *immediately* following this merging).","accounts_in_message":[],"_revision_number":1},{"id":"e47e5675ab47ed41f1c977224bfbb4cdd30bde0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 16:08:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"333a039c65838b881e286f1b84d8a06d2ad1f0f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-03 16:08:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/61/674461/1/gate/openstack-tox-pep8/ceedf92/ : SUCCESS in 3m 40s\n- openstack-tox-docs https://logs.opendev.org/61/674461/1/gate/openstack-tox-docs/1bfad5a/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes https://logs.opendev.org/61/674461/1/gate/build-openstack-releasenotes/0b6edc6/html/ : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43e957f9fe42505b421c882b6706d74132aebd7d","revisions":{"43e957f9fe42505b421c882b6706d74132aebd7d":{"kind":"REWORK","_number":1,"created":"2019-08-03 15:32:58.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/61/674461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/61/674461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/61/674461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/61/674461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/61/674461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/61/674461/1"}}},"commit":{"parents":[{"commit":"05b8ca95fd8b3a63ff46e9b584237cf99b32ac69","subject":"Fix docstring and avoid such errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/05b8ca95fd8b3a63ff46e9b584237cf99b32ac69"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-03 15:31:35.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-03 15:31:35.000000000","tz":120},"subject":"Update for storyboard","message":"Update for storyboard\n\nThis repo uses storyboard since some time and not anymore launchpad:\n\nUpdate the openstackdocstheme options so that the \"Report a bug\" link\npoints to storyboard and not to launchpad.\n\nChange-Id: I5f42dac1987e0f90860beb12fc842e77cd627820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/43e957f9fe42505b421c882b6706d74132aebd7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/43e957f9fe42505b421c882b6706d74132aebd7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
