)]}'
{"id":"openstack%2Foctavia-lib~690836","triplet_id":"openstack%2Foctavia-lib~master~I43572045ea3d054c2993b523ee9ea8a6a71faa61","project":"openstack/octavia-lib","branch":"master","topic":"ussuri","hashtags":[],"change_id":"I43572045ea3d054c2993b523ee9ea8a6a71faa61","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-24 06:17:36.000000000","updated":"2019-11-22 17:00:40.000000000","submitted":"2019-11-22 16:56:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690836-1574441817555-de65efff","meta_rev_id":"a9ca36d0bc674cfa17cedf70c23f3ea259f390e4","_number":690836,"virtual_id_number":690836,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 16:56:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2019-10-24 08:17:52.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2019-10-24 14:50:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":1,"date":"2019-10-24 15:28:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-11-22 16:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-11-22 16:44:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":-2,"date":"2019-10-24 08:17:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 08:17:52.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2019-10-24 14:50:59.000000000","updated_by":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"reviewer":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"state":"REVIEWER"},{"updated":"2019-10-24 15:28:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-22 16:44:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2019-11-22 16:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b83e157f21578414b566fd65bc938a360a3fe598","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-24 06:17:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8236bb7e81ba974ec55b60dba5a4c848f386edf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 07:52:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c526b3704c6b4341980b6081fada364c : SUCCESS in 4m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acbdb528eb96411e8a69b293a8872ab6 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4876441112db45ecb3416c394139544e : SUCCESS in 2m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c78c3512eca34f76859adea3d57ffe7e : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e49202dedbf4816a6d9f9371d129043 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e123c1ae73df4e888eb3a9b054ef01f2 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd840b2926144c99df9524ee9269b53 : SUCCESS in 10m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/f34e25cfe9664d948d6f84424adaf129 : SUCCESS in 9m 41s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/392105ae029b4f979a4ac843a61fcebc : SUCCESS in 9m 24s","accounts_in_message":[],"_revision_number":1},{"id":"4cab9db29580c1de016369d7e700036ff2846dc5","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2019-10-24 08:17:52.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64a2721713ee01c5f70d9548f0dba4644444758b","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-24 14:50:59.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"f09d5bb2ba9113e72890c945a4ae768b26690d95","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-24 15:28:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0263de23551e38719eefb7ace5a4a4dc02ee0f86","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2019-11-22 16:44:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"c4be9f85ae95ab74a48ec25675d577aab6368b48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:46:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8de5a8eb26e09511af087d2b6e4428539618e4b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:56:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd9c42aeee2e45e493ddd8711a0cc3dd : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1dbf24f18b5475da895f11d319a2c9d : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d12a339c525e4fe09993575aee4dce9e : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33ce22689ae74378abef2264679bd653 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9066495fa3f41e39bc6e7481a0bc763 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cb9416e06e49eab1d1efeb19973586 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"72965a2f2587d771a46cc2824af82101de439359","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:56:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"912de13be9eefa3de13de65e922aa2c85084e122","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 17:00:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d964d82c293948c095702f2b93e8adb9 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa68dd2c12c238658f27f802c4052b8266748a40","revisions":{"fa68dd2c12c238658f27f802c4052b8266748a40":{"kind":"REWORK","_number":1,"created":"2019-10-24 06:17:36.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/36/690836/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/36/690836/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/36/690836/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/36/690836/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/36/690836/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/36/690836/1"}}},"commit":{"parents":[{"commit":"079c86613997ac3c2d10e0447c4f305c7d8e82a0","subject":"Merge \"Generate PDF documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/079c86613997ac3c2d10e0447c4f305c7d8e82a0"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-24 06:15:43.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-24 06:15:43.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I43572045ea3d054c2993b523ee9ea8a6a71faa61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/fa68dd2c12c238658f27f802c4052b8266748a40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/fa68dd2c12c238658f27f802c4052b8266748a40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
