)]}'
{"id":"openstack%2Foctavia-lib~692077","triplet_id":"openstack%2Foctavia-lib~master~I412626410710eb4778ce00feb067640a6e725c47","project":"openstack/octavia-lib","branch":"master","topic":"692077","hashtags":[],"change_id":"I412626410710eb4778ce00feb067640a6e725c47","subject":"OpenStack is dropping the py2.7 support in ussuri cycle.","status":"ABANDONED","created":"2019-10-30 08:49:35.000000000","updated":"2019-11-28 16:45:29.000000000","total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"05815c165e5b18e89883526541d9fa5c4f2508f0","_number":692077,"virtual_id_number":692077,"owner":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"date":"2019-11-19 15:32:25.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-11-19 07:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 11:23:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-19 07:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-19 15:32:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"4f5dff9e2d9f635f00b4ea8002c9cc2919066913","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-30 08:49:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fe8f55d4583318517b1042ba2ff48d7804614be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:38:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11def5298fa64d4cb420cd2d741fc59b : SUCCESS in 3m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dd939b901c234c40b30c828849197463 : FAILURE in 3m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0fca8080a144d5c92ba3ecf547ec529 : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb507e4874a843229d99b4b97d5c756e : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65d46f0a56714a2e8cf2c7bcd4969cbd : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd04ebf9fc17419daea6b76703267d44 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4be6d12b6c41aa8d44747e088f9b9d : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/889a42175eca4186bd5789beade2a58c : FAILURE in 3m 11s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/08d9739458d24a31a2fe062a654eb0f5 : SUCCESS in 4m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/b8b3afa5bf254bf5a36eb6b0d363bca2 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"63dfe9be99cb0d3ee5aedc7c1ca9accf7fb55ca1","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-30 11:11:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"91eb3903e0552b60c62a422814bb7a93e9010f2e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-30 11:23:00.000000000","message":"Patch Set 1: Backport-Candidate-1\n\nI guess the code is still supporting py27 so if you want to remove the py27 support you need to remove all the compatibility first\n\nhttps://github.com/openstack/octavia-lib/blob/master/requirements.txt#L8\n\nhttps://github.com/openstack/octavia-lib/search?utf8\u003d%E2%9C%93\u0026q\u003dsix\u0026type\u003d","accounts_in_message":[],"_revision_number":1},{"id":"8287c811541c7913ac905f31aba1d3e8fa8ff9cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-30 14:30:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5b52ec857b7ff6bdef5ac17b63d57a2a1f8100f5","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-10-31 04:48:33.000000000","message":"Patch Set 1: Backport-Candidate-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9e60f0e476c0df2d5c76071fd2312282f0c390dc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-31 13:25:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6f90399ab9ae805033397d4de259a75e85e8101","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-08 05:31:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4f9f74cbf798c7b8e5a0ae9866224f8cdf42d5f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 05:39:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2ff624afc8a6481d8fb41b631f27e922 : SUCCESS in 2m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6c384305b88d46eb8feeee9ca69ce895 : FAILURE in 2m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adcf217234fc4811b7fc60fc7f89a866 : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc11602350714db383bb4292b2e94764 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4967d83054d4c3fa619d85030be3044 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8b2a3f9edf50480cb94df9393dd4c97b : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0feb6c3ba84c949d172e101dd44e23 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4193ab481048445e9d1c25907c8892e5 : FAILURE in 2m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ab68ee14cf574589ab85986c16f8fc73 : SUCCESS in 4m 34s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4c40075b241e4c9b8335fab4a3adcc8e : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"657008191f6fd2c317687f7059cc94fbc85b7bef","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-19 07:26:03.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7b4d588d3bcc99210b9f2fa2a9da06d4671c91ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 07:44:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd885d4e0eda43248a9ea5f40168b286 : SUCCESS in 1m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cec48501bcdd45babe5e9d309a6a4ba2 : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/059e2eb3b0b94243841fcebe763072b0 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/383884761e014cf4b449d3331a323ef8 : SUCCESS in 2m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0194ed6e56854262882bd0ab35508f91 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1daa8ed36f641b6b6b325888020fc70 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/604221141f604015b1ab3247558dccf7 : SUCCESS in 2m 44s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/397f2945ab744fbc8dfdeeb6ac5feee4 : SUCCESS in 5m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fb1dd88612e647d8bc26dcdaf465ff16 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":3},{"id":"5276153f3541178bd3a395905d85a9f9edd64b05","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 15:32:25.000000000","message":"Patch Set 3:\n\nHi Ajay,\n\nduring the PTG I started a new \u0027remove python2\u0027 change for octavia-lib, completely forgetting about this one.  It\u0027s a little more complete:\n\nhttps://review.opendev.org/#/c/693132/","accounts_in_message":[],"_revision_number":3},{"id":"e7194a984cdd39ac4f8f917beee0acb071ce6f8e","author":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"date":"2019-11-28 16:45:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3d1166a5a5106ee159e8e48886aac449b86234fa","revisions":{"053c141cb25a89a303244a5e3bac312cb910d83d":{"kind":"REWORK","_number":1,"created":"2019-10-30 08:49:35.000000000","uploader":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"ref":"refs/changes/77/692077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/77/692077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/1"}}},"commit":{"parents":[{"commit":"3c6b639b685558acc225357546101094a97a0e46","subject":"Batch member update needs pool_id explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3c6b639b685558acc225357546101094a97a0e46"}]}],"author":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-10-30 08:48:10.000000000","tz":330},"committer":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-10-30 08:49:31.000000000","tz":330},"subject":"OpenStack is dropping the py2.7 support in ussuri cycle.","message":"OpenStack is dropping the py2.7 support in ussuri cycle.\n\noctavia-lib is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: I412626410710eb4778ce00feb067640a6e725c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/053c141cb25a89a303244a5e3bac312cb910d83d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/053c141cb25a89a303244a5e3bac312cb910d83d"}]},"branch":"refs/heads/master"},"495b4f731df0380a7dbb7520f6d2baaa494a5401":{"kind":"REWORK","_number":2,"created":"2019-11-08 05:31:28.000000000","uploader":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"ref":"refs/changes/77/692077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/77/692077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/2"}}},"commit":{"parents":[{"commit":"3c6b639b685558acc225357546101094a97a0e46","subject":"Batch member update needs pool_id explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3c6b639b685558acc225357546101094a97a0e46"}]}],"author":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-10-30 08:48:10.000000000","tz":330},"committer":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-11-08 05:29:24.000000000","tz":330},"subject":"OpenStack is dropping the py2.7 support in ussuri cycle.","message":"OpenStack is dropping the py2.7 support in ussuri cycle.\n\noctavia-lib is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: I412626410710eb4778ce00feb067640a6e725c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/495b4f731df0380a7dbb7520f6d2baaa494a5401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/495b4f731df0380a7dbb7520f6d2baaa494a5401"}]},"branch":"refs/heads/master"},"3d1166a5a5106ee159e8e48886aac449b86234fa":{"kind":"REWORK","_number":3,"created":"2019-11-19 07:26:03.000000000","uploader":{"_account_id":30619,"name":"Ajay Kumar","email":"mahajanajay33@gmail.com","username":"ajaykumar33"},"ref":"refs/changes/77/692077/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/77/692077/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/77/692077/3"}}},"commit":{"parents":[{"commit":"3c6b639b685558acc225357546101094a97a0e46","subject":"Batch member update needs pool_id explicitly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3c6b639b685558acc225357546101094a97a0e46"}]}],"author":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-10-30 08:48:10.000000000","tz":330},"committer":{"name":"Ajay Kumar","email":"ajay.k@india.nec.com","date":"2019-11-19 07:25:47.000000000","tz":0},"subject":"OpenStack is dropping the py2.7 support in ussuri cycle.","message":"OpenStack is dropping the py2.7 support in ussuri cycle.\n\noctavia-lib is ready with python 3 and ok to drop the\npython 2.7 support.\n\nComplete discussion \u0026 schedule can be found in\n- http://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal - https://review.opendev.org/#/c/691178/\n\nChange-Id: I412626410710eb4778ce00feb067640a6e725c47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3d1166a5a5106ee159e8e48886aac449b86234fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3d1166a5a5106ee159e8e48886aac449b86234fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
