)]}'
{"id":"openstack%2Foctavia-lib~704054","triplet_id":"openstack%2Foctavia-lib~master~Ib057c917b5703bc461630cab394fa36c882d6ea9","project":"openstack/octavia-lib","branch":"master","topic":"stop_configuring_install_command","hashtags":[],"change_id":"Ib057c917b5703bc461630cab394fa36c882d6ea9","subject":"Stop configuring install_command in tox.","status":"MERGED","created":"2020-01-23 19:13:21.000000000","updated":"2021-03-03 15:43:32.000000000","submitted":"2021-03-03 15:34:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704054-stop_configuring_install_command","meta_rev_id":"ceb5dee981f40585806704d46973ac114f681fd3","_number":704054,"virtual_id_number":704054,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-03 15:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2021-03-03 15:20:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-26 18:01:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2021-03-03 15:20:42.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-14 18:29:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-26 18:01:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-03-03 15:20:42.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"}],"messages":[{"id":"35835a9ccf9e6c47aab997c2577b790cca99c0e2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-01-23 19:13:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e313484a2828df316dea577e8fb9677e0826ece5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 19:33:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3be84aaa37ef4fcab7a99ebece6cd0ee : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf6dcc1b5cc4a938cb66cc9d251c76c : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ddd0409aec42f5b6d76b7498463c3b : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c264fd5c65cd43a19bffe8d3ec2442f8 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6710745a95d4d05b3b3b738241af1c0 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76272ddb0ec94665b3d117c8e0b19e23 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2f8cf211b049c79bc6f05fa0e5bbfe : SUCCESS in 5m 54s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/023f88d44d094ab8b18e1ae36813f76f : SUCCESS in 6m 43s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/fe966a7134734160a30f2b6dfac5b649 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e3acefd6c78e18da6601af849f83a95e6f4eeaad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-23 19:46:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff7bdf1ecca22f06a948bdcce774c4e20f507dc2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-10-14 18:13:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0c951ca80f662912b7d04c9859e814b2cf616e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-14 18:29:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/978498c7959f4312b4fe2ffda587db33 : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e59002f7fc52438f9e0f16c9f0811db6 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b16f1709161249e68213f8b265490b1c : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e611585c06f4c9dbd91b1a69edb5e0a : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f4e39fb0f934b8da2c9a0c42a58e414 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcab809fa7064fc9b758b472266dcbd1 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ceafcf0c6e0044f69122dd5703b09d0e : SUCCESS in 3m 45s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d4d6ac837a814d179147a7a6c5244a78 : SUCCESS in 5m 47s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/106e96f29b17455bbfcf866bc0c87f19 : SUCCESS in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b9ac848a4d46f64909298afc3c4785c4a38036b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-10-26 18:01:27.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"70081a893e7cb201addd1e35d6b7e9195a8ee2d3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2021-03-03 15:20:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"72851da735a1375c88f40e3f82f82b2843ab9153","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:21:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76a2e716caf0d8e47a5e0a41d12de458429d6efc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:34:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5193b2941ff429eb4eec49fd6e4ff0d : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee19a75ac3b4f5bb3fac70688fd9c0c : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d2fe26496024e08bd1484b72583ed72 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db318ea819344ab9bfe5f41fb64f3829 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94539fa2c5024da69d29382c339363aa : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9156e07e9cb747f2a750844f989a8501 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0a20d2e723783f9562dfbf57918b79bbef8b9b8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:34:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ceb5dee981f40585806704d46973ac114f681fd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:43:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4212eae41b6e478a9e2cfab41c739808 : SUCCESS in 1m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14bdd3f0aba04cf993b3eb0305b7c688 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de91d1e1ed297e6ec172b1c9453c3694c1b47fea","revisions":{"466191f39b4f7a3ef3f6da837931f0ee38285e7c":{"kind":"REWORK","_number":1,"created":"2020-01-23 19:13:21.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/54/704054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/54/704054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/1"}}},"commit":{"parents":[{"commit":"36902d90a4db29929dd7e6d292b76561e9782d35","subject":"Fix flake8 tox.ini directive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/36902d90a4db29929dd7e6d292b76561e9782d35"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-23 19:12:49.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-23 19:12:49.000000000","tz":-300},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: Ib057c917b5703bc461630cab394fa36c882d6ea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/466191f39b4f7a3ef3f6da837931f0ee38285e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/466191f39b4f7a3ef3f6da837931f0ee38285e7c"}]},"branch":"refs/heads/master"},"de91d1e1ed297e6ec172b1c9453c3694c1b47fea":{"kind":"REWORK","_number":2,"created":"2020-10-14 18:13:22.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/54/704054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/54/704054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/54/704054/2"}}},"commit":{"parents":[{"commit":"21170fd070b565898530cc70660b44e7eab91a24","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/21170fd070b565898530cc70660b44e7eab91a24"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-01-23 19:12:49.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-10-14 18:13:14.000000000","tz":-240},"subject":"Stop configuring install_command in tox.","message":"Stop configuring install_command in tox.\n\nCurrently, we are overriding \u0027install_command\u0027 to use \u0027pip\u0027. This is\nconsidered poor behavior and \u0027python -m pip\u0027 should be used instead:\n\nhttps://snarky.ca/why-you-should-use-python-m-pip/\n\nIt turns out that this is the the default value provided by tox:\n\nhttps://tox.readthedocs.io/en/latest/config.html#conf-install_command\n\nSo we can remove the line and simply use the default value.\n\nChange-Id: Ib057c917b5703bc461630cab394fa36c882d6ea9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/de91d1e1ed297e6ec172b1c9453c3694c1b47fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/de91d1e1ed297e6ec172b1c9453c3694c1b47fea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
