)]}'
{"id":"openstack%2Foctavia-lib~715329","triplet_id":"openstack%2Foctavia-lib~master~Ie54640201a0f4905d19841af5913972646b317bf","project":"openstack/octavia-lib","branch":"master","topic":"steven","hashtags":[],"change_id":"Ie54640201a0f4905d19841af5913972646b317bf","subject":"Add TLS protocols for listener and pool model","status":"MERGED","created":"2020-03-26 23:41:02.000000000","updated":"2020-04-01 19:43:52.000000000","submitted":"2020-04-01 19:42:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715329-1585770128445-9820b1aa","meta_rev_id":"b33069467942a0477060329cb5666cdd0cf6b344","_number":715329,"virtual_id_number":715329,"owner":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},{"value":0,"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-01 19:42:08.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":2,"date":"2020-04-01 19:17:01.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},{"value":0,"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-01 19:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},{"value":0,"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-01 19:33:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":0,"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},{"value":-2,"date":"2020-04-01 19:17:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},{"value":0,"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-26 23:41:02.000000000","updated_by":{"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},"reviewer":{"_account_id":31560,"name":"Steven Glasford","email":"stevenglasford@gmail.com","username":"stevenglasford"},"state":"REVIEWER"},{"updated":"2020-03-27 20:05:29.000000000","updated_by":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"reviewer":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"state":"REVIEWER"},{"updated":"2020-04-01 19:17:01.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-04-01 19:33:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-04-01 19:42:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1fa89ebbd751c085fdf76263815c713dbf2c8e0","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-26 23:41:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e08839140e9048969ae176d316e88b36554eb849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 23:52:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7251f6216b4eb38f6120566bacb743 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc895925fff14061bd3fe668927aa11e : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728a08dc772d4c30ac7a2ff9eca676e9 : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de9e9def0865448886b4afbf20353baf : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/404f0c9f096e4cb7a18c9f8601f0881d : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f1080ce6c1d44d8bacd1981661b3633 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7885b2a26361479690f3e0b3c2f0970c : SUCCESS in 5m 35s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/74b77cadb9084bb7a1921d7a14dfa220 : SUCCESS in 5m 21s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/5aace54958304805bac5a259169f9f97 : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":1},{"id":"271898f02bfede5ca1ba0f5015354dee27c9f702","author":{"_account_id":31568,"name":"Dawson Coleman","email":"dawsonmcoleman@gmail.com","username":"dawson"},"date":"2020-03-27 20:05:29.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nOther than SSLv3 stuff, looks good to me","accounts_in_message":[],"_revision_number":1},{"id":"e69da9dff5061fba6222d00b099492a753e9ac52","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-29 02:05:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38cc8589fc261f0c88d4027fb25745912f85e772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-29 02:17:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fcb6cfc49634b0db29591c4cc11d58b : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41242279637a4519993da961846c14d5 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/595f5a5bbd9045479aa48d84a24b6974 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf9273c0c2c4c46ac65c446c9935b68 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60a0294facb14f9e80c3f32d75e834fa : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d10acabb761249c992f368153a4a42cf : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62093c2cb5df4905b0a21d8509e49ffa : SUCCESS in 6m 37s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/45e41b53ed7b45f6882d52b4635687fc : SUCCESS in 5m 20s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1bcbe11cf6d34e2491af97e724e8919a : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":2},{"id":"8540acdd6d5ee20c9bd26be72f5f4d84fa8e1a18","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-30 04:18:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"61212798b4c53d97c0250461a6a9e6316bd54536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 04:29:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afd0fcc136db4b77b6a34dbf78a04e3b : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5684623aa64aad94ca960140366378 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925bcf57c23e4c46b06e6c872167918b : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1e0ac2609b04478a42c7fdaa554e0cc : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23537ac6005d40eabea09e09ebaa7af0 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e7b3aecb18741ff974e1b8c0ef94a15 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd536d0493343fba353aed9f766390f : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb5508ffd151473bba22f1ab38ff417e : SUCCESS in 4m 03s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c86cd7928c64465393d4e4b7b7f1d9e3 : SUCCESS in 5m 30s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8ec828c15c7a4e7ea1283642cdc2e166 : FAILURE in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"17a845db5d5538273b0fdff8e4216433a5726453","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-03-30 04:52:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d4f007ecfe12d17080c36283205e64e918408e03","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-31 03:21:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f51347af16bc4ff05c1001f2a5473bf30548ebff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 03:33:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d74cbd36c394b7c8d764f52514d2fab : SUCCESS in 3m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/738d57966b0d49fd84ec393fb66746be : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe46c5385bc0476a96f78053315cf738 : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d41d08cc12624ad8851b2c492a176373 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a286a5806b8443e3a2bc65d7747eb470 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd244d13a17483e9a2d3437995f0382 : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2db4f6928f514c989386d2439eb8a47b : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48bcd8b082384aa9b0b2da7203f96661 : SUCCESS in 4m 39s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/823a192a2e5a4dd1883801374562b064 : SUCCESS in 4m 49s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f811971a06464a0ebb83edf73a20ccfa : FAILURE in 10m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bd7b2266e9f2050603636e5cc42b89962ae85b7e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-03-31 17:44:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5df48a3c5234dafa5beb1f1011c05ef35aa1115c","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-03-31 23:38:24.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nWe had the same thought. But there weren\u0027t any instances of using the constants in the test_data_models.py, so we didn\u0027t know if it was appropriate. There are hard coded \u0027l7policy\u0027, \u0027listener_id\u0027, etc.\n\nWe might be misunderstanding though. Should we use the constants in ref_listener and ref_pool, but not ref_listener_dict nor ref_pool_dict?","accounts_in_message":[],"_revision_number":4},{"id":"179747e1423d4e853e0a20e20c4fe75b6bceae0c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-01 07:54:27.000000000","message":"Patch Set 4: Code-Review+2\n\nWe should be using constants as much as possible everywhere, including in ref_listener, ref_pool, ref_listener_dict and ref_pool_dict. But I get why you did not use them -- constants are not used anywhere in test_data_models.py today. So, I am okay with this as is. At least we are consistent :) Great job, folks!","accounts_in_message":[],"_revision_number":4},{"id":"fcf2078e942d86aacd5408b4c90ca5d89ab05a4e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 16:31:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0778e7cfb74ba5a93cff44db3d53b18a65789989","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 17:05:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f180e5848fb5d5536e80a1e15857b58ff189fbfa","author":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"date":"2020-04-01 19:11:33.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8dda4d4ec74247992b5e3de238afa8a9e2836221","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-04-01 19:17:01.000000000","message":"Patch Set 5: Backport-Candidate-2 Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":5},{"id":"42fea6511de20e9eca70d21bffd1a4db3735a3a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:21:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66b373b49fa042ac9b821e734b40df02 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0813a101a75b469f90240cb916bbb7d4 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8fb2e2ff534623a3c9fc9833608a92 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62bd0deb9f184b29a4b833948eeab984 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28f802237afc40b098b78ae53ff87c9e : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e82301047dff455f93dc80b0497db30d : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84460564372a44ebafa2c30c0c3c53d4 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a32a85f9c6040a1bf6f34816db22f3a : SUCCESS in 4m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8e59c18b3d9642cd86e50effa6419416 : SUCCESS in 5m 54s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dbfae3bb1de4494e843fe508bde395a8 : FAILURE in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9c97b286ba24f9bc985f6386150132d3074b7e26","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-04-01 19:33:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":5},{"id":"0ac544bf4248d0f857aee95994823a5c3b63c3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:33:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e8995d05591d362d461ef9cb5d273468076703f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:42:08.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec544daa039b4082988707cb574e463c : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51acfafd33224bd4a2320e4949bbf137 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae0a441205084eae8d8a15af52b741da : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc0d2bf5aa7b4a1dbd300219d4143a5b : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239fefcf36c84b03bfde8e6b9c81c241 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2948cfa60b354ffc8dba45ad81454d9f : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":5},{"id":"79768d6fad907a33eed2b6e22a75bd88e49997f7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:42:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f9ca4b372c5b2d22a1afbe53411d611186b864b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:43:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/002786666dd144b19a6b2dacd73ef818 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbdcbd1ab7ad42498d65c33b2b3bb142 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6bec133036498ca76969905b69c18892fccd79ec","revisions":{"bd183dccb4d77f60af2afb1326796cb25a789b8e":{"kind":"REWORK","_number":1,"created":"2020-03-26 23:41:02.000000000","uploader":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"ref":"refs/changes/29/715329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/29/715329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/1"}}},"commit":{"parents":[{"commit":"c565e7b2897514f3a7ab76018d8772c8e9ffe1c4","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c565e7b2897514f3a7ab76018d8772c8e9ffe1c4"}]}],"author":{"name":"Steven Glasford","email":"stevenglasford@gmail.com","date":"2020-03-26 23:25:57.000000000","tz":-300},"committer":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-03-26 23:39:51.000000000","tz":-300},"subject":"Add TLS protocol support for listener model","message":"Add TLS protocol support for listener model\n\nParameter added to the listener model for TLS protocols.\n\nChange-Id: Ie54640201a0f4905d19841af5913972646b317bf\nCo-authored-by: Steven Glasford \u003cstevenglasford@gmail.com\u003e\nStory: 2006733\nTask: 37186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bd183dccb4d77f60af2afb1326796cb25a789b8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bd183dccb4d77f60af2afb1326796cb25a789b8e"}]},"branch":"refs/heads/master"},"0f2024016c463abd73926d61ac0c3442c909711c":{"kind":"REWORK","_number":2,"created":"2020-03-29 02:05:59.000000000","uploader":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"ref":"refs/changes/29/715329/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/29/715329/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/2"}}},"commit":{"parents":[{"commit":"c565e7b2897514f3a7ab76018d8772c8e9ffe1c4","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c565e7b2897514f3a7ab76018d8772c8e9ffe1c4"}]}],"author":{"name":"Steven Glasford","email":"stevenglasford@gmail.com","date":"2020-03-26 23:25:57.000000000","tz":-300},"committer":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-03-29 02:04:06.000000000","tz":-300},"subject":"Add TLS protocols for listener and pool model","message":"Add TLS protocols for listener and pool model\n\nParameter added to the listener and pool model for TLS protocols.\n\nChange-Id: Ie54640201a0f4905d19841af5913972646b317bf\nCo-authored-by: Steven Glasford \u003cstevenglasford@gmail.com\u003e\nStory: 2006733\nTask: 37186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0f2024016c463abd73926d61ac0c3442c909711c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/0f2024016c463abd73926d61ac0c3442c909711c"}]},"branch":"refs/heads/master"},"f47446611ff6dc652a0d2a8ddf0a3009db03728f":{"kind":"REWORK","_number":3,"created":"2020-03-30 04:18:19.000000000","uploader":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"ref":"refs/changes/29/715329/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/29/715329/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/3"}}},"commit":{"parents":[{"commit":"c565e7b2897514f3a7ab76018d8772c8e9ffe1c4","subject":"Remove the dependency on the \"mock\" package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/c565e7b2897514f3a7ab76018d8772c8e9ffe1c4"}]}],"author":{"name":"Steven Glasford","email":"stevenglasford@gmail.com","date":"2020-03-26 23:25:57.000000000","tz":-300},"committer":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-03-30 04:18:05.000000000","tz":-300},"subject":"Add TLS protocols for listener and pool model","message":"Add TLS protocols for listener and pool model\n\nParameter added to the listener and pool model for TLS protocols.\n\nChange-Id: Ie54640201a0f4905d19841af5913972646b317bf\nCo-authored-by: Steven Glasford \u003cstevenglasford@gmail.com\u003e\nStory: 2006733\nTask: 37186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f47446611ff6dc652a0d2a8ddf0a3009db03728f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f47446611ff6dc652a0d2a8ddf0a3009db03728f"}]},"branch":"refs/heads/master"},"f54768b3110b693d44538ba48f61db12061b3ba5":{"kind":"REWORK","_number":4,"created":"2020-03-31 03:21:52.000000000","uploader":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"ref":"refs/changes/29/715329/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/29/715329/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/4"}}},"commit":{"parents":[{"commit":"3e1278391fcfb7d14b651e2fdb380dae35cd28cd","subject":"Adding cipher list Support for provider drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3e1278391fcfb7d14b651e2fdb380dae35cd28cd"}]}],"author":{"name":"Steven Glasford","email":"stevenglasford@gmail.com","date":"2020-03-26 23:25:57.000000000","tz":-300},"committer":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-03-31 03:21:24.000000000","tz":-300},"subject":"Add TLS protocols for listener and pool model","message":"Add TLS protocols for listener and pool model\n\nParameter added to the listener and pool model for TLS protocols.\nConstants TLS_VERSIONS, SSL_VERSION_3, TLS_VERSION_1,\nTLS_VERSION_1_1, TLS_VERSION_1_2, and TLS_VERSION_1_3 were added.\n\nChange-Id: Ie54640201a0f4905d19841af5913972646b317bf\nCo-authored-by: Steven Glasford \u003cstevenglasford@gmail.com\u003e\nStory: 2006733\nTask: 37186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f54768b3110b693d44538ba48f61db12061b3ba5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f54768b3110b693d44538ba48f61db12061b3ba5"}]},"branch":"refs/heads/master"},"6bec133036498ca76969905b69c18892fccd79ec":{"kind":"REWORK","_number":5,"created":"2020-04-01 19:11:33.000000000","uploader":{"_account_id":31563,"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","username":"luketollefson"},"ref":"refs/changes/29/715329/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/29/715329/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/29/715329/5"}}},"commit":{"parents":[{"commit":"3e1278391fcfb7d14b651e2fdb380dae35cd28cd","subject":"Adding cipher list Support for provider drivers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3e1278391fcfb7d14b651e2fdb380dae35cd28cd"}]}],"author":{"name":"Steven Glasford","email":"stevenglasford@gmail.com","date":"2020-03-26 23:25:57.000000000","tz":-300},"committer":{"name":"Luke Tollefson","email":"lukewtollefson@gmail.com","date":"2020-04-01 19:11:21.000000000","tz":-300},"subject":"Add TLS protocols for listener and pool model","message":"Add TLS protocols for listener and pool model\n\nParameter added to the listener and pool model for TLS protocols.\nConstants TLS_VERSIONS, SSL_VERSION_3, TLS_VERSION_1,\nTLS_VERSION_1_1, TLS_VERSION_1_2, and TLS_VERSION_1_3 were added.\n\nChange-Id: Ie54640201a0f4905d19841af5913972646b317bf\nCo-authored-by: Steven Glasford \u003cstevenglasford@gmail.com\u003e\nStory: 2006733\nTask: 37186\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/6bec133036498ca76969905b69c18892fccd79ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/6bec133036498ca76969905b69c18892fccd79ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
