)]}'
{"id":"openstack%2Foctavia-lib~741455","triplet_id":"openstack%2Foctavia-lib~master~I7ebdd5eea537a8258f66b5e1bf2ee1aad3f8b10b","project":"openstack/octavia-lib","branch":"master","topic":"update-test-req","hashtags":[],"change_id":"I7ebdd5eea537a8258f66b5e1bf2ee1aad3f8b10b","subject":"Update test-requirements to match Octavia repo","status":"MERGED","created":"2020-07-16 13:51:58.000000000","updated":"2020-07-16 15:03:39.000000000","submitted":"2020-07-16 15:01:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741455-1594911717402-bdb66bb5","meta_rev_id":"0e8cdb2a60aeb65aa3134d02ec1f4d4c93494ce4","_number":741455,"virtual_id_number":741455,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-16 15:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":2,"date":"2020-07-16 14:45:57.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":2,"date":"2020-07-16 14:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"all":[{"value":1,"date":"2020-07-16 14:46:02.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 14:45:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-07-16 14:46:02.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-07-16 15:01:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"986f83cb939c1b35989934ca38f196f94119c244","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 13:51:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1c11a366ecd1c4558012d03ad284bc361bfb17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:11:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e895e81e262c4be8a201cc471568d6f2 : SUCCESS in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30b86ca4904f4e7c846d2c101318c38b : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8c86f4d19704c00bb55613eda16acc8 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91550162f3ce43949756b6ce4a83e1f0 : FAILURE in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a076d14f20cf412ea693bacc57790f00 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5dbe4e1b829949709e950ba59f3a3b46 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6098e4072af1449e84dfab3753558ead : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472b92a8b9a34b099da76bf947f3ed15 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92f7af7b88de4f4887106fe381a712d1 : SUCCESS in 3m 40s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/249c1092a68f4d74840938e2bcf4532e : SUCCESS in 9m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0c22459f7c7345b986b9d664db99c2ef : SUCCESS in 9m 24s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia_lib/api/drivers/driver_lib.py\n  Comments left for invalid file octavia_lib/api/drivers/data_models.py\n  Comments left for invalid file octavia_lib/api/drivers/provider_base.py\n  Comments left for invalid file octavia_lib/hacking/checks.py","accounts_in_message":[],"_revision_number":1},{"id":"60a3c0320bb6d9cfd3ac2814e2eeead0335eae94","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-07-16 14:16:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7ac5af28416a0f6eca4c07f8cf06e1bde1abee8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:34:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c67096ed97b4edeb6ef6e061625f339 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e0a52d87acc48b69177b31cf9a118ec : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f5f0adeb75340338a114e10b38b3aa7 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6db619a8c14427f8a6a9b86d1767be7 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4915989ddfb40cbb9eb8dedb1009c18 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ca749f1726f4b12adef840cc4213eb1 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6461b510b44145ce81655416e6ecefea : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b33114e5aa34234b8f485a0ae3906da : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1057a9110a354305998879fde370f86f : SUCCESS in 4m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/380974fb0f554fdabe6a2110f480cb53 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ff9cac8d5d9d42448777203bcefd4e92 : SUCCESS in 8m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12c2d4361b0e0bfd8973bc6a9a8185ce8d5573fd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-07-16 14:45:53.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"81f692af0d03151f31662e54ff51c502a5daf394","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-16 14:45:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9fef05dc1a53a2f9f1015dfa6d0903110c39a669","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-07-16 14:46:02.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c53318b43464cb9e2ace7e9c66e6cbcf0bf0d387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:46:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8096139159fc5de5771ec996a7627816e5297dc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:01:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/feb4bd02e03d44109f0d7cf8c3794ba2 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51c96f4bc2b34d399a64a55d857e617b : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a2b232d6da455cb8aa62eb948c4194 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f298b35eb8c404a881175fc3a8c8103 : SUCCESS in 7m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a1d18bc6d304f8aaaff7e65cc901fa6 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dcf3b287f394382a4ebe82f8c9c30ae : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b4338e8a05e4adfb9f2e58193147aa2 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f2303e65d51d80ca3cb23bde772eb9aee55fb6d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:01:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ccb7772bb1065bf377ab103c10d4097628876ace","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 15:03:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08090a9aefa4991ba0220764463f0d9 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f79d27c983b84290bc2de3ddb3698f46 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf1824a8b26fe1062f420cb5b0470c6feebaa8ee","revisions":{"27f26079773d4d8100a8502c1748025f51df1e07":{"kind":"REWORK","_number":1,"created":"2020-07-16 13:51:58.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/55/741455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/741455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/1"}}},"commit":{"parents":[{"commit":"90b2b36978a6426b4779a430b4ccbf1d33330f92","subject":"Merge \"Add releasenotes link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/90b2b36978a6426b4779a430b4ccbf1d33330f92"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 13:50:23.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 13:50:23.000000000","tz":-240},"subject":"Update test-requirements to match Octavia repo","message":"Update test-requirements to match Octavia repo\n\npylint needed an update to be \u003e\u003d2.2.0 to fix a pep8 issue,\nupdated bandit requirement to match Octavia repo while\nwe\u0027re at it.\n\nChange-Id: I7ebdd5eea537a8258f66b5e1bf2ee1aad3f8b10b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/27f26079773d4d8100a8502c1748025f51df1e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/27f26079773d4d8100a8502c1748025f51df1e07"}]},"branch":"refs/heads/master"},"bf1824a8b26fe1062f420cb5b0470c6feebaa8ee":{"kind":"REWORK","_number":2,"created":"2020-07-16 14:16:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/55/741455/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/741455/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/741455/2"}}},"commit":{"parents":[{"commit":"90b2b36978a6426b4779a430b4ccbf1d33330f92","subject":"Merge \"Add releasenotes link to README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/90b2b36978a6426b4779a430b4ccbf1d33330f92"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 13:50:23.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-07-16 14:15:29.000000000","tz":-240},"subject":"Update test-requirements to match Octavia repo","message":"Update test-requirements to match Octavia repo\n\npylint needed an update to be \u003e\u003d2.2.0 to fix a pep8 issue,\nupdated bandit requirement to match Octavia repo while\nwe\u0027re at it.\n\nHad to fix some new errors pylint found as a result.\n\nChange-Id: I7ebdd5eea537a8258f66b5e1bf2ee1aad3f8b10b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bf1824a8b26fe1062f420cb5b0470c6feebaa8ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bf1824a8b26fe1062f420cb5b0470c6feebaa8ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
