)]}'
{"id":"openstack%2Foctavia-lib~744519","triplet_id":"openstack%2Foctavia-lib~master~Iea9fdbfaee7c7965adcc65825fe6422f87b06de0","project":"openstack/octavia-lib","branch":"master","topic":"http2","hashtags":[],"change_id":"Iea9fdbfaee7c7965adcc65825fe6422f87b06de0","subject":"Add alpn_protocols to the listener data model","status":"MERGED","created":"2020-08-03 16:12:40.000000000","updated":"2020-08-27 00:00:21.000000000","submitted":"2020-08-26 23:58:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744519-1598486336174-d5824c7c","meta_rev_id":"d9b041a51b832e354862ce3bbcd877e2e614010e","_number":744519,"virtual_id_number":744519,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 23:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":2,"date":"2020-08-26 22:24:11.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":1,"date":"2020-08-26 22:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-26 22:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 22:25:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2020-08-26 22:24:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 19:21:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-08-26 18:16:00.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"},{"updated":"2020-08-26 22:25:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-08-26 23:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2aa928eb2bd89b9675b15db645254f9ee5d9dfc7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-03 16:12:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef09b37c2cab520c2aab209d74ec641166be41be","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-03 16:14:52.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"9c4ff71f683d3775c352a6dca4490dadb04c4f39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:31:52.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2e68d980b3245be8cd2fa2dffa2384a : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7725e49c26d49a1b779a5fc6b94a586 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/365589d728cc4943b175cf3558f0cea0 : FAILURE in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d0852a4273a4bbea028d2be677570d9 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ebe1654ec541859f94a79077b5c435 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bf39e2abe74fedb0e956b8d0b625de : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09a073494dec46ebb9b5573ac6e0be13 : SUCCESS in 3m 05s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/365ad41c54084f3ebd1137fd41e951dc : SUCCESS in 6m 04s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9fc53cf083364930bb4c23ef896298c0 : FAILURE in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a587ca2b991bee341347b2420ad2b615393fb2e0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-04 11:52:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8b0a5b80712c1608faa274439c74ebb33c23edc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 12:05:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ea8fb2f012a43808247662906ceb641 : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec3c9d2155bf4446bdf4f9fab8d36336 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de4988fbadc54f68a174e6b84d877740 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c708e248d63246d8ba36d9c8e3791feb : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17527015242943ee850822572f17d384 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4fafc007fa0420abef8bacab2e4f594 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70620dcd698549c9b0ef52fb6d378ed2 : SUCCESS in 2m 50s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/17317bf7e72540eba9134b99fbc11043 : SUCCESS in 5m 06s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2a31f952eef2401e90548e560f1dc856 : FAILURE in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"146883ef603503465333712daf1fd2b3ef42f8bc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 15:39:35.000000000","message":"Patch Set 2: Backport-Candidate-2 Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"140cdf7fe710e1df72b07d859ba263803953b977","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-08-25 19:21:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b5c337ccb1907fdece31bbcaaf019a4b3046d23","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2020-08-26 18:16:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8afb4b1405fbfd8a5509255755bde4504d264a10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 18:18:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"80a85cd029b377c79aa5dd823846185ce75e6770","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 18:31:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8eda07f582e145edb3379a8bab9e3664 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1d2e6dd6dc41a781bd20002f760356 : FAILURE in 2m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a25cc40e999b4ef7a42609f219780175 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ee1ee72276649d4892b6db6e0edb253 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097d1a915dd74909909f41d891b55a08 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/657b9e51cc7f4e7c8c919dd5245118e5 : SUCCESS in 3m 49s\n\nWarning:\n  Comments left for invalid file octavia_lib/api/drivers/exceptions.py\n  Comments left for invalid file octavia_lib/api/drivers/driver_lib.py","accounts_in_message":[],"_revision_number":2},{"id":"6c39b001be7a376392b7402015caf1169b19f39e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 18:49:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1c0748c90835e6c6c3feaff38d13789e19e2aab5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 19:07:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa8155a93ad84fe1aa9107c62cb7ac3f : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01a52e3c6b21478abd4819e9498eabe1 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3149d144ae70452fa6f87523cf40020b : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ac8417546e143258ce692d5abdee822 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acdac9137b614b8da69b28d789e04eec : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3d7cf8368d4062abb318374e717f35 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/820459c9904e4493ba3328b31886d1e8 : SUCCESS in 4m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/fe47fc5d5d3e434d89967327496d645c : SUCCESS in 7m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/9e42b897a8e94585b2de01efe9e0c8c0 : FAILURE in 8m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f52deef6321f631883a5593b2730288fadc87d25","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-26 20:30:15.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4f680f3dfd94495acd32a12c29fb2896f07cc217","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 21:39:00.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"55bb0c5340d514011f31900b1cce1500a6c7f2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:59:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/545968e892ed486aa5364f98f5a4b4d4 : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07283f76d37e4eab82274482629d3691 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b23689a43140479845ce3bbedac9ab : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a8f26585690464ca6a47ac37319ea37 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b085dd0413844b91932fad010dc4af3e : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cb533576224309b79122078ab08854 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f680f6d7c10d419da256b410f0b538a1 : SUCCESS in 4m 51s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/32656125e72745d79e50eb0e4f1f00cd : SUCCESS in 7m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/ece18cdf01744fb3b1858d8b6a26002b : FAILURE in 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bee322afd8451818316ec66391d9aa7d3362aa5e","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-26 22:24:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"879dbff2d4555573c7e1c5010ac431f68b67dcac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-26 22:25:53.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5b3f9fe0732e182ff610f7a7e2788bb183eba12b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:38:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8297a1856d3d4f7da052659a5635eeea : SUCCESS in 2m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/899a95f1614749659d828c8a077348db : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86bf0688382c4caeb04a40cf0215e0fa : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef318f3c66c74c21affd672a16f1ca2d : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83a5dc692c634ab2a42a2060b4522b3c : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e97d61b8d5d4897b5a950eb45a504da : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4c86ca6a27a4e4892a839f30aefeca9 : SUCCESS in 4m 28s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/dbd0e800ad474f1d9260c661410fcbfa : SUCCESS in 6m 05s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/3a02881220284c1ab6eb811b126b1d30 : FAILURE in 7m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3b7250ebab3458671d202adc14af94f777bb9b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:39:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9674c991c77c3e551588f1a4b4cc6ef968be9bf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:58:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c349555bd1354934bf9a428b7c4b2f5d : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1ecb0ab8aed4157999f291af41e9c97 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9105e12766c549aba8743c3cbdaef185 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7139655e39f493181c60436271e7a68 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f1f7db5129f4ef99c01e6e359377876 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b43da60851504fd79ca0465cd17464a8 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"7e93d396194a9628fee6845242322ff20f939488","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 23:58:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"bca18e23864a93c9e099cb7b8bb3738bf3c812f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 00:00:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f192268b074c4819a1abb9a056ed90fe : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/259a6a79f9594e94b936357211b24b3c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b262fc9ad346aee6c1ac08bfd142451cd55571ea","revisions":{"b942a1ed6a013d74a184fc03b8a9db23c1e3e283":{"kind":"REWORK","_number":1,"created":"2020-08-03 16:12:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/744519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/19/744519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/1"}}},"commit":{"parents":[{"commit":"397653cb6de49d5f9733348231833a12da49531d","subject":"Fix NotFound exception in docstring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/397653cb6de49d5f9733348231833a12da49531d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-02 16:31:32.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-02 21:21:24.000000000","tz":120},"subject":"Add alpn_protocols to the listener data model","message":"Add alpn_protocols to the listener data model\n\nChange-Id: Iea9fdbfaee7c7965adcc65825fe6422f87b06de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b942a1ed6a013d74a184fc03b8a9db23c1e3e283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b942a1ed6a013d74a184fc03b8a9db23c1e3e283"}]},"branch":"refs/heads/master"},"afdc51242662ee8f2639bc5c6b196f4d331e7efa":{"kind":"REWORK","_number":2,"created":"2020-08-04 11:52:14.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/744519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/19/744519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/2"}}},"commit":{"parents":[{"commit":"397653cb6de49d5f9733348231833a12da49531d","subject":"Fix NotFound exception in docstring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/397653cb6de49d5f9733348231833a12da49531d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-02 16:31:32.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-04 11:51:20.000000000","tz":120},"subject":"Add alpn_protocols to the listener data model","message":"Add alpn_protocols to the listener data model\n\nChange-Id: Iea9fdbfaee7c7965adcc65825fe6422f87b06de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/afdc51242662ee8f2639bc5c6b196f4d331e7efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/afdc51242662ee8f2639bc5c6b196f4d331e7efa"}]},"branch":"refs/heads/master"},"31046d99995617b9554b5e38cf03f69713e5f0b3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-26 18:49:01.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/744519/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/19/744519/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/3"}}},"commit":{"parents":[{"commit":"89024085faf7da838735991c5df100f1941d34cf","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/89024085faf7da838735991c5df100f1941d34cf"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-02 16:31:32.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-26 18:48:14.000000000","tz":120},"subject":"Add alpn_protocols to the listener data model","message":"Add alpn_protocols to the listener data model\n\nChange-Id: Iea9fdbfaee7c7965adcc65825fe6422f87b06de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/31046d99995617b9554b5e38cf03f69713e5f0b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/31046d99995617b9554b5e38cf03f69713e5f0b3"}]},"branch":"refs/heads/master"},"b262fc9ad346aee6c1ac08bfd142451cd55571ea":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-26 22:24:11.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/744519/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/19/744519/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/19/744519/4"}}},"commit":{"parents":[{"commit":"b38f83120ced4e2a73c83fb7d650177aca45d541","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b38f83120ced4e2a73c83fb7d650177aca45d541"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-02 16:31:32.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-26 22:24:11.000000000","tz":0},"subject":"Add alpn_protocols to the listener data model","message":"Add alpn_protocols to the listener data model\n\nChange-Id: Iea9fdbfaee7c7965adcc65825fe6422f87b06de0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b262fc9ad346aee6c1ac08bfd142451cd55571ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b262fc9ad346aee6c1ac08bfd142451cd55571ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
