)]}'
{"id":"openstack%2Foctavia-lib~747315","triplet_id":"openstack%2Foctavia-lib~master~Id22bd98892edee74e2db3e0c0fcc88219813fa6b","project":"openstack/octavia-lib","branch":"master","hashtags":[],"change_id":"Id22bd98892edee74e2db3e0c0fcc88219813fa6b","subject":"Add compression offload support","status":"NEW","created":"2020-08-20 23:44:11.000000000","updated":"2020-08-25 15:38:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d5e2c802764985bdf1a1bd642adfe61ac46c9bc2","_number":747315,"virtual_id_number":747315,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-21 00:19:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2020-08-25 15:38:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 00:19:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a31c16b9fee4456851c7d66be37c535ad0061936","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-20 23:44:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f472d026568c461d0283347efadd46a85883cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:57:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55499a6594964550baf0a7cba88af975 : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/409d38376de04895b3f990ae7616a75f : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6397f5bdcde945a48bef2728af8e01a0 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2353e6bb016b4b0e8c44ee08b42ef841 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a83cd27217f4fd09ec98c7e8631bef9 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3bf3bcf1014a1098730a3f3e5fdafe : SUCCESS in 8m 04s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/8257162c89ef4612815969ca9f6eaa0f : SUCCESS in 5m 31s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/bfbe6d9db8ce415aa3f31834a031be0f : SUCCESS in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8799652574705f720702d2549aa293b15d0006a0","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-21 00:01:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"522427206c60aeacd273252ef5fc7a4fa4c5d268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:19:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62688550e95e4986a7a217da9809d92a : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec7dccae9de7440ca668d9d026759a57 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d21f9303a1af412bb2fea275d18ff3a8 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec8a5674eabf4828b243e7fb758fc088 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e266f07c478846e7a788960670952c9c : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d81e951a1f49d9a776916ca33bbffd : SUCCESS in 6m 10s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3cf746e50a744d188c5bf247efff02f3 : SUCCESS in 7m 28s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2bc3de496e1c455f9923778c403b8290 : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5e2c802764985bdf1a1bd642adfe61ac46c9bc2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-08-25 15:38:50.000000000","message":"Patch Set 2: Workflow-1\n\nThis still needs data model changes.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9fd0abad98e5e630aa137f1755b4f8dfe3f6ecea","revisions":{"fb10560a0b114c4deaa7e00bf989277cb3cdbbbf":{"kind":"REWORK","_number":1,"created":"2020-08-20 23:44:11.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/15/747315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/15/747315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/1"}}},"commit":{"parents":[{"commit":"397653cb6de49d5f9733348231833a12da49531d","subject":"Fix NotFound exception in docstring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/397653cb6de49d5f9733348231833a12da49531d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 23:40:01.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 23:43:59.000000000","tz":-420},"subject":"Add compression offload support","message":"Add compression offload support\n\nThis patch adds the required changes to support compression offload.\n\nChange-Id: Id22bd98892edee74e2db3e0c0fcc88219813fa6b\nStory: 2008050\nTask: 40718\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/fb10560a0b114c4deaa7e00bf989277cb3cdbbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/fb10560a0b114c4deaa7e00bf989277cb3cdbbbf"}]},"branch":"refs/heads/master"},"9fd0abad98e5e630aa137f1755b4f8dfe3f6ecea":{"kind":"REWORK","_number":2,"created":"2020-08-21 00:01:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/15/747315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/15/747315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/15/747315/2"}}},"commit":{"parents":[{"commit":"397653cb6de49d5f9733348231833a12da49531d","subject":"Fix NotFound exception in docstring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/397653cb6de49d5f9733348231833a12da49531d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-20 23:40:01.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2020-08-21 00:00:59.000000000","tz":-420},"subject":"Add compression offload support","message":"Add compression offload support\n\nThis patch adds the required changes to support compression offload.\n\nChange-Id: Id22bd98892edee74e2db3e0c0fcc88219813fa6b\nStory: 2008050\nTask: 40718\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9fd0abad98e5e630aa137f1755b4f8dfe3f6ecea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9fd0abad98e5e630aa137f1755b4f8dfe3f6ecea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
