)]}'
{"id":"openstack%2Foctavia-lib~764145","triplet_id":"openstack%2Foctavia-lib~master~Ib10cd48a3a507d97e40e6a54eaad67e479c90a50","project":"openstack/octavia-lib","branch":"master","topic":"remove-six","hashtags":[],"change_id":"Ib10cd48a3a507d97e40e6a54eaad67e479c90a50","subject":"Remove six from lower-constraints.txt","status":"ABANDONED","created":"2020-11-25 09:42:51.000000000","updated":"2021-01-30 13:18:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"79deea9d8501446075a85a245efcfeefe7aa5446","_number":764145,"virtual_id_number":764145,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-25 10:08:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 10:08:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b8f2f1a7276517b53ec3dbc91d295c41b7f83de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-11-25 09:42:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3608d90c1279c58cf9426068f4a5a2d3b8dcec62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 10:08:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6071c16e300c4acaad192cbe3f0b4671 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a7d2a00c36a41ca83d5c18c61d67cc1 : SUCCESS in 3m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1d6dbe316b242a4b7ef07e0ef003fff : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7feaf0e13e7a4f6abe22e19cbabb4fc1 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1398324cdb8247ba910d68ec3e654eb8 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89585845c5694bf3871fc5cf8075b945 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e91d8894310a4c2d84c50eea562d47e1 : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d20991a605f4f1ebabd8a6c868662e4 : SUCCESS in 4m 21s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/d66cbb8cb82a44aaa5753d02197ba5bc : SUCCESS in 4m 50s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/e2aab82307ec457bbcdacea9877dd714 : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79deea9d8501446075a85a245efcfeefe7aa5446","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-01-30 13:18:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16f433543adefd992db20371b0d7bbb3629a624e","revisions":{"16f433543adefd992db20371b0d7bbb3629a624e":{"kind":"REWORK","_number":1,"created":"2020-11-25 09:42:51.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/45/764145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/45/764145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/45/764145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/45/764145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/45/764145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/45/764145/1"}}},"commit":{"parents":[{"commit":"21170fd070b565898530cc70660b44e7eab91a24","subject":"Add Python3 wallaby unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/21170fd070b565898530cc70660b44e7eab91a24"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-11-25 09:42:12.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2020-11-25 09:42:12.000000000","tz":480},"subject":"Remove six from lower-constraints.txt","message":"Remove six from lower-constraints.txt\n\nChange-Id: Ib10cd48a3a507d97e40e6a54eaad67e479c90a50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/16f433543adefd992db20371b0d7bbb3629a624e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/16f433543adefd992db20371b0d7bbb3629a624e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
