)]}'
{"id":"openstack%2Foctavia-lib~781265","triplet_id":"openstack%2Foctavia-lib~master~Ie624d4864959a52e5d7697c8db48772610edfc4e","project":"openstack/octavia-lib","branch":"master","topic":"add-xena-python-jobtemplates","hashtags":[],"change_id":"Ie624d4864959a52e5d7697c8db48772610edfc4e","subject":"Add Python3 xena unit tests","status":"MERGED","created":"2021-03-18 10:24:53.000000000","updated":"2021-08-12 09:51:38.000000000","submitted":"2021-08-12 09:51:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781265-add-xena-python-jobtemplates","meta_rev_id":"6237e0d7c130b988d353f2ebf0fb520310f91b63","_number":781265,"virtual_id_number":781265,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-12 09:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2021-08-12 09:51:38.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2021-08-12 09:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-12 09:51:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2021-08-12 09:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 11:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 20:42:34.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2021-07-08 20:51:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2021-08-12 09:06:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"ee692e9a1710b56edb71d93dd89f1b7357d2e1bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-18 10:24:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60fa975f48d886f5bcc55cb005aa7eaf98868796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 11:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82e0a8074ed44820abe51fa18f5b94a6 : SUCCESS in 5m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07e3b9352edf4c9abe1963b5aa9a380e : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1cf91f60a94ac78987fd803fbc91ec : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b601a6e0b9c340bea22bba301cc58b0a : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fbaef759ac54fda92200510612d6fd6 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e135cb0fa47646f0b5c72fb8f2f8730b : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/986d4578bebc4538ad3a694da9bd28dc : SUCCESS in 6m 02s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/81a036fbc4e84f0f92605f8df4af8e3a : SUCCESS in 6m 24s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/f485e78257854c43a7ab264e07dbf5b7 : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01de55796700641fb77f22b2f23d9b6f04297baa","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2021-07-08 20:42:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3b378f4d362d094270c408106f769668e0df203b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-08 20:51:20.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"34a0ffc8d6530aafb8593dd11d24f9643b2eb943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-07-08 23:37:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"89d3dd7d8dc848d35bd16ae55423c2aa3519f926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 23:51:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a8e016731d49bd8cbf59372d5e3249 : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d8a718bb3e414e955d3497345b90a7 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69b64c90848a4d7c8479594a6769daf3 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c365384775e4da5b28a5fd9076b9135 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a6c500d3e744b3ab236349fa6cd944e : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a03e8e3afb145ecb2579b61c65c71cd : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f004b10446fe45a49e31e92431b31584 : SUCCESS in 5m 17s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/43f9a562591e4db2af9a680c941be546 : SUCCESS in 5m 46s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/886f96cb69df4ddeb7845f53ca2d469e : SUCCESS in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4b0cd1b11a2b4219cd7dfeb78dfd18b5cb58bff","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-08-12 09:06:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e917626dc63bebba3e5f5071c40eebcb3adb2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:07:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f3ad2e6c770ae3beba95330bbd35fd968597a9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:51:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a933384bf3d4679933bda1a2b454797 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cd080fef4314d97b9cf349650107189 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df86b6bf00cc44af9370762a76a7b06c : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa40f7b8b93647808cfba8fd1a66ef96 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1c2db98a47436aabb8eba019067535 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"483e4f20a4d7f403add2ebae35e8922f9965a6ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-12 09:51:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ea61896f1072b85f936916db656fd24c982f9de","revisions":{"eb5a069a74df1e26355230746994379a92145b33":{"kind":"REWORK","_number":1,"created":"2021-03-18 10:24:53.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/65/781265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/65/781265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/1"}}},"commit":{"parents":[{"commit":"53b4b8d3882b4a7e69da1438708834562565cca9","subject":"Update master for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/53b4b8d3882b4a7e69da1438708834562565cca9"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 10:24:43.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 10:24:43.000000000","tz":0},"subject":"Add Python3 xena unit tests","message":"Add Python3 xena unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for xena.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie624d4864959a52e5d7697c8db48772610edfc4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/eb5a069a74df1e26355230746994379a92145b33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/eb5a069a74df1e26355230746994379a92145b33"}]},"branch":"refs/heads/master"},"1ea61896f1072b85f936916db656fd24c982f9de":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-07-08 23:37:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/781265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/65/781265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/65/781265/2"}}},"commit":{"parents":[{"commit":"cdbbacd40fed7768de49898877b441b09ac3e9e8","subject":"Merge \"Update master for stable/wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cdbbacd40fed7768de49898877b441b09ac3e9e8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 10:24:43.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-07-08 23:37:50.000000000","tz":0},"subject":"Add Python3 xena unit tests","message":"Add Python3 xena unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for xena.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie624d4864959a52e5d7697c8db48772610edfc4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/1ea61896f1072b85f936916db656fd24c982f9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/1ea61896f1072b85f936916db656fd24c982f9de"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
