)]}'
{"id":"openstack%2Foctavia-lib~785559","triplet_id":"openstack%2Foctavia-lib~master~I69f61d7d2708be0d3d1248438db3e5b7ee6376ca","project":"openstack/octavia-lib","branch":"master","hashtags":[],"change_id":"I69f61d7d2708be0d3d1248438db3e5b7ee6376ca","subject":"DNM testing driver agent socket","status":"ABANDONED","created":"2021-04-09 06:37:04.000000000","updated":"2021-06-22 09:53:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2378b2511d8fd1e4b3169c97f527edb9c2189fc4","_number":785559,"virtual_id_number":785559,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-09 07:41:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 06:51:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a81d6ee21b83e299a712759a7d53fd9e917b8255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-09 06:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7cdf1cdfdeeee2c961e6af2f8f1bef96debedce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 06:51:59.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2178c7ffd535400187fc638f8699820a : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72f10a9d6e6a44c79aa2f8ab73a25380 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c6c890cbd7436f8c518ab4d29d5d0c : FAILURE in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/539a4b3f040f4835a595370b622767eb : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d2d532d7e95441fad45aba08add3e7d : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87d455be739b4836b59204ddfe468e12 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d544b5b00628459a85620830a10b4f7a : SUCCESS in 6m 38s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/2345145fd60a4656be0057140db81710 : SUCCESS in 6m 39s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/23340121387d43139659ce2ecc9b584c : SUCCESS in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb7bc286351baa2f0ccfe2cff781224837a7d6b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-09 07:10:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6ee1ebb8577504dd53b20fdbb760954e974509d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 07:23:41.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff07a5ee6dce40ad8ee5b371e208dd5c : SUCCESS in 7m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86ebac8087e842b0ad7146c1ba992e29 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a69f556c720493e8354885630c3cd72 : FAILURE in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9edf5b01fc124bb09f1e6389678bcf6a : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39aeeabb8f8f439fa08dedbb7b914bc1 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c6525f934704870983e9c9fb33af3d9 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6785930ad847fcb4dd2b8cdb502cad : SUCCESS in 6m 07s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/a218987123a94dc5904fba8e8d3f4a14 : SUCCESS in 5m 58s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/1a9af1cf4ee04b2e90246e05067f9723 : SUCCESS in 8m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c751cec937a74e178eb2e0bb74658d5d : FAILURE in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ae085042e8c6d4e20ede6065565cd10aef82d635","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-04-09 07:32:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b7a5ce14697dc07cef7a0fa5a44635a9169c96a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 07:41:46.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa954457d3ea4f11aa86cedcfede99de : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09c608329cce49bf89d3e8f46f59940d : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14219563d53a41dc94124a849836285a : FAILURE in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f276b8b4fa0a416297ebfcfa444648dd : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce93bd5ee9140569643273d2931239f : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a4258f9484e4e0abbe7f4eb29b9fe24 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/557ae65bb5954893b0968b1815c3dbf8 : SUCCESS in 5m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9e8943463cba4696848d4313ec3eaf31 : SUCCESS in 5m 56s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/0d2cb9b9450f4e77bc4273ee25d69874 : SUCCESS in 7m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2378b2511d8fd1e4b3169c97f527edb9c2189fc4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-06-22 09:53:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8aa1750e4c575f042ccf5c09668ebe91219f6a8f","revisions":{"513d8dce24ebad03a239743a7ff0d0ff3e71fb9a":{"kind":"REWORK","_number":1,"created":"2021-04-09 06:37:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/785559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/59/785559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/1"}}},"commit":{"parents":[{"commit":"3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075","subject":"Merge \"Stop configuring install_command in tox.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 06:36:29.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 06:36:29.000000000","tz":120},"subject":"DNM testing driver agent socket","message":"DNM testing driver agent socket\n\nChange-Id: I69f61d7d2708be0d3d1248438db3e5b7ee6376ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/513d8dce24ebad03a239743a7ff0d0ff3e71fb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/513d8dce24ebad03a239743a7ff0d0ff3e71fb9a"}]},"branch":"refs/heads/master"},"cffce5a4a55aa8f0f93e83bb3478a0f398b0e2a8":{"kind":"REWORK","_number":2,"created":"2021-04-09 07:10:23.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/785559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/59/785559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/2"}}},"commit":{"parents":[{"commit":"3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075","subject":"Merge \"Stop configuring install_command in tox.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 06:36:29.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 07:10:17.000000000","tz":120},"subject":"DNM testing driver agent socket","message":"DNM testing driver agent socket\n\nChange-Id: I69f61d7d2708be0d3d1248438db3e5b7ee6376ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cffce5a4a55aa8f0f93e83bb3478a0f398b0e2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cffce5a4a55aa8f0f93e83bb3478a0f398b0e2a8"}]},"branch":"refs/heads/master"},"8aa1750e4c575f042ccf5c09668ebe91219f6a8f":{"kind":"REWORK","_number":3,"created":"2021-04-09 07:32:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/59/785559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/59/785559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/59/785559/3"}}},"commit":{"parents":[{"commit":"3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075","subject":"Merge \"Stop configuring install_command in tox.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 06:36:29.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2021-04-09 07:31:30.000000000","tz":120},"subject":"DNM testing driver agent socket","message":"DNM testing driver agent socket\n\nChange-Id: I69f61d7d2708be0d3d1248438db3e5b7ee6376ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/8aa1750e4c575f042ccf5c09668ebe91219f6a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/8aa1750e4c575f042ccf5c09668ebe91219f6a8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
