)]}'
{"id":"openstack%2Foctavia-lib~794364","triplet_id":"openstack%2Foctavia-lib~master~I5ed38a032fe4ac02a03791baf6ff7275688b8b68","project":"openstack/octavia-lib","branch":"master","topic":"docs","hashtags":[],"change_id":"I5ed38a032fe4ac02a03791baf6ff7275688b8b68","subject":"Reuse the docs deps to benefit from constraints","status":"ABANDONED","created":"2021-06-03 01:54:05.000000000","updated":"2021-06-03 08:33:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"69870ae160ee989cc77afd48069d41a82623974c","_number":794364,"virtual_id_number":794364,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-03 02:03:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 02:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7eef68858446dd58c58211f3bf4c30238210968e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-03 01:54:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"242559b503a17cfc75db3950ab3d5eb95547cdec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 02:03:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbf6e7f9402943ae9e398bc575d2e38a : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3c54171a6b645c19cc84858f61359fa : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e562761e848410fb6dc83c41996db2d : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a17e4296c7924dc5baeb5b303fb40a68 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f34eb4a7bc548818a7059891b4ce356 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57ed155b019a4e70b18edc69357ab8f3 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046a70b645e4417984526c4f94a93181 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/060b908cf70d4f628b5a6c76cb17f029 : SUCCESS in 3m 29s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/02b192c7722c4d25b7f75e80a5d84df9 : SUCCESS in 6m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/122600fc69544cf6ba32273fe108fd9a : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69870ae160ee989cc77afd48069d41a82623974c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-03 08:33:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38aba340a876ee423ddc589c3a9a02a7e6f3732f","revisions":{"38aba340a876ee423ddc589c3a9a02a7e6f3732f":{"kind":"REWORK","_number":1,"created":"2021-06-03 01:54:05.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/64/794364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/64/794364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/64/794364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/64/794364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/64/794364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/64/794364/1"}}},"commit":{"parents":[{"commit":"3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075","subject":"Merge \"Stop configuring install_command in tox.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/3208f7fbcae31f84f0d8b0ff05ef63f5f7f2a075"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-06-03 01:52:38.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-06-03 01:52:38.000000000","tz":480},"subject":"Reuse the docs deps to benefit from constraints","message":"Reuse the docs deps to benefit from constraints\n\nTox trying to install latest versions for building releasenotes,\ndoc which may not be supported by stable and lower branches,\nso should be restricted by respective version\u0027s upper-constraints.txt\n\nChange-Id: I5ed38a032fe4ac02a03791baf6ff7275688b8b68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/38aba340a876ee423ddc589c3a9a02a7e6f3732f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/38aba340a876ee423ddc589c3a9a02a7e6f3732f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
