)]}'
{"id":"openstack%2Foctavia-lib~808222","triplet_id":"openstack%2Foctavia-lib~stable%2Fxena~If09845034214e6f26538e8841b621bd1613ac452","project":"openstack/octavia-lib","branch":"stable/xena","topic":"create-xena","hashtags":[],"change_id":"If09845034214e6f26538e8841b621bd1613ac452","subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","status":"MERGED","created":"2021-09-10 14:31:30.000000000","updated":"2021-10-05 08:28:42.000000000","submitted":"2021-10-05 08:25:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808222-create-xena","meta_rev_id":"ac766a9eaf1bc5b199f8ce3cc5676af0e066b16b","_number":808222,"virtual_id_number":808222,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 08:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-09-15 18:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-09-24 11:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-09-24 11:59:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-10 15:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-15 18:18:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-09-24 11:59:30.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"7d72b987605f1c48cbdd14f7217f97ebd86ad7ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-09-10 14:31:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"662c0a98122596687ec3114e6682248114702878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 15:11:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46ce45a5c2c340c7b0942c491d6ea73f : SUCCESS in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/201a17145b8541ec929418b335194c7b : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f8db85e2784f5e8f28a4e752f8d9c0 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4a7cf5a9f04cb99e0fa1b9cc11e0cd : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0199bfe5bde8487ba84285a1669e46e7 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672d9ff0720a44778bde66a629004729 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85da054d0c749fd97acc597beadc001 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d78f14cde51452e836cf395c5238fb8 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e27f47d7dfb62342684fe91b42ed6cadc936bf6d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-15 18:18:03.000000000","message":"Patch Set 1: Code-Review+2\n\nGenerated patch, looks OK.","accounts_in_message":[],"_revision_number":1},{"id":"e1f4e6c65c1002f111c7d56ba5b8e945d0d18c09","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-09-24 11:59:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93f4c36c849f052fa85c5b25e29d448b6830fdca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 11:59:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a696d3ac557c8b7a03788018ec57382afd72a7b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 13:39:44.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"105546b22a651f06609f4978b5ffc56e93485ca0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2021-10-05 07:33:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e80f38a543e774ead2d84dd27f66a0c65b859ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 08:15:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0bba8181be94a3c84f61d032e3dc61f : SUCCESS in 4m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/801e36a712924cc5bbe8ef83d3983059 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d99a8db5d0e478f945458835c60cc31 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cc8f96e6d4448eb887a77a93f38c4d8 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f3eb26b160145248080605345d424d7 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0412bfc4283b45ca96aa752c4b432a8a : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417f3b8e9df349b6a44f22b97ee22f5e : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e1e2b91e2354ee8a967d24d22c80940 : SUCCESS in 4m 23s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/95561b5ec26d4e87b658d08d283d0a05 : SUCCESS in 6m 33s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/773fa89fbb8b4c34b0e5cf69b12a5c9a : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71213e033b5a5d2440b1f7281077445705e83633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 08:15:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"941f1657158511f154b098c52f1e7f962af55190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 08:25:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6f082509a904a6dbdfad318e593b242 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e0781bf566413ab8704ef91f77591e : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d5d569012c94301917e32ccdfc7d290 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bab0a7c04aa24bf7a7aad3bad329551c : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8094d3a71bb4ff7829ee967f16b1e71 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fc541ed99c1408dab902853bd44abfb : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d3f27599d035b9cf6757f1f5d4d6b42282ba752d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 08:25:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ac766a9eaf1bc5b199f8ce3cc5676af0e066b16b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 08:28:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06bb077fb167444d8c0fdc5a03c13b3f : SUCCESS in 1m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/568f1b2e00fc4a4dafe4cbb379b4aa8b : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"40c7b21a173204ac0e97a35d97a724c89c11190c","revisions":{"40c7b21a173204ac0e97a35d97a724c89c11190c":{"kind":"REWORK","_number":1,"created":"2021-09-10 14:31:30.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/22/808222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/22/808222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/808222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/808222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/22/808222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/22/808222/1"}}},"commit":{"parents":[{"commit":"be2de826a7e93450f29a25d25c6d17b980ac00ad","subject":"Update .gitreview for stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/be2de826a7e93450f29a25d25c6d17b980ac00ad"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-10 14:31:29.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-09-10 14:31:29.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/xena","message":"Update TOX_CONSTRAINTS_FILE for stable/xena\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/xena branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: If09845034214e6f26538e8841b621bd1613ac452\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/40c7b21a173204ac0e97a35d97a724c89c11190c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/40c7b21a173204ac0e97a35d97a724c89c11190c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
