)]}'
{"id":"openstack%2Foctavia-lib~812257","triplet_id":"openstack%2Foctavia-lib~master~I313ea875af537f146218eb763ed80038da5d5213","project":"openstack/octavia-lib","branch":"master","topic":"octavia-prometheus","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-01-31 20:30:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I313ea875af537f146218eb763ed80038da5d5213","subject":"Add \"PROMETHEUS\" protocol.","status":"MERGED","created":"2021-10-02 19:52:42.000000000","updated":"2022-01-31 20:31:44.000000000","submitted":"2022-01-31 20:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812257-octavia-prometheus","meta_rev_id":"92bfa7ed1e76efe743421ab139db86f902bb8c95","_number":812257,"virtual_id_number":812257,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-31 20:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2022-01-31 20:21:46.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2022-01-27 16:47:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2022-01-31 20:21:46.000000000","_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-02 20:10:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 07:29:01.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2022-01-27 16:47:11.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2022-01-31 20:21:46.000000000","updated_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"reviewer":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"state":"REVIEWER"}],"messages":[{"id":"58aa7406108be71c3c85da5824752741e9db8ceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2021-10-02 19:52:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a416270cb236627b790faeb27e7ad460d13c240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-02 20:10:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1edbb0c6f7cc4e7e969396434975aaa5 : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72bc226a4066403bbb59f126728ad32f : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb919eab60094d0190ec5421feef4e9d : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2a5b511248342aaaa61bef18bab347a : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e081612b0d243ce99b85a8d0e1305ed : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db11d0d20b2c4fb39bf2f32caacd15ad : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3e9c21c5b44421a75a7b9d69f3f11c : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f1e4279576542dc8db19c92cdc21faa : SUCCESS in 3m 59s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/ef93ac72ba2c41688577cec1f5bbc7c0 : SUCCESS in 7m 55s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/798bcf63c23c4ea8a712f8cfb335abc6 : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7ec3d71c5ddba6fa4c5532074360e3fddc15109","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2021-10-26 07:29:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55f6a19cd26e1f23d2d7f333afdbba56b542f10b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-12 16:08:29.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d3175f8e6b199c252edbefa9c9a473b4b5cc933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 01:56:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ce92ad85ee36cb93a5340fc10c68531b773afc4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 02:00:16.000000000","message":"Topic set to octavia-prometheus","accounts_in_message":[],"_revision_number":2},{"id":"fb27d5677e3eae1fc0584910fe7af361c113ba53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 02:09:31.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7fe9f1e9ac746d8b17e7ba0998a32d5 : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23ffff165fa74ed7bb63dc2c2009cb76 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/980d2f3622e04965bd9cf4e8332bbf00 : FAILURE in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed8cb7018c43487bbeb85afce66ee496 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45168ee5093c4396a2c38d0013462f5a : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0641edec7b2f4c1a9143ad36f27d739d : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/648558fcb133496581eebff8ee1b162b : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc8f0d7f2f8f4dd3bd46cfb87a099f94 : SUCCESS in 3m 18s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/80ef5dd0d7b0436ebdf61fcdab4b6e14 : SUCCESS in 6m 48s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/dffdaaa18a2f47c9986ada6fa74fe309 : SUCCESS in 8m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05e82662014a8c60fa545478268316c5b2eca389","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-01-25 18:07:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"abdf99e6ae01c4451ada18c4d45de06b584949e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 18:26:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa01103ff4c9435a879a1f02619f5bdb : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ddcc566659d4f23b7f8303c1fde3901 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47aab576d5e54f57a65a00f8e51a177a : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5b4726e0e8148b2be9f03a62fea3576 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a1c9daf1e1b4643a3ad06accf783448 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e728bc27fef745dca26b2611d9247fc1 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1390227b34e1401ca0f78cd4adf2223e : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a362d1de45d4238ab14e6b1a3235387 : SUCCESS in 4m 43s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/55596d2ea35647eeb8111ab70fd13411 : SUCCESS in 6m 18s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/6cbd8cb6aa1f41c2a3dc60d72a670f7a : SUCCESS in 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e79021a79ec480d0955f73cf4e1fff45122accf2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-01-27 16:47:11.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d3f95af2c1c274376beb6e4c8b7550fe022c8b7","author":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"},"date":"2022-01-31 20:21:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2d74d22cf7989d8900a72afe6f3c91ca387bee24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:22:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ae88513d11194665b5cff06bbe4b8d3c1ce26d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:30:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db854e5c1c5947c2bed9a8aaca246c20 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34af2c4735994803a9d3f99718043589 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63ab43e8884646c58c927baddbc55fef : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd5c0c3c9ed4bcaa5ab371bdb9fdc5d : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26850bf9ebe4d2299e1bf6a27934ffd : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/463453e0a3c240789281118c78ca7cc9 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4b25c5177cd3269d2780869f00d26e1796f5438d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:30:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"92bfa7ed1e76efe743421ab139db86f902bb8c95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 20:31:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3ecb722a104aa99c62a5d055db487c : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c48ddf8d1434b6d8dbe18e8e647f4ff : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4f217dd4fa7e932d19d756e73469529a5b4f99c6","revisions":{"014da4dda34f8919dc9424d63ca32cf3aa7a5008":{"kind":"REWORK","_number":1,"created":"2021-10-02 19:52:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/812257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/57/812257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/1"}}},"commit":{"parents":[{"commit":"e2c17e0f25a7d166eb51798e45e3fc73422d9685","subject":"Ignore new pylint checker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e2c17e0f25a7d166eb51798e45e3fc73422d9685"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-01 20:18:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-02 19:52:40.000000000","tz":0},"subject":"Add \"STATS\" protocol.","message":"Add \"STATS\" protocol.\n\nThis patch adds a protocol called \"STATS\" as a supported protocol for\nlisteners. It will be used to expose a statistics endpoint through\na listener.\n\nChange-Id: I313ea875af537f146218eb763ed80038da5d5213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/014da4dda34f8919dc9424d63ca32cf3aa7a5008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/014da4dda34f8919dc9424d63ca32cf3aa7a5008"}]},"branch":"refs/heads/master"},"e75bda06012f29919fb38b2e439627b8d5f25916":{"kind":"REWORK","_number":2,"created":"2022-01-25 01:56:07.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/812257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/57/812257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/2"}}},"commit":{"parents":[{"commit":"9fb0ded6657823b5b93c19134c8d98dfabf15691","subject":"Merge \"Update master for stable/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9fb0ded6657823b5b93c19134c8d98dfabf15691"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-01 20:18:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-25 01:56:05.000000000","tz":0},"subject":"Add \"PROMETHEUS\" protocol.","message":"Add \"PROMETHEUS\" protocol.\n\nThis patch adds a protocol called \"PROMETHEUS\" as a supported protocol for\nlisteners. It will be used to expose a Prometheus exporter endpoint through\na listener.\n\nChange-Id: I313ea875af537f146218eb763ed80038da5d5213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e75bda06012f29919fb38b2e439627b8d5f25916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/e75bda06012f29919fb38b2e439627b8d5f25916"}]},"branch":"refs/heads/master"},"4f217dd4fa7e932d19d756e73469529a5b4f99c6":{"kind":"REWORK","_number":3,"created":"2022-01-25 18:07:16.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/57/812257/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/57/812257/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/57/812257/3"}}},"commit":{"parents":[{"commit":"9fb0ded6657823b5b93c19134c8d98dfabf15691","subject":"Merge \"Update master for stable/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9fb0ded6657823b5b93c19134c8d98dfabf15691"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2021-10-01 20:18:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-01-25 18:07:16.000000000","tz":0},"subject":"Add \"PROMETHEUS\" protocol.","message":"Add \"PROMETHEUS\" protocol.\n\nThis patch adds a protocol called \"PROMETHEUS\" as a supported protocol for\nlisteners. It will be used to expose a Prometheus exporter endpoint through\na listener.\n\nChange-Id: I313ea875af537f146218eb763ed80038da5d5213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/4f217dd4fa7e932d19d756e73469529a5b4f99c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/4f217dd4fa7e932d19d756e73469529a5b4f99c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10850,"name":"German Eichberger","email":"german.eichberger@gmail.com","username":"german"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
