)]}'
{"id":"openstack%2Foctavia-lib~855738","triplet_id":"openstack%2Foctavia-lib~master~I169b58c4a02165b5959532e1cd73d1d23aee6676","project":"openstack/octavia-lib","branch":"master","topic":"release-health-check","hashtags":[],"change_id":"I169b58c4a02165b5959532e1cd73d1d23aee6676","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:31:49.000000000","updated":"2022-09-07 09:13:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"13610e91a6d29a6c8189c1510aea8addcad4d14a","_number":855738,"virtual_id_number":855738,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-03 20:48:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 20:48:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a11c2c6514610b57400b6cc16159bf3193d21476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:31:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d928127eec486c2b21ab4fc0794d7942ae714c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 20:48:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/466c84ebf9c549e2afaf8bbaea712c67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f0d25c548284b138601bbd12053c61f : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806e678051554c91be151e4396d9515f : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84d917c8d6134d44b29dc61b785f28ad : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00c43b36de48413eaf4f890aecd88d88 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e05a8080d2b94f338d7d0e26599e99d8 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27390ec76f174f998b4b7bbb140fd244 : SUCCESS in 5m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e459930cedd443fe8bb58841422ba161 : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7f1f2d3ecaf84af19d58ff41b3778424 : SUCCESS in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13610e91a6d29a6c8189c1510aea8addcad4d14a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:13:56.000000000","message":"Abandoned\n\nGate check passed, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5101a1879321cc3ad5e7fd77640aaa2c3fdd1fc","revisions":{"d5101a1879321cc3ad5e7fd77640aaa2c3fdd1fc":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:31:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/38/855738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/38/855738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/855738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/855738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/38/855738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/38/855738/1"}}},"commit":{"parents":[{"commit":"09e9b48d29a10a820369c0dca50e246e27ec6b52","subject":"Merge \"Add new amphora failover stopped status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/09e9b48d29a10a820369c0dca50e246e27ec6b52"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:31:37.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:31:47.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: I169b58c4a02165b5959532e1cd73d1d23aee6676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d5101a1879321cc3ad5e7fd77640aaa2c3fdd1fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/d5101a1879321cc3ad5e7fd77640aaa2c3fdd1fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
