)]}'
{"id":"openstack%2Foctavia-lib~859788","triplet_id":"openstack%2Foctavia-lib~master~Ib185a743937df501e346b9a6e79be590a3dae355","project":"openstack/octavia-lib","branch":"master","topic":"gitignore-dotorig","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-01-11 07:54:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib185a743937df501e346b9a6e79be590a3dae355","subject":"Add *.orig to .gitignore","status":"MERGED","created":"2022-09-29 08:16:11.000000000","updated":"2023-01-11 07:55:05.000000000","submitted":"2023-01-11 07:54:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859788-gitignore-dotorig","meta_rev_id":"232cee5a3072391a73f90bb50babf4f1f18cc79c","_number":859788,"virtual_id_number":859788,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 07:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2022-09-29 17:33:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-01-10 14:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-01-10 14:41:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-29 08:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-29 17:33:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-01-10 14:41:02.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"75bfbbfef45628534ab660c404bcfbc1eda6a1c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 08:16:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57a69a303a2eaf766124c58819709b1446b6a11d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2022-09-29 08:21:24.000000000","message":"Topic set to gitignore-dotorig","accounts_in_message":[],"_revision_number":1},{"id":"ca8780887d09528c12ba61bd1a91dab4e33e110b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 08:30:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/993470860a1743c1aa70c2b3d071145e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cefcb069b62a4515b78f73e291d0ccc9 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99c14dacdf694b5c8928c02f372cdb23 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03128cf7825f4a488a8af858badd5b2c : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d264b7fab454c2182ad1136804bb17a : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/476ebd26cf734ea6aeb4418b2b3c6081 : SUCCESS in 5m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5dd70d2084274d40842bb566cafef0b7 : SUCCESS in 5m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/de5e314f348f4f27a276d0a2833443f4 : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fffa52852a83c7da20e89502dba5c8c2432b199","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-09-29 17:33:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aacde015fd96a5b284bc514f44829cc996805f73","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 14:41:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d32bf5d04ad77c615092da58d11859150139a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:41:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1068e57f4485f2369b1d455d7ce383d8380bccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 14:51:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23da8e85bc4b4571a2a8fd75a5a1c54d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb879c67c9341659c4c6a9403477880 : FAILURE in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09ac9f99cc3048e9b2c52f69a7b8915f : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7838e110a47411fa3cd8446381e772c : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c227ed2e542c439cb9b4c0d3a6d9bbd9 : FAILURE in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0fed4971f224cf38d197eabcd17466f2c3d1d605","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-11 07:30:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7836cd57069176167c2c6809ab2e36e067d2bc3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:44:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8542a6e761bb4526b6e8c9cdc3d7dd80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6f2d7c672ce4921a9616db1af05ac52 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b815aa07f7cd4e9a93a50c5a062e5d39 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce9ab29f872f4bc68e399167408f599c : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/409d8576274f4d34a4261f281a202b70 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288107493fd8492e804aa60e1a943c3f : SUCCESS in 4m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6797e230a2dd46528eadff761f9892dd : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fe5aa1f158924132ad6bbce2154cbb2a : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"822110f6f14c5006f7f8e80a48ebce410a999d8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:44:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"294cff70eb33aa75bbcb720510d65be8aa8b0c5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:54:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edcf4f4715f844139329add980117137\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec56b5a0896b489aa45d4a782db45b37 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb2cdcf84c8b4ff69bc0a335428ad202 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e37cde07d7f46098c6467de3ad6c5fd : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffc38d8ebde470191c84d89b5c0a343 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bdeaf70d237e5895b8685a612e4c295673ef9d75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:54:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"232cee5a3072391a73f90bb50babf4f1f18cc79c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:55:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7482a1d492844d099af1c7d1dd85e2f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/341554b1e52e4ed3903c6fb669a569ab : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8f8ec86ef64a2314ba981fa10c596dbb7fb7840","revisions":{"f8f8ec86ef64a2314ba981fa10c596dbb7fb7840":{"kind":"REWORK","_number":1,"created":"2022-09-29 08:16:11.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/88/859788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/88/859788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/88/859788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/88/859788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/88/859788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/88/859788/1"}}},"commit":{"parents":[{"commit":"09e9b48d29a10a820369c0dca50e246e27ec6b52","subject":"Merge \"Add new amphora failover stopped status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/09e9b48d29a10a820369c0dca50e246e27ec6b52"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-29 08:16:06.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2022-09-29 08:16:06.000000000","tz":120},"subject":"Add *.orig to .gitignore","message":"Add *.orig to .gitignore\n\nChange-Id: Ib185a743937df501e346b9a6e79be590a3dae355\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f8f8ec86ef64a2314ba981fa10c596dbb7fb7840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/f8f8ec86ef64a2314ba981fa10c596dbb7fb7840"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
