)]}'
{"id":"openstack%2Foctavia-lib~868955","triplet_id":"openstack%2Foctavia-lib~master~Ib1e58d792ff12d3c3b72a001e04fa9b8c62d1ef8","project":"openstack/octavia-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"32921":{"account":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"last_update":"2023-01-11 07:29:48.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-01-10 15:56:39.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ib1e58d792ff12d3c3b72a001e04fa9b8c62d1ef8","subject":"Fix tox4 error","status":"MERGED","created":"2022-12-31 08:03:28.000000000","updated":"2023-01-11 07:30:47.000000000","submitted":"2023-01-11 07:29:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"868955","meta_rev_id":"cce6852b31afaa97f83fe1163dd77189c89577b0","_number":868955,"virtual_id_number":868955,"owner":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-11 07:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-11 07:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-01-10 15:43:52.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-11 07:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-31 08:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 11:56:39.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-01-10 15:23:54.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"3b1baa114143e51f160453b61db5e11c107e26b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-12-31 08:03:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f821612af1170be092a569a05c0e6ae0fe47714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 08:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fffb03ce819a42e8aed380a13a3d2062\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04584874aa764cb3a377a98eaea0be6f : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/842b4ef1112a42c295f18c589b4198e1 : FAILURE in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6550fd067db44bb9884e43aeec124f56 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c55ffe7a54c4e7789f848f07ee6c536 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff75224d106b41a7915afc1f76febeab : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/383c31748725423cbe1a592bc4505e97 : SUCCESS in 2m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/68fe94ca98c1440bb2348e1d7b293b2d : FAILURE in 4m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0c015b0425b743079f602226d77df01c : FAILURE in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b12d08b532984fef092e5d3b00072ced6934921e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"date":"2022-12-31 08:32:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3850550baf2ac4cc12a6e88bb70a7708dc2f5890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-31 08:42:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b169ff5aca694ae28dbcf14e0ad4d6a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad12f88a26142de939119bd87bc1ed2 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff3efaaa3914ca196b47b25c98ed26a : FAILURE in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21df9031397947b0a809d9470504cc42 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4f96707d8f84a5887f21a23fd0324b9 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9bd447d503b4b378e37046384264899 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19d6c276faf4df3b062932df3508c22 : SUCCESS in 2m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2164756b5805422a8b5789c902b9162a : FAILURE in 3m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d40aff59eadd4986a03533f6aa65bb95 : FAILURE in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b6f5226e0f7f880a12a6aea10c62cc33bba2426","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-04 11:56:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90b04e910a3a3130d5174d37867616627a2b97e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 15:08:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f704b25789d2c89353fc97516ccd4e5bdfb10d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 15:16:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4efe64a670384df38d87afafcfaa6ffc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d72011b5d65468781cf3d26d52d522f : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ab8dac19fca4330923ec85ba1692459 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0df2a6cfe9b746408509aa53c0259922 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf15d54a4cbb43c9ba46628375042549 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec459f53aee14380b1554c0d2a1075cc : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0861b1f1fe3c439c847647dddd17e8bc : SUCCESS in 2m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/286c93e58fc2437abd609cd3f67b5abe : SUCCESS in 5m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a0e4a1387ee54b9d9110c53b9d8650f9 : FAILURE in 2m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87779dd0e0ff467bc46fc5772bc1eddbd0a8dcf7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-10 15:18:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"429855ba6950885981b168bc56eaf3042c53aed4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 15:23:54.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f66dd765682d9717e2d73a92c27763d1662f667b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 15:25:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"10ace1673fc64158e56b04d0ea65bed3246a3c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 15:42:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bed0b2f72ddc4579aea2fe8647b060d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d592c027ab4e15a614834d0f8e7cc5 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e5111cceca44e408a9cea313684221e : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59114e97b1b344e49b805476f896fa2f : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ad5f1027d57438e83a99f1616189204 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec99721f527b4bb38a57a7ecdd52ea9c : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ad7e3d8660446e0b09ac344827d9f0e : SUCCESS in 4m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b0497c66e5ec4d8f9ccccc64320d4b92 : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/064a915481394e5dbd1ba6d9cfc72049 : FAILURE in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b286516f6dd627689cb224446e2638300c031bc4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-01-10 15:43:52.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"709936c76c77f0f4abdbf6b3406cc46f9567305e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-10 15:56:39.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8cb54c26548f8510558f2690b0cf9d56bfcd5022","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-11 06:01:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"34aa82fa64e457adb7aa5043714258a25483d8d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 06:10:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1b830f04f644028a67c967a6c033bc0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/524bd5454ee8474db0aeeac76854684a : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b34d3304a8a46ffb63b314c62e2e160 : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e6c065281c94d4c88a50b04bf036870 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/949d385215da49b28a70d7d2047dee6c : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/574c4199a4db4bcc9756b4f21e75f73a : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2c401a17e504c8d8bf8a29da70e82c6 : SUCCESS in 2m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4e2aae5f72d2448cabf08d06311f8c91 : SUCCESS in 5m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/62ad502caf3642829eac86e5f11e274d : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f90e467935c26c1d4bbbfd036ee5b958093f6b4a","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-11 06:56:47.000000000","message":"Removed Workflow-1 by \u003cGERRIT_ACCOUNT_29244\u003e\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":4},{"id":"269f887f0be170ec063c2b31b4ba8e8ff1799b5b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-01-11 07:24:17.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9bb5694875f515d38d5ef0d9230523cb48a4deca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:24:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"42c9662235463bc5b77f5af2c91a24ce162f4bcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:29:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac3bfebdf6cf4bb8ad5913a6e395de9d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb73ea4d562045df8e46d046ea874749 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d020dac4476d4e03a4a770e891abac18 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcd4a86fee2d4f99be9ef7ebcdb50d71 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3f1111743a43979a02cef37c0e07d2 : SUCCESS in 4m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0acf6f8d470341daa7254b29832d6c6b : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":4},{"id":"ad6250c01af7c5ba5716fb28e4ead56894c421ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:29:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"cce6852b31afaa97f83fe1163dd77189c89577b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 07:30:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/633e33213809444fa2a5150ea54bd23a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c57c685f0bf04a6d9da68144075677f6 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b05d1c7eb842829868016b35b6f3bc : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"daaef2951b5eb03314582bf94a91b1c3dc097c8f","revisions":{"59f8c1a93771a4e9eb9ab7ccbf5719d39bbf8b5c":{"kind":"REWORK","_number":1,"created":"2022-12-31 08:03:28.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/55/868955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/868955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/1"}}},"commit":{"parents":[{"commit":"84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:02:37.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:03:22.000000000","tz":480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ib1e58d792ff12d3c3b72a001e04fa9b8c62d1ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/59f8c1a93771a4e9eb9ab7ccbf5719d39bbf8b5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/59f8c1a93771a4e9eb9ab7ccbf5719d39bbf8b5c"}]},"branch":"refs/heads/master"},"ffdfceefad7ea53bce538611ea4b058de7b81278":{"kind":"REWORK","_number":2,"created":"2022-12-31 08:32:36.000000000","uploader":{"_account_id":32921,"name":"Reason li","email":"liyou01@inspur.com","username":"liyou01"},"ref":"refs/changes/55/868955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/868955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/2"}}},"commit":{"parents":[{"commit":"84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:02:37.000000000","tz":480},"committer":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:32:16.000000000","tz":480},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ib1e58d792ff12d3c3b72a001e04fa9b8c62d1ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ffdfceefad7ea53bce538611ea4b058de7b81278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ffdfceefad7ea53bce538611ea4b058de7b81278"}]},"branch":"refs/heads/master"},"773d35956f34e47ca25ea0c9226ba8f2d96b1eeb":{"kind":"REWORK","_number":3,"created":"2023-01-10 15:08:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/55/868955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/868955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/3"}}},"commit":{"parents":[{"commit":"84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:02:37.000000000","tz":480},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 15:08:47.000000000","tz":60},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ib1e58d792ff12d3c3b72a001e04fa9b8c62d1ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/773d35956f34e47ca25ea0c9226ba8f2d96b1eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/773d35956f34e47ca25ea0c9226ba8f2d96b1eeb"}]},"branch":"refs/heads/master"},"daaef2951b5eb03314582bf94a91b1c3dc097c8f":{"kind":"REWORK","_number":4,"created":"2023-01-10 15:25:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/55/868955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/55/868955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/55/868955/4"}}},"commit":{"parents":[{"commit":"84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/84c6a3b04ebda18f1bf5fe70db50fb6793d4ca20"}]}],"author":{"name":"liyou01","email":"liyou01@inspur.com","date":"2022-12-31 08:02:37.000000000","tz":480},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-01-10 15:24:12.000000000","tz":60},"subject":"Fix tox4 error","message":"Fix tox4 error\n\ntox.ini started failing with Tox4 which had some\nincompatible changes. One is passenv where we need\nto pass each value in newline otherwise, it fails with\nerror:\n\n failed with pass_env values cannot contain whitespace,\n use comma to have multiple values in a single line,\n invalid values found.....\n\nOther failure is due to the skipsdist \u003d True.\n\nFixing tox.ini for tox4 changes.\n\nChange-Id: Ib1e58d792ff12d3c3b72a001e04fa9b8c62d1ef8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/daaef2951b5eb03314582bf94a91b1c3dc097c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/daaef2951b5eb03314582bf94a91b1c3dc097c8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
