)]}'
{"id":"openstack%2Foctavia-lib~881701","triplet_id":"openstack%2Foctavia-lib~master~I2babd600516cfe98673627c68e97cb581ee54834","project":"openstack/octavia-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-02-21 18:49:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2babd600516cfe98673627c68e97cb581ee54834","subject":"Better __repr__() for data model","status":"MERGED","created":"2023-04-27 08:58:32.000000000","updated":"2024-02-21 18:50:17.000000000","submitted":"2024-02-21 18:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881701","meta_rev_id":"b5b2a0be7410ee5c3d7b750f1564ceb91c9f559b","_number":881701,"virtual_id_number":881701,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-21 18:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-08-24 10:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-08-30 05:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-21 07:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-02-21 18:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-02-21 18:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-05-04 11:03:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 09:11:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 07:56:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-08-24 10:17:28.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-08-30 05:37:40.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-02-21 18:15:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4b1d1f0b9159217183bc5b1cd711b770412767e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-04-27 08:58:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6841ce12df00562cb0a31d0c560311f6fbfcae39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 09:11:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365154b541e040df98513f2c1ace1efc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e8544f2718e43308a4ae0452b264815 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c15d4bfc477485fb64e8c4c2e35a91d : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25a24536127e48cf9c647f2f33f567db : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/931d6575a07a418a8169ad6c3ddfd168 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9788d7c57c7342fea4b0255797300e57 : SUCCESS in 4m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/297b18f620f343bda82718c7596da284 : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7561892c82a7475aa172dc829a500908 : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34ea6620b2a3fceef8bf8426753ff23665bd7c8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-03 08:00:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2a5219a17718ef1fcefdc2a6ee989dbc884f09e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 08:34:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0788d876350446d68d1cb48d28759a1b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cda5853923445c689539e71c82bd6e3 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49e0f1c9698c4bbaafed38ac73f9ba46 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a17e8e6d697474481272da10497d515 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8d369fea7b64c8296f0f0e92a5e62b1 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63500d4caf7b4fb8ac72d94389b13c13 : SUCCESS in 4m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7b92b8c4d6d54512a811509d412585c2 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/304f0aae51c046ec9c9dc5fe4a2d7b4d : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"622981f9c0b198d6d1de097b8904eeddc005fe32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 10:24:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4f2d0183260795973044124cbb1633da1ac36d03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 10:26:13.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"514ca168617823af24d729cc59887c0ed37d013a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 10:36:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43fb6d5c034544feb6fa7e9009b6bb3b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2034f0b94691441eb21b08a832176d33 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9e0c45b1c546b2ac4cdeba63802c3a : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/deee0e3cf5fd4b84bca02b6ebc881fbd : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b065cc25e8b0495aad583b36a91f2d0b : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb87c99a90364bcbad31e663bf211f74 : SUCCESS in 4m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3b59ba15dda64f688d92483a0c985852 : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/084845c4f7a6463d89b8cf8c2bb05bc6 : SUCCESS in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d6c4f87a2a6bd703145e13c4e66cb8d3ed5e8ffb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 10:57:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d01060e5047998f40b2394aa57491b19c86d76a6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-05-04 11:03:05.000000000","message":"Patch Set 5: Backport-Candidate+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"021200f2d23c36ad69f6fd72712e695ff37278b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 11:12:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3637b12cd61f41b8b45b100f12d7f946\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00eeb60e6dda4520b7ce966ddacb5adf : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd118fd3ae247e2b26db5f985d4451b : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e637366c65214bc590b390a68f37ed0f : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a78de6ac65d5443fb51f5e566863afe6 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe8b841454847eb8e6120c6f375abbc : SUCCESS in 4m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0d3a4c9c22b44e018769bd5002307b2f : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1f3c8b6042c547d8b68f009df90ab1ce : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"857f0c408370f8bd57abaf435fd17952036ef6c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-06-28 16:13:57.000000000","message":"Topic hsts-haproxy removed","accounts_in_message":[],"_revision_number":5},{"id":"b923b7a398917bf19b915f8c2ed43aa1374030f8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-21 07:56:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ea031656c6f183a689e8b5d6abd11ae05bfc88b7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-08-24 10:17:28.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8868373d8c1f30f1dc29ae7bdab2739bbe5de858","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-08-30 05:37:40.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"438cb760caa7022fe3b863073007a1fff82cf8a8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-21 18:15:37.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ed07cc70dc3685036515c6292679ff94fb8301a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:16:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6822541d7bb25bd375629e274b4bf592152250b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:49:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c459e3ebe9f43cea9ba7aed7ac7270b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de67ea9b006f4364b300a5ca5b29f180 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/152723a8f399443cb1126a9892c4fd00 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc01a2a84f1e40d1a4c03b068f1d941e : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234710d6afb943bea3369354946c96e2 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":5},{"id":"fc0c8232a967e309c8746531dbc2193f45715437","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:49:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b5b2a0be7410ee5c3d7b750f1564ceb91c9f559b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 18:50:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086317f97e9740d1b3d63dd4193f1d0b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34e5f4501b3b4dd49c2ebfaacc3fcd69 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"93c5e17ba8620d098c5cf58335b1f5e8d3b76cd1","revisions":{"a0cd3a1fcf39a9bb6f817540d9259bf741b1d196":{"kind":"REWORK","_number":1,"created":"2023-04-27 08:58:32.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/01/881701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/01/881701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/1"}}},"commit":{"parents":[{"commit":"53200266b31e64730ab81dc54aa1c45ec5073829","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/53200266b31e64730ab81dc54aa1c45ec5073829"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-27 08:55:15.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-27 08:55:15.000000000","tz":120},"subject":"Better __repr__() for data model","message":"Better __repr__() for data model\n\nThis greatly enhances debugability of data model objects as repr() and\nstr() of objects contain now all attributes.\n\nTODO: Check for secret attributes\nChange-Id: I2babd600516cfe98673627c68e97cb581ee54834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/a0cd3a1fcf39a9bb6f817540d9259bf741b1d196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/a0cd3a1fcf39a9bb6f817540d9259bf741b1d196"}]},"branch":"refs/heads/master"},"cc7797ed33640c45498812271b724d25cea79d28":{"kind":"REWORK","_number":2,"created":"2023-05-03 08:00:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/01/881701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/01/881701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/2"}}},"commit":{"parents":[{"commit":"53200266b31e64730ab81dc54aa1c45ec5073829","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/53200266b31e64730ab81dc54aa1c45ec5073829"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-27 08:55:15.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-03 08:00:10.000000000","tz":120},"subject":"Better __repr__() for data model","message":"Better __repr__() for data model\n\nThis greatly enhances debugability of data model objects as repr() and\nstr() of objects contain now all attributes.\n\nTODO: Check for secret attributes\nChange-Id: I2babd600516cfe98673627c68e97cb581ee54834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cc7797ed33640c45498812271b724d25cea79d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/cc7797ed33640c45498812271b724d25cea79d28"}]},"branch":"refs/heads/master"},"9d912eaa4f7dd2b101a2214ca735b0294e1ed293":{"kind":"REWORK","_number":3,"created":"2023-05-04 10:24:33.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/01/881701/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/01/881701/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/3"}}},"commit":{"parents":[{"commit":"53200266b31e64730ab81dc54aa1c45ec5073829","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/53200266b31e64730ab81dc54aa1c45ec5073829"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-27 08:55:15.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 10:24:19.000000000","tz":120},"subject":"Better __repr__() for data model","message":"Better __repr__() for data model\n\nThis greatly enhances debugability of data model objects as repr() and\nstr() of objects contain now all attributes.\n\nTODO: Check for secret attributes\nChange-Id: I2babd600516cfe98673627c68e97cb581ee54834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9d912eaa4f7dd2b101a2214ca735b0294e1ed293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/9d912eaa4f7dd2b101a2214ca735b0294e1ed293"}]},"branch":"refs/heads/master"},"ea078f9d85b4907c3fa3e7c5d2ff5fa41ed0fef2":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-05-04 10:26:13.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/01/881701/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/01/881701/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/4"}}},"commit":{"parents":[{"commit":"53200266b31e64730ab81dc54aa1c45ec5073829","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/53200266b31e64730ab81dc54aa1c45ec5073829"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-27 08:55:15.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 10:25:01.000000000","tz":120},"subject":"Better __repr__() for data model","message":"Better __repr__() for data model\n\nThis greatly enhances debugability of data model objects as repr() and\nobjects contain now all attributes. Most subclasses implement their\nown __str__() so it shows only the most relevant attributes.\n\nChange-Id: I2babd600516cfe98673627c68e97cb581ee54834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ea078f9d85b4907c3fa3e7c5d2ff5fa41ed0fef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/ea078f9d85b4907c3fa3e7c5d2ff5fa41ed0fef2"}]},"branch":"refs/heads/master"},"93c5e17ba8620d098c5cf58335b1f5e8d3b76cd1":{"kind":"REWORK","_number":5,"created":"2023-05-04 10:57:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/01/881701/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/01/881701/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/01/881701/5"}}},"commit":{"parents":[{"commit":"53200266b31e64730ab81dc54aa1c45ec5073829","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/53200266b31e64730ab81dc54aa1c45ec5073829"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-04-27 08:55:15.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-05-04 10:57:18.000000000","tz":120},"subject":"Better __repr__() for data model","message":"Better __repr__() for data model\n\nThis greatly enhances debugability of data model objects as repr() and\nobjects contain now all attributes. Most subclasses implement their\nown __str__() so it shows only the most relevant attributes.\n\nChange-Id: I2babd600516cfe98673627c68e97cb581ee54834\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/93c5e17ba8620d098c5cf58335b1f5e8d3b76cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/93c5e17ba8620d098c5cf58335b1f5e8d3b76cd1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
