)]}'
{"id":"openstack%2Foctavia-lib~892417","triplet_id":"openstack%2Foctavia-lib~master~I9079030a5fef9dc44da242adab3c568666777451","project":"openstack/octavia-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-09-19 06:27:11.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-19 06:12:02.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-08-30 06:37:30.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I9079030a5fef9dc44da242adab3c568666777451","subject":"Fix a possible receive timeout","status":"MERGED","created":"2023-08-23 00:14:44.000000000","updated":"2023-09-19 06:28:06.000000000","submitted":"2023-09-19 06:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"892417","meta_rev_id":"1c024b603e043668455689da3c3224b36f674b1a","_number":892417,"virtual_id_number":892417,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 06:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"all":[{"value":1,"date":"2023-08-24 08:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-19 06:12:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-08-30 06:37:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2023-08-25 05:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-19 06:12:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-19 06:12:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 00:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-23 06:33:34.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-08-24 08:36:45.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2023-08-25 05:52:12.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-08-30 05:38:10.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"07f641beb5cdca3821578903589768a06821b8cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 00:14:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"561602c7388f0c3d129809dcedace5d79c82f39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 00:29:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc75c1e20eb141308682d99ffe499c7f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50627c4f85674b3a9c1dda510a8412dd : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f061719fca848eb9a0e964b59e5ab77 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ddd678a65e94f6994ffa33f65ad011d : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fa037d33fbd4171baed7fb5407d2818 : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cb7f5c2924a4727b21965030e6d4901 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f16e1e0936cb414484ecfa0f46f8ff06 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fc37224efdb48f0ad29310f08931626 : SUCCESS in 4m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5ff1daa5f4a482b9c1ce42470057aca : SUCCESS in 2m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b7c31f6a127d44e6bd290e6993fdf8b6 : SUCCESS in 4m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b3e5907f87bc452da66e4433d694ada8 : SUCCESS in 7m 52s (non-voting)\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/071180ebc8444f13befb5f9e5a06dfd9 : FAILURE in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"1b9d0091de126d2942495dfc8cc4ebf6d229035a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-23 07:11:15.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"35287c0d00a238d62d945b8589aa0c06bcf8250d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 16:44:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"87f834efcabd247b97f27bc70ad2e0c80771761b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 17:00:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7079b158cce44a99830dac635c792147\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b3bfb4843364df492e0f8a2bee02d0f : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4babaa098b49aa96f5d7a08ed5e4ec : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c0faef414ec47b1877c6341a1c8f91a : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e787b7702dd34002b3cba9a4367f7743 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ec80276c1c44c6f83dab93aba244a03 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69d1dff1b58841e5b8bdd22a63f9a840 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c597dc264a74fbf9055bf9b1d32b007 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1adaf52b3e0d49dfa4ac3b4cc15cd6e7 : SUCCESS in 2m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/55e5b228f5f44c6e95732b72c13184da : SUCCESS in 4m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/862300658f03478fbb3c611f8acbf557 : SUCCESS in 9m 17s (non-voting)\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b96a918c6e004ba89a787bcc8fa914e9 : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"de9385375c18a7fc6b8fd7dab1bcd284661eb975","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 19:47:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7108b19df656f7961563a4209531c75d4fc46548","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 19:59:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4735efeda4134da1b096f2411343675a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/362244e768b04577b7edde2ce5d2ecfe : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8c69e47d33411fb56e4358226354e4 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be28c5d68849439682c99ca4c42b9d10 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f8b537d3b745c3ad5d87505867ab52 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edac02e944e44c6fa28364510b83047a : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94c354f10757488ea5172a53cca4c03b : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65863be554414ea78ccefffe1513649d : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5a022f273314c93860f337555f4cfe9 : SUCCESS in 2m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e46078f09a4e497ca7ab41149fa35176 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/628865cf7a1446a18eb2d09d5243ec8c : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4f8c7c4b8d378734076e9c78357d303419a7bc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-24 06:26:26.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b47be1322865c075b46ad1965a08546a63950e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 06:38:20.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/974bcdbf347a4ebba5613802fe322a0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ae41952f65a4b518f87c1025973b7be : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bd6e1857fb450fa09e16cdd0b9bfe8 : FAILURE in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f500b2660500435e84c45c52488a0518 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee140d758011437da25d81775f9b75b9 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a54570277c394f368cdc511719f53cf7 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1abbdc3042e4410b9a27d782d45e91e8 : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba9e4c4db1e4de692a383b3f1e1aad0 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c41ce0d511cc4192a369c460fcce2617 : SUCCESS in 2m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0c1aa47b93d44fa792e2eae47df85e0a : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3b3c52dad93a42189d00d3af3b434f1b : SUCCESS in 8m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1e570d42a62b972256d710bec5fc59943f4aec75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-08-24 06:56:52.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2529d74918e0c90988612e766e3e95f3208ddbb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 07:07:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5871ea2703ba46078919a58095d82111\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/302327431c5743d89b7d5ad7dad5fadd : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5849ab87fcb346ab8b1ce0d7e732222b : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc144b7c7ba34233bb019e6055ab0461 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecde5ddb627b4ab8875b6bb2835f130a : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16ec555670e045b49eed9ee30fa8ca98 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8670f43199bc43508e2837404b079c4d : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff4101496d34e1a9be01ccb8dd7ad73 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a91290dc2ee84be9aad3ed4385738d49 : SUCCESS in 3m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c66aa438b2e348a5a93483a05672c88a : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bf355150fbd14513b0f9e69fe504ab4a : SUCCESS in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"90fd6293614db811bea51e22bc9015b12450c4cf","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-08-24 08:36:45.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b4fbf45196abc6ce47ddfa9702ed32fe5a109a34","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-08-25 05:52:12.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"63c327e529d6e7c0972e022b5b331367d6a5b773","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-08-30 06:37:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c214fb45b4e06570da40970b8ca610570d3fae63","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-19 06:12:02.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1fc9a1f7e8b2c282fc6688f2df49bbe741b835f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 06:12:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cd63a0d0404f165b56d36fdebf8ba49cd6768331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 06:27:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a39d28082f046f9b3ea7cb4edb43eb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0857ee29b3748f38c5c750a5e06b025 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7baf51260a284dc7b8cf1fe6baf02221 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ca4982a202b447396eaa6256575f9a4 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1adc255639d4045a54397391e2ea836 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf9f51c8cdf4117b5fcf624fefe4dc5 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6892a87ab2eb4fa4ac1b2b0b5d06e46c : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a609be848eecf5bd079153f73b08e19a650a2b12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 06:27:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1c024b603e043668455689da3c3224b36f674b1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 06:28:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d1def6d4ac4500bbc05ae813312181\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177236e347714e028f2304afe7b58a38 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f22d6d21c2c94e46aa45c38457020144 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a84a218ef1f683d5584784b7d9607e7453b0729","revisions":{"1b0fbbae2218c53bec6a9121097f99322f15ae3c":{"kind":"REWORK","_number":1,"created":"2023-08-23 00:14:44.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/892417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/17/892417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/1"}}},"commit":{"parents":[{"commit":"bad19004b09dd6daf1b6db0a6f96545790a08336","subject":"Add support for HTTP Strict Transport Security","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bad19004b09dd6daf1b6db0a6f96545790a08336"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:13:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:14:43.000000000","tz":0},"subject":"Fix a possible receive timeout","message":"Fix a possible receive timeout\n\nThis patch fixes a possible receive timeout caused by a slow response from the\ndriver agent. For example if the database is very slow.\n\nChange-Id: I9079030a5fef9dc44da242adab3c568666777451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/1b0fbbae2218c53bec6a9121097f99322f15ae3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/1b0fbbae2218c53bec6a9121097f99322f15ae3c"}]},"branch":"refs/heads/master"},"b4c210ac58a6da0936c906d076fe71529e6d98f4":{"kind":"REWORK","_number":2,"created":"2023-08-23 16:44:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/17/892417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/17/892417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/2"}}},"commit":{"parents":[{"commit":"bad19004b09dd6daf1b6db0a6f96545790a08336","subject":"Add support for HTTP Strict Transport Security","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bad19004b09dd6daf1b6db0a6f96545790a08336"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:13:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 16:44:21.000000000","tz":0},"subject":"Fix a possible receive timeout","message":"Fix a possible receive timeout\n\nThis patch fixes a possible receive timeout caused by a slow response from the\ndriver agent. For example if the database is very slow.\n\nChange-Id: I9079030a5fef9dc44da242adab3c568666777451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b4c210ac58a6da0936c906d076fe71529e6d98f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b4c210ac58a6da0936c906d076fe71529e6d98f4"}]},"branch":"refs/heads/master"},"b2c3453b138fada741c1ec8bba213cb3e641db1d":{"kind":"REWORK","_number":3,"created":"2023-08-24 06:26:26.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/892417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/17/892417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/3"}}},"commit":{"parents":[{"commit":"bad19004b09dd6daf1b6db0a6f96545790a08336","subject":"Add support for HTTP Strict Transport Security","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bad19004b09dd6daf1b6db0a6f96545790a08336"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:13:39.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-08-24 06:22:54.000000000","tz":-240},"subject":"Fix a possible receive timeout","message":"Fix a possible receive timeout\n\nThis patch fixes a possible receive timeout caused by a slow response from the\ndriver agent. For example if the database is very slow.\n\nCloses-Bug: #2032890\n\nChange-Id: I9079030a5fef9dc44da242adab3c568666777451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b2c3453b138fada741c1ec8bba213cb3e641db1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/b2c3453b138fada741c1ec8bba213cb3e641db1d"}]},"branch":"refs/heads/master"},"2a84a218ef1f683d5584784b7d9607e7453b0729":{"kind":"REWORK","_number":4,"created":"2023-08-24 06:56:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/17/892417/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia-lib","ref":"refs/changes/17/892417/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia-lib refs/changes/17/892417/4"}}},"commit":{"parents":[{"commit":"bad19004b09dd6daf1b6db0a6f96545790a08336","subject":"Add support for HTTP Strict Transport Security","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/bad19004b09dd6daf1b6db0a6f96545790a08336"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 00:13:39.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-08-24 06:38:11.000000000","tz":-240},"subject":"Fix a possible receive timeout","message":"Fix a possible receive timeout\n\nThis patch fixes a possible receive timeout caused by a slow response from the\ndriver agent. For example if the database is very slow.\n\nCloses-Bug: #2032890\n\nChange-Id: I9079030a5fef9dc44da242adab3c568666777451\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/2a84a218ef1f683d5584784b7d9607e7453b0729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia-lib/commit/2a84a218ef1f683d5584784b7d9607e7453b0729"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
